)]}'
{"id":"openstack%2Frequirements~1002117","triplet_id":"openstack%2Frequirements~master~I74e9af76e5076eed0984fc54061202efdd1a4abb","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-08-26 16:08:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I74e9af76e5076eed0984fc54061202efdd1a4abb","subject":"update constraint for oslo.versionedobjects to new release 3.12.0","status":"MERGED","created":"2026-08-24 14:14:25.000000000","updated":"2026-08-26 16:09:38.000000000","submitted":"2026-08-26 16:08:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002117-new-release","meta_rev_id":"fea8224af9d1f32e9880b1ba6274b1bfa493c375","_number":1002117,"virtual_id_number":1002117,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 16:08:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-08-26 13:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-08-26 13:24:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-24 16:05:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-26 06:22:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-08-26 13:24:29.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"50cb7a6e8369167859716573e8f371b14cd951d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-08-24 14:14:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1904ce8e73f40c6ea6631267c428e3944ee0cd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 16:05:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7ae8b4c74f040d58c9011416a7754c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5ecf8587ab46eb92005ac15c8f94ad : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33d829c92d0f4e0ba04e1c3e9935cfb4 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/529e780bfe8d496082314ec58c4c4472 : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/569f615df0c84c9fa91ab0cbd0dd7ae3 : SUCCESS in 2m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b9967129925a4af2ad666674ebaf574b : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06107267de37487cb227c0bd7159216c : SUCCESS in 2m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6c6cdad025f4f658914b611f3fbdb9e : SUCCESS in 3m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4a1480f69d9e4e55aec6b101b4f15d61 : SUCCESS in 8m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7555ecf7d1164efba06f8c06fd3058be : SUCCESS in 5m 21s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c9218a68995644b1883f0b3af82796af : SUCCESS in 5m 14s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/fd1400c96af740769bdfd2caaee5a0f6 : SUCCESS in 11m 43s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/64c17562ce2b4870a5545408151cebba : SUCCESS in 4m 42s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/ba739c8dd7544f2092e8cd0f466302ec : SUCCESS in 4m 17s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/6d86702d8172461ca001a1c9e16bbe97 : SUCCESS in 5m 54s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/130165792bf245b4a93245f660342557 : SUCCESS in 10m 36s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/6ec4e57cf2b5460e9d014eb9baa16ae1 : SUCCESS in 5m 54s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/e361516a19a04d64bf2af02b5ffd21ee : SUCCESS in 5m 13s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/63393a1b68624f01874e26dcb1d039e4 : SUCCESS in 3m 47s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/1d3094eaee8f43b08ff2438a188cf81c : SUCCESS in 7m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d17f6ab12a04d459f731f2063fd2544 : SUCCESS in 4m 17s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/0daeb2f5062e4d238b3d1cf0caa80412 : SUCCESS in 8m 44s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/5352b0ea3c1045c38e12a71444e85969 : SUCCESS in 11m 21s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/e23dc7204db041af9077548375914ed3 : SUCCESS in 2m 30s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/7ca367ebb93d461dac6856be11705315 : SUCCESS in 8m 03s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/25916238a352421998813477dcf17f9c : SUCCESS in 2m 35s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/48a5475f16534d36a476fe0ab440d47e : SUCCESS in 6m 49s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/a7ffa20ae5bd45d6b96424c26a7905db : SUCCESS in 10m 58s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/b65ac5452fa34e2a9bc6d613d99e80aa : SUCCESS in 24m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d6e09604efff40538e49ef7655edeab2 : SUCCESS in 6m 51s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/5448b79ac77840808634fc12b53b2d26 : SUCCESS in 13m 50s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/9a91978e25f9420a9e69a1d64640ed97 : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5954c058ae354f9d8a322d4ef52b9e6f : SUCCESS in 2m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54d65eeadaf242348f43b3cc6ecaa7b8 : FAILURE in 26m 57s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/182ab3a471bf461bbcb34e1bb957606b : SUCCESS in 6m 02s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/55118071725344538434d5b4fcc4e75d : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/61f80b0993a64ab59beefc273e56fb48 : SUCCESS in 7m 05s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/dd432f88b27c4beca0c3567d5b2d2ee6 : SUCCESS in 19m 01s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d37adbe9b6c64bb38a5192624da1e400 : SUCCESS in 5m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f51074df23f4788a903680e42dbd7b9 : SUCCESS in 1h 01m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb4fb4295bde4da990f2bd825e12bcb3 : SUCCESS in 1h 37m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e27a45de3181d3c61e4dff61a1c78b9acb87c73d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-26 06:22:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d712325bbf11633cc14fd36127cf90d798a5afa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 08:13:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb1a83d3ee5e4a1aa0ca73855e0f6ad5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/863ad86b74fc4210a427e55431ad8528 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a2ccf40492d44aea713fbe9337d4af1 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/408b339a8cf9420eb0816fa87feac866 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2665db50c1af4948b20b724568bed939 : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7e746a4c5920489fa6432b5928abad06 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e337c03aac9e4eb598bc0429bf6d3615 : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9cfc8a9dc510417b9008e4c7eaf6aba4 : SUCCESS in 3m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a57dd986d56446f4ad3565d19a3e8a26 : SUCCESS in 5m 38s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8479b85035c145d6ac7663c0b90ffab8 : SUCCESS in 10m 14s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/91c4c8a868e24ba9bd0597718129590a : SUCCESS in 6m 04s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/3a2f28f4b29f4101a7d333963ad6f4c7 : SUCCESS in 9m 23s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/ca11e6e2889348318445cc1440de28df : SUCCESS in 6m 11s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/27a40e05df30402496d1245a8285722b : SUCCESS in 5m 05s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/1378ee1b27f1483882860c49feb40adf : SUCCESS in 6m 59s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/fe7fa44c472245d5899a1c6256fc3847 : SUCCESS in 11m 34s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/d87414ffbe8a4210973377b92fd326a6 : SUCCESS in 5m 43s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/e5c01931fcef4b7195dcbd374f1849f5 : SUCCESS in 5m 05s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/6d4bc1ebc22d4a9cb81df2e817b28cb0 : SUCCESS in 7m 01s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/1668c1515b2f4935adfe3792729e75c8 : SUCCESS in 18m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7fe7a8e0240143b8a69e0a6bd17399c2 : SUCCESS in 4m 56s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/55ac3602fd5d43f2b8cae1d8ce0ee6ca : SUCCESS in 8m 29s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/65f085f7ff0a460f9cff85a716e615fa : SUCCESS in 11m 01s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/5fd6d5b2135f4ac0b48e09b3fb575d6b : SUCCESS in 2m 27s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/24b965467c7d48a78cb543743c063948 : SUCCESS in 8m 08s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/aeeb38aea5954ed9a38682ec493c2742 : SUCCESS in 6m 46s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/e05527ca80594c37a7378dad5388d59b : SUCCESS in 5m 36s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/9ae7b9d72236446bb3041e08f5a8aeb0 : SUCCESS in 11m 07s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/a04db23efb43410ca89c0a344819793b : SUCCESS in 24m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/76d05c50f74a4a3e869ba8ee71012621 : SUCCESS in 7m 06s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/4f35e6ef61db430cb7624474648802c8 : SUCCESS in 17m 10s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/3f4a49fdf22a46bc94b5a817c18718b9 : SUCCESS in 4m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f25bf9fdac83456989daef3812a1d845 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19c1939a05d04e9881eaecb7a184392d : SUCCESS in 23m 37s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/37cb5ba00c184ff5a77b5cfa7cfc13bd : SUCCESS in 6m 01s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/e279854155b84a02b0b0249d174ab8c3 : SUCCESS in 2m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/10f5635e4c904da09c718d36427d2e47 : SUCCESS in 3m 50s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d811af26069d49808651625298fd35eb : SUCCESS in 13m 59s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/7144c313e1fc46ee97155a50c38b2558 : SUCCESS in 6m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ede052faf5d24aadaac34ca121787a75 : SUCCESS in 28m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8b9bfad2e2c4f16b1d06a33568d2f16 : SUCCESS in 1h 39m 35s","accounts_in_message":[],"_revision_number":1},{"id":"81a5cf3aff30eb87a8613fb764cfacaa46451514","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-26 13:24:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"422f5ab33f33762ddf95b60b8c7bbb00da78074f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 13:26:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1555fe662164f917ff60ee763b8d80e01cdfa34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 16:08:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ce9ce55f3e14e768fab7023d4fb1c19\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b91bf234de04615ae6206cbd1f1d777 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1151812075554d3ab0897d9f61a58902 : SUCCESS in 2m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c37edfbd1404da493b08707c3896a55 : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/694d64272cc24957a61fde029029b95a : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0b15e333724b44b9b9aa73a2fb7bdc53 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8199dcdc16734cf3ae830e4456366450 : SUCCESS in 3m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4eed8ff0684e4a688151afd48dc2405c : SUCCESS in 4m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c4255130467f42e8830c7959ac2122d9 : SUCCESS in 4m 08s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/da6a7e9175d747c3a2049d7b54fe3d98 : SUCCESS in 9m 40s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/81c7fde6c73a46919ab221e3de9891ba : SUCCESS in 9m 09s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/2dffd8053fad4a93af03392dfc020b2e : SUCCESS in 4m 22s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/d191d851c03646198bbc8ed5b3394dec : SUCCESS in 3m 46s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/0302470bc062407b8e3d2cdb4e53764b : SUCCESS in 5m 14s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/c0d4f0f9b4ea4a66b2ec1b3721eb64d9 : SUCCESS in 14m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d6a3da8073949ed89cd395f12bdf93e : SUCCESS in 4m 02s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/92dc729fc57c4401947e6c3e3106818b : SUCCESS in 8m 53s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/6059539c94d74f38bd0357aaaf81bc4d : SUCCESS in 11m 46s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/8e0cab99ca854a2583f1747bc84eedbe : SUCCESS in 2m 06s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/1c980a3e453040fbba9a33f7ab63a18f : SUCCESS in 6m 10s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4fe20beb20f143f29a3d0c76ba4b7a2d : SUCCESS in 22m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e8aa1b05e68244e9a1c763bfd099f78a : SUCCESS in 4m 49s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/bd75e9fe16374a7eaede66b220067742 : SUCCESS in 14m 15s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/1a5ea28ce5054062bde50b08d1d6b777 : SUCCESS in 3m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2c32bd7b579a4c359a255772157bad14 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7e967e1df95943c380d019afef0c2660 : SUCCESS in 26m 11s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/35cca7e43b404181acf1a8194d171786 : SUCCESS in 5m 17s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/5e53edebef0f47f189ee60e02a180955 : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/377ddce0cd3842d383ce4f4d11fd4230 : SUCCESS in 8m 08s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/df91cb18c61e436199a1e0b006d9aa7e : SUCCESS in 10m 23s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/85bc750c3c2c4f17a7d4c6698bb14fa0 : SUCCESS in 6m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17b0157f1ef94b49b354bb8a9f70378b : SUCCESS in 1h 06m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44f90c44993943a797e883cdc1f86dbf : SUCCESS in 1h 20m 18s","accounts_in_message":[],"_revision_number":1},{"id":"27d538c5159072b3355e160c5f6e2495b0b0dd3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 16:08:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fea8224af9d1f32e9880b1ba6274b1bfa493c375","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 16:09:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff1fd9f05221431f93632edb303fd291\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c98b12b46d3c4224bb841e7b31103d16 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d54b56ebad1cf361472d01828cfc56f32bd31392","revisions":{"d54b56ebad1cf361472d01828cfc56f32bd31392":{"kind":"REWORK","_number":1,"created":"2026-08-24 14:14:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/17/1002117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/1002117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/1002117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/1002117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/1002117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/1002117/1"}}},"commit":{"parents":[{"commit":"3353677727f8351bc58ba98efafe6bc884a0257c","subject":"Merge \"update constraint for os_vif to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3353677727f8351bc58ba98efafe6bc884a0257c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-24 14:04:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-24 14:04:22.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 3.12.0","message":"update constraint for oslo.versionedobjects to new release 3.12.0\n\nmeta: version: 3.12.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Change-Id: Ia944ec98d2ab3412a5d592c2165cbb490ddba96a\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I74e9af76e5076eed0984fc54061202efdd1a4abb\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d54b56ebad1cf361472d01828cfc56f32bd31392"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d54b56ebad1cf361472d01828cfc56f32bd31392"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
