)]}'
{"id":"openstack%2Frequirements~1003192","triplet_id":"openstack%2Frequirements~master~I927d3cd4a8179f78aa86ba5c51fd4c2655048713","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I927d3cd4a8179f78aa86ba5c51fd4c2655048713","subject":"update constraint for openstacksdk to new release 4.20.0","status":"MERGED","created":"2026-09-01 00:40:33.000000000","updated":"2026-09-01 09:04:45.000000000","submitted":"2026-09-01 09:02:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003192-new-release","meta_rev_id":"af88a64cd4d71b20913191ebc4415d0bb6ed6bdc","_number":1003192,"virtual_id_number":1003192,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-01 09:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2026-09-01 07:30:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2026-09-01 07:30:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}],"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 02:10:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-01 03:01:27.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"CC"},{"updated":"2026-09-01 07:30:38.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"7b120d8a88abf69f3a67924d6d1329cd757d4f5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-01 00:40:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b23f600f7598b919d01fa64b187f15c1707f052f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 02:10:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fea140ba56744127b34dc17b0ca3bba2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a55f1a18f64fcdab96ac973c1f913c : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead6d1e98eb6475da5e86ac66538e506 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af235e53d92b41f5a1fc64ff5ceb993f : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/54c04a797979403b9d2e686954618312 : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3e79b08b6803411ebd437c2f63226a61 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/346e499cfd37416fb790257466ffe846 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ecdbd63c6c2649f3b085c445e9228c84 : SUCCESS in 2m 08s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7474c65341d44d0484bc4b9957780abf : SUCCESS in 6m 32s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/60370d5fd5774f92ba26683bc13865bb : SUCCESS in 6m 09s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/97a5173030db4ab386a8d184fda3eb5a : SUCCESS in 3m 13s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/b164c8be311b419590b3ac8f14f42212 : SUCCESS in 8m 17s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/6cb99c304f5d44a09d27f063b4870e88 : SUCCESS in 3m 47s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/5e3154c6a2d84615b8d32e464c457c06 : SUCCESS in 4m 13s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/260fe420b4834177ab8988658da28a72 : SUCCESS in 4m 14s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/f6c09d9c9f574e84b17e5cb44f9037ff : SUCCESS in 6m 40s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/a9fe5bcc0ab142a59ef68e1655783531 : SUCCESS in 5m 52s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/6c3ddd82a32948c69ebe54e96888d084 : SUCCESS in 6m 00s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/34286f4ca29842818b149e8e5f52058d : SUCCESS in 5m 13s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/0c90dacf2abf45eb96e5233566b451be : SUCCESS in 18m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/69e3c6714f514c87b7bf6046d6cd3de0 : SUCCESS in 7m 56s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/7c87b4a22460495fb56a08cfd40a3884 : SUCCESS in 6m 17s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/d977e9b1b7aa4da4933be314f0f9da04 : SUCCESS in 11m 26s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/8d9f92b9b1844d5ab776a8f72fbacd58 : SUCCESS in 2m 20s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/e3050f148aea452ab30521f0c2a8d326 : SUCCESS in 7m 55s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/715e0a79ca8d454db943bc8158965058 : SUCCESS in 3m 42s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/ffbc28266cd741d4b8cb4da5a151a020 : SUCCESS in 5m 25s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/4a1b40e8a1b34c529f4c0298fe090faa : SUCCESS in 12m 00s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/705139a1e0ec4f14a7749828d984f72b : SUCCESS in 23m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/852c1ab3529a48b783acd25319b16c4d : SUCCESS in 7m 05s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/8cebc2e578214a9081a82f0c0fe0fa8a : SUCCESS in 14m 55s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/4e8acae8cdd040798d37b2387ccc6384 : SUCCESS in 3m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e4577021196b42a8bd48c90fafc267dc : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9936d0f437b4f5fb50a6ab938a12642 : SUCCESS in 21m 24s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/eaf30f228a364f339b4b4e911e65a831 : SUCCESS in 5m 10s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/47284e4ecb324d23ad85c1f2a13d42a3 : SUCCESS in 4m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0974e55a629e4fb398a4fd4a64e88906 : SUCCESS in 6m 30s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/bb8aaaac89e44f77b296eaab288c102c : SUCCESS in 18m 21s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d17231e675df41d5a074b4c50b9b6103 : FAILURE in 4m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfbdde2e3b694ea8831a605dd9f9f20b : SUCCESS in 57m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a9689d1dc26479c83bef8ef961d13d2 : SUCCESS in 1h 27m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5f6eaf4883645b7c45dd89b39709b742d06b1536","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-01 03:01:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13a73495b9afd44efd48d17e8cad46e7d0b8846b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 04:43:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dd7865b35bc4810b2e2a5a4d75160d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f4d6dee684640d5822d67fa752b2516 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/171f747b1e6f49ab88dc31980200cedd : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6eae16d7396b4f41ad7746133cc5205c : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6607534f1cd043e8ad4dbf430aff2f40 : SUCCESS in 2m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/39fa7bc1c1884c0bbb7e10f23928fa0d : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbbdc010820c4866899814cdb621ed2f : SUCCESS in 2m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/16455093680a4aee94b54f112be165a0 : SUCCESS in 3m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a95ccd076c2f435d84904025133d6678 : SUCCESS in 4m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/55042910b7244b179717add859358229 : SUCCESS in 4m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b2457a28236a4b0c8dd309cab7545888 : SUCCESS in 7m 58s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/fb2dfa8d3b91466c899463e9e9d21cf9 : SUCCESS in 4m 56s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/effef326ca5f4aae8d88699b59f3aaa9 : SUCCESS in 4m 20s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/d91bbcda4c81465d82b614c26b0b48dc : SUCCESS in 4m 49s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/23acf8f99f824cdeabdfac831321414e : SUCCESS in 3m 49s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/a3911ba64e2f44988cdd8d4e54c49edc : SUCCESS in 10m 31s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/35d15ebd4ef64b668d27cceeb5fc367d : SUCCESS in 4m 40s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/1f7058703a7146d5812201fa13936035 : SUCCESS in 7m 33s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/9b686d174bd04268a5085dee7fabecf8 : SUCCESS in 5m 00s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/8d2b6b1aa13641b4bad5fcbe4fad4c55 : SUCCESS in 16m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c7c75ecbfa4e43e58593f63ecf73e82b : SUCCESS in 7m 31s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b5848b26c35346d783316e6dc6a28fc1 : SUCCESS in 4m 58s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/8e820bb0d0ef49c0905ce2e3001bc962 : SUCCESS in 11m 56s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/cfc4b42d65b24356814e55ea2613950c : SUCCESS in 2m 52s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/f9b11290b0e240b6834375227631504c : SUCCESS in 5m 20s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/52675d37888f4cb398361cec0f85049f : SUCCESS in 3m 42s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/e7215b1600d145bf9338e9f33399fd99 : SUCCESS in 5m 35s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/edb122e75bbf4029926ec27f9336f65a : SUCCESS in 11m 30s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/8f01195989824b42a77f3657c708e551 : SUCCESS in 25m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/399655a6980042ceab25ce734157be48 : SUCCESS in 7m 17s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/6caa4ab5962d411680465d8cbadf65de : SUCCESS in 9m 02s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/130c96de35fe4f9baf93a31fd0fa884b : SUCCESS in 3m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0e24b382ce834caf965af95c41438b8f : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e8113c60d534f5c9b810dcac14a3bf0 : SUCCESS in 28m 21s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e695e219221e4f1282c7468c4aa7b821 : SUCCESS in 6m 16s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/046da898f9f44750b6736fe1a5682d1b : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96631a1228e94b9e95fd9f82238b7a73 : SUCCESS in 4m 33s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/14c6316cda05478e90d49fd225c6cdc2 : SUCCESS in 19m 49s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/c9b7b0a9858446ee862f7a1575133cbb : SUCCESS in 5m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/692122ea65334a7c9cd4fe2a3f6c3633 : SUCCESS in 1h 02m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/168209e642c045cb98c856dfced70388 : SUCCESS in 1h 28m 51s","accounts_in_message":[],"_revision_number":1},{"id":"704edd8d56ce11291af250688e8b451a2903f417","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-09-01 07:30:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29e2e0b47e85b5e5596b7f3efed5fa2a04ab1f14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 07:31:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6089b4af65428651789ea26d443dfa9316a2f91a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 09:02:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7503f09434b0440bbcbf1b10bed63877\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cbab229e4c4e3ba1028c5bc256d877 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c05e204be4bc49738e2f656da45d2606 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9d388b2a799243b7b3fe2830440e57be : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d887832982b344738e827752a36b66c3 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a68d32398a554eb899b858b17ac67217 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e12946c8352414c936ba32cfd3a7793 : SUCCESS in 2m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7d7db590f2234c0aae4f21bf2309f534 : SUCCESS in 5m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a347793ff2a94b558acaa869d9b8a969 : SUCCESS in 4m 59s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/61a61b199d5e4188b00812ac994ca8f5 : SUCCESS in 7m 50s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/8f0fd6d8a4b04817961df82b41c14154 : SUCCESS in 8m 03s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/76ad7d89070842ad9fe708ded10789d3 : SUCCESS in 4m 54s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/51a0f7a2497f46ffb769e399691b5870 : SUCCESS in 4m 39s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/ef5be810f4bb488bb7175729c73776dc : SUCCESS in 3m 26s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/194ce4fec2994c21928cc0266f393723 : SUCCESS in 13m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fedbe375c39c42eb829de6b26a1565e7 : SUCCESS in 7m 54s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/654b7dfc461e41bd935fe00040e73b62 : SUCCESS in 7m 04s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/881b4e75ab5b469cb6cdc274e5b8dc51 : SUCCESS in 9m 59s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/58b75841d30c426ba00cb74ab62fa7c9 : SUCCESS in 2m 23s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/0583e3ffe7304678983174922ec0b80e : SUCCESS in 5m 46s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4d8a5f2da95849be944005a4db43bd43 : SUCCESS in 25m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9bca305739fc4da6be38893bd2b31377 : SUCCESS in 6m 47s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/3baf91bd962e410eb184efe562b373cd : SUCCESS in 9m 28s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/db2771574dbb452da6d939972bbb08c3 : SUCCESS in 3m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa2f4878b07847639e3c1f82a9b7b424 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1d5b94d7974047d6aa196d611dc9f6bf : SUCCESS in 28m 18s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/38319a2ebf7f446aace71b3ee319f4e7 : SUCCESS in 5m 58s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/16d97e56818146fda0b0c0f1f5079d64 : SUCCESS in 3m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/626fc656206346e99ddafb41f0c39c8a : SUCCESS in 4m 14s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d650722f37f04de0a336593c8ac922e8 : SUCCESS in 17m 33s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/4574132a51de418d85b51809e20cecbc : SUCCESS in 3m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5390c498402c4863b6c41c1665b1dffa : SUCCESS in 47m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f31b42b47d2a41929d41897547448701 : SUCCESS in 1h 30m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ba17dc29555f843cdbe9f66d4756092f7c7b2914","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 09:02:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"af88a64cd4d71b20913191ebc4415d0bb6ed6bdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 09:04:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b8078095497457d8ce8b896853ef91b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f60955fcacd0447391e7c6323d1bf65e : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50a156dd6141fdcd42ab111a58b3ae040e06573d","revisions":{"50a156dd6141fdcd42ab111a58b3ae040e06573d":{"kind":"REWORK","_number":1,"created":"2026-09-01 00:40:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/1003192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/1003192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/1003192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/1003192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/1003192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/1003192/1"}}},"commit":{"parents":[{"commit":"8488704b592ecd80a8287ae51c39375628b6d7b5","subject":"Merge \"update constraint for python-manilaclient to new release 6.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8488704b592ecd80a8287ae51c39375628b6d7b5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-01 00:30:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-01 00:30:31.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 4.20.0","message":"update constraint for openstacksdk to new release 4.20.0\n\nmeta: version: 4.20.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Doug Goldstein \u003ccardoe@cardoe.com\u003e\nmeta: release:Commit: Doug Goldstein \u003ccardoe@cardoe.com\u003e\nmeta: release:Change-Id: I6b106f656ac48a52f697f9b34ca90c92520c865e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I927d3cd4a8179f78aa86ba5c51fd4c2655048713\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50a156dd6141fdcd42ab111a58b3ae040e06573d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50a156dd6141fdcd42ab111a58b3ae040e06573d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
