)]}'
{"id":"openstack%2Frequirements~1003261","triplet_id":"openstack%2Frequirements~master~I8be63428f920b4bf31e8f468cd6f908a660f5537","project":"openstack/requirements","branch":"master","attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-09-17 15:10:18.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"20676":{"account":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"last_update":"2026-09-17 15:05:58.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-17 15:10:18.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-17 15:10:18.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2026-09-17 15:05:58.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2026-09-17 15:10:18.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}},"hashtags":[],"change_id":"I8be63428f920b4bf31e8f468cd6f908a660f5537","subject":"Add nvidia-ml-py to global requirements","status":"NEW","created":"2026-09-01 11:52:28.000000000","updated":"2026-09-17 16:49:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"25f8ac36c79b6d2e0997c88458bef10c980eca06","_number":1003261,"virtual_id_number":1003261,"owner":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-17 16:49:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 13:16:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 05:49:06.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2026-09-02 05:50:06.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2026-09-02 07:26:54.000000000","updated_by":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"reviewer":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"state":"REVIEWER"},{"updated":"2026-09-17 15:10:18.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-09-17 15:10:18.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-09-17 15:10:18.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"fee3a62ff48d41f882f44102dc4ceaf517f11b59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-09-01 11:52:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cff523e53403c09cd7545e5538874d79f20d15fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:16:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55c4c60f82664472bfae522cece28145\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7054a5edb4248f2ba364bfb2f15553b : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87f32b580b8e4da0914773ece682bc43 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63d93d062b4e4258ae04043a435c420c : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4099549efdee4a12b613fe187dd07b80 : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f67002fd7b074a179116ca6b97b270fa : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3486c396429f4947baf742440099cc2a : FAILURE in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5bf25016ea9648f0883fe3b2c05bbdf8 : SUCCESS in 3m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/abac88b5d9114ceeac1d916ceb7a7ad4 : SUCCESS in 9m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c1e66c7f55394f5ca5c4d1e4d47d6752 : SUCCESS in 9m 16s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/c42c8df4a0644ce1b0dfa7f1ed40064b : SUCCESS in 4m 47s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/65fa37b943624e6a9f1ef7958f9c4f18 : SUCCESS in 12m 28s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/af42ba76fee2484c93c7580c66d783ec : FAILURE in 5m 04s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/9c68a4b729324f2aab8116e4d00ba717 : SUCCESS in 4m 09s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/42b62d6e884f4d4c9ecd17d409ec54a6 : SUCCESS in 6m 11s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/fdbb8adf5ba147548fcc5f1ea8e24024 : SUCCESS in 8m 32s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/6116564cce4242c7bd122cfefd0454bd : SUCCESS in 4m 48s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/518ee10b4b39423498768cf79e4503c2 : SUCCESS in 5m 13s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/646d15a71c374fec805681f17710b29a : SUCCESS in 6m 19s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/5aa031edb0c1471b89f7d6b862cb84df : SUCCESS in 18m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fbf4b493b96e40a3920eb0b0cae80ec7 : SUCCESS in 6m 41s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/e802fb8239bb4c4b9456da4d659f012d : SUCCESS in 9m 15s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/0b330a1df0bb4996abbc92a13e16abd3 : SUCCESS in 10m 01s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/9ff833b494f447c4bf77c66b33e9a0c3 : SUCCESS in 2m 32s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/7f2a066433524f978813d97cba07ab80 : SUCCESS in 7m 24s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/87bdd5e1b29b417ca7b47be27ea84d80 : SUCCESS in 4m 44s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/f70ed2368c454165b50f5ff95f6b9813 : SUCCESS in 3m 11s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/b12dc78fe994486b89d10316ae6698b8 : SUCCESS in 9m 41s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/049ae5c49a014f0eb4381be303279fc2 : SUCCESS in 25m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/40f7f67ea6c84391a10a2ce1a8f23e0b : SUCCESS in 6m 57s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/42a4e8a820534090a3b260d51a333b2b : SUCCESS in 13m 10s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/857fbc707d4a4134a96b1089008488e5 : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/402028b3c3614f49853cd616e1591c0d : SUCCESS in 2m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/47e933bb1e81400d8c51828aeb30e180 : SUCCESS in 24m 40s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/b9b730b2175b4f59a49962c564b8af6f : SUCCESS in 7m 26s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/ae0a4f4f36774ad89675981b2006852a : SUCCESS in 7m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/69521b9d5ba24d389c4f18bb64642835 : SUCCESS in 6m 57s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/633a94b68e2d4e55b4965d25182bf3eb : SUCCESS in 16m 31s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d55fec2e3a534cb7a1c66182d66c0351 : SUCCESS in 4m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5b38f50c30249c590b1f7c24632efab : SUCCESS in 42m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a164709e69a48899c61ac325e329ee2 : SUCCESS in 1h 06m 54s","accounts_in_message":[],"_revision_number":1},{"id":"48abe289b4e8d618d860a4114e43ca2fc757d792","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-09-01 13:18:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24491d26149660e53c6d5138714d86dc6f26ced3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 14:55:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7c232dfbff94c119b09d45ff06fe8b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd589783f0d4f12b5231703cca70085 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b0f1f83c3a74aeab529c031919a7f1a : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/626f8fe21645437f9e26be509a6f7b28 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7e55b27820a49139290d694337dba47 : SUCCESS in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/49d9fd9f8d394279b6ea5f68ef9631b5 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/92d7d6abe6904e7c84f91d5f86fe196d : FAILURE in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1f56e2911bbf43ce8115e147b0839501 : SUCCESS in 3m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d69c9e8ecd884835b2cfb1f942a69c81 : SUCCESS in 6m 19s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/357fab5e33cb415ba9c4e68af064dc2c : SUCCESS in 6m 28s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/2c67f543e6884dc88b58d5bd315b7342 : SUCCESS in 5m 57s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/562a36e366574a89a95eb13cd1ab0ac0 : SUCCESS in 14m 13s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/d56d6edd2f05466696a959eeb40f7354 : SUCCESS in 4m 47s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/84b17d5a3e384752ae8937a0cf37a51d : SUCCESS in 5m 09s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/f552aa974bf64700a5cfe0731a83b657 : SUCCESS in 4m 23s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/33086e26ced94001b43305a0b2ed2bbb : SUCCESS in 9m 01s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/ffedd3df57c94280a330a5abe9ec608f : SUCCESS in 4m 46s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/e4c2bb4a2b2349dfbc28737af62bd9fb : SUCCESS in 6m 15s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/fe8273f31b0940d7992b679721756ba3 : SUCCESS in 6m 44s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/5e34c654ca7848dea799a04b15a4d739 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9986611288024a88a89c8a74c19edf1d : SUCCESS in 6m 43s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/82f059f81970493c944eda1ec4108c03 : SUCCESS in 8m 17s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/e549664e9f224869bc356bbe8838ff06 : SUCCESS in 13m 08s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/9bbe818f6ea34a1185342f6ea2018e43 : SUCCESS in 4m 05s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/df5fd55114bd436394c56da3160312b1 : SUCCESS in 7m 54s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/5deb48d38e3e4e6a8cbed9ba21f4270b : SUCCESS in 6m 52s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/de858a8412db42b8aea97568e0e5386c : SUCCESS in 4m 24s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/c3bef3e6c47346ddb836c1edb6d7d7b6 : SUCCESS in 11m 56s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/5eb8fd5df0ac48648c017ce08816bf09 : SUCCESS in 27m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2c2a8d81d1e7408e94a3e64e17a24942 : SUCCESS in 6m 35s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/e5afe521350a4bdfaa2c0d2c184d2966 : SUCCESS in 14m 51s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/51695318710c4abfabc6eadb40ab1a97 : SUCCESS in 3m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/317a1b8d7bb14a5fa0025d9bcb2af325 : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26f61443b003409680f520e1b7210f8f : SUCCESS in 28m 07s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/e2a918230b65400ea2530348029e5408 : SUCCESS in 3m 26s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/257bf10aff454d6691d5c1a370a8177d : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/50d459b7d6cd4e138fbddb5b90ddc69f : SUCCESS in 6m 35s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/dd610237e90841909e8c1792386dac9f : SUCCESS in 9m 40s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/d8c9b1fdc03f477ebe564cb36cd697c2 : SUCCESS in 5m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6d62caa3897458eb4c2b454700be077 : SUCCESS in 43m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03626ceccb3543d6add708a6f8fc8904 : SUCCESS in 1h 33m 52s","accounts_in_message":[],"_revision_number":1},{"id":"73131e84121e9a02d5e717b62aa4f9583c947250","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-09-02 05:49:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70f7ee12007295046357e921a2d0b6cabe1ec7f1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-09-02 05:50:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"290cece8781c57a3f876cd312cf56f66bf64039e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-09-02 06:00:49.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"3102157f3fc55737fcdf2a6b42f6f3feaf2eddda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 07:10:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65b28acbe51445bcaabe752d9f76447c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2375b5f84054767a334fa8c51bc45b1 : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/139cd62e4894407ea0610240eeffde78 : SUCCESS in 1m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56be3009df7a481296ce62f150152371 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/357c904ca4b14096a843b4efacbb1f75 : SUCCESS in 2m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/85649cb7fd0e4d5eaaff996eff9515b0 : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/433ccf7b05c3454bbb68a9ab4c457140 : SUCCESS in 2m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ee06ba5d7874a7e94a7f94be622d038 : SUCCESS in 5m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/29269a46f405426fbc651e3d45f0a38f : SUCCESS in 3m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1dce39a451664ff6887ac3494da1b686 : SUCCESS in 3m 33s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/07e07feef01444be94b30e7309ea00a2 : SUCCESS in 5m 33s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/40591186858040ef9864cfe84ea1a7d8 : SUCCESS in 7m 55s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/b3c52b0a620241e69933b4741f5a64bb : SUCCESS in 4m 35s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/3ac5405f8b114b339158ea0a416d7423 : SUCCESS in 2m 57s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/ab40e4bd4e9f47b2af0ca93a675f35d2 : SUCCESS in 4m 49s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/b21a986a4fe441af963c0b62a88eff04 : SUCCESS in 10m 23s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/97e12e0d94cd48438fc7ac9b8b0718fb : SUCCESS in 3m 21s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/ceb8b928927a45378ab0964d0586cca3 : SUCCESS in 4m 41s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/979f7b57d3fa4ab4838d3882f2e5f78f : SUCCESS in 5m 49s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/4b8e7d0feedb4c9c84d0e324fbe21946 : SUCCESS in 17m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/649bedefd3264b019d51af500ee5ee70 : SUCCESS in 4m 22s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/4c98ec4f711045e88b429efb7f88460a : SUCCESS in 4m 28s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/ff1709c7baa74f7e942a00e12cc49a4b : SUCCESS in 11m 27s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/387d5e0c52c944c5a185f4daba850873 : SUCCESS in 2m 08s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/17dffe4d806b4875bde1cfe94ec355a4 : SUCCESS in 4m 33s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/a77f6212d9bb4d88a04f810070b2d8d8 : SUCCESS in 4m 03s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/7aefc6d05d3443c9ab602be9772722e6 : SUCCESS in 4m 30s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/5602b800f0ac4d5882fb9c60ec0bf9c6 : SUCCESS in 10m 09s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/fb5a94f4ac96431e87401266e45286e1 : SUCCESS in 16m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/495a68b198ff4f2ca8513d1088a521ec : SUCCESS in 6m 02s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/b5d668471f64439b8543bf9425123214 : SUCCESS in 7m 52s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/c3a38ad89bbf45c88e851da82e1988c8 : SUCCESS in 4m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7045d2e382234862acab48106684c984 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1e3e80f566247e69c1646aa2f0dde48 : SUCCESS in 24m 33s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/be6f6b053c684cb49bad7236fd3ee888 : SUCCESS in 5m 12s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/757684bda910477aa94a9f44473ac604 : SUCCESS in 4m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f11ca19c54234bfab93d998cf70c2114 : SUCCESS in 6m 08s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/0456d4f4fe4f48b1b696c76746d705c7 : SUCCESS in 19m 18s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/a0c0d37c971e461eb49f3f03c882fd3d : SUCCESS in 3m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98b92fba7d5444be8e6aedf90470fa2d : SUCCESS in 58m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dbc039a145c941f981d9b2208e2ee693 : SUCCESS in 1h 07m 41s","accounts_in_message":[],"_revision_number":2},{"id":"a8499ed5c455dc87956da8f3e58cf257902d945c","author":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"date":"2026-09-02 07:26:54.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf9af6439a098e26e6a08dce45dfc31f0729e1dc","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-09-02 07:37:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce3512106bdd77d7f07abae36bc740e95ec2eb6c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-09-03 09:39:35.000000000","message":"Abandoned\n\nThe library will not be included in the requirements.txt, it will need to be installed externally, like libvirt and ipmitool","accounts_in_message":[],"_revision_number":2},{"id":"80d2940f71cd0ea119111ed711bc1d5d1f557448","tag":"autogenerated:gerrit:restore","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-09-17 14:59:08.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"adde19fa51c2ebb286ab10d156f6f32ca9dd1758","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-09-17 15:03:18.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"191fdf001af330c8d5daf60b7b27ea90991ef8e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-09-17 15:05:58.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"25f8ac36c79b6d2e0997c88458bef10c980eca06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 16:49:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f3b9bfeba434bfc85aa712da4532d79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb9d2ce84b214a95a828c4050bfba8cf : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a77aa00356494e99beaca91665bed473 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60e221cd4e804347be10cedf9afe96e6 : SUCCESS in 6m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf1e23aac7504f8781092239bc6ca190 : SUCCESS in 2m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eee61c388a8a4ae6ad0206537b2dba5c : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a8f7a568cb84664a66b4329180b7b36 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2d95fb07bdd743c6b571173475680343 : SUCCESS in 3m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6954d26c7735411eb1726fccc110ca20 : SUCCESS in 9m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/83ef7dd790054fb78cb0d6025c32b722 : SUCCESS in 8m 25s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/de419f5194a047949cb86d5017662334 : SUCCESS in 4m 30s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/ff161ce1ae5847738a09e5c39771f027 : SUCCESS in 9m 55s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/ad8c255fcfca48e68999a74c9e68332b : SUCCESS in 4m 50s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/19018f157928461e949e410d9dd52cc9 : SUCCESS in 4m 40s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/4044e476b36f43eeb77fbe5102949d1d : SUCCESS in 5m 35s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/dc6b99efdef04c12854093f3b6010058 : SUCCESS in 7m 55s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/1bce538d65a7441ba903502faa90a106 : SUCCESS in 4m 50s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/a36927151a824d30ba7f09ad00cce11c : SUCCESS in 5m 16s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/530cfcefa4814206ae0d2c34c9e5da17 : SUCCESS in 6m 40s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/80cfdbcf39b042d78316dd7c6ccd0d24 : SUCCESS in 8m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f9cef4d86f1947658a84d813f23816f3 : SUCCESS in 6m 52s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/b0d4f5833ab6462099ebc1de370c8af9 : SUCCESS in 7m 59s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/741cbeabccc94d39835207c64c132527 : SUCCESS in 10m 55s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/395cce7be3054c2bb78b1ebc8b8c5768 : SUCCESS in 2m 39s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/13e93345849b4ed7b15828ad08c487d8 : SUCCESS in 7m 55s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/d44bbf3fa2314af881d762ae43b8c678 : SUCCESS in 4m 17s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/466a444e5c464ea98d2d293f9391a787 : SUCCESS in 4m 35s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/e0c3420c7df4428fb9783c56fdec3580 : SUCCESS in 9m 38s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/df20e0928893494d980c0a5d26061726 : SUCCESS in 26m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/36708799c05e43b0b944b93efe052429 : SUCCESS in 8m 24s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/7bacbb16cb7f4381be81faff17f17c14 : SUCCESS in 16m 10s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/634d9cc59a9945b2a0ea104a3c9a5c99 : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bda80226e42b4483961f1e5c6a1be183 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f461f2aa4d543f3bd0c108347b6eb18 : SUCCESS in 22m 36s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/24bc433deead4241a3e0240290151892 : SUCCESS in 5m 28s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/4a54e96b62df45588f5bc24a3e49b021 : SUCCESS in 3m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b51a27b006c340e8b8f40eff6f3e8dcc : SUCCESS in 7m 49s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/9e3beb91c22a44e39bfcb13bf456ecc0 : SUCCESS in 12m 32s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f45d91bcb87349e5b27958b0ec93ecb2 : SUCCESS in 5m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48ea8d5d61634f8a806edf5210334fdc : SUCCESS in 1h 00m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a0726e6b57d4297a43bf1d698bd9ce1 : SUCCESS in 1h 40m 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7216db430a958a9371087b35e54119739b6ce49c","revisions":{"76b4ada520a9931bfa2940c3593b2e5bb9a5aab3":{"kind":"REWORK","_number":1,"created":"2026-09-01 11:52:28.000000000","uploader":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"ref":"refs/changes/61/1003261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/1003261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/1003261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/1003261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/1003261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/1003261/1"}}},"commit":{"parents":[{"commit":"e7f5b22a07aa8ec5ed47d92507cd16a1d80b99e3","subject":"Merge \"update constraint for keystoneauth1 to new release 5.17.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7f5b22a07aa8ec5ed47d92507cd16a1d80b99e3"}]}],"author":{"name":"Juan Larriba","email":"jlarriba@redhat.com","date":"2026-09-01 11:51:17.000000000","tz":120},"committer":{"name":"Juan Larriba","email":"jlarriba@redhat.com","date":"2026-09-01 11:52:23.000000000","tz":120},"subject":"Add nvidia-ml-py to global requirements","message":"Add nvidia-ml-py to global requirements\n\nAdd nvidia-ml-py (pynvml) to global-requirements.txt and\nupper-constraints.txt. This pure Python library wraps NVIDIA\u0027s\nManagement Library (NVML) and is needed by the new Ceilometer\nGPU pollsters to collect per-instance vGPU metrics from compute\nnodes.\n\nChange-Id: I8be63428f920b4bf31e8f468cd6f908a660f5537\nSigned-off-by: Juan Larriba \u003cjlarriba@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/76b4ada520a9931bfa2940c3593b2e5bb9a5aab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/76b4ada520a9931bfa2940c3593b2e5bb9a5aab3"}]},"branch":"refs/heads/master"},"509b3df846b97df44cc9660e1cd31c350a40f249":{"kind":"REWORK","_number":2,"created":"2026-09-02 06:00:49.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/61/1003261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/1003261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/1003261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/1003261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/1003261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/1003261/2"}}},"commit":{"parents":[{"commit":"e7f5b22a07aa8ec5ed47d92507cd16a1d80b99e3","subject":"Merge \"update constraint for keystoneauth1 to new release 5.17.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7f5b22a07aa8ec5ed47d92507cd16a1d80b99e3"}]}],"author":{"name":"Juan Larriba","email":"jlarriba@redhat.com","date":"2026-09-01 11:51:17.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2026-09-02 06:00:43.000000000","tz":0},"subject":"Add nvidia-ml-py to global requirements","message":"Add nvidia-ml-py to global requirements\n\nAdd nvidia-ml-py (pynvml) to global-requirements.txt and\nupper-constraints.txt. This pure Python library wraps NVIDIA\u0027s\nManagement Library (NVML) and is needed by the new Ceilometer\nGPU pollsters to collect per-instance vGPU metrics from compute\nnodes.\n\nChange-Id: I8be63428f920b4bf31e8f468cd6f908a660f5537\nSigned-off-by: Juan Larriba \u003cjlarriba@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/509b3df846b97df44cc9660e1cd31c350a40f249"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/509b3df846b97df44cc9660e1cd31c350a40f249"}]},"branch":"refs/heads/master"},"7216db430a958a9371087b35e54119739b6ce49c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-17 15:05:58.000000000","uploader":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"ref":"refs/changes/61/1003261/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/1003261/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/1003261/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/1003261/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/1003261/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/1003261/3"}}},"commit":{"parents":[{"commit":"e7f5b22a07aa8ec5ed47d92507cd16a1d80b99e3","subject":"Merge \"update constraint for keystoneauth1 to new release 5.17.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7f5b22a07aa8ec5ed47d92507cd16a1d80b99e3"}]}],"author":{"name":"Juan Larriba","email":"jlarriba@redhat.com","date":"2026-09-01 11:51:17.000000000","tz":120},"committer":{"name":"Juan Larriba","email":"jlarriba@redhat.com","date":"2026-09-17 15:05:58.000000000","tz":0},"subject":"Add nvidia-ml-py to global requirements","message":"Add nvidia-ml-py to global requirements\n\nAdd nvidia-ml-py (pynvml) to global-requirements.txt and\nupper-constraints.txt. This pure Python library wraps NVIDIA\u0027s\nManagement Library (NVML) and is needed by the new Ceilometer\nGPU pollsters to collect per-instance vGPU metrics from compute\nnodes.\n\n* Is the library actively maintained?\n\n    Yes. The most recent release was in Jun, 2026.\n\n* Is the library good code?\n\n    Yes. It is part of the official Nvidia SDK for Python.\n\n* Is the library license compatible?\n\n    Yes, BSD license.\n\n* Is the library already packaged in the distros we target?\n\n    Yes:\n        https://packages.debian.org/trixie/python3-pynvml\n\n* Is the function of this library already covered by other libraries\n  in global-requirements.txt?\n\n    No. There are no Python bindings for Nvidia NVML functionality in any place.\n\n* Is the library required for OpenStack project or related dev or\n  infrastructure setup? (Answer to this should be Yes, of course) Which?\n\n    Yes. Ceilometer will be able to retrieve GPU metrics from the compute nodes via the NVML interface.\n\n* If the library release is managed by the Openstack release process\n  does it use the cycle-with-intermediary release type?\n\n    Not managed by OpenStack\n\nNeeded-By: https://review.opendev.org/c/openstack/ceilometer/+/1003236\n\nChange-Id: I8be63428f920b4bf31e8f468cd6f908a660f5537\nSigned-off-by: Juan Larriba \u003cjlarriba@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7216db430a958a9371087b35e54119739b6ce49c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7216db430a958a9371087b35e54119739b6ce49c"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
