)]}'
{"id":"openstack%2Frequirements~1003285","triplet_id":"openstack%2Frequirements~master~Ifb7d0089872c699f160cd7d20e6102ee0fb8b8fe","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifb7d0089872c699f160cd7d20e6102ee0fb8b8fe","subject":"update constraint for networking-generic-switch to new release 10.1.0","status":"MERGED","created":"2026-09-01 14:12:43.000000000","updated":"2026-09-04 09:20:47.000000000","submitted":"2026-09-04 09:19:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003285-new-release","meta_rev_id":"327acf63ee3306abb38144ac857ad20b4939a8a5","_number":1003285,"virtual_id_number":1003285,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-04 09:19:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-09-04 07:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-09-04 07:28:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 15:37:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 07:28:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"83c1e18395391a2d704b253e952c2672b5ba2661","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-01 14:12:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0d7149ea0213b2f61263da50dbb89966112499e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 15:37:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3346521363df4697bdaaaa7b42bb54dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4311cc74900d4965893ba66fc6008ba1 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66070d6458b34c3bbe65f90010604b37 : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a5c0c14f7fcc4580a2b22258b633db03 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb187dc081574035b4e235ad8bd8f91a : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/973e1e5370c44857aea60b1661793e14 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/025c45bdd4a24746904f540a1d399360 : SUCCESS in 4m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88782671fa404140abea4aefd6829a19 : SUCCESS in 3m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a55fef32570e428e949334a1c8604313 : SUCCESS in 9m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e045c3038955431bba06813de55901ec : SUCCESS in 6m 41s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/d87ea4db486442e4b27debab4effc34e : SUCCESS in 3m 46s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/10583f486e3b495d9d70248a464dcd28 : SUCCESS in 4m 54s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/e13605afa8564464844fd127d7b9a8fa : SUCCESS in 7m 48s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/83586c25bda945199725c2a31a530c22 : SUCCESS in 4m 16s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/417f1dc73ee1450589f3417511161f1c : SUCCESS in 5m 01s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/75bd5d805d2e4269838c54c6bd17f81c : SUCCESS in 9m 56s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/a5fc679fec6d45bab7ebab3de602caa1 : SUCCESS in 6m 14s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/6a84bfdbc28c436d9ff847772f43a023 : SUCCESS in 3m 29s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/513741622e8f4df98fcc479dd07a5309 : SUCCESS in 6m 39s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/e0d84e801cf2446da2c5bbe5047f7bfd : SUCCESS in 16m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fe30057a1c894fc78af19d8a2ba28375 : SUCCESS in 7m 29s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/df045ff8e9524715a7684e33c7494347 : SUCCESS in 8m 43s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/e790c6b76f5542e18c8d8a425e038c9a : SUCCESS in 11m 36s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/317285d35c8f40628ef7443a083c541f : SUCCESS in 2m 28s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/8ac9102772d648b6b9cf646e9747f86a : SUCCESS in 9m 06s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/c675931f313c4442ae9dc8041914e145 : SUCCESS in 3m 30s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/00a81e766056471bacbd96846428e3fa : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/fb21088783484033ba0b8c59cff90a99 : SUCCESS in 9m 25s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/d167466411d740e391202b2aed0e281a : SUCCESS in 27m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b9084776e7e94da3a67efec5199b0022 : SUCCESS in 6m 53s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/816a0dc297e548d7887b8bac983f9379 : SUCCESS in 13m 09s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/2dd722a501554194a37069a2c0395745 : SUCCESS in 5m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0c6a0ff32cae461a84b7984c86930fb8 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/155ac981fe014dabb24644f1aa0e78c1 : SUCCESS in 23m 12s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/5ccf355ad7d54b8fa8ecd0fe373618bb : SUCCESS in 5m 19s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/56ec911b310242869510dedfdeab26c2 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b629743317594e9e9d18b523ea8cddc5 : SUCCESS in 6m 59s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/a010761e176e49cdacb3d22e71216b91 : SUCCESS in 19m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/12b2961475ce46a7a56b2b44972f5455 : SUCCESS in 5m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d2f3c1d8673404d839dbc8ba1818628 : SUCCESS in 1h 01m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc20c2ae14004b7296eca95dcbc6db9b : SUCCESS in 1h 21m 28s","accounts_in_message":[],"_revision_number":1},{"id":"bcdbb8b4dd411081e27909c07c6f6e8a0655afe2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-04 07:28:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15ae9d090a1f31dc5c71511077dcb67aa0527d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 07:29:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d58d6fd6536360ba5fd94de65a631d4116627c48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 09:19:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01f7dfab75d9460e8c9efd40091272c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b6917c2a8494748a2e758ab9626f874 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d89fd8bbe0b42b1ad694480a6d491b1 : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09a8e8edf88945759d20d8031b9d8254 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c164f7bdb064393a0fa09ca6994b58c : SUCCESS in 3m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea4228b8bbf34624b1ba4383bd892024 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fe9d250b18ad41a99301322a764bf668 : SUCCESS in 3m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3170f9894a5b41eebd0a1d5e70ed6eda : SUCCESS in 5m 47s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a38bffa2d4634388a0632fd1e3a5060a : SUCCESS in 8m 32s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/7a8227946d70435bb2becdd42a8f057b : SUCCESS in 6m 14s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/4a451e4db05545abb6d1904705b84e21 : SUCCESS in 10m 58s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/8cf68a0685c04c4f82e7e47ac40dc058 : SUCCESS in 6m 50s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/43f8058d80014d81b284b16acd6f870a : SUCCESS in 5m 09s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/8d3c48577fb242ddba6ea179d1276593 : SUCCESS in 6m 34s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/963754c211f6400c8f6c24d80c3c6ecc : SUCCESS in 15m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6471cb53e28749e38e17f6efa3e241df : SUCCESS in 21m 41s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/7c98ae714cf24aa3bfe4dc20df128913 : SUCCESS in 7m 38s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/2efbda3055174badb1b1cd27f4223e51 : SUCCESS in 12m 18s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/b6ffea4e4c474f7c91579a5a0cc30d08 : SUCCESS in 2m 11s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/7e01c7009cbb41728e7c9f14d05fdf19 : SUCCESS in 7m 04s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/4dc88a94a1fc4cfc8b751dc82354b10f : SUCCESS in 26m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/cc2d48a7fad14b5aa92ac1dcc36c03e7 : SUCCESS in 6m 26s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/7db32b860963488988b9926acb8d2b16 : SUCCESS in 14m 23s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/8e3800fb447b4e828d1b1e7889282f88 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/18d4bf3b982e47e791ba516018e32be9 : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/28a3ee7065964da18a018db6d2bd2072 : SUCCESS in 24m 02s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/d20e3e8df90849e8bd821b4323326a16 : SUCCESS in 3m 39s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/3da8d16eb25e4527a88f51059f6faade : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2dce37ede02d4bd39191d61d29e9d980 : SUCCESS in 7m 38s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/d0abc80ef311493fa66cbfca519f9b63 : SUCCESS in 19m 33s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/eb75f91e37fa45adb80d4555c2ff9352 : SUCCESS in 5m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6957ae5ae934c18b58bf3ce68d9db4a : SUCCESS in 1h 08m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/265e7afdddc545948f9bf3bf2f065d8d : SUCCESS in 1h 44m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3560169cc5106f89ccd64b16c2972f267c3213b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 09:19:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"327acf63ee3306abb38144ac857ad20b4939a8a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 09:20:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40dcc10a4b9f4780ac438aef0aeb1c30\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92eefbf59dba4b62890ce5ec51da4517 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a787522c0e23ddf495bc4b91ef9a97cda01abf0e","revisions":{"a787522c0e23ddf495bc4b91ef9a97cda01abf0e":{"kind":"REWORK","_number":1,"created":"2026-09-01 14:12:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/1003285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/1003285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/1003285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/1003285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/1003285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/1003285/1"}}},"commit":{"parents":[{"commit":"e7f5b22a07aa8ec5ed47d92507cd16a1d80b99e3","subject":"Merge \"update constraint for keystoneauth1 to new release 5.17.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7f5b22a07aa8ec5ed47d92507cd16a1d80b99e3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-01 14:02:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-01 14:02:41.000000000","tz":0},"subject":"update constraint for networking-generic-switch to new release 10.1.0","message":"update constraint for networking-generic-switch to new release 10.1.0\n\nmeta: version: 10.1.0\nmeta: team: ironic\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Iury Gregory Melo Ferreira \u003cimelofer@redhat.com\u003e\nmeta: release:Commit: Iury Gregory Melo Ferreira \u003cimelofer@redhat.com\u003e\nmeta: release:Change-Id: I6b62ffee669c153f0839a69eedd52fd9f0e01363\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ifb7d0089872c699f160cd7d20e6102ee0fb8b8fe\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a787522c0e23ddf495bc4b91ef9a97cda01abf0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a787522c0e23ddf495bc4b91ef9a97cda01abf0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
