)]}'
{"id":"openstack%2Frequirements~1003532","triplet_id":"openstack%2Frequirements~master~I91a122c2ee48c5d1320a2578f831483a364689c7","project":"openstack/requirements","branch":"master","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-03 12:17:24.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I91a122c2ee48c5d1320a2578f831483a364689c7","subject":"Bump imagesize to latest vesion","status":"NEW","created":"2026-09-02 13:54:51.000000000","updated":"2026-09-25 06:18:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f470f015e80309aab14df0cbafcd4e40eeb7c0d7","_number":1003532,"virtual_id_number":1003532,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-02 15:04:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-2,"date":"2026-09-03 12:17:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-09-25 06:18:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 15:04:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 12:17:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-09-25 06:18:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"d1476eb28a31b998701d48d2e48d1bcfdd5074f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-02 13:54:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fb5fab17b6615f673697c5764e0ae36046a46d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 15:04:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c0617afa13d4277be2ec7d19ed0c159\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3b1c96a65742579219e217760b0cf1 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94bfe537b04e42f89ff1a241238c38cf : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/483d20ba237f4574a44d30dfee9b312f : SUCCESS in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/638f65bb358a4597b4dd04bae66ce125 : SUCCESS in 3m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8f4b02abcd494edbadfd718fa2d9cd4c : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4221fce5aef042a9b9694162baf19dc3 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/41d2e963f05a4226958222ab45a9ba18 : SUCCESS in 3m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d288fe22021d42f5a2015b92bb77f1eb : SUCCESS in 5m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/19a6271cfc714108b30ad17a37532a99 : SUCCESS in 10m 15s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/3ac5bf445fc14fca858108b9d12c1676 : SUCCESS in 3m 51s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/bc7fbf94aac8447fa597f4f1b7059b93 : SUCCESS in 4m 30s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/7549511108b241fa9c2e4542e57a311d : SUCCESS in 6m 22s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/8ad40640b83e4900bd306f6847077e80 : SUCCESS in 3m 44s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/955d0c3642c8452ab0995b51718a9a8d : SUCCESS in 4m 06s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/bb2e4f7e18a742c0a22f451947eb519e : SUCCESS in 12m 57s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/905266feed8e4200b59bcbd82b893477 : SUCCESS in 4m 37s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/24ced239e43d45d4ada4ba19580c6ea6 : SUCCESS in 4m 55s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/0488919fa6294632a1b21edc9026ab62 : SUCCESS in 6m 04s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/1b7dd65616794932885d5dbb23983eb9 : SUCCESS in 8m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9fe517985f064ec9acb8b008461f64ff : SUCCESS in 9m 09s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/a8e11a9637144ac0a02eda7dc94a1f7a : SUCCESS in 8m 52s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/4c78940ee22e4b16a671540b95201dc0 : SUCCESS in 14m 09s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/380c1ceab94c43778c57c2c16b349e0d : SUCCESS in 3m 56s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/9f5c31494dbf4881a234bc1a6804a377 : SUCCESS in 10m 37s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/37bf64676cc143c292e3f07f4a86a5d3 : SUCCESS in 5m 40s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/52a8fe59c51741f2953ae0f709adc43d : SUCCESS in 4m 32s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/2af21b8e6f284cb9bd25b21e25d7392d : SUCCESS in 9m 56s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/ae1d5529df0a49039bb88a4792d069fe : SUCCESS in 28m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6d7c0e2884a04273b2b8df07bce41874 : SUCCESS in 8m 06s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/0e1d3dcad4784d7d941166c0e355f550 : SUCCESS in 9m 21s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/ad6bdebeff3c4ce0a0ee0a71b16c4238 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b7d24df1c20944e9bf8983257770808a : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1c537d8cef447e09b9430c7a25f07c2 : SUCCESS in 27m 09s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/43501027655d4271ac2ecab0ad9d8261 : SUCCESS in 6m 10s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/05603e197fc04930afe508b59a560824 : SUCCESS in 3m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/515da9b3ed61402c925e7b0d6a82420e : SUCCESS in 8m 56s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/6c658155062c44768dcf958630422581 : SUCCESS in 16m 51s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/02f3311e471249de87651d0f489a88de : SUCCESS in 4m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45ffa3a9ef4343eca988620c4add1f50 : SUCCESS in 1h 03m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/422e674667164bdd8b17eb3bbd5de679 : SUCCESS in 48m 54s","accounts_in_message":[],"_revision_number":1},{"id":"8abeae58e141476b9a218282c48a6ba59ce5f2a7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-03 12:17:24.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f470f015e80309aab14df0cbafcd4e40eeb7c0d7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 06:18:50.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf149154b729badf3abf926d1c186ad44328392f","revisions":{"cf149154b729badf3abf926d1c186ad44328392f":{"kind":"REWORK","_number":1,"created":"2026-09-02 13:54:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/32/1003532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/1003532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/1003532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/1003532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/1003532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/1003532/1"}}},"commit":{"parents":[{"commit":"e7f5b22a07aa8ec5ed47d92507cd16a1d80b99e3","subject":"Merge \"update constraint for keystoneauth1 to new release 5.17.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7f5b22a07aa8ec5ed47d92507cd16a1d80b99e3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-02 13:54:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-02 13:54:39.000000000","tz":60},"subject":"Bump imagesize to latest vesion","message":"Bump imagesize to latest vesion\n\nThis removes an unnecessary cap on Python version and allows for testing\non Python 3.15.\n\nChange-Id: I91a122c2ee48c5d1320a2578f831483a364689c7\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf149154b729badf3abf926d1c186ad44328392f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf149154b729badf3abf926d1c186ad44328392f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
