)]}'
{"id":"openstack%2Frequirements~1004626","triplet_id":"openstack%2Frequirements~master~I679ca7965d31d1fb1a544f1e7264f3fe2dcf30ee","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I679ca7965d31d1fb1a544f1e7264f3fe2dcf30ee","subject":"update constraint for tosca-parser to new release 2.15.0","status":"ABANDONED","created":"2026-09-08 13:42:30.000000000","updated":"2026-09-28 18:17:11.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b05551111b07ee3741dca83ff97e1d904abcc617","_number":1004626,"virtual_id_number":1004626,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-28 12:08:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2026-09-28 10:39:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-08 14:57:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-27 10:01:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-09-28 08:47:55.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"f59a2b9d4e45980eb8227ac506ac0625c6f804c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-08 13:42:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"743a902fbf85098af01e7b228aec5f03b039ae27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 14:57:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bb4ddcc4a4c405f8701273a09f84c9d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ab14dc1210468a8f960220843f8113 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f965da443e8e45998ee4c40b7028668d : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20daf6a4799247bdb6c8ccab81542052 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4baf78c2ec145338fd86f0878c2a1e9 : SUCCESS in 2m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3abcfe07d63b4f7b847c52b79e47ff4d : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61631613900d4ea1a7ced022afc40626 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3dd3bfd6b4584fca9a14dd5c78c2d7ac : SUCCESS in 2m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/df2e91e428f54f568a4a8abc5ce85893 : SUCCESS in 6m 42s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/db258e2d61ea4f39a34c81dac603eee6 : SUCCESS in 6m 10s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/b3ef5e169ba147d8abf9e0ce337ef02d : SUCCESS in 3m 50s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/b7af4f03e9b84551a3b9dc6ca3368ec3 : SUCCESS in 9m 36s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/c7841469413747539196169852ab745b : SUCCESS in 3m 20s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/7988a9f5d7d74d84ae2445646cd0bd02 : SUCCESS in 5m 05s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/44869170dae8438795dbcaa613412447 : SUCCESS in 4m 55s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/f345752c710d40cb8db2851fd5da9ca2 : FAILURE in 11m 58s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/246001e415ee4882b05aae4872fb2c52 : SUCCESS in 5m 32s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/378283b579af44d29515201d12dd2371 : SUCCESS in 5m 16s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/ea7e4d64feb549a4ab767461b0a2c4a7 : SUCCESS in 6m 41s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/bdb3fa404b4241fcae3b2429bd84861d : SUCCESS in 8m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6434e2896e584bc890805f7d74c9d03e : SUCCESS in 6m 12s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/9af8cd3dcad74affa33c49c4b8f41ff1 : SUCCESS in 8m 09s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/fdb932990a174ad6a8f6e6dc6ad3fa68 : SUCCESS in 9m 55s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/c9806e5e45e540f9bfabe5d32050c747 : SUCCESS in 1m 46s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/fa06974a2552418c8581fada776ad224 : SUCCESS in 7m 57s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/cb640845337d400c915ba213861cabe1 : SUCCESS in 2m 44s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/a72ad9e5f1554693a3a0b89d4d1c99f0 : SUCCESS in 7m 02s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/28df85350915404fb0457818049b21aa : SUCCESS in 9m 27s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/7b39008e22da48c79ab3b757522e7ae7 : SUCCESS in 28m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/41d76583edae40d187e2c4006eba94de : SUCCESS in 6m 46s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/584832b0730c41689b0cc0d21107c930 : SUCCESS in 11m 39s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/dffe386c5027461e8efb5e8bd3db4be1 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/546f051f5e8041f99e86d5435522530a : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1441356bcbfb4079b74f417c7ebc1a95 : SUCCESS in 26m 59s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/54312a5189e74827ad3fac715edfd6e0 : SUCCESS in 6m 42s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/bf8f81cc423b464e8cf7f623931400f1 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e981cb22af634d9c90f90bf06382677b : SUCCESS in 7m 05s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/cc56af8bfa1046748bbf02d650a8f454 : SUCCESS in 19m 05s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/57e1975568564a1b828d7aaa3da2478e : SUCCESS in 7m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1f91f1c3d12455199c93eeb456e64f6 : SUCCESS in 28m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52c45d35d2604163b04dc643fb10cab1 : SUCCESS in 1h 10m 30s","accounts_in_message":[],"_revision_number":1},{"id":"731865605739254deab1ec41c1d4f5bb8fca319c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-27 10:01:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9cc651b10785232ae2a7b2172f0d99301108cce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-27 11:51:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c3af77d94f44a99a834cad680a0b6f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49c7b60da91c42bd8b6eda5a3194554c : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a103b75a0374d50af961ad2622c4868 : SUCCESS in 2m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4e2bc7befaa4e42ba008fa8783231e6 : SUCCESS in 2m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8c604a5464c74cde9df7aa6a999bce8b : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b28a33b1da04b1c8dfd24406696b938 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b26a996a84784e6f8f2c70c20af8d586 : SUCCESS in 2m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fc9f95b4ae134e879da6a6b8a7e778ff : SUCCESS in 8m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/315b8b5bcf214c85af917707c03754e5 : SUCCESS in 7m 07s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/ad64331baad14530858529f386c3d51f : SUCCESS in 8m 57s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/95b93b68f6384b0898936bf8b5994e62 : SUCCESS in 13m 09s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/4cda1020b15b4343b55853b0ab446362 : SUCCESS in 6m 05s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/3c4666452bd34ed08cf976858dfe7f4b : SUCCESS in 4m 33s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/61a939e0392343df836c0d7025aadddb : SUCCESS in 5m 05s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/8d1450ad0fe348fc855658a6b8f6d571 : SUCCESS in 10m 10s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/6b9ca511b4d94909be89e62ba1a1cca7 : SUCCESS in 5m 13s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/0d0fcc85e304406fa4540596fa522b17 : SUCCESS in 6m 22s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/8ae719fda88d45dca1ecb760a3643524 : SUCCESS in 5m 57s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/8b9ce41eb9294436aa99d6f3099b187a : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/edd25f6e493b4090ba90041b90aa94f1 : SUCCESS in 4m 41s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/8270d56aa59849a594fad6e340f215fe : SUCCESS in 5m 43s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/e04b70af90f94514a0b8bcd35fef7d7f : SUCCESS in 11m 22s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/a8b1b35ebf764e6fa1e8b99484b60a78 : SUCCESS in 2m 25s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/8df4a119b6844e06bf9032e6fcae6d48 : SUCCESS in 9m 08s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/e4e595847bf74a57a5f32d824d334551 : SUCCESS in 3m 56s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/5b9ad29da7784197be2911032b7f0165 : SUCCESS in 4m 13s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/0b00f1bc22ff4bd291fbdc26cbff829e : SUCCESS in 9m 29s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/6d6e93bf5b3344b595682e81fe002d80 : SUCCESS in 21m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4dc42fd15fc847c28b2514f05a7cc392 : SUCCESS in 5m 08s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/1757b5156d13453d981ddfc2ad0ba003 : SUCCESS in 7m 55s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/20c6530e38c94bd39a697fbdc1acc257 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dabf1d1ee8b04c6c8eee9155bef8f284 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e1b3c9abf414908b43bb3d76dab6e35 : SUCCESS in 27m 25s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/469a64a1c2a84ac1aed53471e99878c5 : SUCCESS in 5m 56s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/007fd51976494ebf913433c74ed4070c : SUCCESS in 4m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9e29e8cc0aea493ab2a6dc584133347b : SUCCESS in 4m 58s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/b194c2dce247446c8e7bb0cc30f16337 : SUCCESS in 8m 45s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/5e4b7adf96f447b093d124e70a39a4f6 : SUCCESS in 3m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90d3662e41ad4785b70a093fa5681ad8 : SUCCESS in 57m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae0fb8203ff540328d998441bd9b47cf : SUCCESS in 1h 43m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4b428b1e07f0c7681427a3fd4f9de5e4a7080603","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-28 08:47:55.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1aea288052ce1ab5d41755d52142da0526421632","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-28 10:39:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_11131\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"_revision_number":2},{"id":"6d5bb8b4fc1f8dca41a05b8410b934f1038d6b00","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-28 10:39:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8fcea6ae28715f70901fef44107556f282bc9afb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 12:08:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99f22753906a40abb5b23b790e811f2b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b25e450337a84258bd3ab8dab5f61a4b : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/722559253e3f44e89f046f5ce9b2c069 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50563543fc2848aea98c7df443846656 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f981d35268054808b727f01415e28be8 : SUCCESS in 3m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c27487cb9774c1d9d7d90b258ed3540 : SUCCESS in 7m 32s\n- build-wheel-cache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32807cebe5ba44e6b1ca395b88698f71 : SUCCESS in 23m 27s\n- build-wheel-cache-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3babddedcd946db8d3962a20dcdcbfc : SUCCESS in 26m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc8e4b712ce54a5581dce3ccb58870bc : SUCCESS in 5m 34s\n- requirements-check-self https://zuul.opendev.org/t/openstack/build/c06d100d6585453ea038b4a7114000e5 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7188e6cddae24adeba68040734d420b0 : SUCCESS in 3m 38s\n- requirements-tox-bindep https://zuul.opendev.org/t/openstack/build/923207e7c1724957a145383beb904929 : SUCCESS in 4m 11s (non-voting)\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b950895dd0b345139bbc0fb37b428f6d : SUCCESS in 11m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b47847a9e70c401c836654e239c098d4 : SUCCESS in 6m 25s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/a544c1977a524b6e8be14f1706327dfd : SUCCESS in 11m 04s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/2808a3eb4c774d45958141e743468de0 : SUCCESS in 10m 47s (non-voting)\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/a5b76ad751974daeb71db9f84e0a2eae : FAILURE in 4m 07s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/20c72ba756654ffda93577941a4a48ee : SUCCESS in 6m 45s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/9c7e9e5daca34511b8aea410b8e26b17 : SUCCESS in 5m 56s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/76c482db11f443ff87fc68c4873d43b8 : SUCCESS in 5m 35s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/6ab642324f2a41a5ac6f4239f05dfcc0 : SUCCESS in 10m 56s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/60c944cbcafe4f169eb618eefeb03152 : SUCCESS in 3m 44s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/93f61bcac2dd4764a849d792baf6fb69 : SUCCESS in 5m 37s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/e1ad482beaf546d39629a2031a013909 : SUCCESS in 4m 30s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/2d10da4567ed4d2b9fd73bdb96d73352 : SUCCESS in 16m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b7464c0a3e8241a59eb8fa271d3db53d : SUCCESS in 8m 05s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/5b89ceb1d5cf443caf8ae5299521d7a9 : SUCCESS in 10m 49s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/edc2a4cd959940a0938bde430e55b99b : SUCCESS in 11m 08s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/b13cf9d5562e47d2bc7eaadca6ae49c0 : SUCCESS in 2m 10s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/323f8322cae4417b8529465d1183b3af : SUCCESS in 5m 22s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/a5e95d0c4f9d49c99974d5d78d8213a1 : SUCCESS in 6m 01s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/acca4c18cf4d4fa0b63953f272d638b2 : SUCCESS in 5m 03s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/a79277167202461ab89ee19fefa41595 : SUCCESS in 9m 08s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/50837b7d5b1a4f4fa7f4e766f3448608 : SUCCESS in 30m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4b418d60229c4b1ba47e8058ec869ce0 : SUCCESS in 4m 08s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/e95cc897e1b442c2a39117fc856a0b30 : SUCCESS in 12m 13s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/68fe96d3513a4287807ef0a07de7a48e : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f9f0976ab844c6cb0c3e9b4ded429a9 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4fd20baa812d447fbcd5226e08fec057 : SUCCESS in 28m 00s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/5d2482602309434cb06809faca210839 : SUCCESS in 6m 42s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/e0fb140383fc408b89a715596a650fc4 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ddca6aa9ab045429a4ee8c5d39220f9 : SUCCESS in 5m 43s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/10704760bf2c409baf50312c715d6530 : SUCCESS in 20m 15s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/6f9e67dc8ee74ea99f2b58fede1347e0 : SUCCESS in 3m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc2e0056bdf0497684abbbe055f4da8c : SUCCESS in 47m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe410dcffc384a65a23c36b5837aa13d : SUCCESS in 50m 10s","accounts_in_message":[],"_revision_number":2},{"id":"3960924ab59d8e700e020e94df4afa1308d82d64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 12:20:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a503e05e2d164d30acaa94fb69ac8d99\n\n- build-wheel-cache-ubuntu-focal-arm64 https://zuul.opendev.org/t/openstack/build/dbbb9f97a2554e00955b1f820719fe6a : SUCCESS in 17m 24s\n- build-wheel-cache-ubuntu-jammy-arm64 https://zuul.opendev.org/t/openstack/build/da6a546526274fd4b4b98fb281222141 : SUCCESS in 19m 18s","accounts_in_message":[],"_revision_number":2},{"id":"b05551111b07ee3741dca83ff97e1d904abcc617","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-28 18:17:11.000000000","message":"Abandoned\n\nyeah, would have been better to merge this one first, but that\u0027s the way it is","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"15639798314a24e28b0bb25cf867f6b1b24ec089","revisions":{"5e319d2028b8f7fb19dc1d28b67ca7abcb2d5461":{"kind":"REWORK","_number":1,"created":"2026-09-08 13:42:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/26/1004626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/1004626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/1004626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/1004626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/1004626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/1004626/1"}}},"commit":{"parents":[{"commit":"9edc04289eca1a4c18ae86e4430416d82a5b07fa","subject":"Merge \"update constraint for os_vif to new release 5.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9edc04289eca1a4c18ae86e4430416d82a5b07fa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-08 13:32:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-08 13:32:21.000000000","tz":0},"subject":"update constraint for tosca-parser to new release 2.15.0","message":"update constraint for tosca-parser to new release 2.15.0\n\nmeta: version: 2.15.0\nmeta: team: tacker\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iebe065a30946cda548ec22f38e00f959d6fb3059\nmeta: release:Workflow+1: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Code-Review+2: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nChange-Id: I679ca7965d31d1fb1a544f1e7264f3fe2dcf30ee\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e319d2028b8f7fb19dc1d28b67ca7abcb2d5461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5e319d2028b8f7fb19dc1d28b67ca7abcb2d5461"}]},"branch":"refs/heads/master"},"15639798314a24e28b0bb25cf867f6b1b24ec089":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-28 10:39:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"real_uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/1004626/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/1004626/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/1004626/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/1004626/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/1004626/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/1004626/2"}}},"commit":{"parents":[{"commit":"e940761cf4f0d2505ce95f92ce12fb568d2de67a","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e940761cf4f0d2505ce95f92ce12fb568d2de67a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-08 13:32:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-28 10:39:04.000000000","tz":0},"subject":"update constraint for tosca-parser to new release 2.15.0","message":"update constraint for tosca-parser to new release 2.15.0\n\nmeta: version: 2.15.0\nmeta: team: tacker\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iebe065a30946cda548ec22f38e00f959d6fb3059\nmeta: release:Workflow+1: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Code-Review+2: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nChange-Id: I679ca7965d31d1fb1a544f1e7264f3fe2dcf30ee\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15639798314a24e28b0bb25cf867f6b1b24ec089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15639798314a24e28b0bb25cf867f6b1b24ec089"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"9edc04289eca1a4c18ae86e4430416d82a5b07fa","ours":"5e319d2028b8f7fb19dc1d28b67ca7abcb2d5461","theirs":"e940761cf4f0d2505ce95f92ce12fb568d2de67a","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
