)]}'
{"id":"openstack%2Frequirements~1005231","triplet_id":"openstack%2Frequirements~stable%2F2026.1~Ic1b81d2aff31b3487f6613227f9372b48bfc6669","project":"openstack/requirements","branch":"stable/2026.1","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-09-14 13:06:53.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-11 14:15:34.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ic1b81d2aff31b3487f6613227f9372b48bfc6669","subject":"update constraint for glance_store to new release 5.4.1","status":"MERGED","created":"2026-09-11 13:31:19.000000000","updated":"2026-09-14 16:19:59.000000000","submitted":"2026-09-14 16:19:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005231-new-release","meta_rev_id":"c38a0ee9f48e632b8fe753a37309c404b2334963","_number":1005231,"virtual_id_number":1005231,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 16:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-09-14 13:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-09-11 14:15:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-11 14:12:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-09-14 13:06:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-11 14:12:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-11 14:12:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-09-11 14:12:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-09-11 15:06:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"324e62465647b2a910f7a5fb1279e13a19c058e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-11 13:31:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14b41eea1bd4df415336232cfc78ab2b681cbabe","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-11 14:12:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cba36f3fd3670becb59a350412fe7a1875e4aab7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-11 14:15:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31e6e1e88a7fecc503b84cd46d8997b87db4b3fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 15:06:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72490ab3dbd54461aa2677454e4ae2f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d9c53bace4649adb7aa6da82285a2cb : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97f65e1e6722426098a289c4419acd62 : SUCCESS in 2m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8f3b7c913cf47c58db4199df7b4b21a : SUCCESS in 2m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7e4feca2d61e41e3ad8ec1e9e592f77e : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a26d7c69b7b644e29e4ff24b9704321d : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4a4d628c3dbf47c3826f721a8e8fc373 : SUCCESS in 4m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/17341e5491da45f7a8550794c307423d : SUCCESS in 5m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3a4e9aed10744583b11792f95fdba0c8 : SUCCESS in 13m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e583663d2f0841869fe075c83c1349fc : SUCCESS in 5m 55s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/26689c397a12474bb153f88fcc5d79e2 : SUCCESS in 5m 49s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2f15a1402ac041d9b807c44a8736e3a5 : SUCCESS in 3m 19s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/033574af61f54ec9b2c15d676a272c69 : SUCCESS in 4m 26s (non-voting)\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1d61e4de357843abb647f699c0edaaff : SUCCESS in 4m 08s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ea695cb1b3814a9cacaadc9e4d29dc7a : SUCCESS in 11m 35s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/41faea24b02b48149814be999e8f17d6 : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/549322d2a8294477add72a0de4f9cdaa : SUCCESS in 5m 37s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ddbaf736d5f84ce5bf5d359f7d731c0f : SUCCESS in 7m 15s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a611e82aeb6c4272bad8d1723882535a : SUCCESS in 18m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f59be4dbda84374989d9919ed436165 : SUCCESS in 6m 52s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/dbb47061c2f6413aa5c59fcc793b1b3c : SUCCESS in 5m 51s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d33339e5cf3843e1a4efd8783ba5405a : SUCCESS in 11m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7d13a09263894021bdda430af2dc8fd9 : SUCCESS in 3m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7ee8eca174f440abbdb4238c13c8ec83 : SUCCESS in 8m 42s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6cc5673ba1f247f5acd7039429b4e402 : SUCCESS in 4m 19s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/137e675d4b614f0bb0114b510aa52b40 : SUCCESS in 4m 44s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7bd0069ca2a94ee68b74c3d92f9e4361 : SUCCESS in 11m 10s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/228b8e68eed6414699a58087638c2556 : SUCCESS in 27m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0cf0710f6f1b44f5976367e5ff3a4546 : SUCCESS in 7m 48s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/bac0741918aa403c85e3909a8814aa16 : SUCCESS in 15m 25s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3467b402aa3e450abca99b234d7c7998 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9d2898b6a7049ccb1c33180fd5f8d4d : SUCCESS in 4m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7611224256834c8ba2da3028f438f9a2 : SUCCESS in 28m 00s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b2fa0ac2de60416cb64386e00b10bcd9 : SUCCESS in 4m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d92d3d7a83f4485a96f3c09fc6d6107b : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a64d77d626bb47a6a3e6ec581c7125e3 : SUCCESS in 3m 45s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/52dd0e1f43c74327b3eba02232cf8556 : SUCCESS in 14m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/1f2ca2aebfa74f338c01133c0b79ec9e : SUCCESS in 6m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e8ee9f83a78416fa2814a9facbcd509 : SUCCESS in 1h 01m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac6dfedca396434ab9022d254dda495b : SUCCESS in 1h 29m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1b73ff70c17db5e690fd82bcc2a5ff51ecd0552b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-14 13:06:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"74e182845d24f5bc0b9c61c2bdd44eb479375ecc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 13:09:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7039f31bbbda86d8c1965bfe1eb3b22c67e3af2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 16:19:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/314b8549ae8b48f99d2c6f49ed902bf7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94ca98ed6f18427c9477f4fd3512d6ff : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23c6c830be62419aac4e47b979e24a12 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/701a232142eb46c5baeb6dce07bfd807 : SUCCESS in 2m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/92bd70e61cfa4744a255579163d7fe49 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb2177a4b4b64143beb2e5be0d086194 : SUCCESS in 3m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/021ecd08f4aa4697b81e5c0645139936 : SUCCESS in 6m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d8ab814dec38496cabb8a158cb4e4e9d : SUCCESS in 6m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/22caa632458a4f66940b6108bcc9e7d2 : SUCCESS in 8m 05s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/fee391206bac41008047f0ca98cc7f2b : SUCCESS in 5m 25s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/14a60e257884485fb7570705e1a36d05 : SUCCESS in 11m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2f9ddb28e1964d559953d5e2a5986a9d : SUCCESS in 5m 16s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/dab63b12b4534273a4512e93ee36e8b5 : SUCCESS in 7m 36s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f3343e7b184246b4902a83a399b1ac88 : SUCCESS in 7m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0fe46cd3507d4baa9dce2cab96d98e1a : SUCCESS in 15m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/72beb6b3d15d47eb9cb08f41b7ffff55 : SUCCESS in 4m 41s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c328c68c7f7b44b8b810c95d94919fbd : SUCCESS in 8m 27s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/4940b875151a4f2ba6f42f72dd989bf5 : SUCCESS in 11m 31s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/734cb0009fe14511a902e59a56b8d7ae : SUCCESS in 2m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/fd9e5b95dc1e45bdbe47d000b163aa29 : SUCCESS in 9m 31s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/226cfa53729f4108bd516329f9266783 : SUCCESS in 26m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/052031aa66dd4f61ad070c78960d38b3 : SUCCESS in 6m 27s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/57db4d4242ed4d1c84d9f5686d3a812e : SUCCESS in 15m 45s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/59001ef5600e46e8b37c4a59d5e1b2ca : SUCCESS in 5m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5d93d2e9126d47d48356d2e3ac0e09b9 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5f8f99b219c849cabef32bd469acadef : SUCCESS in 28m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c5e4c25daa7547109e15082fd45058a4 : SUCCESS in 7m 07s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/256c8ef3df2d4220b8191ba8531bd563 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f01bc5a673e14baeac651334369e3018 : SUCCESS in 6m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b66ef37df1564219a524403b040819d8 : SUCCESS in 18m 58s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/f853d39d421d45bfb98e20799721f827 : SUCCESS in 6m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f13764d23874d6a9bb9520e0a2ee866 : SUCCESS in 1h 04m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cc0a9dbb5684cfaaa4e516e677a1b40 : SUCCESS in 2h 10m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c38a0ee9f48e632b8fe753a37309c404b2334963","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 16:19:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a307b3fc6eb7afc10cad79244d5e55e82829237","revisions":{"6a307b3fc6eb7afc10cad79244d5e55e82829237":{"kind":"REWORK","_number":1,"created":"2026-09-11 13:31:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/1005231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/1005231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/1005231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/1005231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/1005231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/1005231/1"}}},"commit":{"parents":[{"commit":"0190a07f0add94372d529cbddae77bdc22f3c96c","subject":"Update constraint for gunicorn to 25.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0190a07f0add94372d529cbddae77bdc22f3c96c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-11 13:21:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-11 13:21:18.000000000","tz":0},"subject":"update constraint for glance_store to new release 5.4.1","message":"update constraint for glance_store to new release 5.4.1\n\nmeta: version: 5.4.1\nmeta: team: glance\nmeta: diff-start: -\nmeta: series: gazpacho\nmeta: branch: stable/2026.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\nmeta: release:Commit: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\nmeta: release:Change-Id: I023bbb6049114a4c9bac6e8b77c6e565ced8a3e0\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Code-Review+1: Cyril Roelandt \u003ccyril@redhat.com\u003e\nmeta: release:Code-Review+1: Ivan Anfimov \u003clazekteam@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic1b81d2aff31b3487f6613227f9372b48bfc6669\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a307b3fc6eb7afc10cad79244d5e55e82829237"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a307b3fc6eb7afc10cad79244d5e55e82829237"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
