)]}'
{"id":"openstack%2Frequirements~1007454","triplet_id":"openstack%2Frequirements~stable%2F2026.2~I5d134bd364593245038839d7bd7d6e6c89841292","project":"openstack/requirements","branch":"stable/2026.2","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-10-05 13:14:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5d134bd364593245038839d7bd7d6e6c89841292","subject":"update constraint for oslo.log to new release 8.3.2","status":"MERGED","created":"2026-09-25 11:23:44.000000000","updated":"2026-10-05 13:14:10.000000000","submitted":"2026-10-05 13:14:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007454-new-release","meta_rev_id":"807459ca45d7dc9fef57a3872486453ac3ff6519","_number":1007454,"virtual_id_number":1007454,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-05 13:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"date":"2026-10-02 10:36:48.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-10-05 11:37:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-10-05 11:37:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-25 11:23:44.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-25 12:58:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-05 11:37:04.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"9b91ca1aef467dc62bedd2a13e4e2157e855f75d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-25 11:23:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6617b5a24c6d523140c7dbce6685cd5514fb6f78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 12:58:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/135a9289afb941afacc993605d4bde32\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7970dada3af45eba5cd4e2f0bf574a7 : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62161bd1e3484ce288ca6425b6b4748a : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e858f32718dc4166bf663b7bafd337bb : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/52a5484e5370431da6716267a923a1d7 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24cb5296c9f34c5cab0eed7f82b40250 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/60da512000cc45ccac5af2d9ef2ea0c8 : SUCCESS in 3m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1caadb01f58444fa9b6c29ee0e42c046 : SUCCESS in 9m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2552b053106145a2bad8f63ea3857b06 : SUCCESS in 5m 45s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/5c01b9601c434cf886d8635b0d4a3ce1 : SUCCESS in 5m 34s\n- requirements-tox-py314-check-uc https://zuul.opendev.org/t/openstack/build/4038907784184b0983e317d75824f265 : SUCCESS in 10m 23s (non-voting)\n- cross-aodh-py313 https://zuul.opendev.org/t/openstack/build/b9c312b742a644bea30a1a69e8605577 : SUCCESS in 2m 58s\n- cross-barbican-py313 https://zuul.opendev.org/t/openstack/build/713012e9b8414963b074b3da035785a6 : SUCCESS in 3m 05s\n- cross-ceilometer-py313 https://zuul.opendev.org/t/openstack/build/1518ecb030bd4bceb815871999f2d83f : SUCCESS in 3m 54s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/69ffe1341beb463a879e5844a97ec22b : SUCCESS in 10m 46s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/bb31fb3a5c6b4f55855859f04076c893 : SUCCESS in 5m 00s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/11d158f68a7d4f82ad2a2ffcba292c30 : SUCCESS in 3m 40s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/527a03bffdf648958816123dc882d05d : SUCCESS in 6m 10s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/2f3d6d4869c2443a96dee080e4b5dbb3 : SUCCESS in 17m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3fed334abdd444fa91e4a2d8407789b3 : SUCCESS in 4m 55s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/4e4687efc1504302a3494cf08427e284 : SUCCESS in 5m 05s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/a5c200b7e7844de98105b16056367955 : SUCCESS in 7m 12s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/f195402ced9d4b9f946d237cb3e65582 : SUCCESS in 2m 54s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/d14405bc41c64914bee342a140ddff2c : SUCCESS in 8m 02s\n- cross-magnum-py313 https://zuul.opendev.org/t/openstack/build/b3e6e9ee44fa4b8588e8b3949a9e8323 : SUCCESS in 3m 07s\n- cross-masakari-py313 https://zuul.opendev.org/t/openstack/build/d481df4476114deabc46ef89353ffbe2 : SUCCESS in 4m 05s (non-voting)\n- cross-mistral-py313 https://zuul.opendev.org/t/openstack/build/66e61648f3474ffea777b842fcc88102 : SUCCESS in 10m 38s (non-voting)\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/0399210a6f7e44d191efc89c837412a2 : SUCCESS in 27m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/24e33a0efd614d338444db0696184714 : SUCCESS in 3m 50s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/5e8a1b1ee7d242fe848f1547f3379783 : SUCCESS in 10m 35s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/d3e98750a5f1437d9324d6564f14b6e2 : SUCCESS in 4m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/247db6c4294b4bc9a627944fe86dbd2b : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a82ca0ddfb7d4cba9f4fdce7e01ffe33 : SUCCESS in 26m 55s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/21569a0c4b4f42ba96b56202d9852c13 : SUCCESS in 3m 39s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/846fa284d2354276b34de4cde92c2c83 : SUCCESS in 3m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/875873546cab4e97baff6777441a0fa6 : SUCCESS in 6m 02s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/076a6fb9a2bf492dad41753af5a65cb9 : SUCCESS in 19m 25s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/0e0a2dc4cf99411ba3782fc7a3133776 : SUCCESS in 5m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8635d1c2d40c4068bb1f492c235473d8 : SUCCESS in 1h 02m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02a0d73833be411ba940ca22d32808cd : SUCCESS in 1h 27m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5290ffb9029bee7d6b15a6b9288be4c3695b2278","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-27 10:00:45.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"773ab438bada0d306bb09a692dc90d965e43869d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-10-02 10:36:48.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc819b91cc78cb6d9011b21c2e5c17229a6bd0e9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-10-05 11:37:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c7a001a322de6636e8b223574c7055530ea4c1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 11:38:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"074d0223e857da7dcfceff029a3ca0a54595133f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 13:14:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50668b4397dc4cfe9bb4b5531f5309a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cccd08df73646eb9c894a55c66d6e86 : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/088ef6c8ad894463965893016a9fc85c : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/72e10994327e49f88424c62c0b0c7e7b : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa8fd3541d3748c497ac79f5a51ef7d9 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/01df6917c3374b1ab20f5995ae25b0bd : SUCCESS in 3m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1a5b903547f5482794d168a35571b171 : SUCCESS in 7m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2d850c064b6446c08a9ce264dfabeb30 : SUCCESS in 6m 32s\n- requirements-tox-py313-check-uc https://zuul.opendev.org/t/openstack/build/bb5bb99ede284abdb8f9bddd914bff28 : SUCCESS in 3m 44s\n- cross-cinder-py313 https://zuul.opendev.org/t/openstack/build/4003687216834f4ea25e11b33cfabbed : SUCCESS in 11m 49s\n- cross-designate-py313 https://zuul.opendev.org/t/openstack/build/153f0e9efd2b4b708ffe41ab669e224d : SUCCESS in 5m 46s\n- cross-glance-py313 https://zuul.opendev.org/t/openstack/build/a933f80e4fda4a8b8668c6afff02c525 : SUCCESS in 4m 39s\n- cross-heat-py313 https://zuul.opendev.org/t/openstack/build/df0013ccf2b142bd849ba35b03ac586f : SUCCESS in 5m 58s\n- cross-horizon-py313 https://zuul.opendev.org/t/openstack/build/7e9ee62306404796b3efeaeb5d3820d2 : SUCCESS in 19m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec9156fdd9ce436ba7f49f89626dd01a : SUCCESS in 6m 16s\n- cross-ironic-py313 https://zuul.opendev.org/t/openstack/build/7a3b488091814938a47089a8e801ea6e : SUCCESS in 8m 50s\n- cross-keystone-py313 https://zuul.opendev.org/t/openstack/build/6931c0577c9d4d95903b216598fa385d : SUCCESS in 12m 27s\n- cross-kuryr-py313 https://zuul.opendev.org/t/openstack/build/8ba322176a7a4ea29e65951ca0a03b02 : SUCCESS in 2m 46s\n- cross-manila-py313 https://zuul.opendev.org/t/openstack/build/845a0cde1c44453f9f46b0fbcac32bf6 : SUCCESS in 7m 23s\n- cross-neutron-py313 https://zuul.opendev.org/t/openstack/build/c6b16c07df3e4cbe83b17b5d0993283c : SUCCESS in 25m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4513c28ad99c4ed6887b9584aaa11c11 : SUCCESS in 5m 46s\n- cross-nova-py313 https://zuul.opendev.org/t/openstack/build/9189acf476b9446dbf905f646259f2ac : SUCCESS in 11m 01s\n- cross-placement-py313 https://zuul.opendev.org/t/openstack/build/6d2364127c754d1ba0a2d1d5d4f900c7 : SUCCESS in 3m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/744c9225490b4d65ac74333b1e2f3e2b : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d877bbcb4ca4c43bb98b9c6452f863d : SUCCESS in 27m 09s\n- cross-placement-functional-py313 https://zuul.opendev.org/t/openstack/build/4a8e98e8e0ab4b44a08f82fd9d6bb439 : SUCCESS in 5m 44s\n- cross-octavia-py313 https://zuul.opendev.org/t/openstack/build/af7edddc62b84c2f915d51ab2a6ba5dd : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/93da8b16d3f840bcb9bfba60a624b042 : SUCCESS in 6m 18s\n- cross-swift-py313 https://zuul.opendev.org/t/openstack/build/0edae90e20de4bee8fe2712ddcf5dac5 : SUCCESS in 9m 35s\n- cross-watcher-py3 https://zuul.opendev.org/t/openstack/build/119b554f2e6f4007a821f779deafb6b8 : SUCCESS in 5m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7a0c5b8620a4b3f82636cc0b2585dd3 : SUCCESS in 1h 04m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba5fef5c69e84df8adca8c384ee15c5c : SUCCESS in 1h 10m 50s","accounts_in_message":[],"_revision_number":1},{"id":"807459ca45d7dc9fef57a3872486453ac3ff6519","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 13:14:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a27d4cc44c3fe5864c2ea25362e18c0be1011cd7","revisions":{"a27d4cc44c3fe5864c2ea25362e18c0be1011cd7":{"kind":"REWORK","_number":1,"created":"2026-09-25 11:23:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/54/1007454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/1007454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/1007454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/1007454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/1007454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/1007454/1"}}},"commit":{"parents":[{"commit":"568297f47d9960a0a4ba5466aca5fd8bdd0bbd15","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/568297f47d9960a0a4ba5466aca5fd8bdd0bbd15"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-10 18:28:11.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-09-25 11:23:12.000000000","tz":120},"subject":"update constraint for oslo.log to new release 8.3.2","message":"update constraint for oslo.log to new release 8.3.2\n\nmeta: version: 8.3.2\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: hibiscus\nmeta: branch: stable/2026.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I98717b0e4b7cde86348d071404798cafd718bb4f\nmeta: release:Code-Review+2: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Workflow+1: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nChange-Id: I5d134bd364593245038839d7bd7d6e6c89841292\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a27d4cc44c3fe5864c2ea25362e18c0be1011cd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a27d4cc44c3fe5864c2ea25362e18c0be1011cd7"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
