)]}'
{"id":"openstack%2Frequirements~1008441","triplet_id":"openstack%2Frequirements~master~Id6ddc9d85a27f3bc40b4a680b020a61c29aeb15e","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-10-05 13:13:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id6ddc9d85a27f3bc40b4a680b020a61c29aeb15e","subject":"Allow projects to drop their Windows specifiers","status":"MERGED","created":"2026-10-02 11:44:00.000000000","updated":"2026-10-05 13:15:41.000000000","submitted":"2026-10-05 13:13:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1008441","meta_rev_id":"1a64b7d8ab17512cd3616306ad456e95ae32b0d2","_number":1008441,"virtual_id_number":1008441,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-05 13:13:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-10-02 13:11:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2026-10-02 22:47:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-10-05 11:17:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-10-05 11:17:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-10-02 13:08:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-02 13:11:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-10-02 22:47:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-10-05 11:17:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"389f6e3ae7b9177fdb7a1e7c4cc871e76b1c2f12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-02 11:44:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a4d551a0abe5cd0922b51b82ef0326c52cb2504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 13:08:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/20cae0c4ec3e4eb9940cda35f28b9a5c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/808e610bd8884a6fb06bc94477173f7e : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bb9013cbaeb4febbd1a8f8b222831a5 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f67e2dab350f46d4ada0d152f9f92bfb : SUCCESS in 3m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8ea4dfc5496344bdab755c1a6ca16501 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8b8c2980b314416a806d42dd92f7353 : SUCCESS in 3m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f50ff7eb2e494e7a95f4a2b674f25bf9 : SUCCESS in 1h 00m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38ba6560894c45038fb13d90fb9cbec2 : FAILURE in 58m 42s","accounts_in_message":[],"_revision_number":1},{"id":"314153a137fb7a7cc408768b297077a9fe75c191","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-10-02 13:11:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1b26093b223779c68d83279e60ec1990ce15276d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-02 17:00:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29b1359413389dc05c2314ceb89fdf311c55a277","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 18:45:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c7a72aefe3f43d3a8d6ffcc03247ffd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/827e852c162941c0a4463e94bc82d496 : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46545d393af8435994c6fb48aef518d8 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50d150c8802b4a12b4d6e0248762322f : SUCCESS in 3m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/489a57ec1a904363bdd3a7ff5c162436 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1aa4aaa4f00e4f19a8dcf1c8c501ef04 : SUCCESS in 2m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/715c7720be574b889101ab3ea01e6306 : SUCCESS in 28m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f28384ce34514f76aae2a3a9bd61aab7 : SUCCESS in 1h 30m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d3e5b930a25c3e9596cc1d88afc086bc55eeaab3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-10-02 22:47:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ae845ed57dba5839d02d0274e5188334cbb00d18","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-10-05 11:17:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"72eb57fe01bd01cffee789f0615b44b87bbf7393","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 11:18:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0eda7375d7a19059a0637ce302fae1b5ff8394ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 13:13:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a417f87c76f74626beae13ec38040bd0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7030f0eba84273af2148e3830da128 : SUCCESS in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e72c0a04c82d41cfaadca38a8062a89c : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51e55ce4b18442009b769f19c2e98fcb : SUCCESS in 2m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c61cfba96f414b7a8683ab724f599aa4 : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/355385a0c5fc4326ae91402e6acf01ac : SUCCESS in 3m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72aa70f7747e4460bc45b775362f6665 : SUCCESS in 1h 05m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/935288ae08714406ba78bfb752c119b6 : SUCCESS in 1h 50m 20s","accounts_in_message":[],"_revision_number":1},{"id":"41578cf585c51f5778ffca4477dbe2164f771a31","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 13:13:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1a64b7d8ab17512cd3616306ad456e95ae32b0d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-05 13:15:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57e2796ea6b9480db12b954743db4fb0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86d168d79b244ecc8b704b14f6624530 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d877fc93c4c674e061c2f44d32bea4d2ac762b6","revisions":{"7d877fc93c4c674e061c2f44d32bea4d2ac762b6":{"kind":"REWORK","_number":1,"created":"2026-10-02 11:44:00.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/41/1008441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/1008441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/1008441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/1008441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/1008441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/1008441/1"}}},"commit":{"parents":[{"commit":"abe5957924a34676b3f0a2b791084852cbee5b38","subject":"Remove backports.txt file, backport handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/abe5957924a34676b3f0a2b791084852cbee5b38"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-02 10:45:30.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-02 11:41:51.000000000","tz":60},"subject":"Allow projects to drop their Windows specifiers","message":"Allow projects to drop their Windows specifiers\n\nOpenStack no longer supports Windows. Additionally, pbr has a bug with\nsetuptools \u003e\u003d 83.0.0 where it silently drops dependencies that carry PEP\n508 environment markers like sys_platform!\u003d\u0027win32\u0027 [1]. Ultimately the\nfix here exists in pbr, but there\u0027s no reason we shouldn\u0027t let projects\ndrop these specifiers anyway, given we no longer care about Windows.\n\n[1] https://bugs.launchpad.net/pbr/+bug/2162730\n\nChange-Id: Id6ddc9d85a27f3bc40b4a680b020a61c29aeb15e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d877fc93c4c674e061c2f44d32bea4d2ac762b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d877fc93c4c674e061c2f44d32bea4d2ac762b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
