)]}'
{"id":"openstack%2Frequirements~640710","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I865bc26623d8a460cee7cbf2e0a5e8f9e67e27a3","project":"openstack/requirements","branch":"stable/queens","topic":"story/2004979","hashtags":[],"change_id":"I865bc26623d8a460cee7cbf2e0a5e8f9e67e27a3","subject":"Update requests to 2.20.1 for CVE-2018-18074","status":"ABANDONED","created":"2019-03-04 10:10:21.000000000","updated":"2019-09-22 05:15:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"021e67c0855c89e92afc9b3c7face5db703cfdb8","_number":640710,"virtual_id_number":640710,"owner":{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},{"value":-1,"date":"2019-05-07 22:05:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"disliked":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2019-05-07 20:26:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-07 20:26:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2019-05-14 04:38:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-05-07 20:26:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"date":"2019-05-07 20:26:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-04 17:17:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-03-04 17:17:36.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-03 15:43:43.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-07 22:05:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-14 04:38:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b309141dd6bd21acda22cc6b95241f25c948ab1b","author":{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},"date":"2019-03-04 10:10:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23a41a78d15cea18af83a281f03aa9d5af6054a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-04 11:37:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/10/640710/1/check/openstack-tox-pypy/495571c/ : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/10/640710/1/check/openstack-tox-pep8/0dc4944/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/10/640710/1/check/openstack-tox-py27/0b968bf/ : SUCCESS in 3m 20s\n- openstack-tox-py35 http://logs.openstack.org/10/640710/1/check/openstack-tox-py35/4098075/ : SUCCESS in 3m 32s\n- openstack-tox-validate http://logs.openstack.org/10/640710/1/check/openstack-tox-validate/def519d/ : SUCCESS in 3m 29s\n- requirements-tox-babel http://logs.openstack.org/10/640710/1/check/requirements-tox-babel/520621f/ : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc http://logs.openstack.org/10/640710/1/check/requirements-tox-py27-check-uc/d1f50ff/ : SUCCESS in 6m 01s\n- requirements-tox-py35-check-uc http://logs.openstack.org/10/640710/1/check/requirements-tox-py35-check-uc/3b6addb/ : SUCCESS in 6m 18s\n- cross-cinder-py27 http://logs.openstack.org/10/640710/1/check/cross-cinder-py27/5e413d7/ : SUCCESS in 9m 13s\n- cross-cinder-py35 http://logs.openstack.org/10/640710/1/check/cross-cinder-py35/d342662/ : SUCCESS in 9m 02s\n- cross-glance-py27 http://logs.openstack.org/10/640710/1/check/cross-glance-py27/7ec2434/ : SUCCESS in 5m 03s\n- cross-glance-py35 http://logs.openstack.org/10/640710/1/check/cross-glance-py35/ddd829d/ : SUCCESS in 4m 09s\n- cross-horizon-py27 http://logs.openstack.org/10/640710/1/check/cross-horizon-py27/30e1adc/ : SUCCESS in 12m 28s\n- cross-horizon-py35 http://logs.openstack.org/10/640710/1/check/cross-horizon-py35/bc79ebd/ : SUCCESS in 9m 31s\n- cross-keystone-py27 http://logs.openstack.org/10/640710/1/check/cross-keystone-py27/2ebe2c7/ : SUCCESS in 13m 53s\n- cross-keystone-py35 http://logs.openstack.org/10/640710/1/check/cross-keystone-py35/43d2a7a/ : SUCCESS in 12m 22s\n- cross-neutron-py27 http://logs.openstack.org/10/640710/1/check/cross-neutron-py27/7df7a6b/ : SUCCESS in 30m 53s\n- cross-neutron-py35 http://logs.openstack.org/10/640710/1/check/cross-neutron-py35/5eb1ff1/ : SUCCESS in 25m 23s\n- cross-nova-py27 http://logs.openstack.org/10/640710/1/check/cross-nova-py27/ae5ed14/ : SUCCESS in 12m 45s\n- cross-nova-py35 http://logs.openstack.org/10/640710/1/check/cross-nova-py35/f69a9ba/ : SUCCESS in 12m 23s\n- cross-nova-functional http://logs.openstack.org/10/640710/1/check/cross-nova-functional/a78e5ae/ : SUCCESS in 20m 55s\n- cross-swift-py27 http://logs.openstack.org/10/640710/1/check/cross-swift-py27/001e957/ : SUCCESS in 5m 42s\n- cross-swift-py35 http://logs.openstack.org/10/640710/1/check/cross-swift-py35/162d316/ : SUCCESS in 3m 48s\n- requirements-integration http://logs.openstack.org/10/640710/1/check/requirements-integration/6c2e0ee/ : SUCCESS in 56m 02s\n- neutron-grenade-multinode http://logs.openstack.org/10/640710/1/check/neutron-grenade-multinode/ae3c490/ : SUCCESS in 55m 27s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/10/640710/1/check/legacy-tempest-dsvm-neutron-full/f332eb8/ : SUCCESS in 1h 20m 23s\n- neutron-grenade http://logs.openstack.org/10/640710/1/check/neutron-grenade/686e71d/ : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0e10c0fc8129711b35948d3033c7f87cdac3b9c3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-03-06 21:53:56.000000000","message":"Patch Set 1: Code-Review-1\n\nWe will not be updating upper-constraints for branches past rocky while the following reasons are still valid (links)\n\nhttps://storyboard.openstack.org/#!/story/2004978\nhttps://storyboard.openstack.org/#!/story/2004979\n\nIf we don\u0027t have reasoning to continue with this I\u0027ll close this in a week.  Thanks for submitting and being concerned about this though.","accounts_in_message":[],"_revision_number":1},{"id":"d23ddf3d639234d5176d4f1b9a2a8a1cd2f49320","author":{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},"date":"2019-03-07 09:20:01.000000000","message":"Patch Set 1:\n\n@Matthew Thode - what reason are you trying to highlight from here https://storyboard.openstack.org/#!/story/2004979. That story needs this change-set to be resolved.","accounts_in_message":[],"_revision_number":1},{"id":"b6279c2d9cd033cd80f23af34d43c4a6473400bb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-03-07 13:39:46.000000000","message":"Patch Set 1:\n\nThe reasoning from those linked issues are mainly the large changeset between the version of requests currently in use and 2.20.1.  That the stable team does not support this is also an important note.","accounts_in_message":[],"_revision_number":1},{"id":"1e9957bdd9fdb47191ab1ce333e3f55ee822a543","author":{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},"date":"2019-03-12 18:53:10.000000000","message":"Patch Set 1:\n\n@Matthew Thode - can we do a compatibility check somehow to see whether upgrading requests package to 2.20.1 on queens doesn\u0027t break anything? coz i don\u0027t see any major improvement going in requests library from 2.14.2 to 2.20.1 which might affect anything in openstack.","accounts_in_message":[],"_revision_number":1},{"id":"6dbf255b3a681671885d9414a624ceec6761b222","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-03 15:43:43.000000000","message":"Patch Set 1: Code-Review+1\n\nI am supportive of that change. I am not particularly concerned about the version update in requests, but we probably have to extend testing and inform people a bit more (or potentially search for git commits on newer branches that touched constraints *together* with other code changes)","accounts_in_message":[],"_revision_number":1},{"id":"4039e226d9f4d90797db46f286e634a4a8660d34","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-03 22:48:49.000000000","message":"Patch Set 1:\n\nactually we just need 2.19.1 here according to \nhttps://github.com/pyupio/safety-db/blob/master/data/insecure_full.json#L8205","accounts_in_message":[],"_revision_number":1},{"id":"97b8da12941e9a264bb4c54a14cf1bc795f78db3","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-03 22:50:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f005fa6679e9ee008cf46f2f4569cdee3f88d8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 00:12:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/10/640710/2/check/openstack-tox-pypy/7537d88/ : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/10/640710/2/check/openstack-tox-pep8/937f98b/ : SUCCESS in 2m 44s\n- openstack-tox-py27 http://logs.openstack.org/10/640710/2/check/openstack-tox-py27/b1ee0a6/ : SUCCESS in 3m 05s\n- openstack-tox-py35 http://logs.openstack.org/10/640710/2/check/openstack-tox-py35/644aef6/ : SUCCESS in 3m 18s\n- openstack-tox-validate http://logs.openstack.org/10/640710/2/check/openstack-tox-validate/69bfd05/ : SUCCESS in 3m 20s\n- requirements-tox-babel http://logs.openstack.org/10/640710/2/check/requirements-tox-babel/a6cf3ba/ : SUCCESS in 2m 46s\n- requirements-tox-py27-check-uc http://logs.openstack.org/10/640710/2/check/requirements-tox-py27-check-uc/619d223/ : SUCCESS in 6m 39s\n- requirements-tox-py35-check-uc http://logs.openstack.org/10/640710/2/check/requirements-tox-py35-check-uc/ac3ee53/ : SUCCESS in 7m 46s\n- cross-cinder-py27 http://logs.openstack.org/10/640710/2/check/cross-cinder-py27/8dc27ee/ : SUCCESS in 8m 02s\n- cross-cinder-py35 http://logs.openstack.org/10/640710/2/check/cross-cinder-py35/382f9c0/ : SUCCESS in 10m 16s\n- cross-glance-py27 http://logs.openstack.org/10/640710/2/check/cross-glance-py27/f17daff/ : SUCCESS in 4m 10s\n- cross-glance-py35 http://logs.openstack.org/10/640710/2/check/cross-glance-py35/d1c1d81/ : SUCCESS in 4m 06s\n- cross-horizon-py27 http://logs.openstack.org/10/640710/2/check/cross-horizon-py27/83c5dc3/ : SUCCESS in 11m 50s\n- cross-horizon-py35 http://logs.openstack.org/10/640710/2/check/cross-horizon-py35/b1dc401/ : SUCCESS in 11m 15s\n- cross-keystone-py27 http://logs.openstack.org/10/640710/2/check/cross-keystone-py27/156a590/ : SUCCESS in 12m 25s\n- cross-keystone-py35 http://logs.openstack.org/10/640710/2/check/cross-keystone-py35/f264a2e/ : SUCCESS in 11m 34s\n- cross-neutron-py27 http://logs.openstack.org/10/640710/2/check/cross-neutron-py27/f140ee4/ : SUCCESS in 24m 07s\n- cross-neutron-py35 http://logs.openstack.org/10/640710/2/check/cross-neutron-py35/43490d8/ : SUCCESS in 30m 07s\n- cross-nova-py27 http://logs.openstack.org/10/640710/2/check/cross-nova-py27/2219e6f/ : SUCCESS in 11m 44s\n- cross-nova-py35 http://logs.openstack.org/10/640710/2/check/cross-nova-py35/b684858/ : SUCCESS in 13m 54s\n- cross-nova-functional http://logs.openstack.org/10/640710/2/check/cross-nova-functional/a84e268/ : SUCCESS in 20m 31s\n- cross-swift-py27 http://logs.openstack.org/10/640710/2/check/cross-swift-py27/4a51e2d/ : SUCCESS in 5m 29s\n- cross-swift-py35 http://logs.openstack.org/10/640710/2/check/cross-swift-py35/dd09703/ : SUCCESS in 3m 18s\n- requirements-integration http://logs.openstack.org/10/640710/2/check/requirements-integration/7e9b80f/ : SUCCESS in 1h 02m 32s\n- neutron-grenade-multinode http://logs.openstack.org/10/640710/2/check/neutron-grenade-multinode/9de7f0d/ : SUCCESS in 1h 03m 16s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/10/640710/2/check/legacy-tempest-dsvm-neutron-full/06df813/ : SUCCESS in 1h 17m 37s\n- neutron-grenade http://logs.openstack.org/10/640710/2/check/neutron-grenade/f36177f/ : SUCCESS in 1h 09m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9eea4de8043cae047c1c855c1626f00f9da8bba6","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-06 14:09:13.000000000","message":"Patch Set 2:\n\n\u003e actually we just need 2.19.1 here according to\n \u003e https://github.com/pyupio/safety-db/blob/master/data/insecure_full.json#L8205\n\nDoesn\u0027t that say that all versions \u003c\u003d 2.19.1 are vulnerable meaning that 2.20.0 is the first fixed version but that was broken in other ways and is blocked so I do think that 2.20.1 is the version we want.","accounts_in_message":[],"_revision_number":2},{"id":"5ffd295b0134feae9b07c8a819401c496cb15385","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-07 20:26:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"48c8a23e8f2351c69e814a0b29036c522208315f","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-07 20:40:04.000000000","message":"Patch Set 2:\n\nInteresting, I was reading the description which said \"before 2.19.1\" but the specification actually says \u003c\u003d 2.19.1. checked the code, it is indeed not fixed in 2.19.1","accounts_in_message":[],"_revision_number":2},{"id":"f614f6ff5b2919cf682122b6dd6abf609a5ecd99","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-07 20:46:55.000000000","message":"Patch Set 3:\n\nfixed as https://github.com/pyupio/safety-db/pull/2278","accounts_in_message":[],"_revision_number":3},{"id":"74f9a68302a8d3fd9df8507904af0a08f4a9eba6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 22:05:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/10/640710/3/check/openstack-tox-pypy/d02724b/ : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/10/640710/3/check/openstack-tox-pep8/6c80b1d/ : SUCCESS in 2m 59s\n- openstack-tox-py27 http://logs.openstack.org/10/640710/3/check/openstack-tox-py27/ba37387/ : SUCCESS in 3m 14s\n- openstack-tox-py35 http://logs.openstack.org/10/640710/3/check/openstack-tox-py35/47eb5a6/ : SUCCESS in 3m 53s\n- openstack-tox-validate http://logs.openstack.org/10/640710/3/check/openstack-tox-validate/4fcbdc0/ : SUCCESS in 2m 56s\n- requirements-tox-babel http://logs.openstack.org/10/640710/3/check/requirements-tox-babel/8f7d048/ : SUCCESS in 2m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/10/640710/3/check/requirements-tox-py27-check-uc/ccba64d/ : SUCCESS in 6m 07s\n- requirements-tox-py35-check-uc http://logs.openstack.org/10/640710/3/check/requirements-tox-py35-check-uc/8c0c6b0/ : SUCCESS in 6m 40s\n- cross-cinder-py27 http://logs.openstack.org/10/640710/3/check/cross-cinder-py27/9911c73/ : SUCCESS in 8m 48s\n- cross-cinder-py35 http://logs.openstack.org/10/640710/3/check/cross-cinder-py35/00955fc/ : SUCCESS in 8m 39s\n- cross-glance-py27 http://logs.openstack.org/10/640710/3/check/cross-glance-py27/6bddc41/ : SUCCESS in 5m 05s\n- cross-glance-py35 http://logs.openstack.org/10/640710/3/check/cross-glance-py35/0716b82/ : SUCCESS in 4m 34s\n- cross-horizon-py27 http://logs.openstack.org/10/640710/3/check/cross-horizon-py27/ad89ca1/ : SUCCESS in 18m 38s\n- cross-horizon-py35 http://logs.openstack.org/10/640710/3/check/cross-horizon-py35/2c9c082/ : SUCCESS in 10m 53s\n- cross-keystone-py27 http://logs.openstack.org/10/640710/3/check/cross-keystone-py27/ae0b7df/ : SUCCESS in 12m 17s\n- cross-keystone-py35 http://logs.openstack.org/10/640710/3/check/cross-keystone-py35/7396010/ : SUCCESS in 13m 21s\n- cross-neutron-py27 http://logs.openstack.org/10/640710/3/check/cross-neutron-py27/9610eaa/ : SUCCESS in 27m 45s\n- cross-neutron-py35 http://logs.openstack.org/10/640710/3/check/cross-neutron-py35/739716e/ : SUCCESS in 26m 06s\n- cross-nova-py27 http://logs.openstack.org/10/640710/3/check/cross-nova-py27/a0d9eb4/ : SUCCESS in 12m 56s\n- cross-nova-py35 http://logs.openstack.org/10/640710/3/check/cross-nova-py35/d6f7e39/ : SUCCESS in 15m 02s\n- cross-nova-functional http://logs.openstack.org/10/640710/3/check/cross-nova-functional/fedf86f/ : SUCCESS in 18m 15s\n- cross-swift-py27 http://logs.openstack.org/10/640710/3/check/cross-swift-py27/4f713b4/ : SUCCESS in 5m 45s\n- cross-swift-py35 http://logs.openstack.org/10/640710/3/check/cross-swift-py35/0e52c38/ : SUCCESS in 3m 58s\n- requirements-integration http://logs.openstack.org/10/640710/3/check/requirements-integration/9b312c0/ : SUCCESS in 1h 07m 00s\n- neutron-grenade-multinode http://logs.openstack.org/10/640710/3/check/neutron-grenade-multinode/ffdb452/ : FAILURE in 54m 18s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/10/640710/3/check/legacy-tempest-dsvm-neutron-full/028b7c9/ : SUCCESS in 1h 32m 21s\n- neutron-grenade http://logs.openstack.org/10/640710/3/check/neutron-grenade/76d8786/ : FAILURE in 54m 12s","accounts_in_message":[],"_revision_number":3},{"id":"4ccc3b150c9ddc80e5979ba3745a523de3d30993","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-14 04:38:35.000000000","message":"Patch Set 3: Workflow-1\n\nwe need to answer some outstanding questions as a project before we can move forward here.  At the moment it\u0027s not a technical issue with the PR but about setting a precident  https://etherpad.openstack.org/p/non-openstack-stable-security-constraints-updates","accounts_in_message":[],"_revision_number":3},{"id":"91f1fe49e5135433a8fc369948e1c21577971a5e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 10:19:07.000000000","message":"Abandoned\n\njust merged","accounts_in_message":[],"_revision_number":3},{"id":"b3ff76402cef13c9b4e6569beff1b8d77952165c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 20:59:58.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"35582f10534f3318e91fe262cbb917b23d6d8974","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 21:00:10.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"b8b7fbb1dfb9011663ce86e7108a40dcaa9d743b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-22 05:15:03.000000000","message":"Abandoned\n\nabandoning, it\u0027d be best to rely on downstream packagers (of which projects like OSA can be considered) to manage security updates of dependencies.  Often patches are backported and version numbers largely unchanged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"558ef9368f518deb67806c9d330b3c9128849a12","revisions":{"c07cbe00dfad1012290dbb058178d6fe5e357ee5":{"kind":"REWORK","_number":1,"created":"2019-03-04 10:10:21.000000000","uploader":{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},"ref":"refs/changes/10/640710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/640710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/640710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/640710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/640710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/640710/1"}}},"commit":{"parents":[{"commit":"a1416be0ee111bfcc91d4be45403a6158c355d18","subject":"remove publish-loci-requirements post-job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1416be0ee111bfcc91d4be45403a6158c355d18"}]}],"author":{"name":"Abhishek Sharma","email":"abmsharm@in.ibm.com","date":"2019-03-04 10:06:13.000000000","tz":-300},"committer":{"name":"Abhishek Sharma","email":"abmsharm@in.ibm.com","date":"2019-03-04 10:06:13.000000000","tz":-300},"subject":"Update requests to 2.20.1 for CVE-2018-18074","message":"Update requests to 2.20.1 for CVE-2018-18074\n\nGiven that the OpenStack deployment projects, and many operators\nmake use of upper-constraints for building OpenStack from source,\nit would be prudent to update the requests library to the release\nwhich takes care of this vulnerability.\n\nChange-Id: I865bc26623d8a460cee7cbf2e0a5e8f9e67e27a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c07cbe00dfad1012290dbb058178d6fe5e357ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c07cbe00dfad1012290dbb058178d6fe5e357ee5"}]},"branch":"refs/heads/stable/queens"},"359b8ee4d9b75e80626274a3c934fabdbef3a184":{"kind":"REWORK","_number":2,"created":"2019-05-03 22:50:11.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/10/640710/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/640710/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/640710/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/640710/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/640710/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/640710/2"}}},"commit":{"parents":[{"commit":"a1416be0ee111bfcc91d4be45403a6158c355d18","subject":"remove publish-loci-requirements post-job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1416be0ee111bfcc91d4be45403a6158c355d18"}]}],"author":{"name":"Abhishek Sharma","email":"abmsharm@in.ibm.com","date":"2019-03-04 10:06:13.000000000","tz":-300},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-03 22:49:47.000000000","tz":120},"subject":"Update requests to 2.19.1 for CVE-2018-18074","message":"Update requests to 2.19.1 for CVE-2018-18074\n\nGiven that the OpenStack deployment projects, and many operators\nmake use of upper-constraints for building OpenStack from source,\nit would be prudent to update the requests library to the release\nwhich takes care of this vulnerability.\n\nNOTE(dmllr): Lowered to 2.19.1 for older stable/ releases\n\nChange-Id: I865bc26623d8a460cee7cbf2e0a5e8f9e67e27a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/359b8ee4d9b75e80626274a3c934fabdbef3a184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/359b8ee4d9b75e80626274a3c934fabdbef3a184"}]},"branch":"refs/heads/stable/queens"},"558ef9368f518deb67806c9d330b3c9128849a12":{"kind":"REWORK","_number":3,"created":"2019-05-07 20:26:26.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/10/640710/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/640710/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/640710/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/640710/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/640710/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/640710/3"}}},"commit":{"parents":[{"commit":"a1416be0ee111bfcc91d4be45403a6158c355d18","subject":"remove publish-loci-requirements post-job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1416be0ee111bfcc91d4be45403a6158c355d18"}]}],"author":{"name":"Abhishek Sharma","email":"abmsharm@in.ibm.com","date":"2019-03-04 10:06:13.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-07 20:26:07.000000000","tz":-300},"subject":"Update requests to 2.20.1 for CVE-2018-18074","message":"Update requests to 2.20.1 for CVE-2018-18074\n\nGiven that the OpenStack deployment projects, and many operators\nmake use of upper-constraints for building OpenStack from source,\nit would be prudent to update the requests library to the release\nwhich takes care of this vulnerability.\n\nChange-Id: I865bc26623d8a460cee7cbf2e0a5e8f9e67e27a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/558ef9368f518deb67806c9d330b3c9128849a12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/558ef9368f518deb67806c9d330b3c9128849a12"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
