)]}'
{"id":"openstack%2Frequirements~640727","triplet_id":"openstack%2Frequirements~stable%2Fpike~I0dc7182eadb5a07c571c985d8281a8e85b13fd5c","project":"openstack/requirements","branch":"stable/pike","topic":"story/2004979","hashtags":[],"change_id":"I0dc7182eadb5a07c571c985d8281a8e85b13fd5c","subject":"Update requests to 2.20.1 for CVE-2018-18074","status":"ABANDONED","created":"2019-03-04 10:51:30.000000000","updated":"2019-09-22 05:15:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"026f9f7c1777d9714774b0b787182f1edbf756af","_number":640727,"virtual_id_number":640727,"owner":{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},{"value":1,"date":"2019-05-22 22:39:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2019-05-14 04:38:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2019-05-14 04:38:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"date":"2019-05-10 21:15:39.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-04 17:18:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-03-04 17:18:57.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-03-04 17:19:12.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-10 21:15:39.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2019-05-14 04:38:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-22 22:39:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1680c13ab4f2f18d2fd49ae5ccf76a47ce5e3d8","author":{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},"date":"2019-03-04 10:51:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a5823349d1b69461f9ed44e75a862574aa1f94e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-04 12:16:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/27/640727/1/check/openstack-tox-pypy/29cbe58/ : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/27/640727/1/check/openstack-tox-pep8/bf52b53/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/27/640727/1/check/openstack-tox-py27/5a06b91/ : SUCCESS in 4m 05s\n- openstack-tox-py35 http://logs.openstack.org/27/640727/1/check/openstack-tox-py35/fbeb270/ : SUCCESS in 4m 25s\n- openstack-tox-validate http://logs.openstack.org/27/640727/1/check/openstack-tox-validate/812294e/ : SUCCESS in 4m 18s\n- requirements-tox-babel http://logs.openstack.org/27/640727/1/check/requirements-tox-babel/3b352f0/ : SUCCESS in 4m 14s\n- requirements-tox-py27-check-uc http://logs.openstack.org/27/640727/1/check/requirements-tox-py27-check-uc/e06be29/ : SUCCESS in 6m 44s\n- cross-cinder-py27 http://logs.openstack.org/27/640727/1/check/cross-cinder-py27/9dca1ed/ : SUCCESS in 8m 58s\n- cross-cinder-py35 http://logs.openstack.org/27/640727/1/check/cross-cinder-py35/f874106/ : SUCCESS in 8m 00s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/27/640727/1/check/cross-glance-py27/ffda85c/ : SUCCESS in 9m 49s\n- cross-glance-py35 http://logs.openstack.org/27/640727/1/check/cross-glance-py35/838e93d/ : SUCCESS in 11m 52s\n- cross-horizon-py27 http://logs.openstack.org/27/640727/1/check/cross-horizon-py27/c04212f/ : SUCCESS in 13m 03s\n- cross-horizon-py35 http://logs.openstack.org/27/640727/1/check/cross-horizon-py35/96acd32/ : SUCCESS in 10m 06s\n- cross-keystone-py27 http://logs.openstack.org/27/640727/1/check/cross-keystone-py27/803b6e0/ : SUCCESS in 11m 35s\n- cross-keystone-py35 http://logs.openstack.org/27/640727/1/check/cross-keystone-py35/9b478eb/ : SUCCESS in 14m 24s\n- cross-neutron-py27 http://logs.openstack.org/27/640727/1/check/cross-neutron-py27/009d4cd/ : SUCCESS in 25m 11s\n- cross-neutron-py35 http://logs.openstack.org/27/640727/1/check/cross-neutron-py35/d28837c/ : SUCCESS in 21m 54s\n- cross-nova-py27 http://logs.openstack.org/27/640727/1/check/cross-nova-py27/56f2dbb/ : SUCCESS in 15m 13s\n- cross-nova-py35 http://logs.openstack.org/27/640727/1/check/cross-nova-py35/0336e3b/ : SUCCESS in 12m 50s\n- cross-nova-functional http://logs.openstack.org/27/640727/1/check/cross-nova-functional/2e63094/ : SUCCESS in 21m 15s\n- cross-swift-py27 http://logs.openstack.org/27/640727/1/check/cross-swift-py27/e9b3151/ : SUCCESS in 7m 21s\n- cross-swift-py35 http://logs.openstack.org/27/640727/1/check/cross-swift-py35/e82858d/ : SUCCESS in 4m 11s\n- requirements-integration http://logs.openstack.org/27/640727/1/check/requirements-integration/7d47308/ : SUCCESS in 1h 00m 05s\n- neutron-grenade-multinode http://logs.openstack.org/27/640727/1/check/neutron-grenade-multinode/b8aaa0f/ : SUCCESS in 1h 02m 00s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/27/640727/1/check/legacy-tempest-dsvm-neutron-full/fa66269/ : SUCCESS in 1h 19m 33s\n- neutron-grenade http://logs.openstack.org/27/640727/1/check/neutron-grenade/64d826f/ : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8e64f6b49c83772c9d604f1870ff84183dcaedda","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-03-06 21:53:54.000000000","message":"Patch Set 1: Code-Review-1\n\nWe will not be updating upper-constraints for branches past rocky while the following reasons are still valid (links)\n\nhttps://storyboard.openstack.org/#!/story/2004978\nhttps://storyboard.openstack.org/#!/story/2004979\n\nIf we don\u0027t have reasoning to continue with this I\u0027ll close this in a week.  Thanks for submitting and being concerned about this though.","accounts_in_message":[],"_revision_number":1},{"id":"4e97c5fc0513aa4d8ed248866657a99774006820","author":{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},"date":"2019-03-07 09:26:33.000000000","message":"Patch Set 1:\n\n@Matthew Thode - what reason are you trying to highlight from here https://storyboard.openstack.org/#!/story/2004979. That story needs this change-set to be resolved.","accounts_in_message":[],"_revision_number":1},{"id":"23bc2254992405ff990eb8ff2abdeebad8e70a11","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-03-07 13:39:43.000000000","message":"Patch Set 1:\n\nThe reasoning from those linked issues are mainly the large changeset between the version of requests currently in use and 2.20.1.  That the stable team does not support this is also an important note.","accounts_in_message":[],"_revision_number":1},{"id":"d26e5780f7e7cbea52476d5b2ae9ed14029fa919","author":{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},"date":"2019-03-12 18:53:32.000000000","message":"Patch Set 1:\n\n@Matthew Thode - can we do a compatibility check somehow to see whether upgrading requests package to 2.20.1 on pike doesn\u0027t break anything? coz i don\u0027t see any major improvement going in requests library from 2.14.2 to 2.20.1 which might affect anything in openstack.","accounts_in_message":[],"_revision_number":1},{"id":"014f482b9b7ad76067ec82db302c92d831aa18c8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-05-10 21:15:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6487498ade1a9ee6929bbbc19bf398ade575b78c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 22:47:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/27/640727/2/check/openstack-tox-pypy/6b4ee9a/ : FAILURE in 2m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/27/640727/2/check/openstack-tox-pep8/15d9696/ : FAILURE in 2m 40s\n- openstack-tox-py27 http://logs.openstack.org/27/640727/2/check/openstack-tox-py27/8770d5c/ : FAILURE in 2m 53s\n- openstack-tox-py35 http://logs.openstack.org/27/640727/2/check/openstack-tox-py35/ece945f/ : SUCCESS in 3m 20s\n- openstack-tox-validate http://logs.openstack.org/27/640727/2/check/openstack-tox-validate/56d3686/ : FAILURE in 2m 33s\n- requirements-tox-babel http://logs.openstack.org/27/640727/2/check/requirements-tox-babel/e9b4811/ : SUCCESS in 2m 50s\n- requirements-tox-py27-check-uc http://logs.openstack.org/27/640727/2/check/requirements-tox-py27-check-uc/ab9fb48/ : SUCCESS in 7m 24s\n- cross-cinder-py27 http://logs.openstack.org/27/640727/2/check/cross-cinder-py27/66540ec/ : SUCCESS in 6m 51s\n- cross-cinder-py35 http://logs.openstack.org/27/640727/2/check/cross-cinder-py35/57ac688/ : SUCCESS in 8m 47s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/27/640727/2/check/cross-glance-py27/3ffdd83/ : SUCCESS in 8m 16s\n- cross-glance-py35 http://logs.openstack.org/27/640727/2/check/cross-glance-py35/b6148f2/ : SUCCESS in 9m 40s\n- cross-horizon-py27 http://logs.openstack.org/27/640727/2/check/cross-horizon-py27/43718b2/ : SUCCESS in 11m 12s\n- cross-horizon-py35 http://logs.openstack.org/27/640727/2/check/cross-horizon-py35/80bf065/ : SUCCESS in 9m 33s\n- cross-keystone-py27 http://logs.openstack.org/27/640727/2/check/cross-keystone-py27/0e59aad/ : SUCCESS in 12m 17s\n- cross-keystone-py35 http://logs.openstack.org/27/640727/2/check/cross-keystone-py35/5fb2371/ : SUCCESS in 12m 49s\n- cross-neutron-py27 http://logs.openstack.org/27/640727/2/check/cross-neutron-py27/2172937/ : SUCCESS in 21m 34s\n- cross-neutron-py35 http://logs.openstack.org/27/640727/2/check/cross-neutron-py35/d6e2ca3/ : SUCCESS in 23m 23s\n- cross-nova-py27 http://logs.openstack.org/27/640727/2/check/cross-nova-py27/994e7f6/ : SUCCESS in 12m 57s\n- cross-nova-py35 http://logs.openstack.org/27/640727/2/check/cross-nova-py35/6281025/ : SUCCESS in 12m 15s\n- cross-nova-functional http://logs.openstack.org/27/640727/2/check/cross-nova-functional/74e47c5/ : SUCCESS in 18m 18s\n- cross-swift-py27 http://logs.openstack.org/27/640727/2/check/cross-swift-py27/6c20e19/ : SUCCESS in 7m 01s\n- cross-swift-py35 http://logs.openstack.org/27/640727/2/check/cross-swift-py35/7af8b77/ : SUCCESS in 2m 51s\n- requirements-integration http://logs.openstack.org/27/640727/2/check/requirements-integration/eba5432/ : FAILURE in 28m 28s\n- neutron-grenade-multinode http://logs.openstack.org/27/640727/2/check/neutron-grenade-multinode/4dd7837/ : SUCCESS in 59m 02s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/27/640727/2/check/legacy-tempest-dsvm-neutron-full/043f055/ : SUCCESS in 1h 26m 23s\n- neutron-grenade http://logs.openstack.org/27/640727/2/check/neutron-grenade/44b0805/ : SUCCESS in 56m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9b83336673fdbc5a9c17146388bf355e70978759","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-14 04:38:38.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1\n\nwe need to answer some outstanding questions as a project before we can move forward here.  At the moment it\u0027s not a technical issue with the PR but about setting a precident  https://etherpad.openstack.org/p/non-openstack-stable-security-constraints-updates","accounts_in_message":[],"_revision_number":2},{"id":"f2a1a2d815b4e63e77e1fd9c0d7b0739acaa7675","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 20:35:46.000000000","message":"Abandoned\n\npython is dumb","accounts_in_message":[],"_revision_number":2},{"id":"21fdf9ec48d6ae7922ed267b305001204626aefd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-22 20:58:31.000000000","message":"Restored\n\nabandon\u0027d the wrong patch","accounts_in_message":[],"_revision_number":2},{"id":"be7e82d1943acae27751357a5dddaf1fd4590fba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 22:39:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/27/640727/2/check/openstack-tox-pypy/dd8bb3f/ : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/27/640727/2/check/openstack-tox-pep8/f6d4b40/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/27/640727/2/check/openstack-tox-py27/6d45996/ : SUCCESS in 3m 27s\n- openstack-tox-py35 http://logs.openstack.org/27/640727/2/check/openstack-tox-py35/6c7978a/ : SUCCESS in 3m 45s\n- openstack-tox-validate http://logs.openstack.org/27/640727/2/check/openstack-tox-validate/59d8d33/ : SUCCESS in 3m 23s\n- requirements-tox-babel http://logs.openstack.org/27/640727/2/check/requirements-tox-babel/03282eb/ : SUCCESS in 4m 54s\n- requirements-tox-py27-check-uc http://logs.openstack.org/27/640727/2/check/requirements-tox-py27-check-uc/011b386/ : SUCCESS in 6m 44s\n- cross-cinder-py27 http://logs.openstack.org/27/640727/2/check/cross-cinder-py27/1ff8c45/ : SUCCESS in 7m 57s\n- cross-cinder-py35 http://logs.openstack.org/27/640727/2/check/cross-cinder-py35/e333ef0/ : SUCCESS in 8m 08s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/27/640727/2/check/cross-glance-py27/5b435b3/ : SUCCESS in 8m 33s\n- cross-glance-py35 http://logs.openstack.org/27/640727/2/check/cross-glance-py35/e38e213/ : SUCCESS in 9m 14s\n- cross-horizon-py27 http://logs.openstack.org/27/640727/2/check/cross-horizon-py27/a6e22c1/ : SUCCESS in 13m 19s\n- cross-horizon-py35 http://logs.openstack.org/27/640727/2/check/cross-horizon-py35/d2fbe65/ : SUCCESS in 9m 55s\n- cross-keystone-py27 http://logs.openstack.org/27/640727/2/check/cross-keystone-py27/6ea15a0/ : SUCCESS in 14m 35s\n- cross-keystone-py35 http://logs.openstack.org/27/640727/2/check/cross-keystone-py35/99014e4/ : SUCCESS in 12m 38s\n- cross-neutron-py27 http://logs.openstack.org/27/640727/2/check/cross-neutron-py27/2d747c7/ : SUCCESS in 22m 33s\n- cross-neutron-py35 http://logs.openstack.org/27/640727/2/check/cross-neutron-py35/e67769c/ : SUCCESS in 23m 58s\n- cross-nova-py27 http://logs.openstack.org/27/640727/2/check/cross-nova-py27/0030bd6/ : SUCCESS in 14m 39s\n- cross-nova-py35 http://logs.openstack.org/27/640727/2/check/cross-nova-py35/614aefb/ : SUCCESS in 14m 21s\n- cross-nova-functional http://logs.openstack.org/27/640727/2/check/cross-nova-functional/749508b/ : SUCCESS in 18m 53s\n- cross-swift-py27 http://logs.openstack.org/27/640727/2/check/cross-swift-py27/8cbbeb1/ : SUCCESS in 7m 53s\n- cross-swift-py35 http://logs.openstack.org/27/640727/2/check/cross-swift-py35/1cabf73/ : SUCCESS in 3m 28s\n- requirements-integration http://logs.openstack.org/27/640727/2/check/requirements-integration/4c38081/ : SUCCESS in 51m 17s\n- neutron-grenade-multinode http://logs.openstack.org/27/640727/2/check/neutron-grenade-multinode/a09a8ef/ : SUCCESS in 59m 06s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/27/640727/2/check/legacy-tempest-dsvm-neutron-full/cb606b4/ : SUCCESS in 1h 34m 45s\n- neutron-grenade http://logs.openstack.org/27/640727/2/check/neutron-grenade/47a77b9/ : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":2},{"id":"e9e2d62dbeadd7e883590dd22b8c7c063a101e5f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-22 05:15:51.000000000","message":"Abandoned\n\nabandoning as it\u0027s best to leave dependency security updates to the downstream packagers","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3305f1a4027058ee68d89c61aca34313e5e03c25","revisions":{"307e003947d53b6fba51bf402ba095ea05e60c04":{"kind":"REWORK","_number":1,"created":"2019-03-04 10:51:30.000000000","uploader":{"_account_id":24131,"name":"Abhishek Sharma M","email":"abmsharm@in.ibm.com","username":"abhi.sharma"},"ref":"refs/changes/27/640727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/640727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/640727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/640727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/640727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/640727/1"}}},"commit":{"parents":[{"commit":"b9c1a67dbb1106c19ae99ba331ba3b1fb0b19e80","subject":"remove publish-loci-requirements post-job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9c1a67dbb1106c19ae99ba331ba3b1fb0b19e80"}]}],"author":{"name":"Abhishek Sharma","email":"abmsharm@in.ibm.com","date":"2019-03-04 10:49:37.000000000","tz":-300},"committer":{"name":"Abhishek Sharma","email":"abmsharm@in.ibm.com","date":"2019-03-04 10:49:37.000000000","tz":-300},"subject":"Update requests to 2.20.1 for CVE-2018-18074","message":"Update requests to 2.20.1 for CVE-2018-18074\n\nGiven that the OpenStack deployment projects, and many operators\nmake use of upper-constraints for building OpenStack from source,\nit would be prudent to update the requests library to the release\nwhich takes care of this vulnerability.\n\nChange-Id: I0dc7182eadb5a07c571c985d8281a8e85b13fd5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/307e003947d53b6fba51bf402ba095ea05e60c04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/307e003947d53b6fba51bf402ba095ea05e60c04"}]},"branch":"refs/heads/stable/pike"},"3305f1a4027058ee68d89c61aca34313e5e03c25":{"kind":"REWORK","_number":2,"created":"2019-05-10 21:15:39.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/27/640727/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/640727/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/640727/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/640727/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/640727/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/640727/2"}}},"commit":{"parents":[{"commit":"3f49011b4145023021627b2738c328c09e4ec1fc","subject":"Merge \"update constraint for ironic-lib to new release 2.10.2\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f49011b4145023021627b2738c328c09e4ec1fc"}]}],"author":{"name":"Abhishek Sharma","email":"abmsharm@in.ibm.com","date":"2019-03-04 10:49:37.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-05-10 21:15:38.000000000","tz":-240},"subject":"Update requests to 2.20.1 for CVE-2018-18074","message":"Update requests to 2.20.1 for CVE-2018-18074\n\nGiven that the OpenStack deployment projects, and many operators\nmake use of upper-constraints for building OpenStack from source,\nit would be prudent to update the requests library to the release\nwhich takes care of this vulnerability.\n\nChange-Id: I0dc7182eadb5a07c571c985d8281a8e85b13fd5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3305f1a4027058ee68d89c61aca34313e5e03c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3305f1a4027058ee68d89c61aca34313e5e03c25"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
