)]}'
{"id":"openstack%2Frequirements~650935","triplet_id":"openstack%2Frequirements~master~I3690336b99e6fad53cd6df5e9ef28126c7608c34","project":"openstack/requirements","branch":"master","topic":"I3690336b99e6fad53cd6df5e9ef28126c7608c34","hashtags":[],"change_id":"I3690336b99e6fad53cd6df5e9ef28126c7608c34","subject":"Add rst2txt","status":"MERGED","created":"2019-04-08 15:41:39.000000000","updated":"2020-07-15 16:14:28.000000000","submitted":"2020-07-15 16:12:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"650935-1594829559620-f352bdee","meta_rev_id":"ad4ff7955daf1a1341315832b139e19b96604723","_number":650935,"virtual_id_number":650935,"owner":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-15 16:12:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-07-14 23:46:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2020-07-15 11:58:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2020-07-15 11:58:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-08 15:48:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-04-08 19:23:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-09 07:32:16.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-10 08:20:34.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-07-14 23:46:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-15 11:58:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-15 16:12:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1389e2f15c3457e738edba3547de47d39f26a0d","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-04-08 15:41:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7587fcff6dec0ffb47886b46f98cca689f185683","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-08 15:48:34.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fc03dbb679c857c079102dfe519b157893872e2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 18:19:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/35/650935/1/check/openstack-tox-docs/464542d/html/ : SUCCESS in 4m 07s\n- openstack-tox-pypy http://logs.openstack.org/35/650935/1/check/openstack-tox-pypy/19300b9/ : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/35/650935/1/check/openstack-tox-pep8/3d5e21b/ : SUCCESS in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/35/650935/1/check/openstack-tox-py27/a947894/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/35/650935/1/check/openstack-tox-py35/d823703/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/35/650935/1/check/openstack-tox-py36/f4c2e42/ : SUCCESS in 5m 23s\n- openstack-tox-validate http://logs.openstack.org/35/650935/1/check/openstack-tox-validate/554d957/ : SUCCESS in 4m 56s\n- requirements-tox-babel http://logs.openstack.org/35/650935/1/check/requirements-tox-babel/538d1b3/ : SUCCESS in 5m 24s\n- requirements-tox-py27-check-uc http://logs.openstack.org/35/650935/1/check/requirements-tox-py27-check-uc/c50826b/ : SUCCESS in 9m 11s\n- requirements-tox-py35-check-uc http://logs.openstack.org/35/650935/1/check/requirements-tox-py35-check-uc/ebbc31b/ : SUCCESS in 10m 31s\n- requirements-tox-py36-check-uc http://logs.openstack.org/35/650935/1/check/requirements-tox-py36-check-uc/88fb651/ : SUCCESS in 8m 52s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/35/650935/1/check/cross-cinder-py27/76cb6a8/ : SUCCESS in 11m 04s\n- cross-cinder-py35 http://logs.openstack.org/35/650935/1/check/cross-cinder-py35/0a659c8/ : SUCCESS in 11m 45s\n- cross-glance-py27 http://logs.openstack.org/35/650935/1/check/cross-glance-py27/a8041a4/ : SUCCESS in 8m 05s\n- cross-glance-py35 http://logs.openstack.org/35/650935/1/check/cross-glance-py35/7bc80da/ : SUCCESS in 7m 01s\n- cross-horizon-py27 http://logs.openstack.org/35/650935/1/check/cross-horizon-py27/e36607d/ : SUCCESS in 11m 24s\n- cross-horizon-py35 http://logs.openstack.org/35/650935/1/check/cross-horizon-py35/e51d568/ : SUCCESS in 10m 31s\n- cross-horizon-npm http://logs.openstack.org/35/650935/1/check/cross-horizon-npm/61b4ad0/ : SUCCESS in 8m 06s\n- cross-keystone-py27 http://logs.openstack.org/35/650935/1/check/cross-keystone-py27/46c3b6d/ : SUCCESS in 19m 18s\n- cross-keystone-py35 http://logs.openstack.org/35/650935/1/check/cross-keystone-py35/e233b5a/ : SUCCESS in 25m 42s\n- cross-neutron-py27 http://logs.openstack.org/35/650935/1/check/cross-neutron-py27/33d9038/ : SUCCESS in 23m 08s\n- cross-neutron-py35 http://logs.openstack.org/35/650935/1/check/cross-neutron-py35/6b36dc7/ : SUCCESS in 31m 00s\n- cross-nova-py27 http://logs.openstack.org/35/650935/1/check/cross-nova-py27/8846f7f/ : SUCCESS in 13m 17s\n- cross-nova-py35 http://logs.openstack.org/35/650935/1/check/cross-nova-py35/4ee3a11/ : SUCCESS in 15m 40s\n- cross-nova-functional http://logs.openstack.org/35/650935/1/check/cross-nova-functional/acb2591/ : SUCCESS in 24m 47s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/35/650935/1/check/cross-osc-build-sphinx-docs/f4e8f87/html/ : SUCCESS in 7m 25s\n- cross-swift-py27 http://logs.openstack.org/35/650935/1/check/cross-swift-py27/d5f7109/ : SUCCESS in 15m 27s\n- cross-swift-py35 http://logs.openstack.org/35/650935/1/check/cross-swift-py35/52d4ce3/ : SUCCESS in 9m 35s\n- requirements-integration http://logs.openstack.org/35/650935/1/check/requirements-integration/f2b8c80/ : SUCCESS in 1h 01m 42s\n- neutron-grenade-multinode http://logs.openstack.org/35/650935/1/check/neutron-grenade-multinode/aa752cb/ : SUCCESS in 1h 14m 25s (non-voting)\n- tempest-full http://logs.openstack.org/35/650935/1/check/tempest-full/1a8f427/ : FAILURE in 1h 38m 18s\n- neutron-grenade http://logs.openstack.org/35/650935/1/check/neutron-grenade/880c500/ : FAILURE in 57m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11bfd428ab4282a9d631856decaf0e1cc427733c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-08 19:23:51.000000000","message":"Patch Set 1: Code-Review-1\n\nseems to be missing pygments from being added to upper-constraints.txt\n\nDid you run https://docs.openstack.org/project-team-guide/dependency-management.html#generate-constraints after adding rst2txt to global-requirements?","accounts_in_message":[],"_revision_number":1},{"id":"72cb25906872fd6264d6f2e9f5fa111e10369bef","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-09 07:32:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4c1c02e26df0ca7388022a1dd60703d2c4d77a4","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-04-09 09:38:47.000000000","message":"Patch Set 1:\n\nMatthew, I tried running it, but it was breaking all the time. I\u0027ll spend more time on it.","accounts_in_message":[],"_revision_number":1},{"id":"63fa07888c969a02776a9ace41a6a98f78caa1d7","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-04-10 08:20:34.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0b7bb1184c1eef5fa6c944795591d4bc76d967c0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-02 17:51:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ca1d7c6afc5ee1488377a8735972cc9cc082f5de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 17:54:07.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"881d58849eb5519034467b8b70e3108bd76c51a3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-14 20:43:18.000000000","message":"Patch Set 1:\n\nThis has been awhile. Still needed?","accounts_in_message":[],"_revision_number":1},{"id":"839f2ccc5e39779503ffd117c2bf7620d54b3ea7","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-08-15 12:52:52.000000000","message":"Patch Set 1:\n\nHi Sean, thanks for the reminder. I\u0027ll bring this one to the next oslo meeting and see if there is still interest.","accounts_in_message":[],"_revision_number":1},{"id":"c92b121bb39bd54d28f63ef1149f834aa2e87da0","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2019-08-19 15:34:11.000000000","message":"Abandoned\n\nAbandoning this one for now due to low priority in the oslo patch.","accounts_in_message":[],"_revision_number":1},{"id":"e94243fa28ddc8dd8fe80b45807279606f72d68c","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-07-14 22:00:05.000000000","message":"Restored\n\nraise dead","accounts_in_message":[],"_revision_number":1},{"id":"c997eef4eefd61d359aa0a9b1a03b2b8d0688386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 22:00:18.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"5c5d99885ae25a21659f4a6e457d9f1bb513ff16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 22:00:19.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"9b536378a18dd5b9c0d65792b67f8c82b69df627","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-07-14 22:00:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"20c1f91d944fe075a5a40e730f311d03723b70dd","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-07-14 22:07:12.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"23d678a8ff82f556362f441cdf8a1652b4db81f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 22:51:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1365e23896834ff3abb48bc2b6cf78e0 : SUCCESS in 39m 58s","accounts_in_message":[],"_revision_number":3},{"id":"a593ae2a7dd8a4d9ca214810d5af68cb48e1c483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 23:11:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b171af20c314cf29b38f9f6c75eb6c6 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87994e9a13304367a3723503f858b7b3 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d731103ba964a818502a615fb2763d5 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/304306a0f7c3487d9cf48934e4021450 : SUCCESS in 6m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4ccb0acbae747b8a50d4a1814bab611 : SUCCESS in 5m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e723bd790d914e3283d54302d868920f : SUCCESS in 5m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2a13b981633142bba0f594b91c821fb3 : SUCCESS in 9m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/285fdeb6f39a4e969a6f9e15c59ee923 : SUCCESS in 12m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d7e500fd9f074d2eb4f90368b0d76d90 : SUCCESS in 8m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f0760ed831e94c47bf326d6655a5d04d : SUCCESS in 6m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/26e708f2d69346d4b0b09d40d49c135e : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/188c16c2f4bc4c2799f4ffe66c42a845 : SUCCESS in 8m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/516e73425cf4426d994447fa87bbd098 : SUCCESS in 8m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fc74550c3083453caae6848e0b81c15c : SUCCESS in 29m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/45b22df4eb4c4a44bcd63b0436547382 : SUCCESS in 22m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/876956b4b7814cfb85dbbc33d09338fc : SUCCESS in 18m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d4e0374ad8544a0db5ea16d96bd5dc42 : SUCCESS in 22m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5526f0925541445da3fae5242183d66d : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec8b2562ff5a48cb84e70e5b20d00b1c : SUCCESS in 7m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ae97fd8a3f8442d38ddf60ccabfd34bd : SUCCESS in 18m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab7448a79d3f47e99eb094ca297ed2cf : SUCCESS in 55m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ae654005d9c437d97e2456ca701cf63 : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":3},{"id":"6d37e20a6d577f64ed8353fd33e5462eaa539e40","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-14 23:46:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4749452455c1380cb71d448fd6b8a1247a97f7a2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-15 11:58:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8d33a038a951fbdfd18819f58de6e4ae25f5d09a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 11:59:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8cf5f8801eae3279710fcea2865fc5a3d11a4037","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 16:12:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bae6a1ca51924c3d983a317b476d700b : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a524699b0a0542a4b0376ce21e16b11d : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d131dd403356402eafdf12fbd899fe70 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c84c62bfcaa0472fadd16056eb895a17 : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8beb874689d4145b7186bc5d074e8b1 : SUCCESS in 6m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ab4274e8f824e2eb922e3b5ae3ac946 : SUCCESS in 7m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6eeac3e443cc482ebd4e5e47b6fbda1f : SUCCESS in 11m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/41fd6cc044a04366b607881398c622cd : SUCCESS in 13m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c12e9b5362914315bcc3fabc3749e4a9 : SUCCESS in 9m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d6cf4f39ff0243f39b43e853533494de : SUCCESS in 7m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1b5a0a2da526454f9fb6bea26fa2ac64 : SUCCESS in 12m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a4824438575e44e38eec82c11aa34903 : SUCCESS in 9m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/28bd879550af4fdbb10e22f21dd96927 : SUCCESS in 11m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6aa27487fef04ba7bbe68f37e2b4e47b : SUCCESS in 15m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fc1ef2c657b24956a6b12567f9b30e0c : SUCCESS in 26m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/24b5e3daf3c3407e9b896c4db7732dfc : SUCCESS in 16m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/00e5466d3c9346c6afd43549f6738721 : SUCCESS in 18m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/73a4437f918c4d4bb7fb44152cca5bba : SUCCESS in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/803cde1a738447f59b830fe8912a9e8f : SUCCESS in 8m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d01cda6a45014893912d268c7fff627b : SUCCESS in 14m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e54d231c15a24333befbe43481c75b53 : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fdfb57ab7c1e4a58bcfdfcfab5d9fe92 : SUCCESS in 1h 02m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e751e7ba10344dc3bb4a27257f7e5048 : SUCCESS in 1h 23m 15s","accounts_in_message":[],"_revision_number":3},{"id":"e6697c9acda7f5f0a89f5c81000e96cdacc255b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 16:12:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d076cdd6a0e33fec3257cff7b3ab55a2406cabd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 16:14:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09dc99007a6e449f92d207a163b446d9 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8c314dd51feb5e18f8c51662fac04be13d4820aa","revisions":{"10e0e3e231c777bad0ce997636c401816e2e3c45":{"kind":"REWORK","_number":1,"created":"2019-04-08 15:41:39.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/35/650935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/650935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/650935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/650935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/650935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/650935/1"}}},"commit":{"parents":[{"commit":"8d7059522f7c2f1f92ff770181c1f53554990c9f","subject":"Merge \"Remove cap for jsonschema in global-requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d7059522f7c2f1f92ff770181c1f53554990c9f"}]}],"author":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2019-04-08 15:35:11.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2019-04-08 15:41:37.000000000","tz":120},"subject":"Add rst2txt","message":"Add rst2txt\n\nThis library will be used in oslo.config to convert rst to plaintext\noutput in options\u0027 helper text.\n\n[Q] Is the library actively maintained?\n[A]\n\n[Q] Is the library good code?\n[A]\n\n[Q] Is the library python 3 compatible?\n[A]\n\n[Q] Is the library license compatible?\n[A] rst2txt  # BSD\n\n[Q] Is the library already packaged in the distros we target (Ubuntu\n    latest / Fedora latest)?\n[A] The library distributed through PyPI.\n\n[Q] Is the function of this library already covered by other libraries\n    in global-requirements.txt?\n[A]\n\n[Q] Is the library required for OpenStack project or related dev or\n    infrastructure setup? (Answer to this should be Yes, of course) Which?\n[A] OpenStack project: Olso/Config.\n\n[Q] If the library release is managed by the Openstack release process\n    does it use the cycle-with-intermediary release type?\n[A] N/A\n\nChange-Id: I3690336b99e6fad53cd6df5e9ef28126c7608c34\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10e0e3e231c777bad0ce997636c401816e2e3c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10e0e3e231c777bad0ce997636c401816e2e3c45"}]},"branch":"refs/heads/master"},"13e361ede28bec58b08c52a9d4d25a212d0495a3":{"kind":"REWORK","_number":2,"created":"2020-07-14 22:00:22.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/35/650935/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/650935/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/650935/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/650935/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/650935/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/650935/2"}}},"commit":{"parents":[{"commit":"0d27f316842f38ff558b9bd09a4b48ddb7472f74","subject":"update constraint for python-openstackclient to new release 5.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d27f316842f38ff558b9bd09a4b48ddb7472f74"}]}],"author":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2019-04-08 15:35:11.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-07-14 21:56:28.000000000","tz":120},"subject":"Add rst2txt","message":"Add rst2txt\n\nThis library will be used in oslo.config to convert rst to plaintext\noutput in options\u0027 helper text.\n\n[Q] Is the library actively maintained?\n[A]\n\n[Q] Is the library good code?\n[A]\n\n[Q] Is the library python 3 compatible?\n[A]\n\n[Q] Is the library license compatible?\n[A] rst2txt  # BSD\n\n[Q] Is the library already packaged in the distros we target (Ubuntu\n    latest / Fedora latest)?\n[A] The library distributed through PyPI.\n\n[Q] Is the function of this library already covered by other libraries\n    in global-requirements.txt?\n[A]\n\n[Q] Is the library required for OpenStack project or related dev or\n    infrastructure setup? (Answer to this should be Yes, of course) Which?\n[A] OpenStack project: Olso/Config.\n\n[Q] If the library release is managed by the Openstack release process\n    does it use the cycle-with-intermediary release type?\n[A] N/A\n\nChange-Id: I3690336b99e6fad53cd6df5e9ef28126c7608c34\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13e361ede28bec58b08c52a9d4d25a212d0495a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13e361ede28bec58b08c52a9d4d25a212d0495a3"}]},"branch":"refs/heads/master"},"8c314dd51feb5e18f8c51662fac04be13d4820aa":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-07-14 22:07:12.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/35/650935/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/650935/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/650935/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/650935/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/650935/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/650935/3"}}},"commit":{"parents":[{"commit":"0d27f316842f38ff558b9bd09a4b48ddb7472f74","subject":"update constraint for python-openstackclient to new release 5.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d27f316842f38ff558b9bd09a4b48ddb7472f74"}]}],"author":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2019-04-08 15:35:11.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-07-14 22:01:35.000000000","tz":120},"subject":"Add rst2txt","message":"Add rst2txt\n\nThis library will be used in oslo.config to convert rst to plaintext\noutput in options\u0027 helper text.\n\n[Q] Is the library actively maintained?\n[A] The library is maintained by https://pypi.org/user/stephenfin/ an\nactive Oslo core contributor.\n\n[Q] Is the library good code?\n[A] The library is a small wrapper around docutils, code has both tests\nand documentation.\n\n[Q] Is the library python 3 compatible?\n[A] Yes.\n\n[Q] Is the library license compatible?\n[A] rst2txt  # BSD\n\n[Q] Is the library already packaged in the distros we target (Ubuntu\n    latest / Fedora latest)?\n[A] The library distributed through PyPI.\n\n[Q] Is the function of this library already covered by other libraries\n    in global-requirements.txt?\n[A] No.\n\n[Q] Is the library required for OpenStack project or related dev or\n    infrastructure setup? (Answer to this should be Yes, of course) Which?\n[A] OpenStack project: Olso/Config.\n\n[Q] If the library release is managed by the Openstack release process\n    does it use the cycle-with-intermediary release type?\n[A] N/A\n\nChange-Id: I3690336b99e6fad53cd6df5e9ef28126c7608c34\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c314dd51feb5e18f8c51662fac04be13d4820aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c314dd51feb5e18f8c51662fac04be13d4820aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
