)]}'
{"id":"openstack%2Frequirements~651591","triplet_id":"openstack%2Frequirements~master~I988179f302325ceb5d7411362121c2505864a9ef","project":"openstack/requirements","branch":"master","topic":"bug/1824178","hashtags":[],"change_id":"I988179f302325ceb5d7411362121c2505864a9ef","subject":"Bump SQLAlchemy to 1.3.4","status":"ABANDONED","created":"2019-04-10 15:48:05.000000000","updated":"2019-06-16 20:52:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a5dab66962b7d77de5656aa1efe99396a01b911d","_number":651591,"virtual_id_number":651591,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-05-28 17:30:37.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":-1,"date":"2019-05-28 19:41:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2019-05-28 19:18:33.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-15 15:42:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-28 19:18:33.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-05-28 19:41:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"807c41bdb1432c231f0882527c85d2f2d4dfcec3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-04-10 15:48:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f1d07fdcef953b92ec9450244e075c6607351b9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-04-10 15:48:48.000000000","message":"Patch Set 1: Workflow-1\n\nPatch [1] not merged and neutron-lib version not released yet.\n\n[1] https://review.openstack.org/#/c/651584/","accounts_in_message":[],"_revision_number":1},{"id":"740f73ceae6f6503082fc141c3b966c459e8b431","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 18:12:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/651591/1/check/openstack-tox-docs/9c1b532/html/ : SUCCESS in 3m 42s\n- openstack-tox-pypy http://logs.openstack.org/91/651591/1/check/openstack-tox-pypy/aba0697/ : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/651591/1/check/openstack-tox-pep8/98a4636/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/91/651591/1/check/openstack-tox-py27/e07636e/ : SUCCESS in 4m 17s\n- openstack-tox-py35 http://logs.openstack.org/91/651591/1/check/openstack-tox-py35/df77545/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/91/651591/1/check/openstack-tox-py36/2f2475f/ : SUCCESS in 3m 33s\n- openstack-tox-validate http://logs.openstack.org/91/651591/1/check/openstack-tox-validate/e488ee0/ : SUCCESS in 4m 36s\n- requirements-tox-babel http://logs.openstack.org/91/651591/1/check/requirements-tox-babel/710e778/ : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/651591/1/check/requirements-tox-py27-check-uc/bf634a0/ : SUCCESS in 9m 35s\n- requirements-tox-py35-check-uc http://logs.openstack.org/91/651591/1/check/requirements-tox-py35-check-uc/9182597/ : SUCCESS in 9m 37s\n- requirements-tox-py36-check-uc http://logs.openstack.org/91/651591/1/check/requirements-tox-py36-check-uc/2deb74f/ : SUCCESS in 9m 41s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/91/651591/1/check/cross-cinder-py27/120dee5/ : SUCCESS in 11m 26s\n- cross-cinder-py35 http://logs.openstack.org/91/651591/1/check/cross-cinder-py35/d2354db/ : FAILURE in 10m 57s\n- cross-glance-py27 http://logs.openstack.org/91/651591/1/check/cross-glance-py27/11cff16/ : SUCCESS in 6m 50s\n- cross-glance-py35 http://logs.openstack.org/91/651591/1/check/cross-glance-py35/cebbff8/ : SUCCESS in 4m 51s\n- cross-horizon-py27 http://logs.openstack.org/91/651591/1/check/cross-horizon-py27/f73420b/ : SUCCESS in 12m 00s\n- cross-horizon-py35 http://logs.openstack.org/91/651591/1/check/cross-horizon-py35/99e394b/ : SUCCESS in 11m 21s\n- cross-horizon-npm http://logs.openstack.org/91/651591/1/check/cross-horizon-npm/eb136e9/ : SUCCESS in 6m 12s\n- cross-keystone-py27 http://logs.openstack.org/91/651591/1/check/cross-keystone-py27/e03da65/ : SUCCESS in 21m 42s\n- cross-keystone-py35 http://logs.openstack.org/91/651591/1/check/cross-keystone-py35/73b697e/ : SUCCESS in 25m 59s\n- cross-neutron-py27 http://logs.openstack.org/91/651591/1/check/cross-neutron-py27/3ed79cf/ : FAILURE in 23m 39s\n- cross-neutron-py35 http://logs.openstack.org/91/651591/1/check/cross-neutron-py35/0ad68c7/ : SUCCESS in 23m 26s\n- cross-nova-py27 http://logs.openstack.org/91/651591/1/check/cross-nova-py27/9dbe77e/ : SUCCESS in 14m 11s\n- cross-nova-py35 http://logs.openstack.org/91/651591/1/check/cross-nova-py35/749eb88/ : SUCCESS in 15m 04s\n- cross-nova-functional http://logs.openstack.org/91/651591/1/check/cross-nova-functional/ce00f90/ : SUCCESS in 20m 30s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/91/651591/1/check/cross-osc-build-sphinx-docs/5c13bb6/html/ : SUCCESS in 8m 17s\n- cross-swift-py27 http://logs.openstack.org/91/651591/1/check/cross-swift-py27/cf4ad29/ : SUCCESS in 7m 48s\n- cross-swift-py35 http://logs.openstack.org/91/651591/1/check/cross-swift-py35/29fbd05/ : SUCCESS in 11m 20s\n- requirements-integration http://logs.openstack.org/91/651591/1/check/requirements-integration/17777c8/ : SUCCESS in 1h 03m 31s\n- neutron-grenade-multinode http://logs.openstack.org/91/651591/1/check/neutron-grenade-multinode/46af5c8/ : SUCCESS in 1h 12m 08s (non-voting)\n- tempest-full http://logs.openstack.org/91/651591/1/check/tempest-full/c62e2ef/ : FAILURE in 1h 37m 48s\n- neutron-grenade http://logs.openstack.org/91/651591/1/check/neutron-grenade/35d4e1a/ : SUCCESS in 53m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"976142f168b339a038904839119916198302f955","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-10 19:17:34.000000000","message":"Patch Set 1:\n\nit\u0027s not just neutron, but thanks, I guess this\u0027ll be here until openstack is ready for it","accounts_in_message":[],"_revision_number":1},{"id":"f0b4b1741f021667f22463a13aba2c275cca997d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-13 21:58:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dfb9932db1d38ba3f6c8ec55636dd216904230b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-13 23:26:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/651591/1/check/openstack-tox-docs/9365c64/html/ : SUCCESS in 2m 37s\n- openstack-tox-pypy http://logs.openstack.org/91/651591/1/check/openstack-tox-pypy/66a8f47/ : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/651591/1/check/openstack-tox-pep8/4abdffc/ : SUCCESS in 2m 18s\n- openstack-tox-py27 http://logs.openstack.org/91/651591/1/check/openstack-tox-py27/d245699/ : SUCCESS in 2m 58s\n- openstack-tox-py35 http://logs.openstack.org/91/651591/1/check/openstack-tox-py35/2f82a3a/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/91/651591/1/check/openstack-tox-py36/b4d9fcb/ : SUCCESS in 2m 43s\n- openstack-tox-validate http://logs.openstack.org/91/651591/1/check/openstack-tox-validate/95f2e5a/ : SUCCESS in 2m 43s\n- requirements-tox-babel http://logs.openstack.org/91/651591/1/check/requirements-tox-babel/dd93bd9/ : SUCCESS in 2m 22s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/651591/1/check/requirements-tox-py27-check-uc/b271bc4/ : SUCCESS in 8m 06s\n- requirements-tox-py35-check-uc http://logs.openstack.org/91/651591/1/check/requirements-tox-py35-check-uc/45340e2/ : SUCCESS in 7m 37s\n- requirements-tox-py36-check-uc http://logs.openstack.org/91/651591/1/check/requirements-tox-py36-check-uc/faf3dcc/ : SUCCESS in 8m 23s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/91/651591/1/check/cross-cinder-py27/057118b/ : SUCCESS in 8m 30s\n- cross-cinder-py36 http://logs.openstack.org/91/651591/1/check/cross-cinder-py36/82d2bfe/ : SUCCESS in 8m 33s\n- cross-glance-py27 http://logs.openstack.org/91/651591/1/check/cross-glance-py27/2663cf4/ : SUCCESS in 4m 09s\n- cross-glance-py36 http://logs.openstack.org/91/651591/1/check/cross-glance-py36/eab9c54/ : SUCCESS in 4m 25s\n- cross-horizon-py27 http://logs.openstack.org/91/651591/1/check/cross-horizon-py27/e6ec64a/ : SUCCESS in 9m 25s\n- cross-horizon-py36 http://logs.openstack.org/91/651591/1/check/cross-horizon-py36/a4daa9d/ : SUCCESS in 9m 00s\n- cross-horizon-npm http://logs.openstack.org/91/651591/1/check/cross-horizon-npm/8c14802/ : SUCCESS in 5m 12s\n- cross-keystone-py27 http://logs.openstack.org/91/651591/1/check/cross-keystone-py27/328ba2d/ : SUCCESS in 22m 07s\n- cross-keystone-py36 http://logs.openstack.org/91/651591/1/check/cross-keystone-py36/69fccaf/ : SUCCESS in 23m 04s\n- cross-neutron-py27 http://logs.openstack.org/91/651591/1/check/cross-neutron-py27/29a62bc/ : SUCCESS in 21m 29s\n- cross-neutron-py36 http://logs.openstack.org/91/651591/1/check/cross-neutron-py36/ae7b5f0/ : SUCCESS in 20m 38s\n- cross-nova-py27 http://logs.openstack.org/91/651591/1/check/cross-nova-py27/52fa877/ : SUCCESS in 12m 04s\n- cross-nova-py36 http://logs.openstack.org/91/651591/1/check/cross-nova-py36/8dd7244/ : SUCCESS in 12m 15s\n- cross-nova-functional http://logs.openstack.org/91/651591/1/check/cross-nova-functional/cae84ec/ : SUCCESS in 22m 19s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/91/651591/1/check/cross-osc-build-sphinx-docs/7f609a9/html/ : SUCCESS in 5m 13s\n- cross-swift-py27 http://logs.openstack.org/91/651591/1/check/cross-swift-py27/042ec14/ : SUCCESS in 7m 10s\n- cross-swift-py36 http://logs.openstack.org/91/651591/1/check/cross-swift-py36/1016822/ : SUCCESS in 6m 50s\n- requirements-integration http://logs.openstack.org/91/651591/1/check/requirements-integration/d763906/ : SUCCESS in 57m 21s\n- neutron-grenade-multinode http://logs.openstack.org/91/651591/1/check/neutron-grenade-multinode/6c5026c/ : SUCCESS in 59m 45s\n- tempest-full http://logs.openstack.org/91/651591/1/check/tempest-full/7932e58/ : FAILURE in 1h 22m 15s\n- neutron-grenade http://logs.openstack.org/91/651591/1/check/neutron-grenade/8af1850/ : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d10cb057aefe330bd8b3a8fd91100dfec6a45fb9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-16 18:12:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"11b5380ea937bf172b6c92c5329360b1fd86caf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 22:53:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/651591/1/check/openstack-tox-docs/5e32433/html/ : SUCCESS in 3m 03s\n- openstack-tox-pypy http://logs.openstack.org/91/651591/1/check/openstack-tox-pypy/6fa8d95/ : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/651591/1/check/openstack-tox-pep8/b9ceaad/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/91/651591/1/check/openstack-tox-py27/487ac6d/ : SUCCESS in 4m 40s\n- openstack-tox-py35 http://logs.openstack.org/91/651591/1/check/openstack-tox-py35/9c6ecbc/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/91/651591/1/check/openstack-tox-py36/0cb90e9/ : SUCCESS in 4m 15s\n- openstack-tox-validate http://logs.openstack.org/91/651591/1/check/openstack-tox-validate/0b40b01/ : SUCCESS in 5m 36s\n- requirements-tox-babel http://logs.openstack.org/91/651591/1/check/requirements-tox-babel/ac7d95f/ : SUCCESS in 4m 48s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/651591/1/check/requirements-tox-py27-check-uc/e4364e2/ : SUCCESS in 9m 13s\n- requirements-tox-py35-check-uc http://logs.openstack.org/91/651591/1/check/requirements-tox-py35-check-uc/54f8d8a/ : SUCCESS in 7m 54s\n- requirements-tox-py36-check-uc http://logs.openstack.org/91/651591/1/check/requirements-tox-py36-check-uc/3f13055/ : SUCCESS in 19m 48s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/91/651591/1/check/cross-cinder-py27/e6d3c1a/ : SUCCESS in 14m 46s\n- cross-cinder-py36 http://logs.openstack.org/91/651591/1/check/cross-cinder-py36/f197d21/ : SUCCESS in 15m 26s\n- cross-glance-py27 http://logs.openstack.org/91/651591/1/check/cross-glance-py27/0cabb5b/ : SUCCESS in 6m 35s\n- cross-glance-py36 http://logs.openstack.org/91/651591/1/check/cross-glance-py36/77958d5/ : SUCCESS in 5m 57s\n- cross-horizon-py27 http://logs.openstack.org/91/651591/1/check/cross-horizon-py27/d3b2a94/ : SUCCESS in 10m 54s\n- cross-horizon-py36 http://logs.openstack.org/91/651591/1/check/cross-horizon-py36/f96704a/ : SUCCESS in 8m 29s\n- cross-horizon-npm http://logs.openstack.org/91/651591/1/check/cross-horizon-npm/d32c14b/ : SUCCESS in 7m 14s\n- cross-keystone-py27 http://logs.openstack.org/91/651591/1/check/cross-keystone-py27/1d2c314/ : SUCCESS in 25m 40s\n- cross-keystone-py36 http://logs.openstack.org/91/651591/1/check/cross-keystone-py36/f59525a/ : SUCCESS in 32m 55s\n- cross-neutron-py27 http://logs.openstack.org/91/651591/1/check/cross-neutron-py27/19f7777/ : SUCCESS in 31m 31s\n- cross-neutron-py36 http://logs.openstack.org/91/651591/1/check/cross-neutron-py36/a1c0ec7/ : SUCCESS in 22m 42s\n- cross-nova-py27 http://logs.openstack.org/91/651591/1/check/cross-nova-py27/faf18a3/ : SUCCESS in 11m 18s\n- cross-nova-py36 http://logs.openstack.org/91/651591/1/check/cross-nova-py36/052a6f5/ : SUCCESS in 18m 05s\n- cross-nova-functional http://logs.openstack.org/91/651591/1/check/cross-nova-functional/65a40b3/ : SUCCESS in 23m 43s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/91/651591/1/check/cross-osc-build-sphinx-docs/de9f8a8/html/ : SUCCESS in 5m 40s\n- cross-swift-py27 http://logs.openstack.org/91/651591/1/check/cross-swift-py27/66fa7be/ : SUCCESS in 8m 22s\n- cross-swift-py36 http://logs.openstack.org/91/651591/1/check/cross-swift-py36/854907e/ : SUCCESS in 7m 28s\n- requirements-integration http://logs.openstack.org/91/651591/1/check/requirements-integration/5c7f01a/ : SUCCESS in 1h 03m 15s\n- neutron-grenade-multinode http://logs.openstack.org/91/651591/1/check/neutron-grenade-multinode/7e3d61b/ : SUCCESS in 1h 09m 17s\n- tempest-full http://logs.openstack.org/91/651591/1/check/tempest-full/9a1bcab/ : FAILURE in 1h 19m 27s\n- neutron-grenade http://logs.openstack.org/91/651591/1/check/neutron-grenade/1a36fd9/ : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d6f63a92290b7efc56c33f0ab49067b7b4f8e19f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-02 17:44:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"78f88ad2c459125d529b31a7f5a4998038b44630","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 20:07:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/651591/2/check/openstack-tox-docs/8977597/html/ : SUCCESS in 3m 14s\n- openstack-tox-pypy http://logs.openstack.org/91/651591/2/check/openstack-tox-pypy/c4031d8/ : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/651591/2/check/openstack-tox-pep8/4fdf616/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/91/651591/2/check/openstack-tox-py27/fb840fe/ : SUCCESS in 3m 17s\n- openstack-tox-py36 http://logs.openstack.org/91/651591/2/check/openstack-tox-py36/210a5f7/ : SUCCESS in 5m 01s\n- openstack-tox-validate http://logs.openstack.org/91/651591/2/check/openstack-tox-validate/a648131/ : SUCCESS in 2m 41s\n- requirements-tox-babel http://logs.openstack.org/91/651591/2/check/requirements-tox-babel/73f07bc/ : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/651591/2/check/requirements-tox-py27-check-uc/9fa462c/ : SUCCESS in 10m 40s\n- requirements-tox-py36-check-uc http://logs.openstack.org/91/651591/2/check/requirements-tox-py36-check-uc/01d156b/ : SUCCESS in 7m 55s\n- cross-cinder-py27 http://logs.openstack.org/91/651591/2/check/cross-cinder-py27/4a6dfe6/ : SUCCESS in 8m 48s\n- cross-cinder-py36 http://logs.openstack.org/91/651591/2/check/cross-cinder-py36/ac8a82e/ : SUCCESS in 8m 52s\n- cross-glance-py27 http://logs.openstack.org/91/651591/2/check/cross-glance-py27/6f89a6d/ : SUCCESS in 4m 51s\n- cross-glance-py36 http://logs.openstack.org/91/651591/2/check/cross-glance-py36/5dcbd62/ : SUCCESS in 4m 12s\n- cross-horizon-py27 http://logs.openstack.org/91/651591/2/check/cross-horizon-py27/5dff0b3/ : SUCCESS in 11m 34s\n- cross-horizon-py36 http://logs.openstack.org/91/651591/2/check/cross-horizon-py36/afc63d5/ : SUCCESS in 9m 24s\n- cross-horizon-npm http://logs.openstack.org/91/651591/2/check/cross-horizon-npm/982f74d/ : SUCCESS in 7m 57s\n- cross-keystone-py27 http://logs.openstack.org/91/651591/2/check/cross-keystone-py27/4db3fd3/ : SUCCESS in 23m 10s\n- cross-keystone-py36 http://logs.openstack.org/91/651591/2/check/cross-keystone-py36/5cf19b9/ : SUCCESS in 23m 18s\n- cross-neutron-py27 http://logs.openstack.org/91/651591/2/check/cross-neutron-py27/7947777/ : SUCCESS in 24m 39s\n- cross-neutron-py36 http://logs.openstack.org/91/651591/2/check/cross-neutron-py36/7fd3edb/ : SUCCESS in 23m 30s\n- cross-nova-py27 http://logs.openstack.org/91/651591/2/check/cross-nova-py27/cd90775/ : SUCCESS in 11m 52s\n- cross-nova-py36 http://logs.openstack.org/91/651591/2/check/cross-nova-py36/d34b6f5/ : SUCCESS in 13m 21s\n- cross-nova-functional http://logs.openstack.org/91/651591/2/check/cross-nova-functional/4a8f072/ : SUCCESS in 19m 19s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/91/651591/2/check/cross-osc-build-sphinx-docs/3fd1223/html/ : SUCCESS in 5m 31s\n- cross-swift-py27 http://logs.openstack.org/91/651591/2/check/cross-swift-py27/6257832/ : SUCCESS in 7m 32s\n- cross-swift-py36 http://logs.openstack.org/91/651591/2/check/cross-swift-py36/c8eae0b/ : SUCCESS in 7m 54s\n- requirements-integration http://logs.openstack.org/91/651591/2/check/requirements-integration/0985f78/ : SUCCESS in 1h 10m 33s\n- neutron-grenade-multinode http://logs.openstack.org/91/651591/2/check/neutron-grenade-multinode/067bdc4/ : FAILURE in 28m 59s\n- tempest-full http://logs.openstack.org/91/651591/2/check/tempest-full/5a2926c/ : FAILURE in 1h 41m 40s\n- neutron-grenade http://logs.openstack.org/91/651591/2/check/neutron-grenade/82c07ec/ : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":2},{"id":"41ee38a56d880c88cc1755fc629b9610953e2486","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-15 15:42:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dc499b3bfff770b99074c3413e308de1c0ff7d1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 18:19:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/651591/2/check/openstack-tox-docs/63117dc/html/ : SUCCESS in 3m 17s\n- openstack-tox-pypy http://logs.openstack.org/91/651591/2/check/openstack-tox-pypy/e5f10db/ : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/651591/2/check/openstack-tox-pep8/d3d43c2/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/91/651591/2/check/openstack-tox-py27/e70227d/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/91/651591/2/check/openstack-tox-py36/31d1781/ : SUCCESS in 3m 01s\n- openstack-tox-validate http://logs.openstack.org/91/651591/2/check/openstack-tox-validate/ae59161/ : SUCCESS in 2m 57s\n- requirements-tox-babel http://logs.openstack.org/91/651591/2/check/requirements-tox-babel/a8ec80f/ : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/651591/2/check/requirements-tox-py27-check-uc/88c89b1/ : SUCCESS in 9m 12s\n- requirements-tox-py36-check-uc http://logs.openstack.org/91/651591/2/check/requirements-tox-py36-check-uc/fa90eab/ : SUCCESS in 7m 13s\n- cross-cinder-py27 http://logs.openstack.org/91/651591/2/check/cross-cinder-py27/57c76df/ : SUCCESS in 7m 48s\n- cross-cinder-py36 http://logs.openstack.org/91/651591/2/check/cross-cinder-py36/b39542a/ : SUCCESS in 9m 05s\n- cross-glance-py27 http://logs.openstack.org/91/651591/2/check/cross-glance-py27/d8d97b5/ : SUCCESS in 4m 16s\n- cross-glance-py36 http://logs.openstack.org/91/651591/2/check/cross-glance-py36/974cee0/ : SUCCESS in 3m 45s\n- cross-horizon-py27 http://logs.openstack.org/91/651591/2/check/cross-horizon-py27/9938ce6/ : SUCCESS in 9m 42s\n- cross-horizon-py36 http://logs.openstack.org/91/651591/2/check/cross-horizon-py36/97117c6/ : SUCCESS in 8m 57s\n- cross-horizon-npm http://logs.openstack.org/91/651591/2/check/cross-horizon-npm/01703cc/ : SUCCESS in 5m 18s\n- cross-keystone-py27 http://logs.openstack.org/91/651591/2/check/cross-keystone-py27/6908d8c/ : SUCCESS in 21m 06s\n- cross-keystone-py36 http://logs.openstack.org/91/651591/2/check/cross-keystone-py36/4615ea3/ : SUCCESS in 23m 38s\n- cross-neutron-py27 http://logs.openstack.org/91/651591/2/check/cross-neutron-py27/4ff4de0/ : SUCCESS in 22m 27s\n- cross-neutron-py36 http://logs.openstack.org/91/651591/2/check/cross-neutron-py36/01ea689/ : SUCCESS in 19m 48s\n- cross-nova-py27 http://logs.openstack.org/91/651591/2/check/cross-nova-py27/1063a10/ : SUCCESS in 12m 19s\n- cross-nova-py36 http://logs.openstack.org/91/651591/2/check/cross-nova-py36/051a09e/ : SUCCESS in 11m 44s\n- cross-nova-functional http://logs.openstack.org/91/651591/2/check/cross-nova-functional/24716dd/ : SUCCESS in 23m 40s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/91/651591/2/check/cross-osc-build-sphinx-docs/6bc78db/html/ : SUCCESS in 6m 34s\n- cross-swift-py27 http://logs.openstack.org/91/651591/2/check/cross-swift-py27/1bb8b18/ : SUCCESS in 8m 11s\n- cross-swift-py36 http://logs.openstack.org/91/651591/2/check/cross-swift-py36/0d21054/ : SUCCESS in 8m 14s\n- requirements-integration http://logs.openstack.org/91/651591/2/check/requirements-integration/e0824c8/ : SUCCESS in 52m 43s\n- neutron-grenade-multinode http://logs.openstack.org/91/651591/2/check/neutron-grenade-multinode/def764b/ : SUCCESS in 1h 06m 51s\n- tempest-full http://logs.openstack.org/91/651591/2/check/tempest-full/dce9f7b/ : FAILURE in 1h 30m 15s\n- neutron-grenade http://logs.openstack.org/91/651591/2/check/neutron-grenade/9ac9d61/ : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":2},{"id":"d28066a57eedeeff70895e3c98a1c088f8d1bd25","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-27 17:29:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f573b522bbd89373e2cabaa3eaa1f84728909e63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 18:56:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/651591/3/check/openstack-tox-docs/ff5841c/html/ : SUCCESS in 3m 37s\n- openstack-tox-pypy http://logs.openstack.org/91/651591/3/check/openstack-tox-pypy/2af00a2/ : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/651591/3/check/openstack-tox-pep8/7db1ac6/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/91/651591/3/check/openstack-tox-py27/b92b66f/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/91/651591/3/check/openstack-tox-py36/d852b9d/ : SUCCESS in 4m 25s\n- openstack-tox-validate http://logs.openstack.org/91/651591/3/check/openstack-tox-validate/fc7b52a/ : SUCCESS in 3m 06s\n- requirements-tox-babel http://logs.openstack.org/91/651591/3/check/requirements-tox-babel/ae4b65f/ : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/651591/3/check/requirements-tox-py27-check-uc/21a1ea3/ : SUCCESS in 7m 20s\n- requirements-tox-py36-check-uc http://logs.openstack.org/91/651591/3/check/requirements-tox-py36-check-uc/25c3f9f/ : SUCCESS in 8m 31s\n- cross-cinder-py27 http://logs.openstack.org/91/651591/3/check/cross-cinder-py27/9ec5ede/ : SUCCESS in 8m 55s\n- cross-cinder-py36 http://logs.openstack.org/91/651591/3/check/cross-cinder-py36/fb8450b/ : SUCCESS in 9m 29s\n- cross-glance-py27 http://logs.openstack.org/91/651591/3/check/cross-glance-py27/95570b2/ : SUCCESS in 4m 30s\n- cross-glance-py36 http://logs.openstack.org/91/651591/3/check/cross-glance-py36/53d3fd8/ : SUCCESS in 5m 32s\n- cross-horizon-py27 http://logs.openstack.org/91/651591/3/check/cross-horizon-py27/a35e5c7/ : SUCCESS in 10m 19s\n- cross-horizon-py36 http://logs.openstack.org/91/651591/3/check/cross-horizon-py36/a52581b/ : SUCCESS in 7m 25s\n- cross-horizon-npm http://logs.openstack.org/91/651591/3/check/cross-horizon-npm/7db464e/ : SUCCESS in 5m 43s\n- cross-keystone-py27 http://logs.openstack.org/91/651591/3/check/cross-keystone-py27/bb5565c/ : SUCCESS in 22m 02s\n- cross-keystone-py36 http://logs.openstack.org/91/651591/3/check/cross-keystone-py36/5754800/ : SUCCESS in 23m 41s\n- cross-neutron-py27 http://logs.openstack.org/91/651591/3/check/cross-neutron-py27/36393f0/ : SUCCESS in 20m 52s\n- cross-neutron-py36 http://logs.openstack.org/91/651591/3/check/cross-neutron-py36/24b72c6/ : SUCCESS in 24m 24s\n- cross-nova-py27 http://logs.openstack.org/91/651591/3/check/cross-nova-py27/44d3d34/ : SUCCESS in 11m 23s\n- cross-nova-py36 http://logs.openstack.org/91/651591/3/check/cross-nova-py36/a589f05/ : SUCCESS in 12m 32s\n- cross-nova-functional http://logs.openstack.org/91/651591/3/check/cross-nova-functional/f1c74f4/ : SUCCESS in 18m 26s\n- cross-osc-tox-docs http://logs.openstack.org/91/651591/3/check/cross-osc-tox-docs/97bb158/html/ : SUCCESS in 5m 58s\n- cross-swift-py27 http://logs.openstack.org/91/651591/3/check/cross-swift-py27/8e04360/ : SUCCESS in 10m 20s\n- cross-swift-py36 http://logs.openstack.org/91/651591/3/check/cross-swift-py36/44e2793/ : SUCCESS in 7m 51s\n- requirements-integration http://logs.openstack.org/91/651591/3/check/requirements-integration/981fddf/ : SUCCESS in 1h 04m 44s\n- neutron-grenade-multinode http://logs.openstack.org/91/651591/3/check/neutron-grenade-multinode/fc43bf4/ : SUCCESS in 1h 18m 26s\n- tempest-full http://logs.openstack.org/91/651591/3/check/tempest-full/86582b5/ : FAILURE in 1h 21m 36s\n- neutron-grenade http://logs.openstack.org/91/651591/3/check/neutron-grenade/d828231/ : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":3},{"id":"f8d1c49f1317e93327a12004a985d92eccf62f6f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-28 17:25:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"52cd42b3d6647b6123a2c0515e447e8df9a3fea7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-05-28 17:30:37.000000000","message":"Patch Set 4:\n\n\u003e Uploaded patch set 4.\n\nneutron-lib \u003e 1.26.0 is still not released. I need to talk to boden release a new version.","accounts_in_message":[],"_revision_number":4},{"id":"3468ee9810d18dbaec339d3922a355b3cbcb25aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 19:41:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/651591/4/check/openstack-tox-docs/1b26107/html/ : SUCCESS in 2m 52s\n- openstack-tox-pypy http://logs.openstack.org/91/651591/4/check/openstack-tox-pypy/4a28d1a/ : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/651591/4/check/openstack-tox-pep8/e45fd2e/ : SUCCESS in 2m 36s\n- openstack-tox-py27 http://logs.openstack.org/91/651591/4/check/openstack-tox-py27/83d1584/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/91/651591/4/check/openstack-tox-py36/c312da2/ : SUCCESS in 3m 22s\n- openstack-tox-validate http://logs.openstack.org/91/651591/4/check/openstack-tox-validate/418b876/ : SUCCESS in 2m 44s\n- requirements-tox-babel http://logs.openstack.org/91/651591/4/check/requirements-tox-babel/755aa9d/ : SUCCESS in 3m 24s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/651591/4/check/requirements-tox-py27-check-uc/543af9f/ : SUCCESS in 6m 14s\n- requirements-tox-py36-check-uc http://logs.openstack.org/91/651591/4/check/requirements-tox-py36-check-uc/bcefbe3/ : SUCCESS in 6m 37s\n- cross-cinder-py27 http://logs.openstack.org/91/651591/4/check/cross-cinder-py27/afecf8e/ : SUCCESS in 8m 12s\n- cross-cinder-py36 http://logs.openstack.org/91/651591/4/check/cross-cinder-py36/82c8311/ : SUCCESS in 8m 45s\n- cross-glance-py27 http://logs.openstack.org/91/651591/4/check/cross-glance-py27/939bc55/ : SUCCESS in 4m 30s\n- cross-glance-py36 http://logs.openstack.org/91/651591/4/check/cross-glance-py36/7c6360e/ : SUCCESS in 4m 28s\n- cross-horizon-py27 http://logs.openstack.org/91/651591/4/check/cross-horizon-py27/9c4d6ea/ : SUCCESS in 10m 41s\n- cross-horizon-py36 http://logs.openstack.org/91/651591/4/check/cross-horizon-py36/5d5e7f0/ : SUCCESS in 9m 03s\n- cross-horizon-npm http://logs.openstack.org/91/651591/4/check/cross-horizon-npm/9087da9/ : SUCCESS in 6m 01s\n- cross-keystone-py27 http://logs.openstack.org/91/651591/4/check/cross-keystone-py27/0cdcf33/ : SUCCESS in 25m 12s\n- cross-keystone-py36 http://logs.openstack.org/91/651591/4/check/cross-keystone-py36/7a05fc0/ : SUCCESS in 28m 30s\n- cross-neutron-py27 http://logs.openstack.org/91/651591/4/check/cross-neutron-py27/4a35047/ : SUCCESS in 21m 54s\n- cross-neutron-py36 http://logs.openstack.org/91/651591/4/check/cross-neutron-py36/d5fa959/ : SUCCESS in 23m 09s\n- cross-nova-py27 http://logs.openstack.org/91/651591/4/check/cross-nova-py27/1016187/ : SUCCESS in 13m 14s\n- cross-nova-py36 http://logs.openstack.org/91/651591/4/check/cross-nova-py36/8638d14/ : SUCCESS in 15m 30s\n- cross-nova-functional http://logs.openstack.org/91/651591/4/check/cross-nova-functional/a926f24/ : SUCCESS in 22m 40s\n- cross-osc-tox-docs http://logs.openstack.org/91/651591/4/check/cross-osc-tox-docs/ebf2c10/html/ : SUCCESS in 5m 11s\n- cross-swift-py27 http://logs.openstack.org/91/651591/4/check/cross-swift-py27/2818ebf/ : SUCCESS in 8m 35s\n- cross-swift-py36 http://logs.openstack.org/91/651591/4/check/cross-swift-py36/befb2ad/ : SUCCESS in 8m 01s\n- requirements-integration http://logs.openstack.org/91/651591/4/check/requirements-integration/ae00981/ : SUCCESS in 1h 12m 10s\n- neutron-grenade-multinode http://logs.openstack.org/91/651591/4/check/neutron-grenade-multinode/cefb6f2/ : SUCCESS in 1h 09m 45s\n- tempest-full http://logs.openstack.org/91/651591/4/check/tempest-full/996001e/ : FAILURE in 1h 17m 54s\n- neutron-grenade http://logs.openstack.org/91/651591/4/check/neutron-grenade/60911fb/ : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":4},{"id":"ea2722f67dd1b78a8a0df9cf395a21dfbc2d199d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-16 20:52:19.000000000","message":"Abandoned\n\nthis has been done :D","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e8bced16c2c10f46eaf130956e368a08d3938df7","revisions":{"bb0199410692d4415a41967edd0bedc82ef2689e":{"kind":"REWORK","_number":1,"created":"2019-04-10 15:48:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/651591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/651591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/651591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/651591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/651591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/651591/1"}}},"commit":{"parents":[{"commit":"bb2bd77212d86c2c57c3b16caa1714dececc929a","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb2bd77212d86c2c57c3b16caa1714dececc929a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-04-10 15:44:26.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-04-10 15:48:04.000000000","tz":0},"subject":"Bump SQLAlchemy to 1.3.2","message":"Bump SQLAlchemy to 1.3.2\n\nIn [1] this version was skipped because of a problem in Neutron.\nThis problem was fixed in neutron-lib\u003d\u003dx.x.x (to be completed).\n\nRelated-Bug: #1824178\n\n[1] https://review.openstack.org/#/c/649508\n\nChange-Id: I988179f302325ceb5d7411362121c2505864a9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb0199410692d4415a41967edd0bedc82ef2689e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb0199410692d4415a41967edd0bedc82ef2689e"}]},"branch":"refs/heads/master"},"3d06818ed29f83874d3264fa5109696a320bed42":{"kind":"REWORK","_number":2,"created":"2019-05-02 17:44:05.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/91/651591/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/651591/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/651591/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/651591/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/651591/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/651591/2"}}},"commit":{"parents":[{"commit":"e08870c9a38fa8a066ae535c1f7bb3b6a0593809","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e08870c9a38fa8a066ae535c1f7bb3b6a0593809"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-04-10 15:44:26.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-02 17:43:53.000000000","tz":-300},"subject":"Bump SQLAlchemy to 1.3.3","message":"Bump SQLAlchemy to 1.3.3\n\nIn [1] this version was skipped because of a problem in Neutron.\nThis problem was fixed in neutron-lib\u003d\u003dx.x.x (to be completed).\n\nRelated-Bug: #1824178\n\n[1] https://review.openstack.org/#/c/649508\n\nChange-Id: I988179f302325ceb5d7411362121c2505864a9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d06818ed29f83874d3264fa5109696a320bed42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3d06818ed29f83874d3264fa5109696a320bed42"}]},"branch":"refs/heads/master"},"da9836461fc9d88d9c982b89a2381ec8861cf10d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-27 17:29:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/651591/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/651591/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/651591/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/651591/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/651591/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/651591/3"}}},"commit":{"parents":[{"commit":"3fd78e5eb1dda2d33a73a89a6704b3f29c910847","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fd78e5eb1dda2d33a73a89a6704b3f29c910847"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-04-10 15:44:26.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-05-27 17:29:59.000000000","tz":0},"subject":"Bump SQLAlchemy to 1.3.3","message":"Bump SQLAlchemy to 1.3.3\n\nIn [1] this version was skipped because of a problem in Neutron.\nThis problem was fixed in neutron-lib\u003d\u003dx.x.x (to be completed).\n\nRelated-Bug: #1824178\n\n[1] https://review.openstack.org/#/c/649508\n\nChange-Id: I988179f302325ceb5d7411362121c2505864a9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da9836461fc9d88d9c982b89a2381ec8861cf10d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da9836461fc9d88d9c982b89a2381ec8861cf10d"}]},"branch":"refs/heads/master"},"e8bced16c2c10f46eaf130956e368a08d3938df7":{"kind":"REWORK","_number":4,"created":"2019-05-28 17:25:20.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/91/651591/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/651591/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/651591/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/651591/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/651591/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/651591/4"}}},"commit":{"parents":[{"commit":"3fd78e5eb1dda2d33a73a89a6704b3f29c910847","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fd78e5eb1dda2d33a73a89a6704b3f29c910847"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-04-10 15:44:26.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-28 17:25:02.000000000","tz":-300},"subject":"Bump SQLAlchemy to 1.3.4","message":"Bump SQLAlchemy to 1.3.4\n\nIn [1] this version was skipped because of a problem in Neutron.\nThis problem was fixed in neutron-lib\u003d\u003dx.x.x (to be completed).\n\nRelated-Bug: #1824178\n\n[1] https://review.openstack.org/#/c/649508\n\nChange-Id: I988179f302325ceb5d7411362121c2505864a9ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8bced16c2c10f46eaf130956e368a08d3938df7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e8bced16c2c10f46eaf130956e368a08d3938df7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
