)]}'
{"id":"openstack%2Frequirements~653091","triplet_id":"openstack%2Frequirements~stable%2Fpike~Iadcd75d47bfa7b3452079fa02ed1f3038b6f9373","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"Iadcd75d47bfa7b3452079fa02ed1f3038b6f9373","subject":"update constraint for os-client-config to new release 1.28.1","status":"MERGED","created":"2019-04-16 19:16:48.000000000","updated":"2019-04-26 04:36:17.000000000","submitted":"2019-04-26 04:36:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653091-1556253377907-a42c10a4","meta_rev_id":"96175aa4f7750e294d6c4ebc574819837f0b7460","_number":653091,"virtual_id_number":653091,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-04-26 04:36:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-22 18:28:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-17 05:32:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-04-22 18:28:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 05:32:25.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-22 18:28:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-26 04:36:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"484342ff33ba605f5886b239daf0708d811846be","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-16 19:16:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebb6c3ea2e32a899783e8d805a4207b40399244b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 23:40:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/91/653091/1/check/openstack-tox-pypy/325a4f5/ : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/check/openstack-tox-pep8/72fd41a/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/check/openstack-tox-py27/fb94d82/ : SUCCESS in 4m 22s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/check/openstack-tox-py35/3a8572c/ : SUCCESS in 3m 49s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/check/openstack-tox-validate/48aca07/ : SUCCESS in 3m 02s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/check/requirements-tox-babel/c48f2a2/ : SUCCESS in 3m 49s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/check/requirements-tox-py27-check-uc/e8427a5/ : SUCCESS in 6m 02s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/check/cross-cinder-py27/757c3c7/ : SUCCESS in 7m 20s\n- cross-cinder-py35 http://logs.openstack.org/91/653091/1/check/cross-cinder-py35/00b3bd0/ : SUCCESS in 8m 38s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/check/cross-glance-py27/781883d/ : SUCCESS in 10m 45s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/check/cross-glance-py35/5ac825a/ : SUCCESS in 11m 22s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/check/cross-horizon-py27/5f5edfc/ : SUCCESS in 10m 20s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/check/cross-horizon-py35/a4965fd/ : SUCCESS in 11m 37s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/check/cross-keystone-py27/409e717/ : SUCCESS in 11m 51s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/check/cross-keystone-py35/fdf18a3/ : SUCCESS in 11m 29s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/check/cross-neutron-py27/f08e32d/ : SUCCESS in 23m 08s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/check/cross-neutron-py35/1b12996/ : SUCCESS in 23m 04s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/check/cross-nova-py27/b95f7ef/ : SUCCESS in 13m 02s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/check/cross-nova-py35/4013ed1/ : SUCCESS in 13m 52s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/check/cross-nova-functional/6adfb61/ : SUCCESS in 18m 31s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/check/cross-swift-py27/ff67011/ : SUCCESS in 8m 17s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/check/cross-swift-py35/2722326/ : SUCCESS in 3m 33s\n- requirements-integration http://logs.openstack.org/91/653091/1/check/requirements-integration/97e26cc/ : SUCCESS in 45m 04s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/check/neutron-grenade-multinode/af4ac88/ : SUCCESS in 1h 00m 26s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/91/653091/1/check/legacy-tempest-dsvm-neutron-full/d843ce3/ : SUCCESS in 1h 29m 57s\n- neutron-grenade http://logs.openstack.org/91/653091/1/check/neutron-grenade/6e0b46f/ : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b3dd6191695a36ace270d339d9145521c38f59e4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-16 23:45:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf560d76f68881a165590e7b32f1eeb8c3f6f19e","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-17 05:32:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5a5e14c07e670f830a55d948c3cd3408bcc0c731","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 05:32:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73124ffa3b3d1addf9f36dedccb64140eab5338f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 10:24:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/gate/openstack-tox-pep8/1e2342a/ : SUCCESS in 4m 38s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/gate/openstack-tox-py27/819b790/ : SUCCESS in 4m 57s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/gate/openstack-tox-py35/b5a7053/ : SUCCESS in 5m 00s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/gate/openstack-tox-validate/caecd9e/ : SUCCESS in 4m 44s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/gate/requirements-tox-babel/6fea44c/ : SUCCESS in 4m 38s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/gate/requirements-tox-py27-check-uc/0120c58/ : SUCCESS in 6m 38s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/gate/cross-cinder-py27/7e45bf1/ : SUCCESS in 10m 13s\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/gate/cross-glance-py27/e840009/ : SUCCESS in 11m 07s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/gate/cross-glance-py35/91af673/ : SUCCESS in 11m 28s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/gate/cross-horizon-py27/e9d698f/ : SUCCESS in 11m 27s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/gate/cross-horizon-py35/5a86b6d/ : SUCCESS in 10m 50s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/gate/cross-keystone-py27/287916f/ : SUCCESS in 16m 41s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/gate/cross-keystone-py35/c470429/ : SUCCESS in 11m 04s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/gate/cross-neutron-py27/962da71/ : SUCCESS in 23m 30s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/gate/cross-neutron-py35/63bf564/ : SUCCESS in 22m 28s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/gate/cross-nova-py27/4686a36/ : SUCCESS in 12m 56s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/gate/cross-nova-py35/9252971/ : SUCCESS in 13m 45s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/gate/cross-nova-functional/66b7ea5/ : FAILURE in 19m 06s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/gate/cross-swift-py27/e7f4787/ : SUCCESS in 8m 07s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/gate/cross-swift-py35/d5b63b2/ : SUCCESS in 4m 46s\n- requirements-integration http://logs.openstack.org/91/653091/1/gate/requirements-integration/5c2f85f/ : SUCCESS in 51m 49s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/gate/neutron-grenade-multinode/e87e642/ : SUCCESS in 54m 22s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/91/653091/1/gate/legacy-tempest-dsvm-neutron-full/d5065b2/ : SUCCESS in 1h 27m 04s\n- neutron-grenade http://logs.openstack.org/91/653091/1/gate/neutron-grenade/a80483c/ : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7cabdb4db88925464ec0504bf90939880d3b9f00","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-17 18:40:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a1638b54c032743c82b2fb6cb9a02382ef9f5457","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 02:24:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/91/653091/1/check/openstack-tox-pypy/6497de1/ : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/check/openstack-tox-pep8/3657912/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/check/openstack-tox-py27/845d4f3/ : SUCCESS in 3m 31s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/check/openstack-tox-py35/c7e5c30/ : SUCCESS in 4m 07s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/check/openstack-tox-validate/53df0ae/ : SUCCESS in 3m 36s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/check/requirements-tox-babel/1e6579e/ : SUCCESS in 3m 48s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/check/requirements-tox-py27-check-uc/293ceb5/ : SUCCESS in 9m 30s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/check/cross-cinder-py27/c4bf125/ : SUCCESS in 7m 40s\n- cross-cinder-py35 http://logs.openstack.org/91/653091/1/check/cross-cinder-py35/f919e47/ : SUCCESS in 7m 19s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/check/cross-glance-py27/55fe09e/ : SUCCESS in 8m 53s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/check/cross-glance-py35/ec27d83/ : SUCCESS in 9m 15s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/check/cross-horizon-py27/f3af544/ : SUCCESS in 9m 31s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/check/cross-horizon-py35/18243bc/ : SUCCESS in 9m 59s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/check/cross-keystone-py27/9ddb5ed/ : SUCCESS in 25m 18s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/check/cross-keystone-py35/732166f/ : SUCCESS in 22m 47s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/check/cross-neutron-py27/3554be4/ : SUCCESS in 30m 35s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/check/cross-neutron-py35/22ed29d/ : SUCCESS in 22m 12s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/check/cross-nova-py27/1bc036c/ : SUCCESS in 11m 51s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/check/cross-nova-py35/83be196/ : SUCCESS in 15m 59s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/check/cross-nova-functional/180d4e6/ : SUCCESS in 20m 13s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/check/cross-swift-py27/dd61220/ : SUCCESS in 7m 24s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/check/cross-swift-py35/f8804fb/ : SUCCESS in 3m 43s\n- requirements-integration http://logs.openstack.org/91/653091/1/check/requirements-integration/d017e29/ : SUCCESS in 52m 45s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/check/neutron-grenade-multinode/fa33a90/ : SUCCESS in 1h 02m 22s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/91/653091/1/check/legacy-tempest-dsvm-neutron-full/10253d6/ : SUCCESS in 1h 34m 54s\n- neutron-grenade http://logs.openstack.org/91/653091/1/check/neutron-grenade/047cc75/ : SUCCESS in 51m 57s","accounts_in_message":[],"_revision_number":1},{"id":"50bccc09229df7799fc65f261a21e19249429407","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 02:25:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8522fa83bf9b9cb74be8dc38f37d3be91121e0ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 13:19:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/gate/openstack-tox-pep8/ee77ed9/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/gate/openstack-tox-py27/41391fd/ : SUCCESS in 4m 33s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/gate/openstack-tox-py35/fdf5205/ : SUCCESS in 5m 45s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/gate/openstack-tox-validate/c4d2fdf/ : SUCCESS in 6m 14s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/gate/requirements-tox-babel/6fe4fa6/ : SUCCESS in 4m 05s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/gate/requirements-tox-py27-check-uc/1e61e62/ : SUCCESS in 7m 11s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/gate/cross-cinder-py27/e760f37/ : SUCCESS in 8m 32s\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/gate/cross-glance-py27/186118a/ : SUCCESS in 9m 28s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/gate/cross-glance-py35/7c6f403/ : SUCCESS in 10m 24s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/gate/cross-horizon-py27/c3e44f5/ : SUCCESS in 11m 51s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/gate/cross-horizon-py35/1e9f302/ : SUCCESS in 10m 43s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/gate/cross-keystone-py27/1808449/ : SUCCESS in 13m 30s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/gate/cross-keystone-py35/606022a/ : SUCCESS in 20m 11s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/gate/cross-neutron-py27/ae162e9/ : SUCCESS in 29m 04s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/gate/cross-neutron-py35/0d9456b/ : SUCCESS in 27m 21s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/gate/cross-nova-py27/407b346/ : SUCCESS in 12m 29s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/gate/cross-nova-py35/ad0a7c5/ : SUCCESS in 13m 43s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/gate/cross-nova-functional/e9f170d/ : SUCCESS in 23m 21s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/gate/cross-swift-py27/f9fa667/ : SUCCESS in 8m 03s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/gate/cross-swift-py35/34c5540/ : SUCCESS in 4m 29s\n- requirements-integration http://logs.openstack.org/91/653091/1/gate/requirements-integration/a75a57a/ : SUCCESS in 55m 16s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/gate/neutron-grenade-multinode/a48136d/ : SUCCESS in 1h 13m 46s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/91/653091/1/gate/legacy-tempest-dsvm-neutron-full/b93ead6/ : FAILURE in 1h 23m 29s\n- neutron-grenade http://logs.openstack.org/91/653091/1/gate/neutron-grenade/6dfe655/ : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3d55e2bec991be41926d6ab030895bf825dc95c6","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-18 13:46:30.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"ea1a5b9c12de101ca938b07b4fa7bf5b5a1524d0","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-18 13:46:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0c9d6fbd8800c36553fbcfd647fb27a3333e1207","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-18 13:46:58.000000000","message":"Patch Set 1:\n\nlets not approve this until the rest is through the gate","accounts_in_message":[],"_revision_number":1},{"id":"00e469ccd71349fe38913c47746fe02c97109da2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 01:06:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/91/653091/1/check/openstack-tox-pypy/031d4e8/ : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/check/openstack-tox-pep8/242b119/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/check/openstack-tox-py27/a5fc58a/ : SUCCESS in 4m 24s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/check/openstack-tox-py35/b3c6dfa/ : SUCCESS in 4m 00s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/check/openstack-tox-validate/c6e728a/ : SUCCESS in 3m 54s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/check/requirements-tox-babel/c939030/ : SUCCESS in 3m 33s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/check/requirements-tox-py27-check-uc/c6e672d/ : SUCCESS in 7m 20s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/check/cross-cinder-py27/7819217/ : SUCCESS in 7m 40s\n- cross-cinder-py35 http://logs.openstack.org/91/653091/1/check/cross-cinder-py35/670db22/ : SUCCESS in 7m 22s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/check/cross-glance-py27/e28043b/ : SUCCESS in 9m 47s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/check/cross-glance-py35/68841dc/ : SUCCESS in 12m 00s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/check/cross-horizon-py27/de39cf7/ : SUCCESS in 10m 27s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/check/cross-horizon-py35/b9eca5f/ : SUCCESS in 10m 03s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/check/cross-keystone-py27/11cbae9/ : SUCCESS in 18m 27s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/check/cross-keystone-py35/cf9b8b2/ : SUCCESS in 11m 29s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/check/cross-neutron-py27/b112b64/ : SUCCESS in 25m 23s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/check/cross-neutron-py35/a60f82d/ : SUCCESS in 21m 27s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/check/cross-nova-py27/06b048a/ : SUCCESS in 15m 51s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/check/cross-nova-py35/4e8aa9b/ : SUCCESS in 13m 03s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/check/cross-nova-functional/4a6676b/ : SUCCESS in 24m 29s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/check/cross-swift-py27/2ec7477/ : SUCCESS in 7m 48s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/check/cross-swift-py35/6d57dab/ : SUCCESS in 4m 09s\n- requirements-integration http://logs.openstack.org/91/653091/1/check/requirements-integration/b18d3ad/ : SUCCESS in 47m 10s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/check/neutron-grenade-multinode/655073a/ : SUCCESS in 1h 01m 50s\n- legacy-tempest-dsvm-neutron-full finger://ze11.openstack.org/8628059a673943b480b18ad4b487991c : RETRY_LIMIT in 1h 07m 06s\n- neutron-grenade http://logs.openstack.org/91/653091/1/check/neutron-grenade/72f945f/ : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f6e56ccdbcfa6fcf1d8b09c8d38845451914ea09","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 14:55:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de19d9cbe406e6a384ff6c40b9889a615383ccba","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 16:19:01.000000000","message":"Patch Set 1: -Code-Review\n\nrecheck (and unvote til gate works again)","accounts_in_message":[],"_revision_number":1},{"id":"b4a7f64692d50757c44dc1890dde403f7870a6c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:09:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/91/653091/1/check/openstack-tox-pypy/b6bc635/ : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/check/openstack-tox-pep8/b30cfd6/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/check/openstack-tox-py27/5d865b1/ : SUCCESS in 3m 03s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/check/openstack-tox-py35/b8ec7fd/ : SUCCESS in 3m 41s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/check/openstack-tox-validate/bac0c3f/ : SUCCESS in 3m 33s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/check/requirements-tox-babel/b1ff687/ : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/check/requirements-tox-py27-check-uc/38d55b1/ : SUCCESS in 6m 42s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/check/cross-cinder-py27/6c5fee9/ : SUCCESS in 7m 51s\n- cross-cinder-py35 http://logs.openstack.org/91/653091/1/check/cross-cinder-py35/0692fcb/ : SUCCESS in 8m 10s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/check/cross-glance-py27/8abd7a9/ : SUCCESS in 9m 01s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/check/cross-glance-py35/63802c4/ : SUCCESS in 9m 18s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/check/cross-horizon-py27/21116ac/ : SUCCESS in 11m 08s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/check/cross-horizon-py35/e57a294/ : SUCCESS in 10m 13s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/check/cross-keystone-py27/7469b30/ : SUCCESS in 11m 04s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/check/cross-keystone-py35/5aa99d5/ : SUCCESS in 10m 50s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/check/cross-neutron-py27/25f3317/ : SUCCESS in 22m 10s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/check/cross-neutron-py35/49c93fc/ : SUCCESS in 23m 16s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/check/cross-nova-py27/3c4a888/ : SUCCESS in 12m 16s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/check/cross-nova-py35/df7ba05/ : SUCCESS in 12m 04s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/check/cross-nova-functional/9d42937/ : SUCCESS in 19m 01s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/check/cross-swift-py27/f1baf68/ : SUCCESS in 7m 52s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/check/cross-swift-py35/49aaae3/ : SUCCESS in 3m 44s\n- requirements-integration http://logs.openstack.org/91/653091/1/check/requirements-integration/4c617ac/ : SUCCESS in 55m 23s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/check/neutron-grenade-multinode/4fb6b2f/ : SUCCESS in 1h 03m 43s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/91/653091/1/check/legacy-tempest-dsvm-neutron-full/a5d8cb1/ : SUCCESS in 1h 33m 33s\n- neutron-grenade http://logs.openstack.org/91/653091/1/check/neutron-grenade/1d919e7/ : SUCCESS in 50m 32s","accounts_in_message":[],"_revision_number":1},{"id":"dfb2bbd33b2335ac14458722e8541857fa1eaa5f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 18:28:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42f2443b425790c3050d07b8402afa1ca00e1c3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:28:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5386cd8730ee3a87818736b1439a792c9af4ec56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:10:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/gate/openstack-tox-pep8/70f0c39/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/gate/openstack-tox-py27/bf6c683/ : SUCCESS in 3m 58s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/gate/openstack-tox-py35/ab558b5/ : SUCCESS in 4m 17s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/gate/openstack-tox-validate/ccf5eea/ : SUCCESS in 3m 46s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/gate/requirements-tox-babel/346b1e2/ : SUCCESS in 5m 17s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/gate/requirements-tox-py27-check-uc/52594ee/ : SUCCESS in 7m 20s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/gate/cross-cinder-py27/aab9b67/ : SUCCESS in 7m 23s\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/gate/cross-glance-py27/f4e9b85/ : SUCCESS in 8m 57s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/gate/cross-glance-py35/f201d25/ : SUCCESS in 13m 30s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/gate/cross-horizon-py27/7c90e56/ : SUCCESS in 10m 50s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/gate/cross-horizon-py35/235a3e2/ : SUCCESS in 9m 59s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/gate/cross-keystone-py27/8358add/ : SUCCESS in 18m 32s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/gate/cross-keystone-py35/156f41d/ : TIMED_OUT in 40m 59s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/gate/cross-neutron-py27/aed45ab/ : SUCCESS in 21m 51s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/gate/cross-neutron-py35/cc65c51/ : SUCCESS in 22m 08s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/gate/cross-nova-py27/7b424eb/ : SUCCESS in 14m 06s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/gate/cross-nova-py35/d06281e/ : SUCCESS in 12m 40s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/gate/cross-nova-functional/caf7364/ : SUCCESS in 19m 40s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/gate/cross-swift-py27/69239e3/ : SUCCESS in 8m 49s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/gate/cross-swift-py35/af8621e/ : SUCCESS in 3m 12s\n- requirements-integration http://logs.openstack.org/91/653091/1/gate/requirements-integration/acddb6f/ : SUCCESS in 52m 39s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/gate/neutron-grenade-multinode/22ca3c6/ : SUCCESS in 1h 02m 52s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/91/653091/1/gate/legacy-tempest-dsvm-neutron-full/78dd2fb/ : SUCCESS in 1h 30m 13s\n- neutron-grenade http://logs.openstack.org/91/653091/1/gate/neutron-grenade/19b1c59/ : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ecb51841a8ba807ab80350c5dd8e24eca56a1e1d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 20:15:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"450708d40fc1d39fb7563a20583dc3bc1332ccd6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 01:28:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/91/653091/1/check/openstack-tox-pypy/2bbe8db/ : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/check/openstack-tox-pep8/82b05b5/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/check/openstack-tox-py27/d2b74c7/ : SUCCESS in 4m 23s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/check/openstack-tox-py35/c767304/ : SUCCESS in 4m 08s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/check/openstack-tox-validate/80ed31c/ : SUCCESS in 5m 08s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/check/requirements-tox-babel/2f66f80/ : SUCCESS in 4m 04s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/check/requirements-tox-py27-check-uc/cc8ed34/ : SUCCESS in 5m 59s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/check/cross-cinder-py27/4b1c81f/ : SUCCESS in 7m 32s\n- cross-cinder-py35 http://logs.openstack.org/91/653091/1/check/cross-cinder-py35/f896a5d/ : SUCCESS in 7m 23s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/check/cross-glance-py27/97d6c52/ : SUCCESS in 9m 11s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/check/cross-glance-py35/e92db73/ : SUCCESS in 11m 43s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/check/cross-horizon-py27/ece2bb0/ : SUCCESS in 10m 41s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/check/cross-horizon-py35/5ba1961/ : SUCCESS in 9m 53s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/check/cross-keystone-py27/6e672d1/ : SUCCESS in 12m 42s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/check/cross-keystone-py35/13d3a26/ : SUCCESS in 11m 49s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/check/cross-neutron-py27/adad96a/ : SUCCESS in 29m 24s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/check/cross-neutron-py35/16ab950/ : SUCCESS in 22m 12s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/check/cross-nova-py27/20c8379/ : SUCCESS in 12m 54s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/check/cross-nova-py35/28e66e9/ : SUCCESS in 11m 54s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/check/cross-nova-functional/6af4660/ : SUCCESS in 18m 14s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/check/cross-swift-py27/48e335e/ : SUCCESS in 7m 54s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/check/cross-swift-py35/10d97b6/ : SUCCESS in 3m 20s\n- requirements-integration http://logs.openstack.org/91/653091/1/check/requirements-integration/aa31287/ : SUCCESS in 55m 45s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/check/neutron-grenade-multinode/7cd5cb4/ : SUCCESS in 1h 07m 33s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/91/653091/1/check/legacy-tempest-dsvm-neutron-full/6834fcc/ : FAILURE in 1h 18m 09s\n- neutron-grenade http://logs.openstack.org/91/653091/1/check/neutron-grenade/279b5bc/ : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1f00f2d47c4f23507c5d40053902e0c1aa029a61","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-23 01:34:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cfc9df50ea5a09127c08212c9362d05d48fe858e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 04:28:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/91/653091/1/check/openstack-tox-pypy/86080ca/ : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/check/openstack-tox-pep8/1f79387/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/check/openstack-tox-py27/b592e0a/ : SUCCESS in 3m 44s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/check/openstack-tox-py35/c492625/ : SUCCESS in 3m 35s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/check/openstack-tox-validate/5f2bcdf/ : SUCCESS in 3m 02s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/check/requirements-tox-babel/3795660/ : SUCCESS in 3m 13s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/check/requirements-tox-py27-check-uc/eb34be0/ : SUCCESS in 6m 41s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/check/cross-cinder-py27/c4f64e0/ : SUCCESS in 7m 48s\n- cross-cinder-py35 http://logs.openstack.org/91/653091/1/check/cross-cinder-py35/fe177d1/ : SUCCESS in 9m 08s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/check/cross-glance-py27/a6f2e2f/ : SUCCESS in 9m 10s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/check/cross-glance-py35/ec4d2e5/ : SUCCESS in 8m 47s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/check/cross-horizon-py27/6e08e38/ : SUCCESS in 10m 31s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/check/cross-horizon-py35/52d8df6/ : SUCCESS in 10m 39s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/check/cross-keystone-py27/52f763a/ : SUCCESS in 12m 31s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/check/cross-keystone-py35/1642bcc/ : SUCCESS in 16m 39s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/check/cross-neutron-py27/ec4be86/ : SUCCESS in 22m 48s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/check/cross-neutron-py35/02be9ff/ : SUCCESS in 30m 02s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/check/cross-nova-py27/a558be2/ : SUCCESS in 14m 58s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/check/cross-nova-py35/cb337e6/ : SUCCESS in 12m 06s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/check/cross-nova-functional/2246d08/ : SUCCESS in 24m 38s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/check/cross-swift-py27/0f3b266/ : SUCCESS in 7m 01s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/check/cross-swift-py35/087d476/ : SUCCESS in 3m 45s\n- requirements-integration http://logs.openstack.org/91/653091/1/check/requirements-integration/6e4f341/ : SUCCESS in 48m 31s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/check/neutron-grenade-multinode/0ce090d/ : SUCCESS in 1h 01m 45s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/91/653091/1/check/legacy-tempest-dsvm-neutron-full/1b3977e/ : SUCCESS in 1h 25m 01s\n- neutron-grenade http://logs.openstack.org/91/653091/1/check/neutron-grenade/dd6f1ab/ : SUCCESS in 53m 28s","accounts_in_message":[],"_revision_number":1},{"id":"7fce1570c31e973a530bf44c3295ac0db4fe7484","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 04:28:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49ef3f20cb855ceb73ee98d4b12a0e746dedd93e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 05:53:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/gate/openstack-tox-pep8/42e1948/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/gate/openstack-tox-py27/909468f/ : SUCCESS in 5m 14s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/gate/openstack-tox-py35/23f56f9/ : SUCCESS in 3m 48s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/gate/openstack-tox-validate/1bd59bc/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/gate/requirements-tox-babel/c83bf7d/ : SUCCESS in 4m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/gate/requirements-tox-py27-check-uc/11b4900/ : SUCCESS in 5m 36s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/gate/cross-cinder-py27/597b05d/ : SUCCESS in 10m 01s\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/gate/cross-glance-py27/dec57fc/ : SUCCESS in 12m 40s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/gate/cross-glance-py35/fff825e/ : SUCCESS in 9m 47s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/gate/cross-horizon-py27/3b39ed0/ : SUCCESS in 12m 49s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/gate/cross-horizon-py35/e849e68/ : SUCCESS in 10m 55s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/gate/cross-keystone-py27/f645e98/ : TIMED_OUT in 40m 46s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/gate/cross-keystone-py35/b858452/ : SUCCESS in 23m 15s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/gate/cross-neutron-py27/af8f147/ : SUCCESS in 21m 28s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/gate/cross-neutron-py35/b283df8/ : SUCCESS in 32m 01s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/gate/cross-nova-py27/3e80806/ : SUCCESS in 13m 36s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/gate/cross-nova-py35/6793547/ : SUCCESS in 18m 44s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/gate/cross-nova-functional/143155b/ : FAILURE in 25m 44s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/gate/cross-swift-py27/79bc568/ : SUCCESS in 7m 51s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/gate/cross-swift-py35/a7f540e/ : SUCCESS in 3m 23s\n- requirements-integration http://logs.openstack.org/91/653091/1/gate/requirements-integration/57814b9/ : SUCCESS in 40m 14s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/gate/neutron-grenade-multinode/07f100c/ : SUCCESS in 47m 29s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/91/653091/1/gate/legacy-tempest-dsvm-neutron-full/72edb34/ : SUCCESS in 1h 23m 41s\n- neutron-grenade http://logs.openstack.org/91/653091/1/gate/neutron-grenade/a4b40a6/ : SUCCESS in 46m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a7dc2c89352e3fece6f7941a651c7bb580fbbe80","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-23 14:42:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"218a3e017ac3ded0fcd8b42995f3ed3c9baafc40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 02:45:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/91/653091/1/check/openstack-tox-pypy/a25e331/ : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/check/openstack-tox-pep8/30fad56/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/check/openstack-tox-py27/3b43c89/ : SUCCESS in 3m 44s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/check/openstack-tox-py35/ee44b53/ : SUCCESS in 4m 09s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/check/openstack-tox-validate/2dc26f9/ : SUCCESS in 3m 40s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/check/requirements-tox-babel/23fe12b/ : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/check/requirements-tox-py27-check-uc/18b19b9/ : SUCCESS in 6m 22s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/check/cross-cinder-py27/6a9cdcf/ : SUCCESS in 7m 55s\n- cross-cinder-py35 http://logs.openstack.org/91/653091/1/check/cross-cinder-py35/e7c08c9/ : SUCCESS in 8m 24s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/check/cross-glance-py27/b41050d/ : SUCCESS in 8m 57s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/check/cross-glance-py35/adb0f8c/ : SUCCESS in 9m 48s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/check/cross-horizon-py27/ce620a9/ : SUCCESS in 10m 45s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/check/cross-horizon-py35/dd62ea2/ : SUCCESS in 9m 35s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/check/cross-keystone-py27/bad0d5c/ : SUCCESS in 15m 29s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/check/cross-keystone-py35/6785b47/ : SUCCESS in 16m 15s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/check/cross-neutron-py27/d648e7d/ : SUCCESS in 21m 42s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/check/cross-neutron-py35/790645b/ : SUCCESS in 28m 03s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/check/cross-nova-py27/ea80c6c/ : SUCCESS in 13m 01s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/check/cross-nova-py35/d544cc8/ : SUCCESS in 11m 43s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/check/cross-nova-functional/99a5d70/ : FAILURE in 20m 54s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/check/cross-swift-py27/6519fd5/ : SUCCESS in 7m 59s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/check/cross-swift-py35/fdf2006/ : SUCCESS in 4m 09s\n- requirements-integration http://logs.openstack.org/91/653091/1/check/requirements-integration/54d4bf6/ : SUCCESS in 1h 02m 16s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/check/neutron-grenade-multinode/4b737e5/ : SUCCESS in 51m 43s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/91/653091/1/check/legacy-tempest-dsvm-neutron-full/4c4bec4/ : SUCCESS in 1h 29m 30s\n- neutron-grenade http://logs.openstack.org/91/653091/1/check/neutron-grenade/e52d99e/ : FAILURE in 53m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e673894ad43eea29816f8d1072ebe606ea617dfb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-24 15:18:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f7915cd6753e58cf04b937365693143f9d5acec4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 02:53:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/91/653091/1/check/openstack-tox-pypy/de06500/ : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/check/openstack-tox-pep8/4e88d40/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/check/openstack-tox-py27/429a071/ : SUCCESS in 3m 51s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/check/openstack-tox-py35/c870f8d/ : SUCCESS in 4m 16s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/check/openstack-tox-validate/6403b23/ : SUCCESS in 4m 51s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/check/requirements-tox-babel/60c2ae2/ : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/check/requirements-tox-py27-check-uc/ac3145c/ : SUCCESS in 6m 05s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/check/cross-cinder-py27/362f02f/ : SUCCESS in 9m 08s\n- cross-cinder-py35 http://logs.openstack.org/91/653091/1/check/cross-cinder-py35/fbfdb5d/ : SUCCESS in 7m 29s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/check/cross-glance-py27/bd0fef6/ : SUCCESS in 9m 06s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/check/cross-glance-py35/4bc0df3/ : SUCCESS in 8m 30s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/check/cross-horizon-py27/458ca7c/ : SUCCESS in 10m 09s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/check/cross-horizon-py35/dbb7847/ : SUCCESS in 9m 59s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/check/cross-keystone-py27/7b85630/ : SUCCESS in 14m 08s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/check/cross-keystone-py35/453a8d7/ : SUCCESS in 11m 53s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/check/cross-neutron-py27/d0362c8/ : SUCCESS in 22m 33s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/check/cross-neutron-py35/e3063f6/ : SUCCESS in 19m 14s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/check/cross-nova-py27/33b75a0/ : SUCCESS in 13m 07s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/check/cross-nova-py35/6b15613/ : SUCCESS in 11m 27s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/check/cross-nova-functional/624c776/ : SUCCESS in 22m 21s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/check/cross-swift-py27/23a253e/ : SUCCESS in 7m 30s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/check/cross-swift-py35/2f846b6/ : SUCCESS in 3m 13s\n- requirements-integration http://logs.openstack.org/91/653091/1/check/requirements-integration/30bbfdb/ : SUCCESS in 46m 54s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/check/neutron-grenade-multinode/2016c27/ : SUCCESS in 52m 27s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/91/653091/1/check/legacy-tempest-dsvm-neutron-full/45c5c2c/ : FAILURE in 1h 23m 03s\n- neutron-grenade http://logs.openstack.org/91/653091/1/check/neutron-grenade/19036ba/ : SUCCESS in 49m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8784ed40f8536f3607622928bd253a1a3aa1df51","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-25 13:14:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ca0d1948b0f595bd6983d0a50c1f79cac90b6d57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 20:35:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/91/653091/1/check/openstack-tox-pypy/c6c36a4/ : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/check/openstack-tox-pep8/f0c1ecf/ : SUCCESS in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/check/openstack-tox-py27/1a91032/ : SUCCESS in 3m 37s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/check/openstack-tox-py35/1dea2f2/ : SUCCESS in 3m 36s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/check/openstack-tox-validate/1991033/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/check/requirements-tox-babel/67623ab/ : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/check/requirements-tox-py27-check-uc/6b16826/ : SUCCESS in 6m 02s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/check/cross-cinder-py27/518a72b/ : SUCCESS in 9m 08s\n- cross-cinder-py35 http://logs.openstack.org/91/653091/1/check/cross-cinder-py35/19400ae/ : SUCCESS in 8m 36s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/check/cross-glance-py27/24ef1c8/ : SUCCESS in 8m 43s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/check/cross-glance-py35/5c1bcd4/ : SUCCESS in 10m 50s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/check/cross-horizon-py27/3b5054d/ : SUCCESS in 9m 49s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/check/cross-horizon-py35/d080404/ : SUCCESS in 9m 55s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/check/cross-keystone-py27/a7d7f7b/ : SUCCESS in 18m 04s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/check/cross-keystone-py35/cf7aacd/ : SUCCESS in 18m 22s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/check/cross-neutron-py27/1e8b801/ : SUCCESS in 24m 05s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/check/cross-neutron-py35/b823e08/ : SUCCESS in 29m 25s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/check/cross-nova-py27/0451f4c/ : SUCCESS in 12m 50s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/check/cross-nova-py35/f8497a5/ : SUCCESS in 14m 43s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/check/cross-nova-functional/5cc2c85/ : SUCCESS in 23m 06s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/check/cross-swift-py27/0ac2127/ : SUCCESS in 6m 51s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/check/cross-swift-py35/48cbe46/ : SUCCESS in 3m 03s\n- requirements-integration http://logs.openstack.org/91/653091/1/check/requirements-integration/44aca8f/ : SUCCESS in 51m 20s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/check/neutron-grenade-multinode/0419676/ : SUCCESS in 1h 02m 19s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/91/653091/1/check/legacy-tempest-dsvm-neutron-full/9a49be9/ : SUCCESS in 1h 29m 58s\n- neutron-grenade http://logs.openstack.org/91/653091/1/check/neutron-grenade/526d5a1/ : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7848bdd7acad51f5d440c37174488a7f113e9ba9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 20:36:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25fce44aee0814711c334c1259f7114fc1bb322c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 04:36:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c87e2dddb8a9b5e3adfbb2376afb4aef849371ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 04:36:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/91/653091/1/gate/openstack-tox-pep8/2b6a501/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/91/653091/1/gate/openstack-tox-py27/1d82ff4/ : SUCCESS in 3m 29s\n- openstack-tox-py35 http://logs.openstack.org/91/653091/1/gate/openstack-tox-py35/80aa66e/ : SUCCESS in 3m 38s\n- openstack-tox-validate http://logs.openstack.org/91/653091/1/gate/openstack-tox-validate/ba64a35/ : SUCCESS in 7m 45s\n- requirements-tox-babel http://logs.openstack.org/91/653091/1/gate/requirements-tox-babel/4f27cd7/ : SUCCESS in 5m 35s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653091/1/gate/requirements-tox-py27-check-uc/ce76e38/ : SUCCESS in 7m 43s\n- cross-cinder-py27 http://logs.openstack.org/91/653091/1/gate/cross-cinder-py27/cb9ca50/ : SUCCESS in 7m 06s\n- cross-glance-py27 http://logs.openstack.org/91/653091/1/gate/cross-glance-py27/ae6b0b0/ : SUCCESS in 9m 05s\n- cross-glance-py35 http://logs.openstack.org/91/653091/1/gate/cross-glance-py35/7856bc8/ : SUCCESS in 10m 42s\n- cross-horizon-py27 http://logs.openstack.org/91/653091/1/gate/cross-horizon-py27/66ed1dd/ : SUCCESS in 11m 31s\n- cross-horizon-py35 http://logs.openstack.org/91/653091/1/gate/cross-horizon-py35/69f01da/ : SUCCESS in 10m 02s\n- cross-keystone-py27 http://logs.openstack.org/91/653091/1/gate/cross-keystone-py27/674fc36/ : SUCCESS in 13m 01s\n- cross-keystone-py35 http://logs.openstack.org/91/653091/1/gate/cross-keystone-py35/ce0baa4/ : SUCCESS in 17m 49s\n- cross-neutron-py27 http://logs.openstack.org/91/653091/1/gate/cross-neutron-py27/51fe5a6/ : SUCCESS in 22m 54s\n- cross-neutron-py35 http://logs.openstack.org/91/653091/1/gate/cross-neutron-py35/9509690/ : SUCCESS in 20m 59s\n- cross-nova-py27 http://logs.openstack.org/91/653091/1/gate/cross-nova-py27/c3338a8/ : SUCCESS in 11m 35s\n- cross-nova-py35 http://logs.openstack.org/91/653091/1/gate/cross-nova-py35/5752304/ : SUCCESS in 12m 58s\n- cross-nova-functional http://logs.openstack.org/91/653091/1/gate/cross-nova-functional/bf2d341/ : SUCCESS in 22m 03s\n- cross-swift-py27 http://logs.openstack.org/91/653091/1/gate/cross-swift-py27/f20ffa9/ : SUCCESS in 6m 58s\n- cross-swift-py35 http://logs.openstack.org/91/653091/1/gate/cross-swift-py35/10646b0/ : SUCCESS in 3m 21s\n- requirements-integration http://logs.openstack.org/91/653091/1/gate/requirements-integration/c0ee3b8/ : SUCCESS in 52m 33s\n- neutron-grenade-multinode http://logs.openstack.org/91/653091/1/gate/neutron-grenade-multinode/1c7fe84/ : SUCCESS in 59m 32s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/91/653091/1/gate/legacy-tempest-dsvm-neutron-full/1f75d0d/ : SUCCESS in 1h 44m 24s\n- neutron-grenade http://logs.openstack.org/91/653091/1/gate/neutron-grenade/e2091e2/ : SUCCESS in 44m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f2afdf78759b2ffad76d907451f992de32512a9","revisions":{"2f2afdf78759b2ffad76d907451f992de32512a9":{"kind":"REWORK","_number":1,"created":"2019-04-16 19:16:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/653091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/653091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/653091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/653091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/653091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/653091/1"}}},"commit":{"parents":[{"commit":"03eeccfc2edaaf995010da7a8c2b6b6188439b30","subject":"update constraint for os-brick to new release 1.15.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03eeccfc2edaaf995010da7a8c2b6b6188439b30"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-16 19:16:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-16 19:16:46.000000000","tz":0},"subject":"update constraint for os-client-config to new release 1.28.1","message":"update constraint for os-client-config to new release 1.28.1\n\nChange-Id: Iadcd75d47bfa7b3452079fa02ed1f3038b6f9373\nmeta:version: 1.28.1\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Change-Id: Id510eb5767775defdb903c5f7ecc887361699a04\nmeta:release:Code-Review+1: Dean Troyer \u003cdtroyer@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f2afdf78759b2ffad76d907451f992de32512a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f2afdf78759b2ffad76d907451f992de32512a9"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
