)]}'
{"id":"openstack%2Frequirements~653097","triplet_id":"openstack%2Frequirements~stable%2Fpike~Icd21862779a60ae3ebcd45593eff38fa7515be40","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"Icd21862779a60ae3ebcd45593eff38fa7515be40","subject":"update constraint for oslo.i18n to new release 3.17.2","status":"MERGED","created":"2019-04-16 19:20:03.000000000","updated":"2019-04-24 04:54:28.000000000","submitted":"2019-04-24 04:54:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653097-1556081668120-9fef03c1","meta_rev_id":"6efbe7fe1c875f3af5a273722ade47176502f7a5","_number":653097,"virtual_id_number":653097,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-04-20 08:27:06.000000000","_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-04-24 04:54:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-23 14:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-04-17 06:56:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-04-17 06:56:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 06:56:26.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-04-20 08:27:06.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-23 14:43:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-24 04:54:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf2c066c6430bd243d158f3446352120cb3be389","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-16 19:20:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"664d077a168aceb5f5650ac54bfe9660020eec86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 00:10:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/97/653097/1/check/openstack-tox-pypy/d89427e/ : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/653097/1/check/openstack-tox-pep8/3202163/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/97/653097/1/check/openstack-tox-py27/d685e7e/ : SUCCESS in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/97/653097/1/check/openstack-tox-py35/7785379/ : SUCCESS in 4m 28s\n- openstack-tox-validate http://logs.openstack.org/97/653097/1/check/openstack-tox-validate/7487131/ : SUCCESS in 4m 07s\n- requirements-tox-babel http://logs.openstack.org/97/653097/1/check/requirements-tox-babel/4ed89c8/ : SUCCESS in 4m 19s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/653097/1/check/requirements-tox-py27-check-uc/503b6bb/ : SUCCESS in 10m 55s\n- cross-cinder-py27 http://logs.openstack.org/97/653097/1/check/cross-cinder-py27/5fb1374/ : SUCCESS in 9m 08s\n- cross-cinder-py35 http://logs.openstack.org/97/653097/1/check/cross-cinder-py35/d2efa31/ : SUCCESS in 9m 32s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/97/653097/1/check/cross-glance-py27/7cea423/ : SUCCESS in 9m 32s\n- cross-glance-py35 http://logs.openstack.org/97/653097/1/check/cross-glance-py35/30575a2/ : SUCCESS in 9m 39s\n- cross-horizon-py27 http://logs.openstack.org/97/653097/1/check/cross-horizon-py27/4b2e537/ : SUCCESS in 11m 59s\n- cross-horizon-py35 http://logs.openstack.org/97/653097/1/check/cross-horizon-py35/6c47306/ : SUCCESS in 10m 49s\n- cross-keystone-py27 http://logs.openstack.org/97/653097/1/check/cross-keystone-py27/1b01e26/ : SUCCESS in 13m 18s\n- cross-keystone-py35 http://logs.openstack.org/97/653097/1/check/cross-keystone-py35/cfb59bc/ : SUCCESS in 11m 37s\n- cross-neutron-py27 http://logs.openstack.org/97/653097/1/check/cross-neutron-py27/79be1cd/ : SUCCESS in 24m 45s\n- cross-neutron-py35 http://logs.openstack.org/97/653097/1/check/cross-neutron-py35/b40788a/ : SUCCESS in 34m 39s\n- cross-nova-py27 http://logs.openstack.org/97/653097/1/check/cross-nova-py27/93873aa/ : SUCCESS in 16m 18s\n- cross-nova-py35 http://logs.openstack.org/97/653097/1/check/cross-nova-py35/b944eb0/ : SUCCESS in 13m 31s\n- cross-nova-functional http://logs.openstack.org/97/653097/1/check/cross-nova-functional/8ee605a/ : SUCCESS in 30m 19s\n- cross-swift-py27 http://logs.openstack.org/97/653097/1/check/cross-swift-py27/9cb5fe3/ : SUCCESS in 8m 33s\n- cross-swift-py35 http://logs.openstack.org/97/653097/1/check/cross-swift-py35/919ee78/ : SUCCESS in 3m 43s\n- requirements-integration http://logs.openstack.org/97/653097/1/check/requirements-integration/f4a59eb/ : SUCCESS in 50m 17s\n- neutron-grenade-multinode http://logs.openstack.org/97/653097/1/check/neutron-grenade-multinode/6b20159/ : SUCCESS in 1h 01m 55s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/97/653097/1/check/legacy-tempest-dsvm-neutron-full/5611200/ : SUCCESS in 1h 43m 37s\n- neutron-grenade http://logs.openstack.org/97/653097/1/check/neutron-grenade/eb59636/ : SUCCESS in 52m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d73207eba8d6961b8b7662aa0f5669559e7d013a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-17 00:12:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab8a234cb8325906514dabd2c8209459a01d8bfc","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-17 06:56:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"14bc8350b85d22175a655a682ecb25e63cce2956","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 11:23:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ec3176dd7e750579b1bc150d4e6995fc8503b84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 15:09:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/97/653097/1/gate/openstack-tox-pep8/3d30873/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/97/653097/1/gate/openstack-tox-py27/9ba0114/ : SUCCESS in 4m 43s\n- openstack-tox-py35 http://logs.openstack.org/97/653097/1/gate/openstack-tox-py35/98faf24/ : SUCCESS in 4m 46s\n- openstack-tox-validate http://logs.openstack.org/97/653097/1/gate/openstack-tox-validate/d31d532/ : SUCCESS in 4m 31s\n- requirements-tox-babel http://logs.openstack.org/97/653097/1/gate/requirements-tox-babel/51e573c/ : SUCCESS in 4m 14s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/653097/1/gate/requirements-tox-py27-check-uc/ab8834e/ : SUCCESS in 8m 34s\n- cross-cinder-py27 http://logs.openstack.org/97/653097/1/gate/cross-cinder-py27/b4b0929/ : SUCCESS in 8m 55s\n- cross-glance-py27 http://logs.openstack.org/97/653097/1/gate/cross-glance-py27/c6a5210/ : SUCCESS in 9m 25s\n- cross-glance-py35 http://logs.openstack.org/97/653097/1/gate/cross-glance-py35/c50081c/ : SUCCESS in 10m 05s\n- cross-horizon-py27 http://logs.openstack.org/97/653097/1/gate/cross-horizon-py27/dcd9740/ : SUCCESS in 11m 42s\n- cross-horizon-py35 http://logs.openstack.org/97/653097/1/gate/cross-horizon-py35/6b662aa/ : SUCCESS in 10m 51s\n- cross-keystone-py27 http://logs.openstack.org/97/653097/1/gate/cross-keystone-py27/03969be/ : SUCCESS in 12m 35s\n- cross-keystone-py35 http://logs.openstack.org/97/653097/1/gate/cross-keystone-py35/6eb8ecb/ : SUCCESS in 13m 27s\n- cross-neutron-py27 http://logs.openstack.org/97/653097/1/gate/cross-neutron-py27/c6e8c55/ : SUCCESS in 26m 15s\n- cross-neutron-py35 http://logs.openstack.org/97/653097/1/gate/cross-neutron-py35/4eabea2/ : SUCCESS in 27m 25s\n- cross-nova-py27 http://logs.openstack.org/97/653097/1/gate/cross-nova-py27/3eb018b/ : SUCCESS in 14m 26s\n- cross-nova-py35 http://logs.openstack.org/97/653097/1/gate/cross-nova-py35/b61406e/ : SUCCESS in 12m 35s\n- cross-nova-functional http://logs.openstack.org/97/653097/1/gate/cross-nova-functional/2c885f4/ : FAILURE in 24m 20s\n- cross-swift-py27 http://logs.openstack.org/97/653097/1/gate/cross-swift-py27/f722ba8/ : SUCCESS in 8m 37s\n- cross-swift-py35 http://logs.openstack.org/97/653097/1/gate/cross-swift-py35/c069fbf/ : SUCCESS in 5m 14s\n- requirements-integration http://logs.openstack.org/97/653097/1/gate/requirements-integration/48bd7e6/ : SUCCESS in 49m 56s\n- neutron-grenade-multinode http://logs.openstack.org/97/653097/1/gate/neutron-grenade-multinode/cbe4bfe/ : SUCCESS in 1h 04m 04s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/97/653097/1/gate/legacy-tempest-dsvm-neutron-full/f619b18/ : FAILURE in 1h 20m 41s\n- neutron-grenade http://logs.openstack.org/97/653097/1/gate/neutron-grenade/73d647d/ : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":1},{"id":"630ac0f28d85d2a69e80e94340b5e8cb93031554","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-17 18:40:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7cf8fc55a6209f09a35ce8b5b89585e81ed7310e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 02:26:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/97/653097/1/check/openstack-tox-pypy/13e9395/ : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/653097/1/check/openstack-tox-pep8/ae21254/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/97/653097/1/check/openstack-tox-py27/9ad6cea/ : SUCCESS in 4m 06s\n- openstack-tox-py35 http://logs.openstack.org/97/653097/1/check/openstack-tox-py35/3a9d277/ : SUCCESS in 5m 53s\n- openstack-tox-validate http://logs.openstack.org/97/653097/1/check/openstack-tox-validate/8d9fb26/ : SUCCESS in 2m 57s\n- requirements-tox-babel http://logs.openstack.org/97/653097/1/check/requirements-tox-babel/3311c4e/ : SUCCESS in 3m 51s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/653097/1/check/requirements-tox-py27-check-uc/509d923/ : SUCCESS in 5m 57s\n- cross-cinder-py27 http://logs.openstack.org/97/653097/1/check/cross-cinder-py27/395df64/ : SUCCESS in 9m 17s\n- cross-cinder-py35 http://logs.openstack.org/97/653097/1/check/cross-cinder-py35/a580366/ : SUCCESS in 7m 47s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/97/653097/1/check/cross-glance-py27/68d99b0/ : SUCCESS in 9m 41s\n- cross-glance-py35 http://logs.openstack.org/97/653097/1/check/cross-glance-py35/5ab504a/ : SUCCESS in 9m 06s\n- cross-horizon-py27 http://logs.openstack.org/97/653097/1/check/cross-horizon-py27/8017caf/ : SUCCESS in 10m 55s\n- cross-horizon-py35 http://logs.openstack.org/97/653097/1/check/cross-horizon-py35/ef38eb8/ : SUCCESS in 10m 27s\n- cross-keystone-py27 http://logs.openstack.org/97/653097/1/check/cross-keystone-py27/6666c57/ : SUCCESS in 12m 21s\n- cross-keystone-py35 http://logs.openstack.org/97/653097/1/check/cross-keystone-py35/eebd43e/ : SUCCESS in 12m 10s\n- cross-neutron-py27 http://logs.openstack.org/97/653097/1/check/cross-neutron-py27/c2d3679/ : SUCCESS in 20m 44s\n- cross-neutron-py35 http://logs.openstack.org/97/653097/1/check/cross-neutron-py35/979298b/ : SUCCESS in 21m 36s\n- cross-nova-py27 http://logs.openstack.org/97/653097/1/check/cross-nova-py27/b4e53d3/ : SUCCESS in 13m 07s\n- cross-nova-py35 http://logs.openstack.org/97/653097/1/check/cross-nova-py35/cbbc8d5/ : SUCCESS in 14m 21s\n- cross-nova-functional http://logs.openstack.org/97/653097/1/check/cross-nova-functional/911c685/ : SUCCESS in 19m 33s\n- cross-swift-py27 http://logs.openstack.org/97/653097/1/check/cross-swift-py27/d5a01c9/ : SUCCESS in 8m 19s\n- cross-swift-py35 http://logs.openstack.org/97/653097/1/check/cross-swift-py35/0cc2cd4/ : SUCCESS in 4m 33s\n- requirements-integration http://logs.openstack.org/97/653097/1/check/requirements-integration/5e3c2cf/ : SUCCESS in 50m 24s\n- neutron-grenade-multinode http://logs.openstack.org/97/653097/1/check/neutron-grenade-multinode/3b37556/ : SUCCESS in 1h 01m 04s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/97/653097/1/check/legacy-tempest-dsvm-neutron-full/fc5aad3/ : SUCCESS in 1h 32m 58s\n- neutron-grenade http://logs.openstack.org/97/653097/1/check/neutron-grenade/5d18a37/ : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":1},{"id":"77f1d2c31a98f4cca65ab141eb60e44c7e6e0a9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 02:27:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de33fce043d6ca01567798b51eae54e7f0ec91e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 14:57:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/97/653097/1/gate/openstack-tox-pep8/215c1de/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/97/653097/1/gate/openstack-tox-py27/dee1263/ : SUCCESS in 5m 43s\n- openstack-tox-py35 http://logs.openstack.org/97/653097/1/gate/openstack-tox-py35/5932b9e/ : SUCCESS in 5m 03s\n- openstack-tox-validate http://logs.openstack.org/97/653097/1/gate/openstack-tox-validate/6b8b293/ : SUCCESS in 4m 50s\n- requirements-tox-babel http://logs.openstack.org/97/653097/1/gate/requirements-tox-babel/96bf3a4/ : SUCCESS in 4m 45s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/653097/1/gate/requirements-tox-py27-check-uc/507b54b/ : SUCCESS in 9m 57s\n- cross-cinder-py27 http://logs.openstack.org/97/653097/1/gate/cross-cinder-py27/62c5fcb/ : SUCCESS in 10m 22s\n- cross-glance-py27 http://logs.openstack.org/97/653097/1/gate/cross-glance-py27/caf4957/ : SUCCESS in 10m 05s\n- cross-glance-py35 http://logs.openstack.org/97/653097/1/gate/cross-glance-py35/ed2d40a/ : SUCCESS in 10m 23s\n- cross-horizon-py27 http://logs.openstack.org/97/653097/1/gate/cross-horizon-py27/21e6fc2/ : SUCCESS in 11m 23s\n- cross-horizon-py35 http://logs.openstack.org/97/653097/1/gate/cross-horizon-py35/98577d4/ : SUCCESS in 15m 22s\n- cross-keystone-py27 http://logs.openstack.org/97/653097/1/gate/cross-keystone-py27/edad37d/ : SUCCESS in 12m 19s\n- cross-keystone-py35 http://logs.openstack.org/97/653097/1/gate/cross-keystone-py35/2168019/ : SUCCESS in 12m 49s\n- cross-neutron-py27 http://logs.openstack.org/97/653097/1/gate/cross-neutron-py27/d37880a/ : SUCCESS in 24m 49s\n- cross-neutron-py35 http://logs.openstack.org/97/653097/1/gate/cross-neutron-py35/b45ae49/ : SUCCESS in 31m 00s\n- cross-nova-py27 http://logs.openstack.org/97/653097/1/gate/cross-nova-py27/e218802/ : SUCCESS in 13m 19s\n- cross-nova-py35 http://logs.openstack.org/97/653097/1/gate/cross-nova-py35/12186e3/ : SUCCESS in 13m 52s\n- cross-nova-functional http://logs.openstack.org/97/653097/1/gate/cross-nova-functional/d227b6a/ : SUCCESS in 19m 57s\n- cross-swift-py27 http://logs.openstack.org/97/653097/1/gate/cross-swift-py27/696c720/ : SUCCESS in 11m 55s\n- cross-swift-py35 http://logs.openstack.org/97/653097/1/gate/cross-swift-py35/86dcec1/ : SUCCESS in 5m 11s\n- requirements-integration http://logs.openstack.org/97/653097/1/gate/requirements-integration/67ecb43/ : SUCCESS in 50m 21s\n- neutron-grenade-multinode http://logs.openstack.org/97/653097/1/gate/neutron-grenade-multinode/24de216/ : SUCCESS in 56m 06s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/97/653097/1/gate/legacy-tempest-dsvm-neutron-full/757ce6d/ : FAILURE in 1h 13m 23s\n- neutron-grenade http://logs.openstack.org/97/653097/1/gate/neutron-grenade/19bbf0e/ : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6f0347d6d45bb8c3311d25663d2d28312a0658ac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-18 16:16:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6c681ddfb6322fbf8ff0dbb9cf2597a95b4beac3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 06:41:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/97/653097/1/check/openstack-tox-pypy/5e85853/ : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/653097/1/check/openstack-tox-pep8/d21166a/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/97/653097/1/check/openstack-tox-py27/5d0762d/ : SUCCESS in 3m 54s\n- openstack-tox-py35 http://logs.openstack.org/97/653097/1/check/openstack-tox-py35/d14e08d/ : SUCCESS in 3m 47s\n- openstack-tox-validate http://logs.openstack.org/97/653097/1/check/openstack-tox-validate/55bf3b0/ : SUCCESS in 3m 26s\n- requirements-tox-babel http://logs.openstack.org/97/653097/1/check/requirements-tox-babel/b5d1031/ : SUCCESS in 4m 33s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/653097/1/check/requirements-tox-py27-check-uc/b971612/ : SUCCESS in 6m 04s\n- cross-cinder-py27 http://logs.openstack.org/97/653097/1/check/cross-cinder-py27/28d233c/ : SUCCESS in 10m 19s\n- cross-cinder-py35 http://logs.openstack.org/97/653097/1/check/cross-cinder-py35/39d1335/ : SUCCESS in 9m 49s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/97/653097/1/check/cross-glance-py27/e78d2ae/ : SUCCESS in 10m 53s\n- cross-glance-py35 http://logs.openstack.org/97/653097/1/check/cross-glance-py35/955273a/ : SUCCESS in 10m 27s\n- cross-horizon-py27 http://logs.openstack.org/97/653097/1/check/cross-horizon-py27/a05b554/ : SUCCESS in 10m 55s\n- cross-horizon-py35 http://logs.openstack.org/97/653097/1/check/cross-horizon-py35/046bb0e/ : SUCCESS in 10m 40s\n- cross-keystone-py27 http://logs.openstack.org/97/653097/1/check/cross-keystone-py27/697c602/ : SUCCESS in 10m 59s\n- cross-keystone-py35 http://logs.openstack.org/97/653097/1/check/cross-keystone-py35/ddd4349/ : SUCCESS in 11m 52s\n- cross-neutron-py27 http://logs.openstack.org/97/653097/1/check/cross-neutron-py27/46bf59e/ : SUCCESS in 24m 24s\n- cross-neutron-py35 http://logs.openstack.org/97/653097/1/check/cross-neutron-py35/7f0242f/ : SUCCESS in 23m 40s\n- cross-nova-py27 http://logs.openstack.org/97/653097/1/check/cross-nova-py27/d5a8780/ : SUCCESS in 12m 30s\n- cross-nova-py35 http://logs.openstack.org/97/653097/1/check/cross-nova-py35/9003eab/ : SUCCESS in 12m 41s\n- cross-nova-functional http://logs.openstack.org/97/653097/1/check/cross-nova-functional/f2ea211/ : SUCCESS in 19m 27s\n- cross-swift-py27 http://logs.openstack.org/97/653097/1/check/cross-swift-py27/85042e9/ : SUCCESS in 12m 53s\n- cross-swift-py35 http://logs.openstack.org/97/653097/1/check/cross-swift-py35/22d1471/ : SUCCESS in 5m 58s\n- requirements-integration http://logs.openstack.org/97/653097/1/check/requirements-integration/bad2378/ : SUCCESS in 59m 29s\n- neutron-grenade-multinode http://logs.openstack.org/97/653097/1/check/neutron-grenade-multinode/3d1bc1e/ : POST_FAILURE in 1h 35m 17s\n- legacy-tempest-dsvm-neutron-full finger://ze07.openstack.org/cdee440f6f7e4c7283bfe4ac017a7bc2 : RETRY_LIMIT in 42m 13s\n- neutron-grenade http://logs.openstack.org/97/653097/1/check/neutron-grenade/2857289/ : SUCCESS in 48m 58s","accounts_in_message":[],"_revision_number":1},{"id":"01b306fbe8b8a5adbf07a91848ac043189733d0a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-19 06:42:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce9649e0a1aa6973ec612204e4fc8ab2f77eb4cb","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-20 08:27:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"afc7d77687d5ae320c9e924635174b748660d02f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 08:55:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/97/653097/1/check/openstack-tox-pypy/26b4727/ : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/653097/1/check/openstack-tox-pep8/5f9f01c/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/97/653097/1/check/openstack-tox-py27/a9e4ed6/ : SUCCESS in 3m 47s\n- openstack-tox-py35 http://logs.openstack.org/97/653097/1/check/openstack-tox-py35/ae48a79/ : SUCCESS in 5m 06s\n- openstack-tox-validate http://logs.openstack.org/97/653097/1/check/openstack-tox-validate/9d4bb97/ : SUCCESS in 3m 13s\n- requirements-tox-babel http://logs.openstack.org/97/653097/1/check/requirements-tox-babel/c84136e/ : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/653097/1/check/requirements-tox-py27-check-uc/1031dc9/ : SUCCESS in 8m 31s\n- cross-cinder-py27 http://logs.openstack.org/97/653097/1/check/cross-cinder-py27/664a504/ : SUCCESS in 7m 53s\n- cross-cinder-py35 http://logs.openstack.org/97/653097/1/check/cross-cinder-py35/ee42f15/ : SUCCESS in 7m 46s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/97/653097/1/check/cross-glance-py27/c9d2fed/ : SUCCESS in 8m 54s\n- cross-glance-py35 http://logs.openstack.org/97/653097/1/check/cross-glance-py35/33fbe15/ : SUCCESS in 10m 08s\n- cross-horizon-py27 http://logs.openstack.org/97/653097/1/check/cross-horizon-py27/150c197/ : SUCCESS in 10m 39s\n- cross-horizon-py35 http://logs.openstack.org/97/653097/1/check/cross-horizon-py35/5ab3e8b/ : SUCCESS in 10m 26s\n- cross-keystone-py27 http://logs.openstack.org/97/653097/1/check/cross-keystone-py27/3b93e7b/ : SUCCESS in 12m 43s\n- cross-keystone-py35 http://logs.openstack.org/97/653097/1/check/cross-keystone-py35/1d76a34/ : SUCCESS in 11m 35s\n- cross-neutron-py27 http://logs.openstack.org/97/653097/1/check/cross-neutron-py27/62a305c/ : SUCCESS in 22m 02s\n- cross-neutron-py35 http://logs.openstack.org/97/653097/1/check/cross-neutron-py35/c1c996c/ : SUCCESS in 23m 07s\n- cross-nova-py27 http://logs.openstack.org/97/653097/1/check/cross-nova-py27/96d713c/ : SUCCESS in 12m 19s\n- cross-nova-py35 http://logs.openstack.org/97/653097/1/check/cross-nova-py35/43ecfdc/ : SUCCESS in 11m 45s\n- cross-nova-functional http://logs.openstack.org/97/653097/1/check/cross-nova-functional/b007e5e/ : SUCCESS in 20m 32s\n- cross-swift-py27 http://logs.openstack.org/97/653097/1/check/cross-swift-py27/49a2bf4/ : SUCCESS in 8m 44s\n- cross-swift-py35 http://logs.openstack.org/97/653097/1/check/cross-swift-py35/f147c40/ : SUCCESS in 3m 33s\n- requirements-integration http://logs.openstack.org/97/653097/1/check/requirements-integration/41cbafd/ : FAILURE in 11m 50s\n- neutron-grenade-multinode http://logs.openstack.org/97/653097/1/check/neutron-grenade-multinode/e54284c/ : FAILURE in 9m 54s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/97/653097/1/check/legacy-tempest-dsvm-neutron-full/3cfc8bc/ : FAILURE in 10m 36s\n- neutron-grenade http://logs.openstack.org/97/653097/1/check/neutron-grenade/d982798/ : FAILURE in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c277a68ce71d7ad8059178d273f52541e7c328ba","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 14:55:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"131a1b92afdc5c526738fd5bc211dcec3370f98e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 16:19:07.000000000","message":"Patch Set 1: -Code-Review\n\nrecheck (and unvote til gate works again)","accounts_in_message":[],"_revision_number":1},{"id":"380f2a1d8fb781431192358187b5992c6713d285","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:04:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/97/653097/1/check/openstack-tox-pypy/9ed45ab/ : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/653097/1/check/openstack-tox-pep8/7963ed2/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/97/653097/1/check/openstack-tox-py27/2ff33ab/ : SUCCESS in 4m 57s\n- openstack-tox-py35 http://logs.openstack.org/97/653097/1/check/openstack-tox-py35/203b9ce/ : SUCCESS in 4m 11s\n- openstack-tox-validate http://logs.openstack.org/97/653097/1/check/openstack-tox-validate/ac39ef4/ : SUCCESS in 4m 15s\n- requirements-tox-babel http://logs.openstack.org/97/653097/1/check/requirements-tox-babel/08fe4eb/ : SUCCESS in 4m 02s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/653097/1/check/requirements-tox-py27-check-uc/9370113/ : SUCCESS in 6m 04s\n- cross-cinder-py27 http://logs.openstack.org/97/653097/1/check/cross-cinder-py27/2469671/ : SUCCESS in 7m 41s\n- cross-cinder-py35 http://logs.openstack.org/97/653097/1/check/cross-cinder-py35/59e3e65/ : SUCCESS in 8m 00s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/97/653097/1/check/cross-glance-py27/8b45b00/ : SUCCESS in 9m 24s\n- cross-glance-py35 http://logs.openstack.org/97/653097/1/check/cross-glance-py35/9fa9a95/ : SUCCESS in 9m 52s\n- cross-horizon-py27 http://logs.openstack.org/97/653097/1/check/cross-horizon-py27/457e8d2/ : SUCCESS in 11m 27s\n- cross-horizon-py35 http://logs.openstack.org/97/653097/1/check/cross-horizon-py35/c6d728e/ : SUCCESS in 10m 11s\n- cross-keystone-py27 http://logs.openstack.org/97/653097/1/check/cross-keystone-py27/a70c4a8/ : TIMED_OUT in 40m 55s\n- cross-keystone-py35 http://logs.openstack.org/97/653097/1/check/cross-keystone-py35/e8ebb66/ : SUCCESS in 11m 11s\n- cross-neutron-py27 http://logs.openstack.org/97/653097/1/check/cross-neutron-py27/8c33a0c/ : SUCCESS in 27m 27s\n- cross-neutron-py35 http://logs.openstack.org/97/653097/1/check/cross-neutron-py35/7219312/ : SUCCESS in 23m 49s\n- cross-nova-py27 http://logs.openstack.org/97/653097/1/check/cross-nova-py27/0f6f76b/ : SUCCESS in 13m 08s\n- cross-nova-py35 http://logs.openstack.org/97/653097/1/check/cross-nova-py35/8a9193e/ : SUCCESS in 18m 34s\n- cross-nova-functional http://logs.openstack.org/97/653097/1/check/cross-nova-functional/201b202/ : SUCCESS in 23m 13s\n- cross-swift-py27 http://logs.openstack.org/97/653097/1/check/cross-swift-py27/6777e7a/ : SUCCESS in 6m 57s\n- cross-swift-py35 http://logs.openstack.org/97/653097/1/check/cross-swift-py35/8ef7d0d/ : SUCCESS in 4m 53s\n- requirements-integration http://logs.openstack.org/97/653097/1/check/requirements-integration/941d930/ : SUCCESS in 52m 28s\n- neutron-grenade-multinode http://logs.openstack.org/97/653097/1/check/neutron-grenade-multinode/c43b5d5/ : SUCCESS in 1h 06m 39s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/97/653097/1/check/legacy-tempest-dsvm-neutron-full/7bb92e0/ : SUCCESS in 1h 32m 35s\n- neutron-grenade http://logs.openstack.org/97/653097/1/check/neutron-grenade/d511438/ : SUCCESS in 56m 33s","accounts_in_message":[],"_revision_number":1},{"id":"791a504c54a74096dc7b8a556e959f9bcf23ff5a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-23 14:43:13.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9e0f3936ad1a983ff39782621db7773a84de106b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 02:53:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/97/653097/1/check/openstack-tox-pypy/2d2c211/ : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/653097/1/check/openstack-tox-pep8/c14b2c3/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/97/653097/1/check/openstack-tox-py27/ee77cd8/ : SUCCESS in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/97/653097/1/check/openstack-tox-py35/c68043a/ : SUCCESS in 3m 49s\n- openstack-tox-validate http://logs.openstack.org/97/653097/1/check/openstack-tox-validate/bee67d7/ : SUCCESS in 3m 12s\n- requirements-tox-babel http://logs.openstack.org/97/653097/1/check/requirements-tox-babel/4e22efd/ : SUCCESS in 3m 41s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/653097/1/check/requirements-tox-py27-check-uc/f77a482/ : SUCCESS in 6m 31s\n- cross-cinder-py27 http://logs.openstack.org/97/653097/1/check/cross-cinder-py27/73eda72/ : SUCCESS in 7m 27s\n- cross-cinder-py35 http://logs.openstack.org/97/653097/1/check/cross-cinder-py35/c37c555/ : SUCCESS in 7m 31s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/97/653097/1/check/cross-glance-py27/c3b4a86/ : SUCCESS in 8m 18s\n- cross-glance-py35 http://logs.openstack.org/97/653097/1/check/cross-glance-py35/d3535c8/ : SUCCESS in 8m 58s\n- cross-horizon-py27 http://logs.openstack.org/97/653097/1/check/cross-horizon-py27/65021fb/ : SUCCESS in 10m 15s\n- cross-horizon-py35 http://logs.openstack.org/97/653097/1/check/cross-horizon-py35/843fa71/ : SUCCESS in 9m 52s\n- cross-keystone-py27 http://logs.openstack.org/97/653097/1/check/cross-keystone-py27/d0defaa/ : SUCCESS in 12m 25s\n- cross-keystone-py35 http://logs.openstack.org/97/653097/1/check/cross-keystone-py35/ebe24f2/ : SUCCESS in 15m 42s\n- cross-neutron-py27 http://logs.openstack.org/97/653097/1/check/cross-neutron-py27/2964edb/ : SUCCESS in 22m 30s\n- cross-neutron-py35 http://logs.openstack.org/97/653097/1/check/cross-neutron-py35/be0ddb7/ : SUCCESS in 20m 55s\n- cross-nova-py27 http://logs.openstack.org/97/653097/1/check/cross-nova-py27/9a95074/ : SUCCESS in 13m 53s\n- cross-nova-py35 http://logs.openstack.org/97/653097/1/check/cross-nova-py35/f55310d/ : SUCCESS in 11m 56s\n- cross-nova-functional http://logs.openstack.org/97/653097/1/check/cross-nova-functional/5dfeaeb/ : SUCCESS in 20m 00s\n- cross-swift-py27 http://logs.openstack.org/97/653097/1/check/cross-swift-py27/a78c831/ : SUCCESS in 7m 43s\n- cross-swift-py35 http://logs.openstack.org/97/653097/1/check/cross-swift-py35/3dbbb07/ : SUCCESS in 4m 14s\n- requirements-integration http://logs.openstack.org/97/653097/1/check/requirements-integration/00f0c6c/ : SUCCESS in 52m 19s\n- neutron-grenade-multinode http://logs.openstack.org/97/653097/1/check/neutron-grenade-multinode/917747a/ : SUCCESS in 56m 44s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/97/653097/1/check/legacy-tempest-dsvm-neutron-full/5eabe48/ : SUCCESS in 1h 23m 29s\n- neutron-grenade http://logs.openstack.org/97/653097/1/check/neutron-grenade/f2eb6d0/ : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":1},{"id":"37b327c925f981ef94b5e79bc191f4ce2a6b1faa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 02:53:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1ccc9a855e0f2afc88758084f0c720a65f86b89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 04:54:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/97/653097/1/gate/openstack-tox-pep8/8d13f5e/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/97/653097/1/gate/openstack-tox-py27/d95812a/ : SUCCESS in 3m 59s\n- openstack-tox-py35 http://logs.openstack.org/97/653097/1/gate/openstack-tox-py35/a5ef5f3/ : SUCCESS in 4m 02s\n- openstack-tox-validate http://logs.openstack.org/97/653097/1/gate/openstack-tox-validate/992bd83/ : SUCCESS in 3m 16s\n- requirements-tox-babel http://logs.openstack.org/97/653097/1/gate/requirements-tox-babel/0e67765/ : SUCCESS in 2m 52s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/653097/1/gate/requirements-tox-py27-check-uc/10555cf/ : SUCCESS in 6m 08s\n- cross-cinder-py27 http://logs.openstack.org/97/653097/1/gate/cross-cinder-py27/7d76fcf/ : SUCCESS in 7m 13s\n- cross-glance-py27 http://logs.openstack.org/97/653097/1/gate/cross-glance-py27/4fe769f/ : SUCCESS in 9m 17s\n- cross-glance-py35 http://logs.openstack.org/97/653097/1/gate/cross-glance-py35/423e8ff/ : SUCCESS in 9m 07s\n- cross-horizon-py27 http://logs.openstack.org/97/653097/1/gate/cross-horizon-py27/791d1cb/ : SUCCESS in 10m 23s\n- cross-horizon-py35 http://logs.openstack.org/97/653097/1/gate/cross-horizon-py35/e4002d5/ : SUCCESS in 9m 47s\n- cross-keystone-py27 http://logs.openstack.org/97/653097/1/gate/cross-keystone-py27/e37d701/ : SUCCESS in 11m 53s\n- cross-keystone-py35 http://logs.openstack.org/97/653097/1/gate/cross-keystone-py35/df96f31/ : SUCCESS in 12m 00s\n- cross-neutron-py27 http://logs.openstack.org/97/653097/1/gate/cross-neutron-py27/d6d8915/ : SUCCESS in 20m 30s\n- cross-neutron-py35 http://logs.openstack.org/97/653097/1/gate/cross-neutron-py35/c5a2f7e/ : SUCCESS in 27m 03s\n- cross-nova-py27 http://logs.openstack.org/97/653097/1/gate/cross-nova-py27/8225f0f/ : SUCCESS in 13m 12s\n- cross-nova-py35 http://logs.openstack.org/97/653097/1/gate/cross-nova-py35/523ce40/ : SUCCESS in 12m 26s\n- cross-nova-functional http://logs.openstack.org/97/653097/1/gate/cross-nova-functional/ae6bddc/ : SUCCESS in 17m 12s\n- cross-swift-py27 http://logs.openstack.org/97/653097/1/gate/cross-swift-py27/8a3b913/ : SUCCESS in 7m 17s\n- cross-swift-py35 http://logs.openstack.org/97/653097/1/gate/cross-swift-py35/0963978/ : SUCCESS in 3m 34s\n- requirements-integration http://logs.openstack.org/97/653097/1/gate/requirements-integration/460c6ce/ : SUCCESS in 49m 16s\n- neutron-grenade-multinode http://logs.openstack.org/97/653097/1/gate/neutron-grenade-multinode/f5db247/ : SUCCESS in 57m 28s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/97/653097/1/gate/legacy-tempest-dsvm-neutron-full/0f62b18/ : SUCCESS in 1h 15m 04s\n- neutron-grenade http://logs.openstack.org/97/653097/1/gate/neutron-grenade/5b1023b/ : SUCCESS in 50m 55s","accounts_in_message":[],"_revision_number":1},{"id":"88cff7f5219a52cf3ddacc6f233f9ee87cb68df3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 04:54:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"231cd5b1b5459e5d8a94bec68c8b01e145405d71","revisions":{"231cd5b1b5459e5d8a94bec68c8b01e145405d71":{"kind":"REWORK","_number":1,"created":"2019-04-16 19:20:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/653097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/653097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/653097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/653097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/653097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/653097/1"}}},"commit":{"parents":[{"commit":"03eeccfc2edaaf995010da7a8c2b6b6188439b30","subject":"update constraint for os-brick to new release 1.15.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03eeccfc2edaaf995010da7a8c2b6b6188439b30"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-16 19:20:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-16 19:20:02.000000000","tz":0},"subject":"update constraint for oslo.i18n to new release 3.17.2","message":"update constraint for oslo.i18n to new release 3.17.2\n\nChange-Id: Icd21862779a60ae3ebcd45593eff38fa7515be40\nmeta:version: 3.17.2\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ifcb051db33bffbfe7c183b726e7c54c310c488e1\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/231cd5b1b5459e5d8a94bec68c8b01e145405d71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/231cd5b1b5459e5d8a94bec68c8b01e145405d71"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
