)]}'
{"id":"openstack%2Frequirements~653229","triplet_id":"openstack%2Frequirements~stable%2Fpike~I0eee64dca644214374d9dd93aedf7337a6f521b9","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"I0eee64dca644214374d9dd93aedf7337a6f521b9","subject":"update constraint for python-solumclient to new release 2.5.1","status":"MERGED","created":"2019-04-17 02:29:40.000000000","updated":"2019-04-25 12:04:05.000000000","submitted":"2019-04-25 12:04:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653229-1556193845508-9c933fe4","meta_rev_id":"2fbc0407aaff6c2565c409becb90480b08355192","_number":653229,"virtual_id_number":653229,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-04-25 12:04:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-22 19:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-17 05:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-04-22 19:52:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-17 05:33:38.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-22 19:52:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-25 12:04:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8938ddc9913763a9a3ef8e87b9feda24c7f1ef66","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-17 02:29:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27895409e88cebaf584d00be06ddfc19ec5c856e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 04:01:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/29/653229/1/check/openstack-tox-pypy/4a62e5a/ : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/check/openstack-tox-pep8/f1bb0a4/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/check/openstack-tox-py27/bd45cc5/ : SUCCESS in 3m 51s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/check/openstack-tox-py35/32e0fc6/ : SUCCESS in 3m 58s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/check/openstack-tox-validate/4b7ffb3/ : SUCCESS in 3m 25s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/check/requirements-tox-babel/5a3f34b/ : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/check/requirements-tox-py27-check-uc/bb48869/ : SUCCESS in 7m 03s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/check/cross-cinder-py27/72cf31b/ : SUCCESS in 8m 46s\n- cross-cinder-py35 http://logs.openstack.org/29/653229/1/check/cross-cinder-py35/2e4a093/ : SUCCESS in 8m 26s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/check/cross-glance-py27/c09f848/ : SUCCESS in 9m 51s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/check/cross-glance-py35/411f8eb/ : SUCCESS in 9m 41s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/check/cross-horizon-py27/83d490f/ : SUCCESS in 10m 34s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/check/cross-horizon-py35/621071e/ : SUCCESS in 10m 56s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/check/cross-keystone-py27/4993990/ : SUCCESS in 11m 15s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/check/cross-keystone-py35/615809a/ : SUCCESS in 10m 37s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/check/cross-neutron-py27/e0f10de/ : SUCCESS in 21m 12s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/check/cross-neutron-py35/6487c7f/ : SUCCESS in 24m 31s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/check/cross-nova-py27/2f7e633/ : SUCCESS in 11m 19s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/check/cross-nova-py35/161e61f/ : SUCCESS in 19m 27s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/check/cross-nova-functional/3a4ff90/ : SUCCESS in 21m 06s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/check/cross-swift-py27/41e03c1/ : SUCCESS in 7m 11s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/check/cross-swift-py35/2bf38d3/ : SUCCESS in 3m 31s\n- requirements-integration http://logs.openstack.org/29/653229/1/check/requirements-integration/603bd9e/ : SUCCESS in 50m 52s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/check/neutron-grenade-multinode/3c6500a/ : SUCCESS in 57m 56s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/check/legacy-tempest-dsvm-neutron-full/2ba99ac/ : SUCCESS in 1h 25m 00s\n- neutron-grenade http://logs.openstack.org/29/653229/1/check/neutron-grenade/0ea1c91/ : SUCCESS in 54m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e772c602c1be329170aaa3e4e9895add5c929e51","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-17 04:02:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0661e67316a89fa78ffb5720c6025e8cfcaee9c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-17 05:33:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"67ffb8f9a2b603adf8366c81bdfa774833e9da22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 05:33:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0989f1a6222ccced5ad963fab5153594a6734821","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 11:02:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/gate/openstack-tox-pep8/170c279/ : SUCCESS in 7m 00s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/gate/openstack-tox-py27/0367f64/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/gate/openstack-tox-py35/16bdfd7/ : SUCCESS in 6m 37s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/gate/openstack-tox-validate/60e1280/ : SUCCESS in 6m 02s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/gate/requirements-tox-babel/30c1991/ : SUCCESS in 7m 36s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/gate/requirements-tox-py27-check-uc/acee081/ : SUCCESS in 7m 49s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/gate/cross-cinder-py27/9c59e04/ : SUCCESS in 9m 16s\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/gate/cross-glance-py27/b5915ee/ : SUCCESS in 10m 12s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/gate/cross-glance-py35/06614d5/ : FAILURE in 15m 12s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/gate/cross-horizon-py27/0ac19df/ : SUCCESS in 14m 05s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/gate/cross-horizon-py35/386638f/ : SUCCESS in 13m 11s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/gate/cross-keystone-py27/c6d8c61/ : SUCCESS in 23m 52s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/gate/cross-keystone-py35/c11981f/ : SUCCESS in 15m 29s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/gate/cross-neutron-py27/c2b16db/ : SUCCESS in 23m 23s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/gate/cross-neutron-py35/1065110/ : SUCCESS in 26m 04s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/gate/cross-nova-py27/d1fee0b/ : SUCCESS in 15m 54s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/gate/cross-nova-py35/2faef33/ : SUCCESS in 14m 11s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/gate/cross-nova-functional/4ef3e6f/ : SUCCESS in 19m 52s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/gate/cross-swift-py27/305ae14/ : SUCCESS in 8m 07s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/gate/cross-swift-py35/f04b682/ : SUCCESS in 4m 00s\n- requirements-integration http://logs.openstack.org/29/653229/1/gate/requirements-integration/01e5e69/ : SUCCESS in 49m 20s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/gate/neutron-grenade-multinode/478badc/ : SUCCESS in 59m 11s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/gate/legacy-tempest-dsvm-neutron-full/d25aed6/ : SUCCESS in 1h 43m 18s\n- neutron-grenade http://logs.openstack.org/29/653229/1/gate/neutron-grenade/fb1c7c5/ : SUCCESS in 51m 10s","accounts_in_message":[],"_revision_number":1},{"id":"21e6ba06ebe33a2434c8909eaa3dbe9b56d04e49","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-17 18:40:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1de1c7435aaa129fd217110e1e0003b35d61e9e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 02:12:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/29/653229/1/check/openstack-tox-pypy/b0963a2/ : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/check/openstack-tox-pep8/376a087/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/check/openstack-tox-py27/caf5624/ : SUCCESS in 4m 10s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/check/openstack-tox-py35/820ad26/ : SUCCESS in 3m 40s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/check/openstack-tox-validate/93dc6e6/ : SUCCESS in 3m 22s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/check/requirements-tox-babel/fdbfb9b/ : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/check/requirements-tox-py27-check-uc/ed57662/ : SUCCESS in 6m 22s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/check/cross-cinder-py27/490c98d/ : SUCCESS in 7m 41s\n- cross-cinder-py35 http://logs.openstack.org/29/653229/1/check/cross-cinder-py35/6cacbf4/ : SUCCESS in 8m 21s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/check/cross-glance-py27/5e09a5b/ : SUCCESS in 9m 24s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/check/cross-glance-py35/4f9dcd7/ : SUCCESS in 8m 52s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/check/cross-horizon-py27/65fede9/ : SUCCESS in 10m 33s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/check/cross-horizon-py35/6916e87/ : SUCCESS in 9m 52s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/check/cross-keystone-py27/c9ca94e/ : SUCCESS in 12m 43s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/check/cross-keystone-py35/ceaed2f/ : SUCCESS in 15m 04s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/check/cross-neutron-py27/4d45c23/ : SUCCESS in 23m 54s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/check/cross-neutron-py35/5f3982a/ : SUCCESS in 22m 02s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/check/cross-nova-py27/1eec86e/ : SUCCESS in 12m 28s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/check/cross-nova-py35/c388f3c/ : SUCCESS in 15m 14s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/check/cross-nova-functional/933ceb5/ : FAILURE in 29m 42s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/check/cross-swift-py27/059a605/ : SUCCESS in 7m 25s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/check/cross-swift-py35/cd2a1a6/ : SUCCESS in 3m 04s\n- requirements-integration http://logs.openstack.org/29/653229/1/check/requirements-integration/82482e6/ : SUCCESS in 51m 23s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/check/neutron-grenade-multinode/420f695/ : SUCCESS in 1h 02m 40s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/check/legacy-tempest-dsvm-neutron-full/5779af3/ : SUCCESS in 1h 28m 51s\n- neutron-grenade http://logs.openstack.org/29/653229/1/check/neutron-grenade/f70bd70/ : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":1},{"id":"dee43a69ff4b2042f84caf6852f3baf5529a169f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-18 06:20:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1149cad938f41feddcf384d04b20e1dfafedf7df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 14:37:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/29/653229/1/check/openstack-tox-pypy/c84edd5/ : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/check/openstack-tox-pep8/744f9d5/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/check/openstack-tox-py27/f9a5369/ : SUCCESS in 5m 36s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/check/openstack-tox-py35/fc0da44/ : SUCCESS in 4m 30s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/check/openstack-tox-validate/3bcdac2/ : SUCCESS in 3m 48s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/check/requirements-tox-babel/92d4469/ : SUCCESS in 3m 38s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/check/requirements-tox-py27-check-uc/6ca3557/ : SUCCESS in 6m 00s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/check/cross-cinder-py27/58c4e52/ : SUCCESS in 10m 06s\n- cross-cinder-py35 http://logs.openstack.org/29/653229/1/check/cross-cinder-py35/4e5982f/ : SUCCESS in 7m 35s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/check/cross-glance-py27/35b4d4c/ : SUCCESS in 11m 45s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/check/cross-glance-py35/f135ed7/ : SUCCESS in 14m 15s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/check/cross-horizon-py27/fe0e8f8/ : SUCCESS in 9m 44s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/check/cross-horizon-py35/6fd1670/ : SUCCESS in 9m 16s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/check/cross-keystone-py27/6b21e6a/ : SUCCESS in 19m 42s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/check/cross-keystone-py35/f2a5403/ : SUCCESS in 20m 35s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/check/cross-neutron-py27/2863b09/ : SUCCESS in 24m 54s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/check/cross-neutron-py35/99fefe8/ : SUCCESS in 21m 09s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/check/cross-nova-py27/0d4dbd8/ : SUCCESS in 12m 59s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/check/cross-nova-py35/fb79679/ : SUCCESS in 12m 39s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/check/cross-nova-functional/85682e4/ : SUCCESS in 25m 24s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/check/cross-swift-py27/97cd9a9/ : SUCCESS in 8m 52s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/check/cross-swift-py35/ae91f92/ : SUCCESS in 3m 27s\n- requirements-integration http://logs.openstack.org/29/653229/1/check/requirements-integration/d7a9141/ : SUCCESS in 51m 29s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/check/neutron-grenade-multinode/13c2b85/ : SUCCESS in 59m 27s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/check/legacy-tempest-dsvm-neutron-full/c8d43cb/ : SUCCESS in 1h 32m 33s\n- neutron-grenade http://logs.openstack.org/29/653229/1/check/neutron-grenade/a731d6e/ : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7ecbebb475e813a3aee2bc6cb7ed497549f7ccab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 16:41:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78c0daa71dc9997767c5fb5cd53b53702eaa4711","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 19:45:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/gate/openstack-tox-pep8/d2bf339/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/gate/openstack-tox-py27/ae2abb5/ : SUCCESS in 5m 18s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/gate/openstack-tox-py35/46ae54a/ : SUCCESS in 5m 09s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/gate/openstack-tox-validate/5acd4a9/ : SUCCESS in 4m 35s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/gate/requirements-tox-babel/1183d80/ : SUCCESS in 4m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/gate/requirements-tox-py27-check-uc/c58a1bf/ : SUCCESS in 6m 35s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/gate/cross-cinder-py27/0d483c5/ : SUCCESS in 10m 16s\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/gate/cross-glance-py27/7fc4a7e/ : SUCCESS in 10m 10s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/gate/cross-glance-py35/86af154/ : SUCCESS in 9m 48s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/gate/cross-horizon-py27/a82cb21/ : SUCCESS in 10m 24s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/gate/cross-horizon-py35/17a7adb/ : SUCCESS in 9m 53s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/gate/cross-keystone-py27/0176fd2/ : SUCCESS in 13m 05s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/gate/cross-keystone-py35/9e64ac9/ : SUCCESS in 12m 20s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/gate/cross-neutron-py27/8440192/ : SUCCESS in 22m 09s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/gate/cross-neutron-py35/34c6965/ : SUCCESS in 21m 56s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/gate/cross-nova-py27/9d54378/ : SUCCESS in 12m 40s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/gate/cross-nova-py35/cc54129/ : SUCCESS in 13m 11s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/gate/cross-nova-functional/2e89ac6/ : SUCCESS in 18m 41s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/gate/cross-swift-py27/060347f/ : SUCCESS in 9m 15s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/gate/cross-swift-py35/9d73f77/ : SUCCESS in 6m 20s\n- requirements-integration http://logs.openstack.org/29/653229/1/gate/requirements-integration/e908f07/ : SUCCESS in 51m 46s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/gate/neutron-grenade-multinode/877fc6a/ : SUCCESS in 1h 04m 36s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/gate/legacy-tempest-dsvm-neutron-full/1716977/ : FAILURE in 1h 11m 25s\n- neutron-grenade http://logs.openstack.org/29/653229/1/gate/neutron-grenade/8528678/ : SUCCESS in 53m 24s","accounts_in_message":[],"_revision_number":1},{"id":"95bcc1f15c9fab46000bba2a16859dd8425348e0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-18 22:47:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"04fb08e24a7034da0dbe97af58964c3b860dd004","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 11:00:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/29/653229/1/check/openstack-tox-pypy/629b372/ : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/check/openstack-tox-pep8/fdd886c/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/check/openstack-tox-py27/9820e38/ : SUCCESS in 6m 19s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/check/openstack-tox-py35/11ddc2a/ : SUCCESS in 5m 54s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/check/openstack-tox-validate/f91c5d1/ : SUCCESS in 5m 23s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/check/requirements-tox-babel/5f31262/ : SUCCESS in 4m 41s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/check/requirements-tox-py27-check-uc/6f9b50a/ : SUCCESS in 7m 46s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/check/cross-cinder-py27/0135fd2/ : SUCCESS in 9m 37s\n- cross-cinder-py35 http://logs.openstack.org/29/653229/1/check/cross-cinder-py35/3b34dd4/ : SUCCESS in 10m 47s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/check/cross-glance-py27/a335ad2/ : SUCCESS in 11m 37s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/check/cross-glance-py35/31a4b24/ : SUCCESS in 10m 00s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/check/cross-horizon-py27/9705d1f/ : SUCCESS in 10m 27s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/check/cross-horizon-py35/83ef226/ : SUCCESS in 11m 02s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/check/cross-keystone-py27/e225f9a/ : SUCCESS in 17m 47s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/check/cross-keystone-py35/8c0e7e4/ : SUCCESS in 12m 47s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/check/cross-neutron-py27/812c21d/ : SUCCESS in 26m 49s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/check/cross-neutron-py35/66d570a/ : SUCCESS in 22m 48s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/check/cross-nova-py27/76e5476/ : SUCCESS in 14m 14s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/check/cross-nova-py35/45afff4/ : SUCCESS in 18m 03s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/check/cross-nova-functional/bbf3771/ : SUCCESS in 23m 03s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/check/cross-swift-py27/11b4bf4/ : SUCCESS in 9m 35s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/check/cross-swift-py35/6fad15f/ : SUCCESS in 5m 49s\n- requirements-integration http://logs.openstack.org/29/653229/1/check/requirements-integration/3cf49fa/ : SUCCESS in 51m 40s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/check/neutron-grenade-multinode/d40b890/ : SUCCESS in 58m 37s\n- legacy-tempest-dsvm-neutron-full finger://ze01.openstack.org/cfab0e7ef5d745a099e23ff27110de5c : RETRY_LIMIT in 1h 14m 00s\n- neutron-grenade http://logs.openstack.org/29/653229/1/check/neutron-grenade/bc550f5/ : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e2706f7f555b62937a3050cd572f9dc2cfed1e10","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-20 08:25:59.000000000","message":"Patch Set 1: -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"981b67621af86b588a3043a526aa2aecf9b1aa7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 08:51:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/29/653229/1/check/openstack-tox-pypy/6ef3b81/ : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/check/openstack-tox-pep8/adf9b67/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/check/openstack-tox-py27/aa35a0c/ : SUCCESS in 3m 42s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/check/openstack-tox-py35/b7e89a7/ : SUCCESS in 4m 25s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/check/openstack-tox-validate/e2bfe9d/ : SUCCESS in 3m 34s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/check/requirements-tox-babel/512702e/ : SUCCESS in 3m 16s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/check/requirements-tox-py27-check-uc/77ca14b/ : SUCCESS in 5m 58s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/check/cross-cinder-py27/6957aac/ : SUCCESS in 8m 38s\n- cross-cinder-py35 http://logs.openstack.org/29/653229/1/check/cross-cinder-py35/bcc3fd2/ : SUCCESS in 8m 08s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/check/cross-glance-py27/beb25b3/ : SUCCESS in 7m 58s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/check/cross-glance-py35/4a2e137/ : SUCCESS in 11m 35s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/check/cross-horizon-py27/759b6ea/ : SUCCESS in 10m 31s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/check/cross-horizon-py35/4277686/ : SUCCESS in 10m 17s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/check/cross-keystone-py27/24b8a72/ : SUCCESS in 11m 06s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/check/cross-keystone-py35/2aa7e05/ : SUCCESS in 10m 39s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/check/cross-neutron-py27/f753da9/ : SUCCESS in 20m 33s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/check/cross-neutron-py35/e80a92f/ : SUCCESS in 20m 41s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/check/cross-nova-py27/e8fb242/ : SUCCESS in 11m 34s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/check/cross-nova-py35/99b981b/ : SUCCESS in 16m 37s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/check/cross-nova-functional/4a4a620/ : SUCCESS in 18m 10s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/check/cross-swift-py27/b50f1ed/ : SUCCESS in 8m 36s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/check/cross-swift-py35/7ccbabe/ : SUCCESS in 3m 31s\n- requirements-integration http://logs.openstack.org/29/653229/1/check/requirements-integration/b465b15/ : FAILURE in 12m 28s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/check/neutron-grenade-multinode/6dc2ba1/ : FAILURE in 11m 11s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/check/legacy-tempest-dsvm-neutron-full/1502c2b/ : FAILURE in 10m 19s\n- neutron-grenade http://logs.openstack.org/29/653229/1/check/neutron-grenade/a6bd2e6/ : FAILURE in 9m 42s","accounts_in_message":[],"_revision_number":1},{"id":"06130ad901a7123f0d540d7669da513150f7d067","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 14:55:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"28e80662001de0a2f3d152099c024d2fa7125ae2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 16:18:41.000000000","message":"Patch Set 1: -Code-Review\n\nrecheck (and unvote til gate works again)","accounts_in_message":[],"_revision_number":1},{"id":"e05458b5e2b8dbbe8c0b0a03958d56464f389340","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 19:00:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/29/653229/1/check/openstack-tox-pypy/1e19e2c/ : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/check/openstack-tox-pep8/8174e19/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/check/openstack-tox-py27/f0d7391/ : SUCCESS in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/check/openstack-tox-py35/2f3d39c/ : SUCCESS in 3m 23s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/check/openstack-tox-validate/9f8cdce/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/check/requirements-tox-babel/c105c9f/ : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/check/requirements-tox-py27-check-uc/2d09541/ : SUCCESS in 6m 02s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/check/cross-cinder-py27/3064934/ : SUCCESS in 8m 04s\n- cross-cinder-py35 http://logs.openstack.org/29/653229/1/check/cross-cinder-py35/e5f4334/ : SUCCESS in 8m 38s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/check/cross-glance-py27/08ed6b2/ : SUCCESS in 8m 36s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/check/cross-glance-py35/37a0a5e/ : SUCCESS in 9m 31s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/check/cross-horizon-py27/580d32d/ : SUCCESS in 10m 35s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/check/cross-horizon-py35/2326024/ : SUCCESS in 10m 16s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/check/cross-keystone-py27/d6abc30/ : SUCCESS in 12m 18s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/check/cross-keystone-py35/a2a6970/ : SUCCESS in 10m 50s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/check/cross-neutron-py27/843e9c0/ : SUCCESS in 22m 20s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/check/cross-neutron-py35/9a20733/ : SUCCESS in 19m 40s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/check/cross-nova-py27/8236fc5/ : SUCCESS in 12m 45s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/check/cross-nova-py35/b113fe7/ : SUCCESS in 12m 29s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/check/cross-nova-functional/1a53f93/ : SUCCESS in 18m 41s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/check/cross-swift-py27/5c94c38/ : SUCCESS in 7m 30s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/check/cross-swift-py35/0bf138a/ : SUCCESS in 3m 49s\n- requirements-integration http://logs.openstack.org/29/653229/1/check/requirements-integration/f98c3fa/ : SUCCESS in 50m 56s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/check/neutron-grenade-multinode/fb4da0d/ : SUCCESS in 1h 10m 27s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/check/legacy-tempest-dsvm-neutron-full/8deb8e3/ : SUCCESS in 1h 36m 58s\n- neutron-grenade http://logs.openstack.org/29/653229/1/check/neutron-grenade/485f2f1/ : SUCCESS in 52m 54s","accounts_in_message":[],"_revision_number":1},{"id":"cb02224b853fcc637106691f69ada0f796449d5e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 19:52:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9c7c7440af7e63aae3badfacb8e757641aa19ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 19:52:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcc5f9c1186121be1684b7f50a76ae99982b83cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:28:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/gate/openstack-tox-pep8/8c587f2/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/gate/openstack-tox-py27/659279b/ : SUCCESS in 3m 23s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/gate/openstack-tox-py35/fe54b26/ : SUCCESS in 3m 45s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/gate/openstack-tox-validate/4835007/ : SUCCESS in 5m 20s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/gate/requirements-tox-babel/f2d3cdf/ : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/gate/requirements-tox-py27-check-uc/98fa8ef/ : SUCCESS in 7m 23s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/gate/cross-cinder-py27/e12cd78/ : SUCCESS in 6m 52s\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/gate/cross-glance-py27/b527cf3/ : SUCCESS in 8m 37s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/gate/cross-glance-py35/57b6cc5/ : SUCCESS in 9m 38s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/gate/cross-horizon-py27/485bdd8/ : SUCCESS in 10m 40s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/gate/cross-horizon-py35/ea97a3f/ : SUCCESS in 10m 12s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/gate/cross-keystone-py27/472285a/ : TIMED_OUT in 40m 50s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/gate/cross-keystone-py35/da7e7fd/ : SUCCESS in 12m 00s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/gate/cross-neutron-py27/51fdfa6/ : SUCCESS in 22m 01s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/gate/cross-neutron-py35/38b3cc2/ : SUCCESS in 21m 29s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/gate/cross-nova-py27/c64d8ed/ : SUCCESS in 11m 55s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/gate/cross-nova-py35/cf08541/ : SUCCESS in 15m 37s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/gate/cross-nova-functional/5e855ad/ : SUCCESS in 21m 26s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/gate/cross-swift-py27/672819d/ : SUCCESS in 8m 09s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/gate/cross-swift-py35/cfd4783/ : SUCCESS in 3m 38s\n- requirements-integration http://logs.openstack.org/29/653229/1/gate/requirements-integration/3e31106/ : SUCCESS in 43m 02s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/gate/neutron-grenade-multinode/0c9d5a6/ : SUCCESS in 1h 02m 35s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/gate/legacy-tempest-dsvm-neutron-full/745b25a/ : SUCCESS in 1h 28m 54s\n- neutron-grenade http://logs.openstack.org/29/653229/1/gate/neutron-grenade/d258dd1/ : SUCCESS in 56m 03s","accounts_in_message":[],"_revision_number":1},{"id":"cba2add2ce39e40ae3c43bde629d43c4d6166800","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 23:30:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"84bf5d41a5fcfbb63932c97c4fb628b265a21147","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 03:01:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/29/653229/1/check/openstack-tox-pypy/9cb1c1a/ : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/check/openstack-tox-pep8/61721cd/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/check/openstack-tox-py27/b1e5a0b/ : SUCCESS in 3m 42s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/check/openstack-tox-py35/a5ead88/ : SUCCESS in 4m 09s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/check/openstack-tox-validate/294220e/ : SUCCESS in 3m 18s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/check/requirements-tox-babel/d9a4f54/ : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/check/requirements-tox-py27-check-uc/33496cd/ : SUCCESS in 5m 59s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/check/cross-cinder-py27/e612eaf/ : SUCCESS in 8m 09s\n- cross-cinder-py35 http://logs.openstack.org/29/653229/1/check/cross-cinder-py35/e289156/ : SUCCESS in 8m 18s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/check/cross-glance-py27/15fec9b/ : FAILURE in 9m 30s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/check/cross-glance-py35/ca5870a/ : SUCCESS in 9m 25s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/check/cross-horizon-py27/55f13fc/ : SUCCESS in 11m 38s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/check/cross-horizon-py35/97b5734/ : SUCCESS in 10m 06s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/check/cross-keystone-py27/1f0cae1/ : SUCCESS in 12m 21s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/check/cross-keystone-py35/b4185a9/ : SUCCESS in 21m 36s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/check/cross-neutron-py27/82f4804/ : SUCCESS in 24m 29s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/check/cross-neutron-py35/12f4dc3/ : SUCCESS in 26m 12s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/check/cross-nova-py27/65ae1af/ : SUCCESS in 18m 08s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/check/cross-nova-py35/3859ef4/ : SUCCESS in 19m 31s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/check/cross-nova-functional/4f1ea60/ : SUCCESS in 18m 18s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/check/cross-swift-py27/ba96022/ : SUCCESS in 7m 52s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/check/cross-swift-py35/47c67b6/ : SUCCESS in 3m 23s\n- requirements-integration http://logs.openstack.org/29/653229/1/check/requirements-integration/6de3cff/ : SUCCESS in 51m 31s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/check/neutron-grenade-multinode/a9cf5f8/ : SUCCESS in 1h 05m 12s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/check/legacy-tempest-dsvm-neutron-full/372850f/ : SUCCESS in 1h 22m 22s\n- neutron-grenade http://logs.openstack.org/29/653229/1/check/neutron-grenade/e3dad6b/ : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":1},{"id":"34caed8a778e98bdd2a54fd295bf7470b16b791a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-23 03:06:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"36febbc5db715f3d1aae46c8f93396121275f129","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 04:53:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/29/653229/1/check/openstack-tox-pypy/e868781/ : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/check/openstack-tox-pep8/92f9e61/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/check/openstack-tox-py27/e9ae5a9/ : SUCCESS in 5m 19s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/check/openstack-tox-py35/63befb1/ : SUCCESS in 3m 41s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/check/openstack-tox-validate/d1649bf/ : SUCCESS in 3m 48s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/check/requirements-tox-babel/86add48/ : SUCCESS in 3m 45s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/check/requirements-tox-py27-check-uc/82611c1/ : SUCCESS in 6m 10s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/check/cross-cinder-py27/78db017/ : SUCCESS in 8m 36s\n- cross-cinder-py35 http://logs.openstack.org/29/653229/1/check/cross-cinder-py35/4aadf46/ : SUCCESS in 7m 23s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/check/cross-glance-py27/66342c6/ : SUCCESS in 8m 55s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/check/cross-glance-py35/c566c09/ : SUCCESS in 11m 19s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/check/cross-horizon-py27/e21cf90/ : SUCCESS in 11m 14s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/check/cross-horizon-py35/cfc8683/ : SUCCESS in 9m 32s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/check/cross-keystone-py27/973b9e7/ : SUCCESS in 17m 14s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/check/cross-keystone-py35/6ebb91e/ : SUCCESS in 11m 16s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/check/cross-neutron-py27/9da2113/ : SUCCESS in 32m 39s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/check/cross-neutron-py35/e0bd62c/ : SUCCESS in 33m 35s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/check/cross-nova-py27/1240805/ : SUCCESS in 11m 46s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/check/cross-nova-py35/7145b96/ : SUCCESS in 14m 37s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/check/cross-nova-functional/8c86d0a/ : SUCCESS in 19m 37s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/check/cross-swift-py27/77e1c7a/ : SUCCESS in 7m 22s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/check/cross-swift-py35/437b3e6/ : SUCCESS in 3m 25s\n- requirements-integration http://logs.openstack.org/29/653229/1/check/requirements-integration/cb289c4/ : SUCCESS in 51m 20s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/check/neutron-grenade-multinode/eccd5a6/ : SUCCESS in 1h 00m 00s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/check/legacy-tempest-dsvm-neutron-full/f8f158d/ : SUCCESS in 1h 28m 16s\n- neutron-grenade http://logs.openstack.org/29/653229/1/check/neutron-grenade/fd962af/ : SUCCESS in 43m 31s","accounts_in_message":[],"_revision_number":1},{"id":"71af3a0a5ab2b43965c7f5468d507cbcbba0be1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 04:53:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"312daf1321609985caf95d9df423194f84949162","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 06:10:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/gate/openstack-tox-pep8/d7362e1/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/gate/openstack-tox-py27/583e5d4/ : SUCCESS in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/gate/openstack-tox-py35/0f5e6c5/ : SUCCESS in 3m 18s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/gate/openstack-tox-validate/7e94550/ : SUCCESS in 3m 38s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/gate/requirements-tox-babel/b2af5c0/ : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/gate/requirements-tox-py27-check-uc/d83441f/ : SUCCESS in 6m 00s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/gate/cross-cinder-py27/d162ed8/ : SUCCESS in 7m 51s\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/gate/cross-glance-py27/415dad4/ : SUCCESS in 9m 27s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/gate/cross-glance-py35/7fd74b5/ : SUCCESS in 9m 35s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/gate/cross-horizon-py27/9cdd3e9/ : SUCCESS in 10m 32s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/gate/cross-horizon-py35/086284c/ : SUCCESS in 9m 56s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/gate/cross-keystone-py27/147e007/ : SUCCESS in 10m 52s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/gate/cross-keystone-py35/473bbbb/ : SUCCESS in 11m 43s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/gate/cross-neutron-py27/bf58286/ : SUCCESS in 20m 59s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/gate/cross-neutron-py35/5e0e47c/ : SUCCESS in 20m 20s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/gate/cross-nova-py27/56e4ebe/ : SUCCESS in 11m 41s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/gate/cross-nova-py35/f7cc9d1/ : SUCCESS in 12m 05s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/gate/cross-nova-functional/3f14844/ : SUCCESS in 18m 06s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/gate/cross-swift-py27/8705b33/ : SUCCESS in 7m 07s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/gate/cross-swift-py35/b65bec1/ : SUCCESS in 3m 34s\n- requirements-integration http://logs.openstack.org/29/653229/1/gate/requirements-integration/8e3346f/ : SUCCESS in 51m 35s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/gate/neutron-grenade-multinode/b5bc1b9/ : SUCCESS in 58m 14s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/gate/legacy-tempest-dsvm-neutron-full/ea9b448/ : FAILURE in 1h 10m 12s\n- neutron-grenade http://logs.openstack.org/29/653229/1/gate/neutron-grenade/d2804b4/ : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a8eed78d321632ffc28292e35d09b37942c67a21","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-23 14:42:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"45210be3eb94cc592a2cebd5d13b9fc324da2082","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 02:44:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/29/653229/1/check/openstack-tox-pypy/ef43eee/ : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/check/openstack-tox-pep8/1919db2/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/check/openstack-tox-py27/d5febaa/ : SUCCESS in 3m 45s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/check/openstack-tox-py35/96cdb39/ : SUCCESS in 4m 05s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/check/openstack-tox-validate/0ea8b25/ : SUCCESS in 3m 11s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/check/requirements-tox-babel/3e204b4/ : SUCCESS in 3m 04s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/check/requirements-tox-py27-check-uc/bf9ae02/ : SUCCESS in 6m 01s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/check/cross-cinder-py27/6bdf767/ : SUCCESS in 7m 18s\n- cross-cinder-py35 http://logs.openstack.org/29/653229/1/check/cross-cinder-py35/9efbad1/ : SUCCESS in 7m 22s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/check/cross-glance-py27/e71c588/ : SUCCESS in 10m 03s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/check/cross-glance-py35/09553ee/ : SUCCESS in 10m 18s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/check/cross-horizon-py27/f48f066/ : SUCCESS in 11m 08s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/check/cross-horizon-py35/29db66c/ : SUCCESS in 10m 13s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/check/cross-keystone-py27/e13643f/ : SUCCESS in 18m 08s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/check/cross-keystone-py35/3cf936f/ : SUCCESS in 10m 46s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/check/cross-neutron-py27/bd105cc/ : SUCCESS in 20m 27s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/check/cross-neutron-py35/dc5a7fb/ : SUCCESS in 29m 11s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/check/cross-nova-py27/842ec5a/ : SUCCESS in 11m 43s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/check/cross-nova-py35/b4dae80/ : SUCCESS in 11m 46s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/check/cross-nova-functional/27c62c8/ : SUCCESS in 19m 13s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/check/cross-swift-py27/2cabc03/ : SUCCESS in 8m 17s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/check/cross-swift-py35/36f524b/ : SUCCESS in 2m 56s\n- requirements-integration http://logs.openstack.org/29/653229/1/check/requirements-integration/ce4cfff/ : SUCCESS in 55m 42s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/check/neutron-grenade-multinode/2fe3522/ : SUCCESS in 57m 31s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/check/legacy-tempest-dsvm-neutron-full/ec80fb8/ : SUCCESS in 1h 29m 29s\n- neutron-grenade http://logs.openstack.org/29/653229/1/check/neutron-grenade/a79bf45/ : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":1},{"id":"7167ea5b9214e6f8900364087f70ecdf8d67dfcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 02:45:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3befcf442e6a5ac08594d1660495b9ec60726ceb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 04:10:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/gate/openstack-tox-pep8/2a32faa/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/gate/openstack-tox-py27/9a0b927/ : SUCCESS in 3m 15s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/gate/openstack-tox-py35/85b70c4/ : SUCCESS in 3m 59s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/gate/openstack-tox-validate/e968d94/ : SUCCESS in 3m 23s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/gate/requirements-tox-babel/37a8f66/ : SUCCESS in 4m 02s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/gate/requirements-tox-py27-check-uc/7e2c902/ : SUCCESS in 6m 38s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/gate/cross-cinder-py27/0273516/ : SUCCESS in 7m 39s\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/gate/cross-glance-py27/8cb2cd8/ : SUCCESS in 8m 57s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/gate/cross-glance-py35/d12d353/ : SUCCESS in 8m 54s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/gate/cross-horizon-py27/1ce48ab/ : SUCCESS in 11m 38s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/gate/cross-horizon-py35/5609f5c/ : SUCCESS in 10m 07s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/gate/cross-keystone-py27/37b521d/ : SUCCESS in 11m 54s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/gate/cross-keystone-py35/18ec187/ : TIMED_OUT in 40m 54s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/gate/cross-neutron-py27/b1ea9c6/ : SUCCESS in 24m 08s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/gate/cross-neutron-py35/156bafc/ : SUCCESS in 18m 49s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/gate/cross-nova-py27/7aff654/ : SUCCESS in 12m 48s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/gate/cross-nova-py35/79e25f4/ : SUCCESS in 14m 00s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/gate/cross-nova-functional/7cf33fd/ : SUCCESS in 24m 18s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/gate/cross-swift-py27/34a705e/ : SUCCESS in 8m 18s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/gate/cross-swift-py35/89f8324/ : SUCCESS in 3m 52s\n- requirements-integration http://logs.openstack.org/29/653229/1/gate/requirements-integration/06d777b/ : SUCCESS in 54m 17s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/gate/neutron-grenade-multinode/70de4f9/ : SUCCESS in 54m 38s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/gate/legacy-tempest-dsvm-neutron-full/2f46db3/ : SUCCESS in 1h 21m 26s\n- neutron-grenade http://logs.openstack.org/29/653229/1/gate/neutron-grenade/e3c517a/ : SUCCESS in 51m 34s","accounts_in_message":[],"_revision_number":1},{"id":"883c338c7bff3a27124f6f3f04b6f10948971b76","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-24 15:18:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8736a5d79fe9d5472a52af662dc3592d55ef3486","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 02:32:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/29/653229/1/check/openstack-tox-pypy/cd1173c/ : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/check/openstack-tox-pep8/c511257/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/check/openstack-tox-py27/19b1ab3/ : SUCCESS in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/check/openstack-tox-py35/7da9295/ : SUCCESS in 4m 16s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/check/openstack-tox-validate/13d3817/ : SUCCESS in 3m 32s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/check/requirements-tox-babel/1a2c4dc/ : SUCCESS in 3m 51s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/check/requirements-tox-py27-check-uc/ee658a9/ : SUCCESS in 6m 11s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/check/cross-cinder-py27/f66bbef/ : SUCCESS in 8m 02s\n- cross-cinder-py35 http://logs.openstack.org/29/653229/1/check/cross-cinder-py35/3edc199/ : SUCCESS in 9m 00s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/check/cross-glance-py27/781942c/ : SUCCESS in 9m 12s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/check/cross-glance-py35/7eb8630/ : SUCCESS in 9m 34s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/check/cross-horizon-py27/f9dd74f/ : SUCCESS in 10m 11s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/check/cross-horizon-py35/80ce093/ : SUCCESS in 10m 31s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/check/cross-keystone-py27/882e389/ : SUCCESS in 14m 20s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/check/cross-keystone-py35/ebad46c/ : TIMED_OUT in 40m 46s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/check/cross-neutron-py27/6168d94/ : SUCCESS in 17m 59s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/check/cross-neutron-py35/809ca46/ : SUCCESS in 18m 35s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/check/cross-nova-py27/8d25113/ : SUCCESS in 11m 32s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/check/cross-nova-py35/0d4704e/ : SUCCESS in 11m 06s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/check/cross-nova-functional/977b888/ : SUCCESS in 19m 21s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/check/cross-swift-py27/46bce01/ : SUCCESS in 7m 40s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/check/cross-swift-py35/625759e/ : SUCCESS in 3m 24s\n- requirements-integration http://logs.openstack.org/29/653229/1/check/requirements-integration/30a6c0c/ : SUCCESS in 45m 56s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/check/neutron-grenade-multinode/5ab5374/ : SUCCESS in 1h 13m 15s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/check/legacy-tempest-dsvm-neutron-full/7d9232a/ : SUCCESS in 1h 17m 53s\n- neutron-grenade http://logs.openstack.org/29/653229/1/check/neutron-grenade/2423934/ : SUCCESS in 49m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6af22ddeca5b0b1e884330b60aeaf4be20229b61","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-25 02:32:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ab07338f70775f9f12d22d0316bcedf2056d7200","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 07:16:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/29/653229/1/check/openstack-tox-pypy/ef83040/ : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/check/openstack-tox-pep8/d40df36/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/check/openstack-tox-py27/bf976a9/ : SUCCESS in 3m 14s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/check/openstack-tox-py35/b5a267e/ : SUCCESS in 3m 55s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/check/openstack-tox-validate/6cfe79a/ : SUCCESS in 2m 54s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/check/requirements-tox-babel/30b8e18/ : SUCCESS in 4m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/check/requirements-tox-py27-check-uc/5ec27e9/ : SUCCESS in 7m 20s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/check/cross-cinder-py27/b2b9da9/ : SUCCESS in 7m 26s\n- cross-cinder-py35 http://logs.openstack.org/29/653229/1/check/cross-cinder-py35/494f1ba/ : SUCCESS in 7m 12s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/check/cross-glance-py27/be76da1/ : SUCCESS in 8m 50s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/check/cross-glance-py35/3ab2038/ : SUCCESS in 9m 34s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/check/cross-horizon-py27/46f1d5c/ : SUCCESS in 10m 29s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/check/cross-horizon-py35/78c5d05/ : SUCCESS in 10m 08s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/check/cross-keystone-py27/d168c30/ : SUCCESS in 11m 10s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/check/cross-keystone-py35/cc407b2/ : SUCCESS in 11m 14s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/check/cross-neutron-py27/64ad30a/ : SUCCESS in 20m 28s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/check/cross-neutron-py35/3139241/ : SUCCESS in 21m 29s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/check/cross-nova-py27/f97f745/ : SUCCESS in 12m 13s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/check/cross-nova-py35/ed84067/ : SUCCESS in 12m 34s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/check/cross-nova-functional/5900c41/ : SUCCESS in 18m 54s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/check/cross-swift-py27/3a2b30e/ : SUCCESS in 6m 37s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/check/cross-swift-py35/e37bcc8/ : SUCCESS in 3m 57s\n- requirements-integration http://logs.openstack.org/29/653229/1/check/requirements-integration/2c56b5d/ : SUCCESS in 50m 22s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/check/neutron-grenade-multinode/b82ab31/ : SUCCESS in 58m 35s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/check/legacy-tempest-dsvm-neutron-full/e5e19bc/ : SUCCESS in 1h 17m 44s\n- neutron-grenade http://logs.openstack.org/29/653229/1/check/neutron-grenade/e94f5f8/ : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":1},{"id":"87d415e2763897121354eb2e685e475a7fa655f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 07:17:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9265beea692488e653d63add71d7d3a68247f173","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 12:04:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"24e11dc61a47071f8f72f5b2c3845fa4890f948c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 12:04:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/29/653229/1/gate/openstack-tox-pep8/bb2347e/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/29/653229/1/gate/openstack-tox-py27/97a3df0/ : SUCCESS in 3m 50s\n- openstack-tox-py35 http://logs.openstack.org/29/653229/1/gate/openstack-tox-py35/1910614/ : SUCCESS in 4m 05s\n- openstack-tox-validate http://logs.openstack.org/29/653229/1/gate/openstack-tox-validate/84bc539/ : SUCCESS in 3m 29s\n- requirements-tox-babel http://logs.openstack.org/29/653229/1/gate/requirements-tox-babel/7c5af7c/ : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/653229/1/gate/requirements-tox-py27-check-uc/e543333/ : SUCCESS in 6m 47s\n- cross-cinder-py27 http://logs.openstack.org/29/653229/1/gate/cross-cinder-py27/eebf1c7/ : SUCCESS in 8m 06s\n- cross-glance-py27 http://logs.openstack.org/29/653229/1/gate/cross-glance-py27/6911395/ : SUCCESS in 11m 05s\n- cross-glance-py35 http://logs.openstack.org/29/653229/1/gate/cross-glance-py35/8a280c8/ : SUCCESS in 9m 43s\n- cross-horizon-py27 http://logs.openstack.org/29/653229/1/gate/cross-horizon-py27/70be679/ : SUCCESS in 10m 37s\n- cross-horizon-py35 http://logs.openstack.org/29/653229/1/gate/cross-horizon-py35/b4619e2/ : SUCCESS in 10m 45s\n- cross-keystone-py27 http://logs.openstack.org/29/653229/1/gate/cross-keystone-py27/0b5ee0d/ : SUCCESS in 12m 09s\n- cross-keystone-py35 http://logs.openstack.org/29/653229/1/gate/cross-keystone-py35/de0e911/ : SUCCESS in 32m 50s\n- cross-neutron-py27 http://logs.openstack.org/29/653229/1/gate/cross-neutron-py27/7518bbd/ : SUCCESS in 25m 46s\n- cross-neutron-py35 http://logs.openstack.org/29/653229/1/gate/cross-neutron-py35/e95c6e8/ : SUCCESS in 25m 23s\n- cross-nova-py27 http://logs.openstack.org/29/653229/1/gate/cross-nova-py27/ea67c90/ : SUCCESS in 12m 11s\n- cross-nova-py35 http://logs.openstack.org/29/653229/1/gate/cross-nova-py35/edbc48d/ : SUCCESS in 11m 56s\n- cross-nova-functional http://logs.openstack.org/29/653229/1/gate/cross-nova-functional/182d9d7/ : SUCCESS in 20m 54s\n- cross-swift-py27 http://logs.openstack.org/29/653229/1/gate/cross-swift-py27/14297dd/ : SUCCESS in 7m 28s\n- cross-swift-py35 http://logs.openstack.org/29/653229/1/gate/cross-swift-py35/e5061f2/ : SUCCESS in 3m 38s\n- requirements-integration http://logs.openstack.org/29/653229/1/gate/requirements-integration/2ae4c14/ : SUCCESS in 40m 18s\n- neutron-grenade-multinode http://logs.openstack.org/29/653229/1/gate/neutron-grenade-multinode/a264459/ : SUCCESS in 56m 54s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/29/653229/1/gate/legacy-tempest-dsvm-neutron-full/aa73153/ : SUCCESS in 1h 32m 58s\n- neutron-grenade http://logs.openstack.org/29/653229/1/gate/neutron-grenade/1e295bd/ : SUCCESS in 54m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59d6fc583d48c386d44613cd31a165fb38045d10","revisions":{"59d6fc583d48c386d44613cd31a165fb38045d10":{"kind":"REWORK","_number":1,"created":"2019-04-17 02:29:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/653229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/653229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/653229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/653229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/653229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/653229/1"}}},"commit":{"parents":[{"commit":"03eeccfc2edaaf995010da7a8c2b6b6188439b30","subject":"update constraint for os-brick to new release 1.15.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03eeccfc2edaaf995010da7a8c2b6b6188439b30"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 02:29:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 02:29:39.000000000","tz":0},"subject":"update constraint for python-solumclient to new release 2.5.1","message":"update constraint for python-solumclient to new release 2.5.1\n\nChange-Id: I0eee64dca644214374d9dd93aedf7337a6f521b9\nmeta:version: 2.5.1\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Change-Id: If626d49f5d23fe049b335a92cabfd789f89909e2\nmeta:release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59d6fc583d48c386d44613cd31a165fb38045d10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59d6fc583d48c386d44613cd31a165fb38045d10"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
