)]}'
{"id":"openstack%2Frequirements~653272","triplet_id":"openstack%2Frequirements~stable%2Fpike~I922465e910f57ffaf5a61563a56b026e3d5558ef","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"I922465e910f57ffaf5a61563a56b026e3d5558ef","subject":"update constraint for python-searchlightclient to new release 1.2.1","status":"MERGED","created":"2019-04-17 03:59:16.000000000","updated":"2019-04-27 18:16:49.000000000","submitted":"2019-04-27 18:16:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653272-1556389009952-3a5dd909","meta_rev_id":"0923b6fb8c1ba3a0ce838a53ba311470118cc979","_number":653272,"virtual_id_number":653272,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-04-20 08:27:26.000000000","_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-04-27 18:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-04-18 13:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-04-24 15:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-04-18 13:24:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-18 13:24:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-20 08:27:26.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-24 15:18:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-27 18:16:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c1ce6a92005abc152bf03b7b2bcd98c0a2cadfe","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-17 03:59:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3c1258b0e50993ea654dfa587f2ad2b3fff47a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 06:00:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/72/653272/1/check/openstack-tox-pypy/5a0b559/ : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/1/check/openstack-tox-pep8/238b807/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/1/check/openstack-tox-py27/fbeefcf/ : SUCCESS in 4m 21s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/1/check/openstack-tox-py35/35d0a07/ : SUCCESS in 4m 28s\n- openstack-tox-validate http://logs.openstack.org/72/653272/1/check/openstack-tox-validate/9728145/ : SUCCESS in 4m 39s\n- requirements-tox-babel http://logs.openstack.org/72/653272/1/check/requirements-tox-babel/f4edcf1/ : SUCCESS in 4m 26s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/1/check/requirements-tox-py27-check-uc/903207d/ : SUCCESS in 6m 53s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/1/check/cross-cinder-py27/5d0834e/ : SUCCESS in 8m 52s\n- cross-cinder-py35 http://logs.openstack.org/72/653272/1/check/cross-cinder-py35/4d98cd6/ : SUCCESS in 8m 29s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/72/653272/1/check/cross-glance-py27/4aeb0fd/ : SUCCESS in 12m 09s\n- cross-glance-py35 http://logs.openstack.org/72/653272/1/check/cross-glance-py35/6d36cec/ : SUCCESS in 10m 40s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/1/check/cross-horizon-py27/82a0581/ : SUCCESS in 12m 31s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/1/check/cross-horizon-py35/1e48ee5/ : SUCCESS in 9m 05s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/1/check/cross-keystone-py27/424cd76/ : SUCCESS in 11m 52s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/1/check/cross-keystone-py35/e157e30/ : SUCCESS in 22m 07s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/1/check/cross-neutron-py27/2e378c1/ : SUCCESS in 23m 33s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/1/check/cross-neutron-py35/9a35f70/ : SUCCESS in 19m 56s\n- cross-nova-py27 http://logs.openstack.org/72/653272/1/check/cross-nova-py27/99b374e/ : SUCCESS in 14m 45s\n- cross-nova-py35 http://logs.openstack.org/72/653272/1/check/cross-nova-py35/1ab2e0a/ : SUCCESS in 11m 45s\n- cross-nova-functional http://logs.openstack.org/72/653272/1/check/cross-nova-functional/77a0b3d/ : SUCCESS in 20m 15s\n- cross-swift-py27 http://logs.openstack.org/72/653272/1/check/cross-swift-py27/555bf65/ : SUCCESS in 7m 54s\n- cross-swift-py35 http://logs.openstack.org/72/653272/1/check/cross-swift-py35/bde9d50/ : SUCCESS in 5m 02s\n- requirements-integration http://logs.openstack.org/72/653272/1/check/requirements-integration/c5bcdb6/ : SUCCESS in 49m 35s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/1/check/neutron-grenade-multinode/3c90044/ : SUCCESS in 1h 00m 52s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/1/check/legacy-tempest-dsvm-neutron-full/d09ea00/ : FAILURE in 1h 37m 28s\n- neutron-grenade http://logs.openstack.org/72/653272/1/check/neutron-grenade/e67bbed/ : SUCCESS in 59m 12s","accounts_in_message":[],"_revision_number":1},{"id":"01dbac4d07dcff4a722e780e547be0224f0148d7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-17 06:16:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"beddefe6b188764b5a620cff340891cf7c857d02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 08:57:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/72/653272/1/check/openstack-tox-pypy/a4be098/ : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/1/check/openstack-tox-pep8/f343cb9/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/1/check/openstack-tox-py27/11c58d4/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/1/check/openstack-tox-py35/6e733d7/ : SUCCESS in 4m 37s\n- openstack-tox-validate http://logs.openstack.org/72/653272/1/check/openstack-tox-validate/337a56c/ : SUCCESS in 4m 10s\n- requirements-tox-babel http://logs.openstack.org/72/653272/1/check/requirements-tox-babel/c78abe4/ : SUCCESS in 4m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/1/check/requirements-tox-py27-check-uc/d92da24/ : SUCCESS in 6m 55s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/1/check/cross-cinder-py27/4458367/ : SUCCESS in 8m 48s\n- cross-cinder-py35 http://logs.openstack.org/72/653272/1/check/cross-cinder-py35/a6d52e0/ : SUCCESS in 11m 06s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/72/653272/1/check/cross-glance-py27/0cb7aaf/ : SUCCESS in 11m 33s\n- cross-glance-py35 http://logs.openstack.org/72/653272/1/check/cross-glance-py35/0214b8a/ : SUCCESS in 9m 01s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/1/check/cross-horizon-py27/f216f52/ : SUCCESS in 12m 10s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/1/check/cross-horizon-py35/52396ec/ : SUCCESS in 10m 13s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/1/check/cross-keystone-py27/286e966/ : SUCCESS in 12m 51s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/1/check/cross-keystone-py35/63372ad/ : SUCCESS in 13m 11s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/1/check/cross-neutron-py27/6737af4/ : SUCCESS in 23m 02s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/1/check/cross-neutron-py35/8fc35fe/ : SUCCESS in 23m 56s\n- cross-nova-py27 http://logs.openstack.org/72/653272/1/check/cross-nova-py27/1f96496/ : SUCCESS in 13m 16s\n- cross-nova-py35 http://logs.openstack.org/72/653272/1/check/cross-nova-py35/2063007/ : SUCCESS in 13m 41s\n- cross-nova-functional http://logs.openstack.org/72/653272/1/check/cross-nova-functional/e0834c9/ : SUCCESS in 19m 35s\n- cross-swift-py27 http://logs.openstack.org/72/653272/1/check/cross-swift-py27/2226db6/ : SUCCESS in 8m 05s\n- cross-swift-py35 http://logs.openstack.org/72/653272/1/check/cross-swift-py35/6bef15b/ : SUCCESS in 4m 12s\n- requirements-integration http://logs.openstack.org/72/653272/1/check/requirements-integration/26d02f1/ : SUCCESS in 54m 44s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/1/check/neutron-grenade-multinode/558186d/ : SUCCESS in 59m 20s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/1/check/legacy-tempest-dsvm-neutron-full/2811bff/ : SUCCESS in 1h 32m 39s\n- neutron-grenade http://logs.openstack.org/72/653272/1/check/neutron-grenade/2800556/ : SUCCESS in 52m 11s","accounts_in_message":[],"_revision_number":1},{"id":"11a7ec43e45f1821aa20311eea0780e544b7aed2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-17 11:33:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9fe5a3e33585b4fa9507ed9882c5b4061d2b7f4e","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-17 12:44:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64b5ba1b7410eb76704fed07da888e5ea5e0f244","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-17 18:41:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ddb06ca0a3259a74dc9f58a2828182306dc94b1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 19:23:26.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"30666dbf06f93b538573341733902e950e2468f2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-17 19:46:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"345c3bd466fe5da295c55fa689bdee681e702647","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 05:49:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/72/653272/2/check/openstack-tox-pypy/8a4a330/ : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/check/openstack-tox-pep8/21ce348/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/check/openstack-tox-py27/a1f19d9/ : SUCCESS in 4m 06s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/check/openstack-tox-py35/bedd2f9/ : SUCCESS in 3m 44s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/check/openstack-tox-validate/bd66e54/ : SUCCESS in 3m 29s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/check/requirements-tox-babel/28cb519/ : SUCCESS in 3m 46s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/check/requirements-tox-py27-check-uc/24a580b/ : SUCCESS in 7m 09s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/check/cross-cinder-py27/52ec7ed/ : SUCCESS in 9m 06s\n- cross-cinder-py35 http://logs.openstack.org/72/653272/2/check/cross-cinder-py35/7a70759/ : SUCCESS in 7m 39s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/check/cross-glance-py27/2dc6b48/ : SUCCESS in 8m 29s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/check/cross-glance-py35/b7f2069/ : SUCCESS in 10m 06s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/check/cross-horizon-py27/a88b4ac/ : SUCCESS in 10m 42s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/check/cross-horizon-py35/ec374b4/ : SUCCESS in 10m 36s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/check/cross-keystone-py27/2569fac/ : SUCCESS in 14m 11s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/check/cross-keystone-py35/7be7417/ : SUCCESS in 14m 51s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/check/cross-neutron-py27/900678f/ : SUCCESS in 22m 19s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/check/cross-neutron-py35/488f6ef/ : SUCCESS in 26m 36s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/check/cross-nova-py27/82643fe/ : SUCCESS in 15m 03s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/check/cross-nova-py35/70a3463/ : SUCCESS in 14m 02s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/check/cross-nova-functional/6549f67/ : SUCCESS in 20m 09s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/check/cross-swift-py27/b8d053c/ : SUCCESS in 7m 03s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/check/cross-swift-py35/e4131bc/ : SUCCESS in 4m 31s\n- requirements-integration http://logs.openstack.org/72/653272/2/check/requirements-integration/307b03b/ : SUCCESS in 52m 57s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/check/neutron-grenade-multinode/68f5676/ : SUCCESS in 1h 00m 44s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/2/check/legacy-tempest-dsvm-neutron-full/b457e10/ : SUCCESS in 1h 27m 57s\n- neutron-grenade http://logs.openstack.org/72/653272/2/check/neutron-grenade/a9f04fc/ : SUCCESS in 51m 09s","accounts_in_message":[],"_revision_number":2},{"id":"be6f68a7c8f45d744fc3e160a6f4bc908773e280","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-18 06:20:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b13a1035d4b01808d31e67d3110c768281d4396d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-18 13:24:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5e9a68eb4e3fd0163dfbe4dd9a27dce5188737ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 13:25:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0b2cb8c8dc14a8756cbf7b045e1cbd53179a0188","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 17:32:55.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/gate/openstack-tox-pep8/e57edd1/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/gate/openstack-tox-py27/336389c/ : SUCCESS in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/gate/openstack-tox-py35/649a60b/ : SUCCESS in 5m 23s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/gate/openstack-tox-validate/095cb3e/ : SUCCESS in 6m 54s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/gate/requirements-tox-babel/bb289fb/ : SUCCESS in 4m 29s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/gate/requirements-tox-py27-check-uc/4959ce5/ : SUCCESS in 7m 09s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/gate/cross-cinder-py27/27bd8a1/ : SUCCESS in 9m 52s\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/gate/cross-glance-py27/b72c532/ : SUCCESS in 10m 27s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/gate/cross-glance-py35/f77f2b3/ : SUCCESS in 11m 03s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/gate/cross-horizon-py27/b593b17/ : SUCCESS in 12m 50s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/gate/cross-horizon-py35/e6e5761/ : SUCCESS in 13m 41s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/gate/cross-keystone-py27/50f2952/ : SUCCESS in 30m 07s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/gate/cross-keystone-py35/911bc4a/ : SUCCESS in 12m 20s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/gate/cross-neutron-py27/338cc8d/ : SUCCESS in 23m 29s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/gate/cross-neutron-py35/2362ae2/ : SUCCESS in 26m 11s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/gate/cross-nova-py27/e34f469/ : SUCCESS in 14m 04s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/gate/cross-nova-py35/30ced70/ : SUCCESS in 15m 35s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/gate/cross-nova-functional/6362ce9/ : SUCCESS in 21m 39s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/gate/cross-swift-py27/ec6def3/ : SUCCESS in 10m 56s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/gate/cross-swift-py35/b6b32cd/ : SUCCESS in 4m 16s\n- requirements-integration http://logs.openstack.org/72/653272/2/gate/requirements-integration/266ae63/ : SUCCESS in 56m 27s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/gate/neutron-grenade-multinode/3f1508a/ : SUCCESS in 1h 03m 52s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/2/gate/legacy-tempest-dsvm-neutron-full/f5fb48c/ : FAILURE in 1h 35m 18s\n- neutron-grenade http://logs.openstack.org/72/653272/2/gate/neutron-grenade/128f6f1/ : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":2},{"id":"4434845e5cc04b69fdc6a8b25b1c73b2761ed379","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-18 22:47:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cea13a5a3f8ecf865347c91b314af4473052862f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 10:59:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/72/653272/2/check/openstack-tox-pypy/a212dcd/ : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/check/openstack-tox-pep8/f05d007/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/check/openstack-tox-py27/2cc1797/ : SUCCESS in 5m 42s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/check/openstack-tox-py35/4210b28/ : SUCCESS in 5m 24s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/check/openstack-tox-validate/278222f/ : SUCCESS in 5m 01s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/check/requirements-tox-babel/1a763fe/ : SUCCESS in 5m 58s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/check/requirements-tox-py27-check-uc/75e220d/ : SUCCESS in 8m 44s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/check/cross-cinder-py27/80b8e2b/ : SUCCESS in 12m 14s\n- cross-cinder-py35 http://logs.openstack.org/72/653272/2/check/cross-cinder-py35/b46bc32/ : SUCCESS in 10m 02s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/check/cross-glance-py27/2f7b044/ : SUCCESS in 11m 07s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/check/cross-glance-py35/02895dd/ : SUCCESS in 13m 30s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/check/cross-horizon-py27/ad80660/ : SUCCESS in 12m 24s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/check/cross-horizon-py35/3b75bb8/ : SUCCESS in 11m 43s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/check/cross-keystone-py27/58dab9e/ : SUCCESS in 12m 45s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/check/cross-keystone-py35/b89ea45/ : SUCCESS in 12m 37s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/check/cross-neutron-py27/9e4e6c4/ : SUCCESS in 22m 51s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/check/cross-neutron-py35/b9efb86/ : SUCCESS in 22m 12s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/check/cross-nova-py27/3f552ea/ : SUCCESS in 12m 58s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/check/cross-nova-py35/a96ef32/ : SUCCESS in 14m 16s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/check/cross-nova-functional/aed4d93/ : SUCCESS in 21m 06s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/check/cross-swift-py27/84102d0/ : SUCCESS in 8m 40s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/check/cross-swift-py35/8b1b1c7/ : SUCCESS in 5m 01s\n- requirements-integration http://logs.openstack.org/72/653272/2/check/requirements-integration/5884bb7/ : SUCCESS in 57m 48s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/check/neutron-grenade-multinode/bd88b54/ : SUCCESS in 1h 04m 38s\n- legacy-tempest-dsvm-neutron-full finger://ze06.openstack.org/452f994a214d489f9ba9ae2302363aa8 : RETRY_LIMIT in 1h 17m 48s\n- neutron-grenade http://logs.openstack.org/72/653272/2/check/neutron-grenade/e498058/ : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":2},{"id":"05dbf9cbd2935d8dd6788a03358c2367c1254f27","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-20 08:27:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6c847933cfa9abcd2eb08995223ada18dec314bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 08:53:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/72/653272/2/check/openstack-tox-pypy/7ad0cfb/ : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/check/openstack-tox-pep8/6f2d40a/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/check/openstack-tox-py27/d8f0b0f/ : SUCCESS in 3m 49s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/check/openstack-tox-py35/eb93f21/ : SUCCESS in 3m 45s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/check/openstack-tox-validate/0448272/ : SUCCESS in 3m 30s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/check/requirements-tox-babel/e2fb8fe/ : SUCCESS in 3m 13s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/check/requirements-tox-py27-check-uc/432b8f4/ : SUCCESS in 6m 06s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/check/cross-cinder-py27/8bfb6af/ : SUCCESS in 8m 24s\n- cross-cinder-py35 http://logs.openstack.org/72/653272/2/check/cross-cinder-py35/fe6158f/ : SUCCESS in 7m 12s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/check/cross-glance-py27/2068e8c/ : SUCCESS in 8m 53s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/check/cross-glance-py35/b4a25ae/ : SUCCESS in 10m 05s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/check/cross-horizon-py27/c3de1aa/ : SUCCESS in 10m 03s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/check/cross-horizon-py35/8ce5627/ : SUCCESS in 9m 45s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/check/cross-keystone-py27/6ab2f71/ : SUCCESS in 10m 25s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/check/cross-keystone-py35/809f2cd/ : SUCCESS in 10m 47s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/check/cross-neutron-py27/282b1c4/ : SUCCESS in 19m 56s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/check/cross-neutron-py35/5cfd7ea/ : SUCCESS in 21m 00s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/check/cross-nova-py27/c119055/ : SUCCESS in 12m 26s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/check/cross-nova-py35/fc94b87/ : SUCCESS in 13m 35s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/check/cross-nova-functional/bbbf193/ : SUCCESS in 18m 12s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/check/cross-swift-py27/71aafc7/ : SUCCESS in 7m 42s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/check/cross-swift-py35/6936380/ : SUCCESS in 3m 23s\n- requirements-integration http://logs.openstack.org/72/653272/2/check/requirements-integration/adcec06/ : FAILURE in 12m 14s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/check/neutron-grenade-multinode/d096c70/ : FAILURE in 11m 00s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/2/check/legacy-tempest-dsvm-neutron-full/8eb8dd0/ : FAILURE in 8m 15s\n- neutron-grenade http://logs.openstack.org/72/653272/2/check/neutron-grenade/4727b30/ : FAILURE in 10m 55s","accounts_in_message":[],"_revision_number":2},{"id":"a59ed49de8720eb008a90c1dc5b2208ea64e0dc6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 14:55:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3d179715a062799c0cef379950eb22f9cece23fa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 16:18:44.000000000","message":"Patch Set 2: -Code-Review\n\nrecheck (and unvote til gate works again)","accounts_in_message":[],"_revision_number":2},{"id":"f8348386d6550d661148308a431cc55f70f4a34c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 19:04:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/72/653272/2/check/openstack-tox-pypy/ba73672/ : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/check/openstack-tox-pep8/b923018/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/check/openstack-tox-py27/c80ee5e/ : SUCCESS in 3m 38s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/check/openstack-tox-py35/fdc5e9d/ : SUCCESS in 3m 48s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/check/openstack-tox-validate/0f97b9b/ : SUCCESS in 4m 00s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/check/requirements-tox-babel/8090d07/ : SUCCESS in 4m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/check/requirements-tox-py27-check-uc/de134ce/ : SUCCESS in 5m 44s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/check/cross-cinder-py27/5d5f2e7/ : SUCCESS in 8m 35s\n- cross-cinder-py35 http://logs.openstack.org/72/653272/2/check/cross-cinder-py35/a6f5dcc/ : SUCCESS in 8m 59s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/check/cross-glance-py27/f436a6f/ : SUCCESS in 9m 28s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/check/cross-glance-py35/7a9db69/ : SUCCESS in 10m 03s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/check/cross-horizon-py27/d0287ee/ : SUCCESS in 10m 58s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/check/cross-horizon-py35/5fc1634/ : SUCCESS in 10m 14s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/check/cross-keystone-py27/d54a2fa/ : SUCCESS in 12m 24s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/check/cross-keystone-py35/e3a5968/ : SUCCESS in 16m 04s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/check/cross-neutron-py27/46d79a0/ : SUCCESS in 21m 30s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/check/cross-neutron-py35/1b95083/ : SUCCESS in 26m 01s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/check/cross-nova-py27/0b41ecd/ : SUCCESS in 13m 50s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/check/cross-nova-py35/211dd61/ : SUCCESS in 13m 20s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/check/cross-nova-functional/b644d19/ : SUCCESS in 19m 37s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/check/cross-swift-py27/0162ffa/ : SUCCESS in 7m 14s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/check/cross-swift-py35/3a80c65/ : SUCCESS in 3m 44s\n- requirements-integration http://logs.openstack.org/72/653272/2/check/requirements-integration/47bce96/ : SUCCESS in 43m 35s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/check/neutron-grenade-multinode/7882c79/ : SUCCESS in 55m 31s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/2/check/legacy-tempest-dsvm-neutron-full/4edc217/ : FAILURE in 1h 09m 03s\n- neutron-grenade http://logs.openstack.org/72/653272/2/check/neutron-grenade/9cc8e48/ : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":2},{"id":"3a76b8974a0fc4b82b9c38197cd04002cbb017c2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-23 14:43:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e61548b25f50a6fa7537470dbc2502f7cdd45ad9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 03:03:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/72/653272/2/check/openstack-tox-pypy/3102b0b/ : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/check/openstack-tox-pep8/3c191e9/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/check/openstack-tox-py27/8e72da2/ : SUCCESS in 3m 59s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/check/openstack-tox-py35/b49e80f/ : SUCCESS in 3m 45s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/check/openstack-tox-validate/c80aece/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/check/requirements-tox-babel/1877bce/ : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/check/requirements-tox-py27-check-uc/92810b8/ : SUCCESS in 8m 23s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/check/cross-cinder-py27/668e04c/ : SUCCESS in 7m 05s\n- cross-cinder-py35 http://logs.openstack.org/72/653272/2/check/cross-cinder-py35/4b99120/ : SUCCESS in 7m 06s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/check/cross-glance-py27/458ae09/ : SUCCESS in 9m 44s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/check/cross-glance-py35/1820807/ : SUCCESS in 10m 43s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/check/cross-horizon-py27/5c455e3/ : SUCCESS in 11m 07s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/check/cross-horizon-py35/bc31f9e/ : SUCCESS in 10m 36s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/check/cross-keystone-py27/e5693de/ : SUCCESS in 13m 21s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/check/cross-keystone-py35/04e196f/ : SUCCESS in 14m 03s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/check/cross-neutron-py27/bed7d05/ : SUCCESS in 18m 31s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/check/cross-neutron-py35/3d50289/ : SUCCESS in 21m 49s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/check/cross-nova-py27/00f9ffd/ : SUCCESS in 12m 01s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/check/cross-nova-py35/53c4942/ : SUCCESS in 16m 04s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/check/cross-nova-functional/b02377b/ : SUCCESS in 22m 54s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/check/cross-swift-py27/054197c/ : SUCCESS in 7m 34s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/check/cross-swift-py35/ee34530/ : SUCCESS in 3m 36s\n- requirements-integration http://logs.openstack.org/72/653272/2/check/requirements-integration/bc54c50/ : SUCCESS in 1h 02m 16s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/check/neutron-grenade-multinode/024cbf2/ : FAILURE in 39m 01s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/2/check/legacy-tempest-dsvm-neutron-full/66d00a0/ : SUCCESS in 1h 41m 31s\n- neutron-grenade http://logs.openstack.org/72/653272/2/check/neutron-grenade/8151a0c/ : SUCCESS in 52m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fd6d0afb7fc4694b6ba180184fcb132b39165401","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-24 15:18:46.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"29ead53e311ee2c3495b2d569baa0c228488f675","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 02:59:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/72/653272/2/check/openstack-tox-pypy/ed88a60/ : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/check/openstack-tox-pep8/819d6d7/ : SUCCESS in 3m 49s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/check/openstack-tox-py27/5da580b/ : SUCCESS in 4m 20s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/check/openstack-tox-py35/9612aeb/ : SUCCESS in 3m 55s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/check/openstack-tox-validate/a06e880/ : SUCCESS in 3m 37s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/check/requirements-tox-babel/25ebce3/ : SUCCESS in 4m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/check/requirements-tox-py27-check-uc/3576486/ : SUCCESS in 6m 28s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/check/cross-cinder-py27/22615fb/ : SUCCESS in 7m 56s\n- cross-cinder-py35 http://logs.openstack.org/72/653272/2/check/cross-cinder-py35/1b2cad9/ : SUCCESS in 7m 17s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/check/cross-glance-py27/222a6b4/ : SUCCESS in 8m 56s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/check/cross-glance-py35/f3c6d38/ : SUCCESS in 10m 59s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/check/cross-horizon-py27/8519c92/ : SUCCESS in 10m 40s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/check/cross-horizon-py35/1d1bbbd/ : SUCCESS in 10m 19s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/check/cross-keystone-py27/93abf9c/ : SUCCESS in 11m 19s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/check/cross-keystone-py35/2a94703/ : SUCCESS in 12m 18s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/check/cross-neutron-py27/63cf8e5/ : SUCCESS in 18m 03s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/check/cross-neutron-py35/16f7070/ : SUCCESS in 24m 01s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/check/cross-nova-py27/9a79205/ : SUCCESS in 11m 26s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/check/cross-nova-py35/77e9031/ : SUCCESS in 13m 26s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/check/cross-nova-functional/afc1419/ : SUCCESS in 20m 30s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/check/cross-swift-py27/29e8bf4/ : SUCCESS in 7m 45s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/check/cross-swift-py35/85f8525/ : SUCCESS in 2m 59s\n- requirements-integration http://logs.openstack.org/72/653272/2/check/requirements-integration/e7bad62/ : SUCCESS in 1h 00m 14s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/check/neutron-grenade-multinode/9dca748/ : SUCCESS in 1h 04m 39s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/2/check/legacy-tempest-dsvm-neutron-full/1d5c5b4/ : SUCCESS in 1h 22m 57s\n- neutron-grenade http://logs.openstack.org/72/653272/2/check/neutron-grenade/b70790b/ : SUCCESS in 59m 12s","accounts_in_message":[],"_revision_number":2},{"id":"50cb16dbf5d164f7d98f5368fa48eb585a4d81e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 03:00:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"22ceba6c65d2656693407fc1ae68dc1f0e43777a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 06:04:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/gate/openstack-tox-pep8/b28601e/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/gate/openstack-tox-py27/a9364a0/ : SUCCESS in 3m 38s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/gate/openstack-tox-py35/e98bb6e/ : SUCCESS in 3m 41s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/gate/openstack-tox-validate/972072e/ : SUCCESS in 4m 34s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/gate/requirements-tox-babel/823d55b/ : SUCCESS in 3m 10s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/gate/requirements-tox-py27-check-uc/373e3b8/ : SUCCESS in 6m 25s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/gate/cross-cinder-py27/fce7eb0/ : SUCCESS in 7m 37s\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/gate/cross-glance-py27/15ad2cc/ : SUCCESS in 9m 47s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/gate/cross-glance-py35/6e2db41/ : SUCCESS in 8m 24s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/gate/cross-horizon-py27/f795de1/ : SUCCESS in 10m 46s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/gate/cross-horizon-py35/3a14bac/ : SUCCESS in 9m 59s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/gate/cross-keystone-py27/4356b88/ : SUCCESS in 14m 13s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/gate/cross-keystone-py35/a1ff3c6/ : SUCCESS in 11m 45s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/gate/cross-neutron-py27/1573195/ : SUCCESS in 21m 13s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/gate/cross-neutron-py35/0c0256a/ : SUCCESS in 39m 26s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/gate/cross-nova-py27/26a40d9/ : SUCCESS in 14m 27s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/gate/cross-nova-py35/dfefee8/ : SUCCESS in 11m 48s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/gate/cross-nova-functional/418e556/ : SUCCESS in 29m 24s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/gate/cross-swift-py27/74521ff/ : SUCCESS in 10m 03s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/gate/cross-swift-py35/98eb4e0/ : SUCCESS in 3m 05s\n- requirements-integration http://logs.openstack.org/72/653272/2/gate/requirements-integration/6919dda/ : SUCCESS in 51m 54s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/gate/neutron-grenade-multinode/c43ef4b/ : SUCCESS in 1h 00m 10s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/2/gate/legacy-tempest-dsvm-neutron-full/c364c51/ : FAILURE in 1h 17m 10s\n- neutron-grenade http://logs.openstack.org/72/653272/2/gate/neutron-grenade/7c0a0a5/ : SUCCESS in 50m 54s","accounts_in_message":[],"_revision_number":2},{"id":"1d9e0fe94542a048a9e6ba94e2b3b32aafe3229a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-25 13:14:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"98471747bff471de346ad809bfb530de741c6a89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 21:00:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/72/653272/2/check/openstack-tox-pypy/f7fa3a0/ : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/check/openstack-tox-pep8/284bcd0/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/check/openstack-tox-py27/7c93d81/ : SUCCESS in 3m 52s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/check/openstack-tox-py35/7440563/ : SUCCESS in 4m 16s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/check/openstack-tox-validate/34e7e75/ : SUCCESS in 3m 38s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/check/requirements-tox-babel/ac89204/ : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/check/requirements-tox-py27-check-uc/6dfb65a/ : SUCCESS in 6m 23s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/check/cross-cinder-py27/38cf56f/ : FAILURE in 7m 57s\n- cross-cinder-py35 http://logs.openstack.org/72/653272/2/check/cross-cinder-py35/28003a6/ : SUCCESS in 7m 30s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/check/cross-glance-py27/ba9578b/ : SUCCESS in 9m 23s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/check/cross-glance-py35/e7db64f/ : SUCCESS in 9m 00s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/check/cross-horizon-py27/e3328c9/ : SUCCESS in 11m 02s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/check/cross-horizon-py35/aa03e53/ : SUCCESS in 10m 36s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/check/cross-keystone-py27/5e993cd/ : TIMED_OUT in 40m 56s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/check/cross-keystone-py35/15fe680/ : SUCCESS in 11m 59s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/check/cross-neutron-py27/df1d0d1/ : SUCCESS in 21m 03s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/check/cross-neutron-py35/0993cee/ : SUCCESS in 26m 02s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/check/cross-nova-py27/cde092a/ : SUCCESS in 13m 54s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/check/cross-nova-py35/36c2a27/ : SUCCESS in 12m 41s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/check/cross-nova-functional/c0908a5/ : SUCCESS in 23m 19s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/check/cross-swift-py27/2dee777/ : SUCCESS in 22m 45s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/check/cross-swift-py35/d776c3d/ : SUCCESS in 3m 21s\n- requirements-integration http://logs.openstack.org/72/653272/2/check/requirements-integration/49bb52c/ : SUCCESS in 1h 00m 33s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/check/neutron-grenade-multinode/416598e/ : SUCCESS in 1h 40m 26s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/2/check/legacy-tempest-dsvm-neutron-full/cf17680/ : SUCCESS in 1h 23m 05s\n- neutron-grenade http://logs.openstack.org/72/653272/2/check/neutron-grenade/e33fcfc/ : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":2},{"id":"fd9f1d4c668632e628ee93fbffa249af05f6b000","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-25 21:20:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f141362fe47bd3269d2c810548e5b8974ace14c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 06:09:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/72/653272/2/check/openstack-tox-pypy/52fbf82/ : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/check/openstack-tox-pep8/7d855fe/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/check/openstack-tox-py27/28850b4/ : SUCCESS in 4m 57s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/check/openstack-tox-py35/720be1d/ : SUCCESS in 4m 00s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/check/openstack-tox-validate/792b80e/ : SUCCESS in 4m 36s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/check/requirements-tox-babel/7233676/ : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/check/requirements-tox-py27-check-uc/1973404/ : SUCCESS in 6m 44s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/check/cross-cinder-py27/d8f92a3/ : SUCCESS in 7m 53s\n- cross-cinder-py35 http://logs.openstack.org/72/653272/2/check/cross-cinder-py35/0fd2fb2/ : SUCCESS in 7m 27s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/check/cross-glance-py27/b74e46e/ : SUCCESS in 10m 03s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/check/cross-glance-py35/3932404/ : SUCCESS in 9m 59s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/check/cross-horizon-py27/c8f471a/ : SUCCESS in 10m 21s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/check/cross-horizon-py35/b9d60a2/ : SUCCESS in 9m 24s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/check/cross-keystone-py27/feb4120/ : TIMED_OUT in 40m 52s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/check/cross-keystone-py35/098f435/ : SUCCESS in 16m 43s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/check/cross-neutron-py27/5a9b2c8/ : SUCCESS in 21m 11s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/check/cross-neutron-py35/2e09d77/ : SUCCESS in 22m 18s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/check/cross-nova-py27/05f18cd/ : SUCCESS in 11m 30s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/check/cross-nova-py35/e979753/ : SUCCESS in 11m 15s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/check/cross-nova-functional/2692854/ : SUCCESS in 18m 48s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/check/cross-swift-py27/6bc51af/ : SUCCESS in 7m 26s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/check/cross-swift-py35/382ce9e/ : SUCCESS in 3m 35s\n- requirements-integration http://logs.openstack.org/72/653272/2/check/requirements-integration/2b00c4f/ : SUCCESS in 52m 22s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/check/neutron-grenade-multinode/9e43ca8/ : SUCCESS in 57m 25s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/2/check/legacy-tempest-dsvm-neutron-full/10da8cd/ : SUCCESS in 1h 28m 46s\n- neutron-grenade http://logs.openstack.org/72/653272/2/check/neutron-grenade/62e3413/ : SUCCESS in 53m 03s","accounts_in_message":[],"_revision_number":2},{"id":"27f4944e43ff35c5c47d0c06935caabacbc0b6d0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-26 12:48:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5af49f04c7da5f213a14831e1ed69cfd603f9360","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 15:15:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/72/653272/2/check/openstack-tox-pypy/7359119/ : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/check/openstack-tox-pep8/5aa07cc/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/check/openstack-tox-py27/4e06281/ : SUCCESS in 6m 34s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/check/openstack-tox-py35/de93171/ : SUCCESS in 3m 37s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/check/openstack-tox-validate/1e1c2fd/ : SUCCESS in 3m 07s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/check/requirements-tox-babel/fd9c657/ : SUCCESS in 3m 14s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/check/requirements-tox-py27-check-uc/703412b/ : SUCCESS in 7m 15s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/check/cross-cinder-py27/c3452dc/ : SUCCESS in 7m 37s\n- cross-cinder-py35 http://logs.openstack.org/72/653272/2/check/cross-cinder-py35/17d1bca/ : SUCCESS in 7m 44s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/check/cross-glance-py27/b6a5185/ : SUCCESS in 8m 26s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/check/cross-glance-py35/b37ff2c/ : SUCCESS in 8m 54s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/check/cross-horizon-py27/a9f4e40/ : SUCCESS in 10m 47s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/check/cross-horizon-py35/f98d103/ : SUCCESS in 11m 57s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/check/cross-keystone-py27/fd57fae/ : SUCCESS in 11m 16s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/check/cross-keystone-py35/0d51f4b/ : SUCCESS in 14m 10s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/check/cross-neutron-py27/cac132e/ : SUCCESS in 21m 14s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/check/cross-neutron-py35/952d9b4/ : SUCCESS in 25m 03s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/check/cross-nova-py27/c3cab9f/ : SUCCESS in 13m 05s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/check/cross-nova-py35/801edad/ : SUCCESS in 12m 27s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/check/cross-nova-functional/190cf4f/ : SUCCESS in 19m 46s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/check/cross-swift-py27/d25147c/ : SUCCESS in 14m 59s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/check/cross-swift-py35/c390674/ : SUCCESS in 3m 22s\n- requirements-integration http://logs.openstack.org/72/653272/2/check/requirements-integration/fd2f615/ : SUCCESS in 54m 42s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/check/neutron-grenade-multinode/5aee18c/ : SUCCESS in 1h 05m 13s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/2/check/legacy-tempest-dsvm-neutron-full/7ae7a37/ : SUCCESS in 1h 31m 09s\n- neutron-grenade http://logs.openstack.org/72/653272/2/check/neutron-grenade/087e76f/ : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":2},{"id":"9f2f69e28b3e47214ef538aa7173cb983d8b952a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 15:16:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"90a957aaae21d8338450b627ed6a7483e0cb7e73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 16:05:51.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c4de41809ca22d9f4ea0edb79f2f1c288aad37ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 18:57:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/gate/openstack-tox-pep8/741e7d9/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/gate/openstack-tox-py27/383a254/ : SUCCESS in 3m 24s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/gate/openstack-tox-py35/ac5e8e6/ : SUCCESS in 3m 36s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/gate/openstack-tox-validate/337afe8/ : SUCCESS in 3m 33s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/gate/requirements-tox-babel/7cf3b8a/ : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/gate/requirements-tox-py27-check-uc/f2f037e/ : SUCCESS in 6m 33s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/gate/cross-cinder-py27/a0b916c/ : SUCCESS in 10m 22s\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/gate/cross-glance-py27/5950faa/ : SUCCESS in 9m 09s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/gate/cross-glance-py35/1b99d12/ : SUCCESS in 9m 30s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/gate/cross-horizon-py27/1a3ee75/ : SUCCESS in 11m 12s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/gate/cross-horizon-py35/70af21a/ : SUCCESS in 10m 28s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/gate/cross-keystone-py27/2da5208/ : SUCCESS in 22m 07s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/gate/cross-keystone-py35/8ba651e/ : SUCCESS in 12m 12s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/gate/cross-neutron-py27/f37540b/ : SUCCESS in 20m 34s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/gate/cross-neutron-py35/00a120e/ : SUCCESS in 21m 16s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/gate/cross-nova-py27/e973f0d/ : SUCCESS in 17m 00s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/gate/cross-nova-py35/2aefdb2/ : SUCCESS in 15m 21s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/gate/cross-nova-functional/b0d8fc1/ : SUCCESS in 22m 02s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/gate/cross-swift-py27/c6ef79e/ : SUCCESS in 8m 13s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/gate/cross-swift-py35/183c6e5/ : SUCCESS in 3m 14s\n- requirements-integration http://logs.openstack.org/72/653272/2/gate/requirements-integration/b49184e/ : SUCCESS in 51m 06s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/gate/neutron-grenade-multinode/7e480a4/ : SUCCESS in 1h 03m 41s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/2/gate/legacy-tempest-dsvm-neutron-full/c8a4474/ : FAILURE in 1h 45m 04s\n- neutron-grenade http://logs.openstack.org/72/653272/2/gate/neutron-grenade/1e16ef2/ : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":2},{"id":"4d82c008c16755f4ee27dc02856ed262694f6ed3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-26 22:13:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d34153a895543c648b8608de9e5820142bb06763","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-26 23:28:15.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b44e1a0d4da10f74672bc6105bc0f093bdb80e4f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-26 23:28:26.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"e3433dab32a8a499110090dbb3dd53def00651b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 05:57:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/72/653272/2/check/openstack-tox-pypy/9614a3a/ : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/check/openstack-tox-pep8/b037478/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/check/openstack-tox-py27/93d98ac/ : SUCCESS in 3m 49s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/check/openstack-tox-py35/ec1326c/ : SUCCESS in 4m 50s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/check/openstack-tox-validate/2b737be/ : SUCCESS in 3m 40s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/check/requirements-tox-babel/af6939d/ : SUCCESS in 4m 28s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/check/requirements-tox-py27-check-uc/e781a66/ : SUCCESS in 7m 45s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/check/cross-cinder-py27/905a1e2/ : SUCCESS in 8m 05s\n- cross-cinder-py35 http://logs.openstack.org/72/653272/2/check/cross-cinder-py35/fc7f3b9/ : SUCCESS in 7m 36s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/check/cross-glance-py27/b2c1046/ : SUCCESS in 9m 47s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/check/cross-glance-py35/fc5b7f1/ : SUCCESS in 9m 32s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/check/cross-horizon-py27/6b8f3e0/ : SUCCESS in 10m 57s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/check/cross-horizon-py35/34b3528/ : SUCCESS in 10m 00s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/check/cross-keystone-py27/8a4698b/ : SUCCESS in 17m 17s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/check/cross-keystone-py35/e0d526a/ : SUCCESS in 15m 05s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/check/cross-neutron-py27/e6e38a1/ : SUCCESS in 23m 26s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/check/cross-neutron-py35/ff68972/ : SUCCESS in 24m 18s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/check/cross-nova-py27/7b82a74/ : SUCCESS in 15m 09s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/check/cross-nova-py35/fb1f682/ : SUCCESS in 14m 07s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/check/cross-nova-functional/379c64c/ : SUCCESS in 18m 01s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/check/cross-swift-py27/dce7f94/ : SUCCESS in 7m 25s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/check/cross-swift-py35/22c2e3a/ : SUCCESS in 4m 13s\n- requirements-integration http://logs.openstack.org/72/653272/2/check/requirements-integration/4f94ceb/ : SUCCESS in 53m 36s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/check/neutron-grenade-multinode/cfcda5f/ : SUCCESS in 58m 41s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/2/check/legacy-tempest-dsvm-neutron-full/62eaa25/ : SUCCESS in 1h 26m 54s\n- neutron-grenade http://logs.openstack.org/72/653272/2/check/neutron-grenade/b1d9f0d/ : SUCCESS in 49m 39s","accounts_in_message":[],"_revision_number":2},{"id":"ffee40cf8582a5beae58ae7617028cec93467a62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 09:13:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"88e36828d3681872b8959e5cc34651371f1481dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 18:16:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/72/653272/2/gate/openstack-tox-pep8/a9798d0/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/72/653272/2/gate/openstack-tox-py27/8ab8678/ : SUCCESS in 3m 10s\n- openstack-tox-py35 http://logs.openstack.org/72/653272/2/gate/openstack-tox-py35/506a13b/ : SUCCESS in 6m 03s\n- openstack-tox-validate http://logs.openstack.org/72/653272/2/gate/openstack-tox-validate/f36463a/ : SUCCESS in 5m 30s\n- requirements-tox-babel http://logs.openstack.org/72/653272/2/gate/requirements-tox-babel/0fd3322/ : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/72/653272/2/gate/requirements-tox-py27-check-uc/08b6835/ : SUCCESS in 6m 56s\n- cross-cinder-py27 http://logs.openstack.org/72/653272/2/gate/cross-cinder-py27/ada248c/ : SUCCESS in 7m 48s\n- cross-glance-py27 http://logs.openstack.org/72/653272/2/gate/cross-glance-py27/301f35a/ : SUCCESS in 9m 29s\n- cross-glance-py35 http://logs.openstack.org/72/653272/2/gate/cross-glance-py35/c0491a5/ : SUCCESS in 12m 22s\n- cross-horizon-py27 http://logs.openstack.org/72/653272/2/gate/cross-horizon-py27/5dd7eca/ : SUCCESS in 14m 40s\n- cross-horizon-py35 http://logs.openstack.org/72/653272/2/gate/cross-horizon-py35/7fc2312/ : SUCCESS in 13m 50s\n- cross-keystone-py27 http://logs.openstack.org/72/653272/2/gate/cross-keystone-py27/232d8f0/ : SUCCESS in 12m 01s\n- cross-keystone-py35 http://logs.openstack.org/72/653272/2/gate/cross-keystone-py35/232d8eb/ : SUCCESS in 24m 07s\n- cross-neutron-py27 http://logs.openstack.org/72/653272/2/gate/cross-neutron-py27/04abed9/ : SUCCESS in 23m 54s\n- cross-neutron-py35 http://logs.openstack.org/72/653272/2/gate/cross-neutron-py35/fd1017d/ : SUCCESS in 22m 20s\n- cross-nova-py27 http://logs.openstack.org/72/653272/2/gate/cross-nova-py27/878cd23/ : SUCCESS in 17m 25s\n- cross-nova-py35 http://logs.openstack.org/72/653272/2/gate/cross-nova-py35/f1e5a1b/ : SUCCESS in 15m 11s\n- cross-nova-functional http://logs.openstack.org/72/653272/2/gate/cross-nova-functional/43dff67/ : SUCCESS in 26m 59s\n- cross-swift-py27 http://logs.openstack.org/72/653272/2/gate/cross-swift-py27/4a4d7d7/ : SUCCESS in 7m 20s\n- cross-swift-py35 http://logs.openstack.org/72/653272/2/gate/cross-swift-py35/0df150a/ : SUCCESS in 5m 45s\n- requirements-integration http://logs.openstack.org/72/653272/2/gate/requirements-integration/46a6528/ : SUCCESS in 50m 25s\n- neutron-grenade-multinode http://logs.openstack.org/72/653272/2/gate/neutron-grenade-multinode/bff8b83/ : SUCCESS in 59m 02s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/72/653272/2/gate/legacy-tempest-dsvm-neutron-full/8c806b5/ : SUCCESS in 1h 34m 31s\n- neutron-grenade http://logs.openstack.org/72/653272/2/gate/neutron-grenade/460d413/ : SUCCESS in 51m 32s","accounts_in_message":[],"_revision_number":2},{"id":"cedf87f42f0d06cd27d8933435efbfec52aebb5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 18:16:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"47597c43be6906485c05123ce28cf08d5b28a957","revisions":{"ac751cce203baea27f8a0fd268d734e0192986e2":{"kind":"REWORK","_number":1,"created":"2019-04-17 03:59:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/653272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/653272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/653272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/653272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/653272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/653272/1"}}},"commit":{"parents":[{"commit":"03eeccfc2edaaf995010da7a8c2b6b6188439b30","subject":"update constraint for os-brick to new release 1.15.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03eeccfc2edaaf995010da7a8c2b6b6188439b30"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 03:59:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 03:59:14.000000000","tz":0},"subject":"update constraint for python-searchlightclient to new release 1.2.1","message":"update constraint for python-searchlightclient to new release 1.2.1\n\nChange-Id: I922465e910f57ffaf5a61563a56b026e3d5558ef\nmeta:version: 1.2.1\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Change-Id: I6d72c57fdc58623a7ce89d6320d703b1acfd6210\nmeta:release:Code-Review+1: Trinh Nguyen \u003cdangtrinhnt@gmail.com\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac751cce203baea27f8a0fd268d734e0192986e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac751cce203baea27f8a0fd268d734e0192986e2"}]},"branch":"refs/heads/stable/pike"},"47597c43be6906485c05123ce28cf08d5b28a957":{"kind":"REWORK","_number":2,"created":"2019-04-17 19:46:43.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/72/653272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/653272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/653272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/653272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/653272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/653272/2"}}},"commit":{"parents":[{"commit":"3170a63a11028e9701bc0590fc15a44ede84d30f","subject":"Merge \"update constraint for oslo.reports to new release 1.22.2\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3170a63a11028e9701bc0590fc15a44ede84d30f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 03:59:14.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-04-17 19:46:31.000000000","tz":-300},"subject":"update constraint for python-searchlightclient to new release 1.2.1","message":"update constraint for python-searchlightclient to new release 1.2.1\n\nChange-Id: I922465e910f57ffaf5a61563a56b026e3d5558ef\nmeta:version: 1.2.1\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Change-Id: I6d72c57fdc58623a7ce89d6320d703b1acfd6210\nmeta:release:Code-Review+1: Trinh Nguyen \u003cdangtrinhnt@gmail.com\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47597c43be6906485c05123ce28cf08d5b28a957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47597c43be6906485c05123ce28cf08d5b28a957"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
