)]}'
{"id":"openstack%2Frequirements~653340","triplet_id":"openstack%2Frequirements~stable%2Fpike~I3548aaf0c913a671b14a0e96b3c4cc2c77a65b8c","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"I3548aaf0c913a671b14a0e96b3c4cc2c77a65b8c","subject":"update constraint for python-senlinclient to new release 1.4.1","status":"MERGED","created":"2019-04-17 07:42:39.000000000","updated":"2019-04-23 03:55:22.000000000","submitted":"2019-04-23 03:55:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653340-1555991722877-4f1731f4","meta_rev_id":"08efd8ea22e09d1f6a3c59d2e1c2cbbfad2081fa","_number":653340,"virtual_id_number":653340,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-04-23 03:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-23 00:09:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-23 00:29:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-04-23 00:29:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 00:09:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-23 00:29:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-23 03:55:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf8fc54a4aeb8a79d9b5e34f86ee351835c798ba","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-17 07:42:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d58e7fa696b1a585e17ced9f9095632887cd75f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 13:43:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/40/653340/1/check/openstack-tox-pypy/c06e1d1/ : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/653340/1/check/openstack-tox-pep8/ea5e4da/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/40/653340/1/check/openstack-tox-py27/b0a385c/ : SUCCESS in 5m 17s\n- openstack-tox-py35 http://logs.openstack.org/40/653340/1/check/openstack-tox-py35/0a7aa37/ : SUCCESS in 4m 04s\n- openstack-tox-validate http://logs.openstack.org/40/653340/1/check/openstack-tox-validate/ad1a520/ : SUCCESS in 3m 17s\n- requirements-tox-babel http://logs.openstack.org/40/653340/1/check/requirements-tox-babel/a72f6f8/ : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/653340/1/check/requirements-tox-py27-check-uc/aaf6ded/ : SUCCESS in 8m 12s\n- cross-cinder-py27 http://logs.openstack.org/40/653340/1/check/cross-cinder-py27/b5df3ff/ : SUCCESS in 9m 42s\n- cross-cinder-py35 http://logs.openstack.org/40/653340/1/check/cross-cinder-py35/c285094/ : SUCCESS in 8m 31s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/40/653340/1/check/cross-glance-py27/ee06f37/ : SUCCESS in 10m 03s\n- cross-glance-py35 http://logs.openstack.org/40/653340/1/check/cross-glance-py35/728ea80/ : SUCCESS in 10m 06s\n- cross-horizon-py27 http://logs.openstack.org/40/653340/1/check/cross-horizon-py27/4ca07d4/ : SUCCESS in 11m 40s\n- cross-horizon-py35 http://logs.openstack.org/40/653340/1/check/cross-horizon-py35/447c279/ : SUCCESS in 10m 35s\n- cross-keystone-py27 http://logs.openstack.org/40/653340/1/check/cross-keystone-py27/127b4a8/ : SUCCESS in 16m 05s\n- cross-keystone-py35 http://logs.openstack.org/40/653340/1/check/cross-keystone-py35/57601af/ : SUCCESS in 14m 43s\n- cross-neutron-py27 http://logs.openstack.org/40/653340/1/check/cross-neutron-py27/2191fc0/ : SUCCESS in 26m 40s\n- cross-neutron-py35 http://logs.openstack.org/40/653340/1/check/cross-neutron-py35/854ca71/ : SUCCESS in 23m 22s\n- cross-nova-py27 http://logs.openstack.org/40/653340/1/check/cross-nova-py27/3da5977/ : SUCCESS in 14m 12s\n- cross-nova-py35 http://logs.openstack.org/40/653340/1/check/cross-nova-py35/c201888/ : SUCCESS in 14m 07s\n- cross-nova-functional http://logs.openstack.org/40/653340/1/check/cross-nova-functional/3f88e00/ : SUCCESS in 23m 07s\n- cross-swift-py27 http://logs.openstack.org/40/653340/1/check/cross-swift-py27/c015715/ : SUCCESS in 9m 02s\n- cross-swift-py35 http://logs.openstack.org/40/653340/1/check/cross-swift-py35/aeff7ff/ : SUCCESS in 4m 21s\n- requirements-integration http://logs.openstack.org/40/653340/1/check/requirements-integration/7e99023/ : SUCCESS in 1h 04m 53s\n- neutron-grenade-multinode http://logs.openstack.org/40/653340/1/check/neutron-grenade-multinode/c5428b8/ : SUCCESS in 59m 50s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/40/653340/1/check/legacy-tempest-dsvm-neutron-full/afa1885/ : FAILURE in 1h 08m 53s\n- neutron-grenade http://logs.openstack.org/40/653340/1/check/neutron-grenade/58f68ee/ : SUCCESS in 51m 53s","accounts_in_message":[],"_revision_number":1},{"id":"cdf1eb7dc4fe2c4f7edf1275ef280b82c4a8d493","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-18 06:20:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"56181df7b6ec1609a4132988fa67dd129658be3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 06:20:28.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"fa2423be627e0145350f153dd08577d324ad1e72","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-18 06:23:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"81931dbc4cce710f65e5903d9c5605ae98bbff7c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-18 06:26:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bed45dd529918c2da8eef3df96a793d153b26f8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 15:42:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/40/653340/2/check/openstack-tox-pypy/e59d5ea/ : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/653340/2/check/openstack-tox-pep8/75d8e82/ : SUCCESS in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/40/653340/2/check/openstack-tox-py27/565a521/ : SUCCESS in 4m 17s\n- openstack-tox-py35 http://logs.openstack.org/40/653340/2/check/openstack-tox-py35/d096101/ : SUCCESS in 4m 53s\n- openstack-tox-validate http://logs.openstack.org/40/653340/2/check/openstack-tox-validate/b070b76/ : SUCCESS in 3m 29s\n- requirements-tox-babel http://logs.openstack.org/40/653340/2/check/requirements-tox-babel/1a0b92c/ : SUCCESS in 4m 28s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/653340/2/check/requirements-tox-py27-check-uc/c1ae5b3/ : SUCCESS in 7m 07s\n- cross-cinder-py27 http://logs.openstack.org/40/653340/2/check/cross-cinder-py27/3b22681/ : SUCCESS in 8m 33s\n- cross-cinder-py35 http://logs.openstack.org/40/653340/2/check/cross-cinder-py35/a7373fa/ : SUCCESS in 9m 11s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/40/653340/2/check/cross-glance-py27/a7c5997/ : SUCCESS in 10m 08s\n- cross-glance-py35 http://logs.openstack.org/40/653340/2/check/cross-glance-py35/4e25427/ : SUCCESS in 10m 02s\n- cross-horizon-py27 http://logs.openstack.org/40/653340/2/check/cross-horizon-py27/cc06634/ : SUCCESS in 10m 40s\n- cross-horizon-py35 http://logs.openstack.org/40/653340/2/check/cross-horizon-py35/044eda6/ : SUCCESS in 13m 04s\n- cross-keystone-py27 http://logs.openstack.org/40/653340/2/check/cross-keystone-py27/cb6eb46/ : SUCCESS in 14m 54s\n- cross-keystone-py35 http://logs.openstack.org/40/653340/2/check/cross-keystone-py35/ec79749/ : SUCCESS in 18m 34s\n- cross-neutron-py27 http://logs.openstack.org/40/653340/2/check/cross-neutron-py27/48a8849/ : SUCCESS in 22m 52s\n- cross-neutron-py35 http://logs.openstack.org/40/653340/2/check/cross-neutron-py35/b73c5d6/ : SUCCESS in 23m 55s\n- cross-nova-py27 http://logs.openstack.org/40/653340/2/check/cross-nova-py27/0aab5db/ : SUCCESS in 13m 18s\n- cross-nova-py35 http://logs.openstack.org/40/653340/2/check/cross-nova-py35/a9cb688/ : SUCCESS in 14m 10s\n- cross-nova-functional http://logs.openstack.org/40/653340/2/check/cross-nova-functional/bc86e87/ : SUCCESS in 21m 47s\n- cross-swift-py27 http://logs.openstack.org/40/653340/2/check/cross-swift-py27/c576d91/ : SUCCESS in 8m 05s\n- cross-swift-py35 http://logs.openstack.org/40/653340/2/check/cross-swift-py35/2d9cddd/ : SUCCESS in 4m 35s\n- requirements-integration http://logs.openstack.org/40/653340/2/check/requirements-integration/1aa4b3e/ : SUCCESS in 55m 17s\n- neutron-grenade-multinode http://logs.openstack.org/40/653340/2/check/neutron-grenade-multinode/a699292/ : SUCCESS in 1h 03m 16s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/40/653340/2/check/legacy-tempest-dsvm-neutron-full/768e241/ : FAILURE in 1h 17m 20s\n- neutron-grenade http://logs.openstack.org/40/653340/2/check/neutron-grenade/d86225b/ : SUCCESS in 55m 12s","accounts_in_message":[],"_revision_number":2},{"id":"6ad8d3bc1df49491b299354b8428d244fae1e6a8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-18 16:17:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bcafebc06247e9137429eb46e725f5b69fd01778","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 06:42:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/40/653340/2/check/openstack-tox-pypy/1a7d9ba/ : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/653340/2/check/openstack-tox-pep8/e75cada/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/40/653340/2/check/openstack-tox-py27/0a80975/ : SUCCESS in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/40/653340/2/check/openstack-tox-py35/f410d3a/ : SUCCESS in 4m 40s\n- openstack-tox-validate http://logs.openstack.org/40/653340/2/check/openstack-tox-validate/2555f95/ : SUCCESS in 3m 53s\n- requirements-tox-babel http://logs.openstack.org/40/653340/2/check/requirements-tox-babel/b9848e1/ : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/653340/2/check/requirements-tox-py27-check-uc/8488ba2/ : SUCCESS in 5m 48s\n- cross-cinder-py27 http://logs.openstack.org/40/653340/2/check/cross-cinder-py27/f7fff9a/ : SUCCESS in 9m 47s\n- cross-cinder-py35 http://logs.openstack.org/40/653340/2/check/cross-cinder-py35/009ef02/ : SUCCESS in 9m 27s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/40/653340/2/check/cross-glance-py27/d02d14b/ : SUCCESS in 10m 29s\n- cross-glance-py35 http://logs.openstack.org/40/653340/2/check/cross-glance-py35/701b030/ : SUCCESS in 9m 26s\n- cross-horizon-py27 http://logs.openstack.org/40/653340/2/check/cross-horizon-py27/93660ac/ : SUCCESS in 10m 47s\n- cross-horizon-py35 http://logs.openstack.org/40/653340/2/check/cross-horizon-py35/c24a396/ : SUCCESS in 11m 09s\n- cross-keystone-py27 http://logs.openstack.org/40/653340/2/check/cross-keystone-py27/6dee07e/ : SUCCESS in 11m 45s\n- cross-keystone-py35 http://logs.openstack.org/40/653340/2/check/cross-keystone-py35/5ef4004/ : SUCCESS in 10m 35s\n- cross-neutron-py27 http://logs.openstack.org/40/653340/2/check/cross-neutron-py27/b5c0409/ : SUCCESS in 21m 06s\n- cross-neutron-py35 http://logs.openstack.org/40/653340/2/check/cross-neutron-py35/655a34d/ : SUCCESS in 21m 46s\n- cross-nova-py27 http://logs.openstack.org/40/653340/2/check/cross-nova-py27/405a4d9/ : SUCCESS in 11m 30s\n- cross-nova-py35 http://logs.openstack.org/40/653340/2/check/cross-nova-py35/3edeaab/ : SUCCESS in 11m 41s\n- cross-nova-functional http://logs.openstack.org/40/653340/2/check/cross-nova-functional/72e0bc5/ : SUCCESS in 25m 13s\n- cross-swift-py27 http://logs.openstack.org/40/653340/2/check/cross-swift-py27/3b4658a/ : SUCCESS in 6m 47s\n- cross-swift-py35 http://logs.openstack.org/40/653340/2/check/cross-swift-py35/dd284d1/ : SUCCESS in 5m 22s\n- requirements-integration http://logs.openstack.org/40/653340/2/check/requirements-integration/963ad4b/ : SUCCESS in 53m 10s\n- neutron-grenade-multinode finger://ze11.openstack.org/9c936e070279464081713d139dd2a7c5 : RETRY_LIMIT in 43m 14s\n- legacy-tempest-dsvm-neutron-full finger://ze08.openstack.org/ed90770c68504662a143cda302435e0b : RETRY_LIMIT in 43m 20s\n- neutron-grenade http://logs.openstack.org/40/653340/2/check/neutron-grenade/9c5fd56/ : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b612f763924c7df354da67a30ae294c9ff9dc5b3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-19 06:43:31.000000000","message":"Patch Set 2:\n\nrechech","accounts_in_message":[],"_revision_number":2},{"id":"ee15f48a51ce36f76ab9da776fc9d51eea693b81","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 14:55:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7d624e564eb6839db4b40527079f2db10c4d6e04","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 16:18:55.000000000","message":"Patch Set 2: -Code-Review\n\nrecheck (and unvote til gate works again)","accounts_in_message":[],"_revision_number":2},{"id":"57500adf352355bfb4f2d1dfd95fb0dfbae5705a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 19:27:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/40/653340/2/check/openstack-tox-pypy/0a2e515/ : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/653340/2/check/openstack-tox-pep8/6c5935e/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/40/653340/2/check/openstack-tox-py27/ad033c4/ : SUCCESS in 3m 53s\n- openstack-tox-py35 http://logs.openstack.org/40/653340/2/check/openstack-tox-py35/c3f27b1/ : SUCCESS in 3m 46s\n- openstack-tox-validate http://logs.openstack.org/40/653340/2/check/openstack-tox-validate/102624c/ : SUCCESS in 3m 25s\n- requirements-tox-babel http://logs.openstack.org/40/653340/2/check/requirements-tox-babel/5610615/ : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/653340/2/check/requirements-tox-py27-check-uc/c9f9a2a/ : SUCCESS in 6m 09s\n- cross-cinder-py27 http://logs.openstack.org/40/653340/2/check/cross-cinder-py27/023763d/ : SUCCESS in 9m 13s\n- cross-cinder-py35 http://logs.openstack.org/40/653340/2/check/cross-cinder-py35/242664c/ : SUCCESS in 8m 58s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/40/653340/2/check/cross-glance-py27/3caa0e2/ : SUCCESS in 9m 31s\n- cross-glance-py35 http://logs.openstack.org/40/653340/2/check/cross-glance-py35/3cf8637/ : SUCCESS in 13m 26s\n- cross-horizon-py27 http://logs.openstack.org/40/653340/2/check/cross-horizon-py27/e7601fe/ : SUCCESS in 11m 02s\n- cross-horizon-py35 http://logs.openstack.org/40/653340/2/check/cross-horizon-py35/98d8198/ : SUCCESS in 11m 48s\n- cross-keystone-py27 http://logs.openstack.org/40/653340/2/check/cross-keystone-py27/8678637/ : TIMED_OUT in 40m 59s\n- cross-keystone-py35 http://logs.openstack.org/40/653340/2/check/cross-keystone-py35/8635df5/ : SUCCESS in 25m 53s\n- cross-neutron-py27 http://logs.openstack.org/40/653340/2/check/cross-neutron-py27/cbfdf7d/ : SUCCESS in 21m 01s\n- cross-neutron-py35 http://logs.openstack.org/40/653340/2/check/cross-neutron-py35/c0cd4f7/ : SUCCESS in 23m 08s\n- cross-nova-py27 http://logs.openstack.org/40/653340/2/check/cross-nova-py27/fdf4304/ : SUCCESS in 11m 58s\n- cross-nova-py35 http://logs.openstack.org/40/653340/2/check/cross-nova-py35/e53a2cc/ : SUCCESS in 23m 24s\n- cross-nova-functional http://logs.openstack.org/40/653340/2/check/cross-nova-functional/d4eac04/ : SUCCESS in 23m 32s\n- cross-swift-py27 http://logs.openstack.org/40/653340/2/check/cross-swift-py27/18e1b3a/ : SUCCESS in 7m 40s\n- cross-swift-py35 http://logs.openstack.org/40/653340/2/check/cross-swift-py35/49032e3/ : SUCCESS in 3m 33s\n- requirements-integration http://logs.openstack.org/40/653340/2/check/requirements-integration/446836b/ : SUCCESS in 56m 04s\n- neutron-grenade-multinode http://logs.openstack.org/40/653340/2/check/neutron-grenade-multinode/06ab287/ : SUCCESS in 1h 10m 05s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/40/653340/2/check/legacy-tempest-dsvm-neutron-full/f2c62ae/ : SUCCESS in 1h 25m 55s\n- neutron-grenade http://logs.openstack.org/40/653340/2/check/neutron-grenade/611dafe/ : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":2},{"id":"3fe3ce26d9cabe65b1b0dffc3debe9c5f2b27bbe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 19:53:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cbeac64e2b45c9c3c2741d1a33f82b9830910f9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 00:06:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/40/653340/2/check/openstack-tox-pypy/67aee53/ : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/653340/2/check/openstack-tox-pep8/fde6e22/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/40/653340/2/check/openstack-tox-py27/8ea64b8/ : SUCCESS in 4m 08s\n- openstack-tox-py35 http://logs.openstack.org/40/653340/2/check/openstack-tox-py35/25fb2d6/ : SUCCESS in 4m 04s\n- openstack-tox-validate http://logs.openstack.org/40/653340/2/check/openstack-tox-validate/4a927fb/ : SUCCESS in 3m 18s\n- requirements-tox-babel http://logs.openstack.org/40/653340/2/check/requirements-tox-babel/47896bf/ : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/653340/2/check/requirements-tox-py27-check-uc/403919f/ : SUCCESS in 7m 20s\n- cross-cinder-py27 http://logs.openstack.org/40/653340/2/check/cross-cinder-py27/f6134cc/ : SUCCESS in 9m 33s\n- cross-cinder-py35 http://logs.openstack.org/40/653340/2/check/cross-cinder-py35/069c04c/ : SUCCESS in 7m 06s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/40/653340/2/check/cross-glance-py27/990ecdd/ : SUCCESS in 8m 48s\n- cross-glance-py35 http://logs.openstack.org/40/653340/2/check/cross-glance-py35/a5b84f6/ : SUCCESS in 9m 40s\n- cross-horizon-py27 http://logs.openstack.org/40/653340/2/check/cross-horizon-py27/42c6b7d/ : SUCCESS in 11m 08s\n- cross-horizon-py35 http://logs.openstack.org/40/653340/2/check/cross-horizon-py35/ddd3426/ : SUCCESS in 9m 55s\n- cross-keystone-py27 http://logs.openstack.org/40/653340/2/check/cross-keystone-py27/574c70e/ : SUCCESS in 11m 30s\n- cross-keystone-py35 http://logs.openstack.org/40/653340/2/check/cross-keystone-py35/4a286be/ : SUCCESS in 16m 51s\n- cross-neutron-py27 http://logs.openstack.org/40/653340/2/check/cross-neutron-py27/1878db7/ : SUCCESS in 24m 47s\n- cross-neutron-py35 http://logs.openstack.org/40/653340/2/check/cross-neutron-py35/3739d95/ : SUCCESS in 29m 19s\n- cross-nova-py27 http://logs.openstack.org/40/653340/2/check/cross-nova-py27/5e0ee05/ : SUCCESS in 11m 20s\n- cross-nova-py35 http://logs.openstack.org/40/653340/2/check/cross-nova-py35/67f3b73/ : SUCCESS in 12m 26s\n- cross-nova-functional http://logs.openstack.org/40/653340/2/check/cross-nova-functional/73363a5/ : SUCCESS in 18m 42s\n- cross-swift-py27 http://logs.openstack.org/40/653340/2/check/cross-swift-py27/f6bb024/ : SUCCESS in 6m 42s\n- cross-swift-py35 http://logs.openstack.org/40/653340/2/check/cross-swift-py35/826fac7/ : SUCCESS in 3m 25s\n- requirements-integration http://logs.openstack.org/40/653340/2/check/requirements-integration/120298d/ : SUCCESS in 53m 31s\n- neutron-grenade-multinode http://logs.openstack.org/40/653340/2/check/neutron-grenade-multinode/7e5fbd0/ : SUCCESS in 1h 01m 26s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/40/653340/2/check/legacy-tempest-dsvm-neutron-full/e945348/ : SUCCESS in 1h 22m 45s\n- neutron-grenade http://logs.openstack.org/40/653340/2/check/neutron-grenade/781e3d1/ : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":2},{"id":"9c0dbf0073bc21effa04eb3f396a2c00794467a0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-23 00:09:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e1e2c45eac23349250f17ef56fab55a7bdbac873","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-23 00:29:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0d343937f5efa3c12d1799045266e2d7c6de824e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 00:29:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"72589def3594e49e93710acb12043ac38bb6f31b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 03:55:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/40/653340/2/gate/openstack-tox-pep8/d1ec9de/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/40/653340/2/gate/openstack-tox-py27/5edac74/ : SUCCESS in 3m 41s\n- openstack-tox-py35 http://logs.openstack.org/40/653340/2/gate/openstack-tox-py35/4c76af9/ : SUCCESS in 4m 26s\n- openstack-tox-validate http://logs.openstack.org/40/653340/2/gate/openstack-tox-validate/2ac4726/ : SUCCESS in 5m 09s\n- requirements-tox-babel http://logs.openstack.org/40/653340/2/gate/requirements-tox-babel/5fafd9a/ : SUCCESS in 4m 13s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/653340/2/gate/requirements-tox-py27-check-uc/7692fc4/ : SUCCESS in 6m 58s\n- cross-cinder-py27 http://logs.openstack.org/40/653340/2/gate/cross-cinder-py27/d9141c9/ : SUCCESS in 8m 08s\n- cross-glance-py27 http://logs.openstack.org/40/653340/2/gate/cross-glance-py27/202fe47/ : SUCCESS in 9m 09s\n- cross-glance-py35 http://logs.openstack.org/40/653340/2/gate/cross-glance-py35/cb8e747/ : SUCCESS in 10m 01s\n- cross-horizon-py27 http://logs.openstack.org/40/653340/2/gate/cross-horizon-py27/9a9b8f0/ : SUCCESS in 10m 18s\n- cross-horizon-py35 http://logs.openstack.org/40/653340/2/gate/cross-horizon-py35/2fda2aa/ : SUCCESS in 10m 35s\n- cross-keystone-py27 http://logs.openstack.org/40/653340/2/gate/cross-keystone-py27/9c19200/ : SUCCESS in 11m 16s\n- cross-keystone-py35 http://logs.openstack.org/40/653340/2/gate/cross-keystone-py35/95b5c10/ : SUCCESS in 11m 18s\n- cross-neutron-py27 http://logs.openstack.org/40/653340/2/gate/cross-neutron-py27/e3aeb74/ : SUCCESS in 22m 10s\n- cross-neutron-py35 http://logs.openstack.org/40/653340/2/gate/cross-neutron-py35/2362538/ : SUCCESS in 24m 18s\n- cross-nova-py27 http://logs.openstack.org/40/653340/2/gate/cross-nova-py27/3c0e158/ : SUCCESS in 13m 33s\n- cross-nova-py35 http://logs.openstack.org/40/653340/2/gate/cross-nova-py35/ceb3c9f/ : SUCCESS in 12m 35s\n- cross-nova-functional http://logs.openstack.org/40/653340/2/gate/cross-nova-functional/b761983/ : SUCCESS in 20m 16s\n- cross-swift-py27 http://logs.openstack.org/40/653340/2/gate/cross-swift-py27/f0943ce/ : SUCCESS in 7m 24s\n- cross-swift-py35 http://logs.openstack.org/40/653340/2/gate/cross-swift-py35/69aa0a2/ : SUCCESS in 3m 42s\n- requirements-integration http://logs.openstack.org/40/653340/2/gate/requirements-integration/82c0b47/ : SUCCESS in 55m 46s\n- neutron-grenade-multinode http://logs.openstack.org/40/653340/2/gate/neutron-grenade-multinode/f9f8c2b/ : SUCCESS in 59m 45s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/40/653340/2/gate/legacy-tempest-dsvm-neutron-full/539c4f3/ : SUCCESS in 1h 39m 40s\n- neutron-grenade http://logs.openstack.org/40/653340/2/gate/neutron-grenade/c261192/ : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b9812c5df445c59aa05151fd9a04776deaa0add2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 03:55:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"28f2907b7023b7f0ba0afdfa474979a072d48803","revisions":{"06203981a3647953b53af6ce3c12d59dc0fbffae":{"kind":"REWORK","_number":1,"created":"2019-04-17 07:42:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/653340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/653340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/653340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/653340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/653340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/653340/1"}}},"commit":{"parents":[{"commit":"03eeccfc2edaaf995010da7a8c2b6b6188439b30","subject":"update constraint for os-brick to new release 1.15.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03eeccfc2edaaf995010da7a8c2b6b6188439b30"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 07:42:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 07:42:37.000000000","tz":0},"subject":"update constraint for python-senlinclient to new release 1.4.1","message":"update constraint for python-senlinclient to new release 1.4.1\n\nChange-Id: I3548aaf0c913a671b14a0e96b3c4cc2c77a65b8c\nmeta:version: 1.4.1\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Change-Id: Ie913b0efbb48512a447ac15d1bbe5384b564913e\nmeta:release:Code-Review+1: XueFeng Liu \u003cliu.xuefeng1@zte.com.cn\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06203981a3647953b53af6ce3c12d59dc0fbffae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06203981a3647953b53af6ce3c12d59dc0fbffae"}]},"branch":"refs/heads/stable/pike"},"28f2907b7023b7f0ba0afdfa474979a072d48803":{"kind":"REWORK","_number":2,"created":"2019-04-18 06:23:14.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/40/653340/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/653340/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/653340/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/653340/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/653340/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/653340/2"}}},"commit":{"parents":[{"commit":"3170a63a11028e9701bc0590fc15a44ede84d30f","subject":"Merge \"update constraint for oslo.reports to new release 1.22.2\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3170a63a11028e9701bc0590fc15a44ede84d30f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 07:42:37.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-04-18 06:23:05.000000000","tz":-300},"subject":"update constraint for python-senlinclient to new release 1.4.1","message":"update constraint for python-senlinclient to new release 1.4.1\n\nChange-Id: I3548aaf0c913a671b14a0e96b3c4cc2c77a65b8c\nmeta:version: 1.4.1\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Change-Id: Ie913b0efbb48512a447ac15d1bbe5384b564913e\nmeta:release:Code-Review+1: XueFeng Liu \u003cliu.xuefeng1@zte.com.cn\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28f2907b7023b7f0ba0afdfa474979a072d48803"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/28f2907b7023b7f0ba0afdfa474979a072d48803"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
