)]}'
{"id":"openstack%2Frequirements~653637","triplet_id":"openstack%2Frequirements~stable%2Fpike~Ie757136ef818cf5cb07381c21e149b57e8d949de","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"Ie757136ef818cf5cb07381c21e149b57e8d949de","subject":"update constraint for python-keystoneclient to new release 3.13.1","status":"MERGED","created":"2019-04-17 23:53:13.000000000","updated":"2019-04-23 04:16:28.000000000","submitted":"2019-04-23 04:16:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653637-1555992988748-4f1731f4","meta_rev_id":"6671b6731dfa31c2ad4318fd7d1952277e51a912","_number":653637,"virtual_id_number":653637,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-04-23 04:16:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-04-18 13:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-04-23 01:19:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-18 13:45:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-04-23 01:19:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-18 13:28:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-18 13:45:26.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-23 01:19:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-23 04:16:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce64d623110f09881e2ed7a1086eda98ab67d6c8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-17 23:53:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25ce3c27df643015b529ef0c4dfe0dd9138107ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 10:33:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/37/653637/1/check/openstack-tox-pypy/eac6117/ : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/37/653637/1/check/openstack-tox-pep8/a244f41/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/37/653637/1/check/openstack-tox-py27/d3f0ef4/ : SUCCESS in 6m 45s\n- openstack-tox-py35 http://logs.openstack.org/37/653637/1/check/openstack-tox-py35/e5fb5be/ : SUCCESS in 6m 30s\n- openstack-tox-validate http://logs.openstack.org/37/653637/1/check/openstack-tox-validate/6b6d936/ : SUCCESS in 3m 41s\n- requirements-tox-babel http://logs.openstack.org/37/653637/1/check/requirements-tox-babel/90d2ecb/ : SUCCESS in 5m 02s\n- requirements-tox-py27-check-uc http://logs.openstack.org/37/653637/1/check/requirements-tox-py27-check-uc/1baf358/ : SUCCESS in 7m 28s\n- cross-cinder-py27 http://logs.openstack.org/37/653637/1/check/cross-cinder-py27/3650593/ : SUCCESS in 9m 43s\n- cross-cinder-py35 http://logs.openstack.org/37/653637/1/check/cross-cinder-py35/361154f/ : SUCCESS in 11m 30s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/37/653637/1/check/cross-glance-py27/8f03bd3/ : SUCCESS in 9m 39s\n- cross-glance-py35 http://logs.openstack.org/37/653637/1/check/cross-glance-py35/2f90f84/ : SUCCESS in 13m 03s\n- cross-horizon-py27 http://logs.openstack.org/37/653637/1/check/cross-horizon-py27/562661d/ : SUCCESS in 12m 24s\n- cross-horizon-py35 http://logs.openstack.org/37/653637/1/check/cross-horizon-py35/e0f49a8/ : SUCCESS in 10m 12s\n- cross-keystone-py27 http://logs.openstack.org/37/653637/1/check/cross-keystone-py27/72b518e/ : SUCCESS in 16m 05s\n- cross-keystone-py35 http://logs.openstack.org/37/653637/1/check/cross-keystone-py35/d4e779a/ : SUCCESS in 13m 39s\n- cross-neutron-py27 http://logs.openstack.org/37/653637/1/check/cross-neutron-py27/4d403b2/ : SUCCESS in 24m 59s\n- cross-neutron-py35 http://logs.openstack.org/37/653637/1/check/cross-neutron-py35/81864d8/ : SUCCESS in 25m 05s\n- cross-nova-py27 http://logs.openstack.org/37/653637/1/check/cross-nova-py27/c433ad0/ : SUCCESS in 13m 46s\n- cross-nova-py35 http://logs.openstack.org/37/653637/1/check/cross-nova-py35/3c7a9af/ : SUCCESS in 16m 41s\n- cross-nova-functional http://logs.openstack.org/37/653637/1/check/cross-nova-functional/08ccc11/ : SUCCESS in 20m 03s\n- cross-swift-py27 http://logs.openstack.org/37/653637/1/check/cross-swift-py27/f60f5d0/ : SUCCESS in 8m 31s\n- cross-swift-py35 http://logs.openstack.org/37/653637/1/check/cross-swift-py35/122a193/ : SUCCESS in 6m 29s\n- requirements-integration http://logs.openstack.org/37/653637/1/check/requirements-integration/548a6b8/ : SUCCESS in 52m 23s\n- neutron-grenade-multinode http://logs.openstack.org/37/653637/1/check/neutron-grenade-multinode/9c99d4d/ : SUCCESS in 1h 07m 21s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/37/653637/1/check/legacy-tempest-dsvm-neutron-full/5b72d90/ : SUCCESS in 1h 33m 26s\n- neutron-grenade http://logs.openstack.org/37/653637/1/check/neutron-grenade/54f9aa1/ : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f5143533a6d2bd67335a84e61c421da8e7fd61a3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-18 13:28:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"380103a7f7ad7084aa897e1759c1a17935e192e5","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-18 13:45:26.000000000","message":"Patch Set 1: Code-Review+2\n\nlets not approve this until the rest is through the gate","accounts_in_message":[],"_revision_number":1},{"id":"672a3aa5f4281f48396e1d449c2638b0824d2ef7","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-18 20:42:03.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7e908515302f70b2d6c35ee48c54cf2134c4b72b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 20:43:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a1b33eb7b7d2fcd2bc3e5aa395a5fe28198deb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 00:47:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/37/653637/1/gate/openstack-tox-pep8/50064bb/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/37/653637/1/gate/openstack-tox-py27/b4d91b0/ : SUCCESS in 5m 51s\n- openstack-tox-py35 http://logs.openstack.org/37/653637/1/gate/openstack-tox-py35/54d9d6a/ : SUCCESS in 5m 35s\n- openstack-tox-validate http://logs.openstack.org/37/653637/1/gate/openstack-tox-validate/bfd6d52/ : SUCCESS in 5m 19s\n- requirements-tox-babel http://logs.openstack.org/37/653637/1/gate/requirements-tox-babel/2c00753/ : SUCCESS in 5m 04s\n- requirements-tox-py27-check-uc http://logs.openstack.org/37/653637/1/gate/requirements-tox-py27-check-uc/e551901/ : SUCCESS in 7m 53s\n- cross-cinder-py27 http://logs.openstack.org/37/653637/1/gate/cross-cinder-py27/93932be/ : SUCCESS in 10m 55s\n- cross-glance-py27 http://logs.openstack.org/37/653637/1/gate/cross-glance-py27/9ce168e/ : SUCCESS in 10m 56s\n- cross-glance-py35 http://logs.openstack.org/37/653637/1/gate/cross-glance-py35/f0cad9c/ : SUCCESS in 11m 25s\n- cross-horizon-py27 http://logs.openstack.org/37/653637/1/gate/cross-horizon-py27/0ccfd17/ : SUCCESS in 12m 30s\n- cross-horizon-py35 http://logs.openstack.org/37/653637/1/gate/cross-horizon-py35/0840b84/ : SUCCESS in 12m 12s\n- cross-keystone-py27 http://logs.openstack.org/37/653637/1/gate/cross-keystone-py27/8b71577/ : SUCCESS in 15m 05s\n- cross-keystone-py35 http://logs.openstack.org/37/653637/1/gate/cross-keystone-py35/1bb11b3/ : SUCCESS in 23m 11s\n- cross-neutron-py27 http://logs.openstack.org/37/653637/1/gate/cross-neutron-py27/96826bd/ : SUCCESS in 26m 15s\n- cross-neutron-py35 http://logs.openstack.org/37/653637/1/gate/cross-neutron-py35/222e4f3/ : SUCCESS in 22m 19s\n- cross-nova-py27 http://logs.openstack.org/37/653637/1/gate/cross-nova-py27/e264c04/ : SUCCESS in 13m 20s\n- cross-nova-py35 http://logs.openstack.org/37/653637/1/gate/cross-nova-py35/a381056/ : SUCCESS in 13m 56s\n- cross-nova-functional http://logs.openstack.org/37/653637/1/gate/cross-nova-functional/903a14d/ : SUCCESS in 20m 00s\n- cross-swift-py27 http://logs.openstack.org/37/653637/1/gate/cross-swift-py27/a97c30e/ : SUCCESS in 11m 04s\n- cross-swift-py35 http://logs.openstack.org/37/653637/1/gate/cross-swift-py35/6c35e0e/ : SUCCESS in 5m 00s\n- requirements-integration http://logs.openstack.org/37/653637/1/gate/requirements-integration/70e6844/ : SUCCESS in 44m 11s\n- neutron-grenade-multinode http://logs.openstack.org/37/653637/1/gate/neutron-grenade-multinode/320f3bd/ : SUCCESS in 50m 50s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/37/653637/1/gate/legacy-tempest-dsvm-neutron-full/ee87056/ : FAILURE in 1h 31m 54s\n- neutron-grenade http://logs.openstack.org/37/653637/1/gate/neutron-grenade/ac3d374/ : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7e647e2a667d756ab4c6aace41b880936bc721d1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-19 06:44:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"297d7a77bab4581b4a34ce26adc54cac2d6b0ff4","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-20 08:26:29.000000000","message":"Patch Set 1: -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a62628ee4236c1674daeb25745af8885641c38c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 09:00:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/37/653637/1/check/openstack-tox-pypy/e63f9e8/ : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/37/653637/1/check/openstack-tox-pep8/260e8c8/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/37/653637/1/check/openstack-tox-py27/867c286/ : SUCCESS in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/37/653637/1/check/openstack-tox-py35/ef806a9/ : SUCCESS in 3m 26s\n- openstack-tox-validate http://logs.openstack.org/37/653637/1/check/openstack-tox-validate/2db9dd5/ : SUCCESS in 3m 29s\n- requirements-tox-babel http://logs.openstack.org/37/653637/1/check/requirements-tox-babel/5dbe379/ : SUCCESS in 3m 52s\n- requirements-tox-py27-check-uc http://logs.openstack.org/37/653637/1/check/requirements-tox-py27-check-uc/7e4f1db/ : SUCCESS in 6m 07s\n- cross-cinder-py27 http://logs.openstack.org/37/653637/1/check/cross-cinder-py27/82162a4/ : SUCCESS in 9m 27s\n- cross-cinder-py35 http://logs.openstack.org/37/653637/1/check/cross-cinder-py35/91dda3a/ : SUCCESS in 7m 33s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/37/653637/1/check/cross-glance-py27/529541a/ : SUCCESS in 11m 40s\n- cross-glance-py35 http://logs.openstack.org/37/653637/1/check/cross-glance-py35/ec2dfef/ : SUCCESS in 9m 52s\n- cross-horizon-py27 http://logs.openstack.org/37/653637/1/check/cross-horizon-py27/6affa87/ : SUCCESS in 10m 05s\n- cross-horizon-py35 http://logs.openstack.org/37/653637/1/check/cross-horizon-py35/56c09d1/ : SUCCESS in 10m 41s\n- cross-keystone-py27 http://logs.openstack.org/37/653637/1/check/cross-keystone-py27/87eec1d/ : SUCCESS in 15m 34s\n- cross-keystone-py35 http://logs.openstack.org/37/653637/1/check/cross-keystone-py35/5f53bce/ : SUCCESS in 32m 43s\n- cross-neutron-py27 http://logs.openstack.org/37/653637/1/check/cross-neutron-py27/a5a8085/ : SUCCESS in 21m 11s\n- cross-neutron-py35 http://logs.openstack.org/37/653637/1/check/cross-neutron-py35/5060303/ : SUCCESS in 24m 21s\n- cross-nova-py27 http://logs.openstack.org/37/653637/1/check/cross-nova-py27/8206f3c/ : SUCCESS in 16m 18s\n- cross-nova-py35 http://logs.openstack.org/37/653637/1/check/cross-nova-py35/5838a19/ : SUCCESS in 18m 03s\n- cross-nova-functional http://logs.openstack.org/37/653637/1/check/cross-nova-functional/2432969/ : SUCCESS in 19m 46s\n- cross-swift-py27 http://logs.openstack.org/37/653637/1/check/cross-swift-py27/a81cda7/ : SUCCESS in 14m 19s\n- cross-swift-py35 http://logs.openstack.org/37/653637/1/check/cross-swift-py35/74a2471/ : SUCCESS in 3m 53s\n- requirements-integration http://logs.openstack.org/37/653637/1/check/requirements-integration/2942a11/ : FAILURE in 7m 55s\n- neutron-grenade-multinode http://logs.openstack.org/37/653637/1/check/neutron-grenade-multinode/3fb6d39/ : FAILURE in 11m 05s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/37/653637/1/check/legacy-tempest-dsvm-neutron-full/80058b9/ : FAILURE in 4m 47s\n- neutron-grenade http://logs.openstack.org/37/653637/1/check/neutron-grenade/01182d4/ : FAILURE in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ab137e79e4d67959d532d94b6b6ad7b8553e4354","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 14:55:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"56b620c90d679418bc396813a747cad585a4d11f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 16:19:18.000000000","message":"Patch Set 1:\n\nrecheck (and unvote til gate works again)","accounts_in_message":[],"_revision_number":1},{"id":"f382d77f25edcdfe44526645939de39a1477c776","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 19:46:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/37/653637/1/check/openstack-tox-pypy/47721a1/ : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/37/653637/1/check/openstack-tox-pep8/68a4af8/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/37/653637/1/check/openstack-tox-py27/45436d1/ : SUCCESS in 3m 44s\n- openstack-tox-py35 http://logs.openstack.org/37/653637/1/check/openstack-tox-py35/86347b9/ : SUCCESS in 3m 43s\n- openstack-tox-validate http://logs.openstack.org/37/653637/1/check/openstack-tox-validate/5d3b744/ : SUCCESS in 3m 45s\n- requirements-tox-babel http://logs.openstack.org/37/653637/1/check/requirements-tox-babel/f3806ea/ : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc http://logs.openstack.org/37/653637/1/check/requirements-tox-py27-check-uc/02b06b4/ : SUCCESS in 7m 12s\n- cross-cinder-py27 http://logs.openstack.org/37/653637/1/check/cross-cinder-py27/e8bf2c7/ : SUCCESS in 8m 29s\n- cross-cinder-py35 http://logs.openstack.org/37/653637/1/check/cross-cinder-py35/5d2cb3f/ : SUCCESS in 8m 44s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/37/653637/1/check/cross-glance-py27/ee415ae/ : FAILURE in 15m 19s\n- cross-glance-py35 http://logs.openstack.org/37/653637/1/check/cross-glance-py35/697564e/ : SUCCESS in 9m 35s\n- cross-horizon-py27 http://logs.openstack.org/37/653637/1/check/cross-horizon-py27/fe40cee/ : SUCCESS in 10m 24s\n- cross-horizon-py35 http://logs.openstack.org/37/653637/1/check/cross-horizon-py35/b42f363/ : SUCCESS in 10m 11s\n- cross-keystone-py27 http://logs.openstack.org/37/653637/1/check/cross-keystone-py27/1bb02ad/ : SUCCESS in 13m 08s\n- cross-keystone-py35 http://logs.openstack.org/37/653637/1/check/cross-keystone-py35/7d5a515/ : SUCCESS in 12m 42s\n- cross-neutron-py27 http://logs.openstack.org/37/653637/1/check/cross-neutron-py27/20aa89c/ : SUCCESS in 24m 34s\n- cross-neutron-py35 http://logs.openstack.org/37/653637/1/check/cross-neutron-py35/143e70f/ : SUCCESS in 21m 34s\n- cross-nova-py27 http://logs.openstack.org/37/653637/1/check/cross-nova-py27/40e3ad7/ : SUCCESS in 12m 48s\n- cross-nova-py35 http://logs.openstack.org/37/653637/1/check/cross-nova-py35/88ec2c0/ : SUCCESS in 13m 38s\n- cross-nova-functional http://logs.openstack.org/37/653637/1/check/cross-nova-functional/47de257/ : SUCCESS in 20m 47s\n- cross-swift-py27 http://logs.openstack.org/37/653637/1/check/cross-swift-py27/7b216ce/ : SUCCESS in 7m 55s\n- cross-swift-py35 http://logs.openstack.org/37/653637/1/check/cross-swift-py35/e60cb70/ : SUCCESS in 3m 27s\n- requirements-integration http://logs.openstack.org/37/653637/1/check/requirements-integration/7ee82ee/ : SUCCESS in 53m 36s\n- neutron-grenade-multinode http://logs.openstack.org/37/653637/1/check/neutron-grenade-multinode/99fb234/ : SUCCESS in 1h 11m 45s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/37/653637/1/check/legacy-tempest-dsvm-neutron-full/93ae672/ : SUCCESS in 1h 29m 12s\n- neutron-grenade http://logs.openstack.org/37/653637/1/check/neutron-grenade/d863204/ : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":1},{"id":"14943a555ed0092a1682f5cdeacb1078caf5b78e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 19:54:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d143d677b29ffcb00123594f728666be65a3089","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 01:04:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/37/653637/1/check/openstack-tox-pypy/db135b1/ : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/37/653637/1/check/openstack-tox-pep8/796585c/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/37/653637/1/check/openstack-tox-py27/f027909/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/37/653637/1/check/openstack-tox-py35/786121f/ : SUCCESS in 3m 53s\n- openstack-tox-validate http://logs.openstack.org/37/653637/1/check/openstack-tox-validate/080b3c8/ : SUCCESS in 3m 30s\n- requirements-tox-babel http://logs.openstack.org/37/653637/1/check/requirements-tox-babel/51c6194/ : SUCCESS in 3m 43s\n- requirements-tox-py27-check-uc http://logs.openstack.org/37/653637/1/check/requirements-tox-py27-check-uc/489e982/ : SUCCESS in 6m 24s\n- cross-cinder-py27 http://logs.openstack.org/37/653637/1/check/cross-cinder-py27/8d7603f/ : SUCCESS in 7m 46s\n- cross-cinder-py35 http://logs.openstack.org/37/653637/1/check/cross-cinder-py35/2efeaa7/ : SUCCESS in 8m 01s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/37/653637/1/check/cross-glance-py27/4ed32a0/ : SUCCESS in 11m 06s\n- cross-glance-py35 http://logs.openstack.org/37/653637/1/check/cross-glance-py35/5516c41/ : SUCCESS in 9m 13s\n- cross-horizon-py27 http://logs.openstack.org/37/653637/1/check/cross-horizon-py27/5c65a4e/ : SUCCESS in 11m 22s\n- cross-horizon-py35 http://logs.openstack.org/37/653637/1/check/cross-horizon-py35/3dd3b61/ : SUCCESS in 10m 02s\n- cross-keystone-py27 http://logs.openstack.org/37/653637/1/check/cross-keystone-py27/41d638f/ : SUCCESS in 17m 40s\n- cross-keystone-py35 http://logs.openstack.org/37/653637/1/check/cross-keystone-py35/4125618/ : SUCCESS in 11m 21s\n- cross-neutron-py27 http://logs.openstack.org/37/653637/1/check/cross-neutron-py27/8b92b0a/ : SUCCESS in 23m 56s\n- cross-neutron-py35 http://logs.openstack.org/37/653637/1/check/cross-neutron-py35/00e9c11/ : SUCCESS in 21m 46s\n- cross-nova-py27 http://logs.openstack.org/37/653637/1/check/cross-nova-py27/674fe2d/ : SUCCESS in 12m 24s\n- cross-nova-py35 http://logs.openstack.org/37/653637/1/check/cross-nova-py35/e95fae5/ : SUCCESS in 14m 14s\n- cross-nova-functional http://logs.openstack.org/37/653637/1/check/cross-nova-functional/ae603ba/ : SUCCESS in 19m 52s\n- cross-swift-py27 http://logs.openstack.org/37/653637/1/check/cross-swift-py27/a41e7de/ : SUCCESS in 8m 33s\n- cross-swift-py35 http://logs.openstack.org/37/653637/1/check/cross-swift-py35/be10a60/ : SUCCESS in 3m 48s\n- requirements-integration http://logs.openstack.org/37/653637/1/check/requirements-integration/65d66a2/ : SUCCESS in 47m 50s\n- neutron-grenade-multinode http://logs.openstack.org/37/653637/1/check/neutron-grenade-multinode/56606c6/ : SUCCESS in 57m 07s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/37/653637/1/check/legacy-tempest-dsvm-neutron-full/0df0c9e/ : SUCCESS in 1h 23m 39s\n- neutron-grenade http://logs.openstack.org/37/653637/1/check/neutron-grenade/47d56ca/ : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f464947bd77a17d2c7ab91aea71ce0e2541e54e2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-23 01:19:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a169c0f77ac027dcd861e6f3a2dea3f052aa771c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 01:19:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45a50a88534b0f62023cce7f30e984c726e4cb6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 04:16:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/37/653637/1/gate/openstack-tox-pep8/fe55ef0/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/37/653637/1/gate/openstack-tox-py27/e72a8f7/ : SUCCESS in 5m 34s\n- openstack-tox-py35 http://logs.openstack.org/37/653637/1/gate/openstack-tox-py35/399b187/ : SUCCESS in 3m 37s\n- openstack-tox-validate http://logs.openstack.org/37/653637/1/gate/openstack-tox-validate/42f776b/ : SUCCESS in 3m 14s\n- requirements-tox-babel http://logs.openstack.org/37/653637/1/gate/requirements-tox-babel/39812c4/ : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/37/653637/1/gate/requirements-tox-py27-check-uc/a19c341/ : SUCCESS in 6m 31s\n- cross-cinder-py27 http://logs.openstack.org/37/653637/1/gate/cross-cinder-py27/18919ec/ : SUCCESS in 8m 43s\n- cross-glance-py27 http://logs.openstack.org/37/653637/1/gate/cross-glance-py27/c593b4a/ : SUCCESS in 9m 21s\n- cross-glance-py35 http://logs.openstack.org/37/653637/1/gate/cross-glance-py35/f791784/ : SUCCESS in 10m 13s\n- cross-horizon-py27 http://logs.openstack.org/37/653637/1/gate/cross-horizon-py27/8232f14/ : SUCCESS in 10m 26s\n- cross-horizon-py35 http://logs.openstack.org/37/653637/1/gate/cross-horizon-py35/446ed70/ : SUCCESS in 10m 26s\n- cross-keystone-py27 http://logs.openstack.org/37/653637/1/gate/cross-keystone-py27/5b93ba5/ : SUCCESS in 12m 41s\n- cross-keystone-py35 http://logs.openstack.org/37/653637/1/gate/cross-keystone-py35/f991151/ : SUCCESS in 11m 23s\n- cross-neutron-py27 http://logs.openstack.org/37/653637/1/gate/cross-neutron-py27/bac1221/ : SUCCESS in 22m 29s\n- cross-neutron-py35 http://logs.openstack.org/37/653637/1/gate/cross-neutron-py35/74ed396/ : SUCCESS in 23m 25s\n- cross-nova-py27 http://logs.openstack.org/37/653637/1/gate/cross-nova-py27/7307637/ : SUCCESS in 13m 50s\n- cross-nova-py35 http://logs.openstack.org/37/653637/1/gate/cross-nova-py35/a406e8d/ : SUCCESS in 16m 45s\n- cross-nova-functional http://logs.openstack.org/37/653637/1/gate/cross-nova-functional/5c02911/ : SUCCESS in 21m 17s\n- cross-swift-py27 http://logs.openstack.org/37/653637/1/gate/cross-swift-py27/ae6f395/ : SUCCESS in 7m 04s\n- cross-swift-py35 http://logs.openstack.org/37/653637/1/gate/cross-swift-py35/78d5ff8/ : SUCCESS in 5m 05s\n- requirements-integration http://logs.openstack.org/37/653637/1/gate/requirements-integration/87490f5/ : SUCCESS in 56m 11s\n- neutron-grenade-multinode http://logs.openstack.org/37/653637/1/gate/neutron-grenade-multinode/a5ce7e3/ : SUCCESS in 55m 04s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/37/653637/1/gate/legacy-tempest-dsvm-neutron-full/563b1e2/ : SUCCESS in 1h 38m 23s\n- neutron-grenade http://logs.openstack.org/37/653637/1/gate/neutron-grenade/d940f0a/ : SUCCESS in 47m 30s","accounts_in_message":[],"_revision_number":1},{"id":"510480ba2ae45bf1771ed54e967fc95a7cb3360c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 04:16:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11d258fa51a05177950bb72656fb690743a3196b","revisions":{"11d258fa51a05177950bb72656fb690743a3196b":{"kind":"REWORK","_number":1,"created":"2019-04-17 23:53:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/653637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/653637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/653637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/653637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/653637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/653637/1"}}},"commit":{"parents":[{"commit":"3170a63a11028e9701bc0590fc15a44ede84d30f","subject":"Merge \"update constraint for oslo.reports to new release 1.22.2\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3170a63a11028e9701bc0590fc15a44ede84d30f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 23:53:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 23:53:11.000000000","tz":0},"subject":"update constraint for python-keystoneclient to new release 3.13.1","message":"update constraint for python-keystoneclient to new release 3.13.1\n\nChange-Id: Ie757136ef818cf5cb07381c21e149b57e8d949de\nmeta:version: 3.13.1\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Change-Id: I0ea3e30572fb745950c0eca0770f96bb87e0a53b\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11d258fa51a05177950bb72656fb690743a3196b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11d258fa51a05177950bb72656fb690743a3196b"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
