)]}'
{"id":"openstack%2Frequirements~653638","triplet_id":"openstack%2Frequirements~stable%2Fpike~I1e586ba768b427a6ce1a3785657b5d9d28d6606d","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"I1e586ba768b427a6ce1a3785657b5d9d28d6606d","subject":"update constraint for pycadf to new release 2.6.1","status":"MERGED","created":"2019-04-17 23:53:28.000000000","updated":"2019-04-23 08:50:44.000000000","submitted":"2019-04-23 08:50:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653638-1556009444960-4f1731f4","meta_rev_id":"5bf95f5f7f5f1564e29a6bec57030916bf913d9b","_number":653638,"virtual_id_number":653638,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-04-23 00:47:41.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-04-23 08:50:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-04-18 13:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-18 13:45:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-04-23 07:09:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-18 13:28:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-23 00:47:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-23 07:09:30.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-23 08:50:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dcf4480d5f8669b970199bddd21e2213cc277f3a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-17 23:53:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b706fd9680391e66065f077118ae0e0717776c8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 12:11:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/38/653638/1/check/openstack-tox-pypy/99928aa/ : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/38/653638/1/check/openstack-tox-pep8/e976b16/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/38/653638/1/check/openstack-tox-py27/d6e309f/ : SUCCESS in 6m 25s\n- openstack-tox-py35 http://logs.openstack.org/38/653638/1/check/openstack-tox-py35/d7c86cb/ : SUCCESS in 4m 02s\n- openstack-tox-validate http://logs.openstack.org/38/653638/1/check/openstack-tox-validate/b0d12d2/ : SUCCESS in 3m 34s\n- requirements-tox-babel http://logs.openstack.org/38/653638/1/check/requirements-tox-babel/8205b2b/ : SUCCESS in 3m 14s\n- requirements-tox-py27-check-uc http://logs.openstack.org/38/653638/1/check/requirements-tox-py27-check-uc/4d20976/ : SUCCESS in 9m 38s\n- cross-cinder-py27 http://logs.openstack.org/38/653638/1/check/cross-cinder-py27/64eea7b/ : SUCCESS in 8m 15s\n- cross-cinder-py35 http://logs.openstack.org/38/653638/1/check/cross-cinder-py35/281d4c4/ : SUCCESS in 8m 57s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/38/653638/1/check/cross-glance-py27/6e86830/ : SUCCESS in 10m 54s\n- cross-glance-py35 http://logs.openstack.org/38/653638/1/check/cross-glance-py35/797dcc7/ : SUCCESS in 9m 09s\n- cross-horizon-py27 http://logs.openstack.org/38/653638/1/check/cross-horizon-py27/089bfe6/ : SUCCESS in 11m 11s\n- cross-horizon-py35 http://logs.openstack.org/38/653638/1/check/cross-horizon-py35/f91ca1f/ : SUCCESS in 11m 48s\n- cross-keystone-py27 http://logs.openstack.org/38/653638/1/check/cross-keystone-py27/91e9ab9/ : SUCCESS in 13m 53s\n- cross-keystone-py35 http://logs.openstack.org/38/653638/1/check/cross-keystone-py35/fd644a8/ : SUCCESS in 12m 37s\n- cross-neutron-py27 http://logs.openstack.org/38/653638/1/check/cross-neutron-py27/f3cf069/ : SUCCESS in 21m 25s\n- cross-neutron-py35 http://logs.openstack.org/38/653638/1/check/cross-neutron-py35/381f5f9/ : SUCCESS in 22m 55s\n- cross-nova-py27 http://logs.openstack.org/38/653638/1/check/cross-nova-py27/8f5414a/ : SUCCESS in 12m 30s\n- cross-nova-py35 http://logs.openstack.org/38/653638/1/check/cross-nova-py35/c138441/ : SUCCESS in 13m 13s\n- cross-nova-functional http://logs.openstack.org/38/653638/1/check/cross-nova-functional/eecb252/ : SUCCESS in 27m 53s\n- cross-swift-py27 http://logs.openstack.org/38/653638/1/check/cross-swift-py27/b25ded0/ : SUCCESS in 9m 11s\n- cross-swift-py35 http://logs.openstack.org/38/653638/1/check/cross-swift-py35/e43f46e/ : SUCCESS in 4m 04s\n- requirements-integration http://logs.openstack.org/38/653638/1/check/requirements-integration/5dd39d6/ : SUCCESS in 57m 13s\n- neutron-grenade-multinode http://logs.openstack.org/38/653638/1/check/neutron-grenade-multinode/489ea26/ : SUCCESS in 1h 01m 38s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/38/653638/1/check/legacy-tempest-dsvm-neutron-full/be2b460/ : SUCCESS in 1h 35m 51s\n- neutron-grenade http://logs.openstack.org/38/653638/1/check/neutron-grenade/ce25510/ : SUCCESS in 50m 22s","accounts_in_message":[],"_revision_number":1},{"id":"198d4eca7287693881fa0646ade252e896df1820","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-18 13:28:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f9cc0496bb2518dc951a659362860ad1c19b5e42","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-18 13:45:36.000000000","message":"Patch Set 1: Code-Review+2\n\nlets not approve this until the rest is through the gate","accounts_in_message":[],"_revision_number":1},{"id":"715d2f17900a02cdb8cd7ceb812b265aadba50be","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-18 20:42:23.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f02a2489cfa296b0f8dde47052fa19be85314c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 20:43:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b66a4484ecee4f61c9665cdc647f4fe71f0edcf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 04:09:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/38/653638/1/gate/openstack-tox-pep8/abc2e4a/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/38/653638/1/gate/openstack-tox-py27/0237acf/ : SUCCESS in 5m 33s\n- openstack-tox-py35 http://logs.openstack.org/38/653638/1/gate/openstack-tox-py35/cfa7f80/ : SUCCESS in 4m 58s\n- openstack-tox-validate http://logs.openstack.org/38/653638/1/gate/openstack-tox-validate/e68bbe1/ : SUCCESS in 4m 17s\n- requirements-tox-babel http://logs.openstack.org/38/653638/1/gate/requirements-tox-babel/250eac6/ : SUCCESS in 4m 49s\n- requirements-tox-py27-check-uc http://logs.openstack.org/38/653638/1/gate/requirements-tox-py27-check-uc/f81ffa0/ : SUCCESS in 6m 58s\n- cross-cinder-py27 http://logs.openstack.org/38/653638/1/gate/cross-cinder-py27/be0df09/ : SUCCESS in 10m 38s\n- cross-glance-py27 http://logs.openstack.org/38/653638/1/gate/cross-glance-py27/dd4f47e/ : SUCCESS in 10m 01s\n- cross-glance-py35 http://logs.openstack.org/38/653638/1/gate/cross-glance-py35/8f9006e/ : SUCCESS in 11m 41s\n- cross-horizon-py27 http://logs.openstack.org/38/653638/1/gate/cross-horizon-py27/c2f1f2c/ : SUCCESS in 12m 40s\n- cross-horizon-py35 http://logs.openstack.org/38/653638/1/gate/cross-horizon-py35/2e36fb1/ : SUCCESS in 11m 30s\n- cross-keystone-py27 http://logs.openstack.org/38/653638/1/gate/cross-keystone-py27/2e056c5/ : SUCCESS in 14m 06s\n- cross-keystone-py35 http://logs.openstack.org/38/653638/1/gate/cross-keystone-py35/f801512/ : SUCCESS in 13m 26s\n- cross-neutron-py27 http://logs.openstack.org/38/653638/1/gate/cross-neutron-py27/a13e894/ : SUCCESS in 23m 21s\n- cross-neutron-py35 http://logs.openstack.org/38/653638/1/gate/cross-neutron-py35/86a1373/ : SUCCESS in 23m 34s\n- cross-nova-py27 http://logs.openstack.org/38/653638/1/gate/cross-nova-py27/f7176a1/ : SUCCESS in 14m 31s\n- cross-nova-py35 http://logs.openstack.org/38/653638/1/gate/cross-nova-py35/0ac1e85/ : SUCCESS in 13m 20s\n- cross-nova-functional http://logs.openstack.org/38/653638/1/gate/cross-nova-functional/017aed1/ : SUCCESS in 20m 44s\n- cross-swift-py27 http://logs.openstack.org/38/653638/1/gate/cross-swift-py27/533a5e6/ : SUCCESS in 7m 49s\n- cross-swift-py35 http://logs.openstack.org/38/653638/1/gate/cross-swift-py35/564a55e/ : SUCCESS in 3m 42s\n- requirements-integration finger://ze01.openstack.org/b3597b6507614b00bfbd693971f156b7 : RETRY_LIMIT in 1h 06m 57s\n- neutron-grenade-multinode http://logs.openstack.org/38/653638/1/gate/neutron-grenade-multinode/4b0979d/ : SUCCESS in 52m 24s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/38/653638/1/gate/legacy-tempest-dsvm-neutron-full/9684f98/ : FAILURE in 1h 26m 32s\n- neutron-grenade http://logs.openstack.org/38/653638/1/gate/neutron-grenade/2e780b7/ : SUCCESS in 55m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a3b8377501f40cfa90bab2ed00cac2c3bcc4028e","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-20 08:26:40.000000000","message":"Patch Set 1: -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"50828946b54eb27ad4d06afff9120efe6139c524","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 08:59:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/38/653638/1/check/openstack-tox-pypy/cf37584/ : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/38/653638/1/check/openstack-tox-pep8/f104fa3/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/38/653638/1/check/openstack-tox-py27/6d0809a/ : SUCCESS in 4m 41s\n- openstack-tox-py35 http://logs.openstack.org/38/653638/1/check/openstack-tox-py35/e4c5ebc/ : SUCCESS in 3m 36s\n- openstack-tox-validate http://logs.openstack.org/38/653638/1/check/openstack-tox-validate/8646703/ : SUCCESS in 3m 37s\n- requirements-tox-babel http://logs.openstack.org/38/653638/1/check/requirements-tox-babel/5560e4b/ : SUCCESS in 3m 59s\n- requirements-tox-py27-check-uc http://logs.openstack.org/38/653638/1/check/requirements-tox-py27-check-uc/2c8e129/ : SUCCESS in 6m 11s\n- cross-cinder-py27 http://logs.openstack.org/38/653638/1/check/cross-cinder-py27/4aa9097/ : SUCCESS in 10m 06s\n- cross-cinder-py35 http://logs.openstack.org/38/653638/1/check/cross-cinder-py35/ca013f2/ : SUCCESS in 7m 53s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/38/653638/1/check/cross-glance-py27/8771f39/ : SUCCESS in 9m 49s\n- cross-glance-py35 http://logs.openstack.org/38/653638/1/check/cross-glance-py35/904cb04/ : SUCCESS in 11m 56s\n- cross-horizon-py27 http://logs.openstack.org/38/653638/1/check/cross-horizon-py27/a83a58c/ : SUCCESS in 9m 07s\n- cross-horizon-py35 http://logs.openstack.org/38/653638/1/check/cross-horizon-py35/7fede3e/ : SUCCESS in 9m 49s\n- cross-keystone-py27 http://logs.openstack.org/38/653638/1/check/cross-keystone-py27/8e08dbf/ : SUCCESS in 13m 40s\n- cross-keystone-py35 http://logs.openstack.org/38/653638/1/check/cross-keystone-py35/be4898a/ : SUCCESS in 31m 15s\n- cross-neutron-py27 http://logs.openstack.org/38/653638/1/check/cross-neutron-py27/fceb51a/ : SUCCESS in 27m 28s\n- cross-neutron-py35 http://logs.openstack.org/38/653638/1/check/cross-neutron-py35/f1c8f56/ : SUCCESS in 23m 08s\n- cross-nova-py27 http://logs.openstack.org/38/653638/1/check/cross-nova-py27/fb78331/ : SUCCESS in 11m 51s\n- cross-nova-py35 http://logs.openstack.org/38/653638/1/check/cross-nova-py35/95ac4b2/ : SUCCESS in 13m 45s\n- cross-nova-functional http://logs.openstack.org/38/653638/1/check/cross-nova-functional/f731256/ : SUCCESS in 18m 56s\n- cross-swift-py27 http://logs.openstack.org/38/653638/1/check/cross-swift-py27/467d6b3/ : SUCCESS in 9m 56s\n- cross-swift-py35 http://logs.openstack.org/38/653638/1/check/cross-swift-py35/79849a0/ : SUCCESS in 3m 41s\n- requirements-integration http://logs.openstack.org/38/653638/1/check/requirements-integration/5d4543f/ : FAILURE in 9m 52s\n- neutron-grenade-multinode http://logs.openstack.org/38/653638/1/check/neutron-grenade-multinode/474c661/ : FAILURE in 6m 22s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/38/653638/1/check/legacy-tempest-dsvm-neutron-full/08a11d5/ : FAILURE in 5m 46s\n- neutron-grenade http://logs.openstack.org/38/653638/1/check/neutron-grenade/41ab2c0/ : FAILURE in 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8a23f0b96a38e8084d4cb565a24626797caea242","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 14:55:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9de5b8a9416ad21a41b31db8ace1d3f924f5be84","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 16:19:16.000000000","message":"Patch Set 1:\n\nrecheck (and unvote til gate works again)","accounts_in_message":[],"_revision_number":1},{"id":"18551362c5da9e3ce4fcf1f504a51b42c501a460","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 19:41:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/38/653638/1/check/openstack-tox-pypy/721ee36/ : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/38/653638/1/check/openstack-tox-pep8/3cc5811/ : SUCCESS in 3m 49s\n- openstack-tox-py27 http://logs.openstack.org/38/653638/1/check/openstack-tox-py27/cbc606b/ : SUCCESS in 5m 05s\n- openstack-tox-py35 http://logs.openstack.org/38/653638/1/check/openstack-tox-py35/3945725/ : SUCCESS in 4m 00s\n- openstack-tox-validate http://logs.openstack.org/38/653638/1/check/openstack-tox-validate/b739891/ : SUCCESS in 4m 38s\n- requirements-tox-babel http://logs.openstack.org/38/653638/1/check/requirements-tox-babel/d2e2ac6/ : SUCCESS in 4m 23s\n- requirements-tox-py27-check-uc http://logs.openstack.org/38/653638/1/check/requirements-tox-py27-check-uc/a0ad003/ : SUCCESS in 6m 52s\n- cross-cinder-py27 http://logs.openstack.org/38/653638/1/check/cross-cinder-py27/17257d8/ : SUCCESS in 7m 34s\n- cross-cinder-py35 http://logs.openstack.org/38/653638/1/check/cross-cinder-py35/2bb5e26/ : SUCCESS in 9m 16s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/38/653638/1/check/cross-glance-py27/0b18d02/ : SUCCESS in 12m 28s\n- cross-glance-py35 http://logs.openstack.org/38/653638/1/check/cross-glance-py35/02419a8/ : SUCCESS in 9m 26s\n- cross-horizon-py27 http://logs.openstack.org/38/653638/1/check/cross-horizon-py27/c3d00dc/ : SUCCESS in 9m 48s\n- cross-horizon-py35 http://logs.openstack.org/38/653638/1/check/cross-horizon-py35/f7fe7ec/ : SUCCESS in 10m 43s\n- cross-keystone-py27 http://logs.openstack.org/38/653638/1/check/cross-keystone-py27/3920858/ : TIMED_OUT in 40m 49s\n- cross-keystone-py35 http://logs.openstack.org/38/653638/1/check/cross-keystone-py35/a7a7ec8/ : SUCCESS in 15m 51s\n- cross-neutron-py27 http://logs.openstack.org/38/653638/1/check/cross-neutron-py27/5acc421/ : SUCCESS in 22m 16s\n- cross-neutron-py35 http://logs.openstack.org/38/653638/1/check/cross-neutron-py35/6f8bf13/ : SUCCESS in 27m 54s\n- cross-nova-py27 http://logs.openstack.org/38/653638/1/check/cross-nova-py27/c9f79c6/ : SUCCESS in 11m 40s\n- cross-nova-py35 http://logs.openstack.org/38/653638/1/check/cross-nova-py35/2d3ecda/ : SUCCESS in 11m 47s\n- cross-nova-functional http://logs.openstack.org/38/653638/1/check/cross-nova-functional/8ac1ea3/ : SUCCESS in 20m 13s\n- cross-swift-py27 http://logs.openstack.org/38/653638/1/check/cross-swift-py27/2047715/ : SUCCESS in 6m 53s\n- cross-swift-py35 http://logs.openstack.org/38/653638/1/check/cross-swift-py35/4f3a35a/ : SUCCESS in 3m 22s\n- requirements-integration http://logs.openstack.org/38/653638/1/check/requirements-integration/9c661fb/ : SUCCESS in 54m 25s\n- neutron-grenade-multinode http://logs.openstack.org/38/653638/1/check/neutron-grenade-multinode/9956f4e/ : SUCCESS in 1h 00m 11s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/38/653638/1/check/legacy-tempest-dsvm-neutron-full/cb2a0ab/ : SUCCESS in 1h 31m 09s\n- neutron-grenade http://logs.openstack.org/38/653638/1/check/neutron-grenade/8b20ea4/ : SUCCESS in 53m 54s","accounts_in_message":[],"_revision_number":1},{"id":"620e20a67638ceedbc3c7be9dde3fa6f12657240","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 19:54:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2553a1df64ca181ccc0c3689a08791fa7fe2000e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 00:47:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/38/653638/1/check/openstack-tox-pypy/4d0029c/ : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/38/653638/1/check/openstack-tox-pep8/f1a3629/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/38/653638/1/check/openstack-tox-py27/1e686f9/ : SUCCESS in 5m 56s\n- openstack-tox-py35 http://logs.openstack.org/38/653638/1/check/openstack-tox-py35/395358a/ : SUCCESS in 3m 58s\n- openstack-tox-validate http://logs.openstack.org/38/653638/1/check/openstack-tox-validate/15d5cec/ : SUCCESS in 3m 37s\n- requirements-tox-babel http://logs.openstack.org/38/653638/1/check/requirements-tox-babel/5d965ac/ : SUCCESS in 4m 23s\n- requirements-tox-py27-check-uc http://logs.openstack.org/38/653638/1/check/requirements-tox-py27-check-uc/bc6691e/ : SUCCESS in 6m 30s\n- cross-cinder-py27 http://logs.openstack.org/38/653638/1/check/cross-cinder-py27/bad1127/ : SUCCESS in 7m 12s\n- cross-cinder-py35 http://logs.openstack.org/38/653638/1/check/cross-cinder-py35/450d8b3/ : SUCCESS in 7m 36s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/38/653638/1/check/cross-glance-py27/2a42a9f/ : FAILURE in 13m 36s\n- cross-glance-py35 http://logs.openstack.org/38/653638/1/check/cross-glance-py35/21ba50c/ : SUCCESS in 9m 17s\n- cross-horizon-py27 http://logs.openstack.org/38/653638/1/check/cross-horizon-py27/08be865/ : SUCCESS in 11m 26s\n- cross-horizon-py35 http://logs.openstack.org/38/653638/1/check/cross-horizon-py35/12442c2/ : SUCCESS in 10m 13s\n- cross-keystone-py27 http://logs.openstack.org/38/653638/1/check/cross-keystone-py27/1ffb3d8/ : SUCCESS in 12m 43s\n- cross-keystone-py35 http://logs.openstack.org/38/653638/1/check/cross-keystone-py35/2d5ba97/ : SUCCESS in 11m 07s\n- cross-neutron-py27 http://logs.openstack.org/38/653638/1/check/cross-neutron-py27/02f09bb/ : SUCCESS in 23m 03s\n- cross-neutron-py35 http://logs.openstack.org/38/653638/1/check/cross-neutron-py35/6e70682/ : SUCCESS in 22m 30s\n- cross-nova-py27 http://logs.openstack.org/38/653638/1/check/cross-nova-py27/c51bda1/ : SUCCESS in 14m 05s\n- cross-nova-py35 http://logs.openstack.org/38/653638/1/check/cross-nova-py35/bcbcde0/ : SUCCESS in 13m 01s\n- cross-nova-functional http://logs.openstack.org/38/653638/1/check/cross-nova-functional/08be398/ : SUCCESS in 19m 04s\n- cross-swift-py27 http://logs.openstack.org/38/653638/1/check/cross-swift-py27/85edc62/ : SUCCESS in 9m 11s\n- cross-swift-py35 http://logs.openstack.org/38/653638/1/check/cross-swift-py35/4cdf560/ : SUCCESS in 7m 19s\n- requirements-integration http://logs.openstack.org/38/653638/1/check/requirements-integration/bd26be6/ : SUCCESS in 58m 40s\n- neutron-grenade-multinode http://logs.openstack.org/38/653638/1/check/neutron-grenade-multinode/168efc2/ : SUCCESS in 55m 22s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/38/653638/1/check/legacy-tempest-dsvm-neutron-full/f8dc86c/ : FAILURE in 1h 11m 37s\n- neutron-grenade http://logs.openstack.org/38/653638/1/check/neutron-grenade/123aaf1/ : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":1},{"id":"34f6bfdd05c5002b3325a6c30235ad6fcfa13559","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-23 00:47:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d84a3320932b4d32801a73ed26f184dcfb893050","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 04:01:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/38/653638/1/check/openstack-tox-pypy/944373a/ : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/38/653638/1/check/openstack-tox-pep8/ada4c07/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/38/653638/1/check/openstack-tox-py27/ad2244a/ : SUCCESS in 3m 52s\n- openstack-tox-py35 http://logs.openstack.org/38/653638/1/check/openstack-tox-py35/8381e63/ : SUCCESS in 4m 01s\n- openstack-tox-validate http://logs.openstack.org/38/653638/1/check/openstack-tox-validate/7b4fdfa/ : SUCCESS in 3m 50s\n- requirements-tox-babel http://logs.openstack.org/38/653638/1/check/requirements-tox-babel/5d418c1/ : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/38/653638/1/check/requirements-tox-py27-check-uc/a8a0328/ : SUCCESS in 7m 23s\n- cross-cinder-py27 http://logs.openstack.org/38/653638/1/check/cross-cinder-py27/3050c0e/ : SUCCESS in 7m 39s\n- cross-cinder-py35 http://logs.openstack.org/38/653638/1/check/cross-cinder-py35/514afc6/ : SUCCESS in 7m 19s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/38/653638/1/check/cross-glance-py27/deadb79/ : SUCCESS in 14m 56s\n- cross-glance-py35 http://logs.openstack.org/38/653638/1/check/cross-glance-py35/6bc433f/ : SUCCESS in 12m 47s\n- cross-horizon-py27 http://logs.openstack.org/38/653638/1/check/cross-horizon-py27/497782f/ : SUCCESS in 11m 07s\n- cross-horizon-py35 http://logs.openstack.org/38/653638/1/check/cross-horizon-py35/5ac4bc8/ : SUCCESS in 10m 56s\n- cross-keystone-py27 http://logs.openstack.org/38/653638/1/check/cross-keystone-py27/d04c3b1/ : SUCCESS in 15m 47s\n- cross-keystone-py35 http://logs.openstack.org/38/653638/1/check/cross-keystone-py35/810284d/ : SUCCESS in 11m 28s\n- cross-neutron-py27 http://logs.openstack.org/38/653638/1/check/cross-neutron-py27/6688133/ : SUCCESS in 21m 16s\n- cross-neutron-py35 http://logs.openstack.org/38/653638/1/check/cross-neutron-py35/5259a6f/ : SUCCESS in 22m 27s\n- cross-nova-py27 http://logs.openstack.org/38/653638/1/check/cross-nova-py27/5876bdf/ : SUCCESS in 13m 26s\n- cross-nova-py35 http://logs.openstack.org/38/653638/1/check/cross-nova-py35/cd56d87/ : SUCCESS in 14m 04s\n- cross-nova-functional http://logs.openstack.org/38/653638/1/check/cross-nova-functional/c1aed56/ : SUCCESS in 17m 43s\n- cross-swift-py27 http://logs.openstack.org/38/653638/1/check/cross-swift-py27/d8f0517/ : SUCCESS in 8m 04s\n- cross-swift-py35 http://logs.openstack.org/38/653638/1/check/cross-swift-py35/a56d738/ : SUCCESS in 3m 42s\n- requirements-integration http://logs.openstack.org/38/653638/1/check/requirements-integration/6eac01c/ : SUCCESS in 55m 59s\n- neutron-grenade-multinode http://logs.openstack.org/38/653638/1/check/neutron-grenade-multinode/def4c24/ : SUCCESS in 58m 36s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/38/653638/1/check/legacy-tempest-dsvm-neutron-full/59c583a/ : SUCCESS in 1h 32m 02s\n- neutron-grenade http://logs.openstack.org/38/653638/1/check/neutron-grenade/5659099/ : SUCCESS in 51m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3a097514b01fce26a23765adc10c25eb4318842e","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-23 07:09:30.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b84fef0107e5bead569e8e15ac133ccdf362989d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 07:09:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09f577d23d8a74f05250bd765a604e2110a02a2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 08:50:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/38/653638/1/gate/openstack-tox-pep8/980e3a5/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/38/653638/1/gate/openstack-tox-py27/f6376ce/ : SUCCESS in 6m 36s\n- openstack-tox-py35 http://logs.openstack.org/38/653638/1/gate/openstack-tox-py35/380051f/ : SUCCESS in 8m 34s\n- openstack-tox-validate http://logs.openstack.org/38/653638/1/gate/openstack-tox-validate/9d7d2f4/ : SUCCESS in 4m 28s\n- requirements-tox-babel http://logs.openstack.org/38/653638/1/gate/requirements-tox-babel/1a6a5e8/ : SUCCESS in 3m 59s\n- requirements-tox-py27-check-uc http://logs.openstack.org/38/653638/1/gate/requirements-tox-py27-check-uc/bd3e87c/ : SUCCESS in 8m 25s\n- cross-cinder-py27 http://logs.openstack.org/38/653638/1/gate/cross-cinder-py27/90b12da/ : SUCCESS in 9m 44s\n- cross-glance-py27 http://logs.openstack.org/38/653638/1/gate/cross-glance-py27/d660066/ : SUCCESS in 16m 20s\n- cross-glance-py35 http://logs.openstack.org/38/653638/1/gate/cross-glance-py35/b3f35b9/ : SUCCESS in 11m 51s\n- cross-horizon-py27 http://logs.openstack.org/38/653638/1/gate/cross-horizon-py27/9db3bf1/ : SUCCESS in 10m 58s\n- cross-horizon-py35 http://logs.openstack.org/38/653638/1/gate/cross-horizon-py35/a83a454/ : SUCCESS in 12m 18s\n- cross-keystone-py27 http://logs.openstack.org/38/653638/1/gate/cross-keystone-py27/dafce6a/ : SUCCESS in 40m 29s\n- cross-keystone-py35 http://logs.openstack.org/38/653638/1/gate/cross-keystone-py35/83079ab/ : SUCCESS in 12m 54s\n- cross-neutron-py27 http://logs.openstack.org/38/653638/1/gate/cross-neutron-py27/25a3f87/ : SUCCESS in 23m 08s\n- cross-neutron-py35 http://logs.openstack.org/38/653638/1/gate/cross-neutron-py35/a8f7f01/ : SUCCESS in 24m 39s\n- cross-nova-py27 http://logs.openstack.org/38/653638/1/gate/cross-nova-py27/55296e9/ : SUCCESS in 11m 40s\n- cross-nova-py35 http://logs.openstack.org/38/653638/1/gate/cross-nova-py35/13ed3e3/ : SUCCESS in 12m 40s\n- cross-nova-functional http://logs.openstack.org/38/653638/1/gate/cross-nova-functional/eb4736d/ : SUCCESS in 24m 58s\n- cross-swift-py27 http://logs.openstack.org/38/653638/1/gate/cross-swift-py27/3ea9498/ : SUCCESS in 16m 04s\n- cross-swift-py35 http://logs.openstack.org/38/653638/1/gate/cross-swift-py35/d2a7fca/ : SUCCESS in 4m 41s\n- requirements-integration http://logs.openstack.org/38/653638/1/gate/requirements-integration/c8b8fa4/ : SUCCESS in 1h 14m 57s\n- neutron-grenade-multinode http://logs.openstack.org/38/653638/1/gate/neutron-grenade-multinode/5644a18/ : SUCCESS in 1h 02m 56s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/38/653638/1/gate/legacy-tempest-dsvm-neutron-full/b46f1a6/ : SUCCESS in 1h 28m 33s\n- neutron-grenade http://logs.openstack.org/38/653638/1/gate/neutron-grenade/8c99fed/ : SUCCESS in 1h 07m 20s","accounts_in_message":[],"_revision_number":1},{"id":"bf2aaf720dd3dc52e137821574597d60e7b2dde6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 08:50:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dcbe127797a1676f0a5f88a3ec5177a4a1501b31","revisions":{"dcbe127797a1676f0a5f88a3ec5177a4a1501b31":{"kind":"REWORK","_number":1,"created":"2019-04-17 23:53:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/653638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/653638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/653638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/653638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/653638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/653638/1"}}},"commit":{"parents":[{"commit":"3170a63a11028e9701bc0590fc15a44ede84d30f","subject":"Merge \"update constraint for oslo.reports to new release 1.22.2\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3170a63a11028e9701bc0590fc15a44ede84d30f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 23:53:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 23:53:26.000000000","tz":0},"subject":"update constraint for pycadf to new release 2.6.1","message":"update constraint for pycadf to new release 2.6.1\n\nChange-Id: I1e586ba768b427a6ce1a3785657b5d9d28d6606d\nmeta:version: 2.6.1\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Change-Id: I0ea3e30572fb745950c0eca0770f96bb87e0a53b\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dcbe127797a1676f0a5f88a3ec5177a4a1501b31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dcbe127797a1676f0a5f88a3ec5177a4a1501b31"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
