)]}'
{"id":"openstack%2Frequirements~653640","triplet_id":"openstack%2Frequirements~stable%2Fpike~I93c698f88cfb4e44b6fa98704193359f8f262f7c","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"I93c698f88cfb4e44b6fa98704193359f8f262f7c","subject":"update constraint for keystonemiddleware to new release 4.17.1","status":"MERGED","created":"2019-04-17 23:55:37.000000000","updated":"2019-04-26 11:20:24.000000000","submitted":"2019-04-26 11:20:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653640-1556277624970-a42c10a4","meta_rev_id":"5d93021539e8c50a4476e60750a3e912848daeef","_number":653640,"virtual_id_number":653640,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-04-26 11:20:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-23 01:27:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-25 15:35:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-04-25 15:35:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 01:27:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-25 15:35:28.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-26 11:20:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94645b5f7f984360ceebc17101c6f7d01d1af907","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-17 23:55:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"090637675f182ef101bdbabee54f069fab5885a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 12:01:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/40/653640/1/check/openstack-tox-pypy/94da763/ : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/653640/1/check/openstack-tox-pep8/aa663fb/ : SUCCESS in 3m 40s\n- openstack-tox-py27 http://logs.openstack.org/40/653640/1/check/openstack-tox-py27/b8a6f58/ : SUCCESS in 3m 51s\n- openstack-tox-py35 http://logs.openstack.org/40/653640/1/check/openstack-tox-py35/5ad8f50/ : SUCCESS in 4m 37s\n- openstack-tox-validate http://logs.openstack.org/40/653640/1/check/openstack-tox-validate/cda33fd/ : SUCCESS in 4m 00s\n- requirements-tox-babel http://logs.openstack.org/40/653640/1/check/requirements-tox-babel/7287303/ : SUCCESS in 4m 23s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/653640/1/check/requirements-tox-py27-check-uc/d34fb6a/ : SUCCESS in 8m 23s\n- cross-cinder-py27 http://logs.openstack.org/40/653640/1/check/cross-cinder-py27/d2f8573/ : SUCCESS in 8m 44s\n- cross-cinder-py35 http://logs.openstack.org/40/653640/1/check/cross-cinder-py35/8f0bf83/ : SUCCESS in 12m 22s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/40/653640/1/check/cross-glance-py27/5b289e3/ : SUCCESS in 11m 56s\n- cross-glance-py35 http://logs.openstack.org/40/653640/1/check/cross-glance-py35/d2df6ea/ : SUCCESS in 11m 58s\n- cross-horizon-py27 http://logs.openstack.org/40/653640/1/check/cross-horizon-py27/5b3ee85/ : SUCCESS in 11m 07s\n- cross-horizon-py35 http://logs.openstack.org/40/653640/1/check/cross-horizon-py35/81e1f73/ : SUCCESS in 9m 54s\n- cross-keystone-py27 http://logs.openstack.org/40/653640/1/check/cross-keystone-py27/dd72bdc/ : SUCCESS in 14m 21s\n- cross-keystone-py35 http://logs.openstack.org/40/653640/1/check/cross-keystone-py35/3d1253b/ : SUCCESS in 13m 59s\n- cross-neutron-py27 http://logs.openstack.org/40/653640/1/check/cross-neutron-py27/6dbd857/ : SUCCESS in 22m 05s\n- cross-neutron-py35 http://logs.openstack.org/40/653640/1/check/cross-neutron-py35/ff4420a/ : SUCCESS in 25m 34s\n- cross-nova-py27 http://logs.openstack.org/40/653640/1/check/cross-nova-py27/cc375c2/ : SUCCESS in 14m 05s\n- cross-nova-py35 http://logs.openstack.org/40/653640/1/check/cross-nova-py35/e01f66c/ : SUCCESS in 13m 09s\n- cross-nova-functional http://logs.openstack.org/40/653640/1/check/cross-nova-functional/64adb22/ : SUCCESS in 26m 55s\n- cross-swift-py27 http://logs.openstack.org/40/653640/1/check/cross-swift-py27/7e02714/ : FAILURE in 9m 14s\n- cross-swift-py35 http://logs.openstack.org/40/653640/1/check/cross-swift-py35/3ca666a/ : SUCCESS in 5m 24s\n- requirements-integration http://logs.openstack.org/40/653640/1/check/requirements-integration/69dab23/ : SUCCESS in 53m 05s\n- neutron-grenade-multinode http://logs.openstack.org/40/653640/1/check/neutron-grenade-multinode/89389a5/ : SUCCESS in 1h 01m 23s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/40/653640/1/check/legacy-tempest-dsvm-neutron-full/30835b0/ : SUCCESS in 1h 27m 47s\n- neutron-grenade http://logs.openstack.org/40/653640/1/check/neutron-grenade/cc46318/ : SUCCESS in 52m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7b30aa6f8436a7a3aa73fa4b032f8b5d62525f13","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-18 16:17:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9545a08fc1df8ba46bc9b5de9f4b9db51b075f69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 06:41:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/40/653640/1/check/openstack-tox-pypy/d06753c/ : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/653640/1/check/openstack-tox-pep8/b5ed588/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/40/653640/1/check/openstack-tox-py27/93d2856/ : SUCCESS in 3m 49s\n- openstack-tox-py35 http://logs.openstack.org/40/653640/1/check/openstack-tox-py35/08f1d40/ : SUCCESS in 4m 10s\n- openstack-tox-validate http://logs.openstack.org/40/653640/1/check/openstack-tox-validate/43fed86/ : SUCCESS in 4m 41s\n- requirements-tox-babel http://logs.openstack.org/40/653640/1/check/requirements-tox-babel/3a67065/ : SUCCESS in 5m 19s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/653640/1/check/requirements-tox-py27-check-uc/f05eaae/ : SUCCESS in 6m 59s\n- cross-cinder-py27 http://logs.openstack.org/40/653640/1/check/cross-cinder-py27/1edcbbe/ : SUCCESS in 8m 16s\n- cross-cinder-py35 http://logs.openstack.org/40/653640/1/check/cross-cinder-py35/ffb7cc4/ : SUCCESS in 8m 07s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/40/653640/1/check/cross-glance-py27/b40c1ae/ : SUCCESS in 9m 08s\n- cross-glance-py35 http://logs.openstack.org/40/653640/1/check/cross-glance-py35/9c451b9/ : SUCCESS in 8m 57s\n- cross-horizon-py27 http://logs.openstack.org/40/653640/1/check/cross-horizon-py27/da4d085/ : SUCCESS in 10m 14s\n- cross-horizon-py35 http://logs.openstack.org/40/653640/1/check/cross-horizon-py35/9a91891/ : SUCCESS in 10m 26s\n- cross-keystone-py27 http://logs.openstack.org/40/653640/1/check/cross-keystone-py27/a09d325/ : SUCCESS in 18m 03s\n- cross-keystone-py35 http://logs.openstack.org/40/653640/1/check/cross-keystone-py35/25bc726/ : SUCCESS in 16m 23s\n- cross-neutron-py27 http://logs.openstack.org/40/653640/1/check/cross-neutron-py27/c2d4936/ : SUCCESS in 24m 25s\n- cross-neutron-py35 http://logs.openstack.org/40/653640/1/check/cross-neutron-py35/354db34/ : SUCCESS in 24m 19s\n- cross-nova-py27 http://logs.openstack.org/40/653640/1/check/cross-nova-py27/5b75546/ : SUCCESS in 14m 17s\n- cross-nova-py35 http://logs.openstack.org/40/653640/1/check/cross-nova-py35/4132e73/ : SUCCESS in 13m 46s\n- cross-nova-functional http://logs.openstack.org/40/653640/1/check/cross-nova-functional/6b69298/ : SUCCESS in 19m 05s\n- cross-swift-py27 http://logs.openstack.org/40/653640/1/check/cross-swift-py27/71340af/ : SUCCESS in 7m 46s\n- cross-swift-py35 http://logs.openstack.org/40/653640/1/check/cross-swift-py35/46cdb8e/ : SUCCESS in 3m 35s\n- requirements-integration http://logs.openstack.org/40/653640/1/check/requirements-integration/c21fbaf/ : SUCCESS in 56m 07s\n- neutron-grenade-multinode http://logs.openstack.org/40/653640/1/check/neutron-grenade-multinode/fa7c54e/ : SUCCESS in 1h 04m 42s\n- legacy-tempest-dsvm-neutron-full finger://ze04.openstack.org/fad80ff2835646d5a6ab424ca92d8838 : RETRY_LIMIT in 45m 48s\n- neutron-grenade http://logs.openstack.org/40/653640/1/check/neutron-grenade/6768569/ : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":1},{"id":"bdf337d4a2728d87dbd46f18c39961f3c61fc414","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 14:55:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"26c6a69f8996f8ff46d8184c6cecf2fe74ef2fb3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 16:19:26.000000000","message":"Patch Set 1:\n\nrecheck (and unvote til gate works again)","accounts_in_message":[],"_revision_number":1},{"id":"6220ce2031710ada20cfa4a6db3a8465096eb7fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 19:50:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/40/653640/1/check/openstack-tox-pypy/9cc7d1b/ : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/653640/1/check/openstack-tox-pep8/d9265bc/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/40/653640/1/check/openstack-tox-py27/5410a14/ : SUCCESS in 4m 41s\n- openstack-tox-py35 http://logs.openstack.org/40/653640/1/check/openstack-tox-py35/b402a7d/ : SUCCESS in 3m 54s\n- openstack-tox-validate http://logs.openstack.org/40/653640/1/check/openstack-tox-validate/9a66d68/ : SUCCESS in 4m 58s\n- requirements-tox-babel http://logs.openstack.org/40/653640/1/check/requirements-tox-babel/24ebb7c/ : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/653640/1/check/requirements-tox-py27-check-uc/1da95ab/ : SUCCESS in 5m 54s\n- cross-cinder-py27 http://logs.openstack.org/40/653640/1/check/cross-cinder-py27/2678ed9/ : SUCCESS in 12m 18s\n- cross-cinder-py35 http://logs.openstack.org/40/653640/1/check/cross-cinder-py35/3d12978/ : SUCCESS in 8m 59s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/40/653640/1/check/cross-glance-py27/718f792/ : SUCCESS in 9m 03s\n- cross-glance-py35 http://logs.openstack.org/40/653640/1/check/cross-glance-py35/bdd7298/ : SUCCESS in 9m 00s\n- cross-horizon-py27 http://logs.openstack.org/40/653640/1/check/cross-horizon-py27/0f5e1c1/ : SUCCESS in 11m 00s\n- cross-horizon-py35 http://logs.openstack.org/40/653640/1/check/cross-horizon-py35/c4143b9/ : SUCCESS in 11m 07s\n- cross-keystone-py27 http://logs.openstack.org/40/653640/1/check/cross-keystone-py27/0a68638/ : SUCCESS in 12m 48s\n- cross-keystone-py35 http://logs.openstack.org/40/653640/1/check/cross-keystone-py35/4d6fa62/ : SUCCESS in 11m 23s\n- cross-neutron-py27 http://logs.openstack.org/40/653640/1/check/cross-neutron-py27/7ce60c2/ : SUCCESS in 21m 45s\n- cross-neutron-py35 http://logs.openstack.org/40/653640/1/check/cross-neutron-py35/3185a50/ : TIMED_OUT in 40m 56s\n- cross-nova-py27 http://logs.openstack.org/40/653640/1/check/cross-nova-py27/bb8b1bf/ : SUCCESS in 13m 38s\n- cross-nova-py35 http://logs.openstack.org/40/653640/1/check/cross-nova-py35/bce117b/ : SUCCESS in 11m 45s\n- cross-nova-functional http://logs.openstack.org/40/653640/1/check/cross-nova-functional/53c4e5a/ : SUCCESS in 19m 08s\n- cross-swift-py27 http://logs.openstack.org/40/653640/1/check/cross-swift-py27/5beb896/ : SUCCESS in 7m 19s\n- cross-swift-py35 http://logs.openstack.org/40/653640/1/check/cross-swift-py35/8ff9008/ : SUCCESS in 4m 32s\n- requirements-integration http://logs.openstack.org/40/653640/1/check/requirements-integration/5b73e10/ : SUCCESS in 54m 44s\n- neutron-grenade-multinode http://logs.openstack.org/40/653640/1/check/neutron-grenade-multinode/48b1118/ : SUCCESS in 1h 06m 40s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/40/653640/1/check/legacy-tempest-dsvm-neutron-full/2d287d0/ : SUCCESS in 1h 27m 24s\n- neutron-grenade http://logs.openstack.org/40/653640/1/check/neutron-grenade/e850ac1/ : SUCCESS in 53m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6dfd4d44b495baf544f3bd868366808ab2d999d8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 19:54:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"38e4adcd32d4d63265dce2ba7c2fc224fa6d98ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 01:24:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/40/653640/1/check/openstack-tox-pypy/6253c8c/ : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/653640/1/check/openstack-tox-pep8/e20e4e3/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/40/653640/1/check/openstack-tox-py27/34d9e42/ : SUCCESS in 3m 36s\n- openstack-tox-py35 http://logs.openstack.org/40/653640/1/check/openstack-tox-py35/b46b633/ : SUCCESS in 4m 03s\n- openstack-tox-validate http://logs.openstack.org/40/653640/1/check/openstack-tox-validate/12a72e9/ : SUCCESS in 3m 46s\n- requirements-tox-babel http://logs.openstack.org/40/653640/1/check/requirements-tox-babel/f461637/ : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/653640/1/check/requirements-tox-py27-check-uc/884f28f/ : SUCCESS in 6m 53s\n- cross-cinder-py27 http://logs.openstack.org/40/653640/1/check/cross-cinder-py27/a36b1c7/ : SUCCESS in 7m 26s\n- cross-cinder-py35 http://logs.openstack.org/40/653640/1/check/cross-cinder-py35/1a233da/ : SUCCESS in 9m 09s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/40/653640/1/check/cross-glance-py27/02d18aa/ : SUCCESS in 9m 12s\n- cross-glance-py35 http://logs.openstack.org/40/653640/1/check/cross-glance-py35/ab31a0a/ : SUCCESS in 8m 39s\n- cross-horizon-py27 http://logs.openstack.org/40/653640/1/check/cross-horizon-py27/d056c15/ : SUCCESS in 10m 08s\n- cross-horizon-py35 http://logs.openstack.org/40/653640/1/check/cross-horizon-py35/4facc10/ : SUCCESS in 10m 15s\n- cross-keystone-py27 http://logs.openstack.org/40/653640/1/check/cross-keystone-py27/b42a0bf/ : SUCCESS in 10m 58s\n- cross-keystone-py35 http://logs.openstack.org/40/653640/1/check/cross-keystone-py35/606637b/ : SUCCESS in 10m 40s\n- cross-neutron-py27 http://logs.openstack.org/40/653640/1/check/cross-neutron-py27/eca8b94/ : SUCCESS in 22m 33s\n- cross-neutron-py35 http://logs.openstack.org/40/653640/1/check/cross-neutron-py35/a40bdce/ : SUCCESS in 25m 01s\n- cross-nova-py27 http://logs.openstack.org/40/653640/1/check/cross-nova-py27/34bbdd9/ : SUCCESS in 16m 53s\n- cross-nova-py35 http://logs.openstack.org/40/653640/1/check/cross-nova-py35/077e142/ : SUCCESS in 15m 54s\n- cross-nova-functional http://logs.openstack.org/40/653640/1/check/cross-nova-functional/95c61a4/ : SUCCESS in 18m 28s\n- cross-swift-py27 http://logs.openstack.org/40/653640/1/check/cross-swift-py27/cd3c10d/ : SUCCESS in 7m 28s\n- cross-swift-py35 http://logs.openstack.org/40/653640/1/check/cross-swift-py35/84573b4/ : SUCCESS in 3m 19s\n- requirements-integration http://logs.openstack.org/40/653640/1/check/requirements-integration/df876c7/ : SUCCESS in 56m 04s\n- neutron-grenade-multinode http://logs.openstack.org/40/653640/1/check/neutron-grenade-multinode/b4c9729/ : SUCCESS in 1h 07m 31s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/40/653640/1/check/legacy-tempest-dsvm-neutron-full/761fe51/ : SUCCESS in 1h 29m 32s\n- neutron-grenade http://logs.openstack.org/40/653640/1/check/neutron-grenade/c229bb2/ : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d2af00d0530dc67ae1c6029b397bb047b0e9c27f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-23 01:27:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e9553d79059aaf6901fdb3f7afe4bdad7c226230","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-25 15:35:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"895056a148cfdf969de7570995f94bd97bb15980","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 15:37:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8942cc2e3f70013f851393cc2b54635a4bafb388","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 02:33:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/40/653640/1/gate/openstack-tox-pep8/4b7d7b8/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/40/653640/1/gate/openstack-tox-py27/9b11ddd/ : SUCCESS in 4m 52s\n- openstack-tox-py35 http://logs.openstack.org/40/653640/1/gate/openstack-tox-py35/20f8801/ : SUCCESS in 4m 51s\n- openstack-tox-validate http://logs.openstack.org/40/653640/1/gate/openstack-tox-validate/589db26/ : SUCCESS in 4m 26s\n- requirements-tox-babel http://logs.openstack.org/40/653640/1/gate/requirements-tox-babel/57ec50f/ : SUCCESS in 4m 13s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/653640/1/gate/requirements-tox-py27-check-uc/09a16f8/ : SUCCESS in 7m 10s\n- cross-cinder-py27 http://logs.openstack.org/40/653640/1/gate/cross-cinder-py27/bdd901a/ : SUCCESS in 8m 23s\n- cross-glance-py27 http://logs.openstack.org/40/653640/1/gate/cross-glance-py27/ea6a72b/ : FAILURE in 10m 35s\n- cross-glance-py35 http://logs.openstack.org/40/653640/1/gate/cross-glance-py35/94c3697/ : SUCCESS in 9m 41s\n- cross-horizon-py27 http://logs.openstack.org/40/653640/1/gate/cross-horizon-py27/172f4c2/ : SUCCESS in 10m 39s\n- cross-horizon-py35 http://logs.openstack.org/40/653640/1/gate/cross-horizon-py35/8e435dc/ : SUCCESS in 10m 24s\n- cross-keystone-py27 http://logs.openstack.org/40/653640/1/gate/cross-keystone-py27/f9be68b/ : SUCCESS in 12m 20s\n- cross-keystone-py35 http://logs.openstack.org/40/653640/1/gate/cross-keystone-py35/a9e6715/ : SUCCESS in 12m 56s\n- cross-neutron-py27 http://logs.openstack.org/40/653640/1/gate/cross-neutron-py27/995e450/ : SUCCESS in 24m 09s\n- cross-neutron-py35 http://logs.openstack.org/40/653640/1/gate/cross-neutron-py35/0a6620e/ : SUCCESS in 24m 47s\n- cross-nova-py27 http://logs.openstack.org/40/653640/1/gate/cross-nova-py27/85164b9/ : SUCCESS in 13m 13s\n- cross-nova-py35 http://logs.openstack.org/40/653640/1/gate/cross-nova-py35/fb4d74d/ : SUCCESS in 13m 01s\n- cross-nova-functional http://logs.openstack.org/40/653640/1/gate/cross-nova-functional/97db862/ : SUCCESS in 20m 19s\n- cross-swift-py27 http://logs.openstack.org/40/653640/1/gate/cross-swift-py27/6d755dc/ : SUCCESS in 8m 05s\n- cross-swift-py35 http://logs.openstack.org/40/653640/1/gate/cross-swift-py35/d7293f9/ : SUCCESS in 4m 28s\n- requirements-integration http://logs.openstack.org/40/653640/1/gate/requirements-integration/9c959e2/ : SUCCESS in 1h 01m 57s\n- neutron-grenade-multinode http://logs.openstack.org/40/653640/1/gate/neutron-grenade-multinode/91f41af/ : SUCCESS in 1h 05m 15s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/40/653640/1/gate/legacy-tempest-dsvm-neutron-full/b006d4b/ : FAILURE in 1h 17m 43s\n- neutron-grenade http://logs.openstack.org/40/653640/1/gate/neutron-grenade/6133c17/ : SUCCESS in 54m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1fa2a78e0a6ce3f8da68941e0327ad7f5df41eb1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-26 02:34:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b24dcc918503ee41a26e496d0efddbc0a0e0c710","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 09:33:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/40/653640/1/check/openstack-tox-pypy/a1178ce/ : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/653640/1/check/openstack-tox-pep8/9d68205/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/40/653640/1/check/openstack-tox-py27/1490776/ : SUCCESS in 3m 37s\n- openstack-tox-py35 http://logs.openstack.org/40/653640/1/check/openstack-tox-py35/96c6757/ : SUCCESS in 4m 17s\n- openstack-tox-validate http://logs.openstack.org/40/653640/1/check/openstack-tox-validate/79ff534/ : SUCCESS in 4m 52s\n- requirements-tox-babel http://logs.openstack.org/40/653640/1/check/requirements-tox-babel/45a0564/ : SUCCESS in 3m 05s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/653640/1/check/requirements-tox-py27-check-uc/43e8682/ : SUCCESS in 5m 40s\n- cross-cinder-py27 http://logs.openstack.org/40/653640/1/check/cross-cinder-py27/4c0e98f/ : SUCCESS in 9m 22s\n- cross-cinder-py35 http://logs.openstack.org/40/653640/1/check/cross-cinder-py35/3ff8316/ : SUCCESS in 7m 38s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/40/653640/1/check/cross-glance-py27/8528ca2/ : SUCCESS in 8m 31s\n- cross-glance-py35 http://logs.openstack.org/40/653640/1/check/cross-glance-py35/b1d3df3/ : SUCCESS in 9m 04s\n- cross-horizon-py27 http://logs.openstack.org/40/653640/1/check/cross-horizon-py27/8256186/ : SUCCESS in 10m 48s\n- cross-horizon-py35 http://logs.openstack.org/40/653640/1/check/cross-horizon-py35/1f17f46/ : SUCCESS in 9m 58s\n- cross-keystone-py27 http://logs.openstack.org/40/653640/1/check/cross-keystone-py27/b978cf0/ : SUCCESS in 13m 12s\n- cross-keystone-py35 http://logs.openstack.org/40/653640/1/check/cross-keystone-py35/933d7be/ : SUCCESS in 12m 10s\n- cross-neutron-py27 http://logs.openstack.org/40/653640/1/check/cross-neutron-py27/b3b80d3/ : SUCCESS in 21m 11s\n- cross-neutron-py35 http://logs.openstack.org/40/653640/1/check/cross-neutron-py35/5f25689/ : SUCCESS in 20m 45s\n- cross-nova-py27 http://logs.openstack.org/40/653640/1/check/cross-nova-py27/063b36a/ : SUCCESS in 13m 01s\n- cross-nova-py35 http://logs.openstack.org/40/653640/1/check/cross-nova-py35/8dbb01c/ : SUCCESS in 12m 00s\n- cross-nova-functional http://logs.openstack.org/40/653640/1/check/cross-nova-functional/7b750b6/ : SUCCESS in 21m 51s\n- cross-swift-py27 http://logs.openstack.org/40/653640/1/check/cross-swift-py27/64ec82e/ : SUCCESS in 9m 57s\n- cross-swift-py35 http://logs.openstack.org/40/653640/1/check/cross-swift-py35/8f40546/ : SUCCESS in 3m 40s\n- requirements-integration http://logs.openstack.org/40/653640/1/check/requirements-integration/0e23126/ : SUCCESS in 51m 26s\n- neutron-grenade-multinode http://logs.openstack.org/40/653640/1/check/neutron-grenade-multinode/a01c721/ : SUCCESS in 59m 03s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/40/653640/1/check/legacy-tempest-dsvm-neutron-full/8c2e3da/ : SUCCESS in 1h 25m 09s\n- neutron-grenade http://logs.openstack.org/40/653640/1/check/neutron-grenade/f9b0894/ : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":1},{"id":"cf749935b7bd706534d4c0e4c01e856e1742cc51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 09:34:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91282129051bc8b747cd82ffc6945dde9b47e378","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 11:20:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b134d8fa4bddde3ea849f0a08aa149c5c333df7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 11:20:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/40/653640/1/gate/openstack-tox-pep8/7f0d40c/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/40/653640/1/gate/openstack-tox-py27/6f13dcd/ : SUCCESS in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/40/653640/1/gate/openstack-tox-py35/76a74bc/ : SUCCESS in 3m 51s\n- openstack-tox-validate http://logs.openstack.org/40/653640/1/gate/openstack-tox-validate/447683e/ : SUCCESS in 3m 29s\n- requirements-tox-babel http://logs.openstack.org/40/653640/1/gate/requirements-tox-babel/a2bbde4/ : SUCCESS in 3m 11s\n- requirements-tox-py27-check-uc http://logs.openstack.org/40/653640/1/gate/requirements-tox-py27-check-uc/8ed9c12/ : SUCCESS in 8m 53s\n- cross-cinder-py27 http://logs.openstack.org/40/653640/1/gate/cross-cinder-py27/f97d565/ : SUCCESS in 8m 06s\n- cross-glance-py27 http://logs.openstack.org/40/653640/1/gate/cross-glance-py27/29f7aca/ : SUCCESS in 11m 30s\n- cross-glance-py35 http://logs.openstack.org/40/653640/1/gate/cross-glance-py35/42352dd/ : SUCCESS in 9m 29s\n- cross-horizon-py27 http://logs.openstack.org/40/653640/1/gate/cross-horizon-py27/611d05c/ : SUCCESS in 10m 45s\n- cross-horizon-py35 http://logs.openstack.org/40/653640/1/gate/cross-horizon-py35/07d3ffe/ : SUCCESS in 10m 43s\n- cross-keystone-py27 http://logs.openstack.org/40/653640/1/gate/cross-keystone-py27/6e6307c/ : SUCCESS in 16m 35s\n- cross-keystone-py35 http://logs.openstack.org/40/653640/1/gate/cross-keystone-py35/dec36a3/ : SUCCESS in 12m 02s\n- cross-neutron-py27 http://logs.openstack.org/40/653640/1/gate/cross-neutron-py27/f064a99/ : SUCCESS in 20m 48s\n- cross-neutron-py35 http://logs.openstack.org/40/653640/1/gate/cross-neutron-py35/7d03d5b/ : SUCCESS in 22m 11s\n- cross-nova-py27 http://logs.openstack.org/40/653640/1/gate/cross-nova-py27/c480d16/ : SUCCESS in 12m 28s\n- cross-nova-py35 http://logs.openstack.org/40/653640/1/gate/cross-nova-py35/210132d/ : SUCCESS in 11m 23s\n- cross-nova-functional http://logs.openstack.org/40/653640/1/gate/cross-nova-functional/62725ff/ : SUCCESS in 18m 16s\n- cross-swift-py27 http://logs.openstack.org/40/653640/1/gate/cross-swift-py27/285b37f/ : SUCCESS in 7m 27s\n- cross-swift-py35 http://logs.openstack.org/40/653640/1/gate/cross-swift-py35/f98fbc2/ : SUCCESS in 3m 36s\n- requirements-integration http://logs.openstack.org/40/653640/1/gate/requirements-integration/241ee58/ : SUCCESS in 1h 09m 39s\n- neutron-grenade-multinode http://logs.openstack.org/40/653640/1/gate/neutron-grenade-multinode/c3d6a00/ : SUCCESS in 1h 08m 50s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/40/653640/1/gate/legacy-tempest-dsvm-neutron-full/0fffd09/ : SUCCESS in 1h 32m 51s\n- neutron-grenade http://logs.openstack.org/40/653640/1/gate/neutron-grenade/b6cbaf4/ : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18ab44c714e0dd4b0f91678bee10406f56a1a958","revisions":{"18ab44c714e0dd4b0f91678bee10406f56a1a958":{"kind":"REWORK","_number":1,"created":"2019-04-17 23:55:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/653640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/653640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/653640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/653640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/653640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/653640/1"}}},"commit":{"parents":[{"commit":"3170a63a11028e9701bc0590fc15a44ede84d30f","subject":"Merge \"update constraint for oslo.reports to new release 1.22.2\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3170a63a11028e9701bc0590fc15a44ede84d30f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 23:55:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-17 23:55:36.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 4.17.1","message":"update constraint for keystonemiddleware to new release 4.17.1\n\nChange-Id: I93c698f88cfb4e44b6fa98704193359f8f262f7c\nmeta:version: 4.17.1\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Change-Id: I0ea3e30572fb745950c0eca0770f96bb87e0a53b\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18ab44c714e0dd4b0f91678bee10406f56a1a958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/18ab44c714e0dd4b0f91678bee10406f56a1a958"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
