)]}'
{"id":"openstack%2Frequirements~653787","triplet_id":"openstack%2Frequirements~stable%2Fstein~I5cacdf5186b4604c11c5fc73ef674aebe8d48305","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I5cacdf5186b4604c11c5fc73ef674aebe8d48305","subject":"update constraint for os-collect-config to new release 10.3.0","status":"MERGED","created":"2019-04-18 15:14:09.000000000","updated":"2019-04-26 10:26:59.000000000","submitted":"2019-04-26 10:26:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653787-1556274419995-e0b66468","meta_rev_id":"e3945508ae2b42de75fe107f20edd93843f3fa63","_number":653787,"virtual_id_number":653787,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-04-26 10:26:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-22 23:31:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-25 15:16:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-04-25 15:16:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 23:31:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-25 15:16:36.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-26 10:26:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d13e2b87ff382fb2c95c8b6d6ffc077ba560e13","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-18 15:14:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16fbcb03d52d2ae348f904ecdc4755f959ac4272","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 05:28:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/87/653787/1/check/openstack-tox-pypy/10ba4e2/ : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/87/653787/1/check/openstack-tox-pep8/f2c08c9/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/87/653787/1/check/openstack-tox-py27/4b32c19/ : SUCCESS in 4m 57s\n- openstack-tox-py35 http://logs.openstack.org/87/653787/1/check/openstack-tox-py35/ee72b8e/ : SUCCESS in 5m 06s\n- openstack-tox-py36 http://logs.openstack.org/87/653787/1/check/openstack-tox-py36/49e338f/ : SUCCESS in 5m 29s\n- openstack-tox-validate http://logs.openstack.org/87/653787/1/check/openstack-tox-validate/aefa70d/ : SUCCESS in 4m 44s\n- requirements-tox-babel http://logs.openstack.org/87/653787/1/check/requirements-tox-babel/b9e650a/ : SUCCESS in 4m 11s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/653787/1/check/requirements-tox-py27-check-uc/8ad1dd0/ : SUCCESS in 8m 17s\n- requirements-tox-py35-check-uc http://logs.openstack.org/87/653787/1/check/requirements-tox-py35-check-uc/9119f09/ : SUCCESS in 8m 53s\n- requirements-tox-py36-check-uc http://logs.openstack.org/87/653787/1/check/requirements-tox-py36-check-uc/0e24bd3/ : SUCCESS in 8m 45s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/87/653787/1/check/cross-cinder-py27/80f47fa/ : SUCCESS in 11m 31s\n- cross-cinder-py35 http://logs.openstack.org/87/653787/1/check/cross-cinder-py35/c058435/ : SUCCESS in 10m 41s\n- cross-glance-py27 http://logs.openstack.org/87/653787/1/check/cross-glance-py27/951ecf5/ : SUCCESS in 5m 56s\n- cross-glance-py35 http://logs.openstack.org/87/653787/1/check/cross-glance-py35/38253da/ : SUCCESS in 6m 07s\n- cross-horizon-py27 http://logs.openstack.org/87/653787/1/check/cross-horizon-py27/0f44a75/ : SUCCESS in 11m 16s\n- cross-horizon-py35 http://logs.openstack.org/87/653787/1/check/cross-horizon-py35/3584d95/ : SUCCESS in 10m 47s\n- cross-horizon-npm http://logs.openstack.org/87/653787/1/check/cross-horizon-npm/9fb904e/ : SUCCESS in 7m 55s\n- cross-keystone-py27 http://logs.openstack.org/87/653787/1/check/cross-keystone-py27/b19584a/ : SUCCESS in 23m 42s\n- cross-keystone-py35 http://logs.openstack.org/87/653787/1/check/cross-keystone-py35/696f055/ : SUCCESS in 32m 36s\n- cross-neutron-py27 http://logs.openstack.org/87/653787/1/check/cross-neutron-py27/22a9840/ : SUCCESS in 21m 36s\n- cross-neutron-py35 http://logs.openstack.org/87/653787/1/check/cross-neutron-py35/c8c368c/ : SUCCESS in 23m 48s\n- cross-nova-py27 http://logs.openstack.org/87/653787/1/check/cross-nova-py27/52d61ac/ : SUCCESS in 13m 59s\n- cross-nova-py35 http://logs.openstack.org/87/653787/1/check/cross-nova-py35/fcb4331/ : SUCCESS in 13m 48s\n- cross-nova-functional http://logs.openstack.org/87/653787/1/check/cross-nova-functional/2e6e432/ : SUCCESS in 18m 17s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/87/653787/1/check/cross-osc-build-sphinx-docs/1535de7/html/ : SUCCESS in 6m 08s\n- cross-swift-py27 http://logs.openstack.org/87/653787/1/check/cross-swift-py27/279a686/ : SUCCESS in 9m 18s\n- cross-swift-py35 http://logs.openstack.org/87/653787/1/check/cross-swift-py35/caaf4de/ : SUCCESS in 8m 57s\n- requirements-integration http://logs.openstack.org/87/653787/1/check/requirements-integration/fa85d32/ : SUCCESS in 1h 24m 19s\n- neutron-grenade-multinode http://logs.openstack.org/87/653787/1/check/neutron-grenade-multinode/5a74dc1/ : SUCCESS in 1h 11m 20s\n- tempest-full finger://ze07.openstack.org/fa7b8f011a514904a5f0737ab857bf5b : RETRY_LIMIT in 1h 06m 51s\n- neutron-grenade http://logs.openstack.org/87/653787/1/check/neutron-grenade/f6a121c/ : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":1},{"id":"92305fe12f2f95622ab65edc5658dd5428780e48","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 14:56:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"087b0fbe547ed04e152a5c09c5cc1e9b81de59df","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 16:18:35.000000000","message":"Patch Set 1:\n\nrecheck (and unvote til gate works again)","accounts_in_message":[],"_revision_number":1},{"id":"019fe5d9f7fe19f98c69bbcd8aa15db71ca38987","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:29:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/87/653787/1/check/openstack-tox-pypy/8a38655/ : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/87/653787/1/check/openstack-tox-pep8/6e9a021/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/87/653787/1/check/openstack-tox-py27/e0df745/ : SUCCESS in 3m 53s\n- openstack-tox-py35 http://logs.openstack.org/87/653787/1/check/openstack-tox-py35/131e502/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/87/653787/1/check/openstack-tox-py36/7f1129d/ : SUCCESS in 2m 49s\n- openstack-tox-validate http://logs.openstack.org/87/653787/1/check/openstack-tox-validate/5866970/ : SUCCESS in 4m 49s\n- requirements-tox-babel http://logs.openstack.org/87/653787/1/check/requirements-tox-babel/db1ad48/ : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/653787/1/check/requirements-tox-py27-check-uc/8bf3fba/ : SUCCESS in 7m 50s\n- requirements-tox-py35-check-uc http://logs.openstack.org/87/653787/1/check/requirements-tox-py35-check-uc/7c4d2c9/ : SUCCESS in 7m 29s\n- requirements-tox-py36-check-uc http://logs.openstack.org/87/653787/1/check/requirements-tox-py36-check-uc/7603154/ : SUCCESS in 7m 27s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/87/653787/1/check/cross-cinder-py27/2cd61af/ : SUCCESS in 8m 29s\n- cross-cinder-py35 http://logs.openstack.org/87/653787/1/check/cross-cinder-py35/05239cf/ : SUCCESS in 9m 14s\n- cross-glance-py27 http://logs.openstack.org/87/653787/1/check/cross-glance-py27/6275f3a/ : SUCCESS in 4m 19s\n- cross-glance-py35 http://logs.openstack.org/87/653787/1/check/cross-glance-py35/ba2bd9d/ : SUCCESS in 5m 50s\n- cross-horizon-py27 http://logs.openstack.org/87/653787/1/check/cross-horizon-py27/2d043c1/ : SUCCESS in 11m 21s\n- cross-horizon-py35 http://logs.openstack.org/87/653787/1/check/cross-horizon-py35/4c63c59/ : SUCCESS in 8m 28s\n- cross-horizon-npm http://logs.openstack.org/87/653787/1/check/cross-horizon-npm/0e1d20b/ : SUCCESS in 5m 51s\n- cross-keystone-py27 http://logs.openstack.org/87/653787/1/check/cross-keystone-py27/1ff8aec/ : SUCCESS in 22m 51s\n- cross-keystone-py35 http://logs.openstack.org/87/653787/1/check/cross-keystone-py35/b724fe9/ : SUCCESS in 27m 08s\n- cross-neutron-py27 http://logs.openstack.org/87/653787/1/check/cross-neutron-py27/528e33c/ : FAILURE in 25m 13s\n- cross-neutron-py35 http://logs.openstack.org/87/653787/1/check/cross-neutron-py35/8131032/ : SUCCESS in 26m 49s\n- cross-nova-py27 http://logs.openstack.org/87/653787/1/check/cross-nova-py27/03c56d8/ : SUCCESS in 13m 45s\n- cross-nova-py35 http://logs.openstack.org/87/653787/1/check/cross-nova-py35/ace4de8/ : SUCCESS in 12m 30s\n- cross-nova-functional http://logs.openstack.org/87/653787/1/check/cross-nova-functional/9a92d9c/ : SUCCESS in 20m 54s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/87/653787/1/check/cross-osc-build-sphinx-docs/df8ff18/html/ : SUCCESS in 5m 45s\n- cross-swift-py27 http://logs.openstack.org/87/653787/1/check/cross-swift-py27/dd9e7cb/ : SUCCESS in 7m 37s\n- cross-swift-py35 http://logs.openstack.org/87/653787/1/check/cross-swift-py35/620ecd0/ : SUCCESS in 8m 12s\n- requirements-integration http://logs.openstack.org/87/653787/1/check/requirements-integration/2d490e9/ : SUCCESS in 55m 38s\n- neutron-grenade-multinode http://logs.openstack.org/87/653787/1/check/neutron-grenade-multinode/0a44240/ : SUCCESS in 1h 14m 23s\n- tempest-full http://logs.openstack.org/87/653787/1/check/tempest-full/fba84a1/ : SUCCESS in 1h 36m 02s\n- neutron-grenade http://logs.openstack.org/87/653787/1/check/neutron-grenade/68946ec/ : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":1},{"id":"18362c2cd415375dd9d5ae874824244ba8f8a903","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 23:31:39.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"39a26b9e27ff8d2055e9b5d029524c6620a72112","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 03:53:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/87/653787/1/check/openstack-tox-pypy/17aa4c6/ : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/87/653787/1/check/openstack-tox-pep8/4cda6c8/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/87/653787/1/check/openstack-tox-py27/3a4cc95/ : SUCCESS in 3m 14s\n- openstack-tox-py35 http://logs.openstack.org/87/653787/1/check/openstack-tox-py35/22c5a1c/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/87/653787/1/check/openstack-tox-py36/2741d95/ : SUCCESS in 3m 18s\n- openstack-tox-validate http://logs.openstack.org/87/653787/1/check/openstack-tox-validate/79e3415/ : SUCCESS in 3m 25s\n- requirements-tox-babel http://logs.openstack.org/87/653787/1/check/requirements-tox-babel/b82094f/ : SUCCESS in 2m 33s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/653787/1/check/requirements-tox-py27-check-uc/6b958b2/ : SUCCESS in 7m 17s\n- requirements-tox-py35-check-uc http://logs.openstack.org/87/653787/1/check/requirements-tox-py35-check-uc/15de08d/ : SUCCESS in 7m 28s\n- requirements-tox-py36-check-uc http://logs.openstack.org/87/653787/1/check/requirements-tox-py36-check-uc/ac534c5/ : SUCCESS in 7m 12s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/87/653787/1/check/cross-cinder-py27/a6575c3/ : SUCCESS in 8m 14s\n- cross-cinder-py35 http://logs.openstack.org/87/653787/1/check/cross-cinder-py35/a46f11b/ : SUCCESS in 10m 20s\n- cross-glance-py27 http://logs.openstack.org/87/653787/1/check/cross-glance-py27/578c61b/ : SUCCESS in 4m 20s\n- cross-glance-py35 http://logs.openstack.org/87/653787/1/check/cross-glance-py35/442f51b/ : SUCCESS in 4m 24s\n- cross-horizon-py27 http://logs.openstack.org/87/653787/1/check/cross-horizon-py27/89df65f/ : SUCCESS in 11m 20s\n- cross-horizon-py35 http://logs.openstack.org/87/653787/1/check/cross-horizon-py35/bfee42b/ : SUCCESS in 9m 01s\n- cross-horizon-npm http://logs.openstack.org/87/653787/1/check/cross-horizon-npm/e8d7b1e/ : SUCCESS in 5m 50s\n- cross-keystone-py27 http://logs.openstack.org/87/653787/1/check/cross-keystone-py27/5d095e6/ : SUCCESS in 21m 31s\n- cross-keystone-py35 http://logs.openstack.org/87/653787/1/check/cross-keystone-py35/9c7a648/ : SUCCESS in 25m 03s\n- cross-neutron-py27 http://logs.openstack.org/87/653787/1/check/cross-neutron-py27/c5b1f1a/ : SUCCESS in 20m 45s\n- cross-neutron-py35 http://logs.openstack.org/87/653787/1/check/cross-neutron-py35/0a67a0e/ : SUCCESS in 25m 06s\n- cross-nova-py27 http://logs.openstack.org/87/653787/1/check/cross-nova-py27/2147415/ : SUCCESS in 11m 59s\n- cross-nova-py35 http://logs.openstack.org/87/653787/1/check/cross-nova-py35/66cf5d7/ : SUCCESS in 14m 43s\n- cross-nova-functional http://logs.openstack.org/87/653787/1/check/cross-nova-functional/b810222/ : SUCCESS in 18m 59s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/87/653787/1/check/cross-osc-build-sphinx-docs/a7a165e/html/ : SUCCESS in 5m 34s\n- cross-swift-py27 http://logs.openstack.org/87/653787/1/check/cross-swift-py27/2d3dae2/ : SUCCESS in 7m 34s\n- cross-swift-py35 http://logs.openstack.org/87/653787/1/check/cross-swift-py35/8657d80/ : SUCCESS in 8m 08s\n- requirements-integration http://logs.openstack.org/87/653787/1/check/requirements-integration/a8decec/ : SUCCESS in 1h 00m 54s\n- neutron-grenade-multinode http://logs.openstack.org/87/653787/1/check/neutron-grenade-multinode/7b85b9e/ : SUCCESS in 55m 16s\n- tempest-full http://logs.openstack.org/87/653787/1/check/tempest-full/888f5bf/ : SUCCESS in 1h 43m 50s\n- neutron-grenade http://logs.openstack.org/87/653787/1/check/neutron-grenade/7e124c7/ : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":1},{"id":"99537bf4e144a0b66cc72b930c89fba36e8154ab","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-25 15:16:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca51fff6c4840f5c96c5ebbfaa44d3961a44724c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 15:17:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e033e7bd1d8bdfb944820601093fe1f9360204e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 23:34:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/87/653787/1/gate/openstack-tox-pep8/634b19e/ : SUCCESS in 2m 58s\n- openstack-tox-py27 http://logs.openstack.org/87/653787/1/gate/openstack-tox-py27/3a82519/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/87/653787/1/gate/openstack-tox-py35/354c8c0/ : SUCCESS in 4m 27s\n- openstack-tox-py36 http://logs.openstack.org/87/653787/1/gate/openstack-tox-py36/167800b/ : SUCCESS in 4m 20s\n- openstack-tox-validate http://logs.openstack.org/87/653787/1/gate/openstack-tox-validate/1141859/ : SUCCESS in 3m 13s\n- requirements-tox-babel http://logs.openstack.org/87/653787/1/gate/requirements-tox-babel/476e8c5/ : SUCCESS in 4m 00s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/653787/1/gate/requirements-tox-py27-check-uc/8c8f482/ : SUCCESS in 5m 27s\n- requirements-tox-py35-check-uc http://logs.openstack.org/87/653787/1/gate/requirements-tox-py35-check-uc/660dbcc/ : SUCCESS in 8m 26s\n- cross-cinder-py27 http://logs.openstack.org/87/653787/1/gate/cross-cinder-py27/e29130d/ : FAILURE in 9m 19s\n- cross-cinder-py35 http://logs.openstack.org/87/653787/1/gate/cross-cinder-py35/d74b1d1/ : SUCCESS in 8m 27s\n- cross-glance-py27 http://logs.openstack.org/87/653787/1/gate/cross-glance-py27/4c8a6d2/ : SUCCESS in 3m 50s\n- cross-glance-py35 http://logs.openstack.org/87/653787/1/gate/cross-glance-py35/6b6bdf0/ : SUCCESS in 4m 10s\n- cross-horizon-py27 http://logs.openstack.org/87/653787/1/gate/cross-horizon-py27/3d4f803/ : SUCCESS in 10m 29s\n- cross-horizon-py35 http://logs.openstack.org/87/653787/1/gate/cross-horizon-py35/2420280/ : SUCCESS in 9m 44s\n- cross-horizon-npm http://logs.openstack.org/87/653787/1/gate/cross-horizon-npm/a8613f0/ : SUCCESS in 5m 27s\n- cross-keystone-py27 http://logs.openstack.org/87/653787/1/gate/cross-keystone-py27/2f09422/ : SUCCESS in 23m 35s\n- cross-keystone-py35 http://logs.openstack.org/87/653787/1/gate/cross-keystone-py35/0df9b54/ : SUCCESS in 23m 24s\n- cross-neutron-py27 http://logs.openstack.org/87/653787/1/gate/cross-neutron-py27/79d98d8/ : SUCCESS in 20m 47s\n- cross-neutron-py35 http://logs.openstack.org/87/653787/1/gate/cross-neutron-py35/938cf1b/ : SUCCESS in 24m 24s\n- cross-nova-py27 http://logs.openstack.org/87/653787/1/gate/cross-nova-py27/2e1fef6/ : SUCCESS in 13m 13s\n- cross-nova-py35 http://logs.openstack.org/87/653787/1/gate/cross-nova-py35/ccc9e68/ : SUCCESS in 12m 25s\n- cross-nova-functional http://logs.openstack.org/87/653787/1/gate/cross-nova-functional/c05e454/ : SUCCESS in 17m 21s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/87/653787/1/gate/cross-osc-build-sphinx-docs/1e0f493/html/ : SUCCESS in 4m 53s\n- cross-swift-py27 http://logs.openstack.org/87/653787/1/gate/cross-swift-py27/f6aeae7/ : SUCCESS in 7m 48s\n- cross-swift-py35 http://logs.openstack.org/87/653787/1/gate/cross-swift-py35/bc4ea2a/ : SUCCESS in 7m 49s\n- requirements-integration http://logs.openstack.org/87/653787/1/gate/requirements-integration/fd9aae9/ : SUCCESS in 1h 01m 26s\n- neutron-grenade-multinode http://logs.openstack.org/87/653787/1/gate/neutron-grenade-multinode/3a8e58d/ : SUCCESS in 1h 21m 26s\n- tempest-full http://logs.openstack.org/87/653787/1/gate/tempest-full/ae9a2f8/ : SUCCESS in 1h 53m 33s\n- neutron-grenade http://logs.openstack.org/87/653787/1/gate/neutron-grenade/6c84806/ : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":1},{"id":"f0d123e0db2b703c1ab4e28c30b07b66f7f82bc9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-26 00:47:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3bf15e56bb9d072875bf9c1fe88a62a75d1199a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 07:52:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/87/653787/1/check/openstack-tox-pypy/152b4d3/ : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/87/653787/1/check/openstack-tox-pep8/64a50e3/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/87/653787/1/check/openstack-tox-py27/d4f002e/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/87/653787/1/check/openstack-tox-py35/cd2a99f/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/87/653787/1/check/openstack-tox-py36/5b6fde9/ : SUCCESS in 2m 57s\n- openstack-tox-validate http://logs.openstack.org/87/653787/1/check/openstack-tox-validate/04e4258/ : SUCCESS in 2m 59s\n- requirements-tox-babel http://logs.openstack.org/87/653787/1/check/requirements-tox-babel/ca706c2/ : SUCCESS in 3m 02s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/653787/1/check/requirements-tox-py27-check-uc/936bc79/ : SUCCESS in 7m 11s\n- requirements-tox-py35-check-uc http://logs.openstack.org/87/653787/1/check/requirements-tox-py35-check-uc/7672a64/ : SUCCESS in 6m 15s\n- requirements-tox-py36-check-uc http://logs.openstack.org/87/653787/1/check/requirements-tox-py36-check-uc/01083a9/ : SUCCESS in 6m 37s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/87/653787/1/check/cross-cinder-py27/8cd4bea/ : SUCCESS in 10m 46s\n- cross-cinder-py35 http://logs.openstack.org/87/653787/1/check/cross-cinder-py35/d0d0321/ : SUCCESS in 9m 49s\n- cross-glance-py27 http://logs.openstack.org/87/653787/1/check/cross-glance-py27/d083fb7/ : SUCCESS in 4m 39s\n- cross-glance-py35 http://logs.openstack.org/87/653787/1/check/cross-glance-py35/552bb1e/ : SUCCESS in 4m 33s\n- cross-horizon-py27 http://logs.openstack.org/87/653787/1/check/cross-horizon-py27/01126ab/ : SUCCESS in 9m 19s\n- cross-horizon-py35 http://logs.openstack.org/87/653787/1/check/cross-horizon-py35/5d1ee33/ : SUCCESS in 9m 07s\n- cross-horizon-npm http://logs.openstack.org/87/653787/1/check/cross-horizon-npm/ba0390a/ : SUCCESS in 6m 02s\n- cross-keystone-py27 http://logs.openstack.org/87/653787/1/check/cross-keystone-py27/a4b2d89/ : SUCCESS in 19m 46s\n- cross-keystone-py35 http://logs.openstack.org/87/653787/1/check/cross-keystone-py35/e53d502/ : SUCCESS in 23m 32s\n- cross-neutron-py27 http://logs.openstack.org/87/653787/1/check/cross-neutron-py27/226bb87/ : SUCCESS in 22m 36s\n- cross-neutron-py35 http://logs.openstack.org/87/653787/1/check/cross-neutron-py35/e9cdbf2/ : SUCCESS in 23m 18s\n- cross-nova-py27 http://logs.openstack.org/87/653787/1/check/cross-nova-py27/50fef41/ : SUCCESS in 12m 43s\n- cross-nova-py35 http://logs.openstack.org/87/653787/1/check/cross-nova-py35/093f56f/ : SUCCESS in 13m 06s\n- cross-nova-functional http://logs.openstack.org/87/653787/1/check/cross-nova-functional/8f95d15/ : SUCCESS in 22m 11s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/87/653787/1/check/cross-osc-build-sphinx-docs/408d005/html/ : SUCCESS in 5m 18s\n- cross-swift-py27 http://logs.openstack.org/87/653787/1/check/cross-swift-py27/5b20c8d/ : SUCCESS in 12m 36s\n- cross-swift-py35 http://logs.openstack.org/87/653787/1/check/cross-swift-py35/fe6a2e6/ : SUCCESS in 7m 18s\n- requirements-integration http://logs.openstack.org/87/653787/1/check/requirements-integration/ec84fc8/ : SUCCESS in 1h 01m 15s\n- neutron-grenade-multinode http://logs.openstack.org/87/653787/1/check/neutron-grenade-multinode/4a84c83/ : SUCCESS in 1h 03m 46s\n- tempest-full http://logs.openstack.org/87/653787/1/check/tempest-full/8abd807/ : SUCCESS in 1h 52m 44s\n- neutron-grenade http://logs.openstack.org/87/653787/1/check/neutron-grenade/96d51e4/ : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":1},{"id":"79a6bd6168937b7ca96c530b447c933a6d6747f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 07:53:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87e4e27619ada566421d03f81dd47f7d647378fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 10:26:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b3549884eedd750e3b22c3607f02ab7b5ad285af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 10:26:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/87/653787/1/gate/openstack-tox-pep8/258ea50/ : SUCCESS in 2m 46s\n- openstack-tox-py27 http://logs.openstack.org/87/653787/1/gate/openstack-tox-py27/e9f44f3/ : SUCCESS in 2m 56s\n- openstack-tox-py35 http://logs.openstack.org/87/653787/1/gate/openstack-tox-py35/81b3593/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/87/653787/1/gate/openstack-tox-py36/6f0fc23/ : SUCCESS in 3m 20s\n- openstack-tox-validate http://logs.openstack.org/87/653787/1/gate/openstack-tox-validate/4358fa4/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/87/653787/1/gate/requirements-tox-babel/27ae1f0/ : SUCCESS in 2m 41s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/653787/1/gate/requirements-tox-py27-check-uc/6f74f67/ : SUCCESS in 7m 22s\n- requirements-tox-py35-check-uc http://logs.openstack.org/87/653787/1/gate/requirements-tox-py35-check-uc/46da312/ : SUCCESS in 7m 18s\n- cross-cinder-py27 http://logs.openstack.org/87/653787/1/gate/cross-cinder-py27/2eef26b/ : SUCCESS in 8m 46s\n- cross-cinder-py35 http://logs.openstack.org/87/653787/1/gate/cross-cinder-py35/f925ee5/ : SUCCESS in 11m 13s\n- cross-glance-py27 http://logs.openstack.org/87/653787/1/gate/cross-glance-py27/56572d0/ : SUCCESS in 8m 03s\n- cross-glance-py35 http://logs.openstack.org/87/653787/1/gate/cross-glance-py35/8786d52/ : SUCCESS in 6m 38s\n- cross-horizon-py27 http://logs.openstack.org/87/653787/1/gate/cross-horizon-py27/a71fb16/ : SUCCESS in 10m 49s\n- cross-horizon-py35 http://logs.openstack.org/87/653787/1/gate/cross-horizon-py35/54876ad/ : SUCCESS in 10m 07s\n- cross-horizon-npm http://logs.openstack.org/87/653787/1/gate/cross-horizon-npm/1e5a196/ : SUCCESS in 8m 18s\n- cross-keystone-py27 http://logs.openstack.org/87/653787/1/gate/cross-keystone-py27/b092e64/ : SUCCESS in 20m 43s\n- cross-keystone-py35 http://logs.openstack.org/87/653787/1/gate/cross-keystone-py35/f0618b9/ : SUCCESS in 24m 26s\n- cross-neutron-py27 http://logs.openstack.org/87/653787/1/gate/cross-neutron-py27/99f2383/ : SUCCESS in 20m 31s\n- cross-neutron-py35 http://logs.openstack.org/87/653787/1/gate/cross-neutron-py35/40ef72f/ : SUCCESS in 22m 44s\n- cross-nova-py27 http://logs.openstack.org/87/653787/1/gate/cross-nova-py27/4bb8da6/ : SUCCESS in 12m 32s\n- cross-nova-py35 http://logs.openstack.org/87/653787/1/gate/cross-nova-py35/9cd6e2c/ : SUCCESS in 12m 40s\n- cross-nova-functional http://logs.openstack.org/87/653787/1/gate/cross-nova-functional/ca7d39f/ : SUCCESS in 22m 45s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/87/653787/1/gate/cross-osc-build-sphinx-docs/7e347a8/html/ : SUCCESS in 5m 46s\n- cross-swift-py27 http://logs.openstack.org/87/653787/1/gate/cross-swift-py27/4f56c04/ : SUCCESS in 8m 20s\n- cross-swift-py35 http://logs.openstack.org/87/653787/1/gate/cross-swift-py35/c6b2b6e/ : SUCCESS in 7m 22s\n- requirements-integration http://logs.openstack.org/87/653787/1/gate/requirements-integration/21df7ac/ : SUCCESS in 54m 21s\n- neutron-grenade-multinode http://logs.openstack.org/87/653787/1/gate/neutron-grenade-multinode/58fd696/ : SUCCESS in 1h 12m 14s\n- tempest-full http://logs.openstack.org/87/653787/1/gate/tempest-full/f45d252/ : SUCCESS in 1h 46m 09s\n- neutron-grenade http://logs.openstack.org/87/653787/1/gate/neutron-grenade/6b50e81/ : SUCCESS in 56m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d9f4ac9f759aba462a356ab19be6c356d3bf63f","revisions":{"6d9f4ac9f759aba462a356ab19be6c356d3bf63f":{"kind":"REWORK","_number":1,"created":"2019-04-18 15:14:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/653787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/653787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/653787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/653787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/653787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/653787/1"}}},"commit":{"parents":[{"commit":"f01811923733b376bef96c05cf17c80f445dc774","subject":"update constraint for python-searchlightclient to new release 1.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f01811923733b376bef96c05cf17c80f445dc774"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-18 15:14:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-18 15:14:08.000000000","tz":0},"subject":"update constraint for os-collect-config to new release 10.3.0","message":"update constraint for os-collect-config to new release 10.3.0\n\nChange-Id: I5cacdf5186b4604c11c5fc73ef674aebe8d48305\nmeta:version: 10.3.0\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: Ib8723bbdaab65ea788d3940b6d14b9cdf1b74217\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d9f4ac9f759aba462a356ab19be6c356d3bf63f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d9f4ac9f759aba462a356ab19be6c356d3bf63f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
