)]}'
{"id":"openstack%2Frequirements~653790","triplet_id":"openstack%2Frequirements~stable%2Fstein~I3c3eeb0ff2546b34837fa3677a520b764e51f224","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I3c3eeb0ff2546b34837fa3677a520b764e51f224","subject":"update constraint for tripleo-common to new release 10.7.0","status":"MERGED","created":"2019-04-18 15:17:00.000000000","updated":"2019-04-23 02:44:26.000000000","submitted":"2019-04-23 02:44:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653790-1555987466197-354c625b","meta_rev_id":"0b2bf5263cf66aa17e601cb67fb4fc6d7e742e11","_number":653790,"virtual_id_number":653790,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-23 02:44:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-19 04:25:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-04-22 23:29:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 23:29:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-23 02:44:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf74e015da329ab470c20a3e0223dc4601c4ae1c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-18 15:17:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b911cac0a0311b2abef7d46108921da8b003934f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 00:52:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/90/653790/1/check/openstack-tox-pypy/2f8947b/ : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/90/653790/1/check/openstack-tox-pep8/76f1727/ : SUCCESS in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/90/653790/1/check/openstack-tox-py27/01a1757/ : SUCCESS in 4m 33s\n- openstack-tox-py35 http://logs.openstack.org/90/653790/1/check/openstack-tox-py35/3fefcd0/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/90/653790/1/check/openstack-tox-py36/24ea844/ : SUCCESS in 6m 09s\n- openstack-tox-validate http://logs.openstack.org/90/653790/1/check/openstack-tox-validate/d186b94/ : SUCCESS in 5m 21s\n- requirements-tox-babel http://logs.openstack.org/90/653790/1/check/requirements-tox-babel/ba97f80/ : SUCCESS in 4m 30s\n- requirements-tox-py27-check-uc http://logs.openstack.org/90/653790/1/check/requirements-tox-py27-check-uc/dbb9977/ : SUCCESS in 9m 10s\n- requirements-tox-py35-check-uc http://logs.openstack.org/90/653790/1/check/requirements-tox-py35-check-uc/06d0f4d/ : SUCCESS in 8m 41s\n- requirements-tox-py36-check-uc http://logs.openstack.org/90/653790/1/check/requirements-tox-py36-check-uc/790ce13/ : SUCCESS in 9m 23s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/90/653790/1/check/cross-cinder-py27/ea85f44/ : SUCCESS in 9m 36s\n- cross-cinder-py35 http://logs.openstack.org/90/653790/1/check/cross-cinder-py35/eff39b1/ : SUCCESS in 11m 26s\n- cross-glance-py27 http://logs.openstack.org/90/653790/1/check/cross-glance-py27/5f53e2b/ : SUCCESS in 5m 31s\n- cross-glance-py35 http://logs.openstack.org/90/653790/1/check/cross-glance-py35/8059536/ : SUCCESS in 5m 25s\n- cross-horizon-py27 http://logs.openstack.org/90/653790/1/check/cross-horizon-py27/a3a7af4/ : SUCCESS in 11m 21s\n- cross-horizon-py35 http://logs.openstack.org/90/653790/1/check/cross-horizon-py35/03e3902/ : SUCCESS in 11m 12s\n- cross-horizon-npm http://logs.openstack.org/90/653790/1/check/cross-horizon-npm/c4bb838/ : SUCCESS in 8m 32s\n- cross-keystone-py27 http://logs.openstack.org/90/653790/1/check/cross-keystone-py27/bf60b33/ : SUCCESS in 23m 13s\n- cross-keystone-py35 http://logs.openstack.org/90/653790/1/check/cross-keystone-py35/79c6eec/ : SUCCESS in 26m 18s\n- cross-neutron-py27 http://logs.openstack.org/90/653790/1/check/cross-neutron-py27/2d48782/ : SUCCESS in 22m 02s\n- cross-neutron-py35 http://logs.openstack.org/90/653790/1/check/cross-neutron-py35/e77dd42/ : SUCCESS in 21m 55s\n- cross-nova-py27 http://logs.openstack.org/90/653790/1/check/cross-nova-py27/03359e5/ : SUCCESS in 12m 48s\n- cross-nova-py35 http://logs.openstack.org/90/653790/1/check/cross-nova-py35/f4dffbb/ : SUCCESS in 13m 15s\n- cross-nova-functional http://logs.openstack.org/90/653790/1/check/cross-nova-functional/74c4f8c/ : SUCCESS in 21m 01s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/90/653790/1/check/cross-osc-build-sphinx-docs/12571fa/html/ : SUCCESS in 6m 39s\n- cross-swift-py27 http://logs.openstack.org/90/653790/1/check/cross-swift-py27/ef38851/ : SUCCESS in 8m 47s\n- cross-swift-py35 http://logs.openstack.org/90/653790/1/check/cross-swift-py35/0be001c/ : SUCCESS in 8m 43s\n- requirements-integration http://logs.openstack.org/90/653790/1/check/requirements-integration/bc9bf36/ : SUCCESS in 1h 05m 36s\n- neutron-grenade-multinode http://logs.openstack.org/90/653790/1/check/neutron-grenade-multinode/b3bb926/ : SUCCESS in 1h 12m 42s\n- tempest-full http://logs.openstack.org/90/653790/1/check/tempest-full/329302f/ : SUCCESS in 1h 49m 53s\n- neutron-grenade http://logs.openstack.org/90/653790/1/check/neutron-grenade/2f6a726/ : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":1},{"id":"23a5fd6ed1dad3c8878623c57e5ffb40e3d56fa9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-19 04:25:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d116e047707c0aa9d3a8b25becfef69a1052ae28","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 14:56:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1cb972cab86ebba2a36032d4e7253034c2203b54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:46:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/90/653790/1/check/openstack-tox-pypy/8ed869d/ : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/90/653790/1/check/openstack-tox-pep8/dead32a/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/90/653790/1/check/openstack-tox-py27/c70fdec/ : SUCCESS in 3m 41s\n- openstack-tox-py35 http://logs.openstack.org/90/653790/1/check/openstack-tox-py35/fcbfa99/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/90/653790/1/check/openstack-tox-py36/f76dba3/ : SUCCESS in 3m 32s\n- openstack-tox-validate http://logs.openstack.org/90/653790/1/check/openstack-tox-validate/7456acb/ : SUCCESS in 2m 57s\n- requirements-tox-babel http://logs.openstack.org/90/653790/1/check/requirements-tox-babel/e0a4bd7/ : SUCCESS in 2m 41s\n- requirements-tox-py27-check-uc http://logs.openstack.org/90/653790/1/check/requirements-tox-py27-check-uc/5dfc608/ : SUCCESS in 7m 35s\n- requirements-tox-py35-check-uc http://logs.openstack.org/90/653790/1/check/requirements-tox-py35-check-uc/ec6ff3d/ : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc http://logs.openstack.org/90/653790/1/check/requirements-tox-py36-check-uc/5a93cec/ : SUCCESS in 7m 02s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/90/653790/1/check/cross-cinder-py27/f07710c/ : SUCCESS in 9m 14s\n- cross-cinder-py35 http://logs.openstack.org/90/653790/1/check/cross-cinder-py35/94f437c/ : SUCCESS in 9m 12s\n- cross-glance-py27 http://logs.openstack.org/90/653790/1/check/cross-glance-py27/59391a8/ : SUCCESS in 4m 16s\n- cross-glance-py35 http://logs.openstack.org/90/653790/1/check/cross-glance-py35/cf4a3cf/ : SUCCESS in 4m 49s\n- cross-horizon-py27 http://logs.openstack.org/90/653790/1/check/cross-horizon-py27/1a34774/ : SUCCESS in 11m 35s\n- cross-horizon-py35 http://logs.openstack.org/90/653790/1/check/cross-horizon-py35/4847f45/ : SUCCESS in 10m 08s\n- cross-horizon-npm http://logs.openstack.org/90/653790/1/check/cross-horizon-npm/a1dedcd/ : SUCCESS in 6m 18s\n- cross-keystone-py27 http://logs.openstack.org/90/653790/1/check/cross-keystone-py27/6d6fe9c/ : SUCCESS in 22m 11s\n- cross-keystone-py35 http://logs.openstack.org/90/653790/1/check/cross-keystone-py35/c886e5c/ : SUCCESS in 25m 43s\n- cross-neutron-py27 http://logs.openstack.org/90/653790/1/check/cross-neutron-py27/304eb82/ : SUCCESS in 21m 41s\n- cross-neutron-py35 http://logs.openstack.org/90/653790/1/check/cross-neutron-py35/5eab50d/ : SUCCESS in 25m 44s\n- cross-nova-py27 http://logs.openstack.org/90/653790/1/check/cross-nova-py27/4650bfe/ : SUCCESS in 12m 35s\n- cross-nova-py35 http://logs.openstack.org/90/653790/1/check/cross-nova-py35/f73292a/ : SUCCESS in 12m 06s\n- cross-nova-functional http://logs.openstack.org/90/653790/1/check/cross-nova-functional/fbcb719/ : SUCCESS in 21m 36s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/90/653790/1/check/cross-osc-build-sphinx-docs/d074158/html/ : SUCCESS in 6m 17s\n- cross-swift-py27 http://logs.openstack.org/90/653790/1/check/cross-swift-py27/1dfa863/ : SUCCESS in 8m 08s\n- cross-swift-py35 http://logs.openstack.org/90/653790/1/check/cross-swift-py35/df923a5/ : SUCCESS in 7m 43s\n- requirements-integration http://logs.openstack.org/90/653790/1/check/requirements-integration/060a588/ : SUCCESS in 1h 00m 29s\n- neutron-grenade-multinode http://logs.openstack.org/90/653790/1/check/neutron-grenade-multinode/43ae008/ : SUCCESS in 1h 19m 06s\n- tempest-full http://logs.openstack.org/90/653790/1/check/tempest-full/8e5c092/ : SUCCESS in 1h 42m 46s\n- neutron-grenade http://logs.openstack.org/90/653790/1/check/neutron-grenade/bb9d01a/ : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":1},{"id":"cd4c8c6383a18a0cd1f8888b28431326322e5941","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 23:29:08.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f699ffb7753cbe0cd0957b43d299a3d70e5a914","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 23:30:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b65abe04ece8af8ca1540811859b809a66c162a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 02:44:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/90/653790/1/gate/openstack-tox-pep8/02b4381/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/90/653790/1/gate/openstack-tox-py27/814b421/ : SUCCESS in 3m 56s\n- openstack-tox-py35 http://logs.openstack.org/90/653790/1/gate/openstack-tox-py35/981782e/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/90/653790/1/gate/openstack-tox-py36/e26f4e8/ : SUCCESS in 3m 33s\n- openstack-tox-validate http://logs.openstack.org/90/653790/1/gate/openstack-tox-validate/6aa93d3/ : SUCCESS in 3m 11s\n- requirements-tox-babel http://logs.openstack.org/90/653790/1/gate/requirements-tox-babel/7a896a6/ : SUCCESS in 2m 43s\n- requirements-tox-py27-check-uc http://logs.openstack.org/90/653790/1/gate/requirements-tox-py27-check-uc/8560765/ : SUCCESS in 7m 44s\n- requirements-tox-py35-check-uc http://logs.openstack.org/90/653790/1/gate/requirements-tox-py35-check-uc/e759bc7/ : SUCCESS in 7m 43s\n- cross-cinder-py27 http://logs.openstack.org/90/653790/1/gate/cross-cinder-py27/26d8316/ : SUCCESS in 7m 27s\n- cross-cinder-py35 http://logs.openstack.org/90/653790/1/gate/cross-cinder-py35/8332ab0/ : SUCCESS in 11m 45s\n- cross-glance-py27 http://logs.openstack.org/90/653790/1/gate/cross-glance-py27/dda4243/ : SUCCESS in 4m 57s\n- cross-glance-py35 http://logs.openstack.org/90/653790/1/gate/cross-glance-py35/e416c98/ : SUCCESS in 4m 33s\n- cross-horizon-py27 http://logs.openstack.org/90/653790/1/gate/cross-horizon-py27/aacd2cb/ : SUCCESS in 10m 21s\n- cross-horizon-py35 http://logs.openstack.org/90/653790/1/gate/cross-horizon-py35/5e8c665/ : SUCCESS in 9m 12s\n- cross-horizon-npm http://logs.openstack.org/90/653790/1/gate/cross-horizon-npm/7680758/ : SUCCESS in 6m 21s\n- cross-keystone-py27 http://logs.openstack.org/90/653790/1/gate/cross-keystone-py27/ba2ec60/ : SUCCESS in 23m 41s\n- cross-keystone-py35 http://logs.openstack.org/90/653790/1/gate/cross-keystone-py35/fd2c22c/ : SUCCESS in 26m 10s\n- cross-neutron-py27 http://logs.openstack.org/90/653790/1/gate/cross-neutron-py27/957ef37/ : SUCCESS in 23m 20s\n- cross-neutron-py35 http://logs.openstack.org/90/653790/1/gate/cross-neutron-py35/3509613/ : SUCCESS in 22m 54s\n- cross-nova-py27 http://logs.openstack.org/90/653790/1/gate/cross-nova-py27/fca5088/ : SUCCESS in 11m 14s\n- cross-nova-py35 http://logs.openstack.org/90/653790/1/gate/cross-nova-py35/9c8c210/ : SUCCESS in 14m 56s\n- cross-nova-functional http://logs.openstack.org/90/653790/1/gate/cross-nova-functional/c85a208/ : SUCCESS in 21m 20s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/90/653790/1/gate/cross-osc-build-sphinx-docs/2747cde/html/ : SUCCESS in 5m 20s\n- cross-swift-py27 http://logs.openstack.org/90/653790/1/gate/cross-swift-py27/4b64358/ : SUCCESS in 7m 49s\n- cross-swift-py35 http://logs.openstack.org/90/653790/1/gate/cross-swift-py35/f945bdf/ : SUCCESS in 7m 27s\n- requirements-integration http://logs.openstack.org/90/653790/1/gate/requirements-integration/457a485/ : SUCCESS in 50m 15s\n- neutron-grenade-multinode http://logs.openstack.org/90/653790/1/gate/neutron-grenade-multinode/f318019/ : SUCCESS in 1h 01m 56s\n- tempest-full http://logs.openstack.org/90/653790/1/gate/tempest-full/9ff3aa2/ : SUCCESS in 1h 53m 42s\n- neutron-grenade http://logs.openstack.org/90/653790/1/gate/neutron-grenade/502993f/ : SUCCESS in 59m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6054d8bac481c8861f66bdeed67565ae58522111","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 02:44:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a027361866db582545944a2a444b8fd05991bf3","revisions":{"9a027361866db582545944a2a444b8fd05991bf3":{"kind":"REWORK","_number":1,"created":"2019-04-18 15:17:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/653790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/653790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/653790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/653790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/653790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/653790/1"}}},"commit":{"parents":[{"commit":"f01811923733b376bef96c05cf17c80f445dc774","subject":"update constraint for python-searchlightclient to new release 1.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f01811923733b376bef96c05cf17c80f445dc774"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-18 15:16:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-18 15:16:58.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 10.7.0","message":"update constraint for tripleo-common to new release 10.7.0\n\nChange-Id: I3c3eeb0ff2546b34837fa3677a520b764e51f224\nmeta:version: 10.7.0\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: Ib8723bbdaab65ea788d3940b6d14b9cdf1b74217\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9a027361866db582545944a2a444b8fd05991bf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9a027361866db582545944a2a444b8fd05991bf3"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
