)]}'
{"id":"openstack%2Frequirements~653791","triplet_id":"openstack%2Frequirements~stable%2Fstein~I0ad2ccf07db6ec27e59d30b21474870c638ff125","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I0ad2ccf07db6ec27e59d30b21474870c638ff125","subject":"update constraint for os-net-config to new release 10.4.0","status":"MERGED","created":"2019-04-18 15:19:30.000000000","updated":"2019-04-26 02:33:40.000000000","submitted":"2019-04-26 02:33:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653791-1556246020667-f2659bb3","meta_rev_id":"6fb4851a6af3ab4a8995c6cc1bd808c177299757","_number":653791,"virtual_id_number":653791,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-04-26 02:33:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-19 04:25:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-25 15:16:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-04-25 15:16:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-19 04:25:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-25 15:16:54.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-26 02:33:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f496a89575bab8dfa444f34741dae8ad082a9445","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-18 15:19:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7afa57d02781d0b9bbf4e85c4e2e12ad91b4612","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 01:24:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/91/653791/1/check/openstack-tox-pypy/16917a9/ : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/653791/1/check/openstack-tox-pep8/e5c9fc7/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/91/653791/1/check/openstack-tox-py27/106b182/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/91/653791/1/check/openstack-tox-py35/5ab18ed/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/91/653791/1/check/openstack-tox-py36/dc56322/ : SUCCESS in 4m 48s\n- openstack-tox-validate http://logs.openstack.org/91/653791/1/check/openstack-tox-validate/ddb04a4/ : SUCCESS in 2m 54s\n- requirements-tox-babel http://logs.openstack.org/91/653791/1/check/requirements-tox-babel/6611611/ : SUCCESS in 5m 13s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653791/1/check/requirements-tox-py27-check-uc/66edf82/ : SUCCESS in 8m 10s\n- requirements-tox-py35-check-uc http://logs.openstack.org/91/653791/1/check/requirements-tox-py35-check-uc/0b18d4f/ : SUCCESS in 9m 53s\n- requirements-tox-py36-check-uc http://logs.openstack.org/91/653791/1/check/requirements-tox-py36-check-uc/30de7af/ : SUCCESS in 8m 52s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/91/653791/1/check/cross-cinder-py27/8e820b8/ : SUCCESS in 10m 37s\n- cross-cinder-py35 http://logs.openstack.org/91/653791/1/check/cross-cinder-py35/27e21bf/ : SUCCESS in 10m 04s\n- cross-glance-py27 http://logs.openstack.org/91/653791/1/check/cross-glance-py27/4011dcf/ : SUCCESS in 5m 55s\n- cross-glance-py35 http://logs.openstack.org/91/653791/1/check/cross-glance-py35/2d12ba0/ : SUCCESS in 6m 03s\n- cross-horizon-py27 http://logs.openstack.org/91/653791/1/check/cross-horizon-py27/88b97e4/ : SUCCESS in 12m 18s\n- cross-horizon-py35 http://logs.openstack.org/91/653791/1/check/cross-horizon-py35/78311ea/ : SUCCESS in 9m 41s\n- cross-horizon-npm http://logs.openstack.org/91/653791/1/check/cross-horizon-npm/c30c33f/ : SUCCESS in 8m 28s\n- cross-keystone-py27 http://logs.openstack.org/91/653791/1/check/cross-keystone-py27/e989466/ : SUCCESS in 23m 28s\n- cross-keystone-py35 http://logs.openstack.org/91/653791/1/check/cross-keystone-py35/d1cee7c/ : SUCCESS in 25m 21s\n- cross-neutron-py27 http://logs.openstack.org/91/653791/1/check/cross-neutron-py27/124e13b/ : SUCCESS in 21m 46s\n- cross-neutron-py35 http://logs.openstack.org/91/653791/1/check/cross-neutron-py35/4b64e1a/ : SUCCESS in 23m 10s\n- cross-nova-py27 http://logs.openstack.org/91/653791/1/check/cross-nova-py27/5d3a005/ : SUCCESS in 13m 13s\n- cross-nova-py35 http://logs.openstack.org/91/653791/1/check/cross-nova-py35/1a06f12/ : SUCCESS in 15m 57s\n- cross-nova-functional http://logs.openstack.org/91/653791/1/check/cross-nova-functional/33ad689/ : SUCCESS in 22m 10s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/91/653791/1/check/cross-osc-build-sphinx-docs/caf3d4f/html/ : SUCCESS in 9m 01s\n- cross-swift-py27 http://logs.openstack.org/91/653791/1/check/cross-swift-py27/9c5d804/ : SUCCESS in 8m 22s\n- cross-swift-py35 http://logs.openstack.org/91/653791/1/check/cross-swift-py35/86be294/ : SUCCESS in 8m 38s\n- requirements-integration http://logs.openstack.org/91/653791/1/check/requirements-integration/09c5b5d/ : SUCCESS in 1h 08m 26s\n- neutron-grenade-multinode http://logs.openstack.org/91/653791/1/check/neutron-grenade-multinode/032cbd4/ : SUCCESS in 1h 10m 14s\n- tempest-full http://logs.openstack.org/91/653791/1/check/tempest-full/5ae8f87/ : SUCCESS in 1h 56m 37s\n- neutron-grenade http://logs.openstack.org/91/653791/1/check/neutron-grenade/f6b2f01/ : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b96f592e5f4a783a7cb48d0b6a2746c2b1bc519f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-19 04:25:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"465720e03ba14e6e5f6828a961cf2fa490a7f98b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 14:56:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"532cad5e8b4c140244e671bb055ed6d3e13cc6a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:14:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/91/653791/1/check/openstack-tox-pypy/76f2b8c/ : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/653791/1/check/openstack-tox-pep8/842c490/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/91/653791/1/check/openstack-tox-py27/adaae1f/ : SUCCESS in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/91/653791/1/check/openstack-tox-py35/f6e6240/ : SUCCESS in 3m 45s\n- openstack-tox-py36 http://logs.openstack.org/91/653791/1/check/openstack-tox-py36/d168ec8/ : SUCCESS in 3m 31s\n- openstack-tox-validate http://logs.openstack.org/91/653791/1/check/openstack-tox-validate/ba3ed9c/ : SUCCESS in 3m 11s\n- requirements-tox-babel http://logs.openstack.org/91/653791/1/check/requirements-tox-babel/856f047/ : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653791/1/check/requirements-tox-py27-check-uc/fbd23ee/ : SUCCESS in 7m 41s\n- requirements-tox-py35-check-uc http://logs.openstack.org/91/653791/1/check/requirements-tox-py35-check-uc/98fea10/ : SUCCESS in 7m 10s\n- requirements-tox-py36-check-uc http://logs.openstack.org/91/653791/1/check/requirements-tox-py36-check-uc/62e1e71/ : SUCCESS in 6m 58s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/91/653791/1/check/cross-cinder-py27/b275e96/ : SUCCESS in 9m 04s\n- cross-cinder-py35 http://logs.openstack.org/91/653791/1/check/cross-cinder-py35/9abc4f1/ : SUCCESS in 9m 54s\n- cross-glance-py27 http://logs.openstack.org/91/653791/1/check/cross-glance-py27/e82c71f/ : SUCCESS in 4m 16s\n- cross-glance-py35 http://logs.openstack.org/91/653791/1/check/cross-glance-py35/cdab457/ : SUCCESS in 4m 37s\n- cross-horizon-py27 http://logs.openstack.org/91/653791/1/check/cross-horizon-py27/cb18ae4/ : SUCCESS in 9m 34s\n- cross-horizon-py35 http://logs.openstack.org/91/653791/1/check/cross-horizon-py35/01437ce/ : SUCCESS in 10m 14s\n- cross-horizon-npm http://logs.openstack.org/91/653791/1/check/cross-horizon-npm/a9e2b58/ : SUCCESS in 7m 59s\n- cross-keystone-py27 http://logs.openstack.org/91/653791/1/check/cross-keystone-py27/82b5216/ : SUCCESS in 21m 34s\n- cross-keystone-py35 http://logs.openstack.org/91/653791/1/check/cross-keystone-py35/7189767/ : SUCCESS in 24m 31s\n- cross-neutron-py27 http://logs.openstack.org/91/653791/1/check/cross-neutron-py27/1df8cc8/ : SUCCESS in 25m 16s\n- cross-neutron-py35 http://logs.openstack.org/91/653791/1/check/cross-neutron-py35/5bb26ee/ : SUCCESS in 23m 54s\n- cross-nova-py27 http://logs.openstack.org/91/653791/1/check/cross-nova-py27/3354aa7/ : SUCCESS in 13m 54s\n- cross-nova-py35 http://logs.openstack.org/91/653791/1/check/cross-nova-py35/e9e12da/ : SUCCESS in 13m 51s\n- cross-nova-functional http://logs.openstack.org/91/653791/1/check/cross-nova-functional/7fcf9fb/ : SUCCESS in 19m 05s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/91/653791/1/check/cross-osc-build-sphinx-docs/0bb240d/html/ : SUCCESS in 5m 33s\n- cross-swift-py27 http://logs.openstack.org/91/653791/1/check/cross-swift-py27/d5023a0/ : SUCCESS in 7m 28s\n- cross-swift-py35 http://logs.openstack.org/91/653791/1/check/cross-swift-py35/46fb5f0/ : SUCCESS in 7m 39s\n- requirements-integration http://logs.openstack.org/91/653791/1/check/requirements-integration/e1731f9/ : SUCCESS in 56m 12s\n- neutron-grenade-multinode http://logs.openstack.org/91/653791/1/check/neutron-grenade-multinode/03b8d72/ : SUCCESS in 1h 05m 39s\n- tempest-full http://logs.openstack.org/91/653791/1/check/tempest-full/cd022bb/ : SUCCESS in 1h 47m 23s\n- neutron-grenade http://logs.openstack.org/91/653791/1/check/neutron-grenade/d361248/ : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e5cd2b5a7cd0a147200ff69b74e695c8e1a2c9f4","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-25 15:16:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"acc13f293203b35a4c5710b079b539afdc9b07f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 15:17:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d00e48f11e7ee53c50aaa8f354a6700f89ad82c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 02:33:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/91/653791/1/gate/openstack-tox-pep8/0ae05f6/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/91/653791/1/gate/openstack-tox-py27/be0f82f/ : SUCCESS in 3m 21s\n- openstack-tox-py35 http://logs.openstack.org/91/653791/1/gate/openstack-tox-py35/53f912f/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/91/653791/1/gate/openstack-tox-py36/7e5b073/ : SUCCESS in 3m 23s\n- openstack-tox-validate http://logs.openstack.org/91/653791/1/gate/openstack-tox-validate/4eae0e3/ : SUCCESS in 3m 11s\n- requirements-tox-babel http://logs.openstack.org/91/653791/1/gate/requirements-tox-babel/7246703/ : SUCCESS in 2m 51s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/653791/1/gate/requirements-tox-py27-check-uc/60761ef/ : SUCCESS in 6m 56s\n- requirements-tox-py35-check-uc http://logs.openstack.org/91/653791/1/gate/requirements-tox-py35-check-uc/621674e/ : SUCCESS in 9m 16s\n- cross-cinder-py27 http://logs.openstack.org/91/653791/1/gate/cross-cinder-py27/1fcdbd6/ : SUCCESS in 8m 41s\n- cross-cinder-py35 http://logs.openstack.org/91/653791/1/gate/cross-cinder-py35/735e18c/ : SUCCESS in 10m 57s\n- cross-glance-py27 http://logs.openstack.org/91/653791/1/gate/cross-glance-py27/29c2d48/ : SUCCESS in 3m 59s\n- cross-glance-py35 http://logs.openstack.org/91/653791/1/gate/cross-glance-py35/87bb5bb/ : SUCCESS in 4m 28s\n- cross-horizon-py27 http://logs.openstack.org/91/653791/1/gate/cross-horizon-py27/9de5d4e/ : SUCCESS in 10m 12s\n- cross-horizon-py35 http://logs.openstack.org/91/653791/1/gate/cross-horizon-py35/87c1ef4/ : SUCCESS in 9m 37s\n- cross-horizon-npm http://logs.openstack.org/91/653791/1/gate/cross-horizon-npm/f216f95/ : SUCCESS in 5m 22s\n- cross-keystone-py27 http://logs.openstack.org/91/653791/1/gate/cross-keystone-py27/1ca33d0/ : SUCCESS in 21m 27s\n- cross-keystone-py35 http://logs.openstack.org/91/653791/1/gate/cross-keystone-py35/f23b779/ : SUCCESS in 27m 21s\n- cross-neutron-py27 http://logs.openstack.org/91/653791/1/gate/cross-neutron-py27/abab4da/ : SUCCESS in 22m 24s\n- cross-neutron-py35 http://logs.openstack.org/91/653791/1/gate/cross-neutron-py35/4b42812/ : SUCCESS in 22m 52s\n- cross-nova-py27 http://logs.openstack.org/91/653791/1/gate/cross-nova-py27/c1ecc71/ : SUCCESS in 11m 44s\n- cross-nova-py35 http://logs.openstack.org/91/653791/1/gate/cross-nova-py35/42436ff/ : SUCCESS in 14m 10s\n- cross-nova-functional http://logs.openstack.org/91/653791/1/gate/cross-nova-functional/c28f165/ : SUCCESS in 19m 27s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/91/653791/1/gate/cross-osc-build-sphinx-docs/ebb9b78/html/ : SUCCESS in 5m 23s\n- cross-swift-py27 http://logs.openstack.org/91/653791/1/gate/cross-swift-py27/7b2a58a/ : SUCCESS in 9m 02s\n- cross-swift-py35 http://logs.openstack.org/91/653791/1/gate/cross-swift-py35/d1d4740/ : SUCCESS in 7m 55s\n- requirements-integration http://logs.openstack.org/91/653791/1/gate/requirements-integration/589af39/ : SUCCESS in 50m 31s\n- neutron-grenade-multinode http://logs.openstack.org/91/653791/1/gate/neutron-grenade-multinode/55222cd/ : SUCCESS in 1h 07m 03s\n- tempest-full http://logs.openstack.org/91/653791/1/gate/tempest-full/1bf67c7/ : SUCCESS in 1h 49m 20s\n- neutron-grenade http://logs.openstack.org/91/653791/1/gate/neutron-grenade/62d4669/ : SUCCESS in 46m 04s","accounts_in_message":[],"_revision_number":1},{"id":"42d1ea211785dff49f75eda253ce8486ef01312c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 02:33:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4cf8117fc3f579cf32b1851f751adff5d675900","revisions":{"f4cf8117fc3f579cf32b1851f751adff5d675900":{"kind":"REWORK","_number":1,"created":"2019-04-18 15:19:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/653791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/653791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/653791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/653791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/653791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/653791/1"}}},"commit":{"parents":[{"commit":"f01811923733b376bef96c05cf17c80f445dc774","subject":"update constraint for python-searchlightclient to new release 1.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f01811923733b376bef96c05cf17c80f445dc774"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-18 15:19:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-18 15:19:29.000000000","tz":0},"subject":"update constraint for os-net-config to new release 10.4.0","message":"update constraint for os-net-config to new release 10.4.0\n\nChange-Id: I0ad2ccf07db6ec27e59d30b21474870c638ff125\nmeta:version: 10.4.0\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: Ib8723bbdaab65ea788d3940b6d14b9cdf1b74217\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4cf8117fc3f579cf32b1851f751adff5d675900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4cf8117fc3f579cf32b1851f751adff5d675900"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
