)]}'
{"id":"openstack%2Frequirements~653887","triplet_id":"openstack%2Frequirements~master~Ibf884dbc003e120378089e05e315f9938a705484","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibf884dbc003e120378089e05e315f9938a705484","subject":"update constraint for neutron-lib to new release 1.26.0","status":"MERGED","created":"2019-04-19 00:28:32.000000000","updated":"2019-04-25 22:02:16.000000000","submitted":"2019-04-25 22:02:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653887-1556229736398-1b918c92","meta_rev_id":"d80bed1706262ea8d77b67b8b307f6268a0bf9de","_number":653887,"virtual_id_number":653887,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-04-22 01:23:40.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-04-25 22:02:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-23 14:43:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-04-23 12:10:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-04-25 15:03:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 01:23:40.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-04-23 12:10:34.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-04-25 15:03:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-25 22:02:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63cc1da0bf712d800da00ef627cc349567dae1b7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-19 00:28:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45ab525ceedea3653ec5e72b239fa75bd0d8ad2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 11:00:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/87/653887/1/check/openstack-tox-docs/a0614fe/html/ : SUCCESS in 5m 12s\n- openstack-tox-pypy http://logs.openstack.org/87/653887/1/check/openstack-tox-pypy/a742213/ : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/87/653887/1/check/openstack-tox-pep8/6410be8/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/87/653887/1/check/openstack-tox-py27/f46ce59/ : SUCCESS in 5m 08s\n- openstack-tox-py35 http://logs.openstack.org/87/653887/1/check/openstack-tox-py35/5492a6d/ : SUCCESS in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/87/653887/1/check/openstack-tox-py36/33e2440/ : SUCCESS in 5m 44s\n- openstack-tox-validate http://logs.openstack.org/87/653887/1/check/openstack-tox-validate/fa18df8/ : SUCCESS in 4m 40s\n- requirements-tox-babel http://logs.openstack.org/87/653887/1/check/requirements-tox-babel/e8c1163/ : SUCCESS in 5m 18s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/653887/1/check/requirements-tox-py27-check-uc/c745def/ : SUCCESS in 8m 35s\n- requirements-tox-py35-check-uc http://logs.openstack.org/87/653887/1/check/requirements-tox-py35-check-uc/160245f/ : SUCCESS in 9m 59s\n- requirements-tox-py36-check-uc http://logs.openstack.org/87/653887/1/check/requirements-tox-py36-check-uc/317d3be/ : SUCCESS in 9m 37s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/87/653887/1/check/cross-cinder-py27/a517e9a/ : SUCCESS in 11m 10s\n- cross-cinder-py36 http://logs.openstack.org/87/653887/1/check/cross-cinder-py36/dc02b85/ : SUCCESS in 10m 11s\n- cross-glance-py27 http://logs.openstack.org/87/653887/1/check/cross-glance-py27/bd6ab8c/ : SUCCESS in 6m 12s\n- cross-glance-py36 http://logs.openstack.org/87/653887/1/check/cross-glance-py36/d22e453/ : SUCCESS in 6m 55s\n- cross-horizon-py27 http://logs.openstack.org/87/653887/1/check/cross-horizon-py27/bea597e/ : SUCCESS in 11m 56s\n- cross-horizon-py36 http://logs.openstack.org/87/653887/1/check/cross-horizon-py36/7b0af44/ : SUCCESS in 10m 31s\n- cross-horizon-npm http://logs.openstack.org/87/653887/1/check/cross-horizon-npm/126e5cf/ : SUCCESS in 7m 46s\n- cross-keystone-py27 http://logs.openstack.org/87/653887/1/check/cross-keystone-py27/06307be/ : SUCCESS in 23m 26s\n- cross-keystone-py36 http://logs.openstack.org/87/653887/1/check/cross-keystone-py36/d3017c1/ : SUCCESS in 23m 03s\n- cross-neutron-py27 http://logs.openstack.org/87/653887/1/check/cross-neutron-py27/0130c2f/ : SUCCESS in 21m 39s\n- cross-neutron-py36 http://logs.openstack.org/87/653887/1/check/cross-neutron-py36/38dd4f2/ : SUCCESS in 27m 42s\n- cross-nova-py27 http://logs.openstack.org/87/653887/1/check/cross-nova-py27/ce07c43/ : SUCCESS in 13m 40s\n- cross-nova-py36 http://logs.openstack.org/87/653887/1/check/cross-nova-py36/cc457fa/ : SUCCESS in 15m 59s\n- cross-nova-functional http://logs.openstack.org/87/653887/1/check/cross-nova-functional/f5e3a8b/ : SUCCESS in 22m 27s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/87/653887/1/check/cross-osc-build-sphinx-docs/bf57d93/html/ : SUCCESS in 6m 20s\n- cross-swift-py27 http://logs.openstack.org/87/653887/1/check/cross-swift-py27/94d84c4/ : SUCCESS in 10m 36s\n- cross-swift-py36 http://logs.openstack.org/87/653887/1/check/cross-swift-py36/0598a1c/ : SUCCESS in 8m 58s\n- requirements-integration finger://ze06.openstack.org/832d7963bf9349d3a7927bb1d9219965 : RETRY_LIMIT in 1h 07m 00s\n- neutron-grenade-multinode finger://ze06.openstack.org/1a6de9a7590c4cbaa37eba1c2e7b50f7 : RETRY_LIMIT in 1h 07m 24s\n- tempest-full finger://ze12.openstack.org/f62fb39311d541b1b099485829e7dae7 : RETRY_LIMIT in 1h 10m 10s\n- neutron-grenade http://logs.openstack.org/87/653887/1/check/neutron-grenade/0297a1c/ : FAILURE in 46m 46s","accounts_in_message":[],"_revision_number":1},{"id":"34e1e07a372137934da95492549520f55fdfd3ae","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-04-22 01:23:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"928941815ef076bc06e0d2745fa2a3827fbc200d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 03:10:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/87/653887/1/check/openstack-tox-docs/b55ef5c/html/ : SUCCESS in 2m 41s\n- openstack-tox-pypy http://logs.openstack.org/87/653887/1/check/openstack-tox-pypy/c9276e9/ : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/87/653887/1/check/openstack-tox-pep8/5b944dd/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/87/653887/1/check/openstack-tox-py27/f0dd7be/ : SUCCESS in 6m 03s\n- openstack-tox-py35 http://logs.openstack.org/87/653887/1/check/openstack-tox-py35/7edb5b1/ : SUCCESS in 7m 10s\n- openstack-tox-py36 http://logs.openstack.org/87/653887/1/check/openstack-tox-py36/91ab10e/ : SUCCESS in 4m 05s\n- openstack-tox-validate http://logs.openstack.org/87/653887/1/check/openstack-tox-validate/62990e9/ : SUCCESS in 4m 06s\n- requirements-tox-babel http://logs.openstack.org/87/653887/1/check/requirements-tox-babel/6bfd2c1/ : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/653887/1/check/requirements-tox-py27-check-uc/9ab56e2/ : SUCCESS in 10m 00s\n- requirements-tox-py35-check-uc http://logs.openstack.org/87/653887/1/check/requirements-tox-py35-check-uc/42d2f1d/ : SUCCESS in 10m 59s\n- requirements-tox-py36-check-uc http://logs.openstack.org/87/653887/1/check/requirements-tox-py36-check-uc/f01c72e/ : SUCCESS in 8m 02s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/87/653887/1/check/cross-cinder-py27/3e68d3c/ : SUCCESS in 8m 24s\n- cross-cinder-py36 http://logs.openstack.org/87/653887/1/check/cross-cinder-py36/93e7d87/ : SUCCESS in 8m 52s\n- cross-glance-py27 http://logs.openstack.org/87/653887/1/check/cross-glance-py27/64bf22b/ : SUCCESS in 5m 42s\n- cross-glance-py36 http://logs.openstack.org/87/653887/1/check/cross-glance-py36/cff5aa2/ : SUCCESS in 5m 18s\n- cross-horizon-py27 http://logs.openstack.org/87/653887/1/check/cross-horizon-py27/05d8418/ : SUCCESS in 11m 16s\n- cross-horizon-py36 http://logs.openstack.org/87/653887/1/check/cross-horizon-py36/8ea58df/ : SUCCESS in 8m 50s\n- cross-horizon-npm http://logs.openstack.org/87/653887/1/check/cross-horizon-npm/6af92ee/ : SUCCESS in 7m 08s\n- cross-keystone-py27 http://logs.openstack.org/87/653887/1/check/cross-keystone-py27/badd117/ : SUCCESS in 20m 01s\n- cross-keystone-py36 http://logs.openstack.org/87/653887/1/check/cross-keystone-py36/1e9faf2/ : SUCCESS in 22m 27s\n- cross-neutron-py27 http://logs.openstack.org/87/653887/1/check/cross-neutron-py27/3086cf9/ : SUCCESS in 18m 39s\n- cross-neutron-py36 http://logs.openstack.org/87/653887/1/check/cross-neutron-py36/93f7467/ : SUCCESS in 16m 58s\n- cross-nova-py27 http://logs.openstack.org/87/653887/1/check/cross-nova-py27/be71720/ : SUCCESS in 11m 42s\n- cross-nova-py36 http://logs.openstack.org/87/653887/1/check/cross-nova-py36/5b02141/ : FAILURE in 10m 34s\n- cross-nova-functional http://logs.openstack.org/87/653887/1/check/cross-nova-functional/fb3ebff/ : SUCCESS in 16m 57s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/87/653887/1/check/cross-osc-build-sphinx-docs/0b26ab5/html/ : SUCCESS in 6m 51s\n- cross-swift-py27 http://logs.openstack.org/87/653887/1/check/cross-swift-py27/bd4282a/ : SUCCESS in 7m 17s\n- cross-swift-py36 http://logs.openstack.org/87/653887/1/check/cross-swift-py36/3aa51f5/ : SUCCESS in 6m 38s\n- requirements-integration http://logs.openstack.org/87/653887/1/check/requirements-integration/bb531cb/ : SUCCESS in 46m 31s\n- neutron-grenade-multinode http://logs.openstack.org/87/653887/1/check/neutron-grenade-multinode/951e211/ : SUCCESS in 1h 10m 43s\n- tempest-full http://logs.openstack.org/87/653887/1/check/tempest-full/39a747c/ : SUCCESS in 1h 42m 01s\n- neutron-grenade http://logs.openstack.org/87/653887/1/check/neutron-grenade/aa81697/ : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c9cd8148dc6c2d5b249e822ecb9ff67342ec9bef","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-22 13:09:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e632d5c4d6b60e12b418321314108c8a010cd89","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 14:56:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62460696be2f0f0a0ba3fd65d510b844fbc718a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 15:36:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/87/653887/1/check/openstack-tox-docs/a4683f5/html/ : SUCCESS in 4m 12s\n- openstack-tox-pypy http://logs.openstack.org/87/653887/1/check/openstack-tox-pypy/8593469/ : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/87/653887/1/check/openstack-tox-pep8/8254db3/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/87/653887/1/check/openstack-tox-py27/a7010ee/ : SUCCESS in 3m 17s\n- openstack-tox-py35 http://logs.openstack.org/87/653887/1/check/openstack-tox-py35/16277ed/ : SUCCESS in 3m 06s\n- openstack-tox-py36 http://logs.openstack.org/87/653887/1/check/openstack-tox-py36/b595f1b/ : SUCCESS in 3m 40s\n- openstack-tox-validate http://logs.openstack.org/87/653887/1/check/openstack-tox-validate/1783258/ : SUCCESS in 4m 57s\n- requirements-tox-babel http://logs.openstack.org/87/653887/1/check/requirements-tox-babel/585f13e/ : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/653887/1/check/requirements-tox-py27-check-uc/e7ecf96/ : SUCCESS in 14m 47s\n- requirements-tox-py35-check-uc http://logs.openstack.org/87/653887/1/check/requirements-tox-py35-check-uc/244c721/ : SUCCESS in 16m 06s\n- requirements-tox-py36-check-uc http://logs.openstack.org/87/653887/1/check/requirements-tox-py36-check-uc/2898ba3/ : SUCCESS in 10m 40s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/87/653887/1/check/cross-cinder-py27/4927376/ : SUCCESS in 10m 15s\n- cross-cinder-py36 http://logs.openstack.org/87/653887/1/check/cross-cinder-py36/aba0efc/ : SUCCESS in 9m 58s\n- cross-glance-py27 http://logs.openstack.org/87/653887/1/check/cross-glance-py27/3a597c9/ : SUCCESS in 6m 27s\n- cross-glance-py36 http://logs.openstack.org/87/653887/1/check/cross-glance-py36/a340a69/ : SUCCESS in 6m 51s\n- cross-horizon-py27 http://logs.openstack.org/87/653887/1/check/cross-horizon-py27/1a0f602/ : SUCCESS in 12m 00s\n- cross-horizon-py36 http://logs.openstack.org/87/653887/1/check/cross-horizon-py36/0e54f44/ : SUCCESS in 8m 41s\n- cross-horizon-npm http://logs.openstack.org/87/653887/1/check/cross-horizon-npm/0bd5ea9/ : SUCCESS in 6m 55s\n- cross-keystone-py27 http://logs.openstack.org/87/653887/1/check/cross-keystone-py27/30e8b4e/ : SUCCESS in 25m 36s\n- cross-keystone-py36 http://logs.openstack.org/87/653887/1/check/cross-keystone-py36/956f078/ : SUCCESS in 25m 34s\n- cross-neutron-py27 http://logs.openstack.org/87/653887/1/check/cross-neutron-py27/ee53b1e/ : SUCCESS in 23m 16s\n- cross-neutron-py36 http://logs.openstack.org/87/653887/1/check/cross-neutron-py36/b7d5e00/ : SUCCESS in 27m 47s\n- cross-nova-py27 http://logs.openstack.org/87/653887/1/check/cross-nova-py27/d1f6837/ : SUCCESS in 14m 40s\n- cross-nova-py36 http://logs.openstack.org/87/653887/1/check/cross-nova-py36/35994c5/ : FAILURE in 13m 20s\n- cross-nova-functional http://logs.openstack.org/87/653887/1/check/cross-nova-functional/bad4894/ : SUCCESS in 22m 23s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/87/653887/1/check/cross-osc-build-sphinx-docs/dc473a2/html/ : SUCCESS in 5m 59s\n- cross-swift-py27 http://logs.openstack.org/87/653887/1/check/cross-swift-py27/947cc31/ : SUCCESS in 9m 17s\n- cross-swift-py36 http://logs.openstack.org/87/653887/1/check/cross-swift-py36/8b6fd98/ : SUCCESS in 8m 42s\n- requirements-integration http://logs.openstack.org/87/653887/1/check/requirements-integration/3d92ad4/ : SUCCESS in 58m 24s\n- neutron-grenade-multinode http://logs.openstack.org/87/653887/1/check/neutron-grenade-multinode/1996d07/ : SUCCESS in 1h 11m 09s\n- tempest-full http://logs.openstack.org/87/653887/1/check/tempest-full/aed0f4d/ : SUCCESS in 2h 03m 57s\n- neutron-grenade http://logs.openstack.org/87/653887/1/check/neutron-grenade/f0264a5/ : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ee23295ad159de1e4d7a23f10e5c8ca97c0a3bc8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 16:18:47.000000000","message":"Patch Set 1:\n\nrecheck (and unvote til gate works again)","accounts_in_message":[],"_revision_number":1},{"id":"9caec881f71a1e4bab1ef14e6c2267f4e79dde2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 22:33:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/87/653887/1/check/openstack-tox-docs/cad2b50/html/ : SUCCESS in 2m 42s\n- openstack-tox-pypy http://logs.openstack.org/87/653887/1/check/openstack-tox-pypy/b415b20/ : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/87/653887/1/check/openstack-tox-pep8/f916b88/ : SUCCESS in 2m 40s\n- openstack-tox-py27 http://logs.openstack.org/87/653887/1/check/openstack-tox-py27/9aec3c8/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/87/653887/1/check/openstack-tox-py36/60f3fe0/ : SUCCESS in 3m 08s\n- openstack-tox-validate http://logs.openstack.org/87/653887/1/check/openstack-tox-validate/5522472/ : SUCCESS in 3m 07s\n- requirements-tox-babel http://logs.openstack.org/87/653887/1/check/requirements-tox-babel/0964479/ : SUCCESS in 2m 52s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/653887/1/check/requirements-tox-py27-check-uc/86cff1e/ : SUCCESS in 7m 36s\n- requirements-tox-py36-check-uc http://logs.openstack.org/87/653887/1/check/requirements-tox-py36-check-uc/f75c713/ : SUCCESS in 7m 47s\n- cross-cinder-py27 http://logs.openstack.org/87/653887/1/check/cross-cinder-py27/39fb658/ : SUCCESS in 8m 46s\n- cross-cinder-py36 http://logs.openstack.org/87/653887/1/check/cross-cinder-py36/e4d7385/ : SUCCESS in 8m 21s\n- cross-glance-py27 http://logs.openstack.org/87/653887/1/check/cross-glance-py27/e6b183e/ : SUCCESS in 4m 42s\n- cross-glance-py36 http://logs.openstack.org/87/653887/1/check/cross-glance-py36/2aae8cc/ : SUCCESS in 5m 48s\n- cross-horizon-py27 http://logs.openstack.org/87/653887/1/check/cross-horizon-py27/5ee0dd4/ : SUCCESS in 10m 59s\n- cross-horizon-py36 http://logs.openstack.org/87/653887/1/check/cross-horizon-py36/79f2f43/ : SUCCESS in 9m 21s\n- cross-horizon-npm http://logs.openstack.org/87/653887/1/check/cross-horizon-npm/e5fc438/ : SUCCESS in 5m 53s\n- cross-keystone-py27 http://logs.openstack.org/87/653887/1/check/cross-keystone-py27/afa1d14/ : SUCCESS in 21m 18s\n- cross-keystone-py36 http://logs.openstack.org/87/653887/1/check/cross-keystone-py36/bb69388/ : SUCCESS in 26m 18s\n- cross-neutron-py27 http://logs.openstack.org/87/653887/1/check/cross-neutron-py27/4405bb1/ : SUCCESS in 20m 37s\n- cross-neutron-py36 http://logs.openstack.org/87/653887/1/check/cross-neutron-py36/cd5f7e8/ : FAILURE in 20m 56s\n- cross-nova-py27 http://logs.openstack.org/87/653887/1/check/cross-nova-py27/6e2e393/ : SUCCESS in 13m 42s\n- cross-nova-py36 http://logs.openstack.org/87/653887/1/check/cross-nova-py36/0ae1c2b/ : SUCCESS in 12m 49s\n- cross-nova-functional http://logs.openstack.org/87/653887/1/check/cross-nova-functional/111c3e4/ : FAILURE in 24m 29s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/87/653887/1/check/cross-osc-build-sphinx-docs/38cb6fe/html/ : SUCCESS in 5m 59s\n- cross-swift-py27 http://logs.openstack.org/87/653887/1/check/cross-swift-py27/8cd1ff4/ : SUCCESS in 9m 28s\n- cross-swift-py36 http://logs.openstack.org/87/653887/1/check/cross-swift-py36/a6ef325/ : SUCCESS in 7m 24s\n- requirements-integration http://logs.openstack.org/87/653887/1/check/requirements-integration/5b3d92d/ : SUCCESS in 58m 55s\n- neutron-grenade-multinode http://logs.openstack.org/87/653887/1/check/neutron-grenade-multinode/cfdc6dc/ : SUCCESS in 1h 12m 06s\n- tempest-full http://logs.openstack.org/87/653887/1/check/tempest-full/52a2dcd/ : TIMED_OUT in 2h 07m 55s\n- neutron-grenade http://logs.openstack.org/87/653887/1/check/neutron-grenade/9589857/ : SUCCESS in 1h 06m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5f7a72fa7ed60da84649eab2c17b790b55f59a73","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 23:31:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6e70a5937d1e2a5b4c0b3e127ee40ff800527c23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 03:45:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/87/653887/1/check/openstack-tox-docs/b0eeaab/html/ : SUCCESS in 4m 30s\n- openstack-tox-pypy http://logs.openstack.org/87/653887/1/check/openstack-tox-pypy/ee85ef3/ : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/87/653887/1/check/openstack-tox-pep8/89f6cb5/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/87/653887/1/check/openstack-tox-py27/099feff/ : SUCCESS in 3m 51s\n- openstack-tox-py36 http://logs.openstack.org/87/653887/1/check/openstack-tox-py36/e507971/ : SUCCESS in 3m 15s\n- openstack-tox-validate http://logs.openstack.org/87/653887/1/check/openstack-tox-validate/8b08b2e/ : SUCCESS in 3m 02s\n- requirements-tox-babel http://logs.openstack.org/87/653887/1/check/requirements-tox-babel/04419fe/ : SUCCESS in 3m 13s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/653887/1/check/requirements-tox-py27-check-uc/86069f6/ : SUCCESS in 7m 28s\n- requirements-tox-py36-check-uc http://logs.openstack.org/87/653887/1/check/requirements-tox-py36-check-uc/ccd2611/ : SUCCESS in 7m 40s\n- cross-cinder-py27 http://logs.openstack.org/87/653887/1/check/cross-cinder-py27/227daed/ : SUCCESS in 10m 58s\n- cross-cinder-py36 http://logs.openstack.org/87/653887/1/check/cross-cinder-py36/a8ebdaa/ : SUCCESS in 8m 39s\n- cross-glance-py27 http://logs.openstack.org/87/653887/1/check/cross-glance-py27/413d783/ : SUCCESS in 4m 18s\n- cross-glance-py36 http://logs.openstack.org/87/653887/1/check/cross-glance-py36/89c2809/ : SUCCESS in 4m 53s\n- cross-horizon-py27 http://logs.openstack.org/87/653887/1/check/cross-horizon-py27/4af5749/ : SUCCESS in 10m 44s\n- cross-horizon-py36 http://logs.openstack.org/87/653887/1/check/cross-horizon-py36/3bdbea5/ : SUCCESS in 9m 54s\n- cross-horizon-npm http://logs.openstack.org/87/653887/1/check/cross-horizon-npm/a7d8538/ : SUCCESS in 6m 29s\n- cross-keystone-py27 http://logs.openstack.org/87/653887/1/check/cross-keystone-py27/004988c/ : SUCCESS in 20m 08s\n- cross-keystone-py36 http://logs.openstack.org/87/653887/1/check/cross-keystone-py36/e2e1dbe/ : SUCCESS in 24m 57s\n- cross-neutron-py27 http://logs.openstack.org/87/653887/1/check/cross-neutron-py27/b5872fc/ : SUCCESS in 21m 41s\n- cross-neutron-py36 http://logs.openstack.org/87/653887/1/check/cross-neutron-py36/c3ae53a/ : SUCCESS in 22m 24s\n- cross-nova-py27 http://logs.openstack.org/87/653887/1/check/cross-nova-py27/1953038/ : SUCCESS in 12m 51s\n- cross-nova-py36 http://logs.openstack.org/87/653887/1/check/cross-nova-py36/25fcf3b/ : SUCCESS in 13m 04s\n- cross-nova-functional http://logs.openstack.org/87/653887/1/check/cross-nova-functional/8fe439a/ : SUCCESS in 21m 43s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/87/653887/1/check/cross-osc-build-sphinx-docs/4f6d209/html/ : SUCCESS in 5m 22s\n- cross-swift-py27 http://logs.openstack.org/87/653887/1/check/cross-swift-py27/610f944/ : SUCCESS in 10m 08s\n- cross-swift-py36 http://logs.openstack.org/87/653887/1/check/cross-swift-py36/8a47473/ : SUCCESS in 7m 21s\n- requirements-integration http://logs.openstack.org/87/653887/1/check/requirements-integration/44a3d1e/ : SUCCESS in 56m 41s\n- neutron-grenade-multinode http://logs.openstack.org/87/653887/1/check/neutron-grenade-multinode/e10e9c8/ : SUCCESS in 1h 13m 20s\n- tempest-full http://logs.openstack.org/87/653887/1/check/tempest-full/9312326/ : SUCCESS in 1h 39m 39s\n- neutron-grenade http://logs.openstack.org/87/653887/1/check/neutron-grenade/aeeff05/ : SUCCESS in 1h 06m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e3dfa64564697b2f934e2761deb9aee49a9c2916","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-23 12:10:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f4ed8db5cbfe00e9569b574b7bf76fd08c5e5c5e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-23 14:43:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"803440ce480d98b6f370a0eeba36b323c328f3fc","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-25 14:45:46.000000000","message":"Patch Set 1:\n\nAny chance we can land this so we start to use the new release of neutron-lib for dev?","accounts_in_message":[],"_revision_number":1},{"id":"9cedf2cf460602e6fd7ad9834b2e86f1424706fd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-25 15:03:44.000000000","message":"Patch Set 1: Workflow+1\n\nya, master can be single approved","accounts_in_message":[],"_revision_number":1},{"id":"5a798af7e4411fab71ef006807161f155d03e2b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 15:04:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07f1aaf13e8e7e0e492610766834364c3a864e57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 22:02:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/87/653887/1/gate/openstack-tox-docs/4d08335/html/ : SUCCESS in 2m 37s\n- openstack-tox-pep8 http://logs.openstack.org/87/653887/1/gate/openstack-tox-pep8/8cfaf67/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/87/653887/1/gate/openstack-tox-py27/997f1c5/ : SUCCESS in 2m 49s\n- openstack-tox-py36 http://logs.openstack.org/87/653887/1/gate/openstack-tox-py36/57e8c03/ : SUCCESS in 3m 11s\n- openstack-tox-validate http://logs.openstack.org/87/653887/1/gate/openstack-tox-validate/f6572c7/ : SUCCESS in 4m 29s\n- requirements-tox-babel http://logs.openstack.org/87/653887/1/gate/requirements-tox-babel/64aa6f5/ : SUCCESS in 2m 51s\n- requirements-tox-py27-check-uc http://logs.openstack.org/87/653887/1/gate/requirements-tox-py27-check-uc/59a72bc/ : SUCCESS in 7m 51s\n- requirements-tox-py36-check-uc http://logs.openstack.org/87/653887/1/gate/requirements-tox-py36-check-uc/dda54d3/ : SUCCESS in 7m 10s\n- cross-cinder-py27 http://logs.openstack.org/87/653887/1/gate/cross-cinder-py27/a487e4a/ : SUCCESS in 9m 22s\n- cross-cinder-py36 http://logs.openstack.org/87/653887/1/gate/cross-cinder-py36/be27fca/ : SUCCESS in 8m 53s\n- cross-glance-py27 http://logs.openstack.org/87/653887/1/gate/cross-glance-py27/4f3455c/ : SUCCESS in 4m 45s\n- cross-glance-py36 http://logs.openstack.org/87/653887/1/gate/cross-glance-py36/2828acb/ : SUCCESS in 5m 26s\n- cross-horizon-py27 http://logs.openstack.org/87/653887/1/gate/cross-horizon-py27/ed2ed5a/ : SUCCESS in 8m 56s\n- cross-horizon-py36 http://logs.openstack.org/87/653887/1/gate/cross-horizon-py36/77d27f3/ : SUCCESS in 9m 44s\n- cross-horizon-npm http://logs.openstack.org/87/653887/1/gate/cross-horizon-npm/629dca7/ : SUCCESS in 6m 13s\n- cross-keystone-py27 http://logs.openstack.org/87/653887/1/gate/cross-keystone-py27/19e80b9/ : SUCCESS in 18m 47s\n- cross-keystone-py36 http://logs.openstack.org/87/653887/1/gate/cross-keystone-py36/1e6a1c3/ : SUCCESS in 27m 15s\n- cross-neutron-py27 http://logs.openstack.org/87/653887/1/gate/cross-neutron-py27/ca3ddb3/ : SUCCESS in 21m 21s\n- cross-neutron-py36 http://logs.openstack.org/87/653887/1/gate/cross-neutron-py36/650bef8/ : SUCCESS in 28m 02s\n- cross-nova-py27 http://logs.openstack.org/87/653887/1/gate/cross-nova-py27/2d66cfc/ : SUCCESS in 13m 52s\n- cross-nova-py36 http://logs.openstack.org/87/653887/1/gate/cross-nova-py36/afef42f/ : SUCCESS in 12m 23s\n- cross-nova-functional http://logs.openstack.org/87/653887/1/gate/cross-nova-functional/04595e0/ : SUCCESS in 19m 48s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/87/653887/1/gate/cross-osc-build-sphinx-docs/3689533/html/ : SUCCESS in 5m 07s\n- cross-swift-py27 http://logs.openstack.org/87/653887/1/gate/cross-swift-py27/3d0f50d/ : SUCCESS in 8m 13s\n- cross-swift-py36 http://logs.openstack.org/87/653887/1/gate/cross-swift-py36/eac487e/ : SUCCESS in 7m 53s\n- requirements-integration http://logs.openstack.org/87/653887/1/gate/requirements-integration/d8541d8/ : SUCCESS in 56m 24s\n- neutron-grenade-multinode http://logs.openstack.org/87/653887/1/gate/neutron-grenade-multinode/72e289b/ : SUCCESS in 1h 10m 02s\n- tempest-full http://logs.openstack.org/87/653887/1/gate/tempest-full/8ef2d5c/ : SUCCESS in 1h 43m 44s\n- neutron-grenade http://logs.openstack.org/87/653887/1/gate/neutron-grenade/de55073/ : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6db8f0557267e9bbf9ad9a6195380016d4062bb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 22:02:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7242fabd75df4cff9cf54884ccb1f777123ccf3","revisions":{"f7242fabd75df4cff9cf54884ccb1f777123ccf3":{"kind":"REWORK","_number":1,"created":"2019-04-19 00:28:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/653887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/653887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/653887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/653887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/653887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/653887/1"}}},"commit":{"parents":[{"commit":"901af930f44254f7fd3fd3c0ab34daf98cebf362","subject":"Merge \"Add protobuf to global-requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/901af930f44254f7fd3fd3c0ab34daf98cebf362"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-19 00:28:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-19 00:28:30.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 1.26.0","message":"update constraint for neutron-lib to new release 1.26.0\n\nChange-Id: Ibf884dbc003e120378089e05e315f9938a705484\nmeta:version: 1.26.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Boden R \u003cbodenvmw@gmail.com\u003e\nmeta:release:Commit: Boden R \u003cbodenvmw@gmail.com\u003e\nmeta:release:Change-Id: Id55d7b5c5d099782d8b90d07922496a27b389ddd\nmeta:release:Code-Review+1: Miguel Lavalle \u003cmiguel.lavalle@verizonmedia.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7242fabd75df4cff9cf54884ccb1f777123ccf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7242fabd75df4cff9cf54884ccb1f777123ccf3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
