)]}'
{"id":"openstack%2Frequirements~654487","triplet_id":"openstack%2Frequirements~master~I8ecadff2737a40d5dece4c7511ebe54a5efddf64","project":"openstack/requirements","branch":"master","topic":"update-dft-reqs-dir","hashtags":[],"change_id":"I8ecadff2737a40d5dece4c7511ebe54a5efddf64","subject":"update git.openstack in _DEFAULT_REQS_DIR","status":"MERGED","created":"2019-04-22 15:32:54.000000000","updated":"2019-04-23 12:15:01.000000000","submitted":"2019-04-22 23:54:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654487-1555977267857-7a5aba1f","meta_rev_id":"2cb4d28ad3d7734350c999d5c92cb800378e87fc","_number":654487,"virtual_id_number":654487,"owner":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-04-22 23:54:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-22 15:36:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-22 15:34:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2019-04-22 15:35:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-04-22 15:36:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 15:34:25.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-04-22 15:35:47.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-04-22 15:36:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-22 23:54:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0333b64dab34016ed021bcbdb4a745750cd33c8","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-22 15:32:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c64978aa8a6e3e57d4ab6f550d788bd43e5bfcd3","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-04-22 15:34:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5fe6706393b3ce7a897cdacb8984e46f71279a19","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-22 15:35:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b74dd0183ee31f8857f519e014c59d8715f76e23","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-22 15:36:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"106507c6b0f37f6b53ec710871e0151c4023d757","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-22 16:50:44.000000000","message":"Patch Set 1:\n\nthis change needs backporting to older brnaches - to wherever this file exists. Boden, could you do so, please?","accounts_in_message":[],"_revision_number":1},{"id":"aeac46de5ca8000d4dcaf566049232afae595a2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:59:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/87/654487/1/check/openstack-tox-docs/d7c326f/html/ : SUCCESS in 3m 14s\n- openstack-tox-pypy http://logs.openstack.org/87/654487/1/check/openstack-tox-pypy/5e2d9a3/ : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/87/654487/1/check/openstack-tox-pep8/9ba0eb4/ : SUCCESS in 2m 46s\n- openstack-tox-py27 http://logs.openstack.org/87/654487/1/check/openstack-tox-py27/f006c41/ : SUCCESS in 3m 55s\n- openstack-tox-py36 http://logs.openstack.org/87/654487/1/check/openstack-tox-py36/a673da4/ : SUCCESS in 3m 49s\n- openstack-tox-validate http://logs.openstack.org/87/654487/1/check/openstack-tox-validate/552ad1f/ : SUCCESS in 3m 24s\n- requirements-integration http://logs.openstack.org/87/654487/1/check/requirements-integration/d944b46/ : SUCCESS in 57m 46s\n- neutron-grenade-multinode http://logs.openstack.org/87/654487/1/check/neutron-grenade-multinode/ad02897/ : SUCCESS in 1h 06m 38s\n- tempest-full http://logs.openstack.org/87/654487/1/check/tempest-full/848f653/ : SUCCESS in 1h 44m 44s\n- neutron-grenade http://logs.openstack.org/87/654487/1/check/neutron-grenade/e602b9b/ : SUCCESS in 56m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0872ee3ba29dd51cea10323be891350b1a3d2a8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 21:59:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c360089500d396a716e48506e5178a20b3afdec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 23:54:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6b738062881a955f9e91edf6d65505cd559f4d4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 23:54:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/87/654487/1/gate/openstack-tox-docs/b264c76/html/ : SUCCESS in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/87/654487/1/gate/openstack-tox-pep8/f40007f/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/87/654487/1/gate/openstack-tox-py27/1514f0b/ : SUCCESS in 3m 04s\n- openstack-tox-py36 http://logs.openstack.org/87/654487/1/gate/openstack-tox-py36/899b035/ : SUCCESS in 3m 02s\n- openstack-tox-validate http://logs.openstack.org/87/654487/1/gate/openstack-tox-validate/add580a/ : SUCCESS in 3m 03s\n- requirements-integration http://logs.openstack.org/87/654487/1/gate/requirements-integration/7c1465a/ : SUCCESS in 57m 07s\n- neutron-grenade-multinode http://logs.openstack.org/87/654487/1/gate/neutron-grenade-multinode/11b910f/ : SUCCESS in 1h 13m 40s\n- tempest-full http://logs.openstack.org/87/654487/1/gate/tempest-full/9ab453a/ : SUCCESS in 1h 44m 54s\n- neutron-grenade http://logs.openstack.org/87/654487/1/gate/neutron-grenade/992141b/ : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3e9f264b20b086140e8558184386aaf453f87be7","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-23 12:15:01.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit b0b11d00a188caf51e7dc1c45f24703082aa3fa3","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"674e97a63127b5caa4b9e5f13757603892503d8a","revisions":{"674e97a63127b5caa4b9e5f13757603892503d8a":{"kind":"REWORK","_number":1,"created":"2019-04-22 15:32:54.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/87/654487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/654487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/654487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/654487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/654487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/654487/1"}}},"commit":{"parents":[{"commit":"fb8f620fe3eb4ff3c843c5b4b30ffb29a7004259","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb8f620fe3eb4ff3c843c5b4b30ffb29a7004259"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-04-22 15:32:41.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-04-22 15:32:53.000000000","tz":-360},"subject":"update git.openstack in _DEFAULT_REQS_DIR","message":"update git.openstack in _DEFAULT_REQS_DIR\n\nThis patch changes the _DEFAULT_REQS_DIR constant to use opendev.org\nrather than git.openstack.org in project-requirements-change.py to\naddress failures such as [1].\n\n[1] http://logs.openstack.org/09/653009/5/check/requirements-check/\ncb0e2ad/job-output.txt.gz#_2019-04-22_15_02_24_958181\n\nChange-Id: I8ecadff2737a40d5dece4c7511ebe54a5efddf64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/674e97a63127b5caa4b9e5f13757603892503d8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/674e97a63127b5caa4b9e5f13757603892503d8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
