)]}'
{"id":"openstack%2Frequirements~655498","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I97b909385a8e6778a1a09afb26b8c33be1f294ca","project":"openstack/requirements","branch":"stable/queens","topic":"bump-minimum-version-pylxd","hashtags":[],"change_id":"I97b909385a8e6778a1a09afb26b8c33be1f294ca","subject":"Bump pylxd to 2.2.7","status":"MERGED","created":"2019-04-24 16:54:14.000000000","updated":"2019-04-26 12:02:31.000000000","submitted":"2019-04-26 12:02:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655498-1556280151069-e0b66468","meta_rev_id":"8bd3a4d2fba83ec1390e42a6b600955d580d07d8","_number":655498,"virtual_id_number":655498,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"date":"2019-04-24 19:17:22.000000000","_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},{"value":0,"date":"2019-04-26 07:07:18.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-04-26 12:02:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-04-25 23:06:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-25 23:01:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-04-24 16:55:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":1,"date":"2019-04-25 14:05:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":1,"date":"2019-04-24 17:19:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-24 21:23:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"date":"2019-04-24 16:56:32.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-04-25 23:06:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 16:55:51.000000000","updated_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"reviewer":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"state":"REVIEWER"},{"updated":"2019-04-24 16:56:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-24 17:19:54.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2019-04-24 19:17:22.000000000","updated_by":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"reviewer":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"state":"REVIEWER"},{"updated":"2019-04-24 21:23:17.000000000","updated_by":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"reviewer":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"state":"REVIEWER"},{"updated":"2019-04-25 14:05:08.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2019-04-25 23:01:03.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-04-25 23:06:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-26 12:02:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40c432619558823a7302751822b463b442b21afe","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-24 16:54:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"270bf25bffc7325d570bd2e860f726cbf3509444","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2019-04-24 16:55:51.000000000","message":"Patch Set 1: Code-Review+1\n\nThis will be fine with nova-lxd in queens, so +1 from me.","accounts_in_message":[],"_revision_number":1},{"id":"a80a72f27e9d1287f4ed712cfcb4c1da4fac1e01","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-04-24 17:19:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ac37d2aa14cd53d8657b4697cf3e69bad55c4307","author":{"_account_id":9963,"name":"Jiri Suchomel","email":"jiri.suchomel@suse.com","username":"jsuchome"},"date":"2019-04-24 19:17:22.000000000","message":"Patch Set 1:\n\nCould you explain what has changed, i.e. why can§t we build it \"anymore\" ?","accounts_in_message":[],"_revision_number":1},{"id":"cd3510da476aa9827fe16c818195110620177b3b","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-04-24 21:23:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1ed82121fca50e16ed91f6da1283845bbbeeb610","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 03:47:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/98/655498/1/check/openstack-tox-pypy/6ea8695/ : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/655498/1/check/openstack-tox-pep8/4fad1a9/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/98/655498/1/check/openstack-tox-py27/0ffa5e4/ : SUCCESS in 4m 55s\n- openstack-tox-py35 http://logs.openstack.org/98/655498/1/check/openstack-tox-py35/4ec43e2/ : SUCCESS in 3m 38s\n- openstack-tox-validate http://logs.openstack.org/98/655498/1/check/openstack-tox-validate/6cd7bc2/ : SUCCESS in 3m 02s\n- requirements-tox-babel http://logs.openstack.org/98/655498/1/check/requirements-tox-babel/2202763/ : SUCCESS in 2m 48s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/655498/1/check/requirements-tox-py27-check-uc/31dd6db/ : SUCCESS in 5m 33s\n- requirements-tox-py35-check-uc http://logs.openstack.org/98/655498/1/check/requirements-tox-py35-check-uc/c644374/ : SUCCESS in 6m 08s\n- cross-cinder-py27 http://logs.openstack.org/98/655498/1/check/cross-cinder-py27/ed73a9e/ : SUCCESS in 8m 27s\n- cross-cinder-py35 http://logs.openstack.org/98/655498/1/check/cross-cinder-py35/fd51fb1/ : FAILURE in 9m 25s\n- cross-glance-py27 http://logs.openstack.org/98/655498/1/check/cross-glance-py27/d2edc81/ : SUCCESS in 4m 30s\n- cross-glance-py35 http://logs.openstack.org/98/655498/1/check/cross-glance-py35/735c123/ : SUCCESS in 4m 09s\n- cross-horizon-py27 http://logs.openstack.org/98/655498/1/check/cross-horizon-py27/6aa8c29/ : SUCCESS in 11m 31s\n- cross-horizon-py35 http://logs.openstack.org/98/655498/1/check/cross-horizon-py35/8cd731e/ : SUCCESS in 9m 54s\n- cross-keystone-py27 http://logs.openstack.org/98/655498/1/check/cross-keystone-py27/a9242d6/ : SUCCESS in 12m 42s\n- cross-keystone-py35 http://logs.openstack.org/98/655498/1/check/cross-keystone-py35/8693504/ : SUCCESS in 17m 31s\n- cross-neutron-py27 http://logs.openstack.org/98/655498/1/check/cross-neutron-py27/0993b6b/ : SUCCESS in 22m 48s\n- cross-neutron-py35 http://logs.openstack.org/98/655498/1/check/cross-neutron-py35/87720f0/ : SUCCESS in 26m 27s\n- cross-nova-py27 http://logs.openstack.org/98/655498/1/check/cross-nova-py27/be8df86/ : SUCCESS in 15m 13s\n- cross-nova-py35 http://logs.openstack.org/98/655498/1/check/cross-nova-py35/e86ad12/ : SUCCESS in 12m 15s\n- cross-nova-functional http://logs.openstack.org/98/655498/1/check/cross-nova-functional/2183669/ : SUCCESS in 16m 09s\n- cross-swift-py27 http://logs.openstack.org/98/655498/1/check/cross-swift-py27/7d699ff/ : SUCCESS in 5m 31s\n- cross-swift-py35 http://logs.openstack.org/98/655498/1/check/cross-swift-py35/bb1b64c/ : SUCCESS in 3m 10s\n- requirements-integration http://logs.openstack.org/98/655498/1/check/requirements-integration/8b97d04/ : SUCCESS in 43m 51s\n- neutron-grenade-multinode http://logs.openstack.org/98/655498/1/check/neutron-grenade-multinode/10d6655/ : SUCCESS in 1h 01m 43s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/98/655498/1/check/legacy-tempest-dsvm-neutron-full/3cb34f4/ : SUCCESS in 1h 13m 58s\n- neutron-grenade http://logs.openstack.org/98/655498/1/check/neutron-grenade/ad0aeb8/ : SUCCESS in 45m 40s","accounts_in_message":[],"_revision_number":1},{"id":"87161f0b168d4c17826f8810ecbc086938f7889b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-25 06:29:57.000000000","message":"Patch Set 1:\n\n\u003e Could you explain what has changed, i.e. why can§t we build it\n \u003e \"anymore\" ?\n\nSee commit message. Basically install_requires is uncapped on requests.* , and creating a wheel with latest requests.* packages will have, at the end, conflicting requirements.","accounts_in_message":[],"_revision_number":1},{"id":"a6ec37f39e45da407fb738962133f668661a07ef","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-25 06:30:59.000000000","message":"Patch Set 1:\n\nSee also https://github.com/lxc/pylxd/issues/308","accounts_in_message":[],"_revision_number":1},{"id":"2b81503ac989f3da70287cf857d8b367ee4025ab","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-25 06:31:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"67b2dcf2272e5b103d174aafc68937c84f90c179","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-25 07:08:12.000000000","message":"Patch Set 1:\n\n\u003e Could you explain what has changed, i.e. why can§t we build it\n \u003e \"anymore\" ?\n\nFYI This has started to arise recently when a new version of urllib3 was released https://pypi.org/project/urllib3/#history.","accounts_in_message":[],"_revision_number":1},{"id":"2e50daae974a9fe14605260b9f42739a9e7b8d36","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-25 07:08:44.000000000","message":"Patch Set 1:\n\n\u003e \u003e Could you explain what has changed, i.e. why can§t we build it\n \u003e \u003e \"anymore\" ?\n \u003e \n \u003e FYI This has started to arise recently when a new version of\n \u003e urllib3 was released https://pypi.org/project/urllib3/#history.\n\n... which is pulled as dependency.","accounts_in_message":[],"_revision_number":1},{"id":"843a58484290e0a6a194840f9af3017b27bad6ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 12:22:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/98/655498/1/check/openstack-tox-pypy/69b9188/ : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/655498/1/check/openstack-tox-pep8/9ff59d8/ : SUCCESS in 2m 38s\n- openstack-tox-py27 http://logs.openstack.org/98/655498/1/check/openstack-tox-py27/db08e9a/ : SUCCESS in 3m 09s\n- openstack-tox-py35 http://logs.openstack.org/98/655498/1/check/openstack-tox-py35/81a8113/ : SUCCESS in 5m 51s\n- openstack-tox-validate http://logs.openstack.org/98/655498/1/check/openstack-tox-validate/f9f8bcc/ : SUCCESS in 3m 53s\n- requirements-tox-babel http://logs.openstack.org/98/655498/1/check/requirements-tox-babel/45cbac7/ : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/655498/1/check/requirements-tox-py27-check-uc/d40bec6/ : SUCCESS in 5m 53s\n- requirements-tox-py35-check-uc http://logs.openstack.org/98/655498/1/check/requirements-tox-py35-check-uc/ac17c43/ : SUCCESS in 5m 59s\n- cross-cinder-py27 http://logs.openstack.org/98/655498/1/check/cross-cinder-py27/be20b2f/ : SUCCESS in 7m 43s\n- cross-cinder-py35 http://logs.openstack.org/98/655498/1/check/cross-cinder-py35/73d9c77/ : SUCCESS in 8m 29s\n- cross-glance-py27 http://logs.openstack.org/98/655498/1/check/cross-glance-py27/c2295bf/ : SUCCESS in 4m 07s\n- cross-glance-py35 http://logs.openstack.org/98/655498/1/check/cross-glance-py35/184499c/ : SUCCESS in 4m 01s\n- cross-horizon-py27 http://logs.openstack.org/98/655498/1/check/cross-horizon-py27/72bfeec/ : SUCCESS in 10m 56s\n- cross-horizon-py35 http://logs.openstack.org/98/655498/1/check/cross-horizon-py35/b112546/ : SUCCESS in 10m 52s\n- cross-keystone-py27 http://logs.openstack.org/98/655498/1/check/cross-keystone-py27/1ca39d1/ : SUCCESS in 29m 30s\n- cross-keystone-py35 http://logs.openstack.org/98/655498/1/check/cross-keystone-py35/8b17ab3/ : SUCCESS in 17m 29s\n- cross-neutron-py27 http://logs.openstack.org/98/655498/1/check/cross-neutron-py27/d7cc162/ : SUCCESS in 23m 16s\n- cross-neutron-py35 http://logs.openstack.org/98/655498/1/check/cross-neutron-py35/978bc19/ : SUCCESS in 23m 17s\n- cross-nova-py27 http://logs.openstack.org/98/655498/1/check/cross-nova-py27/8894566/ : SUCCESS in 12m 56s\n- cross-nova-py35 http://logs.openstack.org/98/655498/1/check/cross-nova-py35/cad321e/ : SUCCESS in 13m 48s\n- cross-nova-functional http://logs.openstack.org/98/655498/1/check/cross-nova-functional/12d12d8/ : SUCCESS in 18m 58s\n- cross-swift-py27 http://logs.openstack.org/98/655498/1/check/cross-swift-py27/581ae35/ : SUCCESS in 5m 33s\n- cross-swift-py35 http://logs.openstack.org/98/655498/1/check/cross-swift-py35/28ceb40/ : SUCCESS in 3m 03s\n- requirements-integration http://logs.openstack.org/98/655498/1/check/requirements-integration/0f52c12/ : SUCCESS in 49m 01s\n- neutron-grenade-multinode http://logs.openstack.org/98/655498/1/check/neutron-grenade-multinode/0d2a3a1/ : SUCCESS in 1h 02m 28s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/98/655498/1/check/legacy-tempest-dsvm-neutron-full/3860868/ : SUCCESS in 1h 24m 01s\n- neutron-grenade http://logs.openstack.org/98/655498/1/check/neutron-grenade/f470570/ : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":1},{"id":"02ed46f491ea371ab1aa2e15d44ddaa45ed37ef4","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2019-04-25 14:05:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a1ccf9c5683e89717d482420640a9cdbf129ff92","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-25 14:44:01.000000000","message":"Patch Set 1:\n\ncan 2.2.5 not be built with the queens constraint?","accounts_in_message":[],"_revision_number":1},{"id":"fa3dd87637eaf7ba8e8d629b5edcc03b340b7260","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-04-25 23:01:03.000000000","message":"Patch Set 1: Code-Review+2\n\n\u003e can 2.2.5 not be built with the queens constraint?\n\nIt can but due to what I think is a pip bug it has to be done in 2 steps\n\n virtualenv -p python2 LXD\n LXD/bin/pip install -c https://releases.openstack.org/constraints/upper/queens pylxd\n\n#fail\n\n LXD/bin/pip install -c https://releases.openstack.org/constraints/upper/queens requests\n LXD/bin/pip install -c https://releases.openstack.org/constraints/upper/queens pylxd\n\n# success \n\nI don\u0027t have ready access to see who uses pylxd, and IIRC hound currently isn\u0027t working\n\nI think in general I;m okay with this and it\u0027s supported by the stable policy.","accounts_in_message":[],"_revision_number":1},{"id":"ff9540892a83c95de33b1c2b44e78b4dee4a03b8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-25 23:06:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nonly nova-lxd it seems\n\n| nova-lxd     | requirements.txt                                         |   13 | pylxd\u003e\u003d2.2.6 # Apache-2.0","accounts_in_message":[],"_revision_number":1},{"id":"21aaa3442d24dc50179eced245e625c38da9bf33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 23:07:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e20030fa1fc33c8146a2799cac34ecf301ed06b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 05:13:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/98/655498/1/gate/openstack-tox-pep8/cba3740/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/98/655498/1/gate/openstack-tox-py27/ac98b5d/ : SUCCESS in 4m 23s\n- openstack-tox-py35 http://logs.openstack.org/98/655498/1/gate/openstack-tox-py35/64150c8/ : SUCCESS in 3m 35s\n- openstack-tox-validate http://logs.openstack.org/98/655498/1/gate/openstack-tox-validate/a0e773f/ : SUCCESS in 4m 07s\n- requirements-tox-babel http://logs.openstack.org/98/655498/1/gate/requirements-tox-babel/162f7cc/ : SUCCESS in 3m 16s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/655498/1/gate/requirements-tox-py27-check-uc/7c3bb44/ : SUCCESS in 6m 25s\n- requirements-tox-py35-check-uc http://logs.openstack.org/98/655498/1/gate/requirements-tox-py35-check-uc/44e1de8/ : SUCCESS in 6m 28s\n- cross-cinder-py27 http://logs.openstack.org/98/655498/1/gate/cross-cinder-py27/af5fff6/ : SUCCESS in 9m 18s\n- cross-cinder-py35 http://logs.openstack.org/98/655498/1/gate/cross-cinder-py35/f0c05f2/ : SUCCESS in 7m 41s\n- cross-glance-py27 http://logs.openstack.org/98/655498/1/gate/cross-glance-py27/8c3b890/ : SUCCESS in 6m 25s\n- cross-glance-py35 http://logs.openstack.org/98/655498/1/gate/cross-glance-py35/30a2a2f/ : SUCCESS in 4m 56s\n- cross-horizon-py27 http://logs.openstack.org/98/655498/1/gate/cross-horizon-py27/6eef1d2/ : SUCCESS in 12m 10s\n- cross-horizon-py35 http://logs.openstack.org/98/655498/1/gate/cross-horizon-py35/74b9a29/ : SUCCESS in 14m 13s\n- cross-keystone-py27 http://logs.openstack.org/98/655498/1/gate/cross-keystone-py27/44688a6/ : SUCCESS in 12m 33s\n- cross-keystone-py35 http://logs.openstack.org/98/655498/1/gate/cross-keystone-py35/6acb68f/ : TIMED_OUT in 40m 54s\n- cross-neutron-py27 http://logs.openstack.org/98/655498/1/gate/cross-neutron-py27/9709a78/ : SUCCESS in 22m 30s\n- cross-neutron-py35 http://logs.openstack.org/98/655498/1/gate/cross-neutron-py35/4fdcb46/ : SUCCESS in 23m 59s\n- cross-nova-py27 http://logs.openstack.org/98/655498/1/gate/cross-nova-py27/bbff2fa/ : SUCCESS in 12m 27s\n- cross-nova-py35 http://logs.openstack.org/98/655498/1/gate/cross-nova-py35/2381387/ : SUCCESS in 14m 03s\n- cross-nova-functional http://logs.openstack.org/98/655498/1/gate/cross-nova-functional/3c0085d/ : SUCCESS in 18m 56s\n- cross-swift-py27 http://logs.openstack.org/98/655498/1/gate/cross-swift-py27/509da84/ : SUCCESS in 5m 34s\n- cross-swift-py35 http://logs.openstack.org/98/655498/1/gate/cross-swift-py35/8593170/ : SUCCESS in 3m 50s\n- requirements-integration http://logs.openstack.org/98/655498/1/gate/requirements-integration/6a88a41/ : SUCCESS in 53m 46s\n- neutron-grenade-multinode http://logs.openstack.org/98/655498/1/gate/neutron-grenade-multinode/28beedf/ : SUCCESS in 1h 02m 29s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/98/655498/1/gate/legacy-tempest-dsvm-neutron-full/47a299f/ : SUCCESS in 1h 16m 10s\n- neutron-grenade http://logs.openstack.org/98/655498/1/gate/neutron-grenade/0db5760/ : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":1},{"id":"716b4383505d98b24bff393922b8b5d2339f1a1b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-26 05:16:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"adada3569e7353197c5ce4ddf90dfeadc52dceb5","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-26 07:06:11.000000000","message":"Patch Set 1:\n\n\u003e can 2.2.5 not be built with the queens constraint?\n\nIt can be pip install-ed, but not pip wheel-ed","accounts_in_message":[],"_revision_number":1},{"id":"e04c672ec8315c10058d2df3e55fbd54a73754cd","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-26 07:07:00.000000000","message":"Patch Set 1:\n\n\u003e only nova-lxd it seems\n \u003e \n \u003e | nova-lxd     | requirements.txt                                  \n \u003e       |   13 | pylxd\u003e\u003d2.2.6 # Apache-2.0\n\nCorrect, and tinwood already evaluated this.","accounts_in_message":[],"_revision_number":1},{"id":"aa8ddf0df8f715e45b39e599be8dff50e6a66d7e","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-26 07:07:18.000000000","message":"Patch Set 1:\n\nrecheck -- timeout","accounts_in_message":[],"_revision_number":1},{"id":"d65902fe2027ae94f7943f025dcc99c4817345bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 10:27:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/98/655498/1/check/openstack-tox-pypy/675d206/ : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/655498/1/check/openstack-tox-pep8/1914b11/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/98/655498/1/check/openstack-tox-py27/6160544/ : SUCCESS in 3m 09s\n- openstack-tox-py35 http://logs.openstack.org/98/655498/1/check/openstack-tox-py35/85534b7/ : SUCCESS in 3m 45s\n- openstack-tox-validate http://logs.openstack.org/98/655498/1/check/openstack-tox-validate/d6d67a3/ : SUCCESS in 3m 28s\n- requirements-tox-babel http://logs.openstack.org/98/655498/1/check/requirements-tox-babel/7c2449c/ : SUCCESS in 2m 36s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/655498/1/check/requirements-tox-py27-check-uc/394d85a/ : SUCCESS in 6m 26s\n- requirements-tox-py35-check-uc http://logs.openstack.org/98/655498/1/check/requirements-tox-py35-check-uc/1a86944/ : SUCCESS in 6m 50s\n- cross-cinder-py27 http://logs.openstack.org/98/655498/1/check/cross-cinder-py27/34d4398/ : SUCCESS in 7m 55s\n- cross-cinder-py35 http://logs.openstack.org/98/655498/1/check/cross-cinder-py35/79f76a9/ : SUCCESS in 7m 47s\n- cross-glance-py27 http://logs.openstack.org/98/655498/1/check/cross-glance-py27/56d9e37/ : SUCCESS in 4m 19s\n- cross-glance-py35 http://logs.openstack.org/98/655498/1/check/cross-glance-py35/fe9e806/ : SUCCESS in 5m 36s\n- cross-horizon-py27 http://logs.openstack.org/98/655498/1/check/cross-horizon-py27/3258263/ : SUCCESS in 11m 53s\n- cross-horizon-py35 http://logs.openstack.org/98/655498/1/check/cross-horizon-py35/e39ca44/ : SUCCESS in 11m 51s\n- cross-keystone-py27 http://logs.openstack.org/98/655498/1/check/cross-keystone-py27/7093a54/ : SUCCESS in 11m 06s\n- cross-keystone-py35 http://logs.openstack.org/98/655498/1/check/cross-keystone-py35/76bca4d/ : SUCCESS in 11m 41s\n- cross-neutron-py27 http://logs.openstack.org/98/655498/1/check/cross-neutron-py27/ac6ddff/ : SUCCESS in 22m 29s\n- cross-neutron-py35 http://logs.openstack.org/98/655498/1/check/cross-neutron-py35/677650b/ : SUCCESS in 22m 48s\n- cross-nova-py27 http://logs.openstack.org/98/655498/1/check/cross-nova-py27/006f457/ : SUCCESS in 14m 43s\n- cross-nova-py35 http://logs.openstack.org/98/655498/1/check/cross-nova-py35/9fc2741/ : SUCCESS in 13m 58s\n- cross-nova-functional http://logs.openstack.org/98/655498/1/check/cross-nova-functional/3a92307/ : SUCCESS in 21m 57s\n- cross-swift-py27 http://logs.openstack.org/98/655498/1/check/cross-swift-py27/0330af4/ : SUCCESS in 5m 39s\n- cross-swift-py35 http://logs.openstack.org/98/655498/1/check/cross-swift-py35/6f6b71e/ : SUCCESS in 3m 15s\n- requirements-integration http://logs.openstack.org/98/655498/1/check/requirements-integration/28ee619/ : SUCCESS in 49m 37s\n- neutron-grenade-multinode http://logs.openstack.org/98/655498/1/check/neutron-grenade-multinode/ed562b7/ : SUCCESS in 1h 02m 11s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/98/655498/1/check/legacy-tempest-dsvm-neutron-full/cf001bb/ : SUCCESS in 1h 26m 03s\n- neutron-grenade http://logs.openstack.org/98/655498/1/check/neutron-grenade/1c8fddc/ : SUCCESS in 1h 20m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3192a70b804d3e6df4e18977f2ca0cdd12ec5b71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 10:27:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c04636a78cdbf9c92a001a1a42c715be80d7db9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 12:02:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/98/655498/1/gate/openstack-tox-pep8/ddd8125/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/98/655498/1/gate/openstack-tox-py27/d04a7d9/ : SUCCESS in 3m 07s\n- openstack-tox-py35 http://logs.openstack.org/98/655498/1/gate/openstack-tox-py35/18f0282/ : SUCCESS in 3m 44s\n- openstack-tox-validate http://logs.openstack.org/98/655498/1/gate/openstack-tox-validate/96d7e16/ : SUCCESS in 3m 55s\n- requirements-tox-babel http://logs.openstack.org/98/655498/1/gate/requirements-tox-babel/b67fbf9/ : SUCCESS in 2m 50s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/655498/1/gate/requirements-tox-py27-check-uc/1d2526a/ : SUCCESS in 6m 16s\n- requirements-tox-py35-check-uc http://logs.openstack.org/98/655498/1/gate/requirements-tox-py35-check-uc/7ab9e0b/ : SUCCESS in 8m 35s\n- cross-cinder-py27 http://logs.openstack.org/98/655498/1/gate/cross-cinder-py27/14358f7/ : SUCCESS in 7m 34s\n- cross-cinder-py35 http://logs.openstack.org/98/655498/1/gate/cross-cinder-py35/af198fc/ : SUCCESS in 8m 23s\n- cross-glance-py27 http://logs.openstack.org/98/655498/1/gate/cross-glance-py27/cbfc39d/ : SUCCESS in 4m 54s\n- cross-glance-py35 http://logs.openstack.org/98/655498/1/gate/cross-glance-py35/6718680/ : SUCCESS in 4m 14s\n- cross-horizon-py27 http://logs.openstack.org/98/655498/1/gate/cross-horizon-py27/9ecb3c1/ : SUCCESS in 12m 05s\n- cross-horizon-py35 http://logs.openstack.org/98/655498/1/gate/cross-horizon-py35/8767cd8/ : SUCCESS in 10m 19s\n- cross-keystone-py27 http://logs.openstack.org/98/655498/1/gate/cross-keystone-py27/dd269e6/ : SUCCESS in 15m 15s\n- cross-keystone-py35 http://logs.openstack.org/98/655498/1/gate/cross-keystone-py35/e9d7534/ : SUCCESS in 15m 18s\n- cross-neutron-py27 http://logs.openstack.org/98/655498/1/gate/cross-neutron-py27/ed198db/ : SUCCESS in 27m 25s\n- cross-neutron-py35 http://logs.openstack.org/98/655498/1/gate/cross-neutron-py35/1bb042c/ : SUCCESS in 22m 37s\n- cross-nova-py27 http://logs.openstack.org/98/655498/1/gate/cross-nova-py27/9e04957/ : SUCCESS in 19m 28s\n- cross-nova-py35 http://logs.openstack.org/98/655498/1/gate/cross-nova-py35/d4236e2/ : SUCCESS in 13m 17s\n- cross-nova-functional http://logs.openstack.org/98/655498/1/gate/cross-nova-functional/35d37c5/ : SUCCESS in 17m 27s\n- cross-swift-py27 http://logs.openstack.org/98/655498/1/gate/cross-swift-py27/ffbab3b/ : SUCCESS in 6m 53s\n- cross-swift-py35 http://logs.openstack.org/98/655498/1/gate/cross-swift-py35/6533143/ : SUCCESS in 3m 29s\n- requirements-integration http://logs.openstack.org/98/655498/1/gate/requirements-integration/0b85bda/ : SUCCESS in 56m 34s\n- neutron-grenade-multinode http://logs.openstack.org/98/655498/1/gate/neutron-grenade-multinode/41238ed/ : SUCCESS in 1h 02m 34s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/98/655498/1/gate/legacy-tempest-dsvm-neutron-full/f8644d3/ : SUCCESS in 1h 24m 31s\n- neutron-grenade http://logs.openstack.org/98/655498/1/gate/neutron-grenade/d8e8ddc/ : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ffb1b89c7433a4c995e504798cef6b18d9157e1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 12:02:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f8af1f6d6478a4bed99c12e409a6e8f2faf17d9","revisions":{"7f8af1f6d6478a4bed99c12e409a6e8f2faf17d9":{"kind":"REWORK","_number":1,"created":"2019-04-24 16:54:14.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/98/655498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/655498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/655498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/655498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/655498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/655498/1"}}},"commit":{"parents":[{"commit":"72935c8f0a0ef3302a7c3ee48beb49225d0438e8","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72935c8f0a0ef3302a7c3ee48beb49225d0438e8"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-04-24 16:50:00.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-04-24 16:54:03.000000000","tz":120},"subject":"Bump pylxd to 2.2.7","message":"Bump pylxd to 2.2.7\n\nWe cannot build wheels for pylxd 2.2.5 anymore. Its 2.2.5 setup.py\nfile install_requires have no upper bound on requests, and wheel\ncannot be built with latest requests.\n\nBumping to 2.2.7 is bumping to the lowest version which contains\na patch separating the install and setup_requires [1], allowing\nus to build its wheels. Anything above the Queens branch is not\nimpacted by the issue.\n\n[1]: https://github.com/lxc/pylxd/commit/4b8ab1802f9aee4eb29cf7b119dae0aa47150779\n\nChange-Id: I97b909385a8e6778a1a09afb26b8c33be1f294ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f8af1f6d6478a4bed99c12e409a6e8f2faf17d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f8af1f6d6478a4bed99c12e409a6e8f2faf17d9"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
