)]}'
{"id":"openstack%2Frequirements~655932","triplet_id":"openstack%2Frequirements~stable%2Fpike~Id044ac11a348dfc88ff16e102fd2cb03d1689fb1","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"Id044ac11a348dfc88ff16e102fd2cb03d1689fb1","subject":"update constraint for ironic-lib to new release 2.10.2","status":"MERGED","created":"2019-04-26 15:47:39.000000000","updated":"2019-04-29 19:58:48.000000000","submitted":"2019-04-29 19:58:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655932-1556567928751-332a2ac9","meta_rev_id":"12129a66250c8024d032ff5cb446486ef99bd4f5","_number":655932,"virtual_id_number":655932,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-04-29 19:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-27 05:09:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-29 12:28:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-04-29 12:28:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-27 05:09:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-29 12:28:27.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-29 19:58:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20592f71cddd4e78a6d4cf38e50c517e8fdfc1ee","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-26 15:47:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a84db666559ad14c7ced2389a5cb227df4010295","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 21:15:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/32/655932/1/check/openstack-tox-pypy/2d3927c/ : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/32/655932/1/check/openstack-tox-pep8/8e00075/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/32/655932/1/check/openstack-tox-py27/a063028/ : SUCCESS in 3m 53s\n- openstack-tox-py35 http://logs.openstack.org/32/655932/1/check/openstack-tox-py35/af99aa6/ : SUCCESS in 4m 06s\n- openstack-tox-validate http://logs.openstack.org/32/655932/1/check/openstack-tox-validate/0471895/ : SUCCESS in 3m 16s\n- requirements-tox-babel http://logs.openstack.org/32/655932/1/check/requirements-tox-babel/f543840/ : SUCCESS in 3m 04s\n- requirements-tox-py27-check-uc http://logs.openstack.org/32/655932/1/check/requirements-tox-py27-check-uc/d2adaaa/ : SUCCESS in 8m 47s\n- cross-cinder-py27 http://logs.openstack.org/32/655932/1/check/cross-cinder-py27/8bea73b/ : SUCCESS in 9m 10s\n- cross-cinder-py35 http://logs.openstack.org/32/655932/1/check/cross-cinder-py35/06411c3/ : SUCCESS in 7m 28s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/32/655932/1/check/cross-glance-py27/1deaf16/ : SUCCESS in 8m 27s\n- cross-glance-py35 http://logs.openstack.org/32/655932/1/check/cross-glance-py35/599fe2d/ : SUCCESS in 9m 11s\n- cross-horizon-py27 http://logs.openstack.org/32/655932/1/check/cross-horizon-py27/2f9a168/ : SUCCESS in 11m 03s\n- cross-horizon-py35 http://logs.openstack.org/32/655932/1/check/cross-horizon-py35/938cd2e/ : SUCCESS in 10m 46s\n- cross-keystone-py27 http://logs.openstack.org/32/655932/1/check/cross-keystone-py27/7c11e1d/ : SUCCESS in 11m 56s\n- cross-keystone-py35 http://logs.openstack.org/32/655932/1/check/cross-keystone-py35/8728a9e/ : SUCCESS in 11m 28s\n- cross-neutron-py27 http://logs.openstack.org/32/655932/1/check/cross-neutron-py27/21d07ff/ : SUCCESS in 22m 47s\n- cross-neutron-py35 http://logs.openstack.org/32/655932/1/check/cross-neutron-py35/9900fa8/ : SUCCESS in 25m 45s\n- cross-nova-py27 http://logs.openstack.org/32/655932/1/check/cross-nova-py27/eaf2bae/ : SUCCESS in 11m 24s\n- cross-nova-py35 http://logs.openstack.org/32/655932/1/check/cross-nova-py35/a3eddba/ : SUCCESS in 18m 22s\n- cross-nova-functional http://logs.openstack.org/32/655932/1/check/cross-nova-functional/503db4c/ : SUCCESS in 22m 23s\n- cross-swift-py27 http://logs.openstack.org/32/655932/1/check/cross-swift-py27/f7a7e80/ : SUCCESS in 7m 34s\n- cross-swift-py35 http://logs.openstack.org/32/655932/1/check/cross-swift-py35/7a547ec/ : SUCCESS in 3m 18s\n- requirements-integration http://logs.openstack.org/32/655932/1/check/requirements-integration/14d3285/ : SUCCESS in 55m 50s\n- neutron-grenade-multinode http://logs.openstack.org/32/655932/1/check/neutron-grenade-multinode/79b9020/ : SUCCESS in 1h 03m 14s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/32/655932/1/check/legacy-tempest-dsvm-neutron-full/2eb9f03/ : SUCCESS in 1h 26m 37s\n- neutron-grenade http://logs.openstack.org/32/655932/1/check/neutron-grenade/15ef33b/ : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e8b262fdff8d0893a391f8d0423bd41369463694","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-27 05:09:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"47eb65089d7b937fef1cf9feca57cb0f16cc09b4","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-29 12:28:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7c0a54977bca838fd752270269e5f8b2325f337a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 12:28:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ac8941a1f402ae8068cebd61c1cdb48d9d1be6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 14:17:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/32/655932/1/gate/openstack-tox-pep8/2c43a68/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/32/655932/1/gate/openstack-tox-py27/58fec82/ : SUCCESS in 5m 20s\n- openstack-tox-py35 http://logs.openstack.org/32/655932/1/gate/openstack-tox-py35/b008f72/ : SUCCESS in 4m 48s\n- openstack-tox-validate http://logs.openstack.org/32/655932/1/gate/openstack-tox-validate/014c1a1/ : SUCCESS in 3m 53s\n- requirements-tox-babel http://logs.openstack.org/32/655932/1/gate/requirements-tox-babel/c7b00cf/ : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc http://logs.openstack.org/32/655932/1/gate/requirements-tox-py27-check-uc/3884f3e/ : SUCCESS in 6m 28s\n- cross-cinder-py27 http://logs.openstack.org/32/655932/1/gate/cross-cinder-py27/c6e8bf1/ : SUCCESS in 8m 53s\n- cross-glance-py27 http://logs.openstack.org/32/655932/1/gate/cross-glance-py27/3bbf6a1/ : SUCCESS in 13m 29s\n- cross-glance-py35 http://logs.openstack.org/32/655932/1/gate/cross-glance-py35/3aa6b13/ : SUCCESS in 9m 42s\n- cross-horizon-py27 http://logs.openstack.org/32/655932/1/gate/cross-horizon-py27/546e184/ : SUCCESS in 11m 00s\n- cross-horizon-py35 http://logs.openstack.org/32/655932/1/gate/cross-horizon-py35/579a5ad/ : SUCCESS in 9m 26s\n- cross-keystone-py27 http://logs.openstack.org/32/655932/1/gate/cross-keystone-py27/5cf4959/ : SUCCESS in 12m 54s\n- cross-keystone-py35 http://logs.openstack.org/32/655932/1/gate/cross-keystone-py35/ba02f3e/ : SUCCESS in 13m 36s\n- cross-neutron-py27 http://logs.openstack.org/32/655932/1/gate/cross-neutron-py27/d96f434/ : SUCCESS in 24m 27s\n- cross-neutron-py35 http://logs.openstack.org/32/655932/1/gate/cross-neutron-py35/bea5f61/ : SUCCESS in 22m 44s\n- cross-nova-py27 http://logs.openstack.org/32/655932/1/gate/cross-nova-py27/c967497/ : SUCCESS in 19m 22s\n- cross-nova-py35 http://logs.openstack.org/32/655932/1/gate/cross-nova-py35/87ef549/ : SUCCESS in 14m 13s\n- cross-nova-functional http://logs.openstack.org/32/655932/1/gate/cross-nova-functional/a6c7d3d/ : SUCCESS in 19m 02s\n- cross-swift-py27 http://logs.openstack.org/32/655932/1/gate/cross-swift-py27/ca967bc/ : SUCCESS in 7m 54s\n- cross-swift-py35 http://logs.openstack.org/32/655932/1/gate/cross-swift-py35/01a28ec/ : SUCCESS in 3m 44s\n- requirements-integration http://logs.openstack.org/32/655932/1/gate/requirements-integration/e23b313/ : SUCCESS in 55m 16s\n- neutron-grenade-multinode http://logs.openstack.org/32/655932/1/gate/neutron-grenade-multinode/298ffe3/ : SUCCESS in 58m 53s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/32/655932/1/gate/legacy-tempest-dsvm-neutron-full/66b49fc/ : FAILURE in 1h 17m 05s\n- neutron-grenade http://logs.openstack.org/32/655932/1/gate/neutron-grenade/3998ac3/ : SUCCESS in 59m 46s","accounts_in_message":[],"_revision_number":1},{"id":"acdef1fe645efaa57858ba8effe0029c0ff1fa63","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-29 15:06:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ca7ecbf5071628628663c97aafc0d251cf46caf7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 16:29:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/32/655932/1/check/openstack-tox-pypy/4b995c6/ : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/32/655932/1/check/openstack-tox-pep8/b76a1fe/ : SUCCESS in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/32/655932/1/check/openstack-tox-py27/cf96544/ : SUCCESS in 3m 52s\n- openstack-tox-py35 http://logs.openstack.org/32/655932/1/check/openstack-tox-py35/e197e69/ : SUCCESS in 3m 44s\n- openstack-tox-validate http://logs.openstack.org/32/655932/1/check/openstack-tox-validate/81057a3/ : SUCCESS in 3m 07s\n- requirements-tox-babel http://logs.openstack.org/32/655932/1/check/requirements-tox-babel/fe4e26e/ : SUCCESS in 3m 26s\n- requirements-tox-py27-check-uc http://logs.openstack.org/32/655932/1/check/requirements-tox-py27-check-uc/38f6731/ : SUCCESS in 7m 07s\n- cross-cinder-py27 http://logs.openstack.org/32/655932/1/check/cross-cinder-py27/2a2dadb/ : SUCCESS in 10m 14s\n- cross-cinder-py35 http://logs.openstack.org/32/655932/1/check/cross-cinder-py35/db4f81c/ : SUCCESS in 8m 20s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/32/655932/1/check/cross-glance-py27/c44af4a/ : SUCCESS in 14m 05s\n- cross-glance-py35 http://logs.openstack.org/32/655932/1/check/cross-glance-py35/685ea89/ : SUCCESS in 10m 15s\n- cross-horizon-py27 http://logs.openstack.org/32/655932/1/check/cross-horizon-py27/45f19b2/ : SUCCESS in 10m 29s\n- cross-horizon-py35 http://logs.openstack.org/32/655932/1/check/cross-horizon-py35/f14897f/ : SUCCESS in 10m 10s\n- cross-keystone-py27 http://logs.openstack.org/32/655932/1/check/cross-keystone-py27/a1e9c06/ : SUCCESS in 16m 33s\n- cross-keystone-py35 http://logs.openstack.org/32/655932/1/check/cross-keystone-py35/04935cf/ : SUCCESS in 18m 36s\n- cross-neutron-py27 http://logs.openstack.org/32/655932/1/check/cross-neutron-py27/6c0a10b/ : SUCCESS in 26m 14s\n- cross-neutron-py35 http://logs.openstack.org/32/655932/1/check/cross-neutron-py35/44ee1cc/ : SUCCESS in 23m 38s\n- cross-nova-py27 http://logs.openstack.org/32/655932/1/check/cross-nova-py27/074b1c0/ : SUCCESS in 19m 59s\n- cross-nova-py35 http://logs.openstack.org/32/655932/1/check/cross-nova-py35/de0e5e3/ : SUCCESS in 12m 54s\n- cross-nova-functional http://logs.openstack.org/32/655932/1/check/cross-nova-functional/cd5ceec/ : SUCCESS in 18m 57s\n- cross-swift-py27 http://logs.openstack.org/32/655932/1/check/cross-swift-py27/40e8ce7/ : SUCCESS in 8m 53s\n- cross-swift-py35 http://logs.openstack.org/32/655932/1/check/cross-swift-py35/04cf777/ : SUCCESS in 3m 06s\n- requirements-integration http://logs.openstack.org/32/655932/1/check/requirements-integration/079b0de/ : SUCCESS in 58m 20s\n- neutron-grenade-multinode http://logs.openstack.org/32/655932/1/check/neutron-grenade-multinode/cdb8c5a/ : SUCCESS in 50m 55s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/32/655932/1/check/legacy-tempest-dsvm-neutron-full/704e06f/ : SUCCESS in 1h 21m 47s\n- neutron-grenade http://logs.openstack.org/32/655932/1/check/neutron-grenade/56ba685/ : SUCCESS in 50m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8c68852b676dc43215c43c53d70084c5e51f5fb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 16:29:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e131f2dd3896084e559d0a0dd196d5feb2cb420","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 17:49:36.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e3e7f46113f95f8c1cff210251eefa6fe11fa53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 19:58:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/32/655932/1/gate/openstack-tox-pep8/d26b907/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/32/655932/1/gate/openstack-tox-py27/e59a83a/ : SUCCESS in 5m 17s\n- openstack-tox-py35 http://logs.openstack.org/32/655932/1/gate/openstack-tox-py35/3198fcf/ : SUCCESS in 5m 17s\n- openstack-tox-validate http://logs.openstack.org/32/655932/1/gate/openstack-tox-validate/841bbc5/ : SUCCESS in 5m 32s\n- requirements-tox-babel http://logs.openstack.org/32/655932/1/gate/requirements-tox-babel/4ffdeba/ : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc http://logs.openstack.org/32/655932/1/gate/requirements-tox-py27-check-uc/0b3abcf/ : SUCCESS in 7m 14s\n- cross-cinder-py27 http://logs.openstack.org/32/655932/1/gate/cross-cinder-py27/c1b3cb0/ : SUCCESS in 7m 47s\n- cross-glance-py27 http://logs.openstack.org/32/655932/1/gate/cross-glance-py27/8120ce9/ : SUCCESS in 11m 33s\n- cross-glance-py35 http://logs.openstack.org/32/655932/1/gate/cross-glance-py35/be485d3/ : SUCCESS in 10m 52s\n- cross-horizon-py27 http://logs.openstack.org/32/655932/1/gate/cross-horizon-py27/7b8dc68/ : SUCCESS in 10m 04s\n- cross-horizon-py35 http://logs.openstack.org/32/655932/1/gate/cross-horizon-py35/ae049ed/ : SUCCESS in 11m 22s\n- cross-keystone-py27 http://logs.openstack.org/32/655932/1/gate/cross-keystone-py27/796e955/ : SUCCESS in 20m 45s\n- cross-keystone-py35 http://logs.openstack.org/32/655932/1/gate/cross-keystone-py35/7f9a4a4/ : SUCCESS in 33m 05s\n- cross-neutron-py27 http://logs.openstack.org/32/655932/1/gate/cross-neutron-py27/3e90e2c/ : SUCCESS in 24m 22s\n- cross-neutron-py35 http://logs.openstack.org/32/655932/1/gate/cross-neutron-py35/c8274e5/ : SUCCESS in 24m 12s\n- cross-nova-py27 http://logs.openstack.org/32/655932/1/gate/cross-nova-py27/3d84b63/ : SUCCESS in 21m 38s\n- cross-nova-py35 http://logs.openstack.org/32/655932/1/gate/cross-nova-py35/bcc1616/ : SUCCESS in 12m 09s\n- cross-nova-functional http://logs.openstack.org/32/655932/1/gate/cross-nova-functional/afdb667/ : SUCCESS in 26m 38s\n- cross-swift-py27 http://logs.openstack.org/32/655932/1/gate/cross-swift-py27/1f84b3a/ : SUCCESS in 7m 35s\n- cross-swift-py35 http://logs.openstack.org/32/655932/1/gate/cross-swift-py35/f30ce45/ : SUCCESS in 3m 16s\n- requirements-integration http://logs.openstack.org/32/655932/1/gate/requirements-integration/640d733/ : SUCCESS in 1h 06m 25s\n- neutron-grenade-multinode http://logs.openstack.org/32/655932/1/gate/neutron-grenade-multinode/f7576cc/ : SUCCESS in 1h 00m 59s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/32/655932/1/gate/legacy-tempest-dsvm-neutron-full/b96ffd1/ : SUCCESS in 1h 34m 54s\n- neutron-grenade http://logs.openstack.org/32/655932/1/gate/neutron-grenade/d545892/ : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b99e6bb6fc91f1d2a1355bdba8bd81959336eed9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 19:58:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bee82cd01ee06eb589e96e8fe768c6bef74f2518","revisions":{"bee82cd01ee06eb589e96e8fe768c6bef74f2518":{"kind":"REWORK","_number":1,"created":"2019-04-26 15:47:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/655932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/655932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/655932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/655932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/655932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/655932/1"}}},"commit":{"parents":[{"commit":"6c627edcc32c0e65d75d93084c12e1de30a348c1","subject":"Merge \"update constraint for keystonemiddleware to new release 4.17.1\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c627edcc32c0e65d75d93084c12e1de30a348c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 15:47:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 15:47:38.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 2.10.2","message":"update constraint for ironic-lib to new release 2.10.2\n\nChange-Id: Id044ac11a348dfc88ff16e102fd2cb03d1689fb1\nmeta:version: 2.10.2\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Change-Id: I78721d822e5d113a4ebfea8ec59d8cc1a939043c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bee82cd01ee06eb589e96e8fe768c6bef74f2518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bee82cd01ee06eb589e96e8fe768c6bef74f2518"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
