)]}'
{"id":"openstack%2Frequirements~655948","triplet_id":"openstack%2Frequirements~stable%2Fpike~Ie3ab685614a362b940589ef5485c88cd9351ccde","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"Ie3ab685614a362b940589ef5485c88cd9351ccde","subject":"update constraint for os-win to new release 2.2.1","status":"MERGED","created":"2019-04-26 16:34:52.000000000","updated":"2019-04-29 14:17:27.000000000","submitted":"2019-04-29 14:17:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655948-1556547447522-cdd0ea5d","meta_rev_id":"9636e7d4f4d5857561bdeddc8ecb3066b6cdfede","_number":655948,"virtual_id_number":655948,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-04-29 14:17:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-27 13:40:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-29 12:27:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-04-29 12:27:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-27 13:40:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-29 12:27:27.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-29 14:17:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"069c3d08fc9bbe34e7d193f03ec1b9f28115728d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-26 16:34:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d6a86681d78c9ecc2cb176dc7f5054e8155a722","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 22:29:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/48/655948/1/check/openstack-tox-pypy/a5771d8/ : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/48/655948/1/check/openstack-tox-pep8/74fee79/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/48/655948/1/check/openstack-tox-py27/143af4a/ : SUCCESS in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/48/655948/1/check/openstack-tox-py35/1906fdf/ : SUCCESS in 3m 43s\n- openstack-tox-validate http://logs.openstack.org/48/655948/1/check/openstack-tox-validate/63bb306/ : SUCCESS in 3m 38s\n- requirements-tox-babel http://logs.openstack.org/48/655948/1/check/requirements-tox-babel/1113776/ : SUCCESS in 4m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/48/655948/1/check/requirements-tox-py27-check-uc/bb3cdc6/ : SUCCESS in 8m 08s\n- cross-cinder-py27 http://logs.openstack.org/48/655948/1/check/cross-cinder-py27/c113b5f/ : SUCCESS in 8m 35s\n- cross-cinder-py35 http://logs.openstack.org/48/655948/1/check/cross-cinder-py35/5490452/ : FAILURE in 11m 42s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/48/655948/1/check/cross-glance-py27/3d84a35/ : SUCCESS in 8m 47s\n- cross-glance-py35 http://logs.openstack.org/48/655948/1/check/cross-glance-py35/ce9bc69/ : SUCCESS in 9m 03s\n- cross-horizon-py27 http://logs.openstack.org/48/655948/1/check/cross-horizon-py27/875a66a/ : SUCCESS in 13m 38s\n- cross-horizon-py35 http://logs.openstack.org/48/655948/1/check/cross-horizon-py35/e37b9ed/ : SUCCESS in 10m 46s\n- cross-keystone-py27 http://logs.openstack.org/48/655948/1/check/cross-keystone-py27/753f2f4/ : SUCCESS in 10m 37s\n- cross-keystone-py35 http://logs.openstack.org/48/655948/1/check/cross-keystone-py35/03e1ce9/ : SUCCESS in 11m 57s\n- cross-neutron-py27 http://logs.openstack.org/48/655948/1/check/cross-neutron-py27/0605f02/ : SUCCESS in 20m 44s\n- cross-neutron-py35 http://logs.openstack.org/48/655948/1/check/cross-neutron-py35/819657c/ : SUCCESS in 20m 10s\n- cross-nova-py27 http://logs.openstack.org/48/655948/1/check/cross-nova-py27/f735f9f/ : SUCCESS in 11m 54s\n- cross-nova-py35 http://logs.openstack.org/48/655948/1/check/cross-nova-py35/3b2479d/ : SUCCESS in 12m 05s\n- cross-nova-functional http://logs.openstack.org/48/655948/1/check/cross-nova-functional/b9a3633/ : SUCCESS in 26m 25s\n- cross-swift-py27 http://logs.openstack.org/48/655948/1/check/cross-swift-py27/874cd2b/ : SUCCESS in 7m 45s\n- cross-swift-py35 http://logs.openstack.org/48/655948/1/check/cross-swift-py35/ab92c41/ : SUCCESS in 4m 08s\n- requirements-integration http://logs.openstack.org/48/655948/1/check/requirements-integration/efbaee1/ : SUCCESS in 50m 56s\n- neutron-grenade-multinode http://logs.openstack.org/48/655948/1/check/neutron-grenade-multinode/09a55de/ : SUCCESS in 1h 06m 46s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/655948/1/check/legacy-tempest-dsvm-neutron-full/a54d8f0/ : FAILURE in 1h 29m 53s\n- neutron-grenade http://logs.openstack.org/48/655948/1/check/neutron-grenade/93d8037/ : SUCCESS in 1h 02m 23s","accounts_in_message":[],"_revision_number":1},{"id":"183a1a724d646ec3259e1cc0544384d98a72049d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-27 05:09:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"49f3be4acfc1c62c34f536d727763410c370dcea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 06:58:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/48/655948/1/check/openstack-tox-pypy/0faf528/ : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/48/655948/1/check/openstack-tox-pep8/a21e8c3/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/48/655948/1/check/openstack-tox-py27/17f5c5b/ : SUCCESS in 3m 37s\n- openstack-tox-py35 http://logs.openstack.org/48/655948/1/check/openstack-tox-py35/7f645f8/ : SUCCESS in 3m 32s\n- openstack-tox-validate http://logs.openstack.org/48/655948/1/check/openstack-tox-validate/8ca7180/ : SUCCESS in 2m 58s\n- requirements-tox-babel http://logs.openstack.org/48/655948/1/check/requirements-tox-babel/a993630/ : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/48/655948/1/check/requirements-tox-py27-check-uc/8e1b92f/ : SUCCESS in 6m 20s\n- cross-cinder-py27 http://logs.openstack.org/48/655948/1/check/cross-cinder-py27/81209e6/ : SUCCESS in 7m 18s\n- cross-cinder-py35 http://logs.openstack.org/48/655948/1/check/cross-cinder-py35/448ee99/ : SUCCESS in 8m 04s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/48/655948/1/check/cross-glance-py27/8518e87/ : SUCCESS in 9m 36s\n- cross-glance-py35 http://logs.openstack.org/48/655948/1/check/cross-glance-py35/4339fbe/ : SUCCESS in 9m 22s\n- cross-horizon-py27 http://logs.openstack.org/48/655948/1/check/cross-horizon-py27/f43cf36/ : SUCCESS in 11m 11s\n- cross-horizon-py35 http://logs.openstack.org/48/655948/1/check/cross-horizon-py35/2c07c67/ : SUCCESS in 10m 16s\n- cross-keystone-py27 http://logs.openstack.org/48/655948/1/check/cross-keystone-py27/09a0e8a/ : SUCCESS in 11m 37s\n- cross-keystone-py35 http://logs.openstack.org/48/655948/1/check/cross-keystone-py35/a992edb/ : SUCCESS in 27m 59s\n- cross-neutron-py27 http://logs.openstack.org/48/655948/1/check/cross-neutron-py27/28a1ec3/ : SUCCESS in 27m 56s\n- cross-neutron-py35 http://logs.openstack.org/48/655948/1/check/cross-neutron-py35/0f193da/ : SUCCESS in 23m 10s\n- cross-nova-py27 http://logs.openstack.org/48/655948/1/check/cross-nova-py27/c0e2756/ : SUCCESS in 13m 22s\n- cross-nova-py35 http://logs.openstack.org/48/655948/1/check/cross-nova-py35/ebe82c2/ : SUCCESS in 14m 18s\n- cross-nova-functional http://logs.openstack.org/48/655948/1/check/cross-nova-functional/a213fc1/ : SUCCESS in 18m 24s\n- cross-swift-py27 http://logs.openstack.org/48/655948/1/check/cross-swift-py27/71bb851/ : SUCCESS in 7m 03s\n- cross-swift-py35 http://logs.openstack.org/48/655948/1/check/cross-swift-py35/a0e0a15/ : SUCCESS in 3m 26s\n- requirements-integration http://logs.openstack.org/48/655948/1/check/requirements-integration/ec31a36/ : SUCCESS in 51m 28s\n- neutron-grenade-multinode http://logs.openstack.org/48/655948/1/check/neutron-grenade-multinode/47499c0/ : SUCCESS in 1h 06m 48s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/655948/1/check/legacy-tempest-dsvm-neutron-full/5c5bde1/ : SUCCESS in 1h 42m 06s\n- neutron-grenade http://logs.openstack.org/48/655948/1/check/neutron-grenade/e58211b/ : SUCCESS in 45m 16s","accounts_in_message":[],"_revision_number":1},{"id":"78bcd6b37c31922d5b8ab7d99fec945f79d2335e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-27 13:40:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5ee74a3760be4483feeccfb9968639d00817292e","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-29 12:27:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35fd038117ba107efaef6d0975f438676dc36c03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 12:27:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7d46770db48530a294a4062f85bf516561591a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 14:17:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"04f245f51156de306f54e5fc2669cee6c6843570","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 14:17:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/48/655948/1/gate/openstack-tox-pep8/648c8ff/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/48/655948/1/gate/openstack-tox-py27/9b17792/ : SUCCESS in 4m 06s\n- openstack-tox-py35 http://logs.openstack.org/48/655948/1/gate/openstack-tox-py35/e70d22b/ : SUCCESS in 4m 05s\n- openstack-tox-validate http://logs.openstack.org/48/655948/1/gate/openstack-tox-validate/aa6a856/ : SUCCESS in 4m 13s\n- requirements-tox-babel http://logs.openstack.org/48/655948/1/gate/requirements-tox-babel/6c0cd65/ : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc http://logs.openstack.org/48/655948/1/gate/requirements-tox-py27-check-uc/6b44180/ : SUCCESS in 9m 50s\n- cross-cinder-py27 http://logs.openstack.org/48/655948/1/gate/cross-cinder-py27/acd0e4d/ : SUCCESS in 8m 03s\n- cross-glance-py27 http://logs.openstack.org/48/655948/1/gate/cross-glance-py27/6801604/ : SUCCESS in 8m 14s\n- cross-glance-py35 http://logs.openstack.org/48/655948/1/gate/cross-glance-py35/716ca4a/ : SUCCESS in 9m 11s\n- cross-horizon-py27 http://logs.openstack.org/48/655948/1/gate/cross-horizon-py27/e1242c6/ : SUCCESS in 12m 21s\n- cross-horizon-py35 http://logs.openstack.org/48/655948/1/gate/cross-horizon-py35/4f67c8f/ : SUCCESS in 10m 12s\n- cross-keystone-py27 http://logs.openstack.org/48/655948/1/gate/cross-keystone-py27/87182ac/ : SUCCESS in 11m 49s\n- cross-keystone-py35 http://logs.openstack.org/48/655948/1/gate/cross-keystone-py35/c6fc1ee/ : SUCCESS in 13m 13s\n- cross-neutron-py27 http://logs.openstack.org/48/655948/1/gate/cross-neutron-py27/0807c2c/ : SUCCESS in 25m 35s\n- cross-neutron-py35 http://logs.openstack.org/48/655948/1/gate/cross-neutron-py35/3adbc0d/ : SUCCESS in 25m 02s\n- cross-nova-py27 http://logs.openstack.org/48/655948/1/gate/cross-nova-py27/bbc7f5f/ : SUCCESS in 12m 01s\n- cross-nova-py35 http://logs.openstack.org/48/655948/1/gate/cross-nova-py35/18c3de5/ : SUCCESS in 12m 46s\n- cross-nova-functional http://logs.openstack.org/48/655948/1/gate/cross-nova-functional/af0c2ce/ : SUCCESS in 28m 44s\n- cross-swift-py27 http://logs.openstack.org/48/655948/1/gate/cross-swift-py27/dfa35be/ : SUCCESS in 7m 18s\n- cross-swift-py35 http://logs.openstack.org/48/655948/1/gate/cross-swift-py35/a0570c8/ : SUCCESS in 3m 51s\n- requirements-integration http://logs.openstack.org/48/655948/1/gate/requirements-integration/e87f694/ : SUCCESS in 49m 00s\n- neutron-grenade-multinode http://logs.openstack.org/48/655948/1/gate/neutron-grenade-multinode/6b1f1db/ : SUCCESS in 1h 03m 08s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/48/655948/1/gate/legacy-tempest-dsvm-neutron-full/f4a41ce/ : SUCCESS in 1h 19m 50s\n- neutron-grenade http://logs.openstack.org/48/655948/1/gate/neutron-grenade/93c1004/ : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f665855b6b43d7abea9491c6c1ec19df8155225c","revisions":{"f665855b6b43d7abea9491c6c1ec19df8155225c":{"kind":"REWORK","_number":1,"created":"2019-04-26 16:34:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/655948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/655948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/655948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/655948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/655948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/655948/1"}}},"commit":{"parents":[{"commit":"6c627edcc32c0e65d75d93084c12e1de30a348c1","subject":"Merge \"update constraint for keystonemiddleware to new release 4.17.1\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c627edcc32c0e65d75d93084c12e1de30a348c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 16:34:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 16:34:50.000000000","tz":0},"subject":"update constraint for os-win to new release 2.2.1","message":"update constraint for os-win to new release 2.2.1\n\nChange-Id: Ie3ab685614a362b940589ef5485c88cd9351ccde\nmeta:version: 2.2.1\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Change-Id: I6ed88d6e52f68cd15a73c0b8262d773e9aae1751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f665855b6b43d7abea9491c6c1ec19df8155225c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f665855b6b43d7abea9491c6c1ec19df8155225c"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
