)]}'
{"id":"openstack%2Frequirements~655955","triplet_id":"openstack%2Frequirements~master~I3b1bc3e496e3a1880e3cec140a9279aefd8e7b61","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3b1bc3e496e3a1880e3cec140a9279aefd8e7b61","subject":"update constraint for oslo.versionedobjects to new release 1.36.0","status":"MERGED","created":"2019-04-26 16:55:58.000000000","updated":"2019-04-28 06:37:43.000000000","submitted":"2019-04-28 06:37:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655955-1556433463498-7d7fc31e","meta_rev_id":"fd9a6435ce17182021487c77a9ae9efac36be89e","_number":655955,"virtual_id_number":655955,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-04-28 06:37:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-26 23:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-04-27 19:02:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-04-26 23:29:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-26 23:29:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-27 19:02:27.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-04-28 06:37:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e4d05efe8b9198cc68973f1f2d63a0b0f67f337","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-26 16:55:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50b8fd22ca208f314a3b06812d063ec42ad53bfa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 23:05:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/655955/1/check/openstack-tox-docs/cf8e1e5/html/ : SUCCESS in 2m 54s\n- openstack-tox-pypy http://logs.openstack.org/55/655955/1/check/openstack-tox-pypy/d4aca2d/ : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/55/655955/1/check/openstack-tox-pep8/ae31807/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/55/655955/1/check/openstack-tox-py27/8f49f0f/ : SUCCESS in 2m 58s\n- openstack-tox-py36 http://logs.openstack.org/55/655955/1/check/openstack-tox-py36/190a7c7/ : SUCCESS in 3m 25s\n- openstack-tox-validate http://logs.openstack.org/55/655955/1/check/openstack-tox-validate/4fd1d3b/ : SUCCESS in 3m 11s\n- requirements-tox-babel http://logs.openstack.org/55/655955/1/check/requirements-tox-babel/89b8c13/ : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc http://logs.openstack.org/55/655955/1/check/requirements-tox-py27-check-uc/fcc499b/ : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc http://logs.openstack.org/55/655955/1/check/requirements-tox-py36-check-uc/635494e/ : SUCCESS in 6m 22s\n- cross-cinder-py27 http://logs.openstack.org/55/655955/1/check/cross-cinder-py27/c9e27d8/ : SUCCESS in 9m 05s\n- cross-cinder-py36 http://logs.openstack.org/55/655955/1/check/cross-cinder-py36/bd41e3c/ : SUCCESS in 8m 16s\n- cross-glance-py27 http://logs.openstack.org/55/655955/1/check/cross-glance-py27/a2e3a79/ : SUCCESS in 4m 15s\n- cross-glance-py36 http://logs.openstack.org/55/655955/1/check/cross-glance-py36/2b33843/ : SUCCESS in 5m 00s\n- cross-horizon-py27 http://logs.openstack.org/55/655955/1/check/cross-horizon-py27/5e7e667/ : SUCCESS in 9m 59s\n- cross-horizon-py36 http://logs.openstack.org/55/655955/1/check/cross-horizon-py36/3ac2595/ : SUCCESS in 9m 31s\n- cross-horizon-npm http://logs.openstack.org/55/655955/1/check/cross-horizon-npm/55623c1/ : SUCCESS in 6m 01s\n- cross-keystone-py27 http://logs.openstack.org/55/655955/1/check/cross-keystone-py27/64502e5/ : SUCCESS in 20m 38s\n- cross-keystone-py36 http://logs.openstack.org/55/655955/1/check/cross-keystone-py36/3908d96/ : SUCCESS in 22m 51s\n- cross-neutron-py27 http://logs.openstack.org/55/655955/1/check/cross-neutron-py27/c32516c/ : SUCCESS in 23m 04s\n- cross-neutron-py36 http://logs.openstack.org/55/655955/1/check/cross-neutron-py36/b1cc519/ : SUCCESS in 27m 51s\n- cross-nova-py27 http://logs.openstack.org/55/655955/1/check/cross-nova-py27/0e348e9/ : SUCCESS in 12m 34s\n- cross-nova-py36 http://logs.openstack.org/55/655955/1/check/cross-nova-py36/a3a7c76/ : SUCCESS in 11m 49s\n- cross-nova-functional http://logs.openstack.org/55/655955/1/check/cross-nova-functional/01f8c22/ : SUCCESS in 21m 45s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/55/655955/1/check/cross-osc-build-sphinx-docs/56e3193/html/ : SUCCESS in 7m 23s\n- cross-swift-py27 http://logs.openstack.org/55/655955/1/check/cross-swift-py27/df6085f/ : SUCCESS in 8m 13s\n- cross-swift-py36 http://logs.openstack.org/55/655955/1/check/cross-swift-py36/9d6c1ee/ : SUCCESS in 7m 24s\n- requirements-integration http://logs.openstack.org/55/655955/1/check/requirements-integration/6cf97e6/ : SUCCESS in 50m 37s\n- neutron-grenade-multinode http://logs.openstack.org/55/655955/1/check/neutron-grenade-multinode/445aa3a/ : SUCCESS in 1h 12m 59s\n- tempest-full http://logs.openstack.org/55/655955/1/check/tempest-full/2e55b22/ : SUCCESS in 1h 50m 42s\n- neutron-grenade http://logs.openstack.org/55/655955/1/check/neutron-grenade/9423aeb/ : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":1},{"id":"fb938996b3ea69c4416ae0f699652398e93c1e58","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-26 23:29:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bcfa536ce31635ca1905d0f74348b9db4373f48c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 23:30:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"199905e49b7e494f4565ba766c0fc793bb93c569","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 09:15:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/55/655955/1/gate/openstack-tox-docs/daa2005/html/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/55/655955/1/gate/openstack-tox-pep8/8956112/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/55/655955/1/gate/openstack-tox-py27/c51082b/ : SUCCESS in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/55/655955/1/gate/openstack-tox-py36/a7b0700/ : SUCCESS in 4m 02s\n- openstack-tox-validate http://logs.openstack.org/55/655955/1/gate/openstack-tox-validate/437edf7/ : SUCCESS in 3m 31s\n- requirements-tox-babel http://logs.openstack.org/55/655955/1/gate/requirements-tox-babel/3661dec/ : SUCCESS in 3m 54s\n- requirements-tox-py27-check-uc http://logs.openstack.org/55/655955/1/gate/requirements-tox-py27-check-uc/dae344d/ : SUCCESS in 7m 45s\n- requirements-tox-py36-check-uc http://logs.openstack.org/55/655955/1/gate/requirements-tox-py36-check-uc/de7409b/ : SUCCESS in 8m 21s\n- cross-cinder-py27 http://logs.openstack.org/55/655955/1/gate/cross-cinder-py27/c4e1308/ : SUCCESS in 9m 28s\n- cross-cinder-py36 http://logs.openstack.org/55/655955/1/gate/cross-cinder-py36/9af945f/ : SUCCESS in 8m 05s\n- cross-glance-py27 http://logs.openstack.org/55/655955/1/gate/cross-glance-py27/5da55b0/ : SUCCESS in 5m 01s\n- cross-glance-py36 http://logs.openstack.org/55/655955/1/gate/cross-glance-py36/fd6290e/ : SUCCESS in 5m 21s\n- cross-horizon-py27 http://logs.openstack.org/55/655955/1/gate/cross-horizon-py27/f549828/ : SUCCESS in 10m 32s\n- cross-horizon-py36 http://logs.openstack.org/55/655955/1/gate/cross-horizon-py36/21ceae0/ : SUCCESS in 9m 29s\n- cross-horizon-npm http://logs.openstack.org/55/655955/1/gate/cross-horizon-npm/fa8015b/ : SUCCESS in 6m 35s\n- cross-keystone-py27 http://logs.openstack.org/55/655955/1/gate/cross-keystone-py27/f654066/ : SUCCESS in 22m 08s\n- cross-keystone-py36 http://logs.openstack.org/55/655955/1/gate/cross-keystone-py36/e42e4f9/ : SUCCESS in 22m 33s\n- cross-neutron-py27 http://logs.openstack.org/55/655955/1/gate/cross-neutron-py27/45dd10a/ : SUCCESS in 22m 50s\n- cross-neutron-py36 http://logs.openstack.org/55/655955/1/gate/cross-neutron-py36/104fe36/ : SUCCESS in 25m 20s\n- cross-nova-py27 http://logs.openstack.org/55/655955/1/gate/cross-nova-py27/c23b3ee/ : SUCCESS in 12m 16s\n- cross-nova-py36 http://logs.openstack.org/55/655955/1/gate/cross-nova-py36/0a24053/ : FAILURE in 11m 31s\n- cross-nova-functional http://logs.openstack.org/55/655955/1/gate/cross-nova-functional/f065229/ : SUCCESS in 19m 03s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/55/655955/1/gate/cross-osc-build-sphinx-docs/9678cce/html/ : SUCCESS in 5m 35s\n- cross-swift-py27 http://logs.openstack.org/55/655955/1/gate/cross-swift-py27/16eeaef/ : SUCCESS in 8m 14s\n- cross-swift-py36 http://logs.openstack.org/55/655955/1/gate/cross-swift-py36/4602214/ : SUCCESS in 8m 10s\n- requirements-integration http://logs.openstack.org/55/655955/1/gate/requirements-integration/e327bcc/ : SUCCESS in 1h 01m 19s\n- neutron-grenade-multinode http://logs.openstack.org/55/655955/1/gate/neutron-grenade-multinode/dd401c1/ : SUCCESS in 1h 08m 38s\n- tempest-full http://logs.openstack.org/55/655955/1/gate/tempest-full/7911170/ : SUCCESS in 1h 40m 10s\n- neutron-grenade http://logs.openstack.org/55/655955/1/gate/neutron-grenade/fc81acf/ : SUCCESS in 1h 05m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b4d65f22db01b740b8f4b14eb7de94f441feb3ab","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-27 13:41:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6034777e59dfaeee13907bd6084d70ed8d09021e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-27 13:41:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6dae9539cd2e151837b3135904c869313ba27ed6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 15:24:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/55/655955/1/check/openstack-tox-docs/0176b0b/html/ : SUCCESS in 3m 09s\n- openstack-tox-pypy http://logs.openstack.org/55/655955/1/check/openstack-tox-pypy/038c872/ : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/55/655955/1/check/openstack-tox-pep8/4b99b43/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/55/655955/1/check/openstack-tox-py27/de5b16b/ : SUCCESS in 4m 56s\n- openstack-tox-py36 http://logs.openstack.org/55/655955/1/check/openstack-tox-py36/0f90f47/ : SUCCESS in 4m 02s\n- openstack-tox-validate http://logs.openstack.org/55/655955/1/check/openstack-tox-validate/e06da28/ : SUCCESS in 4m 55s\n- requirements-tox-babel http://logs.openstack.org/55/655955/1/check/requirements-tox-babel/0a3a989/ : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/55/655955/1/check/requirements-tox-py27-check-uc/f92d01e/ : SUCCESS in 8m 07s\n- requirements-tox-py36-check-uc http://logs.openstack.org/55/655955/1/check/requirements-tox-py36-check-uc/02b6ab0/ : SUCCESS in 7m 16s\n- cross-cinder-py27 http://logs.openstack.org/55/655955/1/check/cross-cinder-py27/dbbaea2/ : SUCCESS in 9m 21s\n- cross-cinder-py36 http://logs.openstack.org/55/655955/1/check/cross-cinder-py36/8802c11/ : SUCCESS in 9m 02s\n- cross-glance-py27 http://logs.openstack.org/55/655955/1/check/cross-glance-py27/e499225/ : SUCCESS in 7m 18s\n- cross-glance-py36 http://logs.openstack.org/55/655955/1/check/cross-glance-py36/5eccb92/ : SUCCESS in 5m 25s\n- cross-horizon-py27 http://logs.openstack.org/55/655955/1/check/cross-horizon-py27/e3502e9/ : SUCCESS in 10m 02s\n- cross-horizon-py36 http://logs.openstack.org/55/655955/1/check/cross-horizon-py36/48505b6/ : SUCCESS in 9m 44s\n- cross-horizon-npm http://logs.openstack.org/55/655955/1/check/cross-horizon-npm/38ed06b/ : SUCCESS in 8m 09s\n- cross-keystone-py27 http://logs.openstack.org/55/655955/1/check/cross-keystone-py27/eda0c83/ : SUCCESS in 18m 09s\n- cross-keystone-py36 http://logs.openstack.org/55/655955/1/check/cross-keystone-py36/e80b0aa/ : SUCCESS in 22m 27s\n- cross-neutron-py27 http://logs.openstack.org/55/655955/1/check/cross-neutron-py27/fa3647b/ : SUCCESS in 28m 11s\n- cross-neutron-py36 http://logs.openstack.org/55/655955/1/check/cross-neutron-py36/97f9e6b/ : SUCCESS in 30m 09s\n- cross-nova-py27 http://logs.openstack.org/55/655955/1/check/cross-nova-py27/538636d/ : SUCCESS in 25m 34s\n- cross-nova-py36 http://logs.openstack.org/55/655955/1/check/cross-nova-py36/90d34bf/ : FAILURE in 25m 56s\n- cross-nova-functional http://logs.openstack.org/55/655955/1/check/cross-nova-functional/ae69305/ : SUCCESS in 27m 47s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/55/655955/1/check/cross-osc-build-sphinx-docs/c73e780/html/ : SUCCESS in 7m 48s\n- cross-swift-py27 http://logs.openstack.org/55/655955/1/check/cross-swift-py27/5de036a/ : SUCCESS in 12m 45s\n- cross-swift-py36 http://logs.openstack.org/55/655955/1/check/cross-swift-py36/9cfe49d/ : SUCCESS in 12m 43s\n- requirements-integration http://logs.openstack.org/55/655955/1/check/requirements-integration/0bf2c9d/ : SUCCESS in 1h 11m 56s\n- neutron-grenade-multinode http://logs.openstack.org/55/655955/1/check/neutron-grenade-multinode/be3ff85/ : SUCCESS in 1h 24m 13s\n- tempest-full http://logs.openstack.org/55/655955/1/check/tempest-full/27647eb/ : FAILURE in 1h 39m 41s\n- neutron-grenade http://logs.openstack.org/55/655955/1/check/neutron-grenade/284ef47/ : SUCCESS in 1h 11m 39s","accounts_in_message":[],"_revision_number":1},{"id":"4fe768e79a5c6423dde0224d426bab1a2a9d5d0d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-27 19:02:27.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"260f3756db8512071c554880d39220d49fdd74e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 20:48:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/55/655955/1/check/openstack-tox-docs/7d82df8/html/ : SUCCESS in 4m 15s\n- openstack-tox-pypy http://logs.openstack.org/55/655955/1/check/openstack-tox-pypy/241d3c1/ : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/55/655955/1/check/openstack-tox-pep8/fd95ffd/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/55/655955/1/check/openstack-tox-py27/a500501/ : SUCCESS in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/55/655955/1/check/openstack-tox-py36/8d8269a/ : SUCCESS in 3m 47s\n- openstack-tox-validate http://logs.openstack.org/55/655955/1/check/openstack-tox-validate/e4df5f8/ : SUCCESS in 2m 48s\n- requirements-tox-babel http://logs.openstack.org/55/655955/1/check/requirements-tox-babel/12de291/ : SUCCESS in 2m 59s\n- requirements-tox-py27-check-uc http://logs.openstack.org/55/655955/1/check/requirements-tox-py27-check-uc/38151d3/ : SUCCESS in 6m 57s\n- requirements-tox-py36-check-uc http://logs.openstack.org/55/655955/1/check/requirements-tox-py36-check-uc/4863432/ : SUCCESS in 7m 55s\n- cross-cinder-py27 http://logs.openstack.org/55/655955/1/check/cross-cinder-py27/1130a0f/ : SUCCESS in 8m 17s\n- cross-cinder-py36 http://logs.openstack.org/55/655955/1/check/cross-cinder-py36/bf3fecd/ : SUCCESS in 9m 02s\n- cross-glance-py27 http://logs.openstack.org/55/655955/1/check/cross-glance-py27/cbd802a/ : SUCCESS in 4m 26s\n- cross-glance-py36 http://logs.openstack.org/55/655955/1/check/cross-glance-py36/2c5e610/ : SUCCESS in 4m 55s\n- cross-horizon-py27 http://logs.openstack.org/55/655955/1/check/cross-horizon-py27/528cf6f/ : SUCCESS in 10m 28s\n- cross-horizon-py36 http://logs.openstack.org/55/655955/1/check/cross-horizon-py36/995dcfd/ : SUCCESS in 7m 37s\n- cross-horizon-npm http://logs.openstack.org/55/655955/1/check/cross-horizon-npm/d1069fc/ : SUCCESS in 5m 35s\n- cross-keystone-py27 http://logs.openstack.org/55/655955/1/check/cross-keystone-py27/424419c/ : SUCCESS in 18m 35s\n- cross-keystone-py36 http://logs.openstack.org/55/655955/1/check/cross-keystone-py36/ab5cb43/ : SUCCESS in 21m 40s\n- cross-neutron-py27 http://logs.openstack.org/55/655955/1/check/cross-neutron-py27/cc0b058/ : SUCCESS in 19m 31s\n- cross-neutron-py36 http://logs.openstack.org/55/655955/1/check/cross-neutron-py36/1db590a/ : SUCCESS in 17m 58s\n- cross-nova-py27 http://logs.openstack.org/55/655955/1/check/cross-nova-py27/e58dd7a/ : SUCCESS in 16m 17s\n- cross-nova-py36 http://logs.openstack.org/55/655955/1/check/cross-nova-py36/64e3dcd/ : FAILURE in 14m 22s\n- cross-nova-functional http://logs.openstack.org/55/655955/1/check/cross-nova-functional/f75771b/ : SUCCESS in 18m 21s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/55/655955/1/check/cross-osc-build-sphinx-docs/f12275a/html/ : SUCCESS in 6m 09s\n- cross-swift-py27 http://logs.openstack.org/55/655955/1/check/cross-swift-py27/c778d20/ : SUCCESS in 12m 08s\n- cross-swift-py36 http://logs.openstack.org/55/655955/1/check/cross-swift-py36/e6af571/ : SUCCESS in 6m 29s\n- requirements-integration http://logs.openstack.org/55/655955/1/check/requirements-integration/f7fe1e0/ : SUCCESS in 59m 44s\n- neutron-grenade-multinode http://logs.openstack.org/55/655955/1/check/neutron-grenade-multinode/c675cb0/ : SUCCESS in 1h 11m 51s\n- tempest-full http://logs.openstack.org/55/655955/1/check/tempest-full/e27d633/ : SUCCESS in 1h 41m 32s\n- neutron-grenade http://logs.openstack.org/55/655955/1/check/neutron-grenade/b5bb4bc/ : SUCCESS in 57m 19s","accounts_in_message":[],"_revision_number":1},{"id":"17d1765c66efed36fc9c06886cb9ec8f4f77abd7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-28 02:57:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"61e2a7d09a52e509994695ef2d39009bdef49df7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 04:47:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/655955/1/check/openstack-tox-docs/e100ec6/html/ : SUCCESS in 2m 52s\n- openstack-tox-pypy http://logs.openstack.org/55/655955/1/check/openstack-tox-pypy/01bdcb4/ : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/55/655955/1/check/openstack-tox-pep8/1abcdbc/ : SUCCESS in 2m 46s\n- openstack-tox-py27 http://logs.openstack.org/55/655955/1/check/openstack-tox-py27/d4294b5/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/55/655955/1/check/openstack-tox-py36/47bbf75/ : SUCCESS in 3m 13s\n- openstack-tox-validate http://logs.openstack.org/55/655955/1/check/openstack-tox-validate/c9f7734/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/55/655955/1/check/requirements-tox-babel/fa3715c/ : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/55/655955/1/check/requirements-tox-py27-check-uc/613c097/ : SUCCESS in 8m 00s\n- requirements-tox-py36-check-uc http://logs.openstack.org/55/655955/1/check/requirements-tox-py36-check-uc/c7ff3b0/ : SUCCESS in 9m 00s\n- cross-cinder-py27 http://logs.openstack.org/55/655955/1/check/cross-cinder-py27/19fbcc1/ : SUCCESS in 9m 04s\n- cross-cinder-py36 http://logs.openstack.org/55/655955/1/check/cross-cinder-py36/2a18b67/ : SUCCESS in 9m 25s\n- cross-glance-py27 http://logs.openstack.org/55/655955/1/check/cross-glance-py27/61b47ad/ : SUCCESS in 3m 54s\n- cross-glance-py36 http://logs.openstack.org/55/655955/1/check/cross-glance-py36/63ac1ce/ : SUCCESS in 4m 19s\n- cross-horizon-py27 http://logs.openstack.org/55/655955/1/check/cross-horizon-py27/0147a1e/ : SUCCESS in 10m 59s\n- cross-horizon-py36 http://logs.openstack.org/55/655955/1/check/cross-horizon-py36/1933595/ : SUCCESS in 7m 46s\n- cross-horizon-npm http://logs.openstack.org/55/655955/1/check/cross-horizon-npm/9c6ed11/ : SUCCESS in 6m 09s\n- cross-keystone-py27 http://logs.openstack.org/55/655955/1/check/cross-keystone-py27/5282763/ : SUCCESS in 20m 38s\n- cross-keystone-py36 http://logs.openstack.org/55/655955/1/check/cross-keystone-py36/e4b3e54/ : SUCCESS in 19m 54s\n- cross-neutron-py27 http://logs.openstack.org/55/655955/1/check/cross-neutron-py27/74a22bb/ : SUCCESS in 21m 23s\n- cross-neutron-py36 http://logs.openstack.org/55/655955/1/check/cross-neutron-py36/e1efd2d/ : SUCCESS in 21m 27s\n- cross-nova-py27 http://logs.openstack.org/55/655955/1/check/cross-nova-py27/e726c2b/ : SUCCESS in 20m 13s\n- cross-nova-py36 http://logs.openstack.org/55/655955/1/check/cross-nova-py36/14567fd/ : SUCCESS in 13m 56s\n- cross-nova-functional http://logs.openstack.org/55/655955/1/check/cross-nova-functional/631b019/ : SUCCESS in 26m 21s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/55/655955/1/check/cross-osc-build-sphinx-docs/6d0e32c/html/ : SUCCESS in 4m 51s\n- cross-swift-py27 http://logs.openstack.org/55/655955/1/check/cross-swift-py27/1211d7a/ : SUCCESS in 8m 25s\n- cross-swift-py36 http://logs.openstack.org/55/655955/1/check/cross-swift-py36/7fbbb64/ : SUCCESS in 7m 14s\n- requirements-integration http://logs.openstack.org/55/655955/1/check/requirements-integration/34ffdf9/ : SUCCESS in 58m 47s\n- neutron-grenade-multinode http://logs.openstack.org/55/655955/1/check/neutron-grenade-multinode/920f222/ : SUCCESS in 1h 22m 00s\n- tempest-full http://logs.openstack.org/55/655955/1/check/tempest-full/56bdf70/ : SUCCESS in 1h 47m 02s\n- neutron-grenade http://logs.openstack.org/55/655955/1/check/neutron-grenade/19fd162/ : SUCCESS in 49m 16s","accounts_in_message":[],"_revision_number":1},{"id":"355b71d5af9b48df008d47fc333f4382b0b0341b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 04:47:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ccc94bad3ec5b5e4de266fa618b89b67a1dbe73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 06:37:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/655955/1/gate/openstack-tox-docs/dca9ee4/html/ : SUCCESS in 2m 48s\n- openstack-tox-pep8 http://logs.openstack.org/55/655955/1/gate/openstack-tox-pep8/3d2afbe/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/55/655955/1/gate/openstack-tox-py27/1353347/ : SUCCESS in 4m 37s\n- openstack-tox-py36 http://logs.openstack.org/55/655955/1/gate/openstack-tox-py36/a1eb2be/ : SUCCESS in 3m 10s\n- openstack-tox-validate http://logs.openstack.org/55/655955/1/gate/openstack-tox-validate/ca7f772/ : SUCCESS in 3m 01s\n- requirements-tox-babel http://logs.openstack.org/55/655955/1/gate/requirements-tox-babel/22d8451/ : SUCCESS in 2m 31s\n- requirements-tox-py27-check-uc http://logs.openstack.org/55/655955/1/gate/requirements-tox-py27-check-uc/de989b5/ : SUCCESS in 7m 12s\n- requirements-tox-py36-check-uc http://logs.openstack.org/55/655955/1/gate/requirements-tox-py36-check-uc/581bedf/ : SUCCESS in 7m 11s\n- cross-cinder-py27 http://logs.openstack.org/55/655955/1/gate/cross-cinder-py27/3a5a06e/ : SUCCESS in 9m 41s\n- cross-cinder-py36 http://logs.openstack.org/55/655955/1/gate/cross-cinder-py36/420f968/ : SUCCESS in 8m 54s\n- cross-glance-py27 http://logs.openstack.org/55/655955/1/gate/cross-glance-py27/b7cdf51/ : SUCCESS in 4m 27s\n- cross-glance-py36 http://logs.openstack.org/55/655955/1/gate/cross-glance-py36/c30b230/ : SUCCESS in 4m 20s\n- cross-horizon-py27 http://logs.openstack.org/55/655955/1/gate/cross-horizon-py27/5462c62/ : SUCCESS in 8m 56s\n- cross-horizon-py36 http://logs.openstack.org/55/655955/1/gate/cross-horizon-py36/62c476e/ : SUCCESS in 8m 34s\n- cross-horizon-npm http://logs.openstack.org/55/655955/1/gate/cross-horizon-npm/77f490e/ : SUCCESS in 5m 49s\n- cross-keystone-py27 http://logs.openstack.org/55/655955/1/gate/cross-keystone-py27/92019f6/ : SUCCESS in 20m 08s\n- cross-keystone-py36 http://logs.openstack.org/55/655955/1/gate/cross-keystone-py36/0e1efb1/ : SUCCESS in 21m 36s\n- cross-neutron-py27 http://logs.openstack.org/55/655955/1/gate/cross-neutron-py27/5853285/ : SUCCESS in 29m 48s\n- cross-neutron-py36 http://logs.openstack.org/55/655955/1/gate/cross-neutron-py36/15d7e45/ : SUCCESS in 22m 20s\n- cross-nova-py27 http://logs.openstack.org/55/655955/1/gate/cross-nova-py27/60d59e0/ : SUCCESS in 11m 12s\n- cross-nova-py36 http://logs.openstack.org/55/655955/1/gate/cross-nova-py36/373574b/ : SUCCESS in 15m 23s\n- cross-nova-functional http://logs.openstack.org/55/655955/1/gate/cross-nova-functional/cffc18a/ : SUCCESS in 19m 04s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/55/655955/1/gate/cross-osc-build-sphinx-docs/5aebdef/html/ : SUCCESS in 5m 47s\n- cross-swift-py27 http://logs.openstack.org/55/655955/1/gate/cross-swift-py27/28e7bea/ : SUCCESS in 8m 08s\n- cross-swift-py36 http://logs.openstack.org/55/655955/1/gate/cross-swift-py36/116d8ef/ : SUCCESS in 9m 07s\n- requirements-integration http://logs.openstack.org/55/655955/1/gate/requirements-integration/62ecee0/ : SUCCESS in 55m 18s\n- neutron-grenade-multinode http://logs.openstack.org/55/655955/1/gate/neutron-grenade-multinode/0c78535/ : SUCCESS in 1h 01m 42s\n- tempest-full http://logs.openstack.org/55/655955/1/gate/tempest-full/cee8d05/ : SUCCESS in 1h 45m 36s\n- neutron-grenade http://logs.openstack.org/55/655955/1/gate/neutron-grenade/7e21126/ : SUCCESS in 1h 10m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9eb40e90aca7f49c39ec88db682a6337fe54c0ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 06:37:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90edf80736ed0b8813d0e520e3a583ae3d5d64b8","revisions":{"90edf80736ed0b8813d0e520e3a583ae3d5d64b8":{"kind":"REWORK","_number":1,"created":"2019-04-26 16:55:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/655955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/655955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/655955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/655955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/655955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/655955/1"}}},"commit":{"parents":[{"commit":"c7c3edc2add91936abf98fbefb2e0ae9cb3bb63f","subject":"Merge \"update constraint for neutron-lib to new release 1.26.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7c3edc2add91936abf98fbefb2e0ae9cb3bb63f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 16:55:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 16:55:52.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 1.36.0","message":"update constraint for oslo.versionedobjects to new release 1.36.0\n\nChange-Id: I3b1bc3e496e3a1880e3cec140a9279aefd8e7b61\nmeta:version: 1.36.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I3b9441931e04056a848bab7dbeb80b70a343f1db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90edf80736ed0b8813d0e520e3a583ae3d5d64b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90edf80736ed0b8813d0e520e3a583ae3d5d64b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
