)]}'
{"id":"openstack%2Frequirements~655962","triplet_id":"openstack%2Frequirements~master~Ia825ce9644cb6bfe826dd0e65aecfa3713a95e1a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia825ce9644cb6bfe826dd0e65aecfa3713a95e1a","subject":"update constraint for oslo.config to new release 6.9.0","status":"MERGED","created":"2019-04-26 17:00:48.000000000","updated":"2019-04-27 19:14:39.000000000","submitted":"2019-04-27 19:14:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655962-1556392479511-ad6eae0d","meta_rev_id":"a6948dc9205f5771dd16d233a6b0c81c8e8b2266","_number":655962,"virtual_id_number":655962,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-27 19:14:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-27 13:40:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-04-27 13:40:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-27 13:40:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-27 19:14:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d010dfd3ee0d1e4a84d5fa5c31759d6f04ecd8b1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-26 17:00:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"669eb7837190066b96b80eed8f63cfa88dbdb0b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 23:07:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/62/655962/1/check/openstack-tox-docs/fd53c92/html/ : SUCCESS in 3m 04s\n- openstack-tox-pypy http://logs.openstack.org/62/655962/1/check/openstack-tox-pypy/1348dbd/ : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/62/655962/1/check/openstack-tox-pep8/fe6bad2/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/62/655962/1/check/openstack-tox-py27/f28c7d4/ : SUCCESS in 3m 08s\n- openstack-tox-py36 http://logs.openstack.org/62/655962/1/check/openstack-tox-py36/ec1c66b/ : SUCCESS in 4m 41s\n- openstack-tox-validate http://logs.openstack.org/62/655962/1/check/openstack-tox-validate/ff929fa/ : SUCCESS in 3m 07s\n- requirements-tox-babel http://logs.openstack.org/62/655962/1/check/requirements-tox-babel/d3ca06b/ : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/62/655962/1/check/requirements-tox-py27-check-uc/cbce624/ : SUCCESS in 7m 41s\n- requirements-tox-py36-check-uc http://logs.openstack.org/62/655962/1/check/requirements-tox-py36-check-uc/eedcc8c/ : SUCCESS in 8m 20s\n- cross-cinder-py27 http://logs.openstack.org/62/655962/1/check/cross-cinder-py27/f64016f/ : SUCCESS in 8m 28s\n- cross-cinder-py36 http://logs.openstack.org/62/655962/1/check/cross-cinder-py36/2868968/ : SUCCESS in 8m 19s\n- cross-glance-py27 http://logs.openstack.org/62/655962/1/check/cross-glance-py27/957f6c7/ : SUCCESS in 4m 51s\n- cross-glance-py36 http://logs.openstack.org/62/655962/1/check/cross-glance-py36/6420b38/ : SUCCESS in 4m 51s\n- cross-horizon-py27 http://logs.openstack.org/62/655962/1/check/cross-horizon-py27/2df0a72/ : SUCCESS in 11m 38s\n- cross-horizon-py36 http://logs.openstack.org/62/655962/1/check/cross-horizon-py36/80288ab/ : SUCCESS in 8m 55s\n- cross-horizon-npm http://logs.openstack.org/62/655962/1/check/cross-horizon-npm/c95ec76/ : SUCCESS in 8m 06s\n- cross-keystone-py27 http://logs.openstack.org/62/655962/1/check/cross-keystone-py27/62701b4/ : SUCCESS in 18m 40s\n- cross-keystone-py36 http://logs.openstack.org/62/655962/1/check/cross-keystone-py36/713477e/ : SUCCESS in 21m 07s\n- cross-neutron-py27 http://logs.openstack.org/62/655962/1/check/cross-neutron-py27/b056642/ : SUCCESS in 27m 13s\n- cross-neutron-py36 http://logs.openstack.org/62/655962/1/check/cross-neutron-py36/b41dc92/ : SUCCESS in 21m 27s\n- cross-nova-py27 http://logs.openstack.org/62/655962/1/check/cross-nova-py27/e9b6ee2/ : SUCCESS in 15m 04s\n- cross-nova-py36 http://logs.openstack.org/62/655962/1/check/cross-nova-py36/b6b7a9b/ : FAILURE in 12m 55s\n- cross-nova-functional http://logs.openstack.org/62/655962/1/check/cross-nova-functional/92808c3/ : SUCCESS in 20m 26s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/62/655962/1/check/cross-osc-build-sphinx-docs/9924c63/html/ : SUCCESS in 6m 23s\n- cross-swift-py27 http://logs.openstack.org/62/655962/1/check/cross-swift-py27/06dec7c/ : SUCCESS in 7m 49s\n- cross-swift-py36 http://logs.openstack.org/62/655962/1/check/cross-swift-py36/569db66/ : SUCCESS in 7m 29s\n- requirements-integration http://logs.openstack.org/62/655962/1/check/requirements-integration/4ccf104/ : SUCCESS in 56m 59s\n- neutron-grenade-multinode http://logs.openstack.org/62/655962/1/check/neutron-grenade-multinode/ddd0ae0/ : SUCCESS in 1h 26m 05s\n- tempest-full http://logs.openstack.org/62/655962/1/check/tempest-full/94b8751/ : SUCCESS in 1h 25m 20s\n- neutron-grenade http://logs.openstack.org/62/655962/1/check/neutron-grenade/eb527d1/ : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a3995be19176aa5489f0c65ae587956913a943da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-27 04:34:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"64a7ba34fed9b809b21a69073b6d7547399b52ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 06:45:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/62/655962/1/check/openstack-tox-docs/6d86681/html/ : SUCCESS in 3m 19s\n- openstack-tox-pypy http://logs.openstack.org/62/655962/1/check/openstack-tox-pypy/e8aa548/ : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/62/655962/1/check/openstack-tox-pep8/bdee4a3/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/62/655962/1/check/openstack-tox-py27/478ccec/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/62/655962/1/check/openstack-tox-py36/938daa4/ : SUCCESS in 3m 13s\n- openstack-tox-validate http://logs.openstack.org/62/655962/1/check/openstack-tox-validate/9daab80/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/62/655962/1/check/requirements-tox-babel/b4fea20/ : SUCCESS in 2m 54s\n- requirements-tox-py27-check-uc http://logs.openstack.org/62/655962/1/check/requirements-tox-py27-check-uc/64d2b0c/ : SUCCESS in 6m 48s\n- requirements-tox-py36-check-uc http://logs.openstack.org/62/655962/1/check/requirements-tox-py36-check-uc/4faec1c/ : SUCCESS in 7m 03s\n- cross-cinder-py27 http://logs.openstack.org/62/655962/1/check/cross-cinder-py27/34c6c2d/ : SUCCESS in 9m 47s\n- cross-cinder-py36 http://logs.openstack.org/62/655962/1/check/cross-cinder-py36/b3aed69/ : SUCCESS in 8m 01s\n- cross-glance-py27 http://logs.openstack.org/62/655962/1/check/cross-glance-py27/cee8dda/ : SUCCESS in 4m 08s\n- cross-glance-py36 http://logs.openstack.org/62/655962/1/check/cross-glance-py36/cbc0b11/ : SUCCESS in 4m 16s\n- cross-horizon-py27 http://logs.openstack.org/62/655962/1/check/cross-horizon-py27/784b051/ : SUCCESS in 10m 19s\n- cross-horizon-py36 http://logs.openstack.org/62/655962/1/check/cross-horizon-py36/3f49679/ : SUCCESS in 9m 12s\n- cross-horizon-npm http://logs.openstack.org/62/655962/1/check/cross-horizon-npm/35e76a2/ : SUCCESS in 7m 35s\n- cross-keystone-py27 http://logs.openstack.org/62/655962/1/check/cross-keystone-py27/f948518/ : SUCCESS in 19m 25s\n- cross-keystone-py36 http://logs.openstack.org/62/655962/1/check/cross-keystone-py36/def6bb3/ : SUCCESS in 20m 41s\n- cross-neutron-py27 http://logs.openstack.org/62/655962/1/check/cross-neutron-py27/0d20f46/ : SUCCESS in 22m 51s\n- cross-neutron-py36 http://logs.openstack.org/62/655962/1/check/cross-neutron-py36/c6e1158/ : SUCCESS in 21m 57s\n- cross-nova-py27 http://logs.openstack.org/62/655962/1/check/cross-nova-py27/6eb1e55/ : SUCCESS in 13m 10s\n- cross-nova-py36 http://logs.openstack.org/62/655962/1/check/cross-nova-py36/dcd191e/ : SUCCESS in 12m 00s\n- cross-nova-functional http://logs.openstack.org/62/655962/1/check/cross-nova-functional/8413544/ : SUCCESS in 20m 08s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/62/655962/1/check/cross-osc-build-sphinx-docs/e556ada/html/ : SUCCESS in 5m 15s\n- cross-swift-py27 http://logs.openstack.org/62/655962/1/check/cross-swift-py27/48ba4c5/ : SUCCESS in 7m 34s\n- cross-swift-py36 http://logs.openstack.org/62/655962/1/check/cross-swift-py36/caef0ef/ : SUCCESS in 6m 55s\n- requirements-integration http://logs.openstack.org/62/655962/1/check/requirements-integration/89a8f87/ : SUCCESS in 57m 21s\n- neutron-grenade-multinode http://logs.openstack.org/62/655962/1/check/neutron-grenade-multinode/97036e5/ : SUCCESS in 1h 21m 05s\n- tempest-full http://logs.openstack.org/62/655962/1/check/tempest-full/5006c4f/ : SUCCESS in 1h 36m 30s\n- neutron-grenade http://logs.openstack.org/62/655962/1/check/neutron-grenade/1861af4/ : SUCCESS in 54m 56s","accounts_in_message":[],"_revision_number":1},{"id":"31c455f112aaa09754fb7b530f39aad8250a619a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-27 13:40:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8519f6131768d79573c91f6932aa9907f8571e86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 13:40:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a753167969a33929f254216c343f56ac0be979f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 19:14:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/62/655962/1/gate/openstack-tox-docs/4647013/html/ : SUCCESS in 3m 08s\n- openstack-tox-pep8 http://logs.openstack.org/62/655962/1/gate/openstack-tox-pep8/e2b1803/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/62/655962/1/gate/openstack-tox-py27/6bbea24/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/62/655962/1/gate/openstack-tox-py36/b56d11c/ : SUCCESS in 4m 00s\n- openstack-tox-validate http://logs.openstack.org/62/655962/1/gate/openstack-tox-validate/9c63507/ : SUCCESS in 3m 22s\n- requirements-tox-babel http://logs.openstack.org/62/655962/1/gate/requirements-tox-babel/0b4e9b5/ : SUCCESS in 3m 26s\n- requirements-tox-py27-check-uc http://logs.openstack.org/62/655962/1/gate/requirements-tox-py27-check-uc/416055b/ : SUCCESS in 9m 35s\n- requirements-tox-py36-check-uc http://logs.openstack.org/62/655962/1/gate/requirements-tox-py36-check-uc/a9f0d43/ : SUCCESS in 9m 31s\n- cross-cinder-py27 http://logs.openstack.org/62/655962/1/gate/cross-cinder-py27/6af6318/ : SUCCESS in 12m 01s\n- cross-cinder-py36 http://logs.openstack.org/62/655962/1/gate/cross-cinder-py36/a6d6bdb/ : SUCCESS in 9m 47s\n- cross-glance-py27 http://logs.openstack.org/62/655962/1/gate/cross-glance-py27/130679a/ : SUCCESS in 4m 16s\n- cross-glance-py36 http://logs.openstack.org/62/655962/1/gate/cross-glance-py36/0cf24bb/ : SUCCESS in 6m 06s\n- cross-horizon-py27 http://logs.openstack.org/62/655962/1/gate/cross-horizon-py27/45dea55/ : SUCCESS in 10m 39s\n- cross-horizon-py36 http://logs.openstack.org/62/655962/1/gate/cross-horizon-py36/5839e5e/ : SUCCESS in 9m 23s\n- cross-horizon-npm http://logs.openstack.org/62/655962/1/gate/cross-horizon-npm/0ca20a4/ : SUCCESS in 6m 12s\n- cross-keystone-py27 http://logs.openstack.org/62/655962/1/gate/cross-keystone-py27/f6911a0/ : SUCCESS in 27m 33s\n- cross-keystone-py36 http://logs.openstack.org/62/655962/1/gate/cross-keystone-py36/17837e3/ : SUCCESS in 24m 09s\n- cross-neutron-py27 http://logs.openstack.org/62/655962/1/gate/cross-neutron-py27/37a6bd3/ : SUCCESS in 28m 59s\n- cross-neutron-py36 http://logs.openstack.org/62/655962/1/gate/cross-neutron-py36/bf79b05/ : SUCCESS in 27m 15s\n- cross-nova-py27 http://logs.openstack.org/62/655962/1/gate/cross-nova-py27/5d1890b/ : SUCCESS in 12m 45s\n- cross-nova-py36 http://logs.openstack.org/62/655962/1/gate/cross-nova-py36/11c1f7c/ : SUCCESS in 14m 21s\n- cross-nova-functional http://logs.openstack.org/62/655962/1/gate/cross-nova-functional/c8465d3/ : SUCCESS in 18m 20s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/62/655962/1/gate/cross-osc-build-sphinx-docs/1738d0a/html/ : SUCCESS in 5m 40s\n- cross-swift-py27 http://logs.openstack.org/62/655962/1/gate/cross-swift-py27/f554fa5/ : SUCCESS in 8m 01s\n- cross-swift-py36 http://logs.openstack.org/62/655962/1/gate/cross-swift-py36/858c400/ : SUCCESS in 11m 50s\n- requirements-integration http://logs.openstack.org/62/655962/1/gate/requirements-integration/7f98fb5/ : SUCCESS in 48m 30s\n- neutron-grenade-multinode http://logs.openstack.org/62/655962/1/gate/neutron-grenade-multinode/2d8eda4/ : SUCCESS in 1h 07m 09s\n- tempest-full http://logs.openstack.org/62/655962/1/gate/tempest-full/f0c6159/ : SUCCESS in 1h 50m 52s\n- neutron-grenade http://logs.openstack.org/62/655962/1/gate/neutron-grenade/9eaa342/ : SUCCESS in 1h 11m 11s","accounts_in_message":[],"_revision_number":1},{"id":"cc6a480d6a17a22bc92985c4518b683f60f53182","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 19:14:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2dfa82e73ed6eccb081ed20aaba1acb1e8b577d2","revisions":{"2dfa82e73ed6eccb081ed20aaba1acb1e8b577d2":{"kind":"REWORK","_number":1,"created":"2019-04-26 17:00:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/655962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/655962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/655962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/655962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/655962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/655962/1"}}},"commit":{"parents":[{"commit":"c7c3edc2add91936abf98fbefb2e0ae9cb3bb63f","subject":"Merge \"update constraint for neutron-lib to new release 1.26.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7c3edc2add91936abf98fbefb2e0ae9cb3bb63f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 17:00:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 17:00:46.000000000","tz":0},"subject":"update constraint for oslo.config to new release 6.9.0","message":"update constraint for oslo.config to new release 6.9.0\n\nChange-Id: Ia825ce9644cb6bfe826dd0e65aecfa3713a95e1a\nmeta:version: 6.9.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I3b9441931e04056a848bab7dbeb80b70a343f1db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2dfa82e73ed6eccb081ed20aaba1acb1e8b577d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2dfa82e73ed6eccb081ed20aaba1acb1e8b577d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
