)]}'
{"id":"openstack%2Frequirements~655965","triplet_id":"openstack%2Frequirements~master~I7434ba9f2be2fe1925fdd90edda296e94131b941","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7434ba9f2be2fe1925fdd90edda296e94131b941","subject":"update constraint for oslo.db to new release 4.46.0","status":"MERGED","created":"2019-04-26 17:04:28.000000000","updated":"2019-04-28 02:52:22.000000000","submitted":"2019-04-28 02:52:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655965-1556419942957-fe66b0bc","meta_rev_id":"12f1fe25c6158bb74853447ed9452a7e9a8df6ea","_number":655965,"virtual_id_number":655965,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-04-28 02:52:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-27 04:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-04-27 19:02:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-04-27 04:34:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-27 04:34:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-27 19:02:37.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-04-28 02:52:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72aeb2003d114be28664c11478b97427e8e6dba6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-26 17:04:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f53fc3c268bac1d406ea3b68382286370fefe1cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 23:36:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/65/655965/1/check/openstack-tox-docs/82a11bb/html/ : SUCCESS in 4m 01s\n- openstack-tox-pypy http://logs.openstack.org/65/655965/1/check/openstack-tox-pypy/9275076/ : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/65/655965/1/check/openstack-tox-pep8/a240b2f/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/65/655965/1/check/openstack-tox-py27/a73d7a4/ : SUCCESS in 3m 20s\n- openstack-tox-py36 http://logs.openstack.org/65/655965/1/check/openstack-tox-py36/3fb6123/ : SUCCESS in 4m 25s\n- openstack-tox-validate http://logs.openstack.org/65/655965/1/check/openstack-tox-validate/2ef9a09/ : SUCCESS in 4m 01s\n- requirements-tox-babel http://logs.openstack.org/65/655965/1/check/requirements-tox-babel/8d70f26/ : SUCCESS in 4m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/655965/1/check/requirements-tox-py27-check-uc/8ca9002/ : SUCCESS in 7m 12s\n- requirements-tox-py36-check-uc http://logs.openstack.org/65/655965/1/check/requirements-tox-py36-check-uc/930df74/ : SUCCESS in 7m 33s\n- cross-cinder-py27 http://logs.openstack.org/65/655965/1/check/cross-cinder-py27/3d84f71/ : SUCCESS in 11m 05s\n- cross-cinder-py36 http://logs.openstack.org/65/655965/1/check/cross-cinder-py36/54d2e1c/ : SUCCESS in 9m 31s\n- cross-glance-py27 http://logs.openstack.org/65/655965/1/check/cross-glance-py27/62446d4/ : SUCCESS in 4m 14s\n- cross-glance-py36 http://logs.openstack.org/65/655965/1/check/cross-glance-py36/d0d1ac9/ : SUCCESS in 4m 47s\n- cross-horizon-py27 http://logs.openstack.org/65/655965/1/check/cross-horizon-py27/ca35a43/ : SUCCESS in 10m 51s\n- cross-horizon-py36 http://logs.openstack.org/65/655965/1/check/cross-horizon-py36/cc96eab/ : SUCCESS in 8m 41s\n- cross-horizon-npm http://logs.openstack.org/65/655965/1/check/cross-horizon-npm/5690110/ : SUCCESS in 5m 45s\n- cross-keystone-py27 http://logs.openstack.org/65/655965/1/check/cross-keystone-py27/bfe600f/ : SUCCESS in 29m 25s\n- cross-keystone-py36 http://logs.openstack.org/65/655965/1/check/cross-keystone-py36/a5c2838/ : SUCCESS in 22m 14s\n- cross-neutron-py27 http://logs.openstack.org/65/655965/1/check/cross-neutron-py27/282b2d2/ : SUCCESS in 21m 17s\n- cross-neutron-py36 http://logs.openstack.org/65/655965/1/check/cross-neutron-py36/c1eb9b6/ : SUCCESS in 20m 11s\n- cross-nova-py27 http://logs.openstack.org/65/655965/1/check/cross-nova-py27/5906898/ : SUCCESS in 14m 40s\n- cross-nova-py36 http://logs.openstack.org/65/655965/1/check/cross-nova-py36/44afb6a/ : SUCCESS in 12m 23s\n- cross-nova-functional http://logs.openstack.org/65/655965/1/check/cross-nova-functional/e5a747c/ : SUCCESS in 19m 37s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/65/655965/1/check/cross-osc-build-sphinx-docs/673212e/html/ : SUCCESS in 5m 29s\n- cross-swift-py27 http://logs.openstack.org/65/655965/1/check/cross-swift-py27/17b9437/ : SUCCESS in 8m 36s\n- cross-swift-py36 http://logs.openstack.org/65/655965/1/check/cross-swift-py36/ce0e211/ : SUCCESS in 7m 28s\n- requirements-integration http://logs.openstack.org/65/655965/1/check/requirements-integration/b4a6a7c/ : SUCCESS in 1h 01m 35s\n- neutron-grenade-multinode http://logs.openstack.org/65/655965/1/check/neutron-grenade-multinode/d37d9ce/ : SUCCESS in 1h 12m 34s\n- tempest-full http://logs.openstack.org/65/655965/1/check/tempest-full/523987a/ : SUCCESS in 1h 45m 57s\n- neutron-grenade http://logs.openstack.org/65/655965/1/check/neutron-grenade/8d0b7f8/ : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0d44e37969ec777671d4509fc7527a3520fa4fd3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-27 04:34:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"17d4c0a3bec827b13e85a78d7a9982c86746597d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 06:37:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"026c49fa8f812563b426e911f83f22a2d70ea68a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 13:22:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/65/655965/1/gate/openstack-tox-docs/7208a7e/html/ : SUCCESS in 5m 06s\n- openstack-tox-pep8 http://logs.openstack.org/65/655965/1/gate/openstack-tox-pep8/9ce0813/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/65/655965/1/gate/openstack-tox-py27/01a6f76/ : SUCCESS in 3m 42s\n- openstack-tox-py36 http://logs.openstack.org/65/655965/1/gate/openstack-tox-py36/b9654f2/ : SUCCESS in 4m 20s\n- openstack-tox-validate http://logs.openstack.org/65/655965/1/gate/openstack-tox-validate/c5a14e8/ : SUCCESS in 3m 24s\n- requirements-tox-babel http://logs.openstack.org/65/655965/1/gate/requirements-tox-babel/83ad7d1/ : SUCCESS in 3m 18s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/655965/1/gate/requirements-tox-py27-check-uc/489ca04/ : SUCCESS in 10m 07s\n- requirements-tox-py36-check-uc http://logs.openstack.org/65/655965/1/gate/requirements-tox-py36-check-uc/7b0dd17/ : SUCCESS in 7m 43s\n- cross-cinder-py27 http://logs.openstack.org/65/655965/1/gate/cross-cinder-py27/ef99096/ : SUCCESS in 10m 00s\n- cross-cinder-py36 http://logs.openstack.org/65/655965/1/gate/cross-cinder-py36/b01df5e/ : SUCCESS in 8m 59s\n- cross-glance-py27 http://logs.openstack.org/65/655965/1/gate/cross-glance-py27/c79a6fc/ : SUCCESS in 4m 48s\n- cross-glance-py36 http://logs.openstack.org/65/655965/1/gate/cross-glance-py36/f6ab999/ : SUCCESS in 4m 34s\n- cross-horizon-py27 http://logs.openstack.org/65/655965/1/gate/cross-horizon-py27/321cc34/ : SUCCESS in 10m 02s\n- cross-horizon-py36 http://logs.openstack.org/65/655965/1/gate/cross-horizon-py36/298aaed/ : SUCCESS in 7m 55s\n- cross-horizon-npm http://logs.openstack.org/65/655965/1/gate/cross-horizon-npm/29baaa8/ : SUCCESS in 6m 27s\n- cross-keystone-py27 http://logs.openstack.org/65/655965/1/gate/cross-keystone-py27/221aded/ : SUCCESS in 23m 59s\n- cross-keystone-py36 http://logs.openstack.org/65/655965/1/gate/cross-keystone-py36/019bf5a/ : SUCCESS in 24m 45s\n- cross-neutron-py27 http://logs.openstack.org/65/655965/1/gate/cross-neutron-py27/dbe82f4/ : SUCCESS in 19m 50s\n- cross-neutron-py36 http://logs.openstack.org/65/655965/1/gate/cross-neutron-py36/3870bcd/ : SUCCESS in 19m 21s\n- cross-nova-py27 http://logs.openstack.org/65/655965/1/gate/cross-nova-py27/b1b6e64/ : SUCCESS in 14m 42s\n- cross-nova-py36 http://logs.openstack.org/65/655965/1/gate/cross-nova-py36/07fc408/ : FAILURE in 12m 20s\n- cross-nova-functional http://logs.openstack.org/65/655965/1/gate/cross-nova-functional/8007ff3/ : SUCCESS in 21m 04s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/65/655965/1/gate/cross-osc-build-sphinx-docs/a907a31/html/ : SUCCESS in 6m 53s\n- cross-swift-py27 http://logs.openstack.org/65/655965/1/gate/cross-swift-py27/8fdd1b2/ : SUCCESS in 8m 29s\n- cross-swift-py36 http://logs.openstack.org/65/655965/1/gate/cross-swift-py36/b0486bc/ : SUCCESS in 7m 43s\n- requirements-integration http://logs.openstack.org/65/655965/1/gate/requirements-integration/c654fb9/ : SUCCESS in 56m 18s\n- neutron-grenade-multinode http://logs.openstack.org/65/655965/1/gate/neutron-grenade-multinode/3c36083/ : SUCCESS in 59m 26s\n- tempest-full http://logs.openstack.org/65/655965/1/gate/tempest-full/04622c5/ : FAILURE in 29m 52s\n- neutron-grenade http://logs.openstack.org/65/655965/1/gate/neutron-grenade/bf8fd97/ : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":1},{"id":"78affa78bda83118470f65a731e0837cb02e66a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-27 13:41:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a8d1a69c4169f8cbaad228a84ba0bc44e0c8d513","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 15:24:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/65/655965/1/check/openstack-tox-docs/cd02891/html/ : SUCCESS in 3m 34s\n- openstack-tox-pypy http://logs.openstack.org/65/655965/1/check/openstack-tox-pypy/aeda3cc/ : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/65/655965/1/check/openstack-tox-pep8/c7c6093/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/65/655965/1/check/openstack-tox-py27/d38d044/ : SUCCESS in 4m 04s\n- openstack-tox-py36 http://logs.openstack.org/65/655965/1/check/openstack-tox-py36/643be47/ : SUCCESS in 3m 24s\n- openstack-tox-validate http://logs.openstack.org/65/655965/1/check/openstack-tox-validate/b197709/ : SUCCESS in 3m 11s\n- requirements-tox-babel http://logs.openstack.org/65/655965/1/check/requirements-tox-babel/4b42b32/ : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/655965/1/check/requirements-tox-py27-check-uc/af1a6f4/ : SUCCESS in 8m 08s\n- requirements-tox-py36-check-uc http://logs.openstack.org/65/655965/1/check/requirements-tox-py36-check-uc/d0f3db1/ : SUCCESS in 8m 43s\n- cross-cinder-py27 http://logs.openstack.org/65/655965/1/check/cross-cinder-py27/5cbfe8c/ : SUCCESS in 10m 40s\n- cross-cinder-py36 http://logs.openstack.org/65/655965/1/check/cross-cinder-py36/255c595/ : SUCCESS in 8m 17s\n- cross-glance-py27 http://logs.openstack.org/65/655965/1/check/cross-glance-py27/b3eaee3/ : SUCCESS in 4m 09s\n- cross-glance-py36 http://logs.openstack.org/65/655965/1/check/cross-glance-py36/f0c3d6a/ : SUCCESS in 6m 42s\n- cross-horizon-py27 http://logs.openstack.org/65/655965/1/check/cross-horizon-py27/ffa8477/ : SUCCESS in 10m 25s\n- cross-horizon-py36 http://logs.openstack.org/65/655965/1/check/cross-horizon-py36/d780e41/ : SUCCESS in 8m 21s\n- cross-horizon-npm http://logs.openstack.org/65/655965/1/check/cross-horizon-npm/eaa5019/ : SUCCESS in 5m 54s\n- cross-keystone-py27 http://logs.openstack.org/65/655965/1/check/cross-keystone-py27/43aa2f4/ : SUCCESS in 21m 36s\n- cross-keystone-py36 http://logs.openstack.org/65/655965/1/check/cross-keystone-py36/ccf95b7/ : SUCCESS in 22m 29s\n- cross-neutron-py27 http://logs.openstack.org/65/655965/1/check/cross-neutron-py27/c2845ff/ : SUCCESS in 23m 17s\n- cross-neutron-py36 http://logs.openstack.org/65/655965/1/check/cross-neutron-py36/3e918fe/ : SUCCESS in 21m 16s\n- cross-nova-py27 http://logs.openstack.org/65/655965/1/check/cross-nova-py27/b8d0008/ : SUCCESS in 13m 02s\n- cross-nova-py36 http://logs.openstack.org/65/655965/1/check/cross-nova-py36/11c2c62/ : FAILURE in 12m 30s\n- cross-nova-functional http://logs.openstack.org/65/655965/1/check/cross-nova-functional/aaf239e/ : SUCCESS in 20m 29s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/65/655965/1/check/cross-osc-build-sphinx-docs/5191082/html/ : SUCCESS in 5m 43s\n- cross-swift-py27 http://logs.openstack.org/65/655965/1/check/cross-swift-py27/1e3d763/ : SUCCESS in 8m 28s\n- cross-swift-py36 http://logs.openstack.org/65/655965/1/check/cross-swift-py36/c5b89ac/ : SUCCESS in 7m 21s\n- requirements-integration http://logs.openstack.org/65/655965/1/check/requirements-integration/46240f8/ : SUCCESS in 53m 43s\n- neutron-grenade-multinode http://logs.openstack.org/65/655965/1/check/neutron-grenade-multinode/5736139/ : SUCCESS in 1h 12m 14s\n- tempest-full http://logs.openstack.org/65/655965/1/check/tempest-full/e51e469/ : SUCCESS in 1h 37m 06s\n- neutron-grenade http://logs.openstack.org/65/655965/1/check/neutron-grenade/83f3522/ : SUCCESS in 57m 08s","accounts_in_message":[],"_revision_number":1},{"id":"cebd22da246e536bcc785ecc32744ce8f1b8a601","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-27 19:02:37.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9cf6302379754396411422e44169ba41b6a156f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 21:00:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/65/655965/1/check/openstack-tox-docs/cac369f/html/ : SUCCESS in 2m 59s\n- openstack-tox-pypy http://logs.openstack.org/65/655965/1/check/openstack-tox-pypy/fc48bb9/ : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/65/655965/1/check/openstack-tox-pep8/0f7110f/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/65/655965/1/check/openstack-tox-py27/939cb2e/ : SUCCESS in 3m 02s\n- openstack-tox-py36 http://logs.openstack.org/65/655965/1/check/openstack-tox-py36/111eb90/ : SUCCESS in 3m 19s\n- openstack-tox-validate http://logs.openstack.org/65/655965/1/check/openstack-tox-validate/1f3be5e/ : SUCCESS in 2m 51s\n- requirements-tox-babel http://logs.openstack.org/65/655965/1/check/requirements-tox-babel/b80e319/ : SUCCESS in 2m 42s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/655965/1/check/requirements-tox-py27-check-uc/5511465/ : SUCCESS in 6m 52s\n- requirements-tox-py36-check-uc http://logs.openstack.org/65/655965/1/check/requirements-tox-py36-check-uc/693db9d/ : SUCCESS in 7m 04s\n- cross-cinder-py27 http://logs.openstack.org/65/655965/1/check/cross-cinder-py27/510c814/ : SUCCESS in 7m 55s\n- cross-cinder-py36 http://logs.openstack.org/65/655965/1/check/cross-cinder-py36/1ff02a2/ : SUCCESS in 9m 14s\n- cross-glance-py27 http://logs.openstack.org/65/655965/1/check/cross-glance-py27/58de6a5/ : SUCCESS in 4m 36s\n- cross-glance-py36 http://logs.openstack.org/65/655965/1/check/cross-glance-py36/747a115/ : SUCCESS in 4m 17s\n- cross-horizon-py27 http://logs.openstack.org/65/655965/1/check/cross-horizon-py27/67b80f7/ : SUCCESS in 10m 18s\n- cross-horizon-py36 http://logs.openstack.org/65/655965/1/check/cross-horizon-py36/0dfd28d/ : SUCCESS in 9m 21s\n- cross-horizon-npm http://logs.openstack.org/65/655965/1/check/cross-horizon-npm/dd7e0e4/ : SUCCESS in 7m 40s\n- cross-keystone-py27 http://logs.openstack.org/65/655965/1/check/cross-keystone-py27/35c1dd4/ : SUCCESS in 21m 52s\n- cross-keystone-py36 http://logs.openstack.org/65/655965/1/check/cross-keystone-py36/510700b/ : SUCCESS in 27m 12s\n- cross-neutron-py27 http://logs.openstack.org/65/655965/1/check/cross-neutron-py27/06c7c3c/ : SUCCESS in 22m 49s\n- cross-neutron-py36 http://logs.openstack.org/65/655965/1/check/cross-neutron-py36/94722d1/ : SUCCESS in 30m 36s\n- cross-nova-py27 http://logs.openstack.org/65/655965/1/check/cross-nova-py27/bad48f2/ : SUCCESS in 12m 40s\n- cross-nova-py36 http://logs.openstack.org/65/655965/1/check/cross-nova-py36/c50f003/ : SUCCESS in 11m 42s\n- cross-nova-functional http://logs.openstack.org/65/655965/1/check/cross-nova-functional/44718fd/ : SUCCESS in 18m 56s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/65/655965/1/check/cross-osc-build-sphinx-docs/fd3ddc5/html/ : SUCCESS in 5m 13s\n- cross-swift-py27 http://logs.openstack.org/65/655965/1/check/cross-swift-py27/6428a7f/ : SUCCESS in 8m 19s\n- cross-swift-py36 http://logs.openstack.org/65/655965/1/check/cross-swift-py36/11c6e38/ : SUCCESS in 6m 44s\n- requirements-integration http://logs.openstack.org/65/655965/1/check/requirements-integration/f10bee1/ : SUCCESS in 1h 03m 53s\n- neutron-grenade-multinode http://logs.openstack.org/65/655965/1/check/neutron-grenade-multinode/0c734a4/ : SUCCESS in 1h 12m 00s\n- tempest-full http://logs.openstack.org/65/655965/1/check/tempest-full/2b689e8/ : SUCCESS in 1h 51m 22s\n- neutron-grenade http://logs.openstack.org/65/655965/1/check/neutron-grenade/61b0589/ : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4c8912c0d88313da995c0abbb395b157a1c219c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 21:00:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbb88bf7c62f8e967cc64a83db8bdd7ecf8a7170","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 02:52:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/65/655965/1/gate/openstack-tox-docs/9b56d3a/html/ : SUCCESS in 3m 05s\n- openstack-tox-pep8 http://logs.openstack.org/65/655965/1/gate/openstack-tox-pep8/358637a/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/65/655965/1/gate/openstack-tox-py27/bc6fdc5/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/65/655965/1/gate/openstack-tox-py36/b4aa9e6/ : SUCCESS in 4m 20s\n- openstack-tox-validate http://logs.openstack.org/65/655965/1/gate/openstack-tox-validate/89e02e2/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/65/655965/1/gate/requirements-tox-babel/96e485f/ : SUCCESS in 2m 43s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/655965/1/gate/requirements-tox-py27-check-uc/d5b4268/ : SUCCESS in 6m 59s\n- requirements-tox-py36-check-uc http://logs.openstack.org/65/655965/1/gate/requirements-tox-py36-check-uc/838f968/ : SUCCESS in 8m 13s\n- cross-cinder-py27 http://logs.openstack.org/65/655965/1/gate/cross-cinder-py27/4829585/ : SUCCESS in 8m 57s\n- cross-cinder-py36 http://logs.openstack.org/65/655965/1/gate/cross-cinder-py36/2c093b6/ : SUCCESS in 8m 50s\n- cross-glance-py27 http://logs.openstack.org/65/655965/1/gate/cross-glance-py27/6d0b61f/ : SUCCESS in 6m 10s\n- cross-glance-py36 http://logs.openstack.org/65/655965/1/gate/cross-glance-py36/cc9498f/ : SUCCESS in 5m 20s\n- cross-horizon-py27 http://logs.openstack.org/65/655965/1/gate/cross-horizon-py27/b106a4f/ : SUCCESS in 10m 14s\n- cross-horizon-py36 http://logs.openstack.org/65/655965/1/gate/cross-horizon-py36/7137f06/ : SUCCESS in 8m 58s\n- cross-horizon-npm http://logs.openstack.org/65/655965/1/gate/cross-horizon-npm/df1f474/ : SUCCESS in 7m 28s\n- cross-keystone-py27 http://logs.openstack.org/65/655965/1/gate/cross-keystone-py27/ca40a53/ : SUCCESS in 30m 53s\n- cross-keystone-py36 http://logs.openstack.org/65/655965/1/gate/cross-keystone-py36/c4d78a0/ : SUCCESS in 19m 45s\n- cross-neutron-py27 http://logs.openstack.org/65/655965/1/gate/cross-neutron-py27/ffb90f4/ : SUCCESS in 22m 01s\n- cross-neutron-py36 http://logs.openstack.org/65/655965/1/gate/cross-neutron-py36/42d9b7a/ : SUCCESS in 29m 06s\n- cross-nova-py27 http://logs.openstack.org/65/655965/1/gate/cross-nova-py27/0f303c6/ : SUCCESS in 12m 01s\n- cross-nova-py36 http://logs.openstack.org/65/655965/1/gate/cross-nova-py36/6751657/ : SUCCESS in 11m 05s\n- cross-nova-functional http://logs.openstack.org/65/655965/1/gate/cross-nova-functional/e678659/ : SUCCESS in 25m 05s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/65/655965/1/gate/cross-osc-build-sphinx-docs/5993ea3/html/ : SUCCESS in 5m 44s\n- cross-swift-py27 http://logs.openstack.org/65/655965/1/gate/cross-swift-py27/f64aa95/ : SUCCESS in 9m 23s\n- cross-swift-py36 http://logs.openstack.org/65/655965/1/gate/cross-swift-py36/3e43b57/ : SUCCESS in 8m 10s\n- requirements-integration http://logs.openstack.org/65/655965/1/gate/requirements-integration/7f06548/ : SUCCESS in 1h 02m 48s\n- neutron-grenade-multinode http://logs.openstack.org/65/655965/1/gate/neutron-grenade-multinode/a368b58/ : SUCCESS in 1h 08m 27s\n- tempest-full http://logs.openstack.org/65/655965/1/gate/tempest-full/e1227a7/ : SUCCESS in 1h 43m 35s\n- neutron-grenade http://logs.openstack.org/65/655965/1/gate/neutron-grenade/c35998b/ : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":1},{"id":"25174c6890a5568cd945ce38b443d38059070a0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 02:52:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6652283a80ea9ca4179474c5c51a369cb4556a9","revisions":{"e6652283a80ea9ca4179474c5c51a369cb4556a9":{"kind":"REWORK","_number":1,"created":"2019-04-26 17:04:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/655965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/655965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/655965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/655965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/655965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/655965/1"}}},"commit":{"parents":[{"commit":"c7c3edc2add91936abf98fbefb2e0ae9cb3bb63f","subject":"Merge \"update constraint for neutron-lib to new release 1.26.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7c3edc2add91936abf98fbefb2e0ae9cb3bb63f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 17:04:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 17:04:26.000000000","tz":0},"subject":"update constraint for oslo.db to new release 4.46.0","message":"update constraint for oslo.db to new release 4.46.0\n\nChange-Id: I7434ba9f2be2fe1925fdd90edda296e94131b941\nmeta:version: 4.46.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I3b9441931e04056a848bab7dbeb80b70a343f1db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6652283a80ea9ca4179474c5c51a369cb4556a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6652283a80ea9ca4179474c5c51a369cb4556a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
