)]}'
{"id":"openstack%2Frequirements~655967","triplet_id":"openstack%2Frequirements~master~I60a9d085a0fee9334bd0466ee8950f7d9c025498","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I60a9d085a0fee9334bd0466ee8950f7d9c025498","subject":"update constraint for oslo.middleware to new release 3.38.0","status":"MERGED","created":"2019-04-26 17:04:53.000000000","updated":"2019-04-28 11:53:30.000000000","submitted":"2019-04-28 11:53:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655967-1556452410037-332a2ac9","meta_rev_id":"8a3b9e6bce1e378348768bd8f6dd827377456f66","_number":655967,"virtual_id_number":655967,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-04-28 11:53:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-27 04:33:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-04-28 08:00:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-04-27 04:33:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-27 04:33:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-28 08:00:07.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-04-28 11:53:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebedc3244b5dfc3fdf83e6195d8ae4f465268424","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-26 17:04:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b5ae53a6d65f05d47348fe6deed7fe70566a748","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 23:41:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/655967/1/check/openstack-tox-docs/ac28ac8/html/ : SUCCESS in 2m 48s\n- openstack-tox-pypy http://logs.openstack.org/67/655967/1/check/openstack-tox-pypy/f73e140/ : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/655967/1/check/openstack-tox-pep8/de21080/ : SUCCESS in 2m 58s\n- openstack-tox-py27 http://logs.openstack.org/67/655967/1/check/openstack-tox-py27/43eb55c/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/67/655967/1/check/openstack-tox-py36/50bef14/ : SUCCESS in 3m 21s\n- openstack-tox-validate http://logs.openstack.org/67/655967/1/check/openstack-tox-validate/bb0f501/ : SUCCESS in 3m 56s\n- requirements-tox-babel http://logs.openstack.org/67/655967/1/check/requirements-tox-babel/82d8609/ : SUCCESS in 2m 36s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/655967/1/check/requirements-tox-py27-check-uc/144e052/ : SUCCESS in 10m 33s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/655967/1/check/requirements-tox-py36-check-uc/2ee162b/ : SUCCESS in 7m 44s\n- cross-cinder-py27 http://logs.openstack.org/67/655967/1/check/cross-cinder-py27/a78202b/ : SUCCESS in 11m 46s\n- cross-cinder-py36 http://logs.openstack.org/67/655967/1/check/cross-cinder-py36/b425089/ : SUCCESS in 8m 15s\n- cross-glance-py27 http://logs.openstack.org/67/655967/1/check/cross-glance-py27/22ee79d/ : SUCCESS in 4m 07s\n- cross-glance-py36 http://logs.openstack.org/67/655967/1/check/cross-glance-py36/26a4455/ : SUCCESS in 4m 35s\n- cross-horizon-py27 http://logs.openstack.org/67/655967/1/check/cross-horizon-py27/28e47c3/ : SUCCESS in 11m 18s\n- cross-horizon-py36 http://logs.openstack.org/67/655967/1/check/cross-horizon-py36/b8212d5/ : SUCCESS in 9m 22s\n- cross-horizon-npm http://logs.openstack.org/67/655967/1/check/cross-horizon-npm/a9c69e1/ : SUCCESS in 5m 47s\n- cross-keystone-py27 http://logs.openstack.org/67/655967/1/check/cross-keystone-py27/22fef1b/ : SUCCESS in 21m 52s\n- cross-keystone-py36 http://logs.openstack.org/67/655967/1/check/cross-keystone-py36/6251772/ : SUCCESS in 25m 29s\n- cross-neutron-py27 http://logs.openstack.org/67/655967/1/check/cross-neutron-py27/1e19e6a/ : SUCCESS in 26m 56s\n- cross-neutron-py36 http://logs.openstack.org/67/655967/1/check/cross-neutron-py36/a232516/ : SUCCESS in 22m 16s\n- cross-nova-py27 http://logs.openstack.org/67/655967/1/check/cross-nova-py27/3e3ae26/ : SUCCESS in 12m 00s\n- cross-nova-py36 http://logs.openstack.org/67/655967/1/check/cross-nova-py36/4fa53f5/ : SUCCESS in 15m 18s\n- cross-nova-functional http://logs.openstack.org/67/655967/1/check/cross-nova-functional/d36993d/ : SUCCESS in 23m 23s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/655967/1/check/cross-osc-build-sphinx-docs/9ca5e3f/html/ : SUCCESS in 5m 13s\n- cross-swift-py27 http://logs.openstack.org/67/655967/1/check/cross-swift-py27/277d3ce/ : SUCCESS in 7m 43s\n- cross-swift-py36 http://logs.openstack.org/67/655967/1/check/cross-swift-py36/c528a19/ : SUCCESS in 7m 23s\n- requirements-integration http://logs.openstack.org/67/655967/1/check/requirements-integration/4c61076/ : SUCCESS in 57m 59s\n- neutron-grenade-multinode http://logs.openstack.org/67/655967/1/check/neutron-grenade-multinode/0e61509/ : SUCCESS in 1h 11m 54s\n- tempest-full http://logs.openstack.org/67/655967/1/check/tempest-full/b5e2d27/ : SUCCESS in 1h 46m 37s\n- neutron-grenade http://logs.openstack.org/67/655967/1/check/neutron-grenade/a83f360/ : SUCCESS in 58m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c243d176cbddb69f518f54e6468eef123e16692b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-27 04:33:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"26ef4ff9252d9c24286433143478cd226fb30402","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 04:34:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7313f48ffe5af50a71962026abe72217820b94f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 13:16:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/67/655967/1/gate/openstack-tox-docs/c8ea947/html/ : SUCCESS in 5m 53s\n- openstack-tox-pep8 http://logs.openstack.org/67/655967/1/gate/openstack-tox-pep8/b34e332/ : SUCCESS in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/67/655967/1/gate/openstack-tox-py27/80996d7/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/67/655967/1/gate/openstack-tox-py36/95cd55f/ : SUCCESS in 3m 18s\n- openstack-tox-validate http://logs.openstack.org/67/655967/1/gate/openstack-tox-validate/3a36758/ : SUCCESS in 5m 42s\n- requirements-tox-babel http://logs.openstack.org/67/655967/1/gate/requirements-tox-babel/5dfc85d/ : SUCCESS in 2m 53s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/655967/1/gate/requirements-tox-py27-check-uc/0d06831/ : SUCCESS in 6m 49s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/655967/1/gate/requirements-tox-py36-check-uc/43cefae/ : SUCCESS in 8m 19s\n- cross-cinder-py27 http://logs.openstack.org/67/655967/1/gate/cross-cinder-py27/38ea420/ : SUCCESS in 8m 01s\n- cross-cinder-py36 http://logs.openstack.org/67/655967/1/gate/cross-cinder-py36/6b22177/ : SUCCESS in 8m 08s\n- cross-glance-py27 http://logs.openstack.org/67/655967/1/gate/cross-glance-py27/2b06cea/ : SUCCESS in 5m 19s\n- cross-glance-py36 http://logs.openstack.org/67/655967/1/gate/cross-glance-py36/ad64e92/ : SUCCESS in 5m 04s\n- cross-horizon-py27 http://logs.openstack.org/67/655967/1/gate/cross-horizon-py27/8aa298b/ : SUCCESS in 10m 43s\n- cross-horizon-py36 http://logs.openstack.org/67/655967/1/gate/cross-horizon-py36/53f3afd/ : SUCCESS in 7m 53s\n- cross-horizon-npm http://logs.openstack.org/67/655967/1/gate/cross-horizon-npm/c45f44b/ : SUCCESS in 6m 14s\n- cross-keystone-py27 http://logs.openstack.org/67/655967/1/gate/cross-keystone-py27/388356e/ : SUCCESS in 29m 57s\n- cross-keystone-py36 http://logs.openstack.org/67/655967/1/gate/cross-keystone-py36/f22f1ee/ : SUCCESS in 25m 24s\n- cross-neutron-py27 http://logs.openstack.org/67/655967/1/gate/cross-neutron-py27/931dab3/ : SUCCESS in 23m 03s\n- cross-neutron-py36 http://logs.openstack.org/67/655967/1/gate/cross-neutron-py36/afad4a4/ : SUCCESS in 22m 19s\n- cross-nova-py27 http://logs.openstack.org/67/655967/1/gate/cross-nova-py27/5500795/ : SUCCESS in 18m 21s\n- cross-nova-py36 http://logs.openstack.org/67/655967/1/gate/cross-nova-py36/906b608/ : SUCCESS in 12m 23s\n- cross-nova-functional http://logs.openstack.org/67/655967/1/gate/cross-nova-functional/653fb89/ : SUCCESS in 27m 54s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/655967/1/gate/cross-osc-build-sphinx-docs/07227f9/html/ : SUCCESS in 7m 25s\n- cross-swift-py27 http://logs.openstack.org/67/655967/1/gate/cross-swift-py27/05de7e8/ : SUCCESS in 7m 55s\n- cross-swift-py36 http://logs.openstack.org/67/655967/1/gate/cross-swift-py36/ee0af11/ : SUCCESS in 7m 11s\n- requirements-integration http://logs.openstack.org/67/655967/1/gate/requirements-integration/46bb78a/ : SUCCESS in 47m 19s\n- neutron-grenade-multinode http://logs.openstack.org/67/655967/1/gate/neutron-grenade-multinode/373df9d/ : FAILURE in 31m 05s\n- tempest-full http://logs.openstack.org/67/655967/1/gate/tempest-full/62747e7/ : SUCCESS in 1h 49m 44s\n- neutron-grenade http://logs.openstack.org/67/655967/1/gate/neutron-grenade/316adb0/ : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a1cfdf38fe75f9c37f7232b65a954b9f0b115a4a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-27 13:41:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9baf0ed70f9ddf95f916cd0cc5cc6207f619381f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 15:40:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/655967/1/check/openstack-tox-docs/4be5225/html/ : SUCCESS in 4m 29s\n- openstack-tox-pypy http://logs.openstack.org/67/655967/1/check/openstack-tox-pypy/ce9927a/ : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/655967/1/check/openstack-tox-pep8/6eda9e3/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/67/655967/1/check/openstack-tox-py27/e3b1c7d/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/67/655967/1/check/openstack-tox-py36/36bc843/ : SUCCESS in 3m 40s\n- openstack-tox-validate http://logs.openstack.org/67/655967/1/check/openstack-tox-validate/a3fb209/ : SUCCESS in 2m 53s\n- requirements-tox-babel http://logs.openstack.org/67/655967/1/check/requirements-tox-babel/36ed308/ : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/655967/1/check/requirements-tox-py27-check-uc/ba9629a/ : SUCCESS in 9m 35s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/655967/1/check/requirements-tox-py36-check-uc/816cf7f/ : SUCCESS in 6m 51s\n- cross-cinder-py27 http://logs.openstack.org/67/655967/1/check/cross-cinder-py27/6ff77fd/ : SUCCESS in 9m 53s\n- cross-cinder-py36 http://logs.openstack.org/67/655967/1/check/cross-cinder-py36/98eb9c9/ : SUCCESS in 10m 13s\n- cross-glance-py27 http://logs.openstack.org/67/655967/1/check/cross-glance-py27/f914582/ : SUCCESS in 4m 40s\n- cross-glance-py36 http://logs.openstack.org/67/655967/1/check/cross-glance-py36/dc959e9/ : SUCCESS in 4m 47s\n- cross-horizon-py27 http://logs.openstack.org/67/655967/1/check/cross-horizon-py27/94f0cfd/ : SUCCESS in 11m 53s\n- cross-horizon-py36 http://logs.openstack.org/67/655967/1/check/cross-horizon-py36/a78182f/ : SUCCESS in 9m 14s\n- cross-horizon-npm http://logs.openstack.org/67/655967/1/check/cross-horizon-npm/82ee9ea/ : SUCCESS in 6m 45s\n- cross-keystone-py27 http://logs.openstack.org/67/655967/1/check/cross-keystone-py27/73e9916/ : SUCCESS in 24m 48s\n- cross-keystone-py36 http://logs.openstack.org/67/655967/1/check/cross-keystone-py36/2a1a31e/ : SUCCESS in 25m 26s\n- cross-neutron-py27 http://logs.openstack.org/67/655967/1/check/cross-neutron-py27/7587422/ : SUCCESS in 23m 53s\n- cross-neutron-py36 http://logs.openstack.org/67/655967/1/check/cross-neutron-py36/7d4fd52/ : SUCCESS in 24m 49s\n- cross-nova-py27 http://logs.openstack.org/67/655967/1/check/cross-nova-py27/3e4a054/ : SUCCESS in 13m 45s\n- cross-nova-py36 http://logs.openstack.org/67/655967/1/check/cross-nova-py36/67399a5/ : SUCCESS in 12m 44s\n- cross-nova-functional http://logs.openstack.org/67/655967/1/check/cross-nova-functional/d28d491/ : SUCCESS in 19m 05s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/655967/1/check/cross-osc-build-sphinx-docs/ba954ee/html/ : SUCCESS in 7m 05s\n- cross-swift-py27 http://logs.openstack.org/67/655967/1/check/cross-swift-py27/fffe80b/ : SUCCESS in 8m 58s\n- cross-swift-py36 http://logs.openstack.org/67/655967/1/check/cross-swift-py36/3a0d212/ : SUCCESS in 7m 28s\n- requirements-integration http://logs.openstack.org/67/655967/1/check/requirements-integration/34f8240/ : SUCCESS in 56m 33s\n- neutron-grenade-multinode http://logs.openstack.org/67/655967/1/check/neutron-grenade-multinode/8d38ae1/ : SUCCESS in 1h 06m 15s\n- tempest-full http://logs.openstack.org/67/655967/1/check/tempest-full/de0d142/ : SUCCESS in 1h 51m 32s\n- neutron-grenade http://logs.openstack.org/67/655967/1/check/neutron-grenade/11a9a9c/ : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":1},{"id":"67fa87ec1bbae0ee5b51d42779f07a3b67a6eed0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 15:40:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e99da9e7cb6d6b498e41d0c1ff8affe2b2261ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 21:29:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/67/655967/1/gate/openstack-tox-docs/af8c3bb/html/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/67/655967/1/gate/openstack-tox-pep8/d74da81/ : SUCCESS in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/67/655967/1/gate/openstack-tox-py27/838df5e/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/67/655967/1/gate/openstack-tox-py36/7223d6f/ : SUCCESS in 6m 04s\n- openstack-tox-validate http://logs.openstack.org/67/655967/1/gate/openstack-tox-validate/5d420e3/ : SUCCESS in 4m 08s\n- requirements-tox-babel http://logs.openstack.org/67/655967/1/gate/requirements-tox-babel/8f3f72e/ : SUCCESS in 5m 20s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/655967/1/gate/requirements-tox-py27-check-uc/831f376/ : SUCCESS in 7m 55s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/655967/1/gate/requirements-tox-py36-check-uc/852ea36/ : SUCCESS in 8m 09s\n- cross-cinder-py27 http://logs.openstack.org/67/655967/1/gate/cross-cinder-py27/342f950/ : SUCCESS in 16m 09s\n- cross-cinder-py36 http://logs.openstack.org/67/655967/1/gate/cross-cinder-py36/dbf6e04/ : SUCCESS in 13m 34s\n- cross-glance-py27 http://logs.openstack.org/67/655967/1/gate/cross-glance-py27/9418653/ : SUCCESS in 5m 22s\n- cross-glance-py36 http://logs.openstack.org/67/655967/1/gate/cross-glance-py36/43cef5c/ : SUCCESS in 6m 37s\n- cross-horizon-py27 http://logs.openstack.org/67/655967/1/gate/cross-horizon-py27/b25d93e/ : SUCCESS in 13m 57s\n- cross-horizon-py36 http://logs.openstack.org/67/655967/1/gate/cross-horizon-py36/2be7294/ : SUCCESS in 9m 48s\n- cross-horizon-npm http://logs.openstack.org/67/655967/1/gate/cross-horizon-npm/deb718e/ : SUCCESS in 9m 01s\n- cross-keystone-py27 http://logs.openstack.org/67/655967/1/gate/cross-keystone-py27/7e3d2c2/ : SUCCESS in 24m 48s\n- cross-keystone-py36 http://logs.openstack.org/67/655967/1/gate/cross-keystone-py36/b4948a9/ : SUCCESS in 22m 00s\n- cross-neutron-py27 http://logs.openstack.org/67/655967/1/gate/cross-neutron-py27/d4d75dd/ : SUCCESS in 25m 50s\n- cross-neutron-py36 http://logs.openstack.org/67/655967/1/gate/cross-neutron-py36/ace47ef/ : SUCCESS in 33m 29s\n- cross-nova-py27 http://logs.openstack.org/67/655967/1/gate/cross-nova-py27/6493382/ : SUCCESS in 11m 30s\n- cross-nova-py36 http://logs.openstack.org/67/655967/1/gate/cross-nova-py36/eecea8c/ : SUCCESS in 16m 08s\n- cross-nova-functional http://logs.openstack.org/67/655967/1/gate/cross-nova-functional/5c1fc04/ : SUCCESS in 19m 11s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/655967/1/gate/cross-osc-build-sphinx-docs/8ce348d/html/ : SUCCESS in 5m 04s\n- cross-swift-py27 http://logs.openstack.org/67/655967/1/gate/cross-swift-py27/0803f51/ : SUCCESS in 11m 04s\n- cross-swift-py36 http://logs.openstack.org/67/655967/1/gate/cross-swift-py36/8611c11/ : SUCCESS in 7m 49s\n- requirements-integration http://logs.openstack.org/67/655967/1/gate/requirements-integration/9a809d1/ : SUCCESS in 54m 26s\n- neutron-grenade-multinode http://logs.openstack.org/67/655967/1/gate/neutron-grenade-multinode/c252a82/ : SUCCESS in 1h 14m 04s\n- tempest-full http://logs.openstack.org/67/655967/1/gate/tempest-full/e48479b/ : TIMED_OUT in 2h 04m 13s\n- neutron-grenade http://logs.openstack.org/67/655967/1/gate/neutron-grenade/6774294/ : SUCCESS in 49m 22s","accounts_in_message":[],"_revision_number":1},{"id":"92fce0fb8cb711ce93b2c13887798ea96cc3acf3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-28 02:58:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"87ec16652ea0398fd53cd12468ab374736c66891","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 04:54:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/655967/1/check/openstack-tox-docs/40de19e/html/ : SUCCESS in 4m 18s\n- openstack-tox-pypy http://logs.openstack.org/67/655967/1/check/openstack-tox-pypy/ba11a2d/ : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/655967/1/check/openstack-tox-pep8/acb4be5/ : SUCCESS in 2m 51s\n- openstack-tox-py27 http://logs.openstack.org/67/655967/1/check/openstack-tox-py27/3b2e52c/ : SUCCESS in 4m 35s\n- openstack-tox-py36 http://logs.openstack.org/67/655967/1/check/openstack-tox-py36/4ecc944/ : SUCCESS in 3m 19s\n- openstack-tox-validate http://logs.openstack.org/67/655967/1/check/openstack-tox-validate/3811577/ : SUCCESS in 2m 55s\n- requirements-tox-babel http://logs.openstack.org/67/655967/1/check/requirements-tox-babel/f11e109/ : SUCCESS in 2m 40s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/655967/1/check/requirements-tox-py27-check-uc/130ae9a/ : SUCCESS in 7m 23s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/655967/1/check/requirements-tox-py36-check-uc/95518fa/ : SUCCESS in 7m 22s\n- cross-cinder-py27 http://logs.openstack.org/67/655967/1/check/cross-cinder-py27/b66d4ff/ : SUCCESS in 8m 55s\n- cross-cinder-py36 http://logs.openstack.org/67/655967/1/check/cross-cinder-py36/10125da/ : SUCCESS in 9m 28s\n- cross-glance-py27 http://logs.openstack.org/67/655967/1/check/cross-glance-py27/25a6d8e/ : SUCCESS in 4m 45s\n- cross-glance-py36 http://logs.openstack.org/67/655967/1/check/cross-glance-py36/8e2e962/ : SUCCESS in 5m 17s\n- cross-horizon-py27 http://logs.openstack.org/67/655967/1/check/cross-horizon-py27/f0e5869/ : SUCCESS in 9m 54s\n- cross-horizon-py36 http://logs.openstack.org/67/655967/1/check/cross-horizon-py36/ee31c5c/ : SUCCESS in 8m 50s\n- cross-horizon-npm http://logs.openstack.org/67/655967/1/check/cross-horizon-npm/0dcdace/ : SUCCESS in 6m 13s\n- cross-keystone-py27 http://logs.openstack.org/67/655967/1/check/cross-keystone-py27/ae363cf/ : SUCCESS in 20m 26s\n- cross-keystone-py36 http://logs.openstack.org/67/655967/1/check/cross-keystone-py36/104ca73/ : SUCCESS in 25m 52s\n- cross-neutron-py27 http://logs.openstack.org/67/655967/1/check/cross-neutron-py27/e0749d0/ : SUCCESS in 28m 46s\n- cross-neutron-py36 http://logs.openstack.org/67/655967/1/check/cross-neutron-py36/41801f2/ : SUCCESS in 22m 08s\n- cross-nova-py27 http://logs.openstack.org/67/655967/1/check/cross-nova-py27/9b79e1a/ : SUCCESS in 14m 02s\n- cross-nova-py36 http://logs.openstack.org/67/655967/1/check/cross-nova-py36/c87290e/ : SUCCESS in 11m 06s\n- cross-nova-functional http://logs.openstack.org/67/655967/1/check/cross-nova-functional/1cf8dd3/ : SUCCESS in 19m 31s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/655967/1/check/cross-osc-build-sphinx-docs/5367d69/html/ : SUCCESS in 5m 39s\n- cross-swift-py27 http://logs.openstack.org/67/655967/1/check/cross-swift-py27/63a371e/ : SUCCESS in 8m 13s\n- cross-swift-py36 http://logs.openstack.org/67/655967/1/check/cross-swift-py36/0c9f9bd/ : SUCCESS in 6m 57s\n- requirements-integration http://logs.openstack.org/67/655967/1/check/requirements-integration/d76a8fe/ : SUCCESS in 1h 01m 34s\n- neutron-grenade-multinode http://logs.openstack.org/67/655967/1/check/neutron-grenade-multinode/295fd41/ : SUCCESS in 1h 05m 04s\n- tempest-full http://logs.openstack.org/67/655967/1/check/tempest-full/162ac50/ : SUCCESS in 1h 49m 59s\n- neutron-grenade http://logs.openstack.org/67/655967/1/check/neutron-grenade/13292fb/ : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":1},{"id":"060eba09fe69d93ac4b892a84fd5e7d6d5b9c5e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 04:54:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3c0a6ba6dada0b1fd3389cb53e228bff0dd847a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 07:20:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/67/655967/1/gate/openstack-tox-docs/71d2f54/html/ : SUCCESS in 2m 52s\n- openstack-tox-pep8 http://logs.openstack.org/67/655967/1/gate/openstack-tox-pep8/3b4b4b1/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/67/655967/1/gate/openstack-tox-py27/aa06fec/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/67/655967/1/gate/openstack-tox-py36/035c679/ : SUCCESS in 3m 05s\n- openstack-tox-validate http://logs.openstack.org/67/655967/1/gate/openstack-tox-validate/e377645/ : SUCCESS in 3m 18s\n- requirements-tox-babel http://logs.openstack.org/67/655967/1/gate/requirements-tox-babel/a4d7cac/ : SUCCESS in 2m 35s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/655967/1/gate/requirements-tox-py27-check-uc/0f97cf9/ : SUCCESS in 7m 43s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/655967/1/gate/requirements-tox-py36-check-uc/dcccb0a/ : SUCCESS in 7m 25s\n- cross-cinder-py27 http://logs.openstack.org/67/655967/1/gate/cross-cinder-py27/8de1118/ : SUCCESS in 9m 13s\n- cross-cinder-py36 http://logs.openstack.org/67/655967/1/gate/cross-cinder-py36/61a0fdf/ : SUCCESS in 8m 08s\n- cross-glance-py27 http://logs.openstack.org/67/655967/1/gate/cross-glance-py27/737b1db/ : SUCCESS in 4m 12s\n- cross-glance-py36 http://logs.openstack.org/67/655967/1/gate/cross-glance-py36/0fcb867/ : SUCCESS in 4m 24s\n- cross-horizon-py27 http://logs.openstack.org/67/655967/1/gate/cross-horizon-py27/b7b2ff6/ : SUCCESS in 10m 25s\n- cross-horizon-py36 http://logs.openstack.org/67/655967/1/gate/cross-horizon-py36/b502cf8/ : SUCCESS in 9m 18s\n- cross-horizon-npm http://logs.openstack.org/67/655967/1/gate/cross-horizon-npm/ff78433/ : SUCCESS in 5m 51s\n- cross-keystone-py27 http://logs.openstack.org/67/655967/1/gate/cross-keystone-py27/e8a6334/ : SUCCESS in 22m 26s\n- cross-keystone-py36 http://logs.openstack.org/67/655967/1/gate/cross-keystone-py36/645d4cd/ : SUCCESS in 21m 22s\n- cross-neutron-py27 http://logs.openstack.org/67/655967/1/gate/cross-neutron-py27/005a8bc/ : POST_FAILURE in 22m 13s\n- cross-neutron-py36 http://logs.openstack.org/67/655967/1/gate/cross-neutron-py36/cfbb5c6/ : SUCCESS in 20m 55s\n- cross-nova-py27 http://logs.openstack.org/67/655967/1/gate/cross-nova-py27/5c0e944/ : SUCCESS in 11m 22s\n- cross-nova-py36 http://logs.openstack.org/67/655967/1/gate/cross-nova-py36/e40b159/ : SUCCESS in 11m 42s\n- cross-nova-functional http://logs.openstack.org/67/655967/1/gate/cross-nova-functional/a47bfdb/ : SUCCESS in 20m 42s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/655967/1/gate/cross-osc-build-sphinx-docs/a02b9d7/html/ : SUCCESS in 5m 04s\n- cross-swift-py27 http://logs.openstack.org/67/655967/1/gate/cross-swift-py27/cb8e96e/ : SUCCESS in 7m 32s\n- cross-swift-py36 http://logs.openstack.org/67/655967/1/gate/cross-swift-py36/574cd23/ : SUCCESS in 6m 50s\n- requirements-integration http://logs.openstack.org/67/655967/1/gate/requirements-integration/4cf4bdb/ : SUCCESS in 58m 11s\n- neutron-grenade-multinode http://logs.openstack.org/67/655967/1/gate/neutron-grenade-multinode/096b7af/ : SUCCESS in 1h 09m 55s\n- tempest-full http://logs.openstack.org/67/655967/1/gate/tempest-full/4cd2ab5/ : FAILURE in 1h 52m 50s\n- neutron-grenade http://logs.openstack.org/67/655967/1/gate/neutron-grenade/79c426d/ : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a7e7ea501d58ac9ee4b53a93a3373035434f4817","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-28 08:00:07.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bf6df9f3a70b15ccc4228508dc7d67766866542e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 09:54:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/655967/1/check/openstack-tox-docs/d46ece9/html/ : SUCCESS in 2m 54s\n- openstack-tox-pypy http://logs.openstack.org/67/655967/1/check/openstack-tox-pypy/acbd4d2/ : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/655967/1/check/openstack-tox-pep8/328dd91/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/67/655967/1/check/openstack-tox-py27/bb5be05/ : SUCCESS in 3m 33s\n- openstack-tox-py36 http://logs.openstack.org/67/655967/1/check/openstack-tox-py36/92fae17/ : SUCCESS in 3m 52s\n- openstack-tox-validate http://logs.openstack.org/67/655967/1/check/openstack-tox-validate/14c6091/ : SUCCESS in 3m 01s\n- requirements-tox-babel http://logs.openstack.org/67/655967/1/check/requirements-tox-babel/2cfb552/ : SUCCESS in 2m 41s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/655967/1/check/requirements-tox-py27-check-uc/618072c/ : SUCCESS in 7m 50s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/655967/1/check/requirements-tox-py36-check-uc/22c9340/ : SUCCESS in 6m 46s\n- cross-cinder-py27 http://logs.openstack.org/67/655967/1/check/cross-cinder-py27/5101be9/ : SUCCESS in 8m 44s\n- cross-cinder-py36 http://logs.openstack.org/67/655967/1/check/cross-cinder-py36/9dbf8b8/ : SUCCESS in 8m 11s\n- cross-glance-py27 http://logs.openstack.org/67/655967/1/check/cross-glance-py27/e682983/ : SUCCESS in 5m 05s\n- cross-glance-py36 http://logs.openstack.org/67/655967/1/check/cross-glance-py36/607cbed/ : SUCCESS in 4m 39s\n- cross-horizon-py27 http://logs.openstack.org/67/655967/1/check/cross-horizon-py27/10e9876/ : SUCCESS in 9m 58s\n- cross-horizon-py36 http://logs.openstack.org/67/655967/1/check/cross-horizon-py36/4c3196f/ : SUCCESS in 8m 00s\n- cross-horizon-npm http://logs.openstack.org/67/655967/1/check/cross-horizon-npm/0dc58b4/ : SUCCESS in 5m 24s\n- cross-keystone-py27 http://logs.openstack.org/67/655967/1/check/cross-keystone-py27/fb394fb/ : SUCCESS in 21m 05s\n- cross-keystone-py36 http://logs.openstack.org/67/655967/1/check/cross-keystone-py36/b15f740/ : SUCCESS in 23m 18s\n- cross-neutron-py27 http://logs.openstack.org/67/655967/1/check/cross-neutron-py27/848094e/ : SUCCESS in 22m 14s\n- cross-neutron-py36 http://logs.openstack.org/67/655967/1/check/cross-neutron-py36/973dcde/ : SUCCESS in 21m 04s\n- cross-nova-py27 http://logs.openstack.org/67/655967/1/check/cross-nova-py27/0677610/ : SUCCESS in 11m 43s\n- cross-nova-py36 http://logs.openstack.org/67/655967/1/check/cross-nova-py36/863c6d5/ : SUCCESS in 11m 15s\n- cross-nova-functional http://logs.openstack.org/67/655967/1/check/cross-nova-functional/06e5943/ : SUCCESS in 24m 14s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/655967/1/check/cross-osc-build-sphinx-docs/372174f/html/ : SUCCESS in 5m 14s\n- cross-swift-py27 http://logs.openstack.org/67/655967/1/check/cross-swift-py27/90e5265/ : SUCCESS in 8m 37s\n- cross-swift-py36 http://logs.openstack.org/67/655967/1/check/cross-swift-py36/650b9c1/ : SUCCESS in 11m 20s\n- requirements-integration http://logs.openstack.org/67/655967/1/check/requirements-integration/21ee0de/ : SUCCESS in 53m 03s\n- neutron-grenade-multinode http://logs.openstack.org/67/655967/1/check/neutron-grenade-multinode/d17e322/ : SUCCESS in 1h 16m 14s\n- tempest-full http://logs.openstack.org/67/655967/1/check/tempest-full/dbc6e8d/ : SUCCESS in 1h 53m 47s\n- neutron-grenade http://logs.openstack.org/67/655967/1/check/neutron-grenade/0f65037/ : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2606f177048c53a68f0ec5897e1892a6fbb182b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 09:54:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b3c1bdc2ef48f45205fc709d9b348e90bc767d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 11:53:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/655967/1/gate/openstack-tox-docs/7b9a95c/html/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/67/655967/1/gate/openstack-tox-pep8/3dfde65/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/67/655967/1/gate/openstack-tox-py27/cb7718c/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/67/655967/1/gate/openstack-tox-py36/8548f1e/ : SUCCESS in 3m 47s\n- openstack-tox-validate http://logs.openstack.org/67/655967/1/gate/openstack-tox-validate/c01b413/ : SUCCESS in 2m 56s\n- requirements-tox-babel http://logs.openstack.org/67/655967/1/gate/requirements-tox-babel/7a84fb4/ : SUCCESS in 3m 13s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/655967/1/gate/requirements-tox-py27-check-uc/a8e7061/ : SUCCESS in 13m 04s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/655967/1/gate/requirements-tox-py36-check-uc/c5a4874/ : SUCCESS in 9m 09s\n- cross-cinder-py27 http://logs.openstack.org/67/655967/1/gate/cross-cinder-py27/6ca3c06/ : SUCCESS in 12m 30s\n- cross-cinder-py36 http://logs.openstack.org/67/655967/1/gate/cross-cinder-py36/5017e9b/ : SUCCESS in 7m 34s\n- cross-glance-py27 http://logs.openstack.org/67/655967/1/gate/cross-glance-py27/daeed00/ : SUCCESS in 4m 54s\n- cross-glance-py36 http://logs.openstack.org/67/655967/1/gate/cross-glance-py36/f98f59f/ : SUCCESS in 4m 20s\n- cross-horizon-py27 http://logs.openstack.org/67/655967/1/gate/cross-horizon-py27/b42f721/ : SUCCESS in 10m 43s\n- cross-horizon-py36 http://logs.openstack.org/67/655967/1/gate/cross-horizon-py36/396ae66/ : SUCCESS in 9m 13s\n- cross-horizon-npm http://logs.openstack.org/67/655967/1/gate/cross-horizon-npm/2f3289c/ : SUCCESS in 5m 57s\n- cross-keystone-py27 http://logs.openstack.org/67/655967/1/gate/cross-keystone-py27/b7d5990/ : SUCCESS in 23m 20s\n- cross-keystone-py36 http://logs.openstack.org/67/655967/1/gate/cross-keystone-py36/5516d08/ : SUCCESS in 24m 27s\n- cross-neutron-py27 http://logs.openstack.org/67/655967/1/gate/cross-neutron-py27/85f5c94/ : SUCCESS in 34m 01s\n- cross-neutron-py36 http://logs.openstack.org/67/655967/1/gate/cross-neutron-py36/1aa8d88/ : SUCCESS in 22m 34s\n- cross-nova-py27 http://logs.openstack.org/67/655967/1/gate/cross-nova-py27/0f47af7/ : SUCCESS in 12m 40s\n- cross-nova-py36 http://logs.openstack.org/67/655967/1/gate/cross-nova-py36/4a791be/ : SUCCESS in 10m 56s\n- cross-nova-functional http://logs.openstack.org/67/655967/1/gate/cross-nova-functional/da15d9e/ : SUCCESS in 21m 30s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/655967/1/gate/cross-osc-build-sphinx-docs/a4ba09e/html/ : SUCCESS in 9m 49s\n- cross-swift-py27 http://logs.openstack.org/67/655967/1/gate/cross-swift-py27/fd8dc8a/ : SUCCESS in 7m 16s\n- cross-swift-py36 http://logs.openstack.org/67/655967/1/gate/cross-swift-py36/c585be6/ : SUCCESS in 6m 38s\n- requirements-integration http://logs.openstack.org/67/655967/1/gate/requirements-integration/25d7a53/ : SUCCESS in 57m 36s\n- neutron-grenade-multinode http://logs.openstack.org/67/655967/1/gate/neutron-grenade-multinode/8c71aed/ : SUCCESS in 1h 04m 02s\n- tempest-full http://logs.openstack.org/67/655967/1/gate/tempest-full/a56d2f8/ : SUCCESS in 1h 50m 28s\n- neutron-grenade http://logs.openstack.org/67/655967/1/gate/neutron-grenade/cbd91af/ : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":1},{"id":"64c0c19a7286e316e7a9c2955420cd04d6e4dc7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 11:53:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e13060307959e9dfcc380612e9baefc58affdbd8","revisions":{"e13060307959e9dfcc380612e9baefc58affdbd8":{"kind":"REWORK","_number":1,"created":"2019-04-26 17:04:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/655967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/655967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/655967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/655967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/655967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/655967/1"}}},"commit":{"parents":[{"commit":"c7c3edc2add91936abf98fbefb2e0ae9cb3bb63f","subject":"Merge \"update constraint for neutron-lib to new release 1.26.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c7c3edc2add91936abf98fbefb2e0ae9cb3bb63f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 17:04:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 17:04:48.000000000","tz":0},"subject":"update constraint for oslo.middleware to new release 3.38.0","message":"update constraint for oslo.middleware to new release 3.38.0\n\nChange-Id: I60a9d085a0fee9334bd0466ee8950f7d9c025498\nmeta:version: 3.38.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I3b9441931e04056a848bab7dbeb80b70a343f1db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e13060307959e9dfcc380612e9baefc58affdbd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e13060307959e9dfcc380612e9baefc58affdbd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
