)]}'
{"id":"openstack%2Frequirements~655971","triplet_id":"openstack%2Frequirements~stable%2Fpike~I547a11a6ef881220a7ffdbf7e57067b63917039f","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"I547a11a6ef881220a7ffdbf7e57067b63917039f","subject":"update constraint for python-zaqarclient to new release 1.7.1","status":"MERGED","created":"2019-04-26 17:07:06.000000000","updated":"2019-04-29 15:53:35.000000000","submitted":"2019-04-29 15:53:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655971-1556553215235-af911147","meta_rev_id":"0a2118c042ff57f7e2da293bf646417108498ff1","_number":655971,"virtual_id_number":655971,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-04-29 15:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-27 05:08:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-29 12:28:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-04-29 12:28:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-27 05:08:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-29 12:28:36.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-29 15:53:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8363dd7f6b1daf0d30d0521a4942420b94b02ec7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-26 17:07:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75f9a920b056de6a82901fe8f6bd04e89ff3fd99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 23:29:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/71/655971/1/check/openstack-tox-pypy/c93b971/ : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/71/655971/1/check/openstack-tox-pep8/87f972f/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/71/655971/1/check/openstack-tox-py27/42cb4e2/ : SUCCESS in 3m 28s\n- openstack-tox-py35 http://logs.openstack.org/71/655971/1/check/openstack-tox-py35/c3854a5/ : SUCCESS in 3m 11s\n- openstack-tox-validate http://logs.openstack.org/71/655971/1/check/openstack-tox-validate/d83767e/ : SUCCESS in 3m 24s\n- requirements-tox-babel http://logs.openstack.org/71/655971/1/check/requirements-tox-babel/b176d63/ : SUCCESS in 2m 59s\n- requirements-tox-py27-check-uc http://logs.openstack.org/71/655971/1/check/requirements-tox-py27-check-uc/c781623/ : SUCCESS in 6m 10s\n- cross-cinder-py27 http://logs.openstack.org/71/655971/1/check/cross-cinder-py27/e8e9dba/ : SUCCESS in 8m 05s\n- cross-cinder-py35 http://logs.openstack.org/71/655971/1/check/cross-cinder-py35/d38098f/ : SUCCESS in 7m 30s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/71/655971/1/check/cross-glance-py27/7f5361e/ : SUCCESS in 9m 36s\n- cross-glance-py35 http://logs.openstack.org/71/655971/1/check/cross-glance-py35/359f263/ : SUCCESS in 8m 23s\n- cross-horizon-py27 http://logs.openstack.org/71/655971/1/check/cross-horizon-py27/7b97481/ : SUCCESS in 9m 52s\n- cross-horizon-py35 http://logs.openstack.org/71/655971/1/check/cross-horizon-py35/cff5818/ : SUCCESS in 9m 53s\n- cross-keystone-py27 http://logs.openstack.org/71/655971/1/check/cross-keystone-py27/90dc278/ : SUCCESS in 11m 46s\n- cross-keystone-py35 http://logs.openstack.org/71/655971/1/check/cross-keystone-py35/e6560dc/ : SUCCESS in 11m 09s\n- cross-neutron-py27 http://logs.openstack.org/71/655971/1/check/cross-neutron-py27/4dd3313/ : SUCCESS in 21m 41s\n- cross-neutron-py35 http://logs.openstack.org/71/655971/1/check/cross-neutron-py35/f5fced8/ : SUCCESS in 22m 31s\n- cross-nova-py27 http://logs.openstack.org/71/655971/1/check/cross-nova-py27/483f6a3/ : SUCCESS in 11m 32s\n- cross-nova-py35 http://logs.openstack.org/71/655971/1/check/cross-nova-py35/17c7cd9/ : SUCCESS in 13m 26s\n- cross-nova-functional http://logs.openstack.org/71/655971/1/check/cross-nova-functional/45ec2bd/ : SUCCESS in 21m 26s\n- cross-swift-py27 http://logs.openstack.org/71/655971/1/check/cross-swift-py27/5bc3090/ : SUCCESS in 10m 00s\n- cross-swift-py35 http://logs.openstack.org/71/655971/1/check/cross-swift-py35/0f9878d/ : SUCCESS in 3m 37s\n- requirements-integration http://logs.openstack.org/71/655971/1/check/requirements-integration/eaafc89/ : SUCCESS in 51m 17s\n- neutron-grenade-multinode http://logs.openstack.org/71/655971/1/check/neutron-grenade-multinode/298d6a4/ : SUCCESS in 1h 05m 34s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/71/655971/1/check/legacy-tempest-dsvm-neutron-full/2d0e923/ : SUCCESS in 1h 20m 31s\n- neutron-grenade http://logs.openstack.org/71/655971/1/check/neutron-grenade/f5df60e/ : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1985a89f56687a76a94aba265bb9a083981455d3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-27 05:08:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0ea2de966af04e35c98a7b47b8e4d6a8a61a55d6","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-29 12:28:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"65d9de6568f88ce3fab7dcf290852a96257e7638","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 12:28:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69cb8a85ddb4c6487407fbf6968fa1790f22381a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 15:53:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"30e21921cdae9c7739a3f65f24f71a402ce7a17e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 15:53:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/71/655971/1/gate/openstack-tox-pep8/419d127/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/71/655971/1/gate/openstack-tox-py27/4a0719f/ : SUCCESS in 3m 54s\n- openstack-tox-py35 http://logs.openstack.org/71/655971/1/gate/openstack-tox-py35/87178f1/ : SUCCESS in 3m 13s\n- openstack-tox-validate http://logs.openstack.org/71/655971/1/gate/openstack-tox-validate/f3ebcd5/ : SUCCESS in 3m 32s\n- requirements-tox-babel http://logs.openstack.org/71/655971/1/gate/requirements-tox-babel/c65eda3/ : SUCCESS in 3m 04s\n- requirements-tox-py27-check-uc http://logs.openstack.org/71/655971/1/gate/requirements-tox-py27-check-uc/3d19d20/ : SUCCESS in 7m 30s\n- cross-cinder-py27 http://logs.openstack.org/71/655971/1/gate/cross-cinder-py27/705347c/ : SUCCESS in 7m 59s\n- cross-glance-py27 http://logs.openstack.org/71/655971/1/gate/cross-glance-py27/5faa431/ : SUCCESS in 13m 43s\n- cross-glance-py35 http://logs.openstack.org/71/655971/1/gate/cross-glance-py35/896d7ef/ : SUCCESS in 10m 29s\n- cross-horizon-py27 http://logs.openstack.org/71/655971/1/gate/cross-horizon-py27/ff2f6a2/ : SUCCESS in 10m 09s\n- cross-horizon-py35 http://logs.openstack.org/71/655971/1/gate/cross-horizon-py35/3ed4054/ : SUCCESS in 10m 13s\n- cross-keystone-py27 http://logs.openstack.org/71/655971/1/gate/cross-keystone-py27/276887e/ : SUCCESS in 14m 40s\n- cross-keystone-py35 http://logs.openstack.org/71/655971/1/gate/cross-keystone-py35/5261615/ : SUCCESS in 13m 49s\n- cross-neutron-py27 http://logs.openstack.org/71/655971/1/gate/cross-neutron-py27/ccbe800/ : SUCCESS in 21m 55s\n- cross-neutron-py35 http://logs.openstack.org/71/655971/1/gate/cross-neutron-py35/1406ae9/ : SUCCESS in 22m 25s\n- cross-nova-py27 http://logs.openstack.org/71/655971/1/gate/cross-nova-py27/185b232/ : SUCCESS in 12m 10s\n- cross-nova-py35 http://logs.openstack.org/71/655971/1/gate/cross-nova-py35/4385f88/ : SUCCESS in 13m 12s\n- cross-nova-functional http://logs.openstack.org/71/655971/1/gate/cross-nova-functional/8f875e4/ : SUCCESS in 26m 27s\n- cross-swift-py27 http://logs.openstack.org/71/655971/1/gate/cross-swift-py27/4bbe149/ : SUCCESS in 7m 27s\n- cross-swift-py35 http://logs.openstack.org/71/655971/1/gate/cross-swift-py35/6265d5d/ : SUCCESS in 3m 27s\n- requirements-integration http://logs.openstack.org/71/655971/1/gate/requirements-integration/00bd778/ : SUCCESS in 53m 07s\n- neutron-grenade-multinode http://logs.openstack.org/71/655971/1/gate/neutron-grenade-multinode/3f3bfd6/ : SUCCESS in 1h 17m 25s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/71/655971/1/gate/legacy-tempest-dsvm-neutron-full/1763d9c/ : SUCCESS in 1h 55m 19s\n- neutron-grenade http://logs.openstack.org/71/655971/1/gate/neutron-grenade/98dbed9/ : SUCCESS in 54m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fec3c5221bb6e078ff5862b85267ef02feac5048","revisions":{"fec3c5221bb6e078ff5862b85267ef02feac5048":{"kind":"REWORK","_number":1,"created":"2019-04-26 17:07:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/655971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/655971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/655971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/655971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/655971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/655971/1"}}},"commit":{"parents":[{"commit":"6c627edcc32c0e65d75d93084c12e1de30a348c1","subject":"Merge \"update constraint for keystonemiddleware to new release 4.17.1\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c627edcc32c0e65d75d93084c12e1de30a348c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 17:07:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 17:07:04.000000000","tz":0},"subject":"update constraint for python-zaqarclient to new release 1.7.1","message":"update constraint for python-zaqarclient to new release 1.7.1\n\nChange-Id: I547a11a6ef881220a7ffdbf7e57067b63917039f\nmeta:version: 1.7.1\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Id89d32ba5e8c6f986015b406c88cfc0b0f8b9fc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fec3c5221bb6e078ff5862b85267ef02feac5048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fec3c5221bb6e078ff5862b85267ef02feac5048"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
