)]}'
{"id":"openstack%2Frequirements~655980","triplet_id":"openstack%2Frequirements~stable%2Fpike~I435e62e8d41598314777952b30f4a8de57b3cca2","project":"openstack/requirements","branch":"stable/pike","topic":"new-release","hashtags":[],"change_id":"I435e62e8d41598314777952b30f4a8de57b3cca2","subject":"update constraint for os_vif to new release 1.7.2","status":"MERGED","created":"2019-04-26 17:34:07.000000000","updated":"2019-04-29 16:09:48.000000000","submitted":"2019-04-29 16:09:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655980-1556554188460-fe66b0bc","meta_rev_id":"6a73795445652bf13faf0f37db32d8b8fb6a3c0a","_number":655980,"virtual_id_number":655980,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-04-29 16:09:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-04-27 05:08:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-04-29 12:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-04-29 14:33:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-27 05:08:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-04-29 14:33:25.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-04-29 16:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6b0d2f476c1724953e9bc067f1e8cbcf8138fff","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-26 17:34:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7107134fb85c5e291d9ffb040bc960773248083","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 00:00:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/80/655980/1/check/openstack-tox-pypy/98bb35e/ : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/80/655980/1/check/openstack-tox-pep8/c8a789b/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/80/655980/1/check/openstack-tox-py27/50a97e8/ : SUCCESS in 3m 58s\n- openstack-tox-py35 http://logs.openstack.org/80/655980/1/check/openstack-tox-py35/d140e96/ : SUCCESS in 3m 37s\n- openstack-tox-validate http://logs.openstack.org/80/655980/1/check/openstack-tox-validate/e7de245/ : SUCCESS in 3m 06s\n- requirements-tox-babel http://logs.openstack.org/80/655980/1/check/requirements-tox-babel/b6f61f0/ : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc http://logs.openstack.org/80/655980/1/check/requirements-tox-py27-check-uc/d96cf16/ : SUCCESS in 7m 04s\n- cross-cinder-py27 http://logs.openstack.org/80/655980/1/check/cross-cinder-py27/1472366/ : SUCCESS in 7m 14s\n- cross-cinder-py35 http://logs.openstack.org/80/655980/1/check/cross-cinder-py35/f59f0e5/ : SUCCESS in 7m 13s (non-voting)\n- cross-glance-py27 http://logs.openstack.org/80/655980/1/check/cross-glance-py27/2effbc9/ : SUCCESS in 10m 48s\n- cross-glance-py35 http://logs.openstack.org/80/655980/1/check/cross-glance-py35/affa250/ : SUCCESS in 11m 08s\n- cross-horizon-py27 http://logs.openstack.org/80/655980/1/check/cross-horizon-py27/1be57d6/ : SUCCESS in 11m 30s\n- cross-horizon-py35 http://logs.openstack.org/80/655980/1/check/cross-horizon-py35/11f5e83/ : SUCCESS in 9m 45s\n- cross-keystone-py27 http://logs.openstack.org/80/655980/1/check/cross-keystone-py27/95fbf64/ : SUCCESS in 11m 47s\n- cross-keystone-py35 http://logs.openstack.org/80/655980/1/check/cross-keystone-py35/4c4808f/ : SUCCESS in 11m 50s\n- cross-neutron-py27 http://logs.openstack.org/80/655980/1/check/cross-neutron-py27/b185e0b/ : SUCCESS in 29m 01s\n- cross-neutron-py35 http://logs.openstack.org/80/655980/1/check/cross-neutron-py35/d1f0273/ : SUCCESS in 23m 25s\n- cross-nova-py27 http://logs.openstack.org/80/655980/1/check/cross-nova-py27/312e847/ : SUCCESS in 11m 45s\n- cross-nova-py35 http://logs.openstack.org/80/655980/1/check/cross-nova-py35/17846a4/ : SUCCESS in 12m 58s\n- cross-nova-functional http://logs.openstack.org/80/655980/1/check/cross-nova-functional/eae9a70/ : SUCCESS in 18m 43s\n- cross-swift-py27 http://logs.openstack.org/80/655980/1/check/cross-swift-py27/e7a5615/ : SUCCESS in 7m 02s\n- cross-swift-py35 http://logs.openstack.org/80/655980/1/check/cross-swift-py35/11da0d3/ : SUCCESS in 4m 06s\n- requirements-integration http://logs.openstack.org/80/655980/1/check/requirements-integration/867fed5/ : SUCCESS in 44m 10s\n- neutron-grenade-multinode http://logs.openstack.org/80/655980/1/check/neutron-grenade-multinode/5c0ca63/ : SUCCESS in 1h 00m 05s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/80/655980/1/check/legacy-tempest-dsvm-neutron-full/b7c16c7/ : SUCCESS in 1h 29m 21s\n- neutron-grenade http://logs.openstack.org/80/655980/1/check/neutron-grenade/40ecc50/ : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1918866bf5958c4cab84c8b9c903a876a4e12970","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-04-27 05:08:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd56b5554381e4c32660c0eddfb34f1379e8391b","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-29 12:31:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"57c088d89de8af2a75e6f9c4e00349365d6693a3","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-04-29 14:33:25.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"522e90e787d31fd26e421654ded03fc2c6679a33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 14:33:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"decf732725d052c221c45a55041f0958bf5a4f1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 16:09:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/80/655980/1/gate/openstack-tox-pep8/ea40256/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/80/655980/1/gate/openstack-tox-py27/b8b1cf5/ : SUCCESS in 3m 44s\n- openstack-tox-py35 http://logs.openstack.org/80/655980/1/gate/openstack-tox-py35/dda34e9/ : SUCCESS in 3m 46s\n- openstack-tox-validate http://logs.openstack.org/80/655980/1/gate/openstack-tox-validate/4d25128/ : SUCCESS in 4m 37s\n- requirements-tox-babel http://logs.openstack.org/80/655980/1/gate/requirements-tox-babel/fe38392/ : SUCCESS in 3m 37s\n- requirements-tox-py27-check-uc http://logs.openstack.org/80/655980/1/gate/requirements-tox-py27-check-uc/1b84574/ : SUCCESS in 6m 02s\n- cross-cinder-py27 http://logs.openstack.org/80/655980/1/gate/cross-cinder-py27/5420d52/ : SUCCESS in 7m 50s\n- cross-glance-py27 http://logs.openstack.org/80/655980/1/gate/cross-glance-py27/d376b00/ : SUCCESS in 9m 16s\n- cross-glance-py35 http://logs.openstack.org/80/655980/1/gate/cross-glance-py35/246a366/ : SUCCESS in 8m 56s\n- cross-horizon-py27 http://logs.openstack.org/80/655980/1/gate/cross-horizon-py27/b31e83e/ : SUCCESS in 10m 41s\n- cross-horizon-py35 http://logs.openstack.org/80/655980/1/gate/cross-horizon-py35/a45b88b/ : SUCCESS in 10m 38s\n- cross-keystone-py27 http://logs.openstack.org/80/655980/1/gate/cross-keystone-py27/009319e/ : SUCCESS in 11m 53s\n- cross-keystone-py35 http://logs.openstack.org/80/655980/1/gate/cross-keystone-py35/749fde4/ : SUCCESS in 14m 14s\n- cross-neutron-py27 http://logs.openstack.org/80/655980/1/gate/cross-neutron-py27/22e39fe/ : SUCCESS in 26m 40s\n- cross-neutron-py35 http://logs.openstack.org/80/655980/1/gate/cross-neutron-py35/a3668fc/ : SUCCESS in 28m 05s\n- cross-nova-py27 http://logs.openstack.org/80/655980/1/gate/cross-nova-py27/e8fa8f9/ : SUCCESS in 13m 28s\n- cross-nova-py35 http://logs.openstack.org/80/655980/1/gate/cross-nova-py35/5d9837d/ : SUCCESS in 12m 32s\n- cross-nova-functional http://logs.openstack.org/80/655980/1/gate/cross-nova-functional/4775f3e/ : SUCCESS in 20m 58s\n- cross-swift-py27 http://logs.openstack.org/80/655980/1/gate/cross-swift-py27/f5a2dce/ : SUCCESS in 8m 50s\n- cross-swift-py35 http://logs.openstack.org/80/655980/1/gate/cross-swift-py35/1bfffbb/ : SUCCESS in 3m 37s\n- requirements-integration http://logs.openstack.org/80/655980/1/gate/requirements-integration/78a63da/ : SUCCESS in 53m 33s\n- neutron-grenade-multinode http://logs.openstack.org/80/655980/1/gate/neutron-grenade-multinode/67c6b5e/ : SUCCESS in 1h 04m 17s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/80/655980/1/gate/legacy-tempest-dsvm-neutron-full/240f28f/ : SUCCESS in 1h 32m 51s\n- neutron-grenade http://logs.openstack.org/80/655980/1/gate/neutron-grenade/8449f9b/ : SUCCESS in 43m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e86e7324c3fb9f61751fe912e28f9b3425309b30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 16:09:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc576d74250b853112fbae44e83c9b5f4f3e0d1a","revisions":{"cc576d74250b853112fbae44e83c9b5f4f3e0d1a":{"kind":"REWORK","_number":1,"created":"2019-04-26 17:34:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/655980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/655980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/655980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/655980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/655980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/655980/1"}}},"commit":{"parents":[{"commit":"6c627edcc32c0e65d75d93084c12e1de30a348c1","subject":"Merge \"update constraint for keystonemiddleware to new release 4.17.1\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c627edcc32c0e65d75d93084c12e1de30a348c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 17:34:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-26 17:34:06.000000000","tz":0},"subject":"update constraint for os_vif to new release 1.7.2","message":"update constraint for os_vif to new release 1.7.2\n\nChange-Id: I435e62e8d41598314777952b30f4a8de57b3cca2\nmeta:version: 1.7.2\nmeta:diff-start: -\nmeta:series: pike\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Co-Authored-By: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Change-Id: I64984811e5e3817ceaf053ad0ac8ae3c9e79d87d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc576d74250b853112fbae44e83c9b5f4f3e0d1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc576d74250b853112fbae44e83c9b5f4f3e0d1a"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
