)]}'
{"id":"openstack%2Frequirements~656736","triplet_id":"openstack%2Frequirements~master~I3067f250cf62c40b0f16117b73b0e4580de6c370","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I3067f250cf62c40b0f16117b73b0e4580de6c370","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-05-02 06:11:09.000000000","updated":"2019-05-04 02:40:28.000000000","submitted":"2019-05-04 02:40:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656736-1556937628722-13d10e42","meta_rev_id":"457c251ce3c8723ac608a64cd0b51c523d47dc09","_number":656736,"virtual_id_number":656736,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-04 02:40:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-03 05:54:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-03 05:54:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 05:54:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-04 02:40:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"332ae54310c40698d911221aa10c342c9b53b171","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-02 06:11:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a1b7f397a5d55bb967703aaed0c263f81c7af35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 07:25:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/36/656736/1/check/openstack-tox-docs/dcee540/html/ : SUCCESS in 2m 59s\n- openstack-tox-pypy http://logs.openstack.org/36/656736/1/check/openstack-tox-pypy/d8918c7/ : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/36/656736/1/check/openstack-tox-pep8/898bd93/ : SUCCESS in 2m 50s\n- openstack-tox-py27 http://logs.openstack.org/36/656736/1/check/openstack-tox-py27/efc5f5e/ : SUCCESS in 3m 02s\n- openstack-tox-py36 http://logs.openstack.org/36/656736/1/check/openstack-tox-py36/4dfe177/ : SUCCESS in 3m 28s\n- openstack-tox-validate http://logs.openstack.org/36/656736/1/check/openstack-tox-validate/fa9bb7d/ : SUCCESS in 3m 17s\n- requirements-tox-babel http://logs.openstack.org/36/656736/1/check/requirements-tox-babel/441005f/ : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc http://logs.openstack.org/36/656736/1/check/requirements-tox-py27-check-uc/9b2211a/ : FAILURE in 6m 54s\n- requirements-tox-py36-check-uc http://logs.openstack.org/36/656736/1/check/requirements-tox-py36-check-uc/90dd788/ : FAILURE in 8m 04s\n- cross-cinder-py27 http://logs.openstack.org/36/656736/1/check/cross-cinder-py27/5205be4/ : SUCCESS in 10m 27s\n- cross-cinder-py36 http://logs.openstack.org/36/656736/1/check/cross-cinder-py36/147052e/ : SUCCESS in 10m 12s\n- cross-glance-py27 http://logs.openstack.org/36/656736/1/check/cross-glance-py27/42ca875/ : SUCCESS in 5m 01s\n- cross-glance-py36 http://logs.openstack.org/36/656736/1/check/cross-glance-py36/476ab7a/ : SUCCESS in 4m 37s\n- cross-horizon-py27 http://logs.openstack.org/36/656736/1/check/cross-horizon-py27/692425e/ : SUCCESS in 10m 46s\n- cross-horizon-py36 http://logs.openstack.org/36/656736/1/check/cross-horizon-py36/13ae2f3/ : SUCCESS in 8m 56s\n- cross-horizon-npm http://logs.openstack.org/36/656736/1/check/cross-horizon-npm/1696553/ : SUCCESS in 6m 17s\n- cross-keystone-py27 http://logs.openstack.org/36/656736/1/check/cross-keystone-py27/b712635/ : FAILURE in 24m 33s\n- cross-keystone-py36 http://logs.openstack.org/36/656736/1/check/cross-keystone-py36/ee6d6f0/ : FAILURE in 23m 54s\n- cross-neutron-py27 http://logs.openstack.org/36/656736/1/check/cross-neutron-py27/26f78ef/ : SUCCESS in 21m 01s\n- cross-neutron-py36 http://logs.openstack.org/36/656736/1/check/cross-neutron-py36/e203cf1/ : SUCCESS in 22m 59s\n- cross-nova-py27 http://logs.openstack.org/36/656736/1/check/cross-nova-py27/7b82740/ : FAILURE in 5m 07s\n- cross-nova-py36 http://logs.openstack.org/36/656736/1/check/cross-nova-py36/a5e0bcf/ : FAILURE in 6m 41s\n- cross-nova-functional http://logs.openstack.org/36/656736/1/check/cross-nova-functional/0a9dfa0/ : FAILURE in 19m 05s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/36/656736/1/check/cross-osc-build-sphinx-docs/dd45827/html/ : SUCCESS in 5m 46s\n- cross-swift-py27 http://logs.openstack.org/36/656736/1/check/cross-swift-py27/756f5fa/ : SUCCESS in 6m 19s\n- cross-swift-py36 http://logs.openstack.org/36/656736/1/check/cross-swift-py36/84740ba/ : SUCCESS in 11m 07s\n- requirements-integration http://logs.openstack.org/36/656736/1/check/requirements-integration/f701a69/ : FAILURE in 29m 28s\n- neutron-grenade-multinode http://logs.openstack.org/36/656736/1/check/neutron-grenade-multinode/32d6e5b/ : FAILURE in 1h 00m 58s\n- tempest-full http://logs.openstack.org/36/656736/1/check/tempest-full/009351f/ : FAILURE in 31m 41s\n- neutron-grenade http://logs.openstack.org/36/656736/1/check/neutron-grenade/0411021/ : FAILURE in 48m 52s","accounts_in_message":[],"_revision_number":1},{"id":"113f77374aac6e49434fa7622e5d858844aa12e2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-02 17:48:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"faa5924b9de3721e15bcd7fdc529ddb161188a1d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-02 17:49:59.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"e21a60603c8a0e18c5acff440de19babedec7c82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 20:28:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/36/656736/2/check/openstack-tox-docs/8e4422a/html/ : SUCCESS in 3m 23s\n- openstack-tox-pypy http://logs.openstack.org/36/656736/2/check/openstack-tox-pypy/c281282/ : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/36/656736/2/check/openstack-tox-pep8/cbfdfe2/ : SUCCESS in 2m 32s\n- openstack-tox-py27 http://logs.openstack.org/36/656736/2/check/openstack-tox-py27/4a7dc6d/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/36/656736/2/check/openstack-tox-py36/49e539f/ : SUCCESS in 2m 58s\n- openstack-tox-validate http://logs.openstack.org/36/656736/2/check/openstack-tox-validate/c4ce535/ : SUCCESS in 3m 30s\n- requirements-tox-babel http://logs.openstack.org/36/656736/2/check/requirements-tox-babel/55f4772/ : SUCCESS in 2m 54s\n- requirements-tox-py27-check-uc http://logs.openstack.org/36/656736/2/check/requirements-tox-py27-check-uc/881cc61/ : SUCCESS in 9m 04s\n- requirements-tox-py36-check-uc http://logs.openstack.org/36/656736/2/check/requirements-tox-py36-check-uc/d7a2d73/ : SUCCESS in 8m 00s\n- cross-cinder-py27 http://logs.openstack.org/36/656736/2/check/cross-cinder-py27/14c66cd/ : SUCCESS in 8m 59s\n- cross-cinder-py36 http://logs.openstack.org/36/656736/2/check/cross-cinder-py36/fbd8ac0/ : SUCCESS in 9m 41s\n- cross-glance-py27 http://logs.openstack.org/36/656736/2/check/cross-glance-py27/e71a8a4/ : SUCCESS in 5m 58s\n- cross-glance-py36 http://logs.openstack.org/36/656736/2/check/cross-glance-py36/86e21db/ : SUCCESS in 4m 17s\n- cross-horizon-py27 http://logs.openstack.org/36/656736/2/check/cross-horizon-py27/432896c/ : SUCCESS in 11m 10s\n- cross-horizon-py36 http://logs.openstack.org/36/656736/2/check/cross-horizon-py36/196af76/ : SUCCESS in 8m 30s\n- cross-horizon-npm http://logs.openstack.org/36/656736/2/check/cross-horizon-npm/ee2aa2f/ : SUCCESS in 5m 26s\n- cross-keystone-py27 http://logs.openstack.org/36/656736/2/check/cross-keystone-py27/ef206d9/ : SUCCESS in 22m 40s\n- cross-keystone-py36 http://logs.openstack.org/36/656736/2/check/cross-keystone-py36/095b129/ : SUCCESS in 21m 49s\n- cross-neutron-py27 http://logs.openstack.org/36/656736/2/check/cross-neutron-py27/90a3939/ : SUCCESS in 22m 44s\n- cross-neutron-py36 http://logs.openstack.org/36/656736/2/check/cross-neutron-py36/f52db46/ : SUCCESS in 23m 29s\n- cross-nova-py27 http://logs.openstack.org/36/656736/2/check/cross-nova-py27/5230707/ : SUCCESS in 12m 20s\n- cross-nova-py36 http://logs.openstack.org/36/656736/2/check/cross-nova-py36/caa31b3/ : SUCCESS in 10m 59s\n- cross-nova-functional http://logs.openstack.org/36/656736/2/check/cross-nova-functional/ab88c71/ : SUCCESS in 18m 44s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/36/656736/2/check/cross-osc-build-sphinx-docs/5b3f60c/html/ : SUCCESS in 5m 32s\n- cross-swift-py27 http://logs.openstack.org/36/656736/2/check/cross-swift-py27/9ecd1e1/ : SUCCESS in 7m 15s\n- cross-swift-py36 http://logs.openstack.org/36/656736/2/check/cross-swift-py36/9d256fa/ : SUCCESS in 7m 26s\n- requirements-integration http://logs.openstack.org/36/656736/2/check/requirements-integration/d06ebd0/ : SUCCESS in 1h 08m 32s\n- neutron-grenade-multinode http://logs.openstack.org/36/656736/2/check/neutron-grenade-multinode/433e437/ : SUCCESS in 1h 13m 06s\n- tempest-full http://logs.openstack.org/36/656736/2/check/tempest-full/288053f/ : SUCCESS in 1h 57m 00s\n- neutron-grenade http://logs.openstack.org/36/656736/2/check/neutron-grenade/11f562f/ : FAILURE in 11m 09s","accounts_in_message":[],"_revision_number":2},{"id":"12198face866067d48c3cfc3066b9e249b81941a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-02 20:30:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"408d8b4f10a1d2b5a9f7f235cc41e73fb3e3274c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 22:34:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/36/656736/2/check/openstack-tox-docs/aafcf9c/html/ : SUCCESS in 3m 52s\n- openstack-tox-pypy http://logs.openstack.org/36/656736/2/check/openstack-tox-pypy/4c2abd4/ : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/36/656736/2/check/openstack-tox-pep8/a95e70c/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/36/656736/2/check/openstack-tox-py27/3e40139/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/36/656736/2/check/openstack-tox-py36/1d5e197/ : SUCCESS in 5m 01s\n- openstack-tox-validate http://logs.openstack.org/36/656736/2/check/openstack-tox-validate/bab74c4/ : SUCCESS in 4m 14s\n- requirements-tox-babel http://logs.openstack.org/36/656736/2/check/requirements-tox-babel/810bdc4/ : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc http://logs.openstack.org/36/656736/2/check/requirements-tox-py27-check-uc/9ea21d6/ : SUCCESS in 9m 22s\n- requirements-tox-py36-check-uc http://logs.openstack.org/36/656736/2/check/requirements-tox-py36-check-uc/0d47769/ : SUCCESS in 10m 42s\n- cross-cinder-py27 http://logs.openstack.org/36/656736/2/check/cross-cinder-py27/e4fcc43/ : SUCCESS in 11m 05s\n- cross-cinder-py36 http://logs.openstack.org/36/656736/2/check/cross-cinder-py36/0c75330/ : SUCCESS in 9m 58s\n- cross-glance-py27 http://logs.openstack.org/36/656736/2/check/cross-glance-py27/3b5d286/ : SUCCESS in 5m 43s\n- cross-glance-py36 http://logs.openstack.org/36/656736/2/check/cross-glance-py36/1c11a7b/ : SUCCESS in 5m 22s\n- cross-horizon-py27 http://logs.openstack.org/36/656736/2/check/cross-horizon-py27/0e60f0a/ : SUCCESS in 9m 46s\n- cross-horizon-py36 http://logs.openstack.org/36/656736/2/check/cross-horizon-py36/c984991/ : SUCCESS in 7m 26s\n- cross-horizon-npm http://logs.openstack.org/36/656736/2/check/cross-horizon-npm/52d0dc4/ : SUCCESS in 7m 20s\n- cross-keystone-py27 http://logs.openstack.org/36/656736/2/check/cross-keystone-py27/63c43e1/ : SUCCESS in 32m 07s\n- cross-keystone-py36 http://logs.openstack.org/36/656736/2/check/cross-keystone-py36/768ebad/ : SUCCESS in 29m 48s\n- cross-neutron-py27 http://logs.openstack.org/36/656736/2/check/cross-neutron-py27/ff83f65/ : SUCCESS in 33m 27s\n- cross-neutron-py36 http://logs.openstack.org/36/656736/2/check/cross-neutron-py36/b72edc6/ : SUCCESS in 30m 24s\n- cross-nova-py27 http://logs.openstack.org/36/656736/2/check/cross-nova-py27/72e781d/ : SUCCESS in 23m 52s\n- cross-nova-py36 http://logs.openstack.org/36/656736/2/check/cross-nova-py36/63315a7/ : SUCCESS in 16m 42s\n- cross-nova-functional http://logs.openstack.org/36/656736/2/check/cross-nova-functional/b7911be/ : SUCCESS in 27m 37s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/36/656736/2/check/cross-osc-build-sphinx-docs/b1c2e2f/html/ : SUCCESS in 8m 10s\n- cross-swift-py27 http://logs.openstack.org/36/656736/2/check/cross-swift-py27/741835c/ : SUCCESS in 9m 25s\n- cross-swift-py36 http://logs.openstack.org/36/656736/2/check/cross-swift-py36/1c4052b/ : SUCCESS in 9m 02s\n- requirements-integration http://logs.openstack.org/36/656736/2/check/requirements-integration/2859eeb/ : SUCCESS in 1h 19m 49s\n- neutron-grenade-multinode http://logs.openstack.org/36/656736/2/check/neutron-grenade-multinode/06c9df1/ : SUCCESS in 1h 12m 49s\n- tempest-full http://logs.openstack.org/36/656736/2/check/tempest-full/bdc01cc/ : FAILURE in 1h 59m 36s\n- neutron-grenade http://logs.openstack.org/36/656736/2/check/neutron-grenade/8e18900/ : SUCCESS in 56m 49s","accounts_in_message":[],"_revision_number":2},{"id":"809471b211c29a530278fbb8e7fd0bcf3d95341a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-03 05:54:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"83742eff929518b5913e62e93fab3a6c70227ed0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 07:44:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/656736/2/check/openstack-tox-docs/775aee1/html/ : SUCCESS in 3m 12s\n- openstack-tox-pypy http://logs.openstack.org/36/656736/2/check/openstack-tox-pypy/9bbfbca/ : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/36/656736/2/check/openstack-tox-pep8/bff875c/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/36/656736/2/check/openstack-tox-py27/2cddebe/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/36/656736/2/check/openstack-tox-py36/d22dfcc/ : SUCCESS in 3m 34s\n- openstack-tox-validate http://logs.openstack.org/36/656736/2/check/openstack-tox-validate/56047e9/ : SUCCESS in 3m 18s\n- requirements-tox-babel http://logs.openstack.org/36/656736/2/check/requirements-tox-babel/27af3f9/ : SUCCESS in 4m 43s\n- requirements-tox-py27-check-uc http://logs.openstack.org/36/656736/2/check/requirements-tox-py27-check-uc/53f6a3e/ : SUCCESS in 9m 11s\n- requirements-tox-py36-check-uc http://logs.openstack.org/36/656736/2/check/requirements-tox-py36-check-uc/75e1ea1/ : SUCCESS in 7m 02s\n- cross-cinder-py27 http://logs.openstack.org/36/656736/2/check/cross-cinder-py27/bb5af87/ : SUCCESS in 11m 07s\n- cross-cinder-py36 http://logs.openstack.org/36/656736/2/check/cross-cinder-py36/3bc2ec0/ : SUCCESS in 11m 25s\n- cross-glance-py27 http://logs.openstack.org/36/656736/2/check/cross-glance-py27/060063d/ : SUCCESS in 4m 27s\n- cross-glance-py36 http://logs.openstack.org/36/656736/2/check/cross-glance-py36/1e51601/ : SUCCESS in 4m 04s\n- cross-horizon-py27 http://logs.openstack.org/36/656736/2/check/cross-horizon-py27/cc0f0fe/ : SUCCESS in 10m 27s\n- cross-horizon-py36 http://logs.openstack.org/36/656736/2/check/cross-horizon-py36/5e58639/ : SUCCESS in 7m 18s\n- cross-horizon-npm http://logs.openstack.org/36/656736/2/check/cross-horizon-npm/0890f1f/ : SUCCESS in 6m 12s\n- cross-keystone-py27 http://logs.openstack.org/36/656736/2/check/cross-keystone-py27/2cdfa9a/ : SUCCESS in 19m 56s\n- cross-keystone-py36 http://logs.openstack.org/36/656736/2/check/cross-keystone-py36/0a116ce/ : SUCCESS in 22m 21s\n- cross-neutron-py27 http://logs.openstack.org/36/656736/2/check/cross-neutron-py27/717c107/ : SUCCESS in 22m 24s\n- cross-neutron-py36 http://logs.openstack.org/36/656736/2/check/cross-neutron-py36/8b7d32c/ : SUCCESS in 21m 38s\n- cross-nova-py27 http://logs.openstack.org/36/656736/2/check/cross-nova-py27/2b1a2be/ : SUCCESS in 14m 56s\n- cross-nova-py36 http://logs.openstack.org/36/656736/2/check/cross-nova-py36/a12bf82/ : SUCCESS in 12m 09s\n- cross-nova-functional http://logs.openstack.org/36/656736/2/check/cross-nova-functional/6fd31d0/ : SUCCESS in 17m 41s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/36/656736/2/check/cross-osc-build-sphinx-docs/20bed82/html/ : SUCCESS in 5m 01s\n- cross-swift-py27 http://logs.openstack.org/36/656736/2/check/cross-swift-py27/8df9647/ : SUCCESS in 8m 39s\n- cross-swift-py36 http://logs.openstack.org/36/656736/2/check/cross-swift-py36/c7be42d/ : SUCCESS in 9m 12s\n- requirements-integration http://logs.openstack.org/36/656736/2/check/requirements-integration/faefb70/ : SUCCESS in 1h 10m 32s\n- neutron-grenade-multinode http://logs.openstack.org/36/656736/2/check/neutron-grenade-multinode/9b9cbd9/ : SUCCESS in 1h 07m 14s\n- tempest-full http://logs.openstack.org/36/656736/2/check/tempest-full/b66c9b9/ : SUCCESS in 1h 44m 53s\n- neutron-grenade http://logs.openstack.org/36/656736/2/check/neutron-grenade/750411c/ : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":2},{"id":"27ff43134f7f3769b1edb52d200d44ff3810fabb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 23:31:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"663e5bf5c88bce799d00eb0cdca7a777ba13e9f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 02:40:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/36/656736/2/gate/openstack-tox-docs/253f4db/html/ : SUCCESS in 3m 02s\n- openstack-tox-pep8 http://logs.openstack.org/36/656736/2/gate/openstack-tox-pep8/e26c9ad/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/36/656736/2/gate/openstack-tox-py27/17794e9/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/36/656736/2/gate/openstack-tox-py36/21afcc3/ : SUCCESS in 3m 42s\n- openstack-tox-validate http://logs.openstack.org/36/656736/2/gate/openstack-tox-validate/a689a88/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/36/656736/2/gate/requirements-tox-babel/504e1aa/ : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc http://logs.openstack.org/36/656736/2/gate/requirements-tox-py27-check-uc/3e39f2a/ : SUCCESS in 7m 48s\n- requirements-tox-py36-check-uc http://logs.openstack.org/36/656736/2/gate/requirements-tox-py36-check-uc/dd391b3/ : SUCCESS in 7m 58s\n- cross-cinder-py27 http://logs.openstack.org/36/656736/2/gate/cross-cinder-py27/b047e8d/ : SUCCESS in 10m 07s\n- cross-cinder-py36 http://logs.openstack.org/36/656736/2/gate/cross-cinder-py36/cd8eb6a/ : SUCCESS in 9m 31s\n- cross-glance-py27 http://logs.openstack.org/36/656736/2/gate/cross-glance-py27/cbbb6e9/ : SUCCESS in 4m 18s\n- cross-glance-py36 http://logs.openstack.org/36/656736/2/gate/cross-glance-py36/d2292f0/ : SUCCESS in 4m 55s\n- cross-horizon-py27 http://logs.openstack.org/36/656736/2/gate/cross-horizon-py27/ddf7387/ : SUCCESS in 9m 50s\n- cross-horizon-py36 http://logs.openstack.org/36/656736/2/gate/cross-horizon-py36/ae1692a/ : SUCCESS in 8m 46s\n- cross-horizon-npm http://logs.openstack.org/36/656736/2/gate/cross-horizon-npm/6fa8815/ : SUCCESS in 6m 09s\n- cross-keystone-py27 http://logs.openstack.org/36/656736/2/gate/cross-keystone-py27/d09a39f/ : SUCCESS in 24m 25s\n- cross-keystone-py36 http://logs.openstack.org/36/656736/2/gate/cross-keystone-py36/065c6e6/ : SUCCESS in 24m 12s\n- cross-neutron-py27 http://logs.openstack.org/36/656736/2/gate/cross-neutron-py27/cf1f666/ : SUCCESS in 30m 27s\n- cross-neutron-py36 http://logs.openstack.org/36/656736/2/gate/cross-neutron-py36/c0c39c8/ : SUCCESS in 24m 30s\n- cross-nova-py27 http://logs.openstack.org/36/656736/2/gate/cross-nova-py27/2c1bdf7/ : SUCCESS in 13m 27s\n- cross-nova-py36 http://logs.openstack.org/36/656736/2/gate/cross-nova-py36/81cc862/ : SUCCESS in 14m 40s\n- cross-nova-functional http://logs.openstack.org/36/656736/2/gate/cross-nova-functional/6dd2ca2/ : SUCCESS in 18m 57s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/36/656736/2/gate/cross-osc-build-sphinx-docs/1e92ff1/html/ : SUCCESS in 4m 51s\n- cross-swift-py27 http://logs.openstack.org/36/656736/2/gate/cross-swift-py27/d0ddbfc/ : SUCCESS in 7m 51s\n- cross-swift-py36 http://logs.openstack.org/36/656736/2/gate/cross-swift-py36/0e33f63/ : SUCCESS in 8m 37s\n- requirements-integration http://logs.openstack.org/36/656736/2/gate/requirements-integration/68a3be2/ : SUCCESS in 1h 15m 00s\n- neutron-grenade-multinode http://logs.openstack.org/36/656736/2/gate/neutron-grenade-multinode/ad0e9f0/ : SUCCESS in 1h 07m 07s\n- tempest-full http://logs.openstack.org/36/656736/2/gate/tempest-full/44fbbf0/ : SUCCESS in 1h 45m 07s\n- neutron-grenade http://logs.openstack.org/36/656736/2/gate/neutron-grenade/7c48101/ : SUCCESS in 56m 55s","accounts_in_message":[],"_revision_number":2},{"id":"457c251ce3c8723ac608a64cd0b51c523d47dc09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 02:40:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc07718c9c043bf52d2d0f7206d711a362251e16","revisions":{"934dbce3c487d1d84afd29be72a1693915a24db3":{"kind":"REWORK","_number":1,"created":"2019-05-02 06:11:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/656736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/656736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/656736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/656736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/656736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/656736/1"}}},"commit":{"parents":[{"commit":"75a27084c5af4d43dcc96161f0f5bfe72f1b1115","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75a27084c5af4d43dcc96161f0f5bfe72f1b1115"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-02 06:11:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-02 06:11:08.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I3067f250cf62c40b0f16117b73b0e4580de6c370\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/934dbce3c487d1d84afd29be72a1693915a24db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/934dbce3c487d1d84afd29be72a1693915a24db3"}]},"branch":"refs/heads/master"},"fc07718c9c043bf52d2d0f7206d711a362251e16":{"kind":"REWORK","_number":2,"created":"2019-05-02 17:48:45.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/36/656736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/656736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/656736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/656736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/656736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/656736/2"}}},"commit":{"parents":[{"commit":"e08870c9a38fa8a066ae535c1f7bb3b6a0593809","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e08870c9a38fa8a066ae535c1f7bb3b6a0593809"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-02 06:11:08.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-02 17:48:18.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nDepends-On: https://review.opendev.org/656667\n\nChange-Id: I3067f250cf62c40b0f16117b73b0e4580de6c370\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc07718c9c043bf52d2d0f7206d711a362251e16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc07718c9c043bf52d2d0f7206d711a362251e16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
