)]}'
{"id":"openstack%2Frequirements~656924","triplet_id":"openstack%2Frequirements~master~I69d32ea9d94ec2c34ed173b84a291d91da1b93e2","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I69d32ea9d94ec2c34ed173b84a291d91da1b93e2","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-05-03 06:26:00.000000000","updated":"2019-05-05 02:50:28.000000000","submitted":"2019-05-05 02:50:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656924-1557024628408-7175b152","meta_rev_id":"3df10ad6186ecdfe5bd7dba2e949864479606300","_number":656924,"virtual_id_number":656924,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-05 02:50:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-05 00:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-05 00:52:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-05 00:52:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-05 02:50:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45ee8d58794e1906efe5bbf503ccd0c6bb3166f0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-03 06:26:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08c2b8ca344617c66472da39e7b01fdfa16b4889","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 07:20:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/24/656924/1/check/openstack-tox-docs/8ea3735/html/ : SUCCESS in 2m 45s\n- openstack-tox-pypy http://logs.openstack.org/24/656924/1/check/openstack-tox-pypy/cf44c80/ : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/24/656924/1/check/openstack-tox-pep8/cfd40b2/ : SUCCESS in 2m 33s\n- openstack-tox-py27 http://logs.openstack.org/24/656924/1/check/openstack-tox-py27/aeb0bf7/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/24/656924/1/check/openstack-tox-py36/a0c0b48/ : SUCCESS in 3m 04s\n- openstack-tox-validate http://logs.openstack.org/24/656924/1/check/openstack-tox-validate/09b256b/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/24/656924/1/check/requirements-tox-babel/cf57587/ : SUCCESS in 2m 38s\n- requirements-tox-py27-check-uc http://logs.openstack.org/24/656924/1/check/requirements-tox-py27-check-uc/9bfcefa/ : FAILURE in 13m 12s\n- requirements-tox-py36-check-uc http://logs.openstack.org/24/656924/1/check/requirements-tox-py36-check-uc/ace3d25/ : FAILURE in 13m 33s\n- cross-cinder-py27 http://logs.openstack.org/24/656924/1/check/cross-cinder-py27/bc6a78d/ : SUCCESS in 8m 25s\n- cross-cinder-py36 http://logs.openstack.org/24/656924/1/check/cross-cinder-py36/97457a3/ : SUCCESS in 7m 49s\n- cross-glance-py27 http://logs.openstack.org/24/656924/1/check/cross-glance-py27/c53eda1/ : SUCCESS in 4m 27s\n- cross-glance-py36 http://logs.openstack.org/24/656924/1/check/cross-glance-py36/89e6895/ : SUCCESS in 4m 19s\n- cross-horizon-py27 http://logs.openstack.org/24/656924/1/check/cross-horizon-py27/e3b65a5/ : SUCCESS in 10m 26s\n- cross-horizon-py36 http://logs.openstack.org/24/656924/1/check/cross-horizon-py36/054804c/ : SUCCESS in 9m 08s\n- cross-horizon-npm http://logs.openstack.org/24/656924/1/check/cross-horizon-npm/edfc763/ : SUCCESS in 5m 19s\n- cross-keystone-py27 http://logs.openstack.org/24/656924/1/check/cross-keystone-py27/f7f134e/ : FAILURE in 21m 37s\n- cross-keystone-py36 http://logs.openstack.org/24/656924/1/check/cross-keystone-py36/070c184/ : FAILURE in 22m 21s\n- cross-neutron-py27 http://logs.openstack.org/24/656924/1/check/cross-neutron-py27/53130e4/ : SUCCESS in 21m 05s\n- cross-neutron-py36 http://logs.openstack.org/24/656924/1/check/cross-neutron-py36/b28d05c/ : SUCCESS in 21m 10s\n- cross-nova-py27 http://logs.openstack.org/24/656924/1/check/cross-nova-py27/d5d8274/ : FAILURE in 4m 54s\n- cross-nova-py36 http://logs.openstack.org/24/656924/1/check/cross-nova-py36/5869f54/ : FAILURE in 4m 59s\n- cross-nova-functional http://logs.openstack.org/24/656924/1/check/cross-nova-functional/3a7692e/ : FAILURE in 16m 49s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/24/656924/1/check/cross-osc-build-sphinx-docs/452eda6/html/ : SUCCESS in 5m 19s\n- cross-swift-py27 http://logs.openstack.org/24/656924/1/check/cross-swift-py27/489dcbb/ : SUCCESS in 7m 16s\n- cross-swift-py36 http://logs.openstack.org/24/656924/1/check/cross-swift-py36/33e3c51/ : SUCCESS in 8m 08s\n- requirements-integration http://logs.openstack.org/24/656924/1/check/requirements-integration/6128c3e/ : FAILURE in 30m 13s\n- neutron-grenade-multinode http://logs.openstack.org/24/656924/1/check/neutron-grenade-multinode/497088b/ : FAILURE in 53m 13s\n- tempest-full http://logs.openstack.org/24/656924/1/check/tempest-full/cc796a4/ : FAILURE in 30m 17s\n- neutron-grenade http://logs.openstack.org/24/656924/1/check/neutron-grenade/61efd55/ : FAILURE in 44m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c117c99e11b29dd0d7c1dfdd5245635cd417309b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-04 06:19:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0973fc9a1948bfb1e030f07dbf008eab7f7432fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 07:24:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/24/656924/2/check/openstack-tox-docs/9c5123a/html/ : SUCCESS in 4m 05s\n- openstack-tox-pypy http://logs.openstack.org/24/656924/2/check/openstack-tox-pypy/b381a93/ : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/24/656924/2/check/openstack-tox-pep8/bdb2ed8/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/24/656924/2/check/openstack-tox-py27/ca3f952/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/24/656924/2/check/openstack-tox-py36/5d57207/ : SUCCESS in 4m 47s\n- openstack-tox-validate http://logs.openstack.org/24/656924/2/check/openstack-tox-validate/8039455/ : SUCCESS in 3m 57s\n- requirements-tox-babel http://logs.openstack.org/24/656924/2/check/requirements-tox-babel/33e3a22/ : SUCCESS in 2m 48s\n- requirements-tox-py27-check-uc http://logs.openstack.org/24/656924/2/check/requirements-tox-py27-check-uc/e3a50a5/ : FAILURE in 13m 24s\n- requirements-tox-py36-check-uc http://logs.openstack.org/24/656924/2/check/requirements-tox-py36-check-uc/074c146/ : FAILURE in 12m 33s\n- cross-cinder-py27 http://logs.openstack.org/24/656924/2/check/cross-cinder-py27/9e8a730/ : SUCCESS in 9m 04s\n- cross-cinder-py36 http://logs.openstack.org/24/656924/2/check/cross-cinder-py36/f31d1f3/ : SUCCESS in 8m 35s\n- cross-glance-py27 http://logs.openstack.org/24/656924/2/check/cross-glance-py27/81990ec/ : SUCCESS in 5m 30s\n- cross-glance-py36 http://logs.openstack.org/24/656924/2/check/cross-glance-py36/93ff977/ : SUCCESS in 4m 31s\n- cross-horizon-py27 http://logs.openstack.org/24/656924/2/check/cross-horizon-py27/1778bf1/ : SUCCESS in 15m 09s\n- cross-horizon-py36 http://logs.openstack.org/24/656924/2/check/cross-horizon-py36/9458100/ : SUCCESS in 8m 26s\n- cross-horizon-npm http://logs.openstack.org/24/656924/2/check/cross-horizon-npm/b39d6a4/ : SUCCESS in 7m 54s\n- cross-keystone-py27 http://logs.openstack.org/24/656924/2/check/cross-keystone-py27/c9dd5fd/ : SUCCESS in 21m 54s\n- cross-keystone-py36 http://logs.openstack.org/24/656924/2/check/cross-keystone-py36/7a76697/ : SUCCESS in 20m 43s\n- cross-neutron-py27 http://logs.openstack.org/24/656924/2/check/cross-neutron-py27/7e09ee0/ : SUCCESS in 34m 52s\n- cross-neutron-py36 http://logs.openstack.org/24/656924/2/check/cross-neutron-py36/86fc1ee/ : SUCCESS in 31m 51s\n- cross-nova-py27 http://logs.openstack.org/24/656924/2/check/cross-nova-py27/941e3d8/ : FAILURE in 4m 47s\n- cross-nova-py36 http://logs.openstack.org/24/656924/2/check/cross-nova-py36/7c36de9/ : FAILURE in 4m 47s\n- cross-nova-functional http://logs.openstack.org/24/656924/2/check/cross-nova-functional/82e3767/ : FAILURE in 16m 11s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/24/656924/2/check/cross-osc-build-sphinx-docs/27b9247/html/ : SUCCESS in 5m 30s\n- cross-swift-py27 http://logs.openstack.org/24/656924/2/check/cross-swift-py27/167e8cd/ : SUCCESS in 8m 43s\n- cross-swift-py36 http://logs.openstack.org/24/656924/2/check/cross-swift-py36/a884c91/ : SUCCESS in 7m 05s\n- requirements-integration http://logs.openstack.org/24/656924/2/check/requirements-integration/d6dbe18/ : FAILURE in 26m 42s\n- neutron-grenade-multinode http://logs.openstack.org/24/656924/2/check/neutron-grenade-multinode/3e448fc/ : FAILURE in 59m 08s\n- tempest-full http://logs.openstack.org/24/656924/2/check/tempest-full/767f7c7/ : FAILURE in 25m 41s\n- neutron-grenade http://logs.openstack.org/24/656924/2/check/neutron-grenade/a690434/ : FAILURE in 47m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c4ef2c17985ee560c2a71d41a0b523cfa6e6056b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-04 19:21:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ea9360b3303b7c1656f0de3c261f9aa86014ff7d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-04 19:21:35.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"e91953113dd0a9d6bd2abf089cd97c04694cfa25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 20:54:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/24/656924/3/check/openstack-tox-docs/782d37e/html/ : SUCCESS in 3m 15s\n- openstack-tox-pypy http://logs.openstack.org/24/656924/3/check/openstack-tox-pypy/b2068b2/ : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/24/656924/3/check/openstack-tox-pep8/1509500/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/24/656924/3/check/openstack-tox-py27/23d470b/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/24/656924/3/check/openstack-tox-py36/1685cd7/ : SUCCESS in 3m 03s\n- openstack-tox-validate http://logs.openstack.org/24/656924/3/check/openstack-tox-validate/4dae8e9/ : SUCCESS in 3m 11s\n- requirements-tox-babel http://logs.openstack.org/24/656924/3/check/requirements-tox-babel/34f3690/ : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/24/656924/3/check/requirements-tox-py27-check-uc/dceaaf2/ : SUCCESS in 13m 26s\n- requirements-tox-py36-check-uc http://logs.openstack.org/24/656924/3/check/requirements-tox-py36-check-uc/f435ee3/ : SUCCESS in 11m 19s\n- cross-cinder-py27 http://logs.openstack.org/24/656924/3/check/cross-cinder-py27/56aae69/ : SUCCESS in 9m 05s\n- cross-cinder-py36 http://logs.openstack.org/24/656924/3/check/cross-cinder-py36/0707305/ : SUCCESS in 10m 22s\n- cross-glance-py27 http://logs.openstack.org/24/656924/3/check/cross-glance-py27/0138bb9/ : SUCCESS in 4m 48s\n- cross-glance-py36 http://logs.openstack.org/24/656924/3/check/cross-glance-py36/6137a44/ : SUCCESS in 4m 36s\n- cross-horizon-py27 http://logs.openstack.org/24/656924/3/check/cross-horizon-py27/51753e5/ : SUCCESS in 10m 23s\n- cross-horizon-py36 http://logs.openstack.org/24/656924/3/check/cross-horizon-py36/a7a1284/ : SUCCESS in 7m 24s\n- cross-horizon-npm http://logs.openstack.org/24/656924/3/check/cross-horizon-npm/d05576f/ : SUCCESS in 5m 48s\n- cross-keystone-py27 http://logs.openstack.org/24/656924/3/check/cross-keystone-py27/2cd73e3/ : SUCCESS in 22m 09s\n- cross-keystone-py36 http://logs.openstack.org/24/656924/3/check/cross-keystone-py36/db99064/ : SUCCESS in 22m 53s\n- cross-neutron-py27 http://logs.openstack.org/24/656924/3/check/cross-neutron-py27/4f4dbe8/ : SUCCESS in 24m 04s\n- cross-neutron-py36 http://logs.openstack.org/24/656924/3/check/cross-neutron-py36/18aa1d6/ : SUCCESS in 20m 18s\n- cross-nova-py27 http://logs.openstack.org/24/656924/3/check/cross-nova-py27/22a5c03/ : SUCCESS in 12m 32s\n- cross-nova-py36 http://logs.openstack.org/24/656924/3/check/cross-nova-py36/3b1bf54/ : SUCCESS in 11m 57s\n- cross-nova-functional http://logs.openstack.org/24/656924/3/check/cross-nova-functional/0ebc3a2/ : SUCCESS in 17m 05s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/24/656924/3/check/cross-osc-build-sphinx-docs/0cd6a78/html/ : SUCCESS in 7m 22s\n- cross-swift-py27 http://logs.openstack.org/24/656924/3/check/cross-swift-py27/947c312/ : SUCCESS in 8m 49s\n- cross-swift-py36 http://logs.openstack.org/24/656924/3/check/cross-swift-py36/0222bc4/ : SUCCESS in 7m 12s\n- requirements-integration http://logs.openstack.org/24/656924/3/check/requirements-integration/3c02e71/ : FAILURE in 29m 10s\n- neutron-grenade-multinode http://logs.openstack.org/24/656924/3/check/neutron-grenade-multinode/915667b/ : FAILURE in 32m 01s\n- tempest-full http://logs.openstack.org/24/656924/3/check/tempest-full/e1180e1/ : SUCCESS in 1h 31m 27s\n- neutron-grenade http://logs.openstack.org/24/656924/3/check/neutron-grenade/12a1840/ : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":3},{"id":"cfcace043075a82881961ecc87a6f08487e8ce89","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-04 20:56:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"facb85f5562d2c0530a42b4781cd677610094f9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 22:42:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/24/656924/3/check/openstack-tox-docs/85ac294/html/ : SUCCESS in 3m 00s\n- openstack-tox-pypy http://logs.openstack.org/24/656924/3/check/openstack-tox-pypy/32e0779/ : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/24/656924/3/check/openstack-tox-pep8/cd19bfa/ : SUCCESS in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/24/656924/3/check/openstack-tox-py27/fa1ffbf/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/24/656924/3/check/openstack-tox-py36/f758c83/ : SUCCESS in 3m 24s\n- openstack-tox-validate http://logs.openstack.org/24/656924/3/check/openstack-tox-validate/abddaf1/ : SUCCESS in 2m 59s\n- requirements-tox-babel http://logs.openstack.org/24/656924/3/check/requirements-tox-babel/76c356d/ : SUCCESS in 2m 53s\n- requirements-tox-py27-check-uc http://logs.openstack.org/24/656924/3/check/requirements-tox-py27-check-uc/e1331ac/ : SUCCESS in 10m 26s\n- requirements-tox-py36-check-uc http://logs.openstack.org/24/656924/3/check/requirements-tox-py36-check-uc/8ecf537/ : SUCCESS in 12m 22s\n- cross-cinder-py27 http://logs.openstack.org/24/656924/3/check/cross-cinder-py27/85d89d0/ : SUCCESS in 11m 13s\n- cross-cinder-py36 http://logs.openstack.org/24/656924/3/check/cross-cinder-py36/f631710/ : SUCCESS in 10m 16s\n- cross-glance-py27 http://logs.openstack.org/24/656924/3/check/cross-glance-py27/976164f/ : SUCCESS in 4m 34s\n- cross-glance-py36 http://logs.openstack.org/24/656924/3/check/cross-glance-py36/2ca532e/ : SUCCESS in 4m 14s\n- cross-horizon-py27 http://logs.openstack.org/24/656924/3/check/cross-horizon-py27/9790ec5/ : SUCCESS in 9m 58s\n- cross-horizon-py36 http://logs.openstack.org/24/656924/3/check/cross-horizon-py36/bf1999e/ : SUCCESS in 8m 29s\n- cross-horizon-npm http://logs.openstack.org/24/656924/3/check/cross-horizon-npm/f89b447/ : SUCCESS in 5m 25s\n- cross-keystone-py27 http://logs.openstack.org/24/656924/3/check/cross-keystone-py27/15681ee/ : SUCCESS in 21m 44s\n- cross-keystone-py36 http://logs.openstack.org/24/656924/3/check/cross-keystone-py36/1bb4d09/ : SUCCESS in 23m 57s\n- cross-neutron-py27 http://logs.openstack.org/24/656924/3/check/cross-neutron-py27/8120165/ : SUCCESS in 23m 07s\n- cross-neutron-py36 http://logs.openstack.org/24/656924/3/check/cross-neutron-py36/b1d771c/ : SUCCESS in 29m 16s\n- cross-nova-py27 http://logs.openstack.org/24/656924/3/check/cross-nova-py27/8e406eb/ : SUCCESS in 13m 18s\n- cross-nova-py36 http://logs.openstack.org/24/656924/3/check/cross-nova-py36/27d2702/ : SUCCESS in 11m 09s\n- cross-nova-functional http://logs.openstack.org/24/656924/3/check/cross-nova-functional/9a0993e/ : SUCCESS in 16m 23s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/24/656924/3/check/cross-osc-build-sphinx-docs/df43846/html/ : SUCCESS in 5m 26s\n- cross-swift-py27 http://logs.openstack.org/24/656924/3/check/cross-swift-py27/56bad7b/ : SUCCESS in 7m 50s\n- cross-swift-py36 http://logs.openstack.org/24/656924/3/check/cross-swift-py36/d20b50f/ : SUCCESS in 7m 47s\n- requirements-integration http://logs.openstack.org/24/656924/3/check/requirements-integration/2e696df/ : SUCCESS in 1h 12m 58s\n- neutron-grenade-multinode http://logs.openstack.org/24/656924/3/check/neutron-grenade-multinode/16fc401/ : SUCCESS in 58m 31s\n- tempest-full http://logs.openstack.org/24/656924/3/check/tempest-full/9c4ecae/ : SUCCESS in 1h 45m 02s\n- neutron-grenade http://logs.openstack.org/24/656924/3/check/neutron-grenade/75f3063/ : SUCCESS in 48m 10s","accounts_in_message":[],"_revision_number":3},{"id":"a607e806cce8839cb850bda2c2d0bbdc02251931","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-05 00:52:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"174f3e98dcab2215c5c2a5e18d16a43727df1335","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 00:52:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"04fbc426307ef869d2feb9e8c3d331f97e9a1195","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 02:50:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3df10ad6186ecdfe5bd7dba2e949864479606300","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 02:50:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/24/656924/3/gate/openstack-tox-docs/008f48d/html/ : SUCCESS in 5m 23s\n- openstack-tox-pep8 http://logs.openstack.org/24/656924/3/gate/openstack-tox-pep8/9d48911/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/24/656924/3/gate/openstack-tox-py27/b9c581e/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/24/656924/3/gate/openstack-tox-py36/d20a1f7/ : SUCCESS in 3m 23s\n- openstack-tox-validate http://logs.openstack.org/24/656924/3/gate/openstack-tox-validate/6aa59a0/ : SUCCESS in 2m 58s\n- requirements-tox-babel http://logs.openstack.org/24/656924/3/gate/requirements-tox-babel/9149ef7/ : SUCCESS in 2m 34s\n- requirements-tox-py27-check-uc http://logs.openstack.org/24/656924/3/gate/requirements-tox-py27-check-uc/b2ceb31/ : SUCCESS in 14m 41s\n- requirements-tox-py36-check-uc http://logs.openstack.org/24/656924/3/gate/requirements-tox-py36-check-uc/6a246cf/ : SUCCESS in 12m 57s\n- cross-cinder-py27 http://logs.openstack.org/24/656924/3/gate/cross-cinder-py27/d7bf95f/ : SUCCESS in 10m 51s\n- cross-cinder-py36 http://logs.openstack.org/24/656924/3/gate/cross-cinder-py36/c32dd16/ : SUCCESS in 11m 05s\n- cross-glance-py27 http://logs.openstack.org/24/656924/3/gate/cross-glance-py27/4d7c168/ : SUCCESS in 5m 04s\n- cross-glance-py36 http://logs.openstack.org/24/656924/3/gate/cross-glance-py36/681ffe9/ : SUCCESS in 4m 36s\n- cross-horizon-py27 http://logs.openstack.org/24/656924/3/gate/cross-horizon-py27/f7ad1cf/ : SUCCESS in 10m 30s\n- cross-horizon-py36 http://logs.openstack.org/24/656924/3/gate/cross-horizon-py36/7d01242/ : SUCCESS in 10m 46s\n- cross-horizon-npm http://logs.openstack.org/24/656924/3/gate/cross-horizon-npm/f27b002/ : SUCCESS in 7m 06s\n- cross-keystone-py27 http://logs.openstack.org/24/656924/3/gate/cross-keystone-py27/dbd4ca2/ : SUCCESS in 25m 56s\n- cross-keystone-py36 http://logs.openstack.org/24/656924/3/gate/cross-keystone-py36/d505be8/ : SUCCESS in 25m 52s\n- cross-neutron-py27 http://logs.openstack.org/24/656924/3/gate/cross-neutron-py27/486562b/ : SUCCESS in 22m 28s\n- cross-neutron-py36 http://logs.openstack.org/24/656924/3/gate/cross-neutron-py36/1645588/ : SUCCESS in 24m 40s\n- cross-nova-py27 http://logs.openstack.org/24/656924/3/gate/cross-nova-py27/fb67730/ : SUCCESS in 12m 31s\n- cross-nova-py36 http://logs.openstack.org/24/656924/3/gate/cross-nova-py36/baecca5/ : SUCCESS in 12m 46s\n- cross-nova-functional http://logs.openstack.org/24/656924/3/gate/cross-nova-functional/a2c3d07/ : SUCCESS in 19m 00s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/24/656924/3/gate/cross-osc-build-sphinx-docs/67c1941/html/ : SUCCESS in 6m 07s\n- cross-swift-py27 http://logs.openstack.org/24/656924/3/gate/cross-swift-py27/3bda18e/ : SUCCESS in 9m 02s\n- cross-swift-py36 http://logs.openstack.org/24/656924/3/gate/cross-swift-py36/5ce42cb/ : SUCCESS in 7m 18s\n- requirements-integration http://logs.openstack.org/24/656924/3/gate/requirements-integration/7da426b/ : SUCCESS in 1h 05m 22s\n- neutron-grenade-multinode http://logs.openstack.org/24/656924/3/gate/neutron-grenade-multinode/8b12379/ : SUCCESS in 1h 12m 27s\n- tempest-full http://logs.openstack.org/24/656924/3/gate/tempest-full/b709b80/ : SUCCESS in 1h 50m 15s\n- neutron-grenade http://logs.openstack.org/24/656924/3/gate/neutron-grenade/a627a7e/ : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6affabb05381072509f382af865a54a7916b7664","revisions":{"005a7db5f4eb1dbb4d0be4a9daaecf977df58aee":{"kind":"REWORK","_number":1,"created":"2019-05-03 06:26:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/656924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/656924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/656924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/656924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/656924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/656924/1"}}},"commit":{"parents":[{"commit":"e08870c9a38fa8a066ae535c1f7bb3b6a0593809","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e08870c9a38fa8a066ae535c1f7bb3b6a0593809"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-03 06:25:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-03 06:25:55.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I69d32ea9d94ec2c34ed173b84a291d91da1b93e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/005a7db5f4eb1dbb4d0be4a9daaecf977df58aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/005a7db5f4eb1dbb4d0be4a9daaecf977df58aee"}]},"branch":"refs/heads/master"},"ee773777f8bd90f6c0de2e093b8550c9a2c3cb22":{"kind":"REWORK","_number":2,"created":"2019-05-04 06:19:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/656924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/656924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/656924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/656924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/656924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/656924/2"}}},"commit":{"parents":[{"commit":"fc07718c9c043bf52d2d0f7206d711a362251e16","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc07718c9c043bf52d2d0f7206d711a362251e16"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-04 06:19:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-04 06:19:34.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I69d32ea9d94ec2c34ed173b84a291d91da1b93e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee773777f8bd90f6c0de2e093b8550c9a2c3cb22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee773777f8bd90f6c0de2e093b8550c9a2c3cb22"}]},"branch":"refs/heads/master"},"6affabb05381072509f382af865a54a7916b7664":{"kind":"REWORK","_number":3,"created":"2019-05-04 19:21:33.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/24/656924/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/656924/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/656924/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/656924/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/656924/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/656924/3"}}},"commit":{"parents":[{"commit":"fc07718c9c043bf52d2d0f7206d711a362251e16","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc07718c9c043bf52d2d0f7206d711a362251e16"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-04 06:19:34.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-04 19:21:22.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I69d32ea9d94ec2c34ed173b84a291d91da1b93e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6affabb05381072509f382af865a54a7916b7664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6affabb05381072509f382af865a54a7916b7664"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
