)]}'
{"id":"openstack%2Frequirements~657080","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ib8191602f47641eb047ebe5972ffa92bcfd08de4","project":"openstack/requirements","branch":"stable/stein","hashtags":[],"change_id":"Ib8191602f47641eb047ebe5972ffa92bcfd08de4","subject":"Add safety check output to the linters output","status":"MERGED","created":"2019-05-03 22:56:49.000000000","updated":"2019-05-15 22:18:58.000000000","submitted":"2019-05-15 22:18:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657080-1557958738028-ec554898","meta_rev_id":"12011a324c0e034156527a95ab9ff6390f040ffc","_number":657080,"virtual_id_number":657080,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-05-13 20:16:34.000000000","_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":2,"date":"2019-05-15 22:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-15 19:43:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-15 04:27:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-05-15 10:15:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-15 19:43:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 12:01:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-15 04:27:09.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-15 10:15:52.000000000","updated_by":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"reviewer":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"state":"REVIEWER"},{"updated":"2019-05-15 19:43:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-15 22:18:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8cd1e7c28a67751a6e1cf1fb04c1b172d9eec310","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-03 22:56:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eecbf82b781a973ff58bc601df1cd42e43145f48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 00:52:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/80/657080/1/check/openstack-tox-pypy/2d4cd2e/ : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/80/657080/1/check/openstack-tox-pep8/930a609/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/80/657080/1/check/openstack-tox-py27/2e61fe5/ : SUCCESS in 3m 33s\n- openstack-tox-py35 http://logs.openstack.org/80/657080/1/check/openstack-tox-py35/8b8778c/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/80/657080/1/check/openstack-tox-py36/6792fab/ : SUCCESS in 3m 10s\n- openstack-tox-validate http://logs.openstack.org/80/657080/1/check/openstack-tox-validate/3c619f8/ : SUCCESS in 3m 00s\n- requirements-tox-babel http://logs.openstack.org/80/657080/1/check/requirements-tox-babel/141c9df/ : SUCCESS in 3m 45s\n- requirements-integration http://logs.openstack.org/80/657080/1/check/requirements-integration/671c287/ : SUCCESS in 1h 09m 55s\n- neutron-grenade-multinode http://logs.openstack.org/80/657080/1/check/neutron-grenade-multinode/3689a71/ : SUCCESS in 1h 05m 40s\n- tempest-full http://logs.openstack.org/80/657080/1/check/tempest-full/5b6e123/ : SUCCESS in 1h 49m 44s\n- neutron-grenade http://logs.openstack.org/80/657080/1/check/neutron-grenade/ab1dce2/ : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":1},{"id":"1374cb8b3f7f2e9cd84a449629cbb79fcc91aa1d","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-04 10:24:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7971d7b546815e5b8492dc4c4ac1375c8c0b0f94","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-04 10:24:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8e2f4ed2648f49b2c6878804caf9a5ca98fe67d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 12:08:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/80/657080/3/check/openstack-tox-pypy/6bb946d/ : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/80/657080/3/check/openstack-tox-pep8/4bd4713/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/80/657080/3/check/openstack-tox-py27/5a921e4/ : SUCCESS in 3m 15s\n- openstack-tox-py35 http://logs.openstack.org/80/657080/3/check/openstack-tox-py35/4ffc1da/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/80/657080/3/check/openstack-tox-py36/09c0b0e/ : SUCCESS in 3m 10s\n- openstack-tox-validate http://logs.openstack.org/80/657080/3/check/openstack-tox-validate/d8ee948/ : FAILURE in 2m 41s\n- requirements-tox-babel http://logs.openstack.org/80/657080/3/check/requirements-tox-babel/b7ca2a2/ : SUCCESS in 2m 55s\n- requirements-tox-py27-check-uc http://logs.openstack.org/80/657080/3/check/requirements-tox-py27-check-uc/9cded9e/ : FAILURE in 6m 29s\n- requirements-tox-py35-check-uc http://logs.openstack.org/80/657080/3/check/requirements-tox-py35-check-uc/045d1ac/ : SUCCESS in 6m 32s\n- requirements-tox-py36-check-uc http://logs.openstack.org/80/657080/3/check/requirements-tox-py36-check-uc/6d8a132/ : SUCCESS in 6m 54s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/80/657080/3/check/cross-cinder-py27/f3116a1/ : SUCCESS in 7m 51s\n- cross-cinder-py35 http://logs.openstack.org/80/657080/3/check/cross-cinder-py35/eb35cc0/ : SUCCESS in 8m 33s\n- cross-glance-py27 http://logs.openstack.org/80/657080/3/check/cross-glance-py27/8167cf9/ : SUCCESS in 3m 49s\n- cross-glance-py35 http://logs.openstack.org/80/657080/3/check/cross-glance-py35/f229918/ : SUCCESS in 3m 49s\n- cross-horizon-py27 http://logs.openstack.org/80/657080/3/check/cross-horizon-py27/41de43c/ : SUCCESS in 9m 55s\n- cross-horizon-py35 http://logs.openstack.org/80/657080/3/check/cross-horizon-py35/27e01b7/ : SUCCESS in 9m 00s\n- cross-horizon-npm http://logs.openstack.org/80/657080/3/check/cross-horizon-npm/e690f24/ : SUCCESS in 5m 43s\n- cross-keystone-py27 http://logs.openstack.org/80/657080/3/check/cross-keystone-py27/298db30/ : SUCCESS in 19m 20s\n- cross-keystone-py35 http://logs.openstack.org/80/657080/3/check/cross-keystone-py35/8770ab0/ : SUCCESS in 21m 10s\n- cross-neutron-py27 http://logs.openstack.org/80/657080/3/check/cross-neutron-py27/b09db5d/ : SUCCESS in 18m 07s\n- cross-neutron-py35 http://logs.openstack.org/80/657080/3/check/cross-neutron-py35/3781390/ : SUCCESS in 22m 05s\n- cross-nova-py27 http://logs.openstack.org/80/657080/3/check/cross-nova-py27/b2cc66f/ : SUCCESS in 11m 39s\n- cross-nova-py35 http://logs.openstack.org/80/657080/3/check/cross-nova-py35/be13a30/ : SUCCESS in 11m 08s\n- cross-nova-functional http://logs.openstack.org/80/657080/3/check/cross-nova-functional/7ebab82/ : SUCCESS in 18m 09s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/80/657080/3/check/cross-osc-build-sphinx-docs/aa278fd/html/ : SUCCESS in 4m 39s\n- cross-swift-py27 http://logs.openstack.org/80/657080/3/check/cross-swift-py27/d8048b9/ : SUCCESS in 8m 20s\n- cross-swift-py35 http://logs.openstack.org/80/657080/3/check/cross-swift-py35/890c335/ : SUCCESS in 8m 11s\n- requirements-integration http://logs.openstack.org/80/657080/3/check/requirements-integration/3143c1b/ : SUCCESS in 1h 03m 47s\n- neutron-grenade-multinode http://logs.openstack.org/80/657080/3/check/neutron-grenade-multinode/a4426a7/ : SUCCESS in 1h 03m 41s\n- tempest-full http://logs.openstack.org/80/657080/3/check/tempest-full/7b8c825/ : SUCCESS in 1h 38m 24s\n- neutron-grenade http://logs.openstack.org/80/657080/3/check/neutron-grenade/53bdd2c/ : SUCCESS in 46m 40s","accounts_in_message":[],"_revision_number":3},{"id":"6bacf76859068a99f13da030c40acae204b08d85","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-08 20:28:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5e4ece2ee59b13c2d0a855cf0d4202c0a4efd2a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 22:34:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/80/657080/4/check/openstack-tox-pypy/ac3f3a2/ : FAILURE in 2m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/80/657080/4/check/openstack-tox-pep8/67ab5a5/ : SUCCESS in 2m 54s\n- openstack-tox-py27 http://logs.openstack.org/80/657080/4/check/openstack-tox-py27/8dfa9d9/ : FAILURE in 2m 30s\n- openstack-tox-py35 http://logs.openstack.org/80/657080/4/check/openstack-tox-py35/beb4cd6/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/80/657080/4/check/openstack-tox-py36/497def7/ : SUCCESS in 3m 06s\n- openstack-tox-validate http://logs.openstack.org/80/657080/4/check/openstack-tox-validate/edbd634/ : SUCCESS in 2m 46s\n- requirements-tox-babel http://logs.openstack.org/80/657080/4/check/requirements-tox-babel/0173b13/ : SUCCESS in 3m 14s\n- requirements-integration http://logs.openstack.org/80/657080/4/check/requirements-integration/3fce58d/ : FAILURE in 30m 08s\n- neutron-grenade-multinode http://logs.openstack.org/80/657080/4/check/neutron-grenade-multinode/93439ad/ : SUCCESS in 1h 05m 11s\n- tempest-full http://logs.openstack.org/80/657080/4/check/tempest-full/c698592/ : SUCCESS in 1h 59m 37s\n- neutron-grenade http://logs.openstack.org/80/657080/4/check/neutron-grenade/f1de6d7/ : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":4},{"id":"87eb7b7ab4e596b281ed908d991ee6684da6bc9d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-12 12:01:39.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"263a4ee18eb35db6b3674936c3d2529013f4610d","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 13:27:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7e753f79d806f505b80a306bdb0c58dc1f06804a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 14:31:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/80/657080/5/check/openstack-tox-pypy/df3ddc4/ : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/80/657080/5/check/openstack-tox-pep8/da6e687/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/80/657080/5/check/openstack-tox-py27/b2e7895/ : SUCCESS in 3m 01s\n- openstack-tox-py35 http://logs.openstack.org/80/657080/5/check/openstack-tox-py35/3179e33/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/80/657080/5/check/openstack-tox-py36/becc41d/ : SUCCESS in 3m 30s\n- openstack-tox-validate http://logs.openstack.org/80/657080/5/check/openstack-tox-validate/106994f/ : SUCCESS in 2m 47s\n- requirements-tox-babel http://logs.openstack.org/80/657080/5/check/requirements-tox-babel/75477f1/ : SUCCESS in 2m 36s\n- requirements-integration http://logs.openstack.org/80/657080/5/check/requirements-integration/09564c3/ : SUCCESS in 47m 01s\n- neutron-grenade-multinode http://logs.openstack.org/80/657080/5/check/neutron-grenade-multinode/6eb0041/ : SUCCESS in 1h 02m 39s\n- tempest-full http://logs.openstack.org/80/657080/5/check/tempest-full/5fc914c/ : FAILURE in 27m 47s\n- neutron-grenade http://logs.openstack.org/80/657080/5/check/neutron-grenade/6ed3808/ : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":5},{"id":"afa6f925e5c150f4ec74e162d77f0374e0b11f73","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 20:16:34.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"f2129b81cfb5a74299836f98d976d12519caf9bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 00:20:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/80/657080/5/check/openstack-tox-pypy/2b52555/ : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/80/657080/5/check/openstack-tox-pep8/2b8fa3b/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/80/657080/5/check/openstack-tox-py27/9318d71/ : SUCCESS in 3m 15s\n- openstack-tox-py35 http://logs.openstack.org/80/657080/5/check/openstack-tox-py35/470ef10/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/80/657080/5/check/openstack-tox-py36/de6f23d/ : SUCCESS in 3m 23s\n- openstack-tox-validate http://logs.openstack.org/80/657080/5/check/openstack-tox-validate/82afc6e/ : SUCCESS in 3m 08s\n- requirements-tox-babel http://logs.openstack.org/80/657080/5/check/requirements-tox-babel/4cd5753/ : SUCCESS in 3m 00s\n- requirements-integration http://logs.openstack.org/80/657080/5/check/requirements-integration/97f1e5b/ : SUCCESS in 43m 40s\n- neutron-grenade-multinode http://logs.openstack.org/80/657080/5/check/neutron-grenade-multinode/f0ddba3/ : SUCCESS in 1h 18m 40s\n- tempest-full http://logs.openstack.org/80/657080/5/check/tempest-full/42bd356/ : SUCCESS in 2h 04m 11s\n- neutron-grenade http://logs.openstack.org/80/657080/5/check/neutron-grenade/3e4111d/ : SUCCESS in 56m 21s","accounts_in_message":[],"_revision_number":5},{"id":"34d0d26bcc433d56d6715c51fa38f1a003eb3dba","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-15 04:27:09.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"8dec9f623e9f5f69957d9434e8ddcaee610fa06a","author":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"date":"2019-05-15 10:15:52.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"bccbea904fd76406589e602a20e282b01c35ef2b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-15 19:43:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e0a4955d050f380764862d4d756ba4d92e2eb73a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 19:43:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"162511c3201704368839f770a2d104238cf35943","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 22:18:57.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/80/657080/5/gate/openstack-tox-pep8/66fd925/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/80/657080/5/gate/openstack-tox-py27/9faecaf/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/80/657080/5/gate/openstack-tox-py35/835dd04/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/80/657080/5/gate/openstack-tox-py36/021f95c/ : SUCCESS in 4m 36s\n- openstack-tox-validate http://logs.openstack.org/80/657080/5/gate/openstack-tox-validate/feeb44c/ : SUCCESS in 4m 10s\n- requirements-tox-babel http://logs.openstack.org/80/657080/5/gate/requirements-tox-babel/1761cb0/ : SUCCESS in 4m 16s\n- requirements-integration http://logs.openstack.org/80/657080/5/gate/requirements-integration/bb63118/ : SUCCESS in 1h 00m 20s\n- neutron-grenade-multinode http://logs.openstack.org/80/657080/5/gate/neutron-grenade-multinode/3a022fa/ : SUCCESS in 1h 10m 24s\n- tempest-full http://logs.openstack.org/80/657080/5/gate/tempest-full/ec1bf72/ : SUCCESS in 1h 33m 03s\n- neutron-grenade http://logs.openstack.org/80/657080/5/gate/neutron-grenade/b640671/ : SUCCESS in 49m 08s","accounts_in_message":[],"_revision_number":5},{"id":"12011a324c0e034156527a95ab9ff6390f040ffc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 22:18:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f6ed502758d9ade38584a74fdec018cf6cf31449","revisions":{"87fa7b1e5b8c14efedff32c7834d9389de53e2e4":{"kind":"REWORK","_number":1,"created":"2019-05-03 22:56:49.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/80/657080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/657080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/657080/1"}}},"commit":{"parents":[{"commit":"1baf07a77333fc57d9d4a669d0265b0e687fd031","subject":"Merge \"update constraint for openstack-release-test to new release 1.4.2\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1baf07a77333fc57d9d4a669d0265b0e687fd031"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-03 22:56:15.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-03 22:56:40.000000000","tz":120},"subject":"DNM: do safety check","message":"DNM: do safety check\n\nChange-Id: Ib8191602f47641eb047ebe5972ffa92bcfd08de4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87fa7b1e5b8c14efedff32c7834d9389de53e2e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87fa7b1e5b8c14efedff32c7834d9389de53e2e4"}]},"branch":"refs/heads/stable/stein"},"f9655852d427e63722b7452b6d097d26c68134bb":{"kind":"REWORK","_number":2,"created":"2019-05-04 10:24:04.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/80/657080/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/657080/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/657080/2"}}},"commit":{"parents":[{"commit":"1baf07a77333fc57d9d4a669d0265b0e687fd031","subject":"Merge \"update constraint for openstack-release-test to new release 1.4.2\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1baf07a77333fc57d9d4a669d0265b0e687fd031"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-03 22:56:15.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-04 10:23:54.000000000","tz":120},"subject":"DNM: do safety check","message":"DNM: do safety check\n\nChange-Id: Ib8191602f47641eb047ebe5972ffa92bcfd08de4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9655852d427e63722b7452b6d097d26c68134bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9655852d427e63722b7452b6d097d26c68134bb"}]},"branch":"refs/heads/stable/stein"},"40a59e5ee663a79f1b9b8635565aea711fdc54c9":{"kind":"REWORK","_number":3,"created":"2019-05-04 10:24:29.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/80/657080/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/657080/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/657080/3"}}},"commit":{"parents":[{"commit":"1baf07a77333fc57d9d4a669d0265b0e687fd031","subject":"Merge \"update constraint for openstack-release-test to new release 1.4.2\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1baf07a77333fc57d9d4a669d0265b0e687fd031"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-03 22:56:15.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-04 10:24:19.000000000","tz":120},"subject":"DNM: do safety check","message":"DNM: do safety check\n\nChange-Id: Ib8191602f47641eb047ebe5972ffa92bcfd08de4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40a59e5ee663a79f1b9b8635565aea711fdc54c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40a59e5ee663a79f1b9b8635565aea711fdc54c9"}]},"branch":"refs/heads/stable/stein"},"38fc4bf7caa0ba717c2d04aa1b2645c2eec7cca0":{"kind":"REWORK","_number":4,"created":"2019-05-08 20:28:01.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/80/657080/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/657080/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/657080/4"}}},"commit":{"parents":[{"commit":"1baf07a77333fc57d9d4a669d0265b0e687fd031","subject":"Merge \"update constraint for openstack-release-test to new release 1.4.2\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1baf07a77333fc57d9d4a669d0265b0e687fd031"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-03 22:56:15.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-08 20:27:40.000000000","tz":120},"subject":"Add safety check output to the linters output","message":"Add safety check output to the linters output\n\nCurrently a safety check failure is ignored, the output\nis just for informational purposes. it reports known\nsecurity issues with the versions in upper-constraints.txt\nso that we can determine a corrective action.\n\nChange-Id: Ib8191602f47641eb047ebe5972ffa92bcfd08de4\n(cherry picked from commit a09abdc96c5782d4e17575a8fcc7dc820c7e039f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38fc4bf7caa0ba717c2d04aa1b2645c2eec7cca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38fc4bf7caa0ba717c2d04aa1b2645c2eec7cca0"}]},"branch":"refs/heads/stable/stein"},"f6ed502758d9ade38584a74fdec018cf6cf31449":{"kind":"REWORK","_number":5,"created":"2019-05-12 13:27:45.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/80/657080/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/657080/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/657080/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/657080/5"}}},"commit":{"parents":[{"commit":"1baf07a77333fc57d9d4a669d0265b0e687fd031","subject":"Merge \"update constraint for openstack-release-test to new release 1.4.2\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1baf07a77333fc57d9d4a669d0265b0e687fd031"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-03 22:56:15.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-12 13:27:24.000000000","tz":120},"subject":"Add safety check output to the linters output","message":"Add safety check output to the linters output\n\nCurrently a safety check failure is ignored, the output\nis just for informational purposes. it reports known\nsecurity issues with the versions in upper-constraints.txt\nso that we can determine a corrective action.\n\nChange-Id: Ib8191602f47641eb047ebe5972ffa92bcfd08de4\n(cherry picked from commit b628986912f4cb312c74ff22c1a6370df81bc8d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6ed502758d9ade38584a74fdec018cf6cf31449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6ed502758d9ade38584a74fdec018cf6cf31449"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
