)]}'
{"id":"openstack%2Frequirements~657107","triplet_id":"openstack%2Frequirements~stable%2Frocky~I8cfcc7ddaf59781c498276f66f42a8c6ce335f0b","project":"openstack/requirements","branch":"stable/rocky","topic":"657107","hashtags":[],"change_id":"I8cfcc7ddaf59781c498276f66f42a8c6ce335f0b","subject":"Update django to 1.11.20 / 2.0.13 releases","status":"ABANDONED","created":"2019-05-04 10:55:26.000000000","updated":"2020-05-23 20:18:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5b949bc03c1b7e7c072d2f083ed2bcecad516740","_number":657107,"virtual_id_number":657107,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2019-05-23 00:04:28.000000000","_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"date":"2020-04-17 02:43:41.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-05-13 17:23:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":-2,"date":"2019-09-22 05:21:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-05-15 04:32:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-09-22 09:55:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 17:23:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-15 04:32:27.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-09-22 05:21:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-09-22 09:55:59.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-04-17 02:43:41.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"128ba86063a089275d63163434d27300ea935628","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-04 10:55:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78361a07a726bbd4425f27b55ab9b2b13dd68246","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 12:10:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/07/657107/1/check/openstack-tox-pypy/3957065/ : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/657107/1/check/openstack-tox-pep8/c1913d7/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/07/657107/1/check/openstack-tox-py27/085b92c/ : SUCCESS in 2m 53s\n- openstack-tox-py35 http://logs.openstack.org/07/657107/1/check/openstack-tox-py35/94ff80e/ : SUCCESS in 3m 11s\n- openstack-tox-validate http://logs.openstack.org/07/657107/1/check/openstack-tox-validate/e84706a/ : SUCCESS in 2m 56s\n- requirements-tox-babel http://logs.openstack.org/07/657107/1/check/requirements-tox-babel/229143d/ : SUCCESS in 2m 47s\n- requirements-tox-py27-check-uc http://logs.openstack.org/07/657107/1/check/requirements-tox-py27-check-uc/c898535/ : SUCCESS in 6m 56s\n- requirements-tox-py35-check-uc http://logs.openstack.org/07/657107/1/check/requirements-tox-py35-check-uc/cceebfd/ : SUCCESS in 6m 18s\n- requirements-tox-py36-check-uc http://logs.openstack.org/07/657107/1/check/requirements-tox-py36-check-uc/106e3ee/ : SUCCESS in 8m 12s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/07/657107/1/check/cross-cinder-py27/ed0c9af/ : SUCCESS in 8m 31s\n- cross-cinder-py35 http://logs.openstack.org/07/657107/1/check/cross-cinder-py35/2cfea52/ : SUCCESS in 9m 01s\n- cross-glance-py27 http://logs.openstack.org/07/657107/1/check/cross-glance-py27/48aec4b/ : SUCCESS in 5m 16s\n- cross-glance-py35 http://logs.openstack.org/07/657107/1/check/cross-glance-py35/6cd97c3/ : SUCCESS in 4m 01s\n- cross-horizon-py27 http://logs.openstack.org/07/657107/1/check/cross-horizon-py27/0f8a8d7/ : SUCCESS in 10m 08s\n- cross-horizon-py35 http://logs.openstack.org/07/657107/1/check/cross-horizon-py35/b395ea0/ : SUCCESS in 7m 22s\n- cross-keystone-py27 http://logs.openstack.org/07/657107/1/check/cross-keystone-py27/db7f277/ : SUCCESS in 12m 48s\n- cross-keystone-py35 http://logs.openstack.org/07/657107/1/check/cross-keystone-py35/0a466d6/ : SUCCESS in 13m 32s\n- cross-neutron-py27 http://logs.openstack.org/07/657107/1/check/cross-neutron-py27/1dbd038/ : SUCCESS in 22m 10s\n- cross-neutron-py35 http://logs.openstack.org/07/657107/1/check/cross-neutron-py35/c87940c/ : SUCCESS in 21m 43s\n- cross-nova-py27 http://logs.openstack.org/07/657107/1/check/cross-nova-py27/f9dca56/ : SUCCESS in 12m 19s\n- cross-nova-py35 http://logs.openstack.org/07/657107/1/check/cross-nova-py35/c30761b/ : SUCCESS in 12m 22s\n- cross-nova-functional http://logs.openstack.org/07/657107/1/check/cross-nova-functional/41af0cf/ : SUCCESS in 16m 58s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/07/657107/1/check/cross-osc-build-sphinx-docs/b671f6e/html/ : SUCCESS in 5m 22s\n- cross-swift-py27 http://logs.openstack.org/07/657107/1/check/cross-swift-py27/bcea0ba/ : SUCCESS in 8m 37s\n- cross-swift-py35 http://logs.openstack.org/07/657107/1/check/cross-swift-py35/005f557/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/07/657107/1/check/openstack-tox-py36/30fd542/ : SUCCESS in 3m 14s\n- requirements-integration http://logs.openstack.org/07/657107/1/check/requirements-integration/dabe57f/ : SUCCESS in 1h 05m 07s\n- neutron-grenade-multinode http://logs.openstack.org/07/657107/1/check/neutron-grenade-multinode/7636359/ : SUCCESS in 1h 04m 49s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/07/657107/1/check/legacy-tempest-dsvm-neutron-full/e46fd2c/ : SUCCESS in 1h 13m 42s\n- neutron-grenade http://logs.openstack.org/07/657107/1/check/neutron-grenade/17e6d1b/ : SUCCESS in 52m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7d5ca34cb53bc64e9d4ea76ae34ceea80c0a65d6","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-07 20:54:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd47aa86fe906861c247b07548b282d9ace025f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 22:05:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/07/657107/2/check/openstack-tox-pypy/f27c34b/ : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/657107/2/check/openstack-tox-pep8/6c62be2/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/07/657107/2/check/openstack-tox-py27/ab97e8a/ : SUCCESS in 3m 11s\n- openstack-tox-py35 http://logs.openstack.org/07/657107/2/check/openstack-tox-py35/0889c4f/ : SUCCESS in 2m 51s\n- openstack-tox-validate http://logs.openstack.org/07/657107/2/check/openstack-tox-validate/e259878/ : SUCCESS in 2m 55s\n- requirements-tox-babel http://logs.openstack.org/07/657107/2/check/requirements-tox-babel/2dfdc6b/ : SUCCESS in 2m 27s\n- requirements-tox-py27-check-uc http://logs.openstack.org/07/657107/2/check/requirements-tox-py27-check-uc/a9a68d2/ : SUCCESS in 7m 42s\n- requirements-tox-py35-check-uc http://logs.openstack.org/07/657107/2/check/requirements-tox-py35-check-uc/228d59e/ : SUCCESS in 6m 22s\n- requirements-tox-py36-check-uc http://logs.openstack.org/07/657107/2/check/requirements-tox-py36-check-uc/6a06b4a/ : SUCCESS in 7m 49s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/07/657107/2/check/cross-cinder-py27/d353b30/ : SUCCESS in 8m 42s\n- cross-cinder-py35 http://logs.openstack.org/07/657107/2/check/cross-cinder-py35/7dfea4a/ : SUCCESS in 7m 28s\n- cross-glance-py27 http://logs.openstack.org/07/657107/2/check/cross-glance-py27/7e30995/ : SUCCESS in 5m 18s\n- cross-glance-py35 http://logs.openstack.org/07/657107/2/check/cross-glance-py35/29ab15c/ : SUCCESS in 5m 22s\n- cross-horizon-py27 http://logs.openstack.org/07/657107/2/check/cross-horizon-py27/c9aa19f/ : SUCCESS in 9m 15s\n- cross-horizon-py35 http://logs.openstack.org/07/657107/2/check/cross-horizon-py35/47c7cf9/ : SUCCESS in 8m 44s\n- cross-keystone-py27 http://logs.openstack.org/07/657107/2/check/cross-keystone-py27/f162b02/ : SUCCESS in 13m 23s\n- cross-keystone-py35 http://logs.openstack.org/07/657107/2/check/cross-keystone-py35/76bf95b/ : SUCCESS in 12m 10s\n- cross-neutron-py27 http://logs.openstack.org/07/657107/2/check/cross-neutron-py27/68cce03/ : SUCCESS in 23m 26s\n- cross-neutron-py35 http://logs.openstack.org/07/657107/2/check/cross-neutron-py35/186927a/ : SUCCESS in 29m 14s\n- cross-nova-py27 http://logs.openstack.org/07/657107/2/check/cross-nova-py27/08acb58/ : SUCCESS in 13m 05s\n- cross-nova-py35 http://logs.openstack.org/07/657107/2/check/cross-nova-py35/bbdea41/ : SUCCESS in 12m 27s\n- cross-nova-functional http://logs.openstack.org/07/657107/2/check/cross-nova-functional/e463905/ : SUCCESS in 18m 00s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/07/657107/2/check/cross-osc-build-sphinx-docs/411a07f/html/ : SUCCESS in 5m 18s\n- cross-swift-py27 http://logs.openstack.org/07/657107/2/check/cross-swift-py27/a8f6efb/ : SUCCESS in 8m 05s\n- cross-swift-py35 http://logs.openstack.org/07/657107/2/check/cross-swift-py35/c7c68ca/ : SUCCESS in 4m 37s\n- openstack-tox-py36 http://logs.openstack.org/07/657107/2/check/openstack-tox-py36/4354ea5/ : SUCCESS in 3m 26s\n- requirements-integration http://logs.openstack.org/07/657107/2/check/requirements-integration/69d58d2/ : SUCCESS in 1h 05m 00s\n- neutron-grenade-multinode http://logs.openstack.org/07/657107/2/check/neutron-grenade-multinode/a6a5df4/ : FAILURE in 20m 35s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/07/657107/2/check/legacy-tempest-dsvm-neutron-full/95d8170/ : FAILURE in 19m 26s\n- neutron-grenade http://logs.openstack.org/07/657107/2/check/neutron-grenade/978396f/ : FAILURE in 19m 01s","accounts_in_message":[],"_revision_number":2},{"id":"030ce726afa72ec4804ba7358669a42dcdf59f35","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-08 10:29:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fe9fc5a8c227d677258227fc36bbfa8a0699ba16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 12:08:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/07/657107/2/check/openstack-tox-pypy/1752960/ : SUCCESS in 2m 59s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/657107/2/check/openstack-tox-pep8/ca7564d/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/07/657107/2/check/openstack-tox-py27/d2df300/ : SUCCESS in 3m 17s\n- openstack-tox-py35 http://logs.openstack.org/07/657107/2/check/openstack-tox-py35/f49185e/ : SUCCESS in 3m 53s\n- openstack-tox-validate http://logs.openstack.org/07/657107/2/check/openstack-tox-validate/16bc20f/ : SUCCESS in 2m 28s\n- requirements-tox-babel http://logs.openstack.org/07/657107/2/check/requirements-tox-babel/34d7e90/ : SUCCESS in 3m 26s\n- requirements-tox-py27-check-uc http://logs.openstack.org/07/657107/2/check/requirements-tox-py27-check-uc/264ffde/ : SUCCESS in 8m 02s\n- requirements-tox-py35-check-uc http://logs.openstack.org/07/657107/2/check/requirements-tox-py35-check-uc/78c62c7/ : SUCCESS in 6m 22s\n- requirements-tox-py36-check-uc http://logs.openstack.org/07/657107/2/check/requirements-tox-py36-check-uc/2ef3a2a/ : SUCCESS in 7m 57s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/07/657107/2/check/cross-cinder-py27/50eec0b/ : SUCCESS in 9m 36s\n- cross-cinder-py35 http://logs.openstack.org/07/657107/2/check/cross-cinder-py35/a05c455/ : SUCCESS in 10m 02s\n- cross-glance-py27 http://logs.openstack.org/07/657107/2/check/cross-glance-py27/a1cc9da/ : SUCCESS in 4m 04s\n- cross-glance-py35 http://logs.openstack.org/07/657107/2/check/cross-glance-py35/8b93690/ : SUCCESS in 5m 13s\n- cross-horizon-py27 http://logs.openstack.org/07/657107/2/check/cross-horizon-py27/b111ffb/ : SUCCESS in 9m 52s\n- cross-horizon-py35 http://logs.openstack.org/07/657107/2/check/cross-horizon-py35/7cf231f/ : SUCCESS in 9m 00s\n- cross-keystone-py27 http://logs.openstack.org/07/657107/2/check/cross-keystone-py27/e56b94a/ : SUCCESS in 11m 38s\n- cross-keystone-py35 http://logs.openstack.org/07/657107/2/check/cross-keystone-py35/dd4ba64/ : SUCCESS in 22m 15s\n- cross-neutron-py27 http://logs.openstack.org/07/657107/2/check/cross-neutron-py27/5bf5f2e/ : SUCCESS in 22m 27s\n- cross-neutron-py35 http://logs.openstack.org/07/657107/2/check/cross-neutron-py35/0d560d8/ : SUCCESS in 24m 34s\n- cross-nova-py27 http://logs.openstack.org/07/657107/2/check/cross-nova-py27/d18bd31/ : SUCCESS in 13m 40s\n- cross-nova-py35 http://logs.openstack.org/07/657107/2/check/cross-nova-py35/23737c7/ : SUCCESS in 13m 18s\n- cross-nova-functional http://logs.openstack.org/07/657107/2/check/cross-nova-functional/c731985/ : SUCCESS in 23m 04s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/07/657107/2/check/cross-osc-build-sphinx-docs/362d4fa/html/ : SUCCESS in 4m 42s\n- cross-swift-py27 http://logs.openstack.org/07/657107/2/check/cross-swift-py27/651f597/ : SUCCESS in 8m 32s\n- cross-swift-py35 http://logs.openstack.org/07/657107/2/check/cross-swift-py35/121ea74/ : SUCCESS in 4m 35s\n- openstack-tox-py36 http://logs.openstack.org/07/657107/2/check/openstack-tox-py36/544d155/ : SUCCESS in 3m 15s\n- requirements-integration http://logs.openstack.org/07/657107/2/check/requirements-integration/65bc1b3/ : FAILURE in 31m 38s\n- neutron-grenade-multinode http://logs.openstack.org/07/657107/2/check/neutron-grenade-multinode/e33d97e/ : SUCCESS in 1h 05m 20s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/07/657107/2/check/legacy-tempest-dsvm-neutron-full/131165e/ : SUCCESS in 1h 34m 23s\n- neutron-grenade http://logs.openstack.org/07/657107/2/check/neutron-grenade/b158761/ : SUCCESS in 53m 51s","accounts_in_message":[],"_revision_number":2},{"id":"687df40e488eac60298fd7ab9f526a6ed0d701b2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 05:10:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6f4f88be934c2db18de2dba56811d194cf2aabab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 07:09:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/07/657107/2/check/openstack-tox-pypy/f441a91/ : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/657107/2/check/openstack-tox-pep8/8599933/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/07/657107/2/check/openstack-tox-py27/35e9d1a/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/07/657107/2/check/openstack-tox-py35/326d12c/ : SUCCESS in 3m 38s\n- openstack-tox-validate http://logs.openstack.org/07/657107/2/check/openstack-tox-validate/1e651e5/ : SUCCESS in 3m 26s\n- requirements-tox-babel http://logs.openstack.org/07/657107/2/check/requirements-tox-babel/07886e3/ : SUCCESS in 4m 34s\n- requirements-tox-py27-check-uc http://logs.openstack.org/07/657107/2/check/requirements-tox-py27-check-uc/2e62d06/ : SUCCESS in 5m 59s\n- requirements-tox-py35-check-uc http://logs.openstack.org/07/657107/2/check/requirements-tox-py35-check-uc/e9325a9/ : SUCCESS in 6m 14s\n- requirements-tox-py36-check-uc http://logs.openstack.org/07/657107/2/check/requirements-tox-py36-check-uc/a1c4f76/ : SUCCESS in 6m 55s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/07/657107/2/check/cross-cinder-py27/aadc0ab/ : SUCCESS in 10m 11s\n- cross-cinder-py35 http://logs.openstack.org/07/657107/2/check/cross-cinder-py35/7ca3b74/ : SUCCESS in 8m 51s\n- cross-glance-py27 http://logs.openstack.org/07/657107/2/check/cross-glance-py27/21b891f/ : SUCCESS in 4m 56s\n- cross-glance-py35 http://logs.openstack.org/07/657107/2/check/cross-glance-py35/51748ec/ : SUCCESS in 4m 36s\n- cross-horizon-py27 http://logs.openstack.org/07/657107/2/check/cross-horizon-py27/c4262d8/ : SUCCESS in 9m 23s\n- cross-horizon-py35 http://logs.openstack.org/07/657107/2/check/cross-horizon-py35/9c7c85d/ : SUCCESS in 9m 32s\n- cross-keystone-py27 http://logs.openstack.org/07/657107/2/check/cross-keystone-py27/1b7ad44/ : SUCCESS in 17m 38s\n- cross-keystone-py35 http://logs.openstack.org/07/657107/2/check/cross-keystone-py35/8dee8bc/ : SUCCESS in 18m 57s\n- cross-neutron-py27 http://logs.openstack.org/07/657107/2/check/cross-neutron-py27/4796b28/ : SUCCESS in 21m 49s\n- cross-neutron-py35 http://logs.openstack.org/07/657107/2/check/cross-neutron-py35/77ab4d9/ : SUCCESS in 29m 11s\n- cross-nova-py27 http://logs.openstack.org/07/657107/2/check/cross-nova-py27/1ff5606/ : SUCCESS in 11m 45s\n- cross-nova-py35 http://logs.openstack.org/07/657107/2/check/cross-nova-py35/0251665/ : SUCCESS in 11m 25s\n- cross-nova-functional http://logs.openstack.org/07/657107/2/check/cross-nova-functional/340954c/ : SUCCESS in 16m 52s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/07/657107/2/check/cross-osc-build-sphinx-docs/9bb49e4/html/ : SUCCESS in 4m 57s\n- cross-swift-py27 http://logs.openstack.org/07/657107/2/check/cross-swift-py27/a5f751d/ : SUCCESS in 7m 21s\n- cross-swift-py35 http://logs.openstack.org/07/657107/2/check/cross-swift-py35/1d0ad5a/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/07/657107/2/check/openstack-tox-py36/c5dbf8f/ : SUCCESS in 3m 29s\n- requirements-integration http://logs.openstack.org/07/657107/2/check/requirements-integration/0683b00/ : FAILURE in 31m 13s\n- neutron-grenade-multinode http://logs.openstack.org/07/657107/2/check/neutron-grenade-multinode/319e489/ : SUCCESS in 1h 06m 58s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/07/657107/2/check/legacy-tempest-dsvm-neutron-full/d6be273/ : SUCCESS in 1h 47m 13s\n- neutron-grenade http://logs.openstack.org/07/657107/2/check/neutron-grenade/d4e074c/ : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":2},{"id":"5fe4c3f142e586a6c2eaa709c77f83031a613f2c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 18:55:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d33ead852d2fa42139ad74533f63d59e4a069d48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 20:34:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/07/657107/3/check/openstack-tox-pypy/4748903/ : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/657107/3/check/openstack-tox-pep8/acbae84/ : SUCCESS in 2m 20s\n- openstack-tox-py27 http://logs.openstack.org/07/657107/3/check/openstack-tox-py27/ca45fca/ : SUCCESS in 2m 47s\n- openstack-tox-py35 http://logs.openstack.org/07/657107/3/check/openstack-tox-py35/2a83650/ : SUCCESS in 2m 53s\n- openstack-tox-validate http://logs.openstack.org/07/657107/3/check/openstack-tox-validate/2c36bce/ : SUCCESS in 2m 41s\n- requirements-tox-babel http://logs.openstack.org/07/657107/3/check/requirements-tox-babel/11e73c3/ : SUCCESS in 2m 30s\n- requirements-tox-py27-check-uc http://logs.openstack.org/07/657107/3/check/requirements-tox-py27-check-uc/e0237b8/ : SUCCESS in 6m 21s\n- requirements-tox-py35-check-uc http://logs.openstack.org/07/657107/3/check/requirements-tox-py35-check-uc/74d024c/ : SUCCESS in 6m 50s\n- requirements-tox-py36-check-uc http://logs.openstack.org/07/657107/3/check/requirements-tox-py36-check-uc/11885a2/ : SUCCESS in 6m 11s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/07/657107/3/check/cross-cinder-py27/3eec2cb/ : SUCCESS in 8m 24s\n- cross-cinder-py35 http://logs.openstack.org/07/657107/3/check/cross-cinder-py35/817d566/ : SUCCESS in 8m 55s\n- cross-glance-py27 http://logs.openstack.org/07/657107/3/check/cross-glance-py27/b7af305/ : SUCCESS in 3m 53s\n- cross-glance-py35 http://logs.openstack.org/07/657107/3/check/cross-glance-py35/0d7f024/ : SUCCESS in 4m 33s\n- cross-horizon-py27 http://logs.openstack.org/07/657107/3/check/cross-horizon-py27/9dfce45/ : SUCCESS in 9m 26s\n- cross-horizon-py35 http://logs.openstack.org/07/657107/3/check/cross-horizon-py35/1811213/ : SUCCESS in 9m 03s\n- cross-keystone-py27 http://logs.openstack.org/07/657107/3/check/cross-keystone-py27/cb0d3fb/ : SUCCESS in 11m 15s\n- cross-keystone-py35 http://logs.openstack.org/07/657107/3/check/cross-keystone-py35/f032efa/ : SUCCESS in 11m 12s\n- cross-neutron-py27 http://logs.openstack.org/07/657107/3/check/cross-neutron-py27/85329a4/ : SUCCESS in 25m 21s\n- cross-neutron-py35 http://logs.openstack.org/07/657107/3/check/cross-neutron-py35/7d29441/ : SUCCESS in 23m 59s\n- cross-nova-py27 http://logs.openstack.org/07/657107/3/check/cross-nova-py27/e72cdc2/ : SUCCESS in 12m 09s\n- cross-nova-py35 http://logs.openstack.org/07/657107/3/check/cross-nova-py35/ac4473e/ : SUCCESS in 11m 46s\n- cross-nova-functional http://logs.openstack.org/07/657107/3/check/cross-nova-functional/a58ab75/ : SUCCESS in 17m 48s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/07/657107/3/check/cross-osc-build-sphinx-docs/6c47778/html/ : SUCCESS in 4m 41s\n- cross-swift-py27 http://logs.openstack.org/07/657107/3/check/cross-swift-py27/b318371/ : SUCCESS in 7m 24s\n- cross-swift-py35 http://logs.openstack.org/07/657107/3/check/cross-swift-py35/e519b23/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/07/657107/3/check/openstack-tox-py36/d7b631a/ : SUCCESS in 3m 13s\n- requirements-integration http://logs.openstack.org/07/657107/3/check/requirements-integration/e5f9be3/ : FAILURE in 33m 37s\n- neutron-grenade-multinode http://logs.openstack.org/07/657107/3/check/neutron-grenade-multinode/8b17783/ : SUCCESS in 1h 08m 10s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/07/657107/3/check/legacy-tempest-dsvm-neutron-full/4228e04/ : SUCCESS in 1h 32m 24s\n- neutron-grenade http://logs.openstack.org/07/657107/3/check/neutron-grenade/02c4456/ : SUCCESS in 52m 13s","accounts_in_message":[],"_revision_number":3},{"id":"03e5ffad6f3776ba0b1b1673af779e3302e5e02c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 15:45:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"78f91934d95151e3ea218fca98768cb37f4b31f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 17:23:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/07/657107/3/check/openstack-tox-pypy/17b4484/ : SUCCESS in 2m 58s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/07/657107/3/check/openstack-tox-pep8/920fc39/ : SUCCESS in 6m 12s\n- openstack-tox-py27 http://logs.openstack.org/07/657107/3/check/openstack-tox-py27/a100f98/ : SUCCESS in 3m 19s\n- openstack-tox-py35 http://logs.openstack.org/07/657107/3/check/openstack-tox-py35/5a341ec/ : SUCCESS in 3m 11s\n- openstack-tox-validate http://logs.openstack.org/07/657107/3/check/openstack-tox-validate/84fbf09/ : SUCCESS in 2m 58s\n- requirements-tox-babel http://logs.openstack.org/07/657107/3/check/requirements-tox-babel/9c37f7a/ : SUCCESS in 6m 06s\n- requirements-tox-py27-check-uc http://logs.openstack.org/07/657107/3/check/requirements-tox-py27-check-uc/fd6cedf/ : SUCCESS in 6m 30s\n- requirements-tox-py35-check-uc http://logs.openstack.org/07/657107/3/check/requirements-tox-py35-check-uc/603f0f5/ : SUCCESS in 6m 08s\n- requirements-tox-py36-check-uc http://logs.openstack.org/07/657107/3/check/requirements-tox-py36-check-uc/48d265d/ : SUCCESS in 6m 17s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/07/657107/3/check/cross-cinder-py27/6de9c00/ : SUCCESS in 8m 17s\n- cross-cinder-py35 http://logs.openstack.org/07/657107/3/check/cross-cinder-py35/5ca3ba5/ : SUCCESS in 8m 58s\n- cross-glance-py27 http://logs.openstack.org/07/657107/3/check/cross-glance-py27/d9af77b/ : SUCCESS in 4m 05s\n- cross-glance-py35 http://logs.openstack.org/07/657107/3/check/cross-glance-py35/e38e70a/ : SUCCESS in 4m 09s\n- cross-horizon-py27 http://logs.openstack.org/07/657107/3/check/cross-horizon-py27/f98f2bf/ : SUCCESS in 10m 06s\n- cross-horizon-py35 http://logs.openstack.org/07/657107/3/check/cross-horizon-py35/3cebba4/ : SUCCESS in 9m 27s\n- cross-keystone-py27 http://logs.openstack.org/07/657107/3/check/cross-keystone-py27/c0b4ef7/ : SUCCESS in 11m 41s\n- cross-keystone-py35 http://logs.openstack.org/07/657107/3/check/cross-keystone-py35/0d2aecc/ : SUCCESS in 11m 53s\n- cross-neutron-py27 http://logs.openstack.org/07/657107/3/check/cross-neutron-py27/aec2a5f/ : SUCCESS in 23m 23s\n- cross-neutron-py35 http://logs.openstack.org/07/657107/3/check/cross-neutron-py35/77c4111/ : SUCCESS in 22m 01s\n- cross-nova-py27 http://logs.openstack.org/07/657107/3/check/cross-nova-py27/f1d3f03/ : SUCCESS in 17m 18s\n- cross-nova-py35 http://logs.openstack.org/07/657107/3/check/cross-nova-py35/f6128f6/ : SUCCESS in 23m 07s\n- cross-nova-functional http://logs.openstack.org/07/657107/3/check/cross-nova-functional/ca9cfe9/ : SUCCESS in 16m 57s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/07/657107/3/check/cross-osc-build-sphinx-docs/1f08c07/html/ : SUCCESS in 5m 40s\n- cross-swift-py27 http://logs.openstack.org/07/657107/3/check/cross-swift-py27/500ab99/ : SUCCESS in 7m 36s\n- cross-swift-py35 http://logs.openstack.org/07/657107/3/check/cross-swift-py35/27e3192/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/07/657107/3/check/openstack-tox-py36/a84ea54/ : SUCCESS in 3m 46s\n- requirements-integration http://logs.openstack.org/07/657107/3/check/requirements-integration/d311426/ : SUCCESS in 48m 17s\n- neutron-grenade-multinode http://logs.openstack.org/07/657107/3/check/neutron-grenade-multinode/36d1e6c/ : SUCCESS in 58m 23s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/07/657107/3/check/legacy-tempest-dsvm-neutron-full/bb989e1/ : SUCCESS in 1h 25m 25s\n- neutron-grenade http://logs.openstack.org/07/657107/3/check/neutron-grenade/d6f1053/ : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":3},{"id":"2520c456767dd369ec01b9bd60a2190c240d7488","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-15 04:32:27.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks good to me.\n\nIt\u0027s a little unusual but probably in a good way.\n\nI\u0027ll update my +2 to a +2 if we get an ACK from Horizon PTL or see a no-op change that passes","accounts_in_message":[],"_revision_number":3},{"id":"1e09a4e9847f29c0501729e2594b8c47c7f15942","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-23 00:04:28.000000000","message":"Patch Set 3:\n\nTony, I think thats unneeded, horizon never uses the upper-constraints file for determining which django version to install:\n\nhttp://logs.openstack.org/46/651546/2/check/horizon-openstack-tox-python3-django111/7c4499a/job-output.txt.gz#_2019-05-12_09_15_40_546800\n\nhttps://opendev.org/openstack/horizon/src/branch/stable/rocky/tox.ini#L33","accounts_in_message":[],"_revision_number":3},{"id":"78ceee78f4cc14f3deb9a93835ffb2a739b49344","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-22 05:21:57.000000000","message":"Patch Set 3: Code-Review-2\n\nI\u0027m gonna go with what I said in the other (requests iirc) reviews and say that we should rely on downstream packagers to do their job.","accounts_in_message":[],"_revision_number":3},{"id":"88a2502f018426b33b195e41380c3aa84006fa35","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-09-22 09:55:59.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM from horizon point of view","accounts_in_message":[],"_revision_number":3},{"id":"59408c9566a01acbd8a9ac22e7cbffa2a39e7a79","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-04-17 02:43:41.000000000","message":"Patch Set 3:\n\nDirk Mueller wrote:\n\n\u003e Tony, I think thats unneeded, horizon never uses the upper-\n\u003e constraints file for determining which django version to install:\n\u003e\n\u003e http://logs.openstack.org/46/651546/2/check/horizon-openstack-tox-\n\u003e python3-django111/7c4499a/job-\n\u003e output.txt.gz#_2019-05-12_09_15_40_546800\n\u003e\n\u003e https://opendev.org/openstack/horizon/src/branch/stable/rocky/tox.in\n\u003e i#L33\n\nYour statement is not necessarily correct.\n\nNote that horizon uses upper-constraints for the primary Django version\n(Django 1.11 for py2 and Django 2.0 for py3).\nhttps://opendev.org/openstack/horizon/src/branch/stable/rocky/tox.ini#L15\n\nWhat you referred is an additional test coverage (Django 1.11 for py3).\nIn this case, we cannot use upper-constraints to install a different version of Django.","accounts_in_message":[],"_revision_number":3},{"id":"f686669f5bed478af77c7cfd8692802972672118","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-23 20:18:47.000000000","message":"Abandoned\n\nsecurity releases for EM branches and the like should be handled by downstream packages/distibutors (deployment or otherwise).","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7a59a7f8040adfac464ae5a257f80e37a60df9dc","revisions":{"b1a7cc6d858ee06b1e0a0e54949949e424827401":{"kind":"REWORK","_number":1,"created":"2019-05-04 10:55:26.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/07/657107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/657107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/657107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/657107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/657107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/657107/1"}}},"commit":{"parents":[{"commit":"f01c6b2fcb3124220dc08430c37e14ffa04bf632","subject":"Add safety check output to the linters output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f01c6b2fcb3124220dc08430c37e14ffa04bf632"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-04 10:53:33.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-04 10:53:33.000000000","tz":120},"subject":"Update django to 1.11.14 / 2.0.13 releases","message":"Update django to 1.11.14 / 2.0.13 releases\n\nThis provides only data loss or security fixes:\n\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.13/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.12/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.11/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.10/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.9/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.8/\n\n1.11.x is the LTS branch with the same set of fixes.\n\nChange-Id: I8cfcc7ddaf59781c498276f66f42a8c6ce335f0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1a7cc6d858ee06b1e0a0e54949949e424827401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1a7cc6d858ee06b1e0a0e54949949e424827401"}]},"branch":"refs/heads/stable/rocky"},"91e7dde6fb7663c514f6054200e17cac10b520bf":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2019-05-07 20:54:38.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/07/657107/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/657107/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/657107/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/657107/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/657107/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/657107/2"}}},"commit":{"parents":[{"commit":"964f79d543aae22babac1b93ffbd8694f4a64df7","subject":"update git.openstack in _DEFAULT_REQS_DIR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/964f79d543aae22babac1b93ffbd8694f4a64df7"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-04 10:53:33.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-07 20:54:31.000000000","tz":120},"subject":"Update django to 1.11.20 / 2.0.13 releases","message":"Update django to 1.11.20 / 2.0.13 releases\n\nThis provides only data loss or security fixes:\n\nCVE-2018-14574, CVE-2019-3498, CVE-2019-6975\n\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.13/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.12/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.11/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.10/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.9/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.8/\n\n1.11.20 is the LTS branch with the same set of fixes.\n\nChange-Id: I8cfcc7ddaf59781c498276f66f42a8c6ce335f0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91e7dde6fb7663c514f6054200e17cac10b520bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91e7dde6fb7663c514f6054200e17cac10b520bf"}]},"branch":"refs/heads/stable/rocky"},"7a59a7f8040adfac464ae5a257f80e37a60df9dc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-12 18:55:32.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/07/657107/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/657107/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/657107/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/657107/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/657107/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/657107/3"}}},"commit":{"parents":[{"commit":"ab0875127bd23cf18c549440b9e0130d5162b73d","subject":"Add upper-constraints enforcement for local reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab0875127bd23cf18c549440b9e0130d5162b73d"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-04 10:53:33.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-12 18:55:32.000000000","tz":0},"subject":"Update django to 1.11.20 / 2.0.13 releases","message":"Update django to 1.11.20 / 2.0.13 releases\n\nThis provides only data loss or security fixes:\n\nCVE-2018-14574, CVE-2019-3498, CVE-2019-6975\n\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.13/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.12/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.11/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.10/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.9/\nhttps://docs.djangoproject.com/en/2.2/releases/2.0.8/\n\n1.11.20 is the LTS branch with the same set of fixes.\n\nChange-Id: I8cfcc7ddaf59781c498276f66f42a8c6ce335f0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a59a7f8040adfac464ae5a257f80e37a60df9dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a59a7f8040adfac464ae5a257f80e37a60df9dc"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
