)]}'
{"id":"openstack%2Frequirements~657167","triplet_id":"openstack%2Frequirements~master~I1d95e4bbed051ea7e4bb77fb9e38d3006843f998","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I1d95e4bbed051ea7e4bb77fb9e38d3006843f998","subject":"Add Boto3 to global requirements.","status":"MERGED","created":"2019-05-04 21:52:46.000000000","updated":"2019-05-12 02:30:06.000000000","submitted":"2019-05-12 02:30:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657167-1557628206979-4a183efd","meta_rev_id":"6c74c1296e829ae0159a891be0571ab22dfef540","_number":657167,"virtual_id_number":657167,"owner":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"date":"2019-05-09 21:55:06.000000000","_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-05-12 02:30:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-06 23:16:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-07 20:25:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},{"value":1,"date":"2019-05-10 19:51:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-11 21:13:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 20:25:58.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-10 19:51:51.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-05-11 21:13:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-12 02:30:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1502a12098b018c9fcd201317a3df9f5ffcc8d5d","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2019-05-04 21:52:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c07f3cf488ec3ea76b53890e33ceb3b16adcef2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 23:36:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/67/657167/1/check/openstack-tox-docs/add0c14/html/ : SUCCESS in 2m 34s\n- openstack-tox-pypy http://logs.openstack.org/67/657167/1/check/openstack-tox-pypy/f7145ee/ : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/657167/1/check/openstack-tox-pep8/4e5b847/ : SUCCESS in 2m 31s\n- openstack-tox-py27 http://logs.openstack.org/67/657167/1/check/openstack-tox-py27/c5b7fd8/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/67/657167/1/check/openstack-tox-py36/6710e65/ : SUCCESS in 3m 01s\n- openstack-tox-validate http://logs.openstack.org/67/657167/1/check/openstack-tox-validate/4c5e80c/ : FAILURE in 2m 38s\n- requirements-integration http://logs.openstack.org/67/657167/1/check/requirements-integration/ad9a2a0/ : SUCCESS in 1h 13m 37s\n- neutron-grenade-multinode http://logs.openstack.org/67/657167/1/check/neutron-grenade-multinode/f26d088/ : SUCCESS in 1h 03m 49s\n- tempest-full http://logs.openstack.org/67/657167/1/check/tempest-full/ddf93af/ : SUCCESS in 1h 38m 02s\n- neutron-grenade http://logs.openstack.org/67/657167/1/check/neutron-grenade/384fecd/ : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c2854871e4305bf62ceb336f5e1f94352c843d05","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-06 20:42:29.000000000","message":"Patch Set 1: Code-Review-1\n\nthis will need an upper-constraints.txt entry as well\n\nafter that, we will need the questions from https://docs.openstack.org/project-team-guide/dependency-management.html#for-new-requirements","accounts_in_message":[],"_revision_number":1},{"id":"2b2f37cea4ac2d79927f6624b8c3871a69a2f99f","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2019-05-06 23:09:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a069ce3538c184403f34cfd2f57d64c36ed448fb","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2019-05-06 23:10:59.000000000","message":"Patch Set 2:\n\nUpdated upper-constraints.txt and included the answers to the questions about the library in the commit message.","accounts_in_message":[],"_revision_number":2},{"id":"16a0a0a354d0e432436a8dfb1878a2c627b7b644","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-06 23:16:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e7e1155c96bde09fa5b373bbfad7baa6fe03ff39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 01:15:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/67/657167/2/check/openstack-tox-docs/474b8ba/html/ : SUCCESS in 4m 16s\n- openstack-tox-pypy http://logs.openstack.org/67/657167/2/check/openstack-tox-pypy/0a77981/ : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/657167/2/check/openstack-tox-pep8/3cdfcae/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/67/657167/2/check/openstack-tox-py27/a3e3f6d/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/67/657167/2/check/openstack-tox-py36/bd50e68/ : SUCCESS in 4m 58s\n- openstack-tox-validate http://logs.openstack.org/67/657167/2/check/openstack-tox-validate/b583ea8/ : SUCCESS in 2m 44s\n- requirements-tox-babel http://logs.openstack.org/67/657167/2/check/requirements-tox-babel/562f9f5/ : SUCCESS in 2m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/657167/2/check/requirements-tox-py27-check-uc/a927438/ : SUCCESS in 12m 35s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/657167/2/check/requirements-tox-py36-check-uc/f9c4eb3/ : SUCCESS in 11m 40s\n- cross-cinder-py27 http://logs.openstack.org/67/657167/2/check/cross-cinder-py27/685f1a9/ : SUCCESS in 10m 32s\n- cross-cinder-py36 http://logs.openstack.org/67/657167/2/check/cross-cinder-py36/c9bc94a/ : SUCCESS in 7m 44s\n- cross-glance-py27 http://logs.openstack.org/67/657167/2/check/cross-glance-py27/81b8d92/ : SUCCESS in 4m 08s\n- cross-glance-py36 http://logs.openstack.org/67/657167/2/check/cross-glance-py36/adf0702/ : SUCCESS in 5m 45s\n- cross-horizon-py27 http://logs.openstack.org/67/657167/2/check/cross-horizon-py27/5f55eea/ : SUCCESS in 10m 31s\n- cross-horizon-py36 http://logs.openstack.org/67/657167/2/check/cross-horizon-py36/8d0ae6f/ : SUCCESS in 7m 46s\n- cross-horizon-npm http://logs.openstack.org/67/657167/2/check/cross-horizon-npm/fc8d440/ : SUCCESS in 7m 30s\n- cross-keystone-py27 http://logs.openstack.org/67/657167/2/check/cross-keystone-py27/60331bd/ : SUCCESS in 17m 56s\n- cross-keystone-py36 http://logs.openstack.org/67/657167/2/check/cross-keystone-py36/656fb50/ : SUCCESS in 33m 02s\n- cross-neutron-py27 http://logs.openstack.org/67/657167/2/check/cross-neutron-py27/16f1e2e/ : TIMED_OUT in 40m 47s\n- cross-neutron-py36 http://logs.openstack.org/67/657167/2/check/cross-neutron-py36/f9f8190/ : TIMED_OUT in 40m 57s\n- cross-nova-py27 http://logs.openstack.org/67/657167/2/check/cross-nova-py27/788730c/ : SUCCESS in 11m 55s\n- cross-nova-py36 http://logs.openstack.org/67/657167/2/check/cross-nova-py36/bfa9f43/ : SUCCESS in 12m 32s\n- cross-nova-functional http://logs.openstack.org/67/657167/2/check/cross-nova-functional/076da7d/ : SUCCESS in 27m 02s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/657167/2/check/cross-osc-build-sphinx-docs/55efe43/html/ : SUCCESS in 7m 06s\n- cross-swift-py27 http://logs.openstack.org/67/657167/2/check/cross-swift-py27/405d3fc/ : SUCCESS in 13m 11s\n- cross-swift-py36 http://logs.openstack.org/67/657167/2/check/cross-swift-py36/8904700/ : SUCCESS in 12m 42s\n- requirements-integration http://logs.openstack.org/67/657167/2/check/requirements-integration/8f4329b/ : SUCCESS in 1h 38m 58s\n- neutron-grenade-multinode http://logs.openstack.org/67/657167/2/check/neutron-grenade-multinode/309c778/ : SUCCESS in 1h 04m 54s\n- tempest-full http://logs.openstack.org/67/657167/2/check/tempest-full/891ff03/ : TIMED_OUT in 2h 04m 12s\n- neutron-grenade http://logs.openstack.org/67/657167/2/check/neutron-grenade/110a1d1/ : SUCCESS in 1h 11m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4ac5cba6edd779b5e39009944066d171de387c0e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-07 01:49:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"99394ae405be38456e87af1ff6e2cf586a4dd476","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 03:46:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/657167/2/check/openstack-tox-docs/cb945e8/html/ : SUCCESS in 3m 21s\n- openstack-tox-pypy http://logs.openstack.org/67/657167/2/check/openstack-tox-pypy/7840e15/ : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/657167/2/check/openstack-tox-pep8/c99b747/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/67/657167/2/check/openstack-tox-py27/4e90600/ : SUCCESS in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/67/657167/2/check/openstack-tox-py36/c22bd91/ : SUCCESS in 3m 17s\n- openstack-tox-validate http://logs.openstack.org/67/657167/2/check/openstack-tox-validate/5ed7f44/ : SUCCESS in 2m 59s\n- requirements-tox-babel http://logs.openstack.org/67/657167/2/check/requirements-tox-babel/58392f1/ : SUCCESS in 2m 55s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/657167/2/check/requirements-tox-py27-check-uc/86d231c/ : SUCCESS in 12m 17s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/657167/2/check/requirements-tox-py36-check-uc/74a6ad9/ : SUCCESS in 14m 47s\n- cross-cinder-py27 http://logs.openstack.org/67/657167/2/check/cross-cinder-py27/d989d93/ : SUCCESS in 10m 38s\n- cross-cinder-py36 http://logs.openstack.org/67/657167/2/check/cross-cinder-py36/505e44b/ : SUCCESS in 9m 13s\n- cross-glance-py27 http://logs.openstack.org/67/657167/2/check/cross-glance-py27/f367f15/ : SUCCESS in 5m 59s\n- cross-glance-py36 http://logs.openstack.org/67/657167/2/check/cross-glance-py36/2fdc579/ : SUCCESS in 4m 51s\n- cross-horizon-py27 http://logs.openstack.org/67/657167/2/check/cross-horizon-py27/fac3ad9/ : SUCCESS in 9m 11s\n- cross-horizon-py36 http://logs.openstack.org/67/657167/2/check/cross-horizon-py36/f597554/ : SUCCESS in 8m 54s\n- cross-horizon-npm http://logs.openstack.org/67/657167/2/check/cross-horizon-npm/eea9231/ : SUCCESS in 5m 30s\n- cross-keystone-py27 http://logs.openstack.org/67/657167/2/check/cross-keystone-py27/dab3b1a/ : SUCCESS in 24m 45s\n- cross-keystone-py36 http://logs.openstack.org/67/657167/2/check/cross-keystone-py36/cd26f3b/ : SUCCESS in 20m 18s\n- cross-neutron-py27 http://logs.openstack.org/67/657167/2/check/cross-neutron-py27/4c68431/ : SUCCESS in 22m 13s\n- cross-neutron-py36 http://logs.openstack.org/67/657167/2/check/cross-neutron-py36/d64d212/ : SUCCESS in 24m 51s\n- cross-nova-py27 http://logs.openstack.org/67/657167/2/check/cross-nova-py27/a06686e/ : SUCCESS in 17m 21s\n- cross-nova-py36 http://logs.openstack.org/67/657167/2/check/cross-nova-py36/c1a56bf/ : SUCCESS in 16m 50s\n- cross-nova-functional http://logs.openstack.org/67/657167/2/check/cross-nova-functional/be26672/ : SUCCESS in 19m 23s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/657167/2/check/cross-osc-build-sphinx-docs/4276235/html/ : SUCCESS in 6m 25s\n- cross-swift-py27 http://logs.openstack.org/67/657167/2/check/cross-swift-py27/71d08e8/ : SUCCESS in 7m 35s\n- cross-swift-py36 http://logs.openstack.org/67/657167/2/check/cross-swift-py36/7942cca/ : SUCCESS in 15m 13s\n- requirements-integration http://logs.openstack.org/67/657167/2/check/requirements-integration/acecc74/ : SUCCESS in 1h 08m 20s\n- neutron-grenade-multinode http://logs.openstack.org/67/657167/2/check/neutron-grenade-multinode/1b6873b/ : SUCCESS in 1h 03m 24s\n- tempest-full http://logs.openstack.org/67/657167/2/check/tempest-full/2a20b9c/ : SUCCESS in 1h 53m 55s\n- neutron-grenade http://logs.openstack.org/67/657167/2/check/neutron-grenade/ac108a1/ : SUCCESS in 1h 14m 03s","accounts_in_message":[],"_revision_number":2},{"id":"d56509ac11687d38d282cb9591c09b463f603685","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-07 20:25:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0a2139254605031dd75cac8b9fc7818ee2e9dffc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 20:26:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3b92bd04f893c1eabe3e9a8a581397867a3506c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 22:36:01.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/67/657167/2/gate/openstack-tox-docs/7ee88ef/html/ : SUCCESS in 2m 55s\n- openstack-tox-pep8 http://logs.openstack.org/67/657167/2/gate/openstack-tox-pep8/13b3a2a/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/67/657167/2/gate/openstack-tox-py27/2134f23/ : SUCCESS in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/67/657167/2/gate/openstack-tox-py36/c66bd40/ : SUCCESS in 3m 43s\n- openstack-tox-validate http://logs.openstack.org/67/657167/2/gate/openstack-tox-validate/be38e8f/ : SUCCESS in 3m 13s\n- requirements-tox-babel http://logs.openstack.org/67/657167/2/gate/requirements-tox-babel/c836e58/ : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/657167/2/gate/requirements-tox-py27-check-uc/f477d4f/ : SUCCESS in 12m 29s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/657167/2/gate/requirements-tox-py36-check-uc/8186062/ : SUCCESS in 12m 40s\n- cross-cinder-py27 http://logs.openstack.org/67/657167/2/gate/cross-cinder-py27/f11cf7a/ : SUCCESS in 8m 26s\n- cross-cinder-py36 http://logs.openstack.org/67/657167/2/gate/cross-cinder-py36/922cf7f/ : SUCCESS in 8m 12s\n- cross-glance-py27 http://logs.openstack.org/67/657167/2/gate/cross-glance-py27/61a6d1a/ : SUCCESS in 4m 25s\n- cross-glance-py36 http://logs.openstack.org/67/657167/2/gate/cross-glance-py36/2ee79d4/ : SUCCESS in 4m 44s\n- cross-horizon-py27 http://logs.openstack.org/67/657167/2/gate/cross-horizon-py27/ad05ced/ : SUCCESS in 10m 32s\n- cross-horizon-py36 http://logs.openstack.org/67/657167/2/gate/cross-horizon-py36/cb1f444/ : SUCCESS in 7m 59s\n- cross-horizon-npm http://logs.openstack.org/67/657167/2/gate/cross-horizon-npm/33dcd8d/ : RETRY_LIMIT in 5m 29s\n- cross-keystone-py27 http://logs.openstack.org/67/657167/2/gate/cross-keystone-py27/b6231da/ : SUCCESS in 24m 30s\n- cross-keystone-py36 http://logs.openstack.org/67/657167/2/gate/cross-keystone-py36/7cc1661/ : SUCCESS in 28m 44s\n- cross-neutron-py27 http://logs.openstack.org/67/657167/2/gate/cross-neutron-py27/baec053/ : SUCCESS in 20m 03s\n- cross-neutron-py36 http://logs.openstack.org/67/657167/2/gate/cross-neutron-py36/1b5e7bb/ : SUCCESS in 18m 45s\n- cross-nova-py27 http://logs.openstack.org/67/657167/2/gate/cross-nova-py27/06aa1b1/ : SUCCESS in 14m 45s\n- cross-nova-py36 http://logs.openstack.org/67/657167/2/gate/cross-nova-py36/0acdb38/ : SUCCESS in 12m 56s\n- cross-nova-functional http://logs.openstack.org/67/657167/2/gate/cross-nova-functional/7dccd6c/ : FAILURE in 3m 38s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/657167/2/gate/cross-osc-build-sphinx-docs/27f58ec/html/ : SUCCESS in 5m 56s\n- cross-swift-py27 http://logs.openstack.org/67/657167/2/gate/cross-swift-py27/92c2f78/ : SUCCESS in 9m 41s\n- cross-swift-py36 http://logs.openstack.org/67/657167/2/gate/cross-swift-py36/2dc9f0a/ : SUCCESS in 8m 02s\n- requirements-integration http://logs.openstack.org/67/657167/2/gate/requirements-integration/c6e6b00/ : SUCCESS in 1h 07m 42s\n- neutron-grenade-multinode http://logs.openstack.org/67/657167/2/gate/neutron-grenade-multinode/04d2aad/ : FAILURE in 17m 29s\n- tempest-full http://logs.openstack.org/67/657167/2/gate/tempest-full/2ff667e/ : FAILURE in 15m 44s\n- neutron-grenade http://logs.openstack.org/67/657167/2/gate/neutron-grenade/b7d1ddb/ : FAILURE in 13m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e5dc38d50d4217338e1836fa53b45ae66f85e4b5","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-08 10:29:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fc105461b0d9de4517112da1a89dc02fc5e3cddc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 12:21:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/67/657167/2/check/openstack-tox-docs/80d38ed/html/ : SUCCESS in 3m 11s\n- openstack-tox-pypy http://logs.openstack.org/67/657167/2/check/openstack-tox-pypy/8f3f28f/ : FAILURE in 2m 47s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/657167/2/check/openstack-tox-pep8/e124c20/ : SUCCESS in 2m 32s\n- openstack-tox-py27 http://logs.openstack.org/67/657167/2/check/openstack-tox-py27/5ee1aee/ : FAILURE in 2m 53s\n- openstack-tox-py36 http://logs.openstack.org/67/657167/2/check/openstack-tox-py36/19363a5/ : SUCCESS in 3m 21s\n- openstack-tox-validate http://logs.openstack.org/67/657167/2/check/openstack-tox-validate/00243a0/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/67/657167/2/check/requirements-tox-babel/643c3cc/ : SUCCESS in 2m 47s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/657167/2/check/requirements-tox-py27-check-uc/3aaf13e/ : SUCCESS in 6m 22s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/657167/2/check/requirements-tox-py36-check-uc/a8749f4/ : SUCCESS in 6m 04s\n- cross-cinder-py27 http://logs.openstack.org/67/657167/2/check/cross-cinder-py27/d0fce7a/ : FAILURE in 3m 48s\n- cross-cinder-py36 http://logs.openstack.org/67/657167/2/check/cross-cinder-py36/2a45601/ : SUCCESS in 8m 39s\n- cross-glance-py27 http://logs.openstack.org/67/657167/2/check/cross-glance-py27/ad5cf6e/ : SUCCESS in 3m 47s\n- cross-glance-py36 http://logs.openstack.org/67/657167/2/check/cross-glance-py36/ce8b6d4/ : SUCCESS in 4m 20s\n- cross-horizon-py27 http://logs.openstack.org/67/657167/2/check/cross-horizon-py27/93294b3/ : SUCCESS in 10m 20s\n- cross-horizon-py36 http://logs.openstack.org/67/657167/2/check/cross-horizon-py36/106353a/ : SUCCESS in 8m 31s\n- cross-horizon-npm http://logs.openstack.org/67/657167/2/check/cross-horizon-npm/40016ab/ : SUCCESS in 5m 18s\n- cross-keystone-py27 http://logs.openstack.org/67/657167/2/check/cross-keystone-py27/4cd5be2/ : SUCCESS in 18m 23s\n- cross-keystone-py36 http://logs.openstack.org/67/657167/2/check/cross-keystone-py36/a9976cf/ : SUCCESS in 25m 50s\n- cross-neutron-py27 http://logs.openstack.org/67/657167/2/check/cross-neutron-py27/6f71b70/ : SUCCESS in 20m 56s\n- cross-neutron-py36 http://logs.openstack.org/67/657167/2/check/cross-neutron-py36/69dce70/ : SUCCESS in 20m 19s\n- cross-nova-py27 http://logs.openstack.org/67/657167/2/check/cross-nova-py27/01f4491/ : SUCCESS in 11m 43s\n- cross-nova-py36 http://logs.openstack.org/67/657167/2/check/cross-nova-py36/7841c6d/ : SUCCESS in 15m 09s\n- cross-nova-functional http://logs.openstack.org/67/657167/2/check/cross-nova-functional/de6815f/ : SUCCESS in 18m 50s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/657167/2/check/cross-osc-build-sphinx-docs/14fcd86/html/ : SUCCESS in 4m 33s\n- cross-swift-py27 http://logs.openstack.org/67/657167/2/check/cross-swift-py27/9fbc388/ : SUCCESS in 7m 15s\n- cross-swift-py36 http://logs.openstack.org/67/657167/2/check/cross-swift-py36/b7e6471/ : SUCCESS in 6m 52s\n- requirements-integration http://logs.openstack.org/67/657167/2/check/requirements-integration/9d52644/ : FAILURE in 34m 01s\n- neutron-grenade-multinode http://logs.openstack.org/67/657167/2/check/neutron-grenade-multinode/e507aa0/ : SUCCESS in 1h 06m 51s\n- tempest-full http://logs.openstack.org/67/657167/2/check/tempest-full/352cf95/ : SUCCESS in 1h 46m 31s\n- neutron-grenade http://logs.openstack.org/67/657167/2/check/neutron-grenade/ca869ec/ : SUCCESS in 56m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b8d6f6aab7acea50e675e47fa335ee99b08df5a3","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-08 19:21:06.000000000","message":"Patch Set 2: -Workflow\n\nLooks like we first need to solve the keyring issue","accounts_in_message":[],"_revision_number":2},{"id":"4eb3f750df31fc6429f1494acb52c8b8c9bbe541","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2019-05-09 21:55:06.000000000","message":"Patch Set 2:\n\nI tried reproducing the issue, but instead ran into other problems with python2.7:\nERROR: glance-store 0.28.0 has requirement jsonschema\u003c3.0.0,\u003e\u003d2.6.0, but you\u0027ll have jsonschema 3.0.1 which is incompatible.\nERROR: warlock 1.3.0 has requirement jsonschema\u003c3,\u003e\u003d0.7, but you\u0027ll have jsonschema 3.0.1 which is incompatible.\nERROR: python-ironicclient 2.7.1 has requirement jsonschema\u003c3.0.0,\u003e\u003d2.6.0, but you\u0027ll have jsonschema 3.0.1 which is incompatible.\nERROR: jaeger-client 4.0.0 has requirement tornado\u003c5,\u003e\u003d4.3, but you\u0027ll have tornado 5.1.1 which is incompatible.\nERROR: taskflow 3.4.0 has requirement jsonschema\u003c3.0.0,\u003e\u003d2.6.0, but you\u0027ll have jsonschema 3.0.1 which is incompatible.\nERROR: tempest 20.0.0 has requirement jsonschema\u003c3.0.0,\u003e\u003d2.6.0, but you\u0027ll have jsonschema 3.0.1 which is incompatible.\nERROR: os-net-config 10.4.0 has requirement jsonschema\u003c3.0.0,\u003e\u003d2.0.0, but you\u0027ll have jsonschema 3.0.1 which is incompatible.\nERROR: python-designateclient 2.11.0 has requirement jsonschema\u003c3.0.0,\u003e\u003d2.6.0, but you\u0027ll have jsonschema 3.0.1 which is incompatible.\nERROR: requests 2.21.0 has requirement urllib3\u003c1.25,\u003e\u003d1.21.1, but you\u0027ll have urllib3 1.25.2 which is incompatible.\nERROR: botocore 1.12.145 has requirement urllib3\u003c1.25,\u003e\u003d1.20; python_version \u003d\u003d \"2.7\", but you\u0027ll have urllib3 1.25.2 which is incompatible.\nERROR: openshift 0.8.8 has requirement kubernetes\u003c9.0.0,\u003e\u003d8.0.0, but you\u0027ll have kubernetes 9.0.0 which is incompatible.\nERROR: openshift 0.8.8 has requirement urllib3\u003c1.25, but you\u0027ll have urllib3 1.25.2 which is incompatible.\nERROR: python-solumclient 2.8.0 has requirement jsonschema\u003c3.0.0,\u003e\u003d2.6.0, but you\u0027ll have jsonschema 3.0.1 which is incompatible.\nERROR: python-zaqarclient 1.11.0 has requirement jsonschema\u003c3.0.0,\u003e\u003d2.6.0, but you\u0027ll have jsonschema 3.0.1 which is incompatible.\n\nI tried changing global-constraints to resolve these conflicts, which was then successful. Do you have any insight into what needs to be changed with the keyring package?\n\nMy simple test was setting up virtualenv and running \"./venv/bin/pip install -r global-requirements.txt\".","accounts_in_message":[],"_revision_number":2},{"id":"c3923a6ca1ced21c6662ba1d2b7ae23e7791b509","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-05-10 19:51:51.000000000","message":"Patch Set 2: Code-Review+1\n\nreverify","accounts_in_message":[],"_revision_number":2},{"id":"4a565a68c512b69d0d386c101957b41911cd5064","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-11 19:27:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"620c308048c9ecc223e824ea7c3bf0ba8bd9a39b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 21:08:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/657167/2/check/openstack-tox-docs/3ca940d/html/ : SUCCESS in 3m 06s\n- openstack-tox-pypy http://logs.openstack.org/67/657167/2/check/openstack-tox-pypy/da21613/ : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/657167/2/check/openstack-tox-pep8/91953b0/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/67/657167/2/check/openstack-tox-py27/caa9c58/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/67/657167/2/check/openstack-tox-py36/597d25e/ : SUCCESS in 3m 20s\n- openstack-tox-validate http://logs.openstack.org/67/657167/2/check/openstack-tox-validate/f7c6be3/ : SUCCESS in 3m 08s\n- requirements-tox-babel http://logs.openstack.org/67/657167/2/check/requirements-tox-babel/5376411/ : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/657167/2/check/requirements-tox-py27-check-uc/7f8e867/ : SUCCESS in 5m 40s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/657167/2/check/requirements-tox-py36-check-uc/54282f8/ : SUCCESS in 5m 55s\n- cross-cinder-py27 http://logs.openstack.org/67/657167/2/check/cross-cinder-py27/fdb3e74/ : SUCCESS in 8m 41s\n- cross-cinder-py36 http://logs.openstack.org/67/657167/2/check/cross-cinder-py36/7ed06cf/ : SUCCESS in 7m 55s\n- cross-glance-py27 http://logs.openstack.org/67/657167/2/check/cross-glance-py27/9d975c0/ : SUCCESS in 3m 54s\n- cross-glance-py36 http://logs.openstack.org/67/657167/2/check/cross-glance-py36/d5aef06/ : SUCCESS in 4m 16s\n- cross-horizon-py27 http://logs.openstack.org/67/657167/2/check/cross-horizon-py27/42f3999/ : SUCCESS in 10m 06s\n- cross-horizon-py36 http://logs.openstack.org/67/657167/2/check/cross-horizon-py36/66afe75/ : SUCCESS in 8m 37s\n- cross-horizon-npm http://logs.openstack.org/67/657167/2/check/cross-horizon-npm/9358a45/ : SUCCESS in 5m 10s\n- cross-keystone-py27 http://logs.openstack.org/67/657167/2/check/cross-keystone-py27/ad6402d/ : SUCCESS in 21m 10s\n- cross-keystone-py36 http://logs.openstack.org/67/657167/2/check/cross-keystone-py36/ecbbb59/ : SUCCESS in 19m 33s\n- cross-neutron-py27 http://logs.openstack.org/67/657167/2/check/cross-neutron-py27/ad6ace2/ : SUCCESS in 22m 02s\n- cross-neutron-py36 http://logs.openstack.org/67/657167/2/check/cross-neutron-py36/c75707a/ : SUCCESS in 20m 26s\n- cross-nova-py27 http://logs.openstack.org/67/657167/2/check/cross-nova-py27/bde05c5/ : SUCCESS in 14m 18s\n- cross-nova-py36 http://logs.openstack.org/67/657167/2/check/cross-nova-py36/a08c103/ : SUCCESS in 13m 23s\n- cross-nova-functional http://logs.openstack.org/67/657167/2/check/cross-nova-functional/e3dc1e0/ : SUCCESS in 18m 42s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/657167/2/check/cross-osc-build-sphinx-docs/fd132d9/html/ : SUCCESS in 6m 42s\n- cross-swift-py27 http://logs.openstack.org/67/657167/2/check/cross-swift-py27/37f658d/ : SUCCESS in 8m 14s\n- cross-swift-py36 http://logs.openstack.org/67/657167/2/check/cross-swift-py36/7228296/ : SUCCESS in 7m 36s\n- requirements-integration http://logs.openstack.org/67/657167/2/check/requirements-integration/73f1c80/ : SUCCESS in 54m 11s\n- neutron-grenade-multinode http://logs.openstack.org/67/657167/2/check/neutron-grenade-multinode/644beef/ : SUCCESS in 54m 09s\n- tempest-full http://logs.openstack.org/67/657167/2/check/tempest-full/cd168f8/ : SUCCESS in 1h 40m 24s\n- neutron-grenade http://logs.openstack.org/67/657167/2/check/neutron-grenade/65ec320/ : SUCCESS in 1h 02m 10s","accounts_in_message":[],"_revision_number":2},{"id":"db1ad94b8f8b5261d90deaf2ba7aa25245d40f4d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-11 21:13:37.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ce325007b1f02efc5639e895849cc34e8de0a0a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 21:13:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1677980795be6901037177f8502e1c8855744750","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 22:57:12.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/67/657167/2/gate/openstack-tox-docs/fd04cb7/html/ : SUCCESS in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/67/657167/2/gate/openstack-tox-pep8/7ed42ed/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/67/657167/2/gate/openstack-tox-py27/6a160b1/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/67/657167/2/gate/openstack-tox-py36/41e2913/ : SUCCESS in 3m 01s\n- openstack-tox-validate http://logs.openstack.org/67/657167/2/gate/openstack-tox-validate/c520e27/ : SUCCESS in 2m 41s\n- requirements-tox-babel http://logs.openstack.org/67/657167/2/gate/requirements-tox-babel/ad2515a/ : SUCCESS in 2m 36s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/657167/2/gate/requirements-tox-py27-check-uc/78d5736/ : SUCCESS in 5m 58s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/657167/2/gate/requirements-tox-py36-check-uc/c667ce0/ : SUCCESS in 5m 38s\n- cross-cinder-py27 http://logs.openstack.org/67/657167/2/gate/cross-cinder-py27/0e92084/ : SUCCESS in 7m 59s\n- cross-cinder-py36 http://logs.openstack.org/67/657167/2/gate/cross-cinder-py36/d746bb8/ : SUCCESS in 12m 21s\n- cross-glance-py27 http://logs.openstack.org/67/657167/2/gate/cross-glance-py27/8b7a3fe/ : SUCCESS in 4m 52s\n- cross-glance-py36 http://logs.openstack.org/67/657167/2/gate/cross-glance-py36/7366452/ : SUCCESS in 4m 01s\n- cross-horizon-py27 http://logs.openstack.org/67/657167/2/gate/cross-horizon-py27/72fcc65/ : SUCCESS in 9m 33s\n- cross-horizon-py36 http://logs.openstack.org/67/657167/2/gate/cross-horizon-py36/e91e1b2/ : SUCCESS in 7m 55s\n- cross-horizon-npm http://logs.openstack.org/67/657167/2/gate/cross-horizon-npm/40a13b4/ : SUCCESS in 5m 38s\n- cross-keystone-py27 http://logs.openstack.org/67/657167/2/gate/cross-keystone-py27/deb3448/ : SUCCESS in 20m 50s\n- cross-keystone-py36 http://logs.openstack.org/67/657167/2/gate/cross-keystone-py36/ad6e460/ : SUCCESS in 26m 52s\n- cross-neutron-py27 http://logs.openstack.org/67/657167/2/gate/cross-neutron-py27/647b1a8/ : SUCCESS in 21m 46s\n- cross-neutron-py36 http://logs.openstack.org/67/657167/2/gate/cross-neutron-py36/db56236/ : SUCCESS in 21m 46s\n- cross-nova-py27 http://logs.openstack.org/67/657167/2/gate/cross-nova-py27/76ca3a4/ : SUCCESS in 10m 43s\n- cross-nova-py36 http://logs.openstack.org/67/657167/2/gate/cross-nova-py36/be98802/ : SUCCESS in 16m 04s\n- cross-nova-functional http://logs.openstack.org/67/657167/2/gate/cross-nova-functional/2927fc3/ : SUCCESS in 18m 26s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/657167/2/gate/cross-osc-build-sphinx-docs/3943079/html/ : SUCCESS in 5m 03s\n- cross-swift-py27 http://logs.openstack.org/67/657167/2/gate/cross-swift-py27/e65b11b/ : SUCCESS in 7m 28s\n- cross-swift-py36 http://logs.openstack.org/67/657167/2/gate/cross-swift-py36/ee624c7/ : SUCCESS in 7m 10s\n- requirements-integration http://logs.openstack.org/67/657167/2/gate/requirements-integration/dbb81f8/ : SUCCESS in 1h 06m 54s\n- neutron-grenade-multinode http://logs.openstack.org/67/657167/2/gate/neutron-grenade-multinode/e8b3ef2/ : SUCCESS in 1h 05m 06s\n- tempest-full http://logs.openstack.org/67/657167/2/gate/tempest-full/ed90cfe/ : SUCCESS in 1h 38m 17s\n- neutron-grenade http://logs.openstack.org/67/657167/2/gate/neutron-grenade/7a704a9/ : FAILURE in 54m 44s","accounts_in_message":[],"_revision_number":2},{"id":"3687143659e2692c6e8db3ef7bb8a677ffabea0d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-11 22:59:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"da55810e3d5742954abfeaf32a242800a34293dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 00:44:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/657167/2/check/openstack-tox-docs/604f787/html/ : SUCCESS in 2m 54s\n- openstack-tox-pypy http://logs.openstack.org/67/657167/2/check/openstack-tox-pypy/7ac40d3/ : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/657167/2/check/openstack-tox-pep8/258a7f8/ : SUCCESS in 2m 46s\n- openstack-tox-py27 http://logs.openstack.org/67/657167/2/check/openstack-tox-py27/2d9f448/ : SUCCESS in 3m 20s\n- openstack-tox-py36 http://logs.openstack.org/67/657167/2/check/openstack-tox-py36/a3cb2a8/ : SUCCESS in 3m 26s\n- openstack-tox-validate http://logs.openstack.org/67/657167/2/check/openstack-tox-validate/9b48b7a/ : SUCCESS in 2m 47s\n- requirements-tox-babel http://logs.openstack.org/67/657167/2/check/requirements-tox-babel/c39e685/ : SUCCESS in 2m 40s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/657167/2/check/requirements-tox-py27-check-uc/11602ec/ : SUCCESS in 5m 35s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/657167/2/check/requirements-tox-py36-check-uc/1d32645/ : SUCCESS in 5m 28s\n- cross-cinder-py27 http://logs.openstack.org/67/657167/2/check/cross-cinder-py27/b51dc79/ : SUCCESS in 7m 58s\n- cross-cinder-py36 http://logs.openstack.org/67/657167/2/check/cross-cinder-py36/9437da7/ : SUCCESS in 7m 25s\n- cross-glance-py27 http://logs.openstack.org/67/657167/2/check/cross-glance-py27/c46d555/ : SUCCESS in 3m 59s\n- cross-glance-py36 http://logs.openstack.org/67/657167/2/check/cross-glance-py36/6fee86b/ : SUCCESS in 3m 52s\n- cross-horizon-py27 http://logs.openstack.org/67/657167/2/check/cross-horizon-py27/b0dcedc/ : SUCCESS in 9m 40s\n- cross-horizon-py36 http://logs.openstack.org/67/657167/2/check/cross-horizon-py36/7a5d6ec/ : SUCCESS in 13m 27s\n- cross-horizon-npm http://logs.openstack.org/67/657167/2/check/cross-horizon-npm/f72631e/ : SUCCESS in 5m 30s\n- cross-keystone-py27 http://logs.openstack.org/67/657167/2/check/cross-keystone-py27/7937eb2/ : SUCCESS in 22m 49s\n- cross-keystone-py36 http://logs.openstack.org/67/657167/2/check/cross-keystone-py36/f48cd79/ : SUCCESS in 23m 25s\n- cross-neutron-py27 http://logs.openstack.org/67/657167/2/check/cross-neutron-py27/191bc96/ : SUCCESS in 19m 45s\n- cross-neutron-py36 http://logs.openstack.org/67/657167/2/check/cross-neutron-py36/b49861d/ : SUCCESS in 18m 51s\n- cross-nova-py27 http://logs.openstack.org/67/657167/2/check/cross-nova-py27/326bb72/ : SUCCESS in 11m 05s\n- cross-nova-py36 http://logs.openstack.org/67/657167/2/check/cross-nova-py36/2797d04/ : SUCCESS in 10m 53s\n- cross-nova-functional http://logs.openstack.org/67/657167/2/check/cross-nova-functional/f6645d9/ : SUCCESS in 17m 41s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/657167/2/check/cross-osc-build-sphinx-docs/d2b4d9a/html/ : SUCCESS in 5m 11s\n- cross-swift-py27 http://logs.openstack.org/67/657167/2/check/cross-swift-py27/6b53987/ : SUCCESS in 7m 11s\n- cross-swift-py36 http://logs.openstack.org/67/657167/2/check/cross-swift-py36/d9a03eb/ : SUCCESS in 7m 01s\n- requirements-integration http://logs.openstack.org/67/657167/2/check/requirements-integration/16547d0/ : SUCCESS in 55m 05s\n- neutron-grenade-multinode http://logs.openstack.org/67/657167/2/check/neutron-grenade-multinode/40e7195/ : SUCCESS in 1h 03m 15s\n- tempest-full http://logs.openstack.org/67/657167/2/check/tempest-full/41e1d8c/ : SUCCESS in 1h 40m 37s\n- neutron-grenade http://logs.openstack.org/67/657167/2/check/neutron-grenade/b3194c9/ : SUCCESS in 1h 00m 19s","accounts_in_message":[],"_revision_number":2},{"id":"94c2674a14b222b68d2d29aedb201a525170fb52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 00:45:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"19b0e8e8b82e59be08777f9ebb4b4199f77716bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 02:30:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6c74c1296e829ae0159a891be0571ab22dfef540","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 02:30:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/67/657167/2/gate/openstack-tox-docs/1e0e61e/html/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/67/657167/2/gate/openstack-tox-pep8/ae39c53/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/67/657167/2/gate/openstack-tox-py27/f367ae7/ : SUCCESS in 2m 52s\n- openstack-tox-py36 http://logs.openstack.org/67/657167/2/gate/openstack-tox-py36/a54f6a5/ : SUCCESS in 4m 05s\n- openstack-tox-validate http://logs.openstack.org/67/657167/2/gate/openstack-tox-validate/d8a57aa/ : SUCCESS in 3m 48s\n- requirements-tox-babel http://logs.openstack.org/67/657167/2/gate/requirements-tox-babel/e6e6fe4/ : SUCCESS in 4m 13s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/657167/2/gate/requirements-tox-py27-check-uc/b8b34a8/ : SUCCESS in 5m 58s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/657167/2/gate/requirements-tox-py36-check-uc/bd206c6/ : SUCCESS in 6m 33s\n- cross-cinder-py27 http://logs.openstack.org/67/657167/2/gate/cross-cinder-py27/315c6fd/ : SUCCESS in 8m 19s\n- cross-cinder-py36 http://logs.openstack.org/67/657167/2/gate/cross-cinder-py36/c415adc/ : SUCCESS in 9m 34s\n- cross-glance-py27 http://logs.openstack.org/67/657167/2/gate/cross-glance-py27/76345e0/ : SUCCESS in 4m 40s\n- cross-glance-py36 http://logs.openstack.org/67/657167/2/gate/cross-glance-py36/6c88e0c/ : SUCCESS in 4m 49s\n- cross-horizon-py27 http://logs.openstack.org/67/657167/2/gate/cross-horizon-py27/f8346f7/ : SUCCESS in 8m 32s\n- cross-horizon-py36 http://logs.openstack.org/67/657167/2/gate/cross-horizon-py36/1a57b71/ : SUCCESS in 7m 53s\n- cross-horizon-npm http://logs.openstack.org/67/657167/2/gate/cross-horizon-npm/12349d8/ : SUCCESS in 8m 42s\n- cross-keystone-py27 http://logs.openstack.org/67/657167/2/gate/cross-keystone-py27/6e40c3c/ : SUCCESS in 25m 32s\n- cross-keystone-py36 http://logs.openstack.org/67/657167/2/gate/cross-keystone-py36/1b09dfe/ : SUCCESS in 28m 56s\n- cross-neutron-py27 http://logs.openstack.org/67/657167/2/gate/cross-neutron-py27/33899a7/ : SUCCESS in 29m 35s\n- cross-neutron-py36 http://logs.openstack.org/67/657167/2/gate/cross-neutron-py36/1c88f87/ : SUCCESS in 19m 09s\n- cross-nova-py27 http://logs.openstack.org/67/657167/2/gate/cross-nova-py27/d246f5d/ : SUCCESS in 12m 29s\n- cross-nova-py36 http://logs.openstack.org/67/657167/2/gate/cross-nova-py36/5e2dfd4/ : SUCCESS in 11m 09s\n- cross-nova-functional http://logs.openstack.org/67/657167/2/gate/cross-nova-functional/b4e64bc/ : SUCCESS in 16m 18s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/657167/2/gate/cross-osc-build-sphinx-docs/abcb641/html/ : SUCCESS in 7m 12s\n- cross-swift-py27 http://logs.openstack.org/67/657167/2/gate/cross-swift-py27/fa13f10/ : SUCCESS in 8m 38s\n- cross-swift-py36 http://logs.openstack.org/67/657167/2/gate/cross-swift-py36/2a13aab/ : SUCCESS in 9m 09s\n- requirements-integration http://logs.openstack.org/67/657167/2/gate/requirements-integration/871ddaa/ : SUCCESS in 46m 21s\n- neutron-grenade-multinode http://logs.openstack.org/67/657167/2/gate/neutron-grenade-multinode/726d099/ : SUCCESS in 54m 33s\n- tempest-full http://logs.openstack.org/67/657167/2/gate/tempest-full/15ecdc8/ : SUCCESS in 1h 31m 53s\n- neutron-grenade http://logs.openstack.org/67/657167/2/gate/neutron-grenade/fdd82e9/ : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b98ff65847cc70a82ae8ecaa3979967657222b22","revisions":{"ee73f0832175ee42fb4a0a46207ca4f2a52f16d0":{"kind":"REWORK","_number":1,"created":"2019-05-04 21:52:46.000000000","uploader":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"ref":"refs/changes/67/657167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/657167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/657167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/657167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/657167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/657167/1"}}},"commit":{"parents":[{"commit":"fc07718c9c043bf52d2d0f7206d711a362251e16","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc07718c9c043bf52d2d0f7206d711a362251e16"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-05-04 21:08:46.000000000","tz":-420},"committer":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-05-04 21:52:42.000000000","tz":-420},"subject":"Add boto3 to global requirements.","message":"Add boto3 to global requirements.\n\nSwift is moving to using boto3, as opposed to the unmaintained boto\nlibrary for its S3 tests.\n\nChange-Id: I1d95e4bbed051ea7e4bb77fb9e38d3006843f998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee73f0832175ee42fb4a0a46207ca4f2a52f16d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee73f0832175ee42fb4a0a46207ca4f2a52f16d0"}]},"branch":"refs/heads/master"},"b98ff65847cc70a82ae8ecaa3979967657222b22":{"kind":"REWORK","_number":2,"created":"2019-05-06 23:09:47.000000000","uploader":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"ref":"refs/changes/67/657167/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/657167/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/657167/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/657167/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/657167/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/657167/2"}}},"commit":{"parents":[{"commit":"6affabb05381072509f382af865a54a7916b7664","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6affabb05381072509f382af865a54a7916b7664"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-05-04 21:08:46.000000000","tz":-420},"committer":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2019-05-06 22:41:48.000000000","tz":-420},"subject":"Add Boto3 to global requirements.","message":"Add Boto3 to global requirements.\n\nSwift is moving to using Boto3, as opposed to the unmaintained Boto\nlibrary for its S3 tests.\n\n* Is the library actively maintained?\n\nYes -- it is the recommended library by Amazon for AWS and S3\nspecifically. Last commit: 05/06/2019 (auto-commit). Maintainers respond\nto bug reports.\n\n* Is the library good code?\n\nIt\u0027s the evolution of the Boto library. The code is well documented,\nwith numerous examples of usage. A significant portion is\nauto-generated, which makes working on the library a little more\ndifficult. The library supports extensions and modifications through its\nevent system, which allows its easier usage in the testing\ninfrastructure.\n\n* Is the library python 3 compatible?\n\nYes -- up to python 3.7.\n\n* Is  the library license compatible?\n\nBoto3 is licensed under Apache 2.0.\n\n* Is the library already packaged in the distros we target (Ubuntu\nlatest / Fedora latest)?\n\nYes (Ubuntu 19.04: 1.9.86; Fedora 30: 1.9.101)\n\n* Is the function of this library already covered by other libraries in\nglobal-requirements.txt?\n\nThis would be a replacement for the existing Boto library. The older\nBoto library could be removed if all users transition to Boto3.\n\n* Is the library required for OpenStack project or related dev or\ninfrastructure setup? (Answer to this should be Yes, of course) Which?\n\nYes, it is to be used by Swift: https://review.opendev.org/#/c/657149/.\nThe Boto library has the following warning text:\n\"API updates and all new feature work will be focused on Boto3\"\nand:\n\"As more users make the switch to Boto3, we expect to reduce our\nmaintenance involvement over time\".\nBoto issues/pull requests are currently not addressed (for example,\nhttps://github.com/Boto/Boto/pull/3471 and\nhttps://github.com/Boto/Boto/pull/3513 -- pull requests from a Swift\ncommitter have been open for 3 years with no feedback).\n\nBoto and Boto3 can be used alongside each other to simplify the\ntransition.\n\n* If the library release is managed by the Openstack release process\ndoes it use the cycle-with-intermediary release type?\n\nThis is an external to OpenStack library.\n\nChange-Id: I1d95e4bbed051ea7e4bb77fb9e38d3006843f998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b98ff65847cc70a82ae8ecaa3979967657222b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b98ff65847cc70a82ae8ecaa3979967657222b22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
