)]}'
{"id":"openstack%2Frequirements~657204","triplet_id":"openstack%2Frequirements~master~I68ddaae1845917f4b13bfdcbfe6caf71bcf514a1","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I68ddaae1845917f4b13bfdcbfe6caf71bcf514a1","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-05-05 06:32:10.000000000","updated":"2019-05-11 23:45:10.000000000","submitted":"2019-05-11 23:45:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657204-1557618310642-72f24b91","meta_rev_id":"5f79025e1a0cc7354ffc4245d34871e57ccdcd16","_number":657204,"virtual_id_number":657204,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-11 23:45:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-11 21:18:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-11 21:18:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-11 21:18:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-11 23:45:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6b9bdcc36c81aa274835256d1212a558720e822","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-05 06:32:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5f245d0f669dc807676f11a95ea703b8d912536","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 07:37:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/04/657204/1/check/openstack-tox-docs/2991090/html/ : SUCCESS in 3m 26s\n- openstack-tox-pypy http://logs.openstack.org/04/657204/1/check/openstack-tox-pypy/d00dcfd/ : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/04/657204/1/check/openstack-tox-pep8/9d389ea/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/04/657204/1/check/openstack-tox-py27/31ff796/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/04/657204/1/check/openstack-tox-py36/696d22a/ : SUCCESS in 3m 07s\n- openstack-tox-validate http://logs.openstack.org/04/657204/1/check/openstack-tox-validate/88de7ce/ : SUCCESS in 2m 55s\n- requirements-tox-babel http://logs.openstack.org/04/657204/1/check/requirements-tox-babel/a16b706/ : SUCCESS in 3m 18s\n- requirements-tox-py27-check-uc http://logs.openstack.org/04/657204/1/check/requirements-tox-py27-check-uc/1811654/ : FAILURE in 12m 24s\n- requirements-tox-py36-check-uc http://logs.openstack.org/04/657204/1/check/requirements-tox-py36-check-uc/cb130ed/ : FAILURE in 11m 17s\n- cross-cinder-py27 http://logs.openstack.org/04/657204/1/check/cross-cinder-py27/c06790b/ : SUCCESS in 8m 45s\n- cross-cinder-py36 http://logs.openstack.org/04/657204/1/check/cross-cinder-py36/473d4ef/ : SUCCESS in 8m 10s\n- cross-glance-py27 http://logs.openstack.org/04/657204/1/check/cross-glance-py27/3dc5a79/ : SUCCESS in 5m 09s\n- cross-glance-py36 http://logs.openstack.org/04/657204/1/check/cross-glance-py36/3c13e40/ : SUCCESS in 5m 01s\n- cross-horizon-py27 http://logs.openstack.org/04/657204/1/check/cross-horizon-py27/42d9671/ : SUCCESS in 8m 50s\n- cross-horizon-py36 http://logs.openstack.org/04/657204/1/check/cross-horizon-py36/eaac7f1/ : SUCCESS in 9m 42s\n- cross-horizon-npm http://logs.openstack.org/04/657204/1/check/cross-horizon-npm/c73e6f0/ : SUCCESS in 5m 32s\n- cross-keystone-py27 http://logs.openstack.org/04/657204/1/check/cross-keystone-py27/fdc8243/ : SUCCESS in 36m 10s\n- cross-keystone-py36 http://logs.openstack.org/04/657204/1/check/cross-keystone-py36/2c21e3a/ : SUCCESS in 30m 56s\n- cross-neutron-py27 http://logs.openstack.org/04/657204/1/check/cross-neutron-py27/8bb91f9/ : SUCCESS in 36m 58s\n- cross-neutron-py36 http://logs.openstack.org/04/657204/1/check/cross-neutron-py36/fc25640/ : SUCCESS in 20m 57s\n- cross-nova-py27 http://logs.openstack.org/04/657204/1/check/cross-nova-py27/7ecbc6a/ : FAILURE in 5m 50s\n- cross-nova-py36 http://logs.openstack.org/04/657204/1/check/cross-nova-py36/0ea534f/ : FAILURE in 4m 59s\n- cross-nova-functional http://logs.openstack.org/04/657204/1/check/cross-nova-functional/ffd522d/ : FAILURE in 17m 06s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/04/657204/1/check/cross-osc-build-sphinx-docs/ef7f615/html/ : SUCCESS in 6m 47s\n- cross-swift-py27 http://logs.openstack.org/04/657204/1/check/cross-swift-py27/42393fd/ : SUCCESS in 9m 20s\n- cross-swift-py36 http://logs.openstack.org/04/657204/1/check/cross-swift-py36/e58a839/ : SUCCESS in 9m 40s\n- requirements-integration http://logs.openstack.org/04/657204/1/check/requirements-integration/b864f73/ : FAILURE in 28m 24s\n- neutron-grenade-multinode http://logs.openstack.org/04/657204/1/check/neutron-grenade-multinode/1bd5728/ : FAILURE in 1h 00m 28s\n- tempest-full http://logs.openstack.org/04/657204/1/check/tempest-full/0939545/ : FAILURE in 28m 28s\n- neutron-grenade http://logs.openstack.org/04/657204/1/check/neutron-grenade/e6f43e9/ : FAILURE in 45m 16s","accounts_in_message":[],"_revision_number":1},{"id":"fc9d8f6e4958103a2a4fd3dd678e4a7c0d969597","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-07 06:44:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f2979a20f31d8639194778118f9bd3ed61b5a1ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 07:52:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/04/657204/2/check/openstack-tox-docs/c0e0925/html/ : SUCCESS in 2m 57s\n- openstack-tox-pypy http://logs.openstack.org/04/657204/2/check/openstack-tox-pypy/b69d1df/ : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/04/657204/2/check/openstack-tox-pep8/37b9346/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/04/657204/2/check/openstack-tox-py27/9f7f531/ : SUCCESS in 3m 07s\n- openstack-tox-py36 http://logs.openstack.org/04/657204/2/check/openstack-tox-py36/6647133/ : SUCCESS in 3m 13s\n- openstack-tox-validate http://logs.openstack.org/04/657204/2/check/openstack-tox-validate/5f42de3/ : SUCCESS in 4m 36s\n- requirements-tox-babel http://logs.openstack.org/04/657204/2/check/requirements-tox-babel/dbe9e2b/ : SUCCESS in 2m 59s\n- requirements-tox-py27-check-uc http://logs.openstack.org/04/657204/2/check/requirements-tox-py27-check-uc/bd60e5f/ : FAILURE in 11m 48s\n- requirements-tox-py36-check-uc http://logs.openstack.org/04/657204/2/check/requirements-tox-py36-check-uc/1b08e5e/ : FAILURE in 14m 21s\n- cross-cinder-py27 http://logs.openstack.org/04/657204/2/check/cross-cinder-py27/bf8703b/ : SUCCESS in 8m 55s\n- cross-cinder-py36 http://logs.openstack.org/04/657204/2/check/cross-cinder-py36/9bf8914/ : SUCCESS in 9m 17s\n- cross-glance-py27 http://logs.openstack.org/04/657204/2/check/cross-glance-py27/cfcfd23/ : SUCCESS in 4m 16s\n- cross-glance-py36 http://logs.openstack.org/04/657204/2/check/cross-glance-py36/4b91bcd/ : SUCCESS in 4m 14s\n- cross-horizon-py27 http://logs.openstack.org/04/657204/2/check/cross-horizon-py27/92df0a7/ : SUCCESS in 10m 49s\n- cross-horizon-py36 http://logs.openstack.org/04/657204/2/check/cross-horizon-py36/d18c043/ : SUCCESS in 9m 48s\n- cross-horizon-npm http://logs.openstack.org/04/657204/2/check/cross-horizon-npm/505f657/ : SUCCESS in 5m 47s\n- cross-keystone-py27 http://logs.openstack.org/04/657204/2/check/cross-keystone-py27/b6c0eb7/ : SUCCESS in 23m 51s\n- cross-keystone-py36 http://logs.openstack.org/04/657204/2/check/cross-keystone-py36/ebc82a5/ : SUCCESS in 23m 32s\n- cross-neutron-py27 http://logs.openstack.org/04/657204/2/check/cross-neutron-py27/b3637ba/ : SUCCESS in 24m 15s\n- cross-neutron-py36 http://logs.openstack.org/04/657204/2/check/cross-neutron-py36/167bf93/ : SUCCESS in 21m 03s\n- cross-nova-py27 http://logs.openstack.org/04/657204/2/check/cross-nova-py27/d5b8283/ : FAILURE in 5m 37s\n- cross-nova-py36 http://logs.openstack.org/04/657204/2/check/cross-nova-py36/c9ef1e4/ : FAILURE in 5m 06s\n- cross-nova-functional http://logs.openstack.org/04/657204/2/check/cross-nova-functional/89c1b5d/ : FAILURE in 16m 52s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/04/657204/2/check/cross-osc-build-sphinx-docs/5cbccbf/html/ : SUCCESS in 5m 59s\n- cross-swift-py27 http://logs.openstack.org/04/657204/2/check/cross-swift-py27/5bf4096/ : SUCCESS in 7m 39s\n- cross-swift-py36 http://logs.openstack.org/04/657204/2/check/cross-swift-py36/27d7e35/ : SUCCESS in 7m 07s\n- requirements-integration http://logs.openstack.org/04/657204/2/check/requirements-integration/7bbacc8/ : FAILURE in 31m 21s\n- neutron-grenade-multinode http://logs.openstack.org/04/657204/2/check/neutron-grenade-multinode/c9f0610/ : FAILURE in 1h 01m 49s\n- tempest-full http://logs.openstack.org/04/657204/2/check/tempest-full/037785a/ : FAILURE in 29m 00s\n- neutron-grenade http://logs.openstack.org/04/657204/2/check/neutron-grenade/644a52d/ : FAILURE in 48m 19s","accounts_in_message":[],"_revision_number":2},{"id":"34e2adcaeda6cbb3a06af1c08e7afa961b4f5ecd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-07 14:13:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5cee3abd7771d2ba4fd44437cc8c55619ef2e6a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-07 14:14:09.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"cb3f6463138b38ca1005a17fb681c0ab6b0e477a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 16:23:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/04/657204/3/check/openstack-tox-docs/4073040/html/ : SUCCESS in 3m 23s\n- openstack-tox-pypy http://logs.openstack.org/04/657204/3/check/openstack-tox-pypy/21c5155/ : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/04/657204/3/check/openstack-tox-pep8/66c2d26/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/04/657204/3/check/openstack-tox-py27/be1ea2b/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/04/657204/3/check/openstack-tox-py36/6bd201f/ : SUCCESS in 3m 47s\n- openstack-tox-validate http://logs.openstack.org/04/657204/3/check/openstack-tox-validate/86a6525/ : SUCCESS in 3m 28s\n- requirements-tox-babel http://logs.openstack.org/04/657204/3/check/requirements-tox-babel/4c5e62d/ : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/04/657204/3/check/requirements-tox-py27-check-uc/0d243cf/ : FAILURE in 11m 38s\n- requirements-tox-py36-check-uc http://logs.openstack.org/04/657204/3/check/requirements-tox-py36-check-uc/510aa14/ : FAILURE in 14m 26s\n- cross-cinder-py27 http://logs.openstack.org/04/657204/3/check/cross-cinder-py27/54a8adb/ : SUCCESS in 9m 37s\n- cross-cinder-py36 http://logs.openstack.org/04/657204/3/check/cross-cinder-py36/58dcc4f/ : SUCCESS in 9m 15s\n- cross-glance-py27 http://logs.openstack.org/04/657204/3/check/cross-glance-py27/79e585f/ : SUCCESS in 4m 24s\n- cross-glance-py36 http://logs.openstack.org/04/657204/3/check/cross-glance-py36/58ef2b1/ : SUCCESS in 4m 34s\n- cross-horizon-py27 http://logs.openstack.org/04/657204/3/check/cross-horizon-py27/22d87da/ : SUCCESS in 9m 07s\n- cross-horizon-py36 http://logs.openstack.org/04/657204/3/check/cross-horizon-py36/fa5e715/ : SUCCESS in 8m 06s\n- cross-horizon-npm http://logs.openstack.org/04/657204/3/check/cross-horizon-npm/cd44bd5/ : SUCCESS in 6m 22s\n- cross-keystone-py27 http://logs.openstack.org/04/657204/3/check/cross-keystone-py27/fdc00da/ : SUCCESS in 21m 18s\n- cross-keystone-py36 http://logs.openstack.org/04/657204/3/check/cross-keystone-py36/f330502/ : SUCCESS in 23m 16s\n- cross-neutron-py27 http://logs.openstack.org/04/657204/3/check/cross-neutron-py27/122c562/ : SUCCESS in 21m 39s\n- cross-neutron-py36 http://logs.openstack.org/04/657204/3/check/cross-neutron-py36/0a26617/ : SUCCESS in 31m 47s\n- cross-nova-py27 http://logs.openstack.org/04/657204/3/check/cross-nova-py27/ed4e327/ : SUCCESS in 12m 47s\n- cross-nova-py36 http://logs.openstack.org/04/657204/3/check/cross-nova-py36/12608ae/ : SUCCESS in 11m 20s\n- cross-nova-functional http://logs.openstack.org/04/657204/3/check/cross-nova-functional/ff0188e/ : SUCCESS in 20m 05s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/04/657204/3/check/cross-osc-build-sphinx-docs/9a0fb04/html/ : SUCCESS in 5m 35s\n- cross-swift-py27 http://logs.openstack.org/04/657204/3/check/cross-swift-py27/e4e0da4/ : SUCCESS in 7m 57s\n- cross-swift-py36 http://logs.openstack.org/04/657204/3/check/cross-swift-py36/9c5832c/ : SUCCESS in 9m 02s\n- requirements-integration http://logs.openstack.org/04/657204/3/check/requirements-integration/63471cf/ : SUCCESS in 1h 13m 32s\n- neutron-grenade-multinode http://logs.openstack.org/04/657204/3/check/neutron-grenade-multinode/a7b8540/ : SUCCESS in 1h 14m 04s\n- tempest-full http://logs.openstack.org/04/657204/3/check/tempest-full/6cb0f77/ : SUCCESS in 1h 44m 33s\n- neutron-grenade http://logs.openstack.org/04/657204/3/check/neutron-grenade/63a2992/ : FAILURE in 1h 01m 37s","accounts_in_message":[],"_revision_number":3},{"id":"4995b10904fb95ac933fd4a4683de153b726575a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-09 15:46:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"40f2476d312fecfc6e7d129e191b2b0f26b86ab1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 21:13:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/04/657204/4/check/openstack-tox-docs/57a8032/html/ : SUCCESS in 2m 56s\n- openstack-tox-pypy http://logs.openstack.org/04/657204/4/check/openstack-tox-pypy/c3e9775/ : FAILURE in 3m 50s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/04/657204/4/check/openstack-tox-pep8/01ec095/ : SUCCESS in 2m 35s\n- openstack-tox-py27 http://logs.openstack.org/04/657204/4/check/openstack-tox-py27/6f23113/ : FAILURE in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/04/657204/4/check/openstack-tox-py36/099e8b9/ : SUCCESS in 5m 50s\n- openstack-tox-validate http://logs.openstack.org/04/657204/4/check/openstack-tox-validate/9367164/ : SUCCESS in 3m 04s\n- requirements-tox-babel http://logs.openstack.org/04/657204/4/check/requirements-tox-babel/c04e42a/ : SUCCESS in 5m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/04/657204/4/check/requirements-tox-py27-check-uc/0f64776/ : SUCCESS in 6m 19s\n- requirements-tox-py36-check-uc http://logs.openstack.org/04/657204/4/check/requirements-tox-py36-check-uc/07d5a09/ : SUCCESS in 6m 36s\n- cross-cinder-py27 http://logs.openstack.org/04/657204/4/check/cross-cinder-py27/09ffef8/ : SUCCESS in 9m 34s\n- cross-cinder-py36 http://logs.openstack.org/04/657204/4/check/cross-cinder-py36/4916d71/ : SUCCESS in 8m 39s\n- cross-glance-py27 http://logs.openstack.org/04/657204/4/check/cross-glance-py27/530073c/ : SUCCESS in 4m 26s\n- cross-glance-py36 http://logs.openstack.org/04/657204/4/check/cross-glance-py36/6e1e800/ : SUCCESS in 6m 06s\n- cross-horizon-py27 http://logs.openstack.org/04/657204/4/check/cross-horizon-py27/c6dccd8/ : SUCCESS in 10m 25s\n- cross-horizon-py36 http://logs.openstack.org/04/657204/4/check/cross-horizon-py36/dde36cb/ : SUCCESS in 8m 34s\n- cross-horizon-npm http://logs.openstack.org/04/657204/4/check/cross-horizon-npm/573a63f/ : SUCCESS in 5m 29s\n- cross-keystone-py27 http://logs.openstack.org/04/657204/4/check/cross-keystone-py27/772dea3/ : SUCCESS in 22m 09s\n- cross-keystone-py36 http://logs.openstack.org/04/657204/4/check/cross-keystone-py36/b0b37a2/ : SUCCESS in 21m 33s\n- cross-neutron-py27 http://logs.openstack.org/04/657204/4/check/cross-neutron-py27/f69b631/ : SUCCESS in 22m 52s\n- cross-neutron-py36 http://logs.openstack.org/04/657204/4/check/cross-neutron-py36/5d1d8f1/ : SUCCESS in 23m 44s\n- cross-nova-py27 http://logs.openstack.org/04/657204/4/check/cross-nova-py27/0a8c2f6/ : SUCCESS in 12m 54s\n- cross-nova-py36 http://logs.openstack.org/04/657204/4/check/cross-nova-py36/99fa37d/ : SUCCESS in 12m 40s\n- cross-nova-functional http://logs.openstack.org/04/657204/4/check/cross-nova-functional/8f241c8/ : SUCCESS in 20m 43s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/04/657204/4/check/cross-osc-build-sphinx-docs/5afd5c7/html/ : SUCCESS in 5m 34s\n- cross-swift-py27 http://logs.openstack.org/04/657204/4/check/cross-swift-py27/73bf4f9/ : SUCCESS in 9m 56s\n- cross-swift-py36 http://logs.openstack.org/04/657204/4/check/cross-swift-py36/f6ba91d/ : SUCCESS in 7m 28s\n- requirements-integration http://logs.openstack.org/04/657204/4/check/requirements-integration/f679c98/ : FAILURE in 32m 55s\n- neutron-grenade-multinode http://logs.openstack.org/04/657204/4/check/neutron-grenade-multinode/002faee/ : SUCCESS in 1h 08m 19s\n- tempest-full http://logs.openstack.org/04/657204/4/check/tempest-full/ff9e515/ : SUCCESS in 1h 46m 53s\n- neutron-grenade http://logs.openstack.org/04/657204/4/check/neutron-grenade/2615fe9/ : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":4},{"id":"be3203fc9d747aedf3d7aa9961983cb62046c799","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-09 21:14:59.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fda1339375e9e0fe55dbe4c7dd5bd5785a39ca8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 23:39:54.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/04/657204/4/check/openstack-tox-docs/9f780c2/html/ : SUCCESS in 3m 01s\n- openstack-tox-pypy http://logs.openstack.org/04/657204/4/check/openstack-tox-pypy/2e43ffa/ : FAILURE in 2m 58s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/04/657204/4/check/openstack-tox-pep8/effae05/ : SUCCESS in 2m 58s\n- openstack-tox-py27 http://logs.openstack.org/04/657204/4/check/openstack-tox-py27/b807387/ : FAILURE in 2m 32s\n- openstack-tox-py36 http://logs.openstack.org/04/657204/4/check/openstack-tox-py36/3dbc1a6/ : SUCCESS in 3m 15s\n- openstack-tox-validate http://logs.openstack.org/04/657204/4/check/openstack-tox-validate/2083dfd/ : SUCCESS in 3m 04s\n- requirements-tox-babel http://logs.openstack.org/04/657204/4/check/requirements-tox-babel/44fd701/ : SUCCESS in 2m 50s\n- requirements-tox-py27-check-uc http://logs.openstack.org/04/657204/4/check/requirements-tox-py27-check-uc/9d7d2aa/ : SUCCESS in 5m 53s\n- requirements-tox-py36-check-uc http://logs.openstack.org/04/657204/4/check/requirements-tox-py36-check-uc/f614508/ : SUCCESS in 6m 21s\n- cross-cinder-py27 http://logs.openstack.org/04/657204/4/check/cross-cinder-py27/15490b3/ : SUCCESS in 8m 37s\n- cross-cinder-py36 http://logs.openstack.org/04/657204/4/check/cross-cinder-py36/8d048c2/ : SUCCESS in 9m 00s\n- cross-glance-py27 http://logs.openstack.org/04/657204/4/check/cross-glance-py27/48ce229/ : SUCCESS in 4m 04s\n- cross-glance-py36 http://logs.openstack.org/04/657204/4/check/cross-glance-py36/3b6ef5e/ : SUCCESS in 4m 37s\n- cross-horizon-py27 http://logs.openstack.org/04/657204/4/check/cross-horizon-py27/d0f20e3/ : SUCCESS in 10m 22s\n- cross-horizon-py36 http://logs.openstack.org/04/657204/4/check/cross-horizon-py36/d319352/ : SUCCESS in 9m 29s\n- cross-horizon-npm http://logs.openstack.org/04/657204/4/check/cross-horizon-npm/9b1a315/ : SUCCESS in 6m 10s\n- cross-keystone-py27 http://logs.openstack.org/04/657204/4/check/cross-keystone-py27/0310cf6/ : SUCCESS in 23m 42s\n- cross-keystone-py36 http://logs.openstack.org/04/657204/4/check/cross-keystone-py36/d2ee420/ : SUCCESS in 23m 24s\n- cross-neutron-py27 http://logs.openstack.org/04/657204/4/check/cross-neutron-py27/c655472/ : SUCCESS in 22m 31s\n- cross-neutron-py36 http://logs.openstack.org/04/657204/4/check/cross-neutron-py36/2bef634/ : SUCCESS in 27m 05s\n- cross-nova-py27 http://logs.openstack.org/04/657204/4/check/cross-nova-py27/a1ac678/ : SUCCESS in 12m 32s\n- cross-nova-py36 http://logs.openstack.org/04/657204/4/check/cross-nova-py36/82b2559/ : SUCCESS in 13m 05s\n- cross-nova-functional http://logs.openstack.org/04/657204/4/check/cross-nova-functional/f1e5c66/ : SUCCESS in 21m 04s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/04/657204/4/check/cross-osc-build-sphinx-docs/497e22c/html/ : SUCCESS in 5m 31s\n- cross-swift-py27 http://logs.openstack.org/04/657204/4/check/cross-swift-py27/9068d3f/ : SUCCESS in 10m 09s\n- cross-swift-py36 http://logs.openstack.org/04/657204/4/check/cross-swift-py36/ca791f3/ : SUCCESS in 7m 04s\n- requirements-integration http://logs.openstack.org/04/657204/4/check/requirements-integration/7581092/ : FAILURE in 36m 54s\n- neutron-grenade-multinode http://logs.openstack.org/04/657204/4/check/neutron-grenade-multinode/34be321/ : SUCCESS in 1h 10m 48s\n- tempest-full http://logs.openstack.org/04/657204/4/check/tempest-full/5a741ef/ : SUCCESS in 1h 43m 55s\n- neutron-grenade http://logs.openstack.org/04/657204/4/check/neutron-grenade/fb39491/ : SUCCESS in 56m 43s","accounts_in_message":[],"_revision_number":4},{"id":"794dfda608e3be579a1d2b1e35519842ed108249","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 05:09:00.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"88de3d8edb07372ddd6503994ac25f4bc5fcbfee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 07:08:41.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/04/657204/4/check/openstack-tox-docs/7963b09/html/ : SUCCESS in 5m 32s\n- openstack-tox-pypy http://logs.openstack.org/04/657204/4/check/openstack-tox-pypy/0a093d8/ : FAILURE in 4m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/04/657204/4/check/openstack-tox-pep8/a4c424a/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/04/657204/4/check/openstack-tox-py27/4a2ed33/ : FAILURE in 2m 56s\n- openstack-tox-py36 http://logs.openstack.org/04/657204/4/check/openstack-tox-py36/4383a98/ : SUCCESS in 4m 47s\n- openstack-tox-validate http://logs.openstack.org/04/657204/4/check/openstack-tox-validate/4454262/ : SUCCESS in 4m 43s\n- requirements-tox-babel http://logs.openstack.org/04/657204/4/check/requirements-tox-babel/c315c6b/ : SUCCESS in 4m 19s\n- requirements-tox-py27-check-uc http://logs.openstack.org/04/657204/4/check/requirements-tox-py27-check-uc/6aee505/ : SUCCESS in 8m 35s\n- requirements-tox-py36-check-uc http://logs.openstack.org/04/657204/4/check/requirements-tox-py36-check-uc/9061716/ : SUCCESS in 7m 50s\n- cross-cinder-py27 http://logs.openstack.org/04/657204/4/check/cross-cinder-py27/661ec89/ : SUCCESS in 9m 13s\n- cross-cinder-py36 http://logs.openstack.org/04/657204/4/check/cross-cinder-py36/7abb78a/ : SUCCESS in 10m 27s\n- cross-glance-py27 http://logs.openstack.org/04/657204/4/check/cross-glance-py27/5aae871/ : SUCCESS in 6m 04s\n- cross-glance-py36 http://logs.openstack.org/04/657204/4/check/cross-glance-py36/6eac416/ : SUCCESS in 6m 40s\n- cross-horizon-py27 http://logs.openstack.org/04/657204/4/check/cross-horizon-py27/a7fcc73/ : SUCCESS in 11m 58s\n- cross-horizon-py36 http://logs.openstack.org/04/657204/4/check/cross-horizon-py36/938ba23/ : SUCCESS in 10m 13s\n- cross-horizon-npm http://logs.openstack.org/04/657204/4/check/cross-horizon-npm/93d6c5d/ : SUCCESS in 7m 48s\n- cross-keystone-py27 http://logs.openstack.org/04/657204/4/check/cross-keystone-py27/2651a96/ : SUCCESS in 24m 20s\n- cross-keystone-py36 http://logs.openstack.org/04/657204/4/check/cross-keystone-py36/617df20/ : SUCCESS in 27m 52s\n- cross-neutron-py27 http://logs.openstack.org/04/657204/4/check/cross-neutron-py27/0a3fd6a/ : SUCCESS in 24m 23s\n- cross-neutron-py36 http://logs.openstack.org/04/657204/4/check/cross-neutron-py36/8fa2018/ : SUCCESS in 21m 34s\n- cross-nova-py27 http://logs.openstack.org/04/657204/4/check/cross-nova-py27/ce36d91/ : SUCCESS in 12m 41s\n- cross-nova-py36 http://logs.openstack.org/04/657204/4/check/cross-nova-py36/8514840/ : SUCCESS in 20m 49s\n- cross-nova-functional http://logs.openstack.org/04/657204/4/check/cross-nova-functional/8bc16d9/ : SUCCESS in 18m 45s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/04/657204/4/check/cross-osc-build-sphinx-docs/02d7281/html/ : SUCCESS in 5m 16s\n- cross-swift-py27 http://logs.openstack.org/04/657204/4/check/cross-swift-py27/5c06db2/ : SUCCESS in 8m 42s\n- cross-swift-py36 http://logs.openstack.org/04/657204/4/check/cross-swift-py36/646772e/ : SUCCESS in 8m 46s\n- requirements-integration http://logs.openstack.org/04/657204/4/check/requirements-integration/676eee5/ : FAILURE in 34m 01s\n- neutron-grenade-multinode http://logs.openstack.org/04/657204/4/check/neutron-grenade-multinode/7a587a0/ : SUCCESS in 53m 22s\n- tempest-full http://logs.openstack.org/04/657204/4/check/tempest-full/8cdb70a/ : SUCCESS in 1h 53m 28s\n- neutron-grenade http://logs.openstack.org/04/657204/4/check/neutron-grenade/e48bb11/ : SUCCESS in 59m 06s","accounts_in_message":[],"_revision_number":4},{"id":"a2f3ade8ba83f11501244b9d2cf2f1a620018a04","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 14:21:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5dfed79e5d0e13f85d768416039f52ecaf176e10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 17:40:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/04/657204/5/check/openstack-tox-docs/85fb11c/html/ : SUCCESS in 4m 08s\n- openstack-tox-pypy http://logs.openstack.org/04/657204/5/check/openstack-tox-pypy/04f9f78/ : FAILURE in 2m 36s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/04/657204/5/check/openstack-tox-pep8/1f572d3/ : SUCCESS in 2m 20s\n- openstack-tox-py27 http://logs.openstack.org/04/657204/5/check/openstack-tox-py27/7c01d4d/ : FAILURE in 2m 38s\n- openstack-tox-py36 http://logs.openstack.org/04/657204/5/check/openstack-tox-py36/9fc7a89/ : SUCCESS in 3m 30s\n- openstack-tox-validate http://logs.openstack.org/04/657204/5/check/openstack-tox-validate/0d50d3f/ : SUCCESS in 2m 49s\n- requirements-tox-babel http://logs.openstack.org/04/657204/5/check/requirements-tox-babel/958c4a1/ : SUCCESS in 2m 44s\n- requirements-tox-py27-check-uc http://logs.openstack.org/04/657204/5/check/requirements-tox-py27-check-uc/cf69c4a/ : SUCCESS in 11m 52s\n- requirements-tox-py36-check-uc http://logs.openstack.org/04/657204/5/check/requirements-tox-py36-check-uc/d404d91/ : SUCCESS in 6m 10s\n- cross-cinder-py27 http://logs.openstack.org/04/657204/5/check/cross-cinder-py27/7df0764/ : SUCCESS in 9m 08s\n- cross-cinder-py36 http://logs.openstack.org/04/657204/5/check/cross-cinder-py36/866c527/ : SUCCESS in 7m 56s\n- cross-glance-py27 http://logs.openstack.org/04/657204/5/check/cross-glance-py27/8fb6f48/ : SUCCESS in 4m 23s\n- cross-glance-py36 http://logs.openstack.org/04/657204/5/check/cross-glance-py36/b521dea/ : SUCCESS in 4m 18s\n- cross-horizon-py27 http://logs.openstack.org/04/657204/5/check/cross-horizon-py27/7e6f14d/ : SUCCESS in 10m 00s\n- cross-horizon-py36 http://logs.openstack.org/04/657204/5/check/cross-horizon-py36/3469c0c/ : SUCCESS in 8m 54s\n- cross-horizon-npm http://logs.openstack.org/04/657204/5/check/cross-horizon-npm/4c58594/ : SUCCESS in 5m 56s\n- cross-keystone-py27 http://logs.openstack.org/04/657204/5/check/cross-keystone-py27/bd62a0f/ : SUCCESS in 22m 55s\n- cross-keystone-py36 http://logs.openstack.org/04/657204/5/check/cross-keystone-py36/8122cfa/ : SUCCESS in 23m 19s\n- cross-neutron-py27 http://logs.openstack.org/04/657204/5/check/cross-neutron-py27/7eea76e/ : SUCCESS in 24m 52s\n- cross-neutron-py36 http://logs.openstack.org/04/657204/5/check/cross-neutron-py36/537cb3d/ : SUCCESS in 19m 57s\n- cross-nova-py27 http://logs.openstack.org/04/657204/5/check/cross-nova-py27/670c0d1/ : SUCCESS in 16m 37s\n- cross-nova-py36 http://logs.openstack.org/04/657204/5/check/cross-nova-py36/9e8c3c3/ : SUCCESS in 11m 43s\n- cross-nova-functional http://logs.openstack.org/04/657204/5/check/cross-nova-functional/8624ed3/ : SUCCESS in 24m 04s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/04/657204/5/check/cross-osc-build-sphinx-docs/a0b09e5/html/ : SUCCESS in 5m 19s\n- cross-swift-py27 http://logs.openstack.org/04/657204/5/check/cross-swift-py27/f6735a5/ : SUCCESS in 7m 46s\n- cross-swift-py36 http://logs.openstack.org/04/657204/5/check/cross-swift-py36/e6d3f2d/ : SUCCESS in 7m 05s\n- requirements-integration http://logs.openstack.org/04/657204/5/check/requirements-integration/afa255c/ : FAILURE in 35m 03s\n- neutron-grenade-multinode http://logs.openstack.org/04/657204/5/check/neutron-grenade-multinode/58caf11/ : SUCCESS in 1h 10m 35s\n- tempest-full http://logs.openstack.org/04/657204/5/check/tempest-full/b4af739/ : SUCCESS in 1h 44m 38s\n- neutron-grenade http://logs.openstack.org/04/657204/5/check/neutron-grenade/fa65236/ : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":5},{"id":"5ca4f58f75affbe4814ea8eef034eece387062aa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-11 19:27:57.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d102ab2325538951765416c11ea7fffb0ac41731","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 21:14:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/04/657204/5/check/openstack-tox-docs/bf8b954/html/ : SUCCESS in 4m 13s\n- openstack-tox-pypy http://logs.openstack.org/04/657204/5/check/openstack-tox-pypy/ec9c802/ : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/04/657204/5/check/openstack-tox-pep8/d5f0d2d/ : SUCCESS in 2m 26s\n- openstack-tox-py27 http://logs.openstack.org/04/657204/5/check/openstack-tox-py27/d8fe438/ : SUCCESS in 2m 42s\n- openstack-tox-py36 http://logs.openstack.org/04/657204/5/check/openstack-tox-py36/b80a2d5/ : SUCCESS in 4m 21s\n- openstack-tox-validate http://logs.openstack.org/04/657204/5/check/openstack-tox-validate/234808a/ : SUCCESS in 2m 43s\n- requirements-tox-babel http://logs.openstack.org/04/657204/5/check/requirements-tox-babel/4e436d9/ : SUCCESS in 2m 35s\n- requirements-tox-py27-check-uc http://logs.openstack.org/04/657204/5/check/requirements-tox-py27-check-uc/85c95b7/ : SUCCESS in 7m 43s\n- requirements-tox-py36-check-uc http://logs.openstack.org/04/657204/5/check/requirements-tox-py36-check-uc/345a8f7/ : SUCCESS in 8m 02s\n- cross-cinder-py27 http://logs.openstack.org/04/657204/5/check/cross-cinder-py27/34ab42a/ : SUCCESS in 10m 25s\n- cross-cinder-py36 http://logs.openstack.org/04/657204/5/check/cross-cinder-py36/36530ff/ : SUCCESS in 8m 40s\n- cross-glance-py27 http://logs.openstack.org/04/657204/5/check/cross-glance-py27/da16a94/ : SUCCESS in 3m 41s\n- cross-glance-py36 http://logs.openstack.org/04/657204/5/check/cross-glance-py36/2c9d94e/ : SUCCESS in 4m 36s\n- cross-horizon-py27 http://logs.openstack.org/04/657204/5/check/cross-horizon-py27/78a1efe/ : SUCCESS in 11m 02s\n- cross-horizon-py36 http://logs.openstack.org/04/657204/5/check/cross-horizon-py36/499978c/ : SUCCESS in 9m 14s\n- cross-horizon-npm http://logs.openstack.org/04/657204/5/check/cross-horizon-npm/466544a/ : SUCCESS in 7m 33s\n- cross-keystone-py27 http://logs.openstack.org/04/657204/5/check/cross-keystone-py27/ecd7d6d/ : SUCCESS in 21m 23s\n- cross-keystone-py36 http://logs.openstack.org/04/657204/5/check/cross-keystone-py36/e4207a4/ : SUCCESS in 21m 07s\n- cross-neutron-py27 http://logs.openstack.org/04/657204/5/check/cross-neutron-py27/47bec31/ : SUCCESS in 20m 40s\n- cross-neutron-py36 http://logs.openstack.org/04/657204/5/check/cross-neutron-py36/0b8a22b/ : SUCCESS in 24m 19s\n- cross-nova-py27 http://logs.openstack.org/04/657204/5/check/cross-nova-py27/4eac8e5/ : SUCCESS in 14m 43s\n- cross-nova-py36 http://logs.openstack.org/04/657204/5/check/cross-nova-py36/fdf441e/ : SUCCESS in 11m 18s\n- cross-nova-functional http://logs.openstack.org/04/657204/5/check/cross-nova-functional/7193008/ : SUCCESS in 19m 16s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/04/657204/5/check/cross-osc-build-sphinx-docs/650e1fb/html/ : SUCCESS in 4m 59s\n- cross-swift-py27 http://logs.openstack.org/04/657204/5/check/cross-swift-py27/7018d26/ : SUCCESS in 7m 35s\n- cross-swift-py36 http://logs.openstack.org/04/657204/5/check/cross-swift-py36/ce3bd10/ : SUCCESS in 7m 02s\n- requirements-integration http://logs.openstack.org/04/657204/5/check/requirements-integration/5f48683/ : SUCCESS in 57m 33s\n- neutron-grenade-multinode http://logs.openstack.org/04/657204/5/check/neutron-grenade-multinode/8b9cd7f/ : SUCCESS in 1h 04m 01s\n- tempest-full http://logs.openstack.org/04/657204/5/check/tempest-full/068a8c2/ : SUCCESS in 1h 41m 36s\n- neutron-grenade http://logs.openstack.org/04/657204/5/check/neutron-grenade/6f653c8/ : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":5},{"id":"0203bc99de9b424f90fe6b1ae326323596794e73","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-11 21:18:33.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f31ff37448bb3c0047dbab97f0616fb873d19f2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 21:18:46.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7202f41a822c6b7b6376a4313b0508b552fde337","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 23:45:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"5f79025e1a0cc7354ffc4245d34871e57ccdcd16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 23:45:10.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/04/657204/5/gate/openstack-tox-docs/c95c1f2/html/ : SUCCESS in 3m 10s\n- openstack-tox-pep8 http://logs.openstack.org/04/657204/5/gate/openstack-tox-pep8/4c352be/ : SUCCESS in 2m 44s\n- openstack-tox-py27 http://logs.openstack.org/04/657204/5/gate/openstack-tox-py27/6086620/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/04/657204/5/gate/openstack-tox-py36/96b1d39/ : SUCCESS in 3m 25s\n- openstack-tox-validate http://logs.openstack.org/04/657204/5/gate/openstack-tox-validate/0ec28f4/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/04/657204/5/gate/requirements-tox-babel/503a256/ : SUCCESS in 4m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/04/657204/5/gate/requirements-tox-py27-check-uc/a8f899a/ : SUCCESS in 6m 23s\n- requirements-tox-py36-check-uc http://logs.openstack.org/04/657204/5/gate/requirements-tox-py36-check-uc/86671b2/ : SUCCESS in 5m 43s\n- cross-cinder-py27 http://logs.openstack.org/04/657204/5/gate/cross-cinder-py27/1ae2bc5/ : SUCCESS in 8m 12s\n- cross-cinder-py36 http://logs.openstack.org/04/657204/5/gate/cross-cinder-py36/4e24439/ : SUCCESS in 8m 30s\n- cross-glance-py27 http://logs.openstack.org/04/657204/5/gate/cross-glance-py27/6a96d07/ : SUCCESS in 3m 48s\n- cross-glance-py36 http://logs.openstack.org/04/657204/5/gate/cross-glance-py36/1ca225e/ : SUCCESS in 3m 58s\n- cross-horizon-py27 http://logs.openstack.org/04/657204/5/gate/cross-horizon-py27/d9b4df3/ : SUCCESS in 12m 33s\n- cross-horizon-py36 http://logs.openstack.org/04/657204/5/gate/cross-horizon-py36/2caee8a/ : SUCCESS in 8m 57s\n- cross-horizon-npm http://logs.openstack.org/04/657204/5/gate/cross-horizon-npm/2dcde88/ : SUCCESS in 9m 02s\n- cross-keystone-py27 http://logs.openstack.org/04/657204/5/gate/cross-keystone-py27/609c31c/ : SUCCESS in 30m 08s\n- cross-keystone-py36 http://logs.openstack.org/04/657204/5/gate/cross-keystone-py36/0c8dff8/ : SUCCESS in 17m 56s\n- cross-neutron-py27 http://logs.openstack.org/04/657204/5/gate/cross-neutron-py27/cc9aa72/ : SUCCESS in 26m 15s\n- cross-neutron-py36 http://logs.openstack.org/04/657204/5/gate/cross-neutron-py36/0ef101c/ : SUCCESS in 29m 43s\n- cross-nova-py27 http://logs.openstack.org/04/657204/5/gate/cross-nova-py27/e0488db/ : SUCCESS in 15m 20s\n- cross-nova-py36 http://logs.openstack.org/04/657204/5/gate/cross-nova-py36/3e7f239/ : SUCCESS in 21m 34s\n- cross-nova-functional http://logs.openstack.org/04/657204/5/gate/cross-nova-functional/516d117/ : SUCCESS in 18m 02s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/04/657204/5/gate/cross-osc-build-sphinx-docs/295f136/html/ : SUCCESS in 7m 23s\n- cross-swift-py27 http://logs.openstack.org/04/657204/5/gate/cross-swift-py27/7ef486b/ : SUCCESS in 7m 22s\n- cross-swift-py36 http://logs.openstack.org/04/657204/5/gate/cross-swift-py36/32240f8/ : SUCCESS in 9m 09s\n- requirements-integration http://logs.openstack.org/04/657204/5/gate/requirements-integration/e9de663/ : SUCCESS in 47m 32s\n- neutron-grenade-multinode http://logs.openstack.org/04/657204/5/gate/neutron-grenade-multinode/a8d91ed/ : SUCCESS in 1h 05m 01s\n- tempest-full http://logs.openstack.org/04/657204/5/gate/tempest-full/e5d2454/ : SUCCESS in 1h 31m 01s\n- neutron-grenade http://logs.openstack.org/04/657204/5/gate/neutron-grenade/c451921/ : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d9ebcf0b6232a53ec353450f6209b127e543b061","revisions":{"53d252b87c70d9d42f23f68d212a85aaf6dc1d85":{"kind":"REWORK","_number":1,"created":"2019-05-05 06:32:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/657204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/657204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/657204/1"}}},"commit":{"parents":[{"commit":"6affabb05381072509f382af865a54a7916b7664","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6affabb05381072509f382af865a54a7916b7664"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-05 06:32:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-05 06:32:09.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I68ddaae1845917f4b13bfdcbfe6caf71bcf514a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53d252b87c70d9d42f23f68d212a85aaf6dc1d85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/53d252b87c70d9d42f23f68d212a85aaf6dc1d85"}]},"branch":"refs/heads/master"},"a50eed6a32b3952cde760869326cb074aae055c2":{"kind":"REWORK","_number":2,"created":"2019-05-07 06:44:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/657204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/657204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/657204/2"}}},"commit":{"parents":[{"commit":"16e8fd6a0f179e3cad0ae2197a5c1be18ada2ea1","subject":"update constraint for openstackdocstheme to new release 1.29.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/16e8fd6a0f179e3cad0ae2197a5c1be18ada2ea1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-07 06:44:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-07 06:44:29.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I68ddaae1845917f4b13bfdcbfe6caf71bcf514a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a50eed6a32b3952cde760869326cb074aae055c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a50eed6a32b3952cde760869326cb074aae055c2"}]},"branch":"refs/heads/master"},"0ee595ce71b50391d7c3951fed866569a7666da3":{"kind":"REWORK","_number":3,"created":"2019-05-07 14:13:41.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/04/657204/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/657204/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/657204/3"}}},"commit":{"parents":[{"commit":"16e8fd6a0f179e3cad0ae2197a5c1be18ada2ea1","subject":"update constraint for openstackdocstheme to new release 1.29.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/16e8fd6a0f179e3cad0ae2197a5c1be18ada2ea1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-07 06:44:29.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-07 14:13:28.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I68ddaae1845917f4b13bfdcbfe6caf71bcf514a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ee595ce71b50391d7c3951fed866569a7666da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ee595ce71b50391d7c3951fed866569a7666da3"}]},"branch":"refs/heads/master"},"60647525d5818e2efb9c3ad92ab5bfc2883d3a63":{"kind":"REWORK","_number":4,"created":"2019-05-09 15:46:22.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/04/657204/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/657204/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/657204/4"}}},"commit":{"parents":[{"commit":"16e8fd6a0f179e3cad0ae2197a5c1be18ada2ea1","subject":"update constraint for openstackdocstheme to new release 1.29.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/16e8fd6a0f179e3cad0ae2197a5c1be18ada2ea1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-07 06:44:29.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-09 15:46:07.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I68ddaae1845917f4b13bfdcbfe6caf71bcf514a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/60647525d5818e2efb9c3ad92ab5bfc2883d3a63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/60647525d5818e2efb9c3ad92ab5bfc2883d3a63"}]},"branch":"refs/heads/master"},"d9ebcf0b6232a53ec353450f6209b127e543b061":{"kind":"REWORK","_number":5,"created":"2019-05-10 14:21:26.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/04/657204/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/657204/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/657204/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/657204/5"}}},"commit":{"parents":[{"commit":"16e8fd6a0f179e3cad0ae2197a5c1be18ada2ea1","subject":"update constraint for openstackdocstheme to new release 1.29.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/16e8fd6a0f179e3cad0ae2197a5c1be18ada2ea1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-07 06:44:29.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-05-10 14:21:16.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I68ddaae1845917f4b13bfdcbfe6caf71bcf514a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9ebcf0b6232a53ec353450f6209b127e543b061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9ebcf0b6232a53ec353450f6209b127e543b061"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
