)]}'
{"id":"openstack%2Frequirements~657395","triplet_id":"openstack%2Frequirements~master~Ia8f848a60440d28b20b09f1a83c72a6733e3f523","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Ia8f848a60440d28b20b09f1a83c72a6733e3f523","subject":"Add cinder extras requirements","status":"ABANDONED","created":"2019-05-06 15:03:44.000000000","updated":"2019-05-09 14:26:39.000000000","total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f6243a2288ebc49ae11b038da286028578b3914e","_number":657395,"virtual_id_number":657395,"owner":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"date":"2019-05-08 17:04:25.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2019-05-07 21:37:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-05-08 14:18:03.000000000","_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":-1,"date":"2019-05-08 04:16:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":-1,"date":"2019-05-08 19:24:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-06 15:21:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-07 21:37:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-08 04:16:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-08 17:04:25.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-05-08 19:24:45.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"}],"messages":[{"id":"39cdfb002c4ec19054739019c01a7e95310ec452","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-06 15:03:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50fb0df9f4eed4d2151ed3a66f083b1afc1d7ea0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-06 15:21:03.000000000","message":"Patch Set 1:\n\nTypically when we add a new requirement, we want a set of questions answered about each one:\n\nhttps://docs.openstack.org/project-team-guide/dependency-management.html#for-new-requirements\n\nThis is a little different though. These really have been dependencies of Cinder for a long time now, we just handled the management of them poorly and this is now attempting to get things right.\n\nI\u0027ll leave it up to Matt if we need to break this out for each and have those questions answered.","accounts_in_message":[],"_revision_number":1},{"id":"6e9b5fe437903265d8dcb41c92796b49a3896e2c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-06 15:31:51.000000000","message":"Topic set to driver_exceptions","accounts_in_message":[],"_revision_number":1},{"id":"2e362adf60662b0787dc140bac8b1c4aa9e0105b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-06 15:32:03.000000000","message":"Topic driver_exceptions removed","accounts_in_message":[],"_revision_number":1},{"id":"76084aee2ea1a186e12a4131c706e8a01e6a5005","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 16:49:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/657395/1/check/openstack-tox-docs/7927779/html/ : SUCCESS in 5m 38s\n- openstack-tox-pypy http://logs.openstack.org/95/657395/1/check/openstack-tox-pypy/ac57575/ : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/95/657395/1/check/openstack-tox-pep8/a0a599a/ : SUCCESS in 2m 44s\n- openstack-tox-py27 http://logs.openstack.org/95/657395/1/check/openstack-tox-py27/8eeb440/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/95/657395/1/check/openstack-tox-py36/55ee2fe/ : SUCCESS in 3m 37s\n- openstack-tox-validate http://logs.openstack.org/95/657395/1/check/openstack-tox-validate/e824fc4/ : SUCCESS in 2m 43s\n- requirements-tox-babel http://logs.openstack.org/95/657395/1/check/requirements-tox-babel/da80bb8/ : SUCCESS in 2m 52s\n- requirements-tox-py27-check-uc http://logs.openstack.org/95/657395/1/check/requirements-tox-py27-check-uc/8ee9661/ : FAILURE in 14m 07s\n- requirements-tox-py36-check-uc http://logs.openstack.org/95/657395/1/check/requirements-tox-py36-check-uc/d3cb481/ : FAILURE in 14m 14s\n- cross-cinder-py27 http://logs.openstack.org/95/657395/1/check/cross-cinder-py27/add3ac9/ : SUCCESS in 8m 36s\n- cross-cinder-py36 http://logs.openstack.org/95/657395/1/check/cross-cinder-py36/83b5956/ : SUCCESS in 8m 25s\n- cross-glance-py27 http://logs.openstack.org/95/657395/1/check/cross-glance-py27/feda072/ : SUCCESS in 4m 02s\n- cross-glance-py36 http://logs.openstack.org/95/657395/1/check/cross-glance-py36/e880d34/ : SUCCESS in 4m 27s\n- cross-horizon-py27 http://logs.openstack.org/95/657395/1/check/cross-horizon-py27/74c32a7/ : SUCCESS in 12m 56s\n- cross-horizon-py36 http://logs.openstack.org/95/657395/1/check/cross-horizon-py36/5b3ebcd/ : SUCCESS in 9m 32s\n- cross-horizon-npm http://logs.openstack.org/95/657395/1/check/cross-horizon-npm/25b371b/ : SUCCESS in 5m 39s\n- cross-keystone-py27 http://logs.openstack.org/95/657395/1/check/cross-keystone-py27/5bb180d/ : SUCCESS in 20m 00s\n- cross-keystone-py36 http://logs.openstack.org/95/657395/1/check/cross-keystone-py36/b553eae/ : SUCCESS in 23m 19s\n- cross-neutron-py27 http://logs.openstack.org/95/657395/1/check/cross-neutron-py27/314c4bc/ : SUCCESS in 25m 26s\n- cross-neutron-py36 http://logs.openstack.org/95/657395/1/check/cross-neutron-py36/4f0f74b/ : SUCCESS in 21m 31s\n- cross-nova-py27 http://logs.openstack.org/95/657395/1/check/cross-nova-py27/cfa938c/ : SUCCESS in 13m 21s\n- cross-nova-py36 http://logs.openstack.org/95/657395/1/check/cross-nova-py36/61788ab/ : SUCCESS in 11m 38s\n- cross-nova-functional http://logs.openstack.org/95/657395/1/check/cross-nova-functional/79a0899/ : SUCCESS in 18m 03s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/95/657395/1/check/cross-osc-build-sphinx-docs/cd9acd7/html/ : SUCCESS in 5m 03s\n- cross-swift-py27 http://logs.openstack.org/95/657395/1/check/cross-swift-py27/ca1ded4/ : SUCCESS in 8m 32s\n- cross-swift-py36 http://logs.openstack.org/95/657395/1/check/cross-swift-py36/dfca286/ : SUCCESS in 7m 11s\n- requirements-integration http://logs.openstack.org/95/657395/1/check/requirements-integration/9688e93/ : SUCCESS in 1h 24m 42s\n- neutron-grenade-multinode http://logs.openstack.org/95/657395/1/check/neutron-grenade-multinode/d37f5d8/ : SUCCESS in 1h 08m 14s\n- tempest-full http://logs.openstack.org/95/657395/1/check/tempest-full/4c6e0ba/ : SUCCESS in 1h 43m 25s\n- neutron-grenade http://logs.openstack.org/95/657395/1/check/neutron-grenade/9250bb5/ : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":1},{"id":"243428b8f353468ea8cd171a920ec2b7aa611251","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-06 19:14:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"36381d6a7c5d9bb426d6342effd4bd91e7a271bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 21:05:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/657395/2/check/openstack-tox-docs/1bf3731/html/ : SUCCESS in 2m 31s\n- openstack-tox-pypy http://logs.openstack.org/95/657395/2/check/openstack-tox-pypy/f5a6688/ : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/95/657395/2/check/openstack-tox-pep8/756d871/ : SUCCESS in 2m 36s\n- openstack-tox-py27 http://logs.openstack.org/95/657395/2/check/openstack-tox-py27/fe5add6/ : SUCCESS in 2m 57s\n- openstack-tox-py36 http://logs.openstack.org/95/657395/2/check/openstack-tox-py36/879bb26/ : SUCCESS in 3m 06s\n- openstack-tox-validate http://logs.openstack.org/95/657395/2/check/openstack-tox-validate/b6858c3/ : SUCCESS in 2m 41s\n- requirements-tox-babel http://logs.openstack.org/95/657395/2/check/requirements-tox-babel/93b1400/ : SUCCESS in 2m 32s\n- requirements-tox-py27-check-uc http://logs.openstack.org/95/657395/2/check/requirements-tox-py27-check-uc/208cb32/ : SUCCESS in 16m 29s\n- requirements-tox-py36-check-uc http://logs.openstack.org/95/657395/2/check/requirements-tox-py36-check-uc/5ebdd78/ : SUCCESS in 14m 07s\n- cross-cinder-py27 http://logs.openstack.org/95/657395/2/check/cross-cinder-py27/c6f3498/ : SUCCESS in 8m 20s\n- cross-cinder-py36 http://logs.openstack.org/95/657395/2/check/cross-cinder-py36/8bfe506/ : SUCCESS in 8m 24s\n- cross-glance-py27 http://logs.openstack.org/95/657395/2/check/cross-glance-py27/1c2cb54/ : SUCCESS in 3m 57s\n- cross-glance-py36 http://logs.openstack.org/95/657395/2/check/cross-glance-py36/8d394ab/ : SUCCESS in 4m 17s\n- cross-horizon-py27 http://logs.openstack.org/95/657395/2/check/cross-horizon-py27/b86d569/ : SUCCESS in 11m 10s\n- cross-horizon-py36 http://logs.openstack.org/95/657395/2/check/cross-horizon-py36/3280f2c/ : SUCCESS in 9m 47s\n- cross-horizon-npm http://logs.openstack.org/95/657395/2/check/cross-horizon-npm/0ce8b88/ : SUCCESS in 5m 34s\n- cross-keystone-py27 http://logs.openstack.org/95/657395/2/check/cross-keystone-py27/edda321/ : SUCCESS in 24m 36s\n- cross-keystone-py36 http://logs.openstack.org/95/657395/2/check/cross-keystone-py36/abea8d0/ : SUCCESS in 22m 03s\n- cross-neutron-py27 http://logs.openstack.org/95/657395/2/check/cross-neutron-py27/98537d5/ : SUCCESS in 21m 50s\n- cross-neutron-py36 http://logs.openstack.org/95/657395/2/check/cross-neutron-py36/7b33223/ : SUCCESS in 23m 53s\n- cross-nova-py27 http://logs.openstack.org/95/657395/2/check/cross-nova-py27/5a84e26/ : SUCCESS in 12m 20s\n- cross-nova-py36 http://logs.openstack.org/95/657395/2/check/cross-nova-py36/9fc239d/ : SUCCESS in 11m 41s\n- cross-nova-functional http://logs.openstack.org/95/657395/2/check/cross-nova-functional/1620064/ : SUCCESS in 17m 10s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/95/657395/2/check/cross-osc-build-sphinx-docs/b3f86e6/html/ : SUCCESS in 4m 42s\n- cross-swift-py27 http://logs.openstack.org/95/657395/2/check/cross-swift-py27/5e1e03a/ : SUCCESS in 7m 05s\n- cross-swift-py36 http://logs.openstack.org/95/657395/2/check/cross-swift-py36/d8e1ee1/ : SUCCESS in 7m 26s\n- requirements-integration http://logs.openstack.org/95/657395/2/check/requirements-integration/f2b91bd/ : SUCCESS in 1h 11m 39s\n- neutron-grenade-multinode http://logs.openstack.org/95/657395/2/check/neutron-grenade-multinode/b0759bb/ : SUCCESS in 1h 09m 42s\n- tempest-full http://logs.openstack.org/95/657395/2/check/tempest-full/4213fc7/ : SUCCESS in 1h 44m 33s\n- neutron-grenade http://logs.openstack.org/95/657395/2/check/neutron-grenade/4745204/ : SUCCESS in 1h 00m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b278c7ed9bc37871b653ac92f96b1394d1c677a4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-06 21:14:14.000000000","message":"Patch Set 2:\n\nI think we still need the questions answered, at the very least we need the licences checked, but if we are adding it to GR that kinda means we looked at it at least once for maintainability, testing, activity, etc.  It\u0027d probably be easiest to do this all at once, the licence info is in the commit (GR comments), so that\u0027s one step.","accounts_in_message":[],"_revision_number":2},{"id":"127d0a49375474d6c089eda224854c16cb645db1","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-07 14:25:01.000000000","message":"Patch Set 2:\n\nI don\u0027t know the answers to the questions as I don\u0027t maintain the libraries.   I can vouch for the hpe python libs as I created those years ago, but without this patch, we cannot add the optional extras to cinder to make sure the required libraries are available for our drivers.","accounts_in_message":[],"_revision_number":2},{"id":"e1d323b8b748dc125d8ebcb7ee24a9e8a63c4ca0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-07 14:38:41.000000000","message":"Patch Set 2:\n\nI\u0027m going to start reviewing them today, will likely make one comment per library reviewing them.","accounts_in_message":[],"_revision_number":2},{"id":"50cc94a42dfecedf6bd9fd8c00d3d583f1b69c67","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-07 19:22:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"885969fb3cdea1273f2eb2070ed7078112d4b966","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 21:37:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/657395/3/check/openstack-tox-docs/bb28c02/html/ : SUCCESS in 3m 29s\n- openstack-tox-pypy http://logs.openstack.org/95/657395/3/check/openstack-tox-pypy/780ee3d/ : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/95/657395/3/check/openstack-tox-pep8/6133c24/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/95/657395/3/check/openstack-tox-py27/d66c424/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/95/657395/3/check/openstack-tox-py36/1ac5935/ : SUCCESS in 3m 19s\n- openstack-tox-validate http://logs.openstack.org/95/657395/3/check/openstack-tox-validate/d41ab71/ : SUCCESS in 3m 07s\n- requirements-tox-babel http://logs.openstack.org/95/657395/3/check/requirements-tox-babel/c4029e5/ : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc http://logs.openstack.org/95/657395/3/check/requirements-tox-py27-check-uc/4b4e8f0/ : SUCCESS in 13m 51s\n- requirements-tox-py36-check-uc http://logs.openstack.org/95/657395/3/check/requirements-tox-py36-check-uc/bdabcad/ : SUCCESS in 16m 30s\n- cross-cinder-py27 http://logs.openstack.org/95/657395/3/check/cross-cinder-py27/6cf5c66/ : SUCCESS in 8m 56s\n- cross-cinder-py36 http://logs.openstack.org/95/657395/3/check/cross-cinder-py36/3f57574/ : SUCCESS in 8m 41s\n- cross-glance-py27 http://logs.openstack.org/95/657395/3/check/cross-glance-py27/2c3c9a7/ : SUCCESS in 4m 28s\n- cross-glance-py36 http://logs.openstack.org/95/657395/3/check/cross-glance-py36/59c42c6/ : SUCCESS in 4m 14s\n- cross-horizon-py27 http://logs.openstack.org/95/657395/3/check/cross-horizon-py27/692ed28/ : SUCCESS in 10m 09s\n- cross-horizon-py36 http://logs.openstack.org/95/657395/3/check/cross-horizon-py36/57701e8/ : SUCCESS in 8m 28s\n- cross-horizon-npm http://logs.openstack.org/95/657395/3/check/cross-horizon-npm/28a8047/ : SUCCESS in 7m 08s\n- cross-keystone-py27 http://logs.openstack.org/95/657395/3/check/cross-keystone-py27/b79c4c4/ : SUCCESS in 20m 19s\n- cross-keystone-py36 http://logs.openstack.org/95/657395/3/check/cross-keystone-py36/0e7d09d/ : SUCCESS in 24m 17s\n- cross-neutron-py27 http://logs.openstack.org/95/657395/3/check/cross-neutron-py27/d9bb399/ : SUCCESS in 31m 00s\n- cross-neutron-py36 http://logs.openstack.org/95/657395/3/check/cross-neutron-py36/473bc02/ : SUCCESS in 25m 01s\n- cross-nova-py27 http://logs.openstack.org/95/657395/3/check/cross-nova-py27/045146e/ : SUCCESS in 12m 25s\n- cross-nova-py36 http://logs.openstack.org/95/657395/3/check/cross-nova-py36/71742d3/ : SUCCESS in 11m 38s\n- cross-nova-functional http://logs.openstack.org/95/657395/3/check/cross-nova-functional/82e9d63/ : SUCCESS in 19m 59s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/95/657395/3/check/cross-osc-build-sphinx-docs/6c90369/html/ : SUCCESS in 5m 02s\n- cross-swift-py27 http://logs.openstack.org/95/657395/3/check/cross-swift-py27/85f10e7/ : SUCCESS in 8m 32s\n- cross-swift-py36 http://logs.openstack.org/95/657395/3/check/cross-swift-py36/b0defea/ : SUCCESS in 8m 40s\n- requirements-integration http://logs.openstack.org/95/657395/3/check/requirements-integration/2276206/ : SUCCESS in 1h 13m 34s\n- neutron-grenade-multinode http://logs.openstack.org/95/657395/3/check/neutron-grenade-multinode/8349303/ : SUCCESS in 1h 34m 59s\n- tempest-full http://logs.openstack.org/95/657395/3/check/tempest-full/fc0a75d/ : SUCCESS in 1h 55m 59s\n- neutron-grenade http://logs.openstack.org/95/657395/3/check/neutron-grenade/c21c037/ : SUCCESS in 1h 09m 31s","accounts_in_message":[],"_revision_number":3},{"id":"a924980c089eb2dca56d8ef823a2b852531475ae","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-08 04:16:56.000000000","message":"Patch Set 3: Code-Review-1\n\n(12 comments)\n\nthere are a few problems, mainly with stuff that is not active anymore and not py3, but of those are kinda hard stoppers","accounts_in_message":[],"_revision_number":3},{"id":"63d773965e412fe4eb9350be95709f933e52074c","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-08 11:23:49.000000000","message":"Patch Set 3:\n\nSo which libs can we get in and which can\u0027t get in?\nThe reason we are adding these is so cinder can support\nrunning the volume service in a container image by automatically pulling in the required driver libraries.","accounts_in_message":[],"_revision_number":3},{"id":"d8ca4aaf1d8e26445b856bdd5343810f7d8313f8","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-08 14:18:03.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nshall I break this patch up into individual patches for each library?","accounts_in_message":[],"_revision_number":3},{"id":"c497e3a6620356345caff8925ebb3e03d2172083","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-08 17:04:25.000000000","message":"Patch Set 3:\n\n\u003e (3 comments)\n \u003e \n \u003e shall I break this patch up into individual patches for each\n \u003e library?\n\nI suggest breaking this into 2 changes.  1 with all the ones that are okay) and one for the final 3.\n\nWe can get the first one merged this week while we work through the problems.  On the cinder side you can use the exiting change and make it depend on the second (new) change or make the same split as you\u0027re doing here","accounts_in_message":[],"_revision_number":3},{"id":"b9b72acb99fd58abccefcd37f66ad87a7677b525","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-08 19:24:45.000000000","message":"Patch Set 3: Code-Review-1\n\nCan you fill out the new requirement template for each of those? I guess one review per requirement would be the best solution","accounts_in_message":[],"_revision_number":3},{"id":"f6243a2288ebc49ae11b038da286028578b3914e","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-09 14:26:39.000000000","message":"Abandoned\n\nMigrated this patch into separate patches for each cinder driver\u0027s libraries.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8b811b51b80808ef6628fd8fcaa61b3fd1fe7553","revisions":{"40fb1b691c989d31b1d8b997cf56a28dcafe2519":{"kind":"REWORK","_number":1,"created":"2019-05-06 15:03:44.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/95/657395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/657395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/657395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/657395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/657395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/657395/1"}}},"commit":{"parents":[{"commit":"6affabb05381072509f382af865a54a7916b7664","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6affabb05381072509f382af865a54a7916b7664"}]}],"author":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-06 14:39:36.000000000","tz":0},"committer":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-06 15:04:06.000000000","tz":0},"subject":"Add cinder extras requirements","message":"Add cinder extras requirements\n\nThis patch adds the 3rd party driver requirements for cinder\u0027s\ndrivers that have pypi installable packages.   We are adding\nextras to cinder\u0027s setup.cfg, and need these requirements added\nhere to pass the requirements-check.  The Cinder patch is here:\n\nhttps://review.opendev.org/#/c/656724/\n\nChange-Id: Ia8f848a60440d28b20b09f1a83c72a6733e3f523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40fb1b691c989d31b1d8b997cf56a28dcafe2519"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/40fb1b691c989d31b1d8b997cf56a28dcafe2519"}]},"branch":"refs/heads/master"},"ecdf77947f472aea19fa675aebc2801f96379704":{"kind":"REWORK","_number":2,"created":"2019-05-06 19:14:47.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/95/657395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/657395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/657395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/657395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/657395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/657395/2"}}},"commit":{"parents":[{"commit":"6affabb05381072509f382af865a54a7916b7664","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6affabb05381072509f382af865a54a7916b7664"}]}],"author":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-06 14:39:36.000000000","tz":0},"committer":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-06 19:15:00.000000000","tz":0},"subject":"Add cinder extras requirements","message":"Add cinder extras requirements\n\nThis patch adds the 3rd party driver requirements for cinder\u0027s\ndrivers that have pypi installable packages.   We are adding\nextras to cinder\u0027s setup.cfg, and need these requirements added\nhere to pass the requirements-check.  The Cinder patch is here:\n\nhttps://review.opendev.org/#/c/656724/\n\nChange-Id: Ia8f848a60440d28b20b09f1a83c72a6733e3f523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecdf77947f472aea19fa675aebc2801f96379704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecdf77947f472aea19fa675aebc2801f96379704"}]},"branch":"refs/heads/master"},"8b811b51b80808ef6628fd8fcaa61b3fd1fe7553":{"kind":"REWORK","_number":3,"created":"2019-05-07 19:22:03.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/95/657395/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/657395/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/657395/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/657395/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/657395/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/657395/3"}}},"commit":{"parents":[{"commit":"16e8fd6a0f179e3cad0ae2197a5c1be18ada2ea1","subject":"update constraint for openstackdocstheme to new release 1.29.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/16e8fd6a0f179e3cad0ae2197a5c1be18ada2ea1"}]}],"author":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-06 14:39:36.000000000","tz":0},"committer":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-07 19:22:18.000000000","tz":0},"subject":"Add cinder extras requirements","message":"Add cinder extras requirements\n\nThis patch adds the 3rd party driver requirements for cinder\u0027s\ndrivers that have pypi installable packages.   We are adding\nextras to cinder\u0027s setup.cfg, and need these requirements added\nhere to pass the requirements-check.  The Cinder patch is here:\n\nhttps://review.opendev.org/#/c/656724/\n\nChange-Id: Ia8f848a60440d28b20b09f1a83c72a6733e3f523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b811b51b80808ef6628fd8fcaa61b3fd1fe7553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b811b51b80808ef6628fd8fcaa61b3fd1fe7553"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
