)]}'
{"id":"openstack%2Frequirements~658095","triplet_id":"openstack%2Frequirements~master~I227bf514d15e301224a01252c5f5f784f040ad34","project":"openstack/requirements","branch":"master","topic":"cinder_extras","hashtags":[],"change_id":"I227bf514d15e301224a01252c5f5f784f040ad34","subject":"Add cinder extras storops library","status":"MERGED","created":"2019-05-09 14:11:01.000000000","updated":"2019-05-14 23:24:02.000000000","submitted":"2019-05-14 23:24:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658095-1557876242906-71816a1a","meta_rev_id":"2776d9161082fde3542e171e95aeabb049b72ca6","_number":658095,"virtual_id_number":658095,"owner":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-05-14 23:24:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-14 18:28:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-14 21:08:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-14 21:08:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-11 19:36:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-13 14:32:48.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-05-14 18:28:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-14 21:08:25.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-14 23:24:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4c5772d0c8ed775ad5ad6010068d8f8d336121f","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-09 14:11:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5588a8d3a43eb385e82c48ed359b5dab420526b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 18:01:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/658095/1/check/openstack-tox-docs/88bc545/html/ : SUCCESS in 2m 51s\n- openstack-tox-pypy http://logs.openstack.org/95/658095/1/check/openstack-tox-pypy/ff74255/ : FAILURE in 2m 58s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/95/658095/1/check/openstack-tox-pep8/d350f3f/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/95/658095/1/check/openstack-tox-py27/9e67e15/ : FAILURE in 2m 36s\n- openstack-tox-py36 http://logs.openstack.org/95/658095/1/check/openstack-tox-py36/2c2a7e9/ : SUCCESS in 3m 41s\n- openstack-tox-validate http://logs.openstack.org/95/658095/1/check/openstack-tox-validate/be0cf5b/ : SUCCESS in 2m 42s\n- requirements-tox-babel http://logs.openstack.org/95/658095/1/check/requirements-tox-babel/09ed5f1/ : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc http://logs.openstack.org/95/658095/1/check/requirements-tox-py27-check-uc/1b582a6/ : SUCCESS in 6m 31s\n- requirements-tox-py36-check-uc http://logs.openstack.org/95/658095/1/check/requirements-tox-py36-check-uc/7b91591/ : SUCCESS in 5m 30s\n- cross-cinder-py27 http://logs.openstack.org/95/658095/1/check/cross-cinder-py27/2466559/ : SUCCESS in 13m 41s\n- cross-cinder-py36 http://logs.openstack.org/95/658095/1/check/cross-cinder-py36/f0ad181/ : SUCCESS in 11m 25s\n- cross-glance-py27 http://logs.openstack.org/95/658095/1/check/cross-glance-py27/bf83165/ : SUCCESS in 5m 25s\n- cross-glance-py36 http://logs.openstack.org/95/658095/1/check/cross-glance-py36/0540b55/ : SUCCESS in 4m 20s\n- cross-horizon-py27 http://logs.openstack.org/95/658095/1/check/cross-horizon-py27/a1f82e8/ : SUCCESS in 10m 03s\n- cross-horizon-py36 http://logs.openstack.org/95/658095/1/check/cross-horizon-py36/9d2c8b9/ : SUCCESS in 9m 04s\n- cross-horizon-npm http://logs.openstack.org/95/658095/1/check/cross-horizon-npm/68cc7c3/ : SUCCESS in 5m 42s\n- cross-keystone-py27 http://logs.openstack.org/95/658095/1/check/cross-keystone-py27/87b1855/ : SUCCESS in 20m 44s\n- cross-keystone-py36 http://logs.openstack.org/95/658095/1/check/cross-keystone-py36/5991ccc/ : SUCCESS in 31m 22s\n- cross-neutron-py27 http://logs.openstack.org/95/658095/1/check/cross-neutron-py27/87e1182/ : SUCCESS in 21m 34s\n- cross-neutron-py36 http://logs.openstack.org/95/658095/1/check/cross-neutron-py36/aa9c646/ : SUCCESS in 32m 57s\n- cross-nova-py27 http://logs.openstack.org/95/658095/1/check/cross-nova-py27/14fb9b6/ : SUCCESS in 13m 50s\n- cross-nova-py36 http://logs.openstack.org/95/658095/1/check/cross-nova-py36/e36c825/ : SUCCESS in 12m 23s\n- cross-nova-functional http://logs.openstack.org/95/658095/1/check/cross-nova-functional/1c31963/ : SUCCESS in 21m 06s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/95/658095/1/check/cross-osc-build-sphinx-docs/e56e34e/html/ : SUCCESS in 5m 16s\n- cross-swift-py27 http://logs.openstack.org/95/658095/1/check/cross-swift-py27/f9171bc/ : SUCCESS in 7m 20s\n- cross-swift-py36 http://logs.openstack.org/95/658095/1/check/cross-swift-py36/755d49a/ : SUCCESS in 9m 11s\n- requirements-integration http://logs.openstack.org/95/658095/1/check/requirements-integration/74af461/ : FAILURE in 35m 03s\n- neutron-grenade-multinode http://logs.openstack.org/95/658095/1/check/neutron-grenade-multinode/7355619/ : SUCCESS in 1h 16m 56s\n- tempest-full http://logs.openstack.org/95/658095/1/check/tempest-full/b38ca03/ : SUCCESS in 1h 35m 53s\n- neutron-grenade http://logs.openstack.org/95/658095/1/check/neutron-grenade/c499290/ : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1aebb000642fd39ce7d06c72b75b6e5fee66ab65","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 05:09:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4a5f1e76a3af9ab08b8602fa5bd6e4429eed410","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 06:29:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/658095/1/check/openstack-tox-docs/951bd85/html/ : SUCCESS in 3m 27s\n- openstack-tox-pypy http://logs.openstack.org/95/658095/1/check/openstack-tox-pypy/4df2281/ : FAILURE in 6m 00s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/95/658095/1/check/openstack-tox-pep8/8a8bf09/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/95/658095/1/check/openstack-tox-py27/829dd0e/ : FAILURE in 5m 34s\n- openstack-tox-py36 http://logs.openstack.org/95/658095/1/check/openstack-tox-py36/b406c72/ : SUCCESS in 3m 55s\n- openstack-tox-validate http://logs.openstack.org/95/658095/1/check/openstack-tox-validate/7c746d5/ : SUCCESS in 5m 06s\n- requirements-tox-babel http://logs.openstack.org/95/658095/1/check/requirements-tox-babel/adf7b25/ : SUCCESS in 3m 28s\n- requirements-tox-py27-check-uc http://logs.openstack.org/95/658095/1/check/requirements-tox-py27-check-uc/b874bd8/ : SUCCESS in 10m 41s\n- requirements-tox-py36-check-uc http://logs.openstack.org/95/658095/1/check/requirements-tox-py36-check-uc/988f4c4/ : SUCCESS in 7m 21s\n- cross-cinder-py27 http://logs.openstack.org/95/658095/1/check/cross-cinder-py27/b26c25b/ : SUCCESS in 12m 30s\n- cross-cinder-py36 http://logs.openstack.org/95/658095/1/check/cross-cinder-py36/0f20741/ : SUCCESS in 8m 54s\n- cross-glance-py27 http://logs.openstack.org/95/658095/1/check/cross-glance-py27/29431bc/ : SUCCESS in 7m 15s\n- cross-glance-py36 http://logs.openstack.org/95/658095/1/check/cross-glance-py36/4be34bf/ : SUCCESS in 5m 34s\n- cross-horizon-py27 http://logs.openstack.org/95/658095/1/check/cross-horizon-py27/bb947e0/ : SUCCESS in 11m 25s\n- cross-horizon-py36 http://logs.openstack.org/95/658095/1/check/cross-horizon-py36/e86808e/ : SUCCESS in 9m 50s\n- cross-horizon-npm http://logs.openstack.org/95/658095/1/check/cross-horizon-npm/33181c8/ : SUCCESS in 9m 44s\n- cross-keystone-py27 http://logs.openstack.org/95/658095/1/check/cross-keystone-py27/ac91f15/ : SUCCESS in 20m 53s\n- cross-keystone-py36 http://logs.openstack.org/95/658095/1/check/cross-keystone-py36/9488388/ : SUCCESS in 22m 16s\n- cross-neutron-py27 http://logs.openstack.org/95/658095/1/check/cross-neutron-py27/de3a4e2/ : SUCCESS in 21m 48s\n- cross-neutron-py36 http://logs.openstack.org/95/658095/1/check/cross-neutron-py36/47bd278/ : SUCCESS in 20m 02s\n- cross-nova-py27 http://logs.openstack.org/95/658095/1/check/cross-nova-py27/f5aebe8/ : SUCCESS in 12m 37s\n- cross-nova-py36 http://logs.openstack.org/95/658095/1/check/cross-nova-py36/bfc5852/ : SUCCESS in 12m 53s\n- cross-nova-functional http://logs.openstack.org/95/658095/1/check/cross-nova-functional/4828870/ : SUCCESS in 22m 28s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/95/658095/1/check/cross-osc-build-sphinx-docs/e823d4f/html/ : SUCCESS in 6m 58s\n- cross-swift-py27 http://logs.openstack.org/95/658095/1/check/cross-swift-py27/be863de/ : SUCCESS in 9m 14s\n- cross-swift-py36 http://logs.openstack.org/95/658095/1/check/cross-swift-py36/b054e46/ : SUCCESS in 9m 05s\n- requirements-integration http://logs.openstack.org/95/658095/1/check/requirements-integration/c3dde51/ : FAILURE in 32m 09s\n- neutron-grenade-multinode http://logs.openstack.org/95/658095/1/check/neutron-grenade-multinode/86d35fc/ : SUCCESS in 1h 13m 06s\n- tempest-full http://logs.openstack.org/95/658095/1/check/tempest-full/c232f65/ : FAILURE in 43m 15s\n- neutron-grenade http://logs.openstack.org/95/658095/1/check/neutron-grenade/2a604ab/ : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":1},{"id":"08d96142d3801f407bb18b3f1874a8b67b5eeff6","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-10 11:17:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a6c273210d23b5f8f64c19f3d5c3ea159f003c96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 13:12:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/658095/1/check/openstack-tox-docs/22c3d3c/html/ : SUCCESS in 3m 38s\n- openstack-tox-pypy http://logs.openstack.org/95/658095/1/check/openstack-tox-pypy/0ee68b5/ : FAILURE in 3m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/95/658095/1/check/openstack-tox-pep8/2ed3dfd/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/95/658095/1/check/openstack-tox-py27/6d6a7c9/ : FAILURE in 2m 55s\n- openstack-tox-py36 http://logs.openstack.org/95/658095/1/check/openstack-tox-py36/9af754c/ : SUCCESS in 3m 33s\n- openstack-tox-validate http://logs.openstack.org/95/658095/1/check/openstack-tox-validate/9f1f504/ : SUCCESS in 2m 57s\n- requirements-tox-babel http://logs.openstack.org/95/658095/1/check/requirements-tox-babel/76753a0/ : SUCCESS in 2m 55s\n- requirements-tox-py27-check-uc http://logs.openstack.org/95/658095/1/check/requirements-tox-py27-check-uc/690c2b9/ : SUCCESS in 5m 48s\n- requirements-tox-py36-check-uc http://logs.openstack.org/95/658095/1/check/requirements-tox-py36-check-uc/8bab1f5/ : SUCCESS in 8m 39s\n- cross-cinder-py27 http://logs.openstack.org/95/658095/1/check/cross-cinder-py27/84b7630/ : SUCCESS in 10m 49s\n- cross-cinder-py36 http://logs.openstack.org/95/658095/1/check/cross-cinder-py36/8047a84/ : SUCCESS in 8m 02s\n- cross-glance-py27 http://logs.openstack.org/95/658095/1/check/cross-glance-py27/8387cd6/ : SUCCESS in 4m 08s\n- cross-glance-py36 http://logs.openstack.org/95/658095/1/check/cross-glance-py36/adf3ffa/ : SUCCESS in 4m 02s\n- cross-horizon-py27 http://logs.openstack.org/95/658095/1/check/cross-horizon-py27/373994a/ : SUCCESS in 9m 21s\n- cross-horizon-py36 http://logs.openstack.org/95/658095/1/check/cross-horizon-py36/898d597/ : SUCCESS in 7m 10s\n- cross-horizon-npm http://logs.openstack.org/95/658095/1/check/cross-horizon-npm/51a2cc1/ : SUCCESS in 6m 50s\n- cross-keystone-py27 http://logs.openstack.org/95/658095/1/check/cross-keystone-py27/a6f416b/ : SUCCESS in 19m 36s\n- cross-keystone-py36 http://logs.openstack.org/95/658095/1/check/cross-keystone-py36/22b1701/ : SUCCESS in 23m 40s\n- cross-neutron-py27 http://logs.openstack.org/95/658095/1/check/cross-neutron-py27/9770280/ : SUCCESS in 22m 10s\n- cross-neutron-py36 http://logs.openstack.org/95/658095/1/check/cross-neutron-py36/c19073f/ : SUCCESS in 23m 00s\n- cross-nova-py27 http://logs.openstack.org/95/658095/1/check/cross-nova-py27/8247969/ : SUCCESS in 11m 35s\n- cross-nova-py36 http://logs.openstack.org/95/658095/1/check/cross-nova-py36/3f99057/ : SUCCESS in 11m 58s\n- cross-nova-functional http://logs.openstack.org/95/658095/1/check/cross-nova-functional/ef42253/ : SUCCESS in 19m 08s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/95/658095/1/check/cross-osc-build-sphinx-docs/d9f6609/html/ : SUCCESS in 5m 37s\n- cross-swift-py27 http://logs.openstack.org/95/658095/1/check/cross-swift-py27/cb816f9/ : SUCCESS in 7m 33s\n- cross-swift-py36 http://logs.openstack.org/95/658095/1/check/cross-swift-py36/3940f09/ : SUCCESS in 6m 58s\n- requirements-integration http://logs.openstack.org/95/658095/1/check/requirements-integration/b2e8cd2/ : FAILURE in 31m 35s\n- neutron-grenade-multinode http://logs.openstack.org/95/658095/1/check/neutron-grenade-multinode/08e508d/ : SUCCESS in 1h 23m 14s\n- tempest-full http://logs.openstack.org/95/658095/1/check/tempest-full/104b6d1/ : SUCCESS in 1h 47m 55s\n- neutron-grenade http://logs.openstack.org/95/658095/1/check/neutron-grenade/274e3bf/ : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"42080e2ef7282ba7a1b078ab5017d82930356c62","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-11 19:36:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2516f5f3e93f48e8e4b229dc93bb96fa8f721d0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 21:15:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/658095/1/check/openstack-tox-docs/b392ce8/html/ : SUCCESS in 3m 31s\n- openstack-tox-pypy http://logs.openstack.org/95/658095/1/check/openstack-tox-pypy/ae6ffac/ : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/95/658095/1/check/openstack-tox-pep8/9fcd33b/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/95/658095/1/check/openstack-tox-py27/e9c572f/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/95/658095/1/check/openstack-tox-py36/7cbec03/ : SUCCESS in 3m 03s\n- openstack-tox-validate http://logs.openstack.org/95/658095/1/check/openstack-tox-validate/29f6b28/ : SUCCESS in 2m 50s\n- requirements-tox-babel http://logs.openstack.org/95/658095/1/check/requirements-tox-babel/3e6bd18/ : SUCCESS in 3m 32s\n- requirements-tox-py27-check-uc http://logs.openstack.org/95/658095/1/check/requirements-tox-py27-check-uc/a5057ef/ : SUCCESS in 6m 31s\n- requirements-tox-py36-check-uc http://logs.openstack.org/95/658095/1/check/requirements-tox-py36-check-uc/6f569c7/ : SUCCESS in 6m 30s\n- cross-cinder-py27 http://logs.openstack.org/95/658095/1/check/cross-cinder-py27/2c18319/ : SUCCESS in 8m 29s\n- cross-cinder-py36 http://logs.openstack.org/95/658095/1/check/cross-cinder-py36/35840f6/ : SUCCESS in 9m 09s\n- cross-glance-py27 http://logs.openstack.org/95/658095/1/check/cross-glance-py27/f0e55c2/ : SUCCESS in 4m 03s\n- cross-glance-py36 http://logs.openstack.org/95/658095/1/check/cross-glance-py36/4105842/ : SUCCESS in 4m 04s\n- cross-horizon-py27 http://logs.openstack.org/95/658095/1/check/cross-horizon-py27/96269a5/ : SUCCESS in 10m 24s\n- cross-horizon-py36 http://logs.openstack.org/95/658095/1/check/cross-horizon-py36/77bb9da/ : SUCCESS in 9m 36s\n- cross-horizon-npm http://logs.openstack.org/95/658095/1/check/cross-horizon-npm/5e017e2/ : SUCCESS in 5m 34s\n- cross-keystone-py27 http://logs.openstack.org/95/658095/1/check/cross-keystone-py27/5bb2fb5/ : SUCCESS in 21m 10s\n- cross-keystone-py36 http://logs.openstack.org/95/658095/1/check/cross-keystone-py36/6898981/ : SUCCESS in 29m 06s\n- cross-neutron-py27 http://logs.openstack.org/95/658095/1/check/cross-neutron-py27/27c2777/ : SUCCESS in 21m 39s\n- cross-neutron-py36 http://logs.openstack.org/95/658095/1/check/cross-neutron-py36/a3a9195/ : SUCCESS in 23m 42s\n- cross-nova-py27 http://logs.openstack.org/95/658095/1/check/cross-nova-py27/498900d/ : SUCCESS in 14m 09s\n- cross-nova-py36 http://logs.openstack.org/95/658095/1/check/cross-nova-py36/86c61af/ : SUCCESS in 12m 48s\n- cross-nova-functional http://logs.openstack.org/95/658095/1/check/cross-nova-functional/d324a4a/ : SUCCESS in 18m 41s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/95/658095/1/check/cross-osc-build-sphinx-docs/896e482/html/ : SUCCESS in 5m 13s\n- cross-swift-py27 http://logs.openstack.org/95/658095/1/check/cross-swift-py27/fd84bba/ : SUCCESS in 7m 09s\n- cross-swift-py36 http://logs.openstack.org/95/658095/1/check/cross-swift-py36/859115b/ : SUCCESS in 7m 19s\n- requirements-integration http://logs.openstack.org/95/658095/1/check/requirements-integration/e0a6c27/ : SUCCESS in 52m 37s\n- neutron-grenade-multinode http://logs.openstack.org/95/658095/1/check/neutron-grenade-multinode/061984f/ : SUCCESS in 1h 11m 17s\n- tempest-full http://logs.openstack.org/95/658095/1/check/tempest-full/0e67306/ : SUCCESS in 1h 34m 26s\n- neutron-grenade http://logs.openstack.org/95/658095/1/check/neutron-grenade/2b60991/ : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c327f6e9ca80df76d4bc7218479e13c44b4b1cce","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-11 21:20:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ef0ce35e7d6c43b7ee763d540fb10826a33b7b6a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 13:12:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"31f105bc231414b21d0b156d9bf856679457ad97","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-12 20:06:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"11095069d6c138972f1bbfdc3c2eda7fcd7371a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 21:49:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/658095/2/check/openstack-tox-docs/5556f9c/html/ : SUCCESS in 2m 43s\n- openstack-tox-pypy http://logs.openstack.org/95/658095/2/check/openstack-tox-pypy/757bc6e/ : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/95/658095/2/check/openstack-tox-pep8/ee8b2b1/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/95/658095/2/check/openstack-tox-py27/529762d/ : SUCCESS in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/95/658095/2/check/openstack-tox-py36/9341bac/ : SUCCESS in 4m 27s\n- openstack-tox-validate http://logs.openstack.org/95/658095/2/check/openstack-tox-validate/fe894c6/ : SUCCESS in 3m 45s\n- requirements-tox-babel http://logs.openstack.org/95/658095/2/check/requirements-tox-babel/64b5db6/ : SUCCESS in 2m 36s\n- requirements-tox-py27-check-uc http://logs.openstack.org/95/658095/2/check/requirements-tox-py27-check-uc/41294c5/ : SUCCESS in 5m 24s\n- requirements-tox-py36-check-uc http://logs.openstack.org/95/658095/2/check/requirements-tox-py36-check-uc/2878cf6/ : SUCCESS in 5m 41s\n- cross-cinder-py27 http://logs.openstack.org/95/658095/2/check/cross-cinder-py27/49e8aa7/ : SUCCESS in 8m 49s\n- cross-cinder-py36 http://logs.openstack.org/95/658095/2/check/cross-cinder-py36/8a1f524/ : SUCCESS in 8m 07s\n- cross-glance-py27 http://logs.openstack.org/95/658095/2/check/cross-glance-py27/31fa5b8/ : SUCCESS in 3m 59s\n- cross-glance-py36 http://logs.openstack.org/95/658095/2/check/cross-glance-py36/1bfe2a7/ : SUCCESS in 3m 56s\n- cross-horizon-py27 http://logs.openstack.org/95/658095/2/check/cross-horizon-py27/cffd36d/ : SUCCESS in 8m 33s\n- cross-horizon-py36 http://logs.openstack.org/95/658095/2/check/cross-horizon-py36/c138702/ : SUCCESS in 7m 21s\n- cross-horizon-npm http://logs.openstack.org/95/658095/2/check/cross-horizon-npm/9589736/ : SUCCESS in 5m 27s\n- cross-keystone-py27 http://logs.openstack.org/95/658095/2/check/cross-keystone-py27/0858c78/ : SUCCESS in 23m 46s\n- cross-keystone-py36 http://logs.openstack.org/95/658095/2/check/cross-keystone-py36/3caf160/ : SUCCESS in 20m 32s\n- cross-neutron-py27 http://logs.openstack.org/95/658095/2/check/cross-neutron-py27/78bd50e/ : SUCCESS in 20m 32s\n- cross-neutron-py36 http://logs.openstack.org/95/658095/2/check/cross-neutron-py36/810ad85/ : SUCCESS in 18m 43s\n- cross-nova-py27 http://logs.openstack.org/95/658095/2/check/cross-nova-py27/15efe31/ : SUCCESS in 11m 27s\n- cross-nova-py36 http://logs.openstack.org/95/658095/2/check/cross-nova-py36/11c58fa/ : SUCCESS in 10m 58s\n- cross-nova-functional http://logs.openstack.org/95/658095/2/check/cross-nova-functional/09ba8d9/ : SUCCESS in 20m 30s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/95/658095/2/check/cross-osc-build-sphinx-docs/5ff27f8/html/ : SUCCESS in 5m 31s\n- cross-swift-py27 http://logs.openstack.org/95/658095/2/check/cross-swift-py27/a9bf3ca/ : SUCCESS in 9m 31s\n- cross-swift-py36 http://logs.openstack.org/95/658095/2/check/cross-swift-py36/ae513e6/ : SUCCESS in 7m 21s\n- requirements-integration http://logs.openstack.org/95/658095/2/check/requirements-integration/fefdbd7/ : SUCCESS in 52m 28s\n- neutron-grenade-multinode http://logs.openstack.org/95/658095/2/check/neutron-grenade-multinode/47398cc/ : SUCCESS in 58m 24s\n- tempest-full http://logs.openstack.org/95/658095/2/check/tempest-full/f6e513e/ : SUCCESS in 1h 37m 53s\n- neutron-grenade http://logs.openstack.org/95/658095/2/check/neutron-grenade/5622a9d/ : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":2},{"id":"c810c1cd25e51b474fb2140d41202708d284de50","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-12 21:53:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"35d205090a482b78315774d833cc013b996342e3","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-05-13 14:32:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5620b8d61a37a8414a6f4668607238d4a4c520f9","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-14 16:26:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"603c974ab61524ad305db82fdd109138746d36e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 18:25:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/658095/3/check/openstack-tox-docs/b98e7e8/html/ : SUCCESS in 3m 23s\n- openstack-tox-pypy http://logs.openstack.org/95/658095/3/check/openstack-tox-pypy/16b1c25/ : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/95/658095/3/check/openstack-tox-pep8/720180f/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/95/658095/3/check/openstack-tox-py27/2e45995/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/95/658095/3/check/openstack-tox-py36/5b7580f/ : SUCCESS in 3m 32s\n- openstack-tox-validate http://logs.openstack.org/95/658095/3/check/openstack-tox-validate/a6bf815/ : SUCCESS in 3m 55s\n- requirements-tox-babel http://logs.openstack.org/95/658095/3/check/requirements-tox-babel/040fca2/ : SUCCESS in 3m 13s\n- requirements-tox-py27-check-uc http://logs.openstack.org/95/658095/3/check/requirements-tox-py27-check-uc/214cd20/ : SUCCESS in 11m 39s\n- requirements-tox-py36-check-uc http://logs.openstack.org/95/658095/3/check/requirements-tox-py36-check-uc/138a8d4/ : SUCCESS in 12m 53s\n- cross-cinder-py27 http://logs.openstack.org/95/658095/3/check/cross-cinder-py27/c62f739/ : SUCCESS in 8m 18s\n- cross-cinder-py36 http://logs.openstack.org/95/658095/3/check/cross-cinder-py36/4c1387b/ : SUCCESS in 7m 57s\n- cross-glance-py27 http://logs.openstack.org/95/658095/3/check/cross-glance-py27/82dbe6b/ : SUCCESS in 4m 32s\n- cross-glance-py36 http://logs.openstack.org/95/658095/3/check/cross-glance-py36/3c5c52e/ : SUCCESS in 4m 00s\n- cross-horizon-py27 http://logs.openstack.org/95/658095/3/check/cross-horizon-py27/bf59b0b/ : SUCCESS in 10m 18s\n- cross-horizon-py36 http://logs.openstack.org/95/658095/3/check/cross-horizon-py36/ad2acd1/ : SUCCESS in 9m 37s\n- cross-horizon-npm http://logs.openstack.org/95/658095/3/check/cross-horizon-npm/691e5cd/ : SUCCESS in 5m 45s\n- cross-keystone-py27 http://logs.openstack.org/95/658095/3/check/cross-keystone-py27/8792dae/ : SUCCESS in 30m 21s\n- cross-keystone-py36 http://logs.openstack.org/95/658095/3/check/cross-keystone-py36/1a46177/ : SUCCESS in 33m 04s\n- cross-neutron-py27 http://logs.openstack.org/95/658095/3/check/cross-neutron-py27/328488b/ : SUCCESS in 21m 38s\n- cross-neutron-py36 http://logs.openstack.org/95/658095/3/check/cross-neutron-py36/b1b2f86/ : SUCCESS in 25m 40s\n- cross-nova-py27 http://logs.openstack.org/95/658095/3/check/cross-nova-py27/65a2923/ : SUCCESS in 11m 57s\n- cross-nova-py36 http://logs.openstack.org/95/658095/3/check/cross-nova-py36/1207f10/ : SUCCESS in 11m 18s\n- cross-nova-functional http://logs.openstack.org/95/658095/3/check/cross-nova-functional/ae30ae8/ : SUCCESS in 23m 12s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/95/658095/3/check/cross-osc-build-sphinx-docs/5d4328e/html/ : SUCCESS in 5m 49s\n- cross-swift-py27 http://logs.openstack.org/95/658095/3/check/cross-swift-py27/448fe54/ : SUCCESS in 9m 15s\n- cross-swift-py36 http://logs.openstack.org/95/658095/3/check/cross-swift-py36/ba68663/ : SUCCESS in 9m 46s\n- requirements-integration http://logs.openstack.org/95/658095/3/check/requirements-integration/5e3201b/ : SUCCESS in 1h 26m 11s\n- neutron-grenade-multinode http://logs.openstack.org/95/658095/3/check/neutron-grenade-multinode/de400b7/ : SUCCESS in 1h 17m 00s\n- tempest-full http://logs.openstack.org/95/658095/3/check/tempest-full/4903ab1/ : SUCCESS in 1h 44m 32s\n- neutron-grenade http://logs.openstack.org/95/658095/3/check/neutron-grenade/ce588d1/ : SUCCESS in 48m 02s","accounts_in_message":[],"_revision_number":3},{"id":"a209a060b912b846ef015bb2e3fa1fa4837e6c6b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-14 18:28:12.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8a1f1eec6e27cadcbdb58826746311d97e27d745","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-14 21:08:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThis code has frequent updates on github and regular releases. also seems to explicitly declare python 2.x and 3.x compatibility and looks okaish.","accounts_in_message":[],"_revision_number":3},{"id":"2096da1b8a9af8afe6b2061b2ad51d52f7ee028d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 21:08:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2f520e998bab9b77e5d1755ccc7ebdb152c728ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 23:24:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/658095/3/gate/openstack-tox-docs/b83ab08/html/ : SUCCESS in 3m 00s\n- openstack-tox-pep8 http://logs.openstack.org/95/658095/3/gate/openstack-tox-pep8/d9a409e/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/95/658095/3/gate/openstack-tox-py27/97ffc2e/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/95/658095/3/gate/openstack-tox-py36/2f23e4d/ : SUCCESS in 3m 44s\n- openstack-tox-validate http://logs.openstack.org/95/658095/3/gate/openstack-tox-validate/10d2cef/ : SUCCESS in 3m 05s\n- requirements-tox-babel http://logs.openstack.org/95/658095/3/gate/requirements-tox-babel/21ed3a7/ : SUCCESS in 3m 02s\n- requirements-tox-py27-check-uc http://logs.openstack.org/95/658095/3/gate/requirements-tox-py27-check-uc/01a5484/ : SUCCESS in 6m 35s\n- requirements-tox-py36-check-uc http://logs.openstack.org/95/658095/3/gate/requirements-tox-py36-check-uc/1288a46/ : SUCCESS in 5m 35s\n- cross-cinder-py27 http://logs.openstack.org/95/658095/3/gate/cross-cinder-py27/b33e69b/ : SUCCESS in 8m 33s\n- cross-cinder-py36 http://logs.openstack.org/95/658095/3/gate/cross-cinder-py36/eaed484/ : SUCCESS in 7m 05s\n- cross-glance-py27 http://logs.openstack.org/95/658095/3/gate/cross-glance-py27/1ccc410/ : SUCCESS in 3m 59s\n- cross-glance-py36 http://logs.openstack.org/95/658095/3/gate/cross-glance-py36/cc88e63/ : SUCCESS in 4m 07s\n- cross-horizon-py27 http://logs.openstack.org/95/658095/3/gate/cross-horizon-py27/3e1cde5/ : SUCCESS in 10m 51s\n- cross-horizon-py36 http://logs.openstack.org/95/658095/3/gate/cross-horizon-py36/f5bc4e7/ : SUCCESS in 8m 09s\n- cross-horizon-npm http://logs.openstack.org/95/658095/3/gate/cross-horizon-npm/8e1f1c4/ : SUCCESS in 6m 11s\n- cross-keystone-py27 http://logs.openstack.org/95/658095/3/gate/cross-keystone-py27/82d63f3/ : SUCCESS in 20m 00s\n- cross-keystone-py36 http://logs.openstack.org/95/658095/3/gate/cross-keystone-py36/1d882be/ : SUCCESS in 26m 02s\n- cross-neutron-py27 http://logs.openstack.org/95/658095/3/gate/cross-neutron-py27/c96077c/ : SUCCESS in 24m 46s\n- cross-neutron-py36 http://logs.openstack.org/95/658095/3/gate/cross-neutron-py36/f713773/ : SUCCESS in 24m 38s\n- cross-nova-py27 http://logs.openstack.org/95/658095/3/gate/cross-nova-py27/23532bf/ : SUCCESS in 13m 09s\n- cross-nova-py36 http://logs.openstack.org/95/658095/3/gate/cross-nova-py36/ff64cee/ : SUCCESS in 11m 41s\n- cross-nova-functional http://logs.openstack.org/95/658095/3/gate/cross-nova-functional/4a97cd4/ : SUCCESS in 18m 34s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/95/658095/3/gate/cross-osc-build-sphinx-docs/9040fae/html/ : SUCCESS in 5m 54s\n- cross-swift-py27 http://logs.openstack.org/95/658095/3/gate/cross-swift-py27/f1be404/ : SUCCESS in 7m 52s\n- cross-swift-py36 http://logs.openstack.org/95/658095/3/gate/cross-swift-py36/3afd01a/ : SUCCESS in 7m 39s\n- requirements-integration http://logs.openstack.org/95/658095/3/gate/requirements-integration/b4b8dfc/ : SUCCESS in 54m 31s\n- neutron-grenade-multinode http://logs.openstack.org/95/658095/3/gate/neutron-grenade-multinode/263947c/ : SUCCESS in 1h 11m 46s\n- tempest-full http://logs.openstack.org/95/658095/3/gate/tempest-full/cecdf7d/ : SUCCESS in 1h 47m 41s\n- neutron-grenade http://logs.openstack.org/95/658095/3/gate/neutron-grenade/6bf51ec/ : SUCCESS in 54m 14s","accounts_in_message":[],"_revision_number":3},{"id":"88020a0ea0f52803e8a64763375078d529835fab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 23:24:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"522c48a7e434231132d1793d581bba9bb483eeea","revisions":{"b99b877ba0aea9cf56d05626f2e117ce95a3d2b1":{"kind":"REWORK","_number":1,"created":"2019-05-09 14:11:01.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/95/658095/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/658095/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/658095/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/658095/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/658095/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/658095/1"}}},"commit":{"parents":[{"commit":"842ed7bbb06c09dc01785265b589c228b5f928af","subject":"Merge \"Add missing format arg for missing g-r reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/842ed7bbb06c09dc01785265b589c228b5f928af"}]}],"author":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-09 14:09:44.000000000","tz":0},"committer":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-09 14:09:44.000000000","tz":0},"subject":"Add cinder extras storops library","message":"Add cinder extras storops library\n\nThis patch adds the cinder extras storops library needed by the\nDell EMC VNX driver\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: eh?\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nChange-Id: I227bf514d15e301224a01252c5f5f784f040ad34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b99b877ba0aea9cf56d05626f2e117ce95a3d2b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b99b877ba0aea9cf56d05626f2e117ce95a3d2b1"}]},"branch":"refs/heads/master"},"f14a3da4cffda9d424a311dfc08d394bcbb1038e":{"kind":"REWORK","_number":2,"created":"2019-05-12 20:06:12.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/95/658095/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/658095/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/658095/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/658095/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/658095/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/658095/2"}}},"commit":{"parents":[{"commit":"61b95f944d701f5eff02bf8aea4d6b3e767e5822","subject":"Merge \"Fix docs environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61b95f944d701f5eff02bf8aea4d6b3e767e5822"}]}],"author":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-09 14:09:44.000000000","tz":0},"committer":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-12 20:06:06.000000000","tz":-420},"subject":"Add cinder extras storops library","message":"Add cinder extras storops library\n\nThis patch adds the cinder extras storops library needed by the\nDell EMC VNX driver\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: eh?\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nChange-Id: I227bf514d15e301224a01252c5f5f784f040ad34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f14a3da4cffda9d424a311dfc08d394bcbb1038e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f14a3da4cffda9d424a311dfc08d394bcbb1038e"}]},"branch":"refs/heads/master"},"522c48a7e434231132d1793d581bba9bb483eeea":{"kind":"REWORK","_number":3,"created":"2019-05-14 16:26:54.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/95/658095/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/658095/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/658095/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/658095/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/658095/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/658095/3"}}},"commit":{"parents":[{"commit":"85f9aad798323c5f6dd8ce1a3dd5e009c2f944a1","subject":"Merge \"Set upper-constraint for scikit-learn\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85f9aad798323c5f6dd8ce1a3dd5e009c2f944a1"}]}],"author":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-09 14:09:44.000000000","tz":0},"committer":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-14 16:27:09.000000000","tz":0},"subject":"Add cinder extras storops library","message":"Add cinder extras storops library\n\nThis patch adds the cinder extras storops library needed by the\nDell EMC VNX driver\n\nactive: yes\ngood code: seems good\npy3: yes\nlicense compat: yes\npackaged: eh?\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nChange-Id: I227bf514d15e301224a01252c5f5f784f040ad34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/522c48a7e434231132d1793d581bba9bb483eeea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/522c48a7e434231132d1793d581bba9bb483eeea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
