)]}'
{"id":"openstack%2Frequirements~658096","triplet_id":"openstack%2Frequirements~master~I65de3e178d573d0275fe1d944ce1231e2ef6c935","project":"openstack/requirements","branch":"master","topic":"cinder_extras","hashtags":[],"change_id":"I65de3e178d573d0275fe1d944ce1231e2ef6c935","subject":"Add cinder extras purestorage library","status":"MERGED","created":"2019-05-09 14:13:20.000000000","updated":"2019-05-20 20:07:23.000000000","submitted":"2019-05-20 20:07:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658096-1558382843269-d1a6a7d1","meta_rev_id":"c77a3dd94a6a3cd2dedea775d8dea370c89469c5","_number":658096,"virtual_id_number":658096,"owner":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-20 20:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-17 02:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-17 00:11:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2019-05-20 16:16:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2019-05-20 16:16:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-11 19:35:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-17 00:11:53.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2019-05-17 02:10:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-20 16:16:48.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-20 20:07:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf98cbf3bf7d6ebfef0f812b6684aa4fb1c74d36","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-09 14:13:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4fe8baf22bede7360e2633eb83264c81a0492f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 18:13:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/96/658096/1/check/openstack-tox-docs/77f1b7e/html/ : SUCCESS in 3m 07s\n- openstack-tox-pypy http://logs.openstack.org/96/658096/1/check/openstack-tox-pypy/1e17051/ : FAILURE in 3m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/96/658096/1/check/openstack-tox-pep8/b7e92e5/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/96/658096/1/check/openstack-tox-py27/242f8e7/ : FAILURE in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/96/658096/1/check/openstack-tox-py36/c3e208f/ : SUCCESS in 4m 28s\n- openstack-tox-validate http://logs.openstack.org/96/658096/1/check/openstack-tox-validate/d3cadc8/ : SUCCESS in 3m 30s\n- requirements-tox-babel http://logs.openstack.org/96/658096/1/check/requirements-tox-babel/ec26473/ : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc http://logs.openstack.org/96/658096/1/check/requirements-tox-py27-check-uc/f9619f0/ : SUCCESS in 6m 41s\n- requirements-tox-py36-check-uc http://logs.openstack.org/96/658096/1/check/requirements-tox-py36-check-uc/d94ad1f/ : SUCCESS in 6m 40s\n- cross-cinder-py27 http://logs.openstack.org/96/658096/1/check/cross-cinder-py27/b36b5e5/ : SUCCESS in 9m 07s\n- cross-cinder-py36 http://logs.openstack.org/96/658096/1/check/cross-cinder-py36/9fd0c42/ : SUCCESS in 9m 40s\n- cross-glance-py27 http://logs.openstack.org/96/658096/1/check/cross-glance-py27/39ac0ca/ : SUCCESS in 5m 59s\n- cross-glance-py36 http://logs.openstack.org/96/658096/1/check/cross-glance-py36/223b77f/ : SUCCESS in 3m 51s\n- cross-horizon-py27 http://logs.openstack.org/96/658096/1/check/cross-horizon-py27/1050766/ : SUCCESS in 9m 56s\n- cross-horizon-py36 http://logs.openstack.org/96/658096/1/check/cross-horizon-py36/409ca50/ : SUCCESS in 10m 12s\n- cross-horizon-npm http://logs.openstack.org/96/658096/1/check/cross-horizon-npm/22d694d/ : SUCCESS in 5m 25s\n- cross-keystone-py27 http://logs.openstack.org/96/658096/1/check/cross-keystone-py27/6f7221b/ : SUCCESS in 26m 38s\n- cross-keystone-py36 http://logs.openstack.org/96/658096/1/check/cross-keystone-py36/45f3b06/ : SUCCESS in 22m 24s\n- cross-neutron-py27 http://logs.openstack.org/96/658096/1/check/cross-neutron-py27/27354b6/ : SUCCESS in 24m 23s\n- cross-neutron-py36 http://logs.openstack.org/96/658096/1/check/cross-neutron-py36/06f4391/ : SUCCESS in 33m 18s\n- cross-nova-py27 http://logs.openstack.org/96/658096/1/check/cross-nova-py27/8f9a429/ : SUCCESS in 13m 05s\n- cross-nova-py36 http://logs.openstack.org/96/658096/1/check/cross-nova-py36/356ef33/ : SUCCESS in 28m 07s\n- cross-nova-functional http://logs.openstack.org/96/658096/1/check/cross-nova-functional/e5f18f9/ : SUCCESS in 18m 21s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/96/658096/1/check/cross-osc-build-sphinx-docs/19da1ed/html/ : SUCCESS in 4m 56s\n- cross-swift-py27 http://logs.openstack.org/96/658096/1/check/cross-swift-py27/ca94b58/ : SUCCESS in 6m 48s\n- cross-swift-py36 http://logs.openstack.org/96/658096/1/check/cross-swift-py36/3f0bd99/ : SUCCESS in 10m 35s\n- requirements-integration http://logs.openstack.org/96/658096/1/check/requirements-integration/31eb61d/ : FAILURE in 31m 08s\n- neutron-grenade-multinode http://logs.openstack.org/96/658096/1/check/neutron-grenade-multinode/ff1bf2d/ : SUCCESS in 1h 07m 33s\n- tempest-full http://logs.openstack.org/96/658096/1/check/tempest-full/63cc145/ : SUCCESS in 1h 45m 56s\n- neutron-grenade http://logs.openstack.org/96/658096/1/check/neutron-grenade/871ce58/ : SUCCESS in 1h 05m 18s","accounts_in_message":[],"_revision_number":1},{"id":"380e6fd4cf7f5a9d41392008cd5e2f564db1421f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 05:09:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8300da3fa61cf33adea4792d9bc88f05887c2b20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 06:41:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/96/658096/1/check/openstack-tox-docs/bdc8a0b/html/ : SUCCESS in 4m 55s\n- openstack-tox-pypy http://logs.openstack.org/96/658096/1/check/openstack-tox-pypy/b6d5ad5/ : FAILURE in 4m 45s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/96/658096/1/check/openstack-tox-pep8/d085acf/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/96/658096/1/check/openstack-tox-py27/7a983ae/ : FAILURE in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/96/658096/1/check/openstack-tox-py36/ad73f85/ : SUCCESS in 5m 02s\n- openstack-tox-validate http://logs.openstack.org/96/658096/1/check/openstack-tox-validate/cc45c6d/ : SUCCESS in 4m 44s\n- requirements-tox-babel http://logs.openstack.org/96/658096/1/check/requirements-tox-babel/81e7774/ : SUCCESS in 4m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/96/658096/1/check/requirements-tox-py27-check-uc/5cd3bc2/ : SUCCESS in 7m 08s\n- requirements-tox-py36-check-uc http://logs.openstack.org/96/658096/1/check/requirements-tox-py36-check-uc/f5dd2a9/ : SUCCESS in 7m 21s\n- cross-cinder-py27 http://logs.openstack.org/96/658096/1/check/cross-cinder-py27/e20446d/ : SUCCESS in 9m 01s\n- cross-cinder-py36 http://logs.openstack.org/96/658096/1/check/cross-cinder-py36/0620d5d/ : SUCCESS in 8m 54s\n- cross-glance-py27 http://logs.openstack.org/96/658096/1/check/cross-glance-py27/1320257/ : SUCCESS in 6m 35s\n- cross-glance-py36 http://logs.openstack.org/96/658096/1/check/cross-glance-py36/248ffc7/ : SUCCESS in 5m 18s\n- cross-horizon-py27 http://logs.openstack.org/96/658096/1/check/cross-horizon-py27/86f627d/ : SUCCESS in 12m 32s\n- cross-horizon-py36 http://logs.openstack.org/96/658096/1/check/cross-horizon-py36/6848702/ : SUCCESS in 10m 11s\n- cross-horizon-npm http://logs.openstack.org/96/658096/1/check/cross-horizon-npm/4e0b31e/ : SUCCESS in 5m 56s\n- cross-keystone-py27 http://logs.openstack.org/96/658096/1/check/cross-keystone-py27/690697e/ : SUCCESS in 24m 40s\n- cross-keystone-py36 http://logs.openstack.org/96/658096/1/check/cross-keystone-py36/9fc46d7/ : SUCCESS in 23m 12s\n- cross-neutron-py27 http://logs.openstack.org/96/658096/1/check/cross-neutron-py27/397eb65/ : SUCCESS in 25m 06s\n- cross-neutron-py36 http://logs.openstack.org/96/658096/1/check/cross-neutron-py36/2527287/ : SUCCESS in 20m 29s\n- cross-nova-py27 http://logs.openstack.org/96/658096/1/check/cross-nova-py27/c3aafa8/ : SUCCESS in 12m 52s\n- cross-nova-py36 http://logs.openstack.org/96/658096/1/check/cross-nova-py36/2bc8954/ : SUCCESS in 12m 09s\n- cross-nova-functional http://logs.openstack.org/96/658096/1/check/cross-nova-functional/130c795/ : SUCCESS in 19m 42s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/96/658096/1/check/cross-osc-build-sphinx-docs/31c01d4/html/ : SUCCESS in 6m 42s\n- cross-swift-py27 http://logs.openstack.org/96/658096/1/check/cross-swift-py27/b899c29/ : SUCCESS in 9m 31s\n- cross-swift-py36 http://logs.openstack.org/96/658096/1/check/cross-swift-py36/bfdc68a/ : SUCCESS in 9m 02s\n- requirements-integration http://logs.openstack.org/96/658096/1/check/requirements-integration/e080bb7/ : FAILURE in 34m 54s\n- neutron-grenade-multinode http://logs.openstack.org/96/658096/1/check/neutron-grenade-multinode/36fccbc/ : SUCCESS in 1h 06m 35s\n- tempest-full http://logs.openstack.org/96/658096/1/check/tempest-full/decb869/ : FAILURE in 1h 24m 57s\n- neutron-grenade http://logs.openstack.org/96/658096/1/check/neutron-grenade/ea37895/ : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7e4a7df36bcf4c791b115b18736a25c50aaf6e35","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-11 19:35:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"867c3e330eb6ff738ea6434ed90649849f50fb95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 21:26:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/96/658096/1/check/openstack-tox-docs/433f951/html/ : SUCCESS in 3m 32s\n- openstack-tox-pypy http://logs.openstack.org/96/658096/1/check/openstack-tox-pypy/9f7b1ba/ : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/96/658096/1/check/openstack-tox-pep8/18f6e34/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/96/658096/1/check/openstack-tox-py27/907c57a/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/96/658096/1/check/openstack-tox-py36/9c9002b/ : SUCCESS in 4m 19s\n- openstack-tox-validate http://logs.openstack.org/96/658096/1/check/openstack-tox-validate/55d023c/ : SUCCESS in 3m 35s\n- requirements-tox-babel http://logs.openstack.org/96/658096/1/check/requirements-tox-babel/bd2afdd/ : SUCCESS in 3m 04s\n- requirements-tox-py27-check-uc http://logs.openstack.org/96/658096/1/check/requirements-tox-py27-check-uc/971c400/ : SUCCESS in 8m 06s\n- requirements-tox-py36-check-uc http://logs.openstack.org/96/658096/1/check/requirements-tox-py36-check-uc/2531313/ : SUCCESS in 5m 59s\n- cross-cinder-py27 http://logs.openstack.org/96/658096/1/check/cross-cinder-py27/9f6767e/ : SUCCESS in 10m 12s\n- cross-cinder-py36 http://logs.openstack.org/96/658096/1/check/cross-cinder-py36/f15da08/ : SUCCESS in 8m 59s\n- cross-glance-py27 http://logs.openstack.org/96/658096/1/check/cross-glance-py27/47d0f84/ : SUCCESS in 4m 54s\n- cross-glance-py36 http://logs.openstack.org/96/658096/1/check/cross-glance-py36/575ad15/ : SUCCESS in 4m 53s\n- cross-horizon-py27 http://logs.openstack.org/96/658096/1/check/cross-horizon-py27/0c42b67/ : SUCCESS in 11m 30s\n- cross-horizon-py36 http://logs.openstack.org/96/658096/1/check/cross-horizon-py36/918efd6/ : SUCCESS in 9m 09s\n- cross-horizon-npm http://logs.openstack.org/96/658096/1/check/cross-horizon-npm/11e6c04/ : SUCCESS in 6m 25s\n- cross-keystone-py27 http://logs.openstack.org/96/658096/1/check/cross-keystone-py27/006fdf5/ : SUCCESS in 21m 32s\n- cross-keystone-py36 http://logs.openstack.org/96/658096/1/check/cross-keystone-py36/1253c10/ : SUCCESS in 31m 01s\n- cross-neutron-py27 http://logs.openstack.org/96/658096/1/check/cross-neutron-py27/7e27695/ : SUCCESS in 24m 08s\n- cross-neutron-py36 http://logs.openstack.org/96/658096/1/check/cross-neutron-py36/f03b6ae/ : SUCCESS in 31m 09s\n- cross-nova-py27 http://logs.openstack.org/96/658096/1/check/cross-nova-py27/b462d22/ : SUCCESS in 11m 51s\n- cross-nova-py36 http://logs.openstack.org/96/658096/1/check/cross-nova-py36/f3781f8/ : SUCCESS in 19m 53s\n- cross-nova-functional http://logs.openstack.org/96/658096/1/check/cross-nova-functional/25bae8a/ : SUCCESS in 20m 30s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/96/658096/1/check/cross-osc-build-sphinx-docs/d97a98d/html/ : SUCCESS in 6m 43s\n- cross-swift-py27 http://logs.openstack.org/96/658096/1/check/cross-swift-py27/2f7d682/ : SUCCESS in 8m 07s\n- cross-swift-py36 http://logs.openstack.org/96/658096/1/check/cross-swift-py36/f5427f5/ : SUCCESS in 7m 35s\n- requirements-integration http://logs.openstack.org/96/658096/1/check/requirements-integration/29a7f15/ : SUCCESS in 52m 59s\n- neutron-grenade-multinode http://logs.openstack.org/96/658096/1/check/neutron-grenade-multinode/a611ba5/ : SUCCESS in 1h 13m 50s\n- tempest-full http://logs.openstack.org/96/658096/1/check/tempest-full/012aa71/ : SUCCESS in 1h 45m 07s\n- neutron-grenade http://logs.openstack.org/96/658096/1/check/neutron-grenade/20686bb/ : SUCCESS in 55m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fa6cd6ffd402dca349e475749fab0c91660f0018","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-11 21:28:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a0248fbc61a51487ce13e181c40f5c72aba3b162","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-12 20:07:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"caa3dee711001c85b69bc13c18735cc99d3670d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 21:51:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/96/658096/2/check/openstack-tox-docs/ce5a966/html/ : SUCCESS in 3m 11s\n- openstack-tox-pypy http://logs.openstack.org/96/658096/2/check/openstack-tox-pypy/f265a91/ : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/96/658096/2/check/openstack-tox-pep8/7c12a9b/ : SUCCESS in 2m 31s\n- openstack-tox-py27 http://logs.openstack.org/96/658096/2/check/openstack-tox-py27/5d66dee/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/96/658096/2/check/openstack-tox-py36/38677b7/ : SUCCESS in 3m 17s\n- openstack-tox-validate http://logs.openstack.org/96/658096/2/check/openstack-tox-validate/dbc35ae/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/96/658096/2/check/requirements-tox-babel/c6150f2/ : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/96/658096/2/check/requirements-tox-py27-check-uc/45271c4/ : SUCCESS in 5m 55s\n- requirements-tox-py36-check-uc http://logs.openstack.org/96/658096/2/check/requirements-tox-py36-check-uc/915dee5/ : SUCCESS in 6m 50s\n- cross-cinder-py27 http://logs.openstack.org/96/658096/2/check/cross-cinder-py27/9443bd1/ : SUCCESS in 10m 12s\n- cross-cinder-py36 http://logs.openstack.org/96/658096/2/check/cross-cinder-py36/70ea635/ : SUCCESS in 7m 46s\n- cross-glance-py27 http://logs.openstack.org/96/658096/2/check/cross-glance-py27/f6714ff/ : SUCCESS in 3m 43s\n- cross-glance-py36 http://logs.openstack.org/96/658096/2/check/cross-glance-py36/f686b19/ : SUCCESS in 4m 05s\n- cross-horizon-py27 http://logs.openstack.org/96/658096/2/check/cross-horizon-py27/03ee5f6/ : SUCCESS in 9m 18s\n- cross-horizon-py36 http://logs.openstack.org/96/658096/2/check/cross-horizon-py36/e142880/ : SUCCESS in 9m 21s\n- cross-horizon-npm http://logs.openstack.org/96/658096/2/check/cross-horizon-npm/a322834/ : SUCCESS in 4m 42s\n- cross-keystone-py27 http://logs.openstack.org/96/658096/2/check/cross-keystone-py27/75664d2/ : SUCCESS in 17m 04s\n- cross-keystone-py36 http://logs.openstack.org/96/658096/2/check/cross-keystone-py36/06575d5/ : SUCCESS in 22m 41s\n- cross-neutron-py27 http://logs.openstack.org/96/658096/2/check/cross-neutron-py27/0769a49/ : SUCCESS in 26m 41s\n- cross-neutron-py36 http://logs.openstack.org/96/658096/2/check/cross-neutron-py36/5d4bd68/ : SUCCESS in 19m 46s\n- cross-nova-py27 http://logs.openstack.org/96/658096/2/check/cross-nova-py27/a0ff9ee/ : SUCCESS in 17m 04s\n- cross-nova-py36 http://logs.openstack.org/96/658096/2/check/cross-nova-py36/d397eee/ : SUCCESS in 10m 35s\n- cross-nova-functional http://logs.openstack.org/96/658096/2/check/cross-nova-functional/54c71b8/ : SUCCESS in 19m 47s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/96/658096/2/check/cross-osc-build-sphinx-docs/ca89fb9/html/ : SUCCESS in 4m 55s\n- cross-swift-py27 http://logs.openstack.org/96/658096/2/check/cross-swift-py27/aac76c5/ : SUCCESS in 8m 07s\n- cross-swift-py36 http://logs.openstack.org/96/658096/2/check/cross-swift-py36/a81bcf0/ : SUCCESS in 6m 57s\n- requirements-integration http://logs.openstack.org/96/658096/2/check/requirements-integration/01c8b90/ : SUCCESS in 53m 11s\n- neutron-grenade-multinode http://logs.openstack.org/96/658096/2/check/neutron-grenade-multinode/b8e58cb/ : SUCCESS in 54m 41s\n- tempest-full http://logs.openstack.org/96/658096/2/check/tempest-full/9b5f652/ : SUCCESS in 1h 43m 15s\n- neutron-grenade http://logs.openstack.org/96/658096/2/check/neutron-grenade/f0e13df/ : SUCCESS in 45m 24s","accounts_in_message":[],"_revision_number":2},{"id":"5ee0b79e106b2278a8304ebd69afa9c4237771a0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-12 21:53:39.000000000","message":"Patch Set 2: Code-Review-1\n\nI think we need clarification on the py3 support","accounts_in_message":[],"_revision_number":2},{"id":"fef738152de92799c6d9af62237d969cec0473c6","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2019-05-14 01:04:59.000000000","message":"Patch Set 2: Code-Review+1\n\nGood from a Pure Storage perspective. We are good with py3","accounts_in_message":[],"_revision_number":2},{"id":"3497916f2aa0dc1a6095a8354b165335c34f3b14","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-14 16:27:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"46ff470fc9586880f2f693f5799fd63bd14df58f","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2019-05-14 16:55:01.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6dbf0b1bf0edc9a5edcab0743f5d7733461d37bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 18:17:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/96/658096/3/check/openstack-tox-docs/9cd5aa5/html/ : SUCCESS in 2m 48s\n- openstack-tox-pypy http://logs.openstack.org/96/658096/3/check/openstack-tox-pypy/4f419b7/ : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/96/658096/3/check/openstack-tox-pep8/513bf18/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/96/658096/3/check/openstack-tox-py27/a8d805d/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/96/658096/3/check/openstack-tox-py36/99b0d28/ : SUCCESS in 3m 14s\n- openstack-tox-validate http://logs.openstack.org/96/658096/3/check/openstack-tox-validate/807c1e8/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/96/658096/3/check/requirements-tox-babel/d5923d0/ : SUCCESS in 3m 39s\n- requirements-tox-py27-check-uc http://logs.openstack.org/96/658096/3/check/requirements-tox-py27-check-uc/421c740/ : SUCCESS in 7m 03s\n- requirements-tox-py36-check-uc http://logs.openstack.org/96/658096/3/check/requirements-tox-py36-check-uc/3205d98/ : SUCCESS in 7m 59s\n- cross-cinder-py27 http://logs.openstack.org/96/658096/3/check/cross-cinder-py27/1e55896/ : SUCCESS in 10m 36s\n- cross-cinder-py36 http://logs.openstack.org/96/658096/3/check/cross-cinder-py36/2a261ad/ : SUCCESS in 8m 07s\n- cross-glance-py27 http://logs.openstack.org/96/658096/3/check/cross-glance-py27/76f0066/ : SUCCESS in 5m 03s\n- cross-glance-py36 http://logs.openstack.org/96/658096/3/check/cross-glance-py36/7ffcbc6/ : SUCCESS in 4m 01s\n- cross-horizon-py27 http://logs.openstack.org/96/658096/3/check/cross-horizon-py27/d340e3d/ : SUCCESS in 9m 02s\n- cross-horizon-py36 http://logs.openstack.org/96/658096/3/check/cross-horizon-py36/7261c50/ : SUCCESS in 10m 07s\n- cross-horizon-npm http://logs.openstack.org/96/658096/3/check/cross-horizon-npm/4604fed/ : SUCCESS in 5m 38s\n- cross-keystone-py27 http://logs.openstack.org/96/658096/3/check/cross-keystone-py27/e2a0908/ : SUCCESS in 24m 07s\n- cross-keystone-py36 http://logs.openstack.org/96/658096/3/check/cross-keystone-py36/023c211/ : SUCCESS in 21m 36s\n- cross-neutron-py27 http://logs.openstack.org/96/658096/3/check/cross-neutron-py27/325b491/ : SUCCESS in 21m 28s\n- cross-neutron-py36 http://logs.openstack.org/96/658096/3/check/cross-neutron-py36/c6bf227/ : SUCCESS in 28m 43s\n- cross-nova-py27 http://logs.openstack.org/96/658096/3/check/cross-nova-py27/a7ab47d/ : SUCCESS in 12m 55s\n- cross-nova-py36 http://logs.openstack.org/96/658096/3/check/cross-nova-py36/7662a5f/ : SUCCESS in 14m 48s\n- cross-nova-functional http://logs.openstack.org/96/658096/3/check/cross-nova-functional/bf32259/ : SUCCESS in 18m 50s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/96/658096/3/check/cross-osc-build-sphinx-docs/66790b3/html/ : SUCCESS in 5m 00s\n- cross-swift-py27 http://logs.openstack.org/96/658096/3/check/cross-swift-py27/b6ced60/ : SUCCESS in 7m 00s\n- cross-swift-py36 http://logs.openstack.org/96/658096/3/check/cross-swift-py36/cdecc20/ : SUCCESS in 7m 35s\n- requirements-integration http://logs.openstack.org/96/658096/3/check/requirements-integration/d867d1b/ : SUCCESS in 58m 36s\n- neutron-grenade-multinode http://logs.openstack.org/96/658096/3/check/neutron-grenade-multinode/908c4bc/ : SUCCESS in 1h 08m 20s\n- tempest-full http://logs.openstack.org/96/658096/3/check/tempest-full/9d8169f/ : FAILURE in 28m 55s\n- neutron-grenade http://logs.openstack.org/96/658096/3/check/neutron-grenade/24e7322/ : SUCCESS in 51m 24s","accounts_in_message":[],"_revision_number":3},{"id":"06aea795d2492ab7f23947be40bc98dbbb33da5e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-14 19:03:04.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"95668ae1ea26af3956295abef097ef420c22f396","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 20:58:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/96/658096/3/check/openstack-tox-docs/c1f5dc6/html/ : SUCCESS in 3m 06s\n- openstack-tox-pypy http://logs.openstack.org/96/658096/3/check/openstack-tox-pypy/d2ce5c5/ : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/96/658096/3/check/openstack-tox-pep8/1e76ac9/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/96/658096/3/check/openstack-tox-py27/c175bb2/ : SUCCESS in 3m 06s\n- openstack-tox-py36 http://logs.openstack.org/96/658096/3/check/openstack-tox-py36/7529c2d/ : SUCCESS in 3m 40s\n- openstack-tox-validate http://logs.openstack.org/96/658096/3/check/openstack-tox-validate/cce70f2/ : SUCCESS in 2m 43s\n- requirements-tox-babel http://logs.openstack.org/96/658096/3/check/requirements-tox-babel/02bfabf/ : SUCCESS in 2m 51s\n- requirements-tox-py27-check-uc http://logs.openstack.org/96/658096/3/check/requirements-tox-py27-check-uc/1c375f0/ : SUCCESS in 9m 14s\n- requirements-tox-py36-check-uc http://logs.openstack.org/96/658096/3/check/requirements-tox-py36-check-uc/ec02ffd/ : SUCCESS in 7m 23s\n- cross-cinder-py27 http://logs.openstack.org/96/658096/3/check/cross-cinder-py27/deb431a/ : SUCCESS in 7m 39s\n- cross-cinder-py36 http://logs.openstack.org/96/658096/3/check/cross-cinder-py36/e804303/ : SUCCESS in 12m 13s\n- cross-glance-py27 http://logs.openstack.org/96/658096/3/check/cross-glance-py27/a609a08/ : SUCCESS in 4m 22s\n- cross-glance-py36 http://logs.openstack.org/96/658096/3/check/cross-glance-py36/d5102eb/ : SUCCESS in 4m 03s\n- cross-horizon-py27 http://logs.openstack.org/96/658096/3/check/cross-horizon-py27/253b5bb/ : SUCCESS in 11m 00s\n- cross-horizon-py36 http://logs.openstack.org/96/658096/3/check/cross-horizon-py36/1108fd6/ : SUCCESS in 7m 59s\n- cross-horizon-npm http://logs.openstack.org/96/658096/3/check/cross-horizon-npm/8a1e8e9/ : SUCCESS in 5m 54s\n- cross-keystone-py27 http://logs.openstack.org/96/658096/3/check/cross-keystone-py27/5376184/ : SUCCESS in 27m 03s\n- cross-keystone-py36 http://logs.openstack.org/96/658096/3/check/cross-keystone-py36/838604a/ : SUCCESS in 22m 55s\n- cross-neutron-py27 http://logs.openstack.org/96/658096/3/check/cross-neutron-py27/5cf0db6/ : SUCCESS in 23m 07s\n- cross-neutron-py36 http://logs.openstack.org/96/658096/3/check/cross-neutron-py36/96ea5d5/ : SUCCESS in 27m 32s\n- cross-nova-py27 http://logs.openstack.org/96/658096/3/check/cross-nova-py27/d0cb8bf/ : SUCCESS in 12m 43s\n- cross-nova-py36 http://logs.openstack.org/96/658096/3/check/cross-nova-py36/f4dac13/ : SUCCESS in 11m 04s\n- cross-nova-functional http://logs.openstack.org/96/658096/3/check/cross-nova-functional/911b9fd/ : SUCCESS in 22m 01s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/96/658096/3/check/cross-osc-build-sphinx-docs/2736956/html/ : SUCCESS in 6m 59s\n- cross-swift-py27 http://logs.openstack.org/96/658096/3/check/cross-swift-py27/b0a2a08/ : SUCCESS in 7m 47s\n- cross-swift-py36 http://logs.openstack.org/96/658096/3/check/cross-swift-py36/828c64d/ : SUCCESS in 7m 30s\n- requirements-integration http://logs.openstack.org/96/658096/3/check/requirements-integration/f06f584/ : SUCCESS in 50m 58s\n- neutron-grenade-multinode http://logs.openstack.org/96/658096/3/check/neutron-grenade-multinode/b23be45/ : SUCCESS in 1h 32m 55s\n- tempest-full http://logs.openstack.org/96/658096/3/check/tempest-full/2d52ca2/ : SUCCESS in 1h 49m 46s\n- neutron-grenade http://logs.openstack.org/96/658096/3/check/neutron-grenade/ba9ec71/ : SUCCESS in 53m 36s","accounts_in_message":[],"_revision_number":3},{"id":"42777d44f406141e5596f19ba6d5d7c38195075a","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-15 16:46:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dc51209d02fe28594c0f40aa31f28bcd569dd482","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2019-05-15 16:52:52.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8f983badd5333182bf516f55d9c89830a177adba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 19:34:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/96/658096/4/check/openstack-tox-docs/2304a60/html/ : SUCCESS in 2m 34s\n- openstack-tox-pypy http://logs.openstack.org/96/658096/4/check/openstack-tox-pypy/b50b4af/ : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/96/658096/4/check/openstack-tox-pep8/01baa6f/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/96/658096/4/check/openstack-tox-py27/871a408/ : SUCCESS in 2m 48s\n- openstack-tox-py36 http://logs.openstack.org/96/658096/4/check/openstack-tox-py36/8b302c4/ : SUCCESS in 3m 20s\n- openstack-tox-validate http://logs.openstack.org/96/658096/4/check/openstack-tox-validate/7620b97/ : SUCCESS in 3m 01s\n- requirements-tox-babel http://logs.openstack.org/96/658096/4/check/requirements-tox-babel/2c131d7/ : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc http://logs.openstack.org/96/658096/4/check/requirements-tox-py27-check-uc/493a2ac/ : SUCCESS in 7m 14s\n- requirements-tox-py36-check-uc http://logs.openstack.org/96/658096/4/check/requirements-tox-py36-check-uc/52ed9e3/ : SUCCESS in 6m 55s\n- cross-cinder-py27 http://logs.openstack.org/96/658096/4/check/cross-cinder-py27/9cb4c87/ : SUCCESS in 11m 04s\n- cross-cinder-py36 http://logs.openstack.org/96/658096/4/check/cross-cinder-py36/2f63021/ : SUCCESS in 7m 46s\n- cross-glance-py27 http://logs.openstack.org/96/658096/4/check/cross-glance-py27/9ff50fb/ : SUCCESS in 4m 07s\n- cross-glance-py36 http://logs.openstack.org/96/658096/4/check/cross-glance-py36/be9a82d/ : SUCCESS in 4m 10s\n- cross-horizon-py27 http://logs.openstack.org/96/658096/4/check/cross-horizon-py27/87f7ae2/ : SUCCESS in 9m 56s\n- cross-horizon-py36 http://logs.openstack.org/96/658096/4/check/cross-horizon-py36/3fd4a38/ : SUCCESS in 9m 31s\n- cross-horizon-npm http://logs.openstack.org/96/658096/4/check/cross-horizon-npm/4fd4ba3/ : SUCCESS in 5m 12s\n- cross-keystone-py27 http://logs.openstack.org/96/658096/4/check/cross-keystone-py27/c849321/ : SUCCESS in 23m 29s\n- cross-keystone-py36 http://logs.openstack.org/96/658096/4/check/cross-keystone-py36/4ac1571/ : SUCCESS in 23m 35s\n- cross-neutron-py27 http://logs.openstack.org/96/658096/4/check/cross-neutron-py27/834d888/ : SUCCESS in 22m 17s\n- cross-neutron-py36 http://logs.openstack.org/96/658096/4/check/cross-neutron-py36/7a9e1d8/ : SUCCESS in 22m 07s\n- cross-nova-py27 http://logs.openstack.org/96/658096/4/check/cross-nova-py27/9f7fe69/ : SUCCESS in 11m 46s\n- cross-nova-py36 http://logs.openstack.org/96/658096/4/check/cross-nova-py36/64d31b4/ : SUCCESS in 12m 28s\n- cross-nova-functional http://logs.openstack.org/96/658096/4/check/cross-nova-functional/dac5810/ : SUCCESS in 19m 33s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/96/658096/4/check/cross-osc-build-sphinx-docs/238d196/html/ : SUCCESS in 5m 09s\n- cross-swift-py27 http://logs.openstack.org/96/658096/4/check/cross-swift-py27/8e03c37/ : SUCCESS in 7m 42s\n- cross-swift-py36 http://logs.openstack.org/96/658096/4/check/cross-swift-py36/85ba359/ : SUCCESS in 8m 16s\n- requirements-integration http://logs.openstack.org/96/658096/4/check/requirements-integration/e759429/ : SUCCESS in 57m 52s\n- neutron-grenade-multinode http://logs.openstack.org/96/658096/4/check/neutron-grenade-multinode/c1be852/ : SUCCESS in 1h 05m 39s\n- tempest-full http://logs.openstack.org/96/658096/4/check/tempest-full/dc09cfc/ : TIMED_OUT in 2h 05m 41s\n- neutron-grenade http://logs.openstack.org/96/658096/4/check/neutron-grenade/349cd18/ : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":4},{"id":"36db972f284bcab9f170ade246584b2a84637906","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-15 20:15:21.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b538ab8e436c27b109602f1962365b373383db4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 22:14:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/96/658096/4/check/openstack-tox-docs/0e0024f/html/ : SUCCESS in 2m 53s\n- openstack-tox-pypy http://logs.openstack.org/96/658096/4/check/openstack-tox-pypy/c39253d/ : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/96/658096/4/check/openstack-tox-pep8/8a91bf9/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/96/658096/4/check/openstack-tox-py27/969d0de/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/96/658096/4/check/openstack-tox-py36/da1509d/ : SUCCESS in 3m 59s\n- openstack-tox-validate http://logs.openstack.org/96/658096/4/check/openstack-tox-validate/1074daf/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/96/658096/4/check/requirements-tox-babel/281f314/ : SUCCESS in 2m 35s\n- requirements-tox-py27-check-uc http://logs.openstack.org/96/658096/4/check/requirements-tox-py27-check-uc/77f65e2/ : SUCCESS in 6m 58s\n- requirements-tox-py36-check-uc http://logs.openstack.org/96/658096/4/check/requirements-tox-py36-check-uc/09c80f9/ : SUCCESS in 5m 52s\n- cross-cinder-py27 http://logs.openstack.org/96/658096/4/check/cross-cinder-py27/768b8bd/ : SUCCESS in 8m 57s\n- cross-cinder-py36 http://logs.openstack.org/96/658096/4/check/cross-cinder-py36/57c6c4f/ : SUCCESS in 7m 14s\n- cross-glance-py27 http://logs.openstack.org/96/658096/4/check/cross-glance-py27/025263b/ : SUCCESS in 4m 07s\n- cross-glance-py36 http://logs.openstack.org/96/658096/4/check/cross-glance-py36/960e4e7/ : SUCCESS in 4m 08s\n- cross-horizon-py27 http://logs.openstack.org/96/658096/4/check/cross-horizon-py27/ca0ea0e/ : SUCCESS in 10m 10s\n- cross-horizon-py36 http://logs.openstack.org/96/658096/4/check/cross-horizon-py36/e4b8523/ : SUCCESS in 9m 32s\n- cross-horizon-npm http://logs.openstack.org/96/658096/4/check/cross-horizon-npm/05ab640/ : SUCCESS in 5m 54s\n- cross-keystone-py27 http://logs.openstack.org/96/658096/4/check/cross-keystone-py27/164a134/ : SUCCESS in 25m 49s\n- cross-keystone-py36 http://logs.openstack.org/96/658096/4/check/cross-keystone-py36/1272be9/ : SUCCESS in 23m 57s\n- cross-neutron-py27 http://logs.openstack.org/96/658096/4/check/cross-neutron-py27/e58ca20/ : SUCCESS in 27m 46s\n- cross-neutron-py36 http://logs.openstack.org/96/658096/4/check/cross-neutron-py36/1937908/ : SUCCESS in 21m 33s\n- cross-nova-py27 http://logs.openstack.org/96/658096/4/check/cross-nova-py27/ee6f1e1/ : SUCCESS in 22m 44s\n- cross-nova-py36 http://logs.openstack.org/96/658096/4/check/cross-nova-py36/9d6a489/ : SUCCESS in 10m 29s\n- cross-nova-functional http://logs.openstack.org/96/658096/4/check/cross-nova-functional/234fccb/ : SUCCESS in 20m 05s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/96/658096/4/check/cross-osc-build-sphinx-docs/99a5004/html/ : SUCCESS in 4m 56s\n- cross-swift-py27 http://logs.openstack.org/96/658096/4/check/cross-swift-py27/eca24ac/ : SUCCESS in 7m 20s\n- cross-swift-py36 http://logs.openstack.org/96/658096/4/check/cross-swift-py36/f9de0dc/ : SUCCESS in 8m 29s\n- requirements-integration http://logs.openstack.org/96/658096/4/check/requirements-integration/b718d33/ : SUCCESS in 58m 19s\n- neutron-grenade-multinode http://logs.openstack.org/96/658096/4/check/neutron-grenade-multinode/48c2df5/ : SUCCESS in 1h 07m 13s\n- tempest-full http://logs.openstack.org/96/658096/4/check/tempest-full/1c6d538/ : SUCCESS in 1h 53m 03s\n- neutron-grenade http://logs.openstack.org/96/658096/4/check/neutron-grenade/d9f715f/ : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":4},{"id":"f4cf1c2563ae50bd1906f2c1bc419c2a61cd3328","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-17 00:06:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b3a85d8fb7347500f324fa4a31b7fe6a09f8babc","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2019-05-17 00:11:53.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"210b23b078695a5033cfcde008b93e3ebc1c7c92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 02:06:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/96/658096/5/check/openstack-tox-docs/bd34188/html/ : SUCCESS in 3m 07s\n- openstack-tox-pypy http://logs.openstack.org/96/658096/5/check/openstack-tox-pypy/873331f/ : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/96/658096/5/check/openstack-tox-pep8/b597b95/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/96/658096/5/check/openstack-tox-py27/b543fac/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/96/658096/5/check/openstack-tox-py36/94885bc/ : SUCCESS in 3m 49s\n- openstack-tox-validate http://logs.openstack.org/96/658096/5/check/openstack-tox-validate/d719774/ : SUCCESS in 4m 17s\n- requirements-tox-babel http://logs.openstack.org/96/658096/5/check/requirements-tox-babel/d411e5f/ : SUCCESS in 2m 45s\n- requirements-tox-py27-check-uc http://logs.openstack.org/96/658096/5/check/requirements-tox-py27-check-uc/f4805f3/ : SUCCESS in 7m 37s\n- requirements-tox-py36-check-uc http://logs.openstack.org/96/658096/5/check/requirements-tox-py36-check-uc/6ab1d98/ : SUCCESS in 7m 10s\n- cross-cinder-py27 http://logs.openstack.org/96/658096/5/check/cross-cinder-py27/675e431/ : SUCCESS in 9m 37s\n- cross-cinder-py36 http://logs.openstack.org/96/658096/5/check/cross-cinder-py36/13b461b/ : SUCCESS in 8m 01s\n- cross-glance-py27 http://logs.openstack.org/96/658096/5/check/cross-glance-py27/0830b47/ : SUCCESS in 4m 16s\n- cross-glance-py36 http://logs.openstack.org/96/658096/5/check/cross-glance-py36/9716b93/ : SUCCESS in 4m 48s\n- cross-horizon-py27 http://logs.openstack.org/96/658096/5/check/cross-horizon-py27/386be47/ : SUCCESS in 9m 33s\n- cross-horizon-py36 http://logs.openstack.org/96/658096/5/check/cross-horizon-py36/299c223/ : SUCCESS in 8m 49s\n- cross-horizon-npm http://logs.openstack.org/96/658096/5/check/cross-horizon-npm/b5e6367/ : SUCCESS in 5m 36s\n- cross-keystone-py27 http://logs.openstack.org/96/658096/5/check/cross-keystone-py27/06106c5/ : SUCCESS in 24m 37s\n- cross-keystone-py36 http://logs.openstack.org/96/658096/5/check/cross-keystone-py36/bd2596b/ : SUCCESS in 22m 49s\n- cross-neutron-py27 http://logs.openstack.org/96/658096/5/check/cross-neutron-py27/f0043cc/ : SUCCESS in 25m 14s\n- cross-neutron-py36 http://logs.openstack.org/96/658096/5/check/cross-neutron-py36/7211fc7/ : SUCCESS in 21m 26s\n- cross-nova-py27 http://logs.openstack.org/96/658096/5/check/cross-nova-py27/5787e5a/ : SUCCESS in 10m 59s\n- cross-nova-py36 http://logs.openstack.org/96/658096/5/check/cross-nova-py36/f745100/ : SUCCESS in 11m 22s\n- cross-nova-functional http://logs.openstack.org/96/658096/5/check/cross-nova-functional/953e187/ : SUCCESS in 17m 38s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/96/658096/5/check/cross-osc-build-sphinx-docs/55912b7/html/ : SUCCESS in 6m 00s\n- cross-swift-py27 http://logs.openstack.org/96/658096/5/check/cross-swift-py27/08ed615/ : SUCCESS in 10m 03s\n- cross-swift-py36 http://logs.openstack.org/96/658096/5/check/cross-swift-py36/90e6645/ : SUCCESS in 7m 21s\n- requirements-integration http://logs.openstack.org/96/658096/5/check/requirements-integration/ece48ff/ : SUCCESS in 58m 49s\n- neutron-grenade-multinode http://logs.openstack.org/96/658096/5/check/neutron-grenade-multinode/7d37736/ : SUCCESS in 1h 04m 05s\n- tempest-full http://logs.openstack.org/96/658096/5/check/tempest-full/6633c2f/ : SUCCESS in 1h 44m 49s\n- neutron-grenade http://logs.openstack.org/96/658096/5/check/neutron-grenade/852241d/ : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":5},{"id":"c497526dee2131020aab0a14b0bff39a2e722d6a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-17 02:10:09.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"99e67e56d064c6a714002a45d89ec6fc4f4fab5c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-20 16:16:48.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d9b8c1faa579e7e3cb96a98bd2d87179e03c33f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 16:17:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2963c8dcd8894ee482c76cfeafe8185e4a1ff22b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 20:07:22.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/96/658096/5/gate/openstack-tox-docs/e48c815/html/ : SUCCESS in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/96/658096/5/gate/openstack-tox-pep8/f4cec68/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/96/658096/5/gate/openstack-tox-py27/1977723/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/96/658096/5/gate/openstack-tox-py36/36b28ac/ : SUCCESS in 3m 25s\n- openstack-tox-validate http://logs.openstack.org/96/658096/5/gate/openstack-tox-validate/de52eb7/ : SUCCESS in 3m 06s\n- requirements-tox-babel http://logs.openstack.org/96/658096/5/gate/requirements-tox-babel/3d242fb/ : SUCCESS in 5m 15s\n- requirements-tox-py27-check-uc http://logs.openstack.org/96/658096/5/gate/requirements-tox-py27-check-uc/145aa49/ : SUCCESS in 7m 00s\n- requirements-tox-py36-check-uc http://logs.openstack.org/96/658096/5/gate/requirements-tox-py36-check-uc/a18acd5/ : SUCCESS in 6m 53s\n- cross-cinder-py27 http://logs.openstack.org/96/658096/5/gate/cross-cinder-py27/f2acdf2/ : SUCCESS in 10m 38s\n- cross-cinder-py36 http://logs.openstack.org/96/658096/5/gate/cross-cinder-py36/34ec4f7/ : SUCCESS in 8m 14s\n- cross-glance-py27 http://logs.openstack.org/96/658096/5/gate/cross-glance-py27/2970ad9/ : SUCCESS in 4m 23s\n- cross-glance-py36 http://logs.openstack.org/96/658096/5/gate/cross-glance-py36/da893f6/ : SUCCESS in 4m 24s\n- cross-horizon-py27 http://logs.openstack.org/96/658096/5/gate/cross-horizon-py27/a838ce2/ : SUCCESS in 11m 21s\n- cross-horizon-py36 http://logs.openstack.org/96/658096/5/gate/cross-horizon-py36/dbbcf48/ : SUCCESS in 9m 23s\n- cross-horizon-npm http://logs.openstack.org/96/658096/5/gate/cross-horizon-npm/0d251bd/ : SUCCESS in 6m 46s\n- cross-keystone-py27 http://logs.openstack.org/96/658096/5/gate/cross-keystone-py27/8cb2ad3/ : SUCCESS in 30m 06s\n- cross-keystone-py36 http://logs.openstack.org/96/658096/5/gate/cross-keystone-py36/630c53d/ : SUCCESS in 21m 51s\n- cross-neutron-py27 http://logs.openstack.org/96/658096/5/gate/cross-neutron-py27/d735366/ : SUCCESS in 29m 46s\n- cross-neutron-py36 http://logs.openstack.org/96/658096/5/gate/cross-neutron-py36/d231d48/ : SUCCESS in 23m 15s\n- cross-nova-py27 http://logs.openstack.org/96/658096/5/gate/cross-nova-py27/6644984/ : SUCCESS in 11m 17s\n- cross-nova-py36 http://logs.openstack.org/96/658096/5/gate/cross-nova-py36/027d030/ : SUCCESS in 11m 18s\n- cross-nova-functional http://logs.openstack.org/96/658096/5/gate/cross-nova-functional/163ffe1/ : SUCCESS in 17m 42s\n- cross-osc-tox-docs http://logs.openstack.org/96/658096/5/gate/cross-osc-tox-docs/477eda9/html/ : SUCCESS in 5m 32s\n- cross-swift-py27 http://logs.openstack.org/96/658096/5/gate/cross-swift-py27/35f193a/ : SUCCESS in 8m 53s\n- cross-swift-py36 http://logs.openstack.org/96/658096/5/gate/cross-swift-py36/a2a221d/ : SUCCESS in 8m 26s\n- requirements-integration http://logs.openstack.org/96/658096/5/gate/requirements-integration/04920f0/ : SUCCESS in 1h 01m 02s\n- neutron-grenade-multinode http://logs.openstack.org/96/658096/5/gate/neutron-grenade-multinode/33ff4c4/ : SUCCESS in 1h 14m 16s\n- tempest-full http://logs.openstack.org/96/658096/5/gate/tempest-full/dccf5a1/ : SUCCESS in 1h 59m 52s\n- neutron-grenade http://logs.openstack.org/96/658096/5/gate/neutron-grenade/f9e0f5a/ : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":5},{"id":"037d35c071cce2dfc679d6a3ba39ebda7163ec35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 20:07:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"098e63b1c858d1dc419ae0b11d1f1c3936e77e3a","revisions":{"004f7b273024ab82d5423f12a62f968b8fa6114d":{"kind":"REWORK","_number":1,"created":"2019-05-09 14:13:20.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/96/658096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/658096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/658096/1"}}},"commit":{"parents":[{"commit":"842ed7bbb06c09dc01785265b589c228b5f928af","subject":"Merge \"Add missing format arg for missing g-r reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/842ed7bbb06c09dc01785265b589c228b5f928af"}]}],"author":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-09 14:12:44.000000000","tz":0},"committer":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-09 14:12:44.000000000","tz":0},"subject":"Add cinder extras purestorage library","message":"Add cinder extras purestorage library\n\nThis patch adds the cinder extras purestorage library for the cinder\npurestorage driver.\n\nactive: yes\ngood code: seems good\npy3: yes? nothing specifies it\nlicense compat: yes\npackaged: eh?\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nChange-Id: I65de3e178d573d0275fe1d944ce1231e2ef6c935\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/004f7b273024ab82d5423f12a62f968b8fa6114d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/004f7b273024ab82d5423f12a62f968b8fa6114d"}]},"branch":"refs/heads/master"},"a4a8a6e9aad2095da452d8449f161d6e0191b591":{"kind":"REWORK","_number":2,"created":"2019-05-12 20:07:07.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/96/658096/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/658096/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/658096/2"}}},"commit":{"parents":[{"commit":"61b95f944d701f5eff02bf8aea4d6b3e767e5822","subject":"Merge \"Fix docs environment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61b95f944d701f5eff02bf8aea4d6b3e767e5822"}]}],"author":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-09 14:12:44.000000000","tz":0},"committer":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-12 20:07:01.000000000","tz":-420},"subject":"Add cinder extras purestorage library","message":"Add cinder extras purestorage library\n\nThis patch adds the cinder extras purestorage library for the cinder\npurestorage driver.\n\nactive: yes\ngood code: seems good\npy3: yes? nothing specifies it\nlicense compat: yes\npackaged: eh?\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nChange-Id: I65de3e178d573d0275fe1d944ce1231e2ef6c935\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4a8a6e9aad2095da452d8449f161d6e0191b591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4a8a6e9aad2095da452d8449f161d6e0191b591"}]},"branch":"refs/heads/master"},"8510be3012d565009f91af62ebc509c918dd587b":{"kind":"REWORK","_number":3,"created":"2019-05-14 16:27:54.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/96/658096/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/658096/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/658096/3"}}},"commit":{"parents":[{"commit":"85f9aad798323c5f6dd8ce1a3dd5e009c2f944a1","subject":"Merge \"Set upper-constraint for scikit-learn\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/85f9aad798323c5f6dd8ce1a3dd5e009c2f944a1"}]}],"author":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-09 14:12:44.000000000","tz":0},"committer":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-14 16:28:14.000000000","tz":0},"subject":"Add cinder extras purestorage library","message":"Add cinder extras purestorage library\n\nThis patch adds the cinder extras purestorage library for the cinder\npurestorage driver.\n\nactive: yes\ngood code: seems good\npy3: yes? nothing specifies it\nlicense compat: yes\npackaged: eh?\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nChange-Id: I65de3e178d573d0275fe1d944ce1231e2ef6c935\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8510be3012d565009f91af62ebc509c918dd587b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8510be3012d565009f91af62ebc509c918dd587b"}]},"branch":"refs/heads/master"},"d296b65bd7c52ca4d9c4c0f8a7e8baa414445177":{"kind":"REWORK","_number":4,"created":"2019-05-15 16:46:50.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/96/658096/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/658096/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/658096/4"}}},"commit":{"parents":[{"commit":"abda9d216df814cf5ddb6282357419227a470467","subject":"Merge \"Add cinder extras storops library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/abda9d216df814cf5ddb6282357419227a470467"}]}],"author":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-09 14:12:44.000000000","tz":0},"committer":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-15 16:46:58.000000000","tz":0},"subject":"Add cinder extras purestorage library","message":"Add cinder extras purestorage library\n\nThis patch adds the cinder extras purestorage library for the cinder\npurestorage driver.\n\nactive: yes\ngood code: seems good\npy3: yes? nothing specifies it\nlicense compat: yes\npackaged: eh?\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nChange-Id: I65de3e178d573d0275fe1d944ce1231e2ef6c935\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d296b65bd7c52ca4d9c4c0f8a7e8baa414445177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d296b65bd7c52ca4d9c4c0f8a7e8baa414445177"}]},"branch":"refs/heads/master"},"098e63b1c858d1dc419ae0b11d1f1c3936e77e3a":{"kind":"REWORK","_number":5,"created":"2019-05-17 00:06:12.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/96/658096/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/658096/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/658096/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/658096/5"}}},"commit":{"parents":[{"commit":"3c3e972b7a1231f0ce625276e2d62f974fca3e01","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c3e972b7a1231f0ce625276e2d62f974fca3e01"}]}],"author":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-09 14:12:44.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-05-17 00:05:34.000000000","tz":-300},"subject":"Add cinder extras purestorage library","message":"Add cinder extras purestorage library\n\nThis patch adds the cinder extras purestorage library for the cinder\npurestorage driver.\n\nactive: yes\ngood code: seems good\npy3: yes? nothing specifies it\nlicense compat: yes\npackaged: eh?\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nChange-Id: I65de3e178d573d0275fe1d944ce1231e2ef6c935\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/098e63b1c858d1dc419ae0b11d1f1c3936e77e3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/098e63b1c858d1dc419ae0b11d1f1c3936e77e3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
