)]}'
{"id":"openstack%2Frequirements~658097","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ibc4e0e99c6d9b37e05ee33ab30440a09228c4cfd","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ibc4e0e99c6d9b37e05ee33ab30440a09228c4cfd","subject":"update constraint for python-monascaclient to new release 1.15.0","status":"MERGED","created":"2019-05-09 14:14:19.000000000","updated":"2019-05-14 09:32:17.000000000","submitted":"2019-05-14 09:32:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658097-1557826337689-a4b81bb3","meta_rev_id":"bb2110f5a3fb227f6df1af84a623deeeb1c282c3","_number":658097,"virtual_id_number":658097,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-14 09:32:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-13 11:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-13 07:21:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-05-13 11:46:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 05:09:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-12 13:26:23.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-13 11:46:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-14 09:32:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d8f8dfa24e832b19a333b54d8fedf64abfc9946","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 14:14:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb1c878c4c23f178e671988d3094eeae7f8926b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 18:10:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/97/658097/1/check/openstack-tox-pypy/2642a0c/ : FAILURE in 3m 02s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/658097/1/check/openstack-tox-pep8/858c89c/ : SUCCESS in 2m 34s\n- openstack-tox-py27 http://logs.openstack.org/97/658097/1/check/openstack-tox-py27/b2c46e2/ : FAILURE in 3m 21s\n- openstack-tox-py35 http://logs.openstack.org/97/658097/1/check/openstack-tox-py35/fffe3a1/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/97/658097/1/check/openstack-tox-py36/c9aa396/ : SUCCESS in 3m 13s\n- openstack-tox-validate http://logs.openstack.org/97/658097/1/check/openstack-tox-validate/26f7e12/ : SUCCESS in 2m 58s\n- requirements-tox-babel http://logs.openstack.org/97/658097/1/check/requirements-tox-babel/da412d5/ : SUCCESS in 3m 06s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/658097/1/check/requirements-tox-py27-check-uc/1c4c27f/ : SUCCESS in 8m 21s\n- requirements-tox-py35-check-uc http://logs.openstack.org/97/658097/1/check/requirements-tox-py35-check-uc/9e7c7a5/ : SUCCESS in 16m 44s\n- requirements-tox-py36-check-uc http://logs.openstack.org/97/658097/1/check/requirements-tox-py36-check-uc/1f258d2/ : SUCCESS in 7m 35s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/97/658097/1/check/cross-cinder-py27/622ad9e/ : FAILURE in 3m 38s\n- cross-cinder-py35 http://logs.openstack.org/97/658097/1/check/cross-cinder-py35/f299387/ : SUCCESS in 9m 12s\n- cross-glance-py27 http://logs.openstack.org/97/658097/1/check/cross-glance-py27/c7e7c86/ : SUCCESS in 4m 01s\n- cross-glance-py35 http://logs.openstack.org/97/658097/1/check/cross-glance-py35/e27e4a9/ : SUCCESS in 4m 07s\n- cross-horizon-py27 http://logs.openstack.org/97/658097/1/check/cross-horizon-py27/4c0b0c3/ : SUCCESS in 9m 58s\n- cross-horizon-py35 http://logs.openstack.org/97/658097/1/check/cross-horizon-py35/3b96c99/ : SUCCESS in 8m 35s\n- cross-horizon-npm http://logs.openstack.org/97/658097/1/check/cross-horizon-npm/32da877/ : SUCCESS in 6m 07s\n- cross-keystone-py27 http://logs.openstack.org/97/658097/1/check/cross-keystone-py27/346cce1/ : SUCCESS in 20m 19s\n- cross-keystone-py35 http://logs.openstack.org/97/658097/1/check/cross-keystone-py35/adc6826/ : SUCCESS in 25m 42s\n- cross-neutron-py27 http://logs.openstack.org/97/658097/1/check/cross-neutron-py27/4878393/ : SUCCESS in 24m 19s\n- cross-neutron-py35 http://logs.openstack.org/97/658097/1/check/cross-neutron-py35/666def9/ : SUCCESS in 23m 37s\n- cross-nova-py27 http://logs.openstack.org/97/658097/1/check/cross-nova-py27/4007b90/ : SUCCESS in 14m 00s\n- cross-nova-py35 http://logs.openstack.org/97/658097/1/check/cross-nova-py35/e01d0ff/ : SUCCESS in 12m 57s\n- cross-nova-functional http://logs.openstack.org/97/658097/1/check/cross-nova-functional/ed51e9c/ : SUCCESS in 19m 56s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/97/658097/1/check/cross-osc-build-sphinx-docs/c4ddca4/html/ : SUCCESS in 7m 45s\n- cross-swift-py27 http://logs.openstack.org/97/658097/1/check/cross-swift-py27/b2adcbf/ : SUCCESS in 7m 19s\n- cross-swift-py35 http://logs.openstack.org/97/658097/1/check/cross-swift-py35/1bbe1de/ : SUCCESS in 8m 04s\n- requirements-integration http://logs.openstack.org/97/658097/1/check/requirements-integration/0466849/ : FAILURE in 33m 51s\n- neutron-grenade-multinode http://logs.openstack.org/97/658097/1/check/neutron-grenade-multinode/d980553/ : SUCCESS in 1h 12m 22s\n- tempest-full http://logs.openstack.org/97/658097/1/check/tempest-full/19e9c0b/ : SUCCESS in 1h 36m 11s\n- neutron-grenade http://logs.openstack.org/97/658097/1/check/neutron-grenade/6213fb1/ : SUCCESS in 47m 21s","accounts_in_message":[],"_revision_number":1},{"id":"115aa95971ba68c779b219cb3c80d4454aaba558","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 05:09:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"73b5fca59f990858624963e54eccbbd2ffcfdeaf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 07:06:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/97/658097/1/check/openstack-tox-pypy/53cdc41/ : FAILURE in 4m 43s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/658097/1/check/openstack-tox-pep8/2b6d2c1/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/97/658097/1/check/openstack-tox-py27/0cd1b87/ : FAILURE in 4m 13s\n- openstack-tox-py35 http://logs.openstack.org/97/658097/1/check/openstack-tox-py35/4fdc30f/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/97/658097/1/check/openstack-tox-py36/6eed753/ : SUCCESS in 5m 26s\n- openstack-tox-validate http://logs.openstack.org/97/658097/1/check/openstack-tox-validate/fce62b4/ : SUCCESS in 4m 39s\n- requirements-tox-babel http://logs.openstack.org/97/658097/1/check/requirements-tox-babel/ae06254/ : SUCCESS in 3m 13s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/658097/1/check/requirements-tox-py27-check-uc/94278b7/ : SUCCESS in 7m 40s\n- requirements-tox-py35-check-uc http://logs.openstack.org/97/658097/1/check/requirements-tox-py35-check-uc/42ad1b2/ : SUCCESS in 7m 46s\n- requirements-tox-py36-check-uc http://logs.openstack.org/97/658097/1/check/requirements-tox-py36-check-uc/1b8d2c3/ : SUCCESS in 6m 13s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/97/658097/1/check/cross-cinder-py27/3fa912f/ : FAILURE in 5m 45s\n- cross-cinder-py35 http://logs.openstack.org/97/658097/1/check/cross-cinder-py35/ddf7afa/ : SUCCESS in 9m 47s\n- cross-glance-py27 http://logs.openstack.org/97/658097/1/check/cross-glance-py27/3ea42dc/ : SUCCESS in 5m 35s\n- cross-glance-py35 http://logs.openstack.org/97/658097/1/check/cross-glance-py35/88ffbdd/ : SUCCESS in 4m 19s\n- cross-horizon-py27 http://logs.openstack.org/97/658097/1/check/cross-horizon-py27/089ca2b/ : SUCCESS in 11m 29s\n- cross-horizon-py35 http://logs.openstack.org/97/658097/1/check/cross-horizon-py35/e0c3609/ : SUCCESS in 10m 24s\n- cross-horizon-npm http://logs.openstack.org/97/658097/1/check/cross-horizon-npm/cbd3b9c/ : SUCCESS in 7m 11s\n- cross-keystone-py27 http://logs.openstack.org/97/658097/1/check/cross-keystone-py27/4292d3f/ : SUCCESS in 21m 51s\n- cross-keystone-py35 http://logs.openstack.org/97/658097/1/check/cross-keystone-py35/0270f1b/ : SUCCESS in 24m 43s\n- cross-neutron-py27 http://logs.openstack.org/97/658097/1/check/cross-neutron-py27/57e4fc4/ : SUCCESS in 21m 55s\n- cross-neutron-py35 http://logs.openstack.org/97/658097/1/check/cross-neutron-py35/46e0169/ : SUCCESS in 24m 09s\n- cross-nova-py27 http://logs.openstack.org/97/658097/1/check/cross-nova-py27/003c82c/ : SUCCESS in 12m 52s\n- cross-nova-py35 http://logs.openstack.org/97/658097/1/check/cross-nova-py35/24fc7d1/ : SUCCESS in 14m 10s\n- cross-nova-functional http://logs.openstack.org/97/658097/1/check/cross-nova-functional/3bee6fe/ : SUCCESS in 19m 58s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/97/658097/1/check/cross-osc-build-sphinx-docs/a89506e/html/ : SUCCESS in 6m 24s\n- cross-swift-py27 http://logs.openstack.org/97/658097/1/check/cross-swift-py27/1944c2e/ : SUCCESS in 9m 26s\n- cross-swift-py35 http://logs.openstack.org/97/658097/1/check/cross-swift-py35/bbc98f7/ : SUCCESS in 8m 43s\n- requirements-integration http://logs.openstack.org/97/658097/1/check/requirements-integration/390f66b/ : FAILURE in 32m 05s\n- neutron-grenade-multinode http://logs.openstack.org/97/658097/1/check/neutron-grenade-multinode/23b9384/ : SUCCESS in 1h 11m 23s\n- tempest-full http://logs.openstack.org/97/658097/1/check/tempest-full/2e108ad/ : SUCCESS in 1h 50m 36s\n- neutron-grenade http://logs.openstack.org/97/658097/1/check/neutron-grenade/e2fc963/ : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":1},{"id":"151a11a3b856cfa2968f6d45c2d39c7c3e981eb2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-12 12:02:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8f5b9aea1b3bfd69ea8449c75c4800677520ac78","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 13:26:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"761f50f0e4264661b9554b390a88e4f22a0d7cd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 13:44:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/97/658097/1/check/openstack-tox-pypy/ff86bba/ : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/658097/1/check/openstack-tox-pep8/705e674/ : SUCCESS in 2m 51s\n- openstack-tox-py27 http://logs.openstack.org/97/658097/1/check/openstack-tox-py27/2781b49/ : SUCCESS in 3m 12s\n- openstack-tox-py35 http://logs.openstack.org/97/658097/1/check/openstack-tox-py35/99c82f2/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/97/658097/1/check/openstack-tox-py36/2983fb8/ : SUCCESS in 3m 09s\n- openstack-tox-validate http://logs.openstack.org/97/658097/1/check/openstack-tox-validate/77595aa/ : SUCCESS in 3m 01s\n- requirements-tox-babel http://logs.openstack.org/97/658097/1/check/requirements-tox-babel/ad980aa/ : SUCCESS in 2m 38s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/658097/1/check/requirements-tox-py27-check-uc/aef7726/ : SUCCESS in 5m 49s\n- requirements-tox-py35-check-uc http://logs.openstack.org/97/658097/1/check/requirements-tox-py35-check-uc/2a8cd37/ : SUCCESS in 6m 28s\n- requirements-tox-py36-check-uc http://logs.openstack.org/97/658097/1/check/requirements-tox-py36-check-uc/df75025/ : SUCCESS in 6m 01s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/97/658097/1/check/cross-cinder-py27/cedf1ac/ : FAILURE in 3m 52s\n- cross-cinder-py35 http://logs.openstack.org/97/658097/1/check/cross-cinder-py35/db34b82/ : SUCCESS in 7m 15s\n- cross-glance-py27 http://logs.openstack.org/97/658097/1/check/cross-glance-py27/89f1fa8/ : SUCCESS in 3m 32s\n- cross-glance-py35 http://logs.openstack.org/97/658097/1/check/cross-glance-py35/5f31998/ : SUCCESS in 4m 32s\n- cross-horizon-py27 http://logs.openstack.org/97/658097/1/check/cross-horizon-py27/f60ef7f/ : SUCCESS in 10m 08s\n- cross-horizon-py35 http://logs.openstack.org/97/658097/1/check/cross-horizon-py35/faed64a/ : SUCCESS in 9m 22s\n- cross-horizon-npm http://logs.openstack.org/97/658097/1/check/cross-horizon-npm/fd945aa/ : SUCCESS in 5m 16s\n- cross-keystone-py27 http://logs.openstack.org/97/658097/1/check/cross-keystone-py27/684e58d/ : SUCCESS in 17m 53s\n- cross-keystone-py35 http://logs.openstack.org/97/658097/1/check/cross-keystone-py35/855dce8/ : SUCCESS in 22m 38s\n- cross-neutron-py27 http://logs.openstack.org/97/658097/1/check/cross-neutron-py27/b5ed454/ : SUCCESS in 22m 40s\n- cross-neutron-py35 http://logs.openstack.org/97/658097/1/check/cross-neutron-py35/dab03fc/ : SUCCESS in 24m 58s\n- cross-nova-py27 http://logs.openstack.org/97/658097/1/check/cross-nova-py27/376ef10/ : SUCCESS in 12m 45s\n- cross-nova-py35 http://logs.openstack.org/97/658097/1/check/cross-nova-py35/092150e/ : SUCCESS in 14m 47s\n- cross-nova-functional http://logs.openstack.org/97/658097/1/check/cross-nova-functional/ebfe2e3/ : SUCCESS in 18m 49s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/97/658097/1/check/cross-osc-build-sphinx-docs/873516e/html/ : SUCCESS in 5m 32s\n- cross-swift-py27 http://logs.openstack.org/97/658097/1/check/cross-swift-py27/7f57655/ : SUCCESS in 7m 37s\n- cross-swift-py35 http://logs.openstack.org/97/658097/1/check/cross-swift-py35/8d63716/ : SUCCESS in 7m 18s\n- requirements-integration http://logs.openstack.org/97/658097/1/check/requirements-integration/afc504c/ : FAILURE in 43m 46s\n- neutron-grenade-multinode http://logs.openstack.org/97/658097/1/check/neutron-grenade-multinode/289bb85/ : SUCCESS in 1h 07m 33s\n- tempest-full http://logs.openstack.org/97/658097/1/check/tempest-full/fcc46cd/ : SUCCESS in 1h 41m 02s\n- neutron-grenade http://logs.openstack.org/97/658097/1/check/neutron-grenade/8c87f89/ : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e7c3f419409d0cc11ea83e4c21f00bce34525324","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 07:21:12.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"28e93cb662460c186621fe9b67f479077de8b601","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 09:28:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/97/658097/2/check/openstack-tox-pypy/980015f/ : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/658097/2/check/openstack-tox-pep8/3b19daf/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/97/658097/2/check/openstack-tox-py27/01215b8/ : SUCCESS in 3m 05s\n- openstack-tox-py35 http://logs.openstack.org/97/658097/2/check/openstack-tox-py35/737b47b/ : SUCCESS in 3m 39s\n- openstack-tox-py36 http://logs.openstack.org/97/658097/2/check/openstack-tox-py36/22e4e69/ : SUCCESS in 2m 47s\n- openstack-tox-validate http://logs.openstack.org/97/658097/2/check/openstack-tox-validate/98a1764/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/97/658097/2/check/requirements-tox-babel/1870f48/ : SUCCESS in 2m 45s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/658097/2/check/requirements-tox-py27-check-uc/f7c3618/ : SUCCESS in 6m 14s\n- requirements-tox-py35-check-uc http://logs.openstack.org/97/658097/2/check/requirements-tox-py35-check-uc/9d2f5ec/ : SUCCESS in 6m 07s\n- requirements-tox-py36-check-uc http://logs.openstack.org/97/658097/2/check/requirements-tox-py36-check-uc/a471f68/ : SUCCESS in 6m 11s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/97/658097/2/check/cross-cinder-py27/5ef6ce7/ : FAILURE in 3m 30s\n- cross-cinder-py35 http://logs.openstack.org/97/658097/2/check/cross-cinder-py35/6b57673/ : SUCCESS in 10m 14s\n- cross-glance-py27 http://logs.openstack.org/97/658097/2/check/cross-glance-py27/1cf2ea6/ : SUCCESS in 4m 19s\n- cross-glance-py35 http://logs.openstack.org/97/658097/2/check/cross-glance-py35/8343b96/ : SUCCESS in 3m 59s\n- cross-horizon-py27 http://logs.openstack.org/97/658097/2/check/cross-horizon-py27/0988994/ : SUCCESS in 9m 17s\n- cross-horizon-py35 http://logs.openstack.org/97/658097/2/check/cross-horizon-py35/4e49154/ : SUCCESS in 9m 50s\n- cross-horizon-npm http://logs.openstack.org/97/658097/2/check/cross-horizon-npm/b61ab5a/ : SUCCESS in 5m 09s\n- cross-keystone-py27 http://logs.openstack.org/97/658097/2/check/cross-keystone-py27/51357e4/ : SUCCESS in 21m 31s\n- cross-keystone-py35 http://logs.openstack.org/97/658097/2/check/cross-keystone-py35/3ca2ce3/ : SUCCESS in 25m 56s\n- cross-neutron-py27 http://logs.openstack.org/97/658097/2/check/cross-neutron-py27/3a0b4fc/ : SUCCESS in 23m 18s\n- cross-neutron-py35 http://logs.openstack.org/97/658097/2/check/cross-neutron-py35/d377398/ : SUCCESS in 25m 31s\n- cross-nova-py27 http://logs.openstack.org/97/658097/2/check/cross-nova-py27/5dfff67/ : SUCCESS in 12m 36s\n- cross-nova-py35 http://logs.openstack.org/97/658097/2/check/cross-nova-py35/9d5a67b/ : SUCCESS in 12m 01s\n- cross-nova-functional http://logs.openstack.org/97/658097/2/check/cross-nova-functional/776bd03/ : SUCCESS in 17m 54s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/97/658097/2/check/cross-osc-build-sphinx-docs/aac483b/html/ : SUCCESS in 5m 10s\n- cross-swift-py27 http://logs.openstack.org/97/658097/2/check/cross-swift-py27/5d702cb/ : SUCCESS in 7m 34s\n- cross-swift-py35 http://logs.openstack.org/97/658097/2/check/cross-swift-py35/2496543/ : SUCCESS in 7m 31s\n- requirements-integration http://logs.openstack.org/97/658097/2/check/requirements-integration/67946f6/ : SUCCESS in 44m 21s\n- neutron-grenade-multinode http://logs.openstack.org/97/658097/2/check/neutron-grenade-multinode/e71e855/ : SUCCESS in 1h 02m 38s\n- tempest-full http://logs.openstack.org/97/658097/2/check/tempest-full/938daad/ : SUCCESS in 2h 01m 59s\n- neutron-grenade http://logs.openstack.org/97/658097/2/check/neutron-grenade/eee1dbe/ : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":2},{"id":"131d511835ca2f277e77d3310a30f2748c95bda5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-13 11:46:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThese will need to wait for https://review.opendev.org/#/c/658297 to make it through. Recheck should clear it then.","accounts_in_message":[],"_revision_number":2},{"id":"b10b5b34681bc65d201c22156858e1faedd51ab1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-13 13:00:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"73162d03bf5aa1d02c944db33d12dc4a1c7a3608","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 13:31:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/97/658097/2/check/openstack-tox-pypy/2c85e20/ : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/658097/2/check/openstack-tox-pep8/43128ae/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/97/658097/2/check/openstack-tox-py27/1f7b4c9/ : SUCCESS in 2m 45s\n- openstack-tox-py35 http://logs.openstack.org/97/658097/2/check/openstack-tox-py35/5d05187/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/97/658097/2/check/openstack-tox-py36/3f7865d/ : SUCCESS in 3m 19s\n- openstack-tox-validate http://logs.openstack.org/97/658097/2/check/openstack-tox-validate/02138d1/ : SUCCESS in 3m 19s\n- requirements-tox-babel http://logs.openstack.org/97/658097/2/check/requirements-tox-babel/dbd44de/ : SUCCESS in 2m 47s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/658097/2/check/requirements-tox-py27-check-uc/d9b4e1f/ : SUCCESS in 6m 41s\n- requirements-tox-py35-check-uc http://logs.openstack.org/97/658097/2/check/requirements-tox-py35-check-uc/ba69697/ : SUCCESS in 5m 51s\n- requirements-tox-py36-check-uc http://logs.openstack.org/97/658097/2/check/requirements-tox-py36-check-uc/d1cef09/ : SUCCESS in 6m 54s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/97/658097/2/check/cross-cinder-py27/ef20de1/ : FAILURE in 4m 08s\n- cross-cinder-py35 http://logs.openstack.org/97/658097/2/check/cross-cinder-py35/1facae7/ : SUCCESS in 8m 57s\n- cross-glance-py27 http://logs.openstack.org/97/658097/2/check/cross-glance-py27/36465ef/ : SUCCESS in 4m 23s\n- cross-glance-py35 http://logs.openstack.org/97/658097/2/check/cross-glance-py35/3bb38bd/ : SUCCESS in 4m 37s\n- cross-horizon-py27 http://logs.openstack.org/97/658097/2/check/cross-horizon-py27/cfcd0df/ : SUCCESS in 10m 34s\n- cross-horizon-py35 http://logs.openstack.org/97/658097/2/check/cross-horizon-py35/8913cd6/ : SUCCESS in 10m 14s\n- cross-horizon-npm http://logs.openstack.org/97/658097/2/check/cross-horizon-npm/6a5a07d/ : SUCCESS in 5m 58s\n- cross-keystone-py27 http://logs.openstack.org/97/658097/2/check/cross-keystone-py27/61c61d4/ : SUCCESS in 24m 31s\n- cross-keystone-py35 http://logs.openstack.org/97/658097/2/check/cross-keystone-py35/2a89be8/ : SUCCESS in 30m 49s\n- cross-neutron-py27 http://logs.openstack.org/97/658097/2/check/cross-neutron-py27/239d4f7/ : SUCCESS in 24m 17s\n- cross-neutron-py35 http://logs.openstack.org/97/658097/2/check/cross-neutron-py35/57eeb0d/ : SUCCESS in 23m 09s\n- cross-nova-py27 http://logs.openstack.org/97/658097/2/check/cross-nova-py27/9f7f272/ : SUCCESS in 12m 09s\n- cross-nova-py35 http://logs.openstack.org/97/658097/2/check/cross-nova-py35/32bc2e3/ : SUCCESS in 13m 44s\n- cross-nova-functional http://logs.openstack.org/97/658097/2/check/cross-nova-functional/37b8791/ : SUCCESS in 20m 45s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/97/658097/2/check/cross-osc-build-sphinx-docs/cc6128e/html/ : SUCCESS in 5m 07s\n- cross-swift-py27 http://logs.openstack.org/97/658097/2/check/cross-swift-py27/f372f8d/ : SUCCESS in 7m 20s\n- cross-swift-py35 http://logs.openstack.org/97/658097/2/check/cross-swift-py35/d0de7ee/ : SUCCESS in 7m 35s\n- requirements-integration http://logs.openstack.org/97/658097/2/check/requirements-integration/433d9e7/ : SUCCESS in 52m 02s\n- neutron-grenade-multinode http://logs.openstack.org/97/658097/2/check/neutron-grenade-multinode/502569b/ : SUCCESS in 1h 17m 23s\n- tempest-full http://logs.openstack.org/97/658097/2/check/tempest-full/47d3e0b/ : SUCCESS in 1h 39m 56s\n- neutron-grenade http://logs.openstack.org/97/658097/2/check/neutron-grenade/1e56727/ : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e7f06b7e64f1d9638c228666f4c0025c7cc3e0de","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 20:15:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1eb9a5c6e395c46a35fd552410fd286863839f22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 23:47:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/97/658097/2/check/openstack-tox-pypy/80c54d2/ : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/658097/2/check/openstack-tox-pep8/404eed6/ : SUCCESS in 2m 35s\n- openstack-tox-py27 http://logs.openstack.org/97/658097/2/check/openstack-tox-py27/6fbf91d/ : SUCCESS in 3m 27s\n- openstack-tox-py35 http://logs.openstack.org/97/658097/2/check/openstack-tox-py35/27a6e11/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/97/658097/2/check/openstack-tox-py36/34fb51b/ : SUCCESS in 3m 26s\n- openstack-tox-validate http://logs.openstack.org/97/658097/2/check/openstack-tox-validate/1f6c1e1/ : SUCCESS in 3m 00s\n- requirements-tox-babel http://logs.openstack.org/97/658097/2/check/requirements-tox-babel/6a9a307/ : SUCCESS in 4m 19s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/658097/2/check/requirements-tox-py27-check-uc/8544fea/ : SUCCESS in 6m 47s\n- requirements-tox-py35-check-uc http://logs.openstack.org/97/658097/2/check/requirements-tox-py35-check-uc/18ecac0/ : SUCCESS in 7m 09s\n- requirements-tox-py36-check-uc http://logs.openstack.org/97/658097/2/check/requirements-tox-py36-check-uc/86e56bf/ : SUCCESS in 6m 32s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/97/658097/2/check/cross-cinder-py27/b515f7e/ : FAILURE in 8m 44s\n- cross-cinder-py35 http://logs.openstack.org/97/658097/2/check/cross-cinder-py35/f9c0cc8/ : SUCCESS in 9m 57s\n- cross-glance-py27 http://logs.openstack.org/97/658097/2/check/cross-glance-py27/8db02b0/ : SUCCESS in 4m 21s\n- cross-glance-py35 http://logs.openstack.org/97/658097/2/check/cross-glance-py35/c62e5b5/ : SUCCESS in 4m 42s\n- cross-horizon-py27 http://logs.openstack.org/97/658097/2/check/cross-horizon-py27/60c3998/ : SUCCESS in 10m 55s\n- cross-horizon-py35 http://logs.openstack.org/97/658097/2/check/cross-horizon-py35/2cd5919/ : SUCCESS in 10m 14s\n- cross-horizon-npm http://logs.openstack.org/97/658097/2/check/cross-horizon-npm/cbaa165/ : SUCCESS in 5m 14s\n- cross-keystone-py27 http://logs.openstack.org/97/658097/2/check/cross-keystone-py27/deba2f9/ : SUCCESS in 40m 41s\n- cross-keystone-py35 http://logs.openstack.org/97/658097/2/check/cross-keystone-py35/e6a079b/ : TIMED_OUT in 41m 19s\n- cross-neutron-py27 http://logs.openstack.org/97/658097/2/check/cross-neutron-py27/6760287/ : SUCCESS in 25m 30s\n- cross-neutron-py35 http://logs.openstack.org/97/658097/2/check/cross-neutron-py35/710e92c/ : SUCCESS in 28m 03s\n- cross-nova-py27 http://logs.openstack.org/97/658097/2/check/cross-nova-py27/552255c/ : SUCCESS in 12m 44s\n- cross-nova-py35 http://logs.openstack.org/97/658097/2/check/cross-nova-py35/210d1e4/ : SUCCESS in 14m 09s\n- cross-nova-functional http://logs.openstack.org/97/658097/2/check/cross-nova-functional/4588f35/ : SUCCESS in 21m 59s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/97/658097/2/check/cross-osc-build-sphinx-docs/9bd1225/html/ : SUCCESS in 6m 06s\n- cross-swift-py27 http://logs.openstack.org/97/658097/2/check/cross-swift-py27/734ff99/ : SUCCESS in 7m 38s\n- cross-swift-py35 http://logs.openstack.org/97/658097/2/check/cross-swift-py35/ad4a526/ : SUCCESS in 8m 48s\n- requirements-integration http://logs.openstack.org/97/658097/2/check/requirements-integration/f76d3aa/ : SUCCESS in 52m 28s\n- neutron-grenade-multinode http://logs.openstack.org/97/658097/2/check/neutron-grenade-multinode/b39ec03/ : SUCCESS in 1h 17m 18s\n- tempest-full http://logs.openstack.org/97/658097/2/check/tempest-full/6513c41/ : SUCCESS in 1h 31m 34s\n- neutron-grenade http://logs.openstack.org/97/658097/2/check/neutron-grenade/64879b7/ : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":2},{"id":"322fb78eabf52315a38f340f5c6d2ae9ceb98d21","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-14 05:45:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dc61514152f28095dd2c68bfe4944b8955d2c40e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 07:51:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/97/658097/2/check/openstack-tox-pypy/9c5dd49/ : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/97/658097/2/check/openstack-tox-pep8/bdff1d7/ : SUCCESS in 2m 42s\n- openstack-tox-py27 http://logs.openstack.org/97/658097/2/check/openstack-tox-py27/f392633/ : SUCCESS in 3m 20s\n- openstack-tox-py35 http://logs.openstack.org/97/658097/2/check/openstack-tox-py35/2ffa5ea/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/97/658097/2/check/openstack-tox-py36/b1eac83/ : SUCCESS in 2m 51s\n- openstack-tox-validate http://logs.openstack.org/97/658097/2/check/openstack-tox-validate/af6bda5/ : SUCCESS in 2m 52s\n- requirements-tox-babel http://logs.openstack.org/97/658097/2/check/requirements-tox-babel/533c839/ : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/658097/2/check/requirements-tox-py27-check-uc/5ae4873/ : SUCCESS in 5m 58s\n- requirements-tox-py35-check-uc http://logs.openstack.org/97/658097/2/check/requirements-tox-py35-check-uc/40c270d/ : SUCCESS in 9m 11s\n- requirements-tox-py36-check-uc http://logs.openstack.org/97/658097/2/check/requirements-tox-py36-check-uc/d218f76/ : SUCCESS in 6m 58s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/97/658097/2/check/cross-cinder-py27/1da0763/ : SUCCESS in 7m 44s\n- cross-cinder-py35 http://logs.openstack.org/97/658097/2/check/cross-cinder-py35/15edbde/ : SUCCESS in 9m 04s\n- cross-glance-py27 http://logs.openstack.org/97/658097/2/check/cross-glance-py27/80ad77e/ : SUCCESS in 4m 25s\n- cross-glance-py35 http://logs.openstack.org/97/658097/2/check/cross-glance-py35/555457c/ : SUCCESS in 4m 23s\n- cross-horizon-py27 http://logs.openstack.org/97/658097/2/check/cross-horizon-py27/f275b6a/ : SUCCESS in 10m 08s\n- cross-horizon-py35 http://logs.openstack.org/97/658097/2/check/cross-horizon-py35/75fcfcc/ : SUCCESS in 9m 43s\n- cross-horizon-npm http://logs.openstack.org/97/658097/2/check/cross-horizon-npm/46ee81c/ : SUCCESS in 6m 28s\n- cross-keystone-py27 http://logs.openstack.org/97/658097/2/check/cross-keystone-py27/3d31081/ : SUCCESS in 21m 51s\n- cross-keystone-py35 http://logs.openstack.org/97/658097/2/check/cross-keystone-py35/f4981dd/ : SUCCESS in 33m 06s\n- cross-neutron-py27 http://logs.openstack.org/97/658097/2/check/cross-neutron-py27/f3c8963/ : SUCCESS in 22m 03s\n- cross-neutron-py35 http://logs.openstack.org/97/658097/2/check/cross-neutron-py35/9d40dd8/ : SUCCESS in 32m 03s\n- cross-nova-py27 http://logs.openstack.org/97/658097/2/check/cross-nova-py27/6cb2ddb/ : SUCCESS in 11m 36s\n- cross-nova-py35 http://logs.openstack.org/97/658097/2/check/cross-nova-py35/5f0ad89/ : SUCCESS in 11m 41s\n- cross-nova-functional http://logs.openstack.org/97/658097/2/check/cross-nova-functional/36493d1/ : SUCCESS in 20m 16s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/97/658097/2/check/cross-osc-build-sphinx-docs/a57c1dd/html/ : SUCCESS in 5m 18s\n- cross-swift-py27 http://logs.openstack.org/97/658097/2/check/cross-swift-py27/1f15072/ : SUCCESS in 10m 03s\n- cross-swift-py35 http://logs.openstack.org/97/658097/2/check/cross-swift-py35/856ac24/ : SUCCESS in 9m 00s\n- requirements-integration http://logs.openstack.org/97/658097/2/check/requirements-integration/daac17a/ : SUCCESS in 53m 17s\n- neutron-grenade-multinode http://logs.openstack.org/97/658097/2/check/neutron-grenade-multinode/00bb4be/ : SUCCESS in 1h 12m 45s\n- tempest-full http://logs.openstack.org/97/658097/2/check/tempest-full/dabcecc/ : SUCCESS in 1h 52m 05s\n- neutron-grenade http://logs.openstack.org/97/658097/2/check/neutron-grenade/54c70b6/ : SUCCESS in 51m 02s","accounts_in_message":[],"_revision_number":2},{"id":"7bff6b1db0134741d6156f8b5dd6303b0786dac9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 07:51:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b63d6d9bf9052a0138d7f44a26691de907993592","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 09:32:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"605a258fbc40bf6dc1170685c36afbd836762d7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 09:32:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/97/658097/2/gate/openstack-tox-pep8/d65681e/ : SUCCESS in 2m 44s\n- openstack-tox-py27 http://logs.openstack.org/97/658097/2/gate/openstack-tox-py27/27135d6/ : SUCCESS in 4m 19s\n- openstack-tox-py35 http://logs.openstack.org/97/658097/2/gate/openstack-tox-py35/66651dc/ : SUCCESS in 5m 27s\n- openstack-tox-py36 http://logs.openstack.org/97/658097/2/gate/openstack-tox-py36/e262e87/ : SUCCESS in 2m 46s\n- openstack-tox-validate http://logs.openstack.org/97/658097/2/gate/openstack-tox-validate/586e46d/ : SUCCESS in 4m 56s\n- requirements-tox-babel http://logs.openstack.org/97/658097/2/gate/requirements-tox-babel/eab1d2b/ : SUCCESS in 2m 36s\n- requirements-tox-py27-check-uc http://logs.openstack.org/97/658097/2/gate/requirements-tox-py27-check-uc/9c47a00/ : SUCCESS in 9m 09s\n- requirements-tox-py35-check-uc http://logs.openstack.org/97/658097/2/gate/requirements-tox-py35-check-uc/4e04b5b/ : SUCCESS in 6m 56s\n- cross-cinder-py27 http://logs.openstack.org/97/658097/2/gate/cross-cinder-py27/83a21e5/ : SUCCESS in 8m 16s\n- cross-cinder-py35 http://logs.openstack.org/97/658097/2/gate/cross-cinder-py35/b81badc/ : SUCCESS in 8m 17s\n- cross-glance-py27 http://logs.openstack.org/97/658097/2/gate/cross-glance-py27/c0a2c5c/ : SUCCESS in 4m 04s\n- cross-glance-py35 http://logs.openstack.org/97/658097/2/gate/cross-glance-py35/971fdd2/ : SUCCESS in 4m 20s\n- cross-horizon-py27 http://logs.openstack.org/97/658097/2/gate/cross-horizon-py27/3527606/ : SUCCESS in 10m 14s\n- cross-horizon-py35 http://logs.openstack.org/97/658097/2/gate/cross-horizon-py35/516fe2a/ : SUCCESS in 7m 06s\n- cross-horizon-npm http://logs.openstack.org/97/658097/2/gate/cross-horizon-npm/592cbdc/ : SUCCESS in 5m 46s\n- cross-keystone-py27 http://logs.openstack.org/97/658097/2/gate/cross-keystone-py27/a359646/ : SUCCESS in 24m 25s\n- cross-keystone-py35 http://logs.openstack.org/97/658097/2/gate/cross-keystone-py35/c6f5935/ : SUCCESS in 26m 01s\n- cross-neutron-py27 http://logs.openstack.org/97/658097/2/gate/cross-neutron-py27/7d66fe8/ : SUCCESS in 20m 57s\n- cross-neutron-py35 http://logs.openstack.org/97/658097/2/gate/cross-neutron-py35/3f5a906/ : SUCCESS in 23m 23s\n- cross-nova-py27 http://logs.openstack.org/97/658097/2/gate/cross-nova-py27/59ff28f/ : SUCCESS in 11m 37s\n- cross-nova-py35 http://logs.openstack.org/97/658097/2/gate/cross-nova-py35/f94d58d/ : SUCCESS in 12m 50s\n- cross-nova-functional http://logs.openstack.org/97/658097/2/gate/cross-nova-functional/20b36e8/ : SUCCESS in 17m 08s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/97/658097/2/gate/cross-osc-build-sphinx-docs/fbeeee8/html/ : SUCCESS in 4m 59s\n- cross-swift-py27 http://logs.openstack.org/97/658097/2/gate/cross-swift-py27/c8efc2a/ : SUCCESS in 8m 11s\n- cross-swift-py35 http://logs.openstack.org/97/658097/2/gate/cross-swift-py35/1ab6e25/ : SUCCESS in 7m 37s\n- requirements-integration http://logs.openstack.org/97/658097/2/gate/requirements-integration/a9b6a0e/ : SUCCESS in 55m 07s\n- neutron-grenade-multinode http://logs.openstack.org/97/658097/2/gate/neutron-grenade-multinode/0f501b1/ : SUCCESS in 1h 01m 47s\n- tempest-full http://logs.openstack.org/97/658097/2/gate/tempest-full/c7a3002/ : SUCCESS in 1h 36m 01s\n- neutron-grenade http://logs.openstack.org/97/658097/2/gate/neutron-grenade/0d82fba/ : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44e985b22273c3b63301ba80c32ac0b070d596c1","revisions":{"1001f539e82a7e313c25870e87a49ddd49053fff":{"kind":"REWORK","_number":1,"created":"2019-05-09 14:14:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/658097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/658097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/658097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/658097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/658097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/658097/1"}}},"commit":{"parents":[{"commit":"1baf07a77333fc57d9d4a669d0265b0e687fd031","subject":"Merge \"update constraint for openstack-release-test to new release 1.4.2\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1baf07a77333fc57d9d4a669d0265b0e687fd031"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 14:14:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 14:14:17.000000000","tz":0},"subject":"update constraint for python-monascaclient to new release 1.15.0","message":"update constraint for python-monascaclient to new release 1.15.0\n\nChange-Id: Ibc4e0e99c6d9b37e05ee33ab30440a09228c4cfd\nmeta:version: 1.15.0\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Doug Szumski \u003cdoug@stackhpc.com\u003e\nmeta:release:Commit: Doug Szumski \u003cdoug@stackhpc.com\u003e\nmeta:release:Change-Id: I768f7c1fcd43fb605a714cf4f1da2943d10866ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1001f539e82a7e313c25870e87a49ddd49053fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1001f539e82a7e313c25870e87a49ddd49053fff"}]},"branch":"refs/heads/stable/stein"},"44e985b22273c3b63301ba80c32ac0b070d596c1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-13 07:21:12.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/97/658097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/658097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/658097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/658097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/658097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/658097/2"}}},"commit":{"parents":[{"commit":"aa266cbc630bf853c0bfe9e3333fbf0e82157981","subject":"Add upper-constraints enforcement for local reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa266cbc630bf853c0bfe9e3333fbf0e82157981"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 14:14:17.000000000","tz":0},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-13 07:21:12.000000000","tz":0},"subject":"update constraint for python-monascaclient to new release 1.15.0","message":"update constraint for python-monascaclient to new release 1.15.0\n\nChange-Id: Ibc4e0e99c6d9b37e05ee33ab30440a09228c4cfd\nmeta:version: 1.15.0\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Doug Szumski \u003cdoug@stackhpc.com\u003e\nmeta:release:Commit: Doug Szumski \u003cdoug@stackhpc.com\u003e\nmeta:release:Change-Id: I768f7c1fcd43fb605a714cf4f1da2943d10866ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44e985b22273c3b63301ba80c32ac0b070d596c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44e985b22273c3b63301ba80c32ac0b070d596c1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
