)]}'
{"id":"openstack%2Frequirements~658099","triplet_id":"openstack%2Frequirements~stable%2Frocky~I2cdf8d32a20ca57cd923cfca4aacd3500c085b7a","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"I2cdf8d32a20ca57cd923cfca4aacd3500c085b7a","subject":"update constraint for os-brick to new release 2.5.7","status":"MERGED","created":"2019-05-09 14:18:34.000000000","updated":"2019-05-13 19:48:12.000000000","submitted":"2019-05-13 19:48:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658099-1557776892869-71816a1a","meta_rev_id":"9673427a8f1d36e66619a51a9eb1b398d3f7ebd2","_number":658099,"virtual_id_number":658099,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-13 19:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-05-13 17:59:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-05-12 18:54:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-05-13 17:59:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 13:52:07.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-13 17:59:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-13 19:48:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2bbb069f19243a484657c929418445a7ba196099","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 14:18:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f72b6a551fe3bac48742628f43c6e4fd3f7715de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 18:13:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/99/658099/1/check/openstack-tox-pypy/d1269ff/ : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658099/1/check/openstack-tox-pep8/8165b54/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/99/658099/1/check/openstack-tox-py27/b7d949c/ : SUCCESS in 3m 12s\n- openstack-tox-py35 http://logs.openstack.org/99/658099/1/check/openstack-tox-py35/8182e12/ : SUCCESS in 3m 01s\n- openstack-tox-validate http://logs.openstack.org/99/658099/1/check/openstack-tox-validate/003e175/ : SUCCESS in 2m 44s\n- requirements-tox-babel http://logs.openstack.org/99/658099/1/check/requirements-tox-babel/1b44bbf/ : SUCCESS in 2m 40s\n- requirements-tox-py27-check-uc http://logs.openstack.org/99/658099/1/check/requirements-tox-py27-check-uc/47afde3/ : SUCCESS in 7m 11s\n- requirements-tox-py35-check-uc http://logs.openstack.org/99/658099/1/check/requirements-tox-py35-check-uc/7c9e563/ : SUCCESS in 11m 38s\n- requirements-tox-py36-check-uc http://logs.openstack.org/99/658099/1/check/requirements-tox-py36-check-uc/06f05e7/ : SUCCESS in 12m 03s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/99/658099/1/check/cross-cinder-py27/c68a0d5/ : SUCCESS in 11m 40s\n- cross-cinder-py35 http://logs.openstack.org/99/658099/1/check/cross-cinder-py35/cc10005/ : SUCCESS in 9m 09s\n- cross-glance-py27 http://logs.openstack.org/99/658099/1/check/cross-glance-py27/abe92f0/ : SUCCESS in 6m 13s\n- cross-glance-py35 http://logs.openstack.org/99/658099/1/check/cross-glance-py35/e1c07c2/ : SUCCESS in 5m 49s\n- cross-horizon-py27 http://logs.openstack.org/99/658099/1/check/cross-horizon-py27/a2f91ac/ : SUCCESS in 8m 11s\n- cross-horizon-py35 http://logs.openstack.org/99/658099/1/check/cross-horizon-py35/ad7d276/ : SUCCESS in 7m 57s\n- cross-keystone-py27 http://logs.openstack.org/99/658099/1/check/cross-keystone-py27/39ed1b5/ : SUCCESS in 12m 39s\n- cross-keystone-py35 http://logs.openstack.org/99/658099/1/check/cross-keystone-py35/6bb659e/ : SUCCESS in 12m 15s\n- cross-neutron-py27 http://logs.openstack.org/99/658099/1/check/cross-neutron-py27/7bb7f7e/ : SUCCESS in 24m 43s\n- cross-neutron-py35 http://logs.openstack.org/99/658099/1/check/cross-neutron-py35/82fe27e/ : SUCCESS in 34m 36s\n- cross-nova-py27 http://logs.openstack.org/99/658099/1/check/cross-nova-py27/a83357b/ : SUCCESS in 15m 47s\n- cross-nova-py35 http://logs.openstack.org/99/658099/1/check/cross-nova-py35/fb56acf/ : SUCCESS in 16m 25s\n- cross-nova-functional http://logs.openstack.org/99/658099/1/check/cross-nova-functional/f6667bc/ : SUCCESS in 25m 14s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/99/658099/1/check/cross-osc-build-sphinx-docs/f0e8fe7/html/ : SUCCESS in 5m 32s\n- cross-swift-py27 http://logs.openstack.org/99/658099/1/check/cross-swift-py27/f8c8c2b/ : SUCCESS in 7m 59s\n- cross-swift-py35 http://logs.openstack.org/99/658099/1/check/cross-swift-py35/5e9ed9b/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/99/658099/1/check/openstack-tox-py36/132886e/ : SUCCESS in 4m 39s\n- requirements-integration http://logs.openstack.org/99/658099/1/check/requirements-integration/232c4a5/ : FAILURE in 38m 44s\n- neutron-grenade-multinode http://logs.openstack.org/99/658099/1/check/neutron-grenade-multinode/753a035/ : SUCCESS in 1h 09m 17s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658099/1/check/legacy-tempest-dsvm-neutron-full/e82d879/ : SUCCESS in 1h 36m 25s\n- neutron-grenade http://logs.openstack.org/99/658099/1/check/neutron-grenade/5ed82eb/ : SUCCESS in 1h 06m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4fb87dbf2b36e96e1087c63fad0bf49c79f4533d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 05:09:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6eae8e39b2de4b4f961d5aafdb84594f073128b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 06:42:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/99/658099/1/check/openstack-tox-pypy/0843a85/ : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658099/1/check/openstack-tox-pep8/69ad13d/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/99/658099/1/check/openstack-tox-py27/2641d7e/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/99/658099/1/check/openstack-tox-py35/2e3f3c2/ : SUCCESS in 5m 22s\n- openstack-tox-validate http://logs.openstack.org/99/658099/1/check/openstack-tox-validate/a1bf362/ : SUCCESS in 3m 17s\n- requirements-tox-babel http://logs.openstack.org/99/658099/1/check/requirements-tox-babel/2dd86bf/ : SUCCESS in 5m 19s\n- requirements-tox-py27-check-uc http://logs.openstack.org/99/658099/1/check/requirements-tox-py27-check-uc/4f7af50/ : SUCCESS in 7m 13s\n- requirements-tox-py35-check-uc http://logs.openstack.org/99/658099/1/check/requirements-tox-py35-check-uc/f254bcf/ : SUCCESS in 7m 19s\n- requirements-tox-py36-check-uc http://logs.openstack.org/99/658099/1/check/requirements-tox-py36-check-uc/ea6e628/ : SUCCESS in 6m 55s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/99/658099/1/check/cross-cinder-py27/764e214/ : SUCCESS in 9m 41s\n- cross-cinder-py35 http://logs.openstack.org/99/658099/1/check/cross-cinder-py35/a9c545e/ : SUCCESS in 8m 27s\n- cross-glance-py27 http://logs.openstack.org/99/658099/1/check/cross-glance-py27/4f580b7/ : SUCCESS in 6m 39s\n- cross-glance-py35 http://logs.openstack.org/99/658099/1/check/cross-glance-py35/f33c946/ : SUCCESS in 6m 18s\n- cross-horizon-py27 http://logs.openstack.org/99/658099/1/check/cross-horizon-py27/7c5f3a1/ : SUCCESS in 11m 01s\n- cross-horizon-py35 http://logs.openstack.org/99/658099/1/check/cross-horizon-py35/77b0e8c/ : SUCCESS in 10m 27s\n- cross-keystone-py27 http://logs.openstack.org/99/658099/1/check/cross-keystone-py27/11ca6c7/ : SUCCESS in 11m 49s\n- cross-keystone-py35 http://logs.openstack.org/99/658099/1/check/cross-keystone-py35/16affe2/ : SUCCESS in 11m 25s\n- cross-neutron-py27 http://logs.openstack.org/99/658099/1/check/cross-neutron-py27/d957198/ : SUCCESS in 25m 22s\n- cross-neutron-py35 http://logs.openstack.org/99/658099/1/check/cross-neutron-py35/722bf28/ : SUCCESS in 29m 03s\n- cross-nova-py27 http://logs.openstack.org/99/658099/1/check/cross-nova-py27/4f3963e/ : SUCCESS in 13m 17s\n- cross-nova-py35 http://logs.openstack.org/99/658099/1/check/cross-nova-py35/81b63fc/ : SUCCESS in 13m 33s\n- cross-nova-functional http://logs.openstack.org/99/658099/1/check/cross-nova-functional/f1776c4/ : SUCCESS in 16m 34s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/99/658099/1/check/cross-osc-build-sphinx-docs/7b527a9/html/ : SUCCESS in 5m 56s\n- cross-swift-py27 http://logs.openstack.org/99/658099/1/check/cross-swift-py27/036d5b2/ : SUCCESS in 8m 43s\n- cross-swift-py35 http://logs.openstack.org/99/658099/1/check/cross-swift-py35/30f91bf/ : SUCCESS in 6m 23s\n- openstack-tox-py36 http://logs.openstack.org/99/658099/1/check/openstack-tox-py36/0fe6777/ : SUCCESS in 3m 32s\n- requirements-integration http://logs.openstack.org/99/658099/1/check/requirements-integration/f192670/ : FAILURE in 31m 58s\n- neutron-grenade-multinode http://logs.openstack.org/99/658099/1/check/neutron-grenade-multinode/d7296ef/ : SUCCESS in 1h 05m 03s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658099/1/check/legacy-tempest-dsvm-neutron-full/f7c9bb1/ : SUCCESS in 1h 21m 12s\n- neutron-grenade http://logs.openstack.org/99/658099/1/check/neutron-grenade/144ec69/ : SUCCESS in 58m 42s","accounts_in_message":[],"_revision_number":1},{"id":"98193387079a2559332d0e713ba1d80ab00eb590","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 13:52:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6c0575efdc01060abd44ee59eb89ca291c6646ed","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-12 18:54:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8072021fb9316957631eaf5e2b6edfbfba953fd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 20:22:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/99/658099/2/check/openstack-tox-pypy/626693b/ : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658099/2/check/openstack-tox-pep8/2ebd30b/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/99/658099/2/check/openstack-tox-py27/30d21f5/ : SUCCESS in 3m 18s\n- openstack-tox-py35 http://logs.openstack.org/99/658099/2/check/openstack-tox-py35/dc910da/ : SUCCESS in 3m 19s\n- openstack-tox-validate http://logs.openstack.org/99/658099/2/check/openstack-tox-validate/5c44f81/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/99/658099/2/check/requirements-tox-babel/c6b0c8d/ : SUCCESS in 2m 45s\n- requirements-tox-py27-check-uc http://logs.openstack.org/99/658099/2/check/requirements-tox-py27-check-uc/6906f18/ : SUCCESS in 7m 07s\n- requirements-tox-py35-check-uc http://logs.openstack.org/99/658099/2/check/requirements-tox-py35-check-uc/76a05ec/ : SUCCESS in 5m 54s\n- requirements-tox-py36-check-uc http://logs.openstack.org/99/658099/2/check/requirements-tox-py36-check-uc/5bf6baf/ : SUCCESS in 5m 29s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/99/658099/2/check/cross-cinder-py27/4661d4d/ : SUCCESS in 8m 00s\n- cross-cinder-py35 http://logs.openstack.org/99/658099/2/check/cross-cinder-py35/473f78c/ : SUCCESS in 8m 03s\n- cross-glance-py27 http://logs.openstack.org/99/658099/2/check/cross-glance-py27/aa940f7/ : SUCCESS in 4m 04s\n- cross-glance-py35 http://logs.openstack.org/99/658099/2/check/cross-glance-py35/931ee55/ : SUCCESS in 4m 42s\n- cross-horizon-py27 http://logs.openstack.org/99/658099/2/check/cross-horizon-py27/c3e05c7/ : SUCCESS in 9m 55s\n- cross-horizon-py35 http://logs.openstack.org/99/658099/2/check/cross-horizon-py35/23c4053/ : SUCCESS in 7m 51s\n- cross-keystone-py27 http://logs.openstack.org/99/658099/2/check/cross-keystone-py27/6f2793b/ : SUCCESS in 11m 29s\n- cross-keystone-py35 http://logs.openstack.org/99/658099/2/check/cross-keystone-py35/ad25c72/ : SUCCESS in 12m 52s\n- cross-neutron-py27 http://logs.openstack.org/99/658099/2/check/cross-neutron-py27/db1f88b/ : SUCCESS in 24m 57s\n- cross-neutron-py35 http://logs.openstack.org/99/658099/2/check/cross-neutron-py35/310d6a9/ : SUCCESS in 24m 18s\n- cross-nova-py27 http://logs.openstack.org/99/658099/2/check/cross-nova-py27/39e1a70/ : SUCCESS in 12m 20s\n- cross-nova-py35 http://logs.openstack.org/99/658099/2/check/cross-nova-py35/5a67906/ : SUCCESS in 17m 29s\n- cross-nova-functional http://logs.openstack.org/99/658099/2/check/cross-nova-functional/1e61ccf/ : SUCCESS in 17m 28s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/99/658099/2/check/cross-osc-build-sphinx-docs/8853435/html/ : SUCCESS in 6m 04s\n- cross-swift-py27 http://logs.openstack.org/99/658099/2/check/cross-swift-py27/a7bf30c/ : SUCCESS in 7m 13s\n- cross-swift-py35 http://logs.openstack.org/99/658099/2/check/cross-swift-py35/85208cc/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/99/658099/2/check/openstack-tox-py36/108352b/ : SUCCESS in 4m 16s\n- requirements-integration http://logs.openstack.org/99/658099/2/check/requirements-integration/b4e5f5c/ : FAILURE in 30m 07s\n- neutron-grenade-multinode http://logs.openstack.org/99/658099/2/check/neutron-grenade-multinode/402fb0a/ : SUCCESS in 1h 00m 52s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658099/2/check/legacy-tempest-dsvm-neutron-full/245f35d/ : SUCCESS in 1h 22m 30s\n- neutron-grenade http://logs.openstack.org/99/658099/2/check/neutron-grenade/06ae74d/ : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":2},{"id":"6910a5227b2eb7595d9df11b3b0f2bb74db5e46a","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 15:46:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"94691abad9ac7c724c6e5a825209079162a99222","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 17:44:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/99/658099/2/check/openstack-tox-pypy/8c9d8c1/ : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/99/658099/2/check/openstack-tox-pep8/559e9de/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/99/658099/2/check/openstack-tox-py27/d8299e4/ : SUCCESS in 2m 48s\n- openstack-tox-py35 http://logs.openstack.org/99/658099/2/check/openstack-tox-py35/8eb1ffe/ : SUCCESS in 3m 13s\n- openstack-tox-validate http://logs.openstack.org/99/658099/2/check/openstack-tox-validate/4a55a6a/ : SUCCESS in 3m 17s\n- requirements-tox-babel http://logs.openstack.org/99/658099/2/check/requirements-tox-babel/0dae4b0/ : SUCCESS in 2m 46s\n- requirements-tox-py27-check-uc http://logs.openstack.org/99/658099/2/check/requirements-tox-py27-check-uc/1df7244/ : SUCCESS in 7m 51s\n- requirements-tox-py35-check-uc http://logs.openstack.org/99/658099/2/check/requirements-tox-py35-check-uc/d92c2a9/ : SUCCESS in 9m 31s\n- requirements-tox-py36-check-uc http://logs.openstack.org/99/658099/2/check/requirements-tox-py36-check-uc/4560b82/ : SUCCESS in 8m 49s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/99/658099/2/check/cross-cinder-py27/faad3b7/ : SUCCESS in 9m 20s\n- cross-cinder-py35 http://logs.openstack.org/99/658099/2/check/cross-cinder-py35/497a110/ : SUCCESS in 8m 26s\n- cross-glance-py27 http://logs.openstack.org/99/658099/2/check/cross-glance-py27/135d5c1/ : SUCCESS in 4m 14s\n- cross-glance-py35 http://logs.openstack.org/99/658099/2/check/cross-glance-py35/37c15e1/ : SUCCESS in 4m 49s\n- cross-horizon-py27 http://logs.openstack.org/99/658099/2/check/cross-horizon-py27/3814309/ : SUCCESS in 10m 12s\n- cross-horizon-py35 http://logs.openstack.org/99/658099/2/check/cross-horizon-py35/ea8f0f2/ : SUCCESS in 8m 43s\n- cross-keystone-py27 http://logs.openstack.org/99/658099/2/check/cross-keystone-py27/ccd71f2/ : SUCCESS in 11m 57s\n- cross-keystone-py35 http://logs.openstack.org/99/658099/2/check/cross-keystone-py35/12c4e2b/ : SUCCESS in 12m 56s\n- cross-neutron-py27 http://logs.openstack.org/99/658099/2/check/cross-neutron-py27/328c3b8/ : SUCCESS in 23m 44s\n- cross-neutron-py35 http://logs.openstack.org/99/658099/2/check/cross-neutron-py35/c882416/ : SUCCESS in 28m 25s\n- cross-nova-py27 http://logs.openstack.org/99/658099/2/check/cross-nova-py27/525dd75/ : SUCCESS in 13m 35s\n- cross-nova-py35 http://logs.openstack.org/99/658099/2/check/cross-nova-py35/a9d5d65/ : SUCCESS in 12m 55s\n- cross-nova-functional http://logs.openstack.org/99/658099/2/check/cross-nova-functional/f94957b/ : SUCCESS in 17m 07s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/99/658099/2/check/cross-osc-build-sphinx-docs/f275881/html/ : SUCCESS in 4m 42s\n- cross-swift-py27 http://logs.openstack.org/99/658099/2/check/cross-swift-py27/9d64034/ : SUCCESS in 7m 44s\n- cross-swift-py35 http://logs.openstack.org/99/658099/2/check/cross-swift-py35/6d5fd2c/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/99/658099/2/check/openstack-tox-py36/15c1ce2/ : SUCCESS in 3m 10s\n- requirements-integration http://logs.openstack.org/99/658099/2/check/requirements-integration/012f110/ : SUCCESS in 1h 01m 13s\n- neutron-grenade-multinode http://logs.openstack.org/99/658099/2/check/neutron-grenade-multinode/093f117/ : SUCCESS in 1h 09m 07s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658099/2/check/legacy-tempest-dsvm-neutron-full/fbcca57/ : SUCCESS in 1h 41m 12s\n- neutron-grenade http://logs.openstack.org/99/658099/2/check/neutron-grenade/81cde11/ : SUCCESS in 56m 34s","accounts_in_message":[],"_revision_number":2},{"id":"fdbc2bbbd033637ffd0836957bba57b699a88179","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-13 17:59:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a3fe6f2fd45c65a0e03c60774bf58a7cc5d38186","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 17:59:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7b3bdb2374aef7a2c620b631ad9edac32b88a1d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 19:48:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"23c2b9e544cfc2fc87abd35857375c165e19165e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 19:48:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/99/658099/2/gate/openstack-tox-pep8/679a368/ : SUCCESS in 2m 27s\n- openstack-tox-py27 http://logs.openstack.org/99/658099/2/gate/openstack-tox-py27/75db532/ : SUCCESS in 4m 43s\n- openstack-tox-py35 http://logs.openstack.org/99/658099/2/gate/openstack-tox-py35/31e3942/ : SUCCESS in 3m 27s\n- openstack-tox-validate http://logs.openstack.org/99/658099/2/gate/openstack-tox-validate/9bb0d55/ : SUCCESS in 3m 30s\n- requirements-tox-babel http://logs.openstack.org/99/658099/2/gate/requirements-tox-babel/65c885b/ : SUCCESS in 2m 43s\n- requirements-tox-py27-check-uc http://logs.openstack.org/99/658099/2/gate/requirements-tox-py27-check-uc/60637d3/ : SUCCESS in 7m 45s\n- requirements-tox-py35-check-uc http://logs.openstack.org/99/658099/2/gate/requirements-tox-py35-check-uc/2381cf0/ : SUCCESS in 5m 58s\n- cross-cinder-py27 http://logs.openstack.org/99/658099/2/gate/cross-cinder-py27/5a5c7d3/ : SUCCESS in 8m 00s\n- cross-cinder-py35 http://logs.openstack.org/99/658099/2/gate/cross-cinder-py35/7bd9eb8/ : SUCCESS in 7m 45s\n- cross-glance-py27 http://logs.openstack.org/99/658099/2/gate/cross-glance-py27/b19a74a/ : SUCCESS in 4m 11s\n- cross-glance-py35 http://logs.openstack.org/99/658099/2/gate/cross-glance-py35/dda7132/ : SUCCESS in 4m 31s\n- cross-horizon-py27 http://logs.openstack.org/99/658099/2/gate/cross-horizon-py27/d492909/ : SUCCESS in 10m 07s\n- cross-horizon-py35 http://logs.openstack.org/99/658099/2/gate/cross-horizon-py35/ac1427a/ : SUCCESS in 7m 48s\n- cross-keystone-py27 http://logs.openstack.org/99/658099/2/gate/cross-keystone-py27/d4386d4/ : SUCCESS in 14m 12s\n- cross-keystone-py35 http://logs.openstack.org/99/658099/2/gate/cross-keystone-py35/c0c8417/ : SUCCESS in 21m 48s\n- cross-neutron-py27 http://logs.openstack.org/99/658099/2/gate/cross-neutron-py27/0d730a5/ : SUCCESS in 22m 55s\n- cross-neutron-py35 http://logs.openstack.org/99/658099/2/gate/cross-neutron-py35/376dcb6/ : SUCCESS in 28m 20s\n- cross-nova-py27 http://logs.openstack.org/99/658099/2/gate/cross-nova-py27/472c611/ : SUCCESS in 12m 29s\n- cross-nova-py35 http://logs.openstack.org/99/658099/2/gate/cross-nova-py35/f8f4c7d/ : SUCCESS in 13m 03s\n- cross-nova-functional http://logs.openstack.org/99/658099/2/gate/cross-nova-functional/430f314/ : SUCCESS in 18m 18s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/99/658099/2/gate/cross-osc-build-sphinx-docs/6f7a1de/html/ : SUCCESS in 4m 50s\n- cross-swift-py27 http://logs.openstack.org/99/658099/2/gate/cross-swift-py27/8d72baf/ : SUCCESS in 7m 37s\n- cross-swift-py35 http://logs.openstack.org/99/658099/2/gate/cross-swift-py35/fb36663/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/99/658099/2/gate/openstack-tox-py36/2387bcc/ : SUCCESS in 3m 32s\n- requirements-integration http://logs.openstack.org/99/658099/2/gate/requirements-integration/a978cfb/ : SUCCESS in 57m 49s\n- neutron-grenade-multinode http://logs.openstack.org/99/658099/2/gate/neutron-grenade-multinode/6b380bf/ : SUCCESS in 1h 09m 46s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/99/658099/2/gate/legacy-tempest-dsvm-neutron-full/73bc0c4/ : SUCCESS in 1h 30m 46s\n- neutron-grenade http://logs.openstack.org/99/658099/2/gate/neutron-grenade/9b4cea9/ : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"705b3a7ca82be96e2cd35ab5326bbdbf4c25d36f","revisions":{"6905e11c92f8f01693228b2c9fb3338f45fd364a":{"kind":"REWORK","_number":1,"created":"2019-05-09 14:18:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/658099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/658099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/658099/1"}}},"commit":{"parents":[{"commit":"964f79d543aae22babac1b93ffbd8694f4a64df7","subject":"update git.openstack in _DEFAULT_REQS_DIR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/964f79d543aae22babac1b93ffbd8694f4a64df7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 14:18:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 14:18:32.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.5.7","message":"update constraint for os-brick to new release 2.5.7\n\nChange-Id: I2cdf8d32a20ca57cd923cfca4aacd3500c085b7a\nmeta:version: 2.5.7\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: Idba62bc46f9199ebb27d81cac83bf13a6fe74869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6905e11c92f8f01693228b2c9fb3338f45fd364a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6905e11c92f8f01693228b2c9fb3338f45fd364a"}]},"branch":"refs/heads/stable/rocky"},"705b3a7ca82be96e2cd35ab5326bbdbf4c25d36f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-12 18:54:25.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/99/658099/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/658099/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658099/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658099/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/658099/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/658099/2"}}},"commit":{"parents":[{"commit":"ab0875127bd23cf18c549440b9e0130d5162b73d","subject":"Add upper-constraints enforcement for local reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab0875127bd23cf18c549440b9e0130d5162b73d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 14:18:32.000000000","tz":0},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-05-12 18:54:25.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.5.7","message":"update constraint for os-brick to new release 2.5.7\n\nChange-Id: I2cdf8d32a20ca57cd923cfca4aacd3500c085b7a\nmeta:version: 2.5.7\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: Idba62bc46f9199ebb27d81cac83bf13a6fe74869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/705b3a7ca82be96e2cd35ab5326bbdbf4c25d36f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/705b3a7ca82be96e2cd35ab5326bbdbf4c25d36f"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
