)]}'
{"id":"openstack%2Frequirements~658104","triplet_id":"openstack%2Frequirements~master~Ibf615dd966261e211c36aa4b0a446abfccfecf5d","project":"openstack/requirements","branch":"master","topic":"cinder_extras","hashtags":[],"change_id":"Ibf615dd966261e211c36aa4b0a446abfccfecf5d","subject":"Add cinder extras pyxcli library","status":"ABANDONED","created":"2019-05-09 14:23:10.000000000","updated":"2019-09-22 05:13:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"767a3634ae3bbc7657b583bbd2168a025d79c1ce","_number":658104,"virtual_id_number":658104,"owner":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2019-05-11 19:36:24.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-11 21:30:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-05-10 16:33:24.000000000","_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"date":"2019-05-21 20:55:07.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2019-05-11 21:46:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-11 19:36:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-11 21:30:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-11 21:46:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-21 20:55:07.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"636fe7769608ce828d897e11fdb388464bf0ac14","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-09 14:23:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adff0955f63fdab07e54d2d0a49a344a3f23c9f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 18:42:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/04/658104/1/check/openstack-tox-docs/4185999/html/ : SUCCESS in 4m 12s\n- openstack-tox-pypy http://logs.openstack.org/04/658104/1/check/openstack-tox-pypy/4e1c13e/ : FAILURE in 4m 20s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/04/658104/1/check/openstack-tox-pep8/324ce64/ : SUCCESS in 2m 43s\n- openstack-tox-py27 http://logs.openstack.org/04/658104/1/check/openstack-tox-py27/aee56e4/ : FAILURE in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/04/658104/1/check/openstack-tox-py36/1311ebb/ : SUCCESS in 5m 17s\n- openstack-tox-validate http://logs.openstack.org/04/658104/1/check/openstack-tox-validate/f9c6ea5/ : SUCCESS in 5m 35s\n- requirements-tox-babel http://logs.openstack.org/04/658104/1/check/requirements-tox-babel/d0874f5/ : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/04/658104/1/check/requirements-tox-py27-check-uc/eb5d2fc/ : SUCCESS in 16m 32s\n- requirements-tox-py36-check-uc http://logs.openstack.org/04/658104/1/check/requirements-tox-py36-check-uc/8b63181/ : SUCCESS in 19m 37s\n- cross-cinder-py27 http://logs.openstack.org/04/658104/1/check/cross-cinder-py27/179c8a4/ : SUCCESS in 12m 03s\n- cross-cinder-py36 http://logs.openstack.org/04/658104/1/check/cross-cinder-py36/f2516e4/ : SUCCESS in 7m 23s\n- cross-glance-py27 http://logs.openstack.org/04/658104/1/check/cross-glance-py27/95ef0de/ : SUCCESS in 5m 27s\n- cross-glance-py36 http://logs.openstack.org/04/658104/1/check/cross-glance-py36/bb82432/ : SUCCESS in 5m 34s\n- cross-horizon-py27 http://logs.openstack.org/04/658104/1/check/cross-horizon-py27/ddf9c38/ : SUCCESS in 9m 10s\n- cross-horizon-py36 http://logs.openstack.org/04/658104/1/check/cross-horizon-py36/45e073a/ : SUCCESS in 10m 15s\n- cross-horizon-npm http://logs.openstack.org/04/658104/1/check/cross-horizon-npm/05eba48/ : SUCCESS in 10m 31s\n- cross-keystone-py27 http://logs.openstack.org/04/658104/1/check/cross-keystone-py27/5c77639/ : SUCCESS in 23m 10s\n- cross-keystone-py36 http://logs.openstack.org/04/658104/1/check/cross-keystone-py36/c353911/ : SUCCESS in 22m 26s\n- cross-neutron-py27 http://logs.openstack.org/04/658104/1/check/cross-neutron-py27/bbd99c4/ : SUCCESS in 23m 06s\n- cross-neutron-py36 http://logs.openstack.org/04/658104/1/check/cross-neutron-py36/b87acbd/ : SUCCESS in 21m 33s\n- cross-nova-py27 http://logs.openstack.org/04/658104/1/check/cross-nova-py27/fcef5b1/ : SUCCESS in 13m 50s\n- cross-nova-py36 http://logs.openstack.org/04/658104/1/check/cross-nova-py36/852ed26/ : SUCCESS in 19m 23s\n- cross-nova-functional http://logs.openstack.org/04/658104/1/check/cross-nova-functional/25ba594/ : SUCCESS in 17m 52s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/04/658104/1/check/cross-osc-build-sphinx-docs/577b3c2/html/ : SUCCESS in 4m 46s\n- cross-swift-py27 http://logs.openstack.org/04/658104/1/check/cross-swift-py27/41416ee/ : SUCCESS in 7m 22s\n- cross-swift-py36 http://logs.openstack.org/04/658104/1/check/cross-swift-py36/4354a1c/ : SUCCESS in 9m 29s\n- requirements-integration http://logs.openstack.org/04/658104/1/check/requirements-integration/4383eca/ : FAILURE in 43m 26s\n- neutron-grenade-multinode http://logs.openstack.org/04/658104/1/check/neutron-grenade-multinode/e1bddce/ : SUCCESS in 1h 21m 12s\n- tempest-full http://logs.openstack.org/04/658104/1/check/tempest-full/bf80f68/ : SUCCESS in 1h 53m 57s\n- neutron-grenade http://logs.openstack.org/04/658104/1/check/neutron-grenade/91fef95/ : SUCCESS in 54m 35s","accounts_in_message":[],"_revision_number":1},{"id":"32a03fae9f1a5ddecbd661f67e5466985ece2e2c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 05:09:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d2f1206eaef81938c29208eadc72914c7094fa4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 07:15:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/04/658104/1/check/openstack-tox-docs/fd62cea/html/ : SUCCESS in 5m 25s\n- openstack-tox-pypy http://logs.openstack.org/04/658104/1/check/openstack-tox-pypy/c6a3b6b/ : FAILURE in 3m 20s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/04/658104/1/check/openstack-tox-pep8/9d693c2/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/04/658104/1/check/openstack-tox-py27/247ac4a/ : FAILURE in 2m 56s\n- openstack-tox-py36 http://logs.openstack.org/04/658104/1/check/openstack-tox-py36/658cadf/ : SUCCESS in 5m 38s\n- openstack-tox-validate http://logs.openstack.org/04/658104/1/check/openstack-tox-validate/8fb055a/ : SUCCESS in 3m 28s\n- requirements-tox-babel http://logs.openstack.org/04/658104/1/check/requirements-tox-babel/b7fed51/ : SUCCESS in 4m 48s\n- requirements-tox-py27-check-uc http://logs.openstack.org/04/658104/1/check/requirements-tox-py27-check-uc/54dd975/ : SUCCESS in 7m 28s\n- requirements-tox-py36-check-uc http://logs.openstack.org/04/658104/1/check/requirements-tox-py36-check-uc/c0551ff/ : SUCCESS in 7m 30s\n- cross-cinder-py27 http://logs.openstack.org/04/658104/1/check/cross-cinder-py27/3d16f3e/ : SUCCESS in 10m 10s\n- cross-cinder-py36 http://logs.openstack.org/04/658104/1/check/cross-cinder-py36/0797a09/ : SUCCESS in 9m 32s\n- cross-glance-py27 http://logs.openstack.org/04/658104/1/check/cross-glance-py27/00d5679/ : SUCCESS in 5m 27s\n- cross-glance-py36 http://logs.openstack.org/04/658104/1/check/cross-glance-py36/4d46866/ : SUCCESS in 4m 59s\n- cross-horizon-py27 http://logs.openstack.org/04/658104/1/check/cross-horizon-py27/a3326d8/ : SUCCESS in 10m 58s\n- cross-horizon-py36 http://logs.openstack.org/04/658104/1/check/cross-horizon-py36/1dbf976/ : SUCCESS in 10m 46s\n- cross-horizon-npm http://logs.openstack.org/04/658104/1/check/cross-horizon-npm/f51e4b4/ : SUCCESS in 8m 26s\n- cross-keystone-py27 http://logs.openstack.org/04/658104/1/check/cross-keystone-py27/55d7912/ : SUCCESS in 25m 53s\n- cross-keystone-py36 http://logs.openstack.org/04/658104/1/check/cross-keystone-py36/9ecc6d0/ : SUCCESS in 19m 35s\n- cross-neutron-py27 http://logs.openstack.org/04/658104/1/check/cross-neutron-py27/8152858/ : SUCCESS in 18m 14s\n- cross-neutron-py36 http://logs.openstack.org/04/658104/1/check/cross-neutron-py36/f19ce8c/ : SUCCESS in 19m 50s\n- cross-nova-py27 http://logs.openstack.org/04/658104/1/check/cross-nova-py27/c03213a/ : SUCCESS in 13m 48s\n- cross-nova-py36 http://logs.openstack.org/04/658104/1/check/cross-nova-py36/449a366/ : SUCCESS in 12m 32s\n- cross-nova-functional http://logs.openstack.org/04/658104/1/check/cross-nova-functional/9cf7691/ : SUCCESS in 17m 52s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/04/658104/1/check/cross-osc-build-sphinx-docs/7d4fdcd/html/ : SUCCESS in 6m 11s\n- cross-swift-py27 http://logs.openstack.org/04/658104/1/check/cross-swift-py27/9c66560/ : SUCCESS in 8m 44s\n- cross-swift-py36 http://logs.openstack.org/04/658104/1/check/cross-swift-py36/a1bcc8d/ : SUCCESS in 9m 16s\n- requirements-integration http://logs.openstack.org/04/658104/1/check/requirements-integration/945b861/ : FAILURE in 34m 45s\n- neutron-grenade-multinode http://logs.openstack.org/04/658104/1/check/neutron-grenade-multinode/9bed6e9/ : SUCCESS in 1h 12m 54s\n- tempest-full http://logs.openstack.org/04/658104/1/check/tempest-full/117d69c/ : SUCCESS in 1h 58m 15s\n- neutron-grenade http://logs.openstack.org/04/658104/1/check/neutron-grenade/14efabd/ : SUCCESS in 1h 02m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2ebf415d5ed7b2f4aafc428dc7bb01df5d589da2","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-10 16:33:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8314e2b692ec20b81a24841c1c9e73ed3ad561d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:16:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/04/658104/1/check/openstack-tox-docs/b3dd661/html/ : SUCCESS in 3m 25s\n- openstack-tox-pypy http://logs.openstack.org/04/658104/1/check/openstack-tox-pypy/dba7813/ : FAILURE in 2m 44s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/04/658104/1/check/openstack-tox-pep8/9bdbcf7/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/04/658104/1/check/openstack-tox-py27/be1cbc3/ : FAILURE in 2m 53s\n- openstack-tox-py36 http://logs.openstack.org/04/658104/1/check/openstack-tox-py36/294719e/ : SUCCESS in 4m 05s\n- openstack-tox-validate http://logs.openstack.org/04/658104/1/check/openstack-tox-validate/2a3030c/ : SUCCESS in 4m 26s\n- requirements-tox-babel http://logs.openstack.org/04/658104/1/check/requirements-tox-babel/35e1af5/ : SUCCESS in 8m 20s\n- requirements-tox-py27-check-uc http://logs.openstack.org/04/658104/1/check/requirements-tox-py27-check-uc/3c08716/ : SUCCESS in 6m 30s\n- requirements-tox-py36-check-uc http://logs.openstack.org/04/658104/1/check/requirements-tox-py36-check-uc/df2fff0/ : SUCCESS in 6m 29s\n- cross-cinder-py27 http://logs.openstack.org/04/658104/1/check/cross-cinder-py27/8ff8c9e/ : SUCCESS in 11m 36s\n- cross-cinder-py36 http://logs.openstack.org/04/658104/1/check/cross-cinder-py36/59e54d6/ : SUCCESS in 7m 56s\n- cross-glance-py27 http://logs.openstack.org/04/658104/1/check/cross-glance-py27/5111300/ : SUCCESS in 4m 17s\n- cross-glance-py36 http://logs.openstack.org/04/658104/1/check/cross-glance-py36/5b17707/ : SUCCESS in 3m 54s\n- cross-horizon-py27 http://logs.openstack.org/04/658104/1/check/cross-horizon-py27/7b64439/ : SUCCESS in 9m 41s\n- cross-horizon-py36 http://logs.openstack.org/04/658104/1/check/cross-horizon-py36/406c56c/ : SUCCESS in 8m 20s\n- cross-horizon-npm http://logs.openstack.org/04/658104/1/check/cross-horizon-npm/a7640f2/ : SUCCESS in 5m 54s\n- cross-keystone-py27 http://logs.openstack.org/04/658104/1/check/cross-keystone-py27/fd7fe0f/ : SUCCESS in 20m 24s\n- cross-keystone-py36 http://logs.openstack.org/04/658104/1/check/cross-keystone-py36/04a36c5/ : SUCCESS in 27m 34s\n- cross-neutron-py27 http://logs.openstack.org/04/658104/1/check/cross-neutron-py27/410531f/ : SUCCESS in 22m 11s\n- cross-neutron-py36 http://logs.openstack.org/04/658104/1/check/cross-neutron-py36/48f5638/ : SUCCESS in 24m 17s\n- cross-nova-py27 http://logs.openstack.org/04/658104/1/check/cross-nova-py27/34883ef/ : SUCCESS in 13m 31s\n- cross-nova-py36 http://logs.openstack.org/04/658104/1/check/cross-nova-py36/2c2c821/ : SUCCESS in 11m 12s\n- cross-nova-functional http://logs.openstack.org/04/658104/1/check/cross-nova-functional/f37e5e4/ : SUCCESS in 18m 20s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/04/658104/1/check/cross-osc-build-sphinx-docs/91f71ce/html/ : SUCCESS in 5m 30s\n- cross-swift-py27 http://logs.openstack.org/04/658104/1/check/cross-swift-py27/f96016a/ : SUCCESS in 6m 42s\n- cross-swift-py36 http://logs.openstack.org/04/658104/1/check/cross-swift-py36/8bc0141/ : SUCCESS in 7m 48s\n- requirements-integration http://logs.openstack.org/04/658104/1/check/requirements-integration/dae5444/ : FAILURE in 32m 35s\n- neutron-grenade-multinode http://logs.openstack.org/04/658104/1/check/neutron-grenade-multinode/e74140c/ : SUCCESS in 1h 05m 29s\n- tempest-full http://logs.openstack.org/04/658104/1/check/tempest-full/00b9b08/ : SUCCESS in 1h 52m 07s\n- neutron-grenade http://logs.openstack.org/04/658104/1/check/neutron-grenade/c6c29df/ : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e82b4d2bb7cb5668a1f0d2bff955bb9433bac641","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-11 19:36:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c8470283c213cd3c1c292b57251e1f2ecdfff6f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 21:30:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/04/658104/1/check/openstack-tox-docs/9140cd8/html/ : SUCCESS in 3m 40s\n- openstack-tox-pypy http://logs.openstack.org/04/658104/1/check/openstack-tox-pypy/0d76b2f/ : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/04/658104/1/check/openstack-tox-pep8/487682d/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/04/658104/1/check/openstack-tox-py27/712752c/ : SUCCESS in 2m 49s\n- openstack-tox-py36 http://logs.openstack.org/04/658104/1/check/openstack-tox-py36/f199405/ : SUCCESS in 3m 57s\n- openstack-tox-validate http://logs.openstack.org/04/658104/1/check/openstack-tox-validate/11160f4/ : SUCCESS in 3m 24s\n- requirements-tox-babel http://logs.openstack.org/04/658104/1/check/requirements-tox-babel/79307ff/ : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc http://logs.openstack.org/04/658104/1/check/requirements-tox-py27-check-uc/e1bce2f/ : SUCCESS in 6m 18s\n- requirements-tox-py36-check-uc http://logs.openstack.org/04/658104/1/check/requirements-tox-py36-check-uc/2acb2a2/ : SUCCESS in 7m 05s\n- cross-cinder-py27 http://logs.openstack.org/04/658104/1/check/cross-cinder-py27/fba1bb8/ : SUCCESS in 9m 43s\n- cross-cinder-py36 http://logs.openstack.org/04/658104/1/check/cross-cinder-py36/3062280/ : SUCCESS in 8m 26s\n- cross-glance-py27 http://logs.openstack.org/04/658104/1/check/cross-glance-py27/c1bfc40/ : SUCCESS in 3m 58s\n- cross-glance-py36 http://logs.openstack.org/04/658104/1/check/cross-glance-py36/ffdf5a3/ : SUCCESS in 4m 35s\n- cross-horizon-py27 http://logs.openstack.org/04/658104/1/check/cross-horizon-py27/5e69145/ : SUCCESS in 10m 42s\n- cross-horizon-py36 http://logs.openstack.org/04/658104/1/check/cross-horizon-py36/a343872/ : SUCCESS in 7m 37s\n- cross-horizon-npm http://logs.openstack.org/04/658104/1/check/cross-horizon-npm/c56a7f7/ : SUCCESS in 5m 28s\n- cross-keystone-py27 http://logs.openstack.org/04/658104/1/check/cross-keystone-py27/a1029e5/ : SUCCESS in 21m 57s\n- cross-keystone-py36 http://logs.openstack.org/04/658104/1/check/cross-keystone-py36/310705e/ : SUCCESS in 21m 57s\n- cross-neutron-py27 http://logs.openstack.org/04/658104/1/check/cross-neutron-py27/70daabe/ : SUCCESS in 22m 12s\n- cross-neutron-py36 http://logs.openstack.org/04/658104/1/check/cross-neutron-py36/f2b11f8/ : SUCCESS in 20m 22s\n- cross-nova-py27 http://logs.openstack.org/04/658104/1/check/cross-nova-py27/dd5951c/ : SUCCESS in 14m 51s\n- cross-nova-py36 http://logs.openstack.org/04/658104/1/check/cross-nova-py36/519bab2/ : SUCCESS in 11m 23s\n- cross-nova-functional http://logs.openstack.org/04/658104/1/check/cross-nova-functional/82e1d2c/ : SUCCESS in 16m 24s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/04/658104/1/check/cross-osc-build-sphinx-docs/f8408dd/html/ : SUCCESS in 5m 54s\n- cross-swift-py27 http://logs.openstack.org/04/658104/1/check/cross-swift-py27/99fc77e/ : SUCCESS in 8m 16s\n- cross-swift-py36 http://logs.openstack.org/04/658104/1/check/cross-swift-py36/f6c6081/ : SUCCESS in 8m 07s\n- requirements-integration http://logs.openstack.org/04/658104/1/check/requirements-integration/27313e9/ : SUCCESS in 50m 49s\n- neutron-grenade-multinode http://logs.openstack.org/04/658104/1/check/neutron-grenade-multinode/1e0b94d/ : SUCCESS in 1h 06m 43s\n- tempest-full http://logs.openstack.org/04/658104/1/check/tempest-full/9d601f9/ : SUCCESS in 1h 41m 44s\n- neutron-grenade http://logs.openstack.org/04/658104/1/check/neutron-grenade/4ddcff1/ : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2c0f26c86ec375fb15f2595637e9bbfee2812a58","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-11 21:46:49.000000000","message":"Patch Set 1: Code-Review-1\n\nno py3","accounts_in_message":[],"_revision_number":1},{"id":"6afaaa51e6d739110c13f0d9d121b46249fcb9e5","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-05-21 20:55:07.000000000","message":"Patch Set 1:\n\nJust as an FYI I have heard back from IBM that they are planning to get this updated for Py3 support.","accounts_in_message":[],"_revision_number":1},{"id":"767e2fa7942a435436ecfe56e03749e7e6315b95","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-21 20:57:05.000000000","message":"Patch Set 1:\n\ngood to hear","accounts_in_message":[],"_revision_number":1},{"id":"fe62b644a27a43ee3b9d97d3088149a7af446b83","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-16 20:51:54.000000000","message":"Patch Set 1:\n\njust a ping to see if there\u0027s been progress upstream for these (so they are not forgotten)","accounts_in_message":[],"_revision_number":1},{"id":"82e6fc984d45ed7d3f55665174852694b63a85da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-09-22 05:13:15.000000000","message":"Abandoned\n\nabandoning as it\u0027s been ~4-5 months","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b7284233cdc17f31aaf5eaac37458a1f7e040fc","revisions":{"3b7284233cdc17f31aaf5eaac37458a1f7e040fc":{"kind":"REWORK","_number":1,"created":"2019-05-09 14:23:10.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/04/658104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/658104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/658104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/658104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/658104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/658104/1"}}},"commit":{"parents":[{"commit":"842ed7bbb06c09dc01785265b589c228b5f928af","subject":"Merge \"Add missing format arg for missing g-r reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/842ed7bbb06c09dc01785265b589c228b5f928af"}]}],"author":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-09 14:22:27.000000000","tz":0},"committer":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-09 14:22:27.000000000","tz":0},"subject":"Add cinder extras pyxcli library","message":"Add cinder extras pyxcli library\n\nThis patch adds the cinder IBM XIV driver\u0027s required library.\n\nactive: seems so\ngood code: seems ok\npy3: NO\nlicense compat: yes\npackaged: eh?\ncovered by other libs: no\nneeded by: cinder\nmanaged by openstack: no\n\nChange-Id: Ibf615dd966261e211c36aa4b0a446abfccfecf5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b7284233cdc17f31aaf5eaac37458a1f7e040fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b7284233cdc17f31aaf5eaac37458a1f7e040fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
