)]}'
{"id":"openstack%2Frequirements~658121","triplet_id":"openstack%2Frequirements~stable%2Fpike~Ib5865e1c36e90e3a6c4d7f4044c86ea878f7ca5e","project":"openstack/requirements","branch":"stable/pike","topic":"opendev-path","hashtags":[],"change_id":"Ib5865e1c36e90e3a6c4d7f4044c86ea878f7ca5e","subject":"Use the new opendev path for requirements-check","status":"MERGED","created":"2019-05-09 14:33:00.000000000","updated":"2019-05-14 02:36:15.000000000","submitted":"2019-05-14 02:36:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658121-1557801375497-96febb6b","meta_rev_id":"c5cb8097e6ccc8fdb8b1da0b3565ca8cba36e47d","_number":658121,"virtual_id_number":658121,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-05-10 05:09:15.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-05-13 20:38:23.000000000","_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-05-14 02:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-09 14:37:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-05-09 14:42:24.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-05-09 14:42:24.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 14:37:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-09 14:42:24.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-05-10 05:09:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-13 20:38:23.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-14 02:36:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f94684d9f4daf260a837316aa36e5b99d4bc038","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-09 14:33:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/queens.","accounts_in_message":[],"_revision_number":1},{"id":"429f17228ca4edae48fd4105d2a34c1de7e5e00f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-09 14:33:12.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ocata as commit 90445dc8d8f0e3f3292ef0181fce91db0210d4e1","accounts_in_message":[],"_revision_number":1},{"id":"4547b28e6df952b1f67937d98d9f6b322c78408c","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-09 14:33:36.000000000","message":"Topic changed from opendev-path-stable/pike to opendev-path","accounts_in_message":[],"_revision_number":1},{"id":"4987e5e494722e7485144ecd8f8d6c4111856d62","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-09 14:37:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b041c4ac181974d1850490fe5719eb4246855dbb","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-05-09 14:42:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"475ec935725ae2fc55526b64bb9289fd60ee4193","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 19:39:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/21/658121/1/check/openstack-tox-pypy/23af4f6/ : FAILURE in 2m 59s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/21/658121/1/check/openstack-tox-pep8/00183c7/ : FAILURE in 2m 32s\n- openstack-tox-py27 http://logs.openstack.org/21/658121/1/check/openstack-tox-py27/13c0231/ : FAILURE in 3m 03s\n- openstack-tox-py35 http://logs.openstack.org/21/658121/1/check/openstack-tox-py35/03af740/ : SUCCESS in 5m 19s\n- openstack-tox-validate http://logs.openstack.org/21/658121/1/check/openstack-tox-validate/1cc8c38/ : FAILURE in 3m 14s\n- requirements-integration http://logs.openstack.org/21/658121/1/check/requirements-integration/0530d88/ : FAILURE in 32m 28s\n- neutron-grenade-multinode http://logs.openstack.org/21/658121/1/check/neutron-grenade-multinode/5f57f65/ : SUCCESS in 59m 34s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/21/658121/1/check/legacy-tempest-dsvm-neutron-full/ee0cf3d/ : SUCCESS in 1h 28m 29s\n- neutron-grenade http://logs.openstack.org/21/658121/1/check/neutron-grenade/f182202/ : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ceb378d9eee1a19a3fbdcb301aa01afbc49c69e9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 05:09:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c1338b06b75b442b71c2f825f5c2f64cdd0838e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 06:44:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/21/658121/1/check/openstack-tox-pypy/81f09e0/ : FAILURE in 4m 39s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/21/658121/1/check/openstack-tox-pep8/23c7726/ : FAILURE in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/21/658121/1/check/openstack-tox-py27/d36fc21/ : FAILURE in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/21/658121/1/check/openstack-tox-py35/3f6fc55/ : SUCCESS in 4m 21s\n- openstack-tox-validate http://logs.openstack.org/21/658121/1/check/openstack-tox-validate/6681eb6/ : FAILURE in 3m 03s\n- requirements-integration http://logs.openstack.org/21/658121/1/check/requirements-integration/f93b820/ : FAILURE in 31m 20s\n- neutron-grenade-multinode http://logs.openstack.org/21/658121/1/check/neutron-grenade-multinode/f71c2b6/ : SUCCESS in 1h 07m 17s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/21/658121/1/check/legacy-tempest-dsvm-neutron-full/fcfa8c2/ : SUCCESS in 1h 30m 08s\n- neutron-grenade http://logs.openstack.org/21/658121/1/check/neutron-grenade/11908db/ : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":1},{"id":"03ec571ca0056138704e253b43e4084adc2cf96e","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 20:38:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f34b80bdf68b6f5583ffc78c56e193bff1a24490","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 00:27:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/21/658121/1/check/openstack-tox-pypy/02c4cd5/ : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/21/658121/1/check/openstack-tox-pep8/79b9c68/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/21/658121/1/check/openstack-tox-py27/5e54cfc/ : SUCCESS in 5m 33s\n- openstack-tox-py35 http://logs.openstack.org/21/658121/1/check/openstack-tox-py35/b9dae63/ : SUCCESS in 3m 36s\n- openstack-tox-validate http://logs.openstack.org/21/658121/1/check/openstack-tox-validate/3442b0a/ : SUCCESS in 3m 26s\n- requirements-integration http://logs.openstack.org/21/658121/1/check/requirements-integration/76f861d/ : SUCCESS in 51m 23s\n- neutron-grenade-multinode http://logs.openstack.org/21/658121/1/check/neutron-grenade-multinode/845ae7e/ : SUCCESS in 1h 00m 06s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/21/658121/1/check/legacy-tempest-dsvm-neutron-full/116d420/ : SUCCESS in 1h 39m 49s\n- neutron-grenade http://logs.openstack.org/21/658121/1/check/neutron-grenade/b52c4ac/ : SUCCESS in 52m 24s","accounts_in_message":[],"_revision_number":1},{"id":"7382a2a02d6e4388b24f4661ff7a9e6344822848","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 00:28:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e752037efa8bd72029d600cfb482bbe265eee53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 02:36:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/21/658121/1/gate/openstack-tox-pep8/a954a19/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/21/658121/1/gate/openstack-tox-py27/9bf45f2/ : SUCCESS in 4m 09s\n- openstack-tox-py35 http://logs.openstack.org/21/658121/1/gate/openstack-tox-py35/a4b1a3b/ : SUCCESS in 3m 33s\n- openstack-tox-validate http://logs.openstack.org/21/658121/1/gate/openstack-tox-validate/77fe17c/ : SUCCESS in 3m 22s\n- requirements-integration http://logs.openstack.org/21/658121/1/gate/requirements-integration/38ddd78/ : SUCCESS in 54m 34s\n- neutron-grenade-multinode http://logs.openstack.org/21/658121/1/gate/neutron-grenade-multinode/bad3b68/ : SUCCESS in 47m 23s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/21/658121/1/gate/legacy-tempest-dsvm-neutron-full/8c3fdb1/ : SUCCESS in 1h 49m 07s\n- neutron-grenade http://logs.openstack.org/21/658121/1/gate/neutron-grenade/223ff8d/ : SUCCESS in 51m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c5cb8097e6ccc8fdb8b1da0b3565ca8cba36e47d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 02:36:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9bebcdddb29416567317bcac96d2246cb09b2e8","revisions":{"e9bebcdddb29416567317bcac96d2246cb09b2e8":{"kind":"REWORK","_number":1,"created":"2019-05-09 14:33:00.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/21/658121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/21/658121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/658121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/658121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/658121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/21/658121/1"}}},"commit":{"parents":[{"commit":"3f49011b4145023021627b2738c328c09e4ec1fc","subject":"Merge \"update constraint for ironic-lib to new release 2.10.2\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f49011b4145023021627b2738c328c09e4ec1fc"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-09 14:31:54.000000000","tz":-240},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-09 14:33:00.000000000","tz":0},"subject":"Use the new opendev path for requirements-check","message":"Use the new opendev path for requirements-check\n\nChange-Id: Ib5865e1c36e90e3a6c4d7f4044c86ea878f7ca5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9bebcdddb29416567317bcac96d2246cb09b2e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9bebcdddb29416567317bcac96d2246cb09b2e8"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
