)]}'
{"id":"openstack%2Frequirements~658122","triplet_id":"openstack%2Frequirements~stable%2Focata~Ib5865e1c36e90e3a6c4d7f4044c86ea878f7ca5e","project":"openstack/requirements","branch":"stable/ocata","topic":"opendev-path","hashtags":[],"change_id":"Ib5865e1c36e90e3a6c4d7f4044c86ea878f7ca5e","subject":"Use the new opendev path for requirements-check","status":"MERGED","created":"2019-05-09 14:33:12.000000000","updated":"2019-05-24 07:10:50.000000000","submitted":"2019-05-24 07:10:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658122-1558681850681-fdcadcda","meta_rev_id":"80ea5ceed19457be958d65feb82e7599d4580a74","_number":658122,"virtual_id_number":658122,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-05-24 07:10:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-24 00:10:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-05-24 02:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-05-24 02:07:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 14:37:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-13 20:38:35.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-24 02:07:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-05-24 07:10:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c70a3af7709e14707fba31f62e0e3ed2a95a25c","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-09 14:33:12.000000000","message":"Patch Set 1: Cherry Picked from branch stable/pike.","accounts_in_message":[],"_revision_number":1},{"id":"b1163a811f2159eea7619f9e2fd116ca8db72c8d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-09 14:33:47.000000000","message":"Topic changed from opendev-path-stable/pike-stable/ocata to opendev-path","accounts_in_message":[],"_revision_number":1},{"id":"275f02fe35303d76b131f59f91665556fa831710","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-09 14:37:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"064ad68bfe03c39bd14128d2330878f1c5aab253","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 19:49:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/22/658122/1/check/openstack-tox-pypy/9b2c82f/ : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/22/658122/1/check/openstack-tox-pep8/fb0c0ab/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/22/658122/1/check/openstack-tox-py27/7ff54e2/ : SUCCESS in 6m 57s\n- openstack-tox-py35 http://logs.openstack.org/22/658122/1/check/openstack-tox-py35/1107da6/ : SUCCESS in 3m 39s\n- openstack-tox-validate http://logs.openstack.org/22/658122/1/check/openstack-tox-validate/c989c83/ : SUCCESS in 2m 59s\n- requirements-integration http://logs.openstack.org/22/658122/1/check/requirements-integration/d605dbe/ : FAILURE in 40m 08s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/22/658122/1/check/legacy-tempest-dsvm-neutron-full/ff59a04/ : SUCCESS in 1h 35m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ba5e51d7c77314ef3e2230fd944490aff0add6cf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-10 05:09:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0ec6c9f50f5ed7ed3e7ba0ef73a2c7841e193550","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 06:39:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/22/658122/1/check/openstack-tox-pypy/32d2fe6/ : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/22/658122/1/check/openstack-tox-pep8/61f3f94/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/22/658122/1/check/openstack-tox-py27/66d79bb/ : SUCCESS in 6m 02s\n- openstack-tox-py35 http://logs.openstack.org/22/658122/1/check/openstack-tox-py35/2e00c7b/ : SUCCESS in 5m 47s\n- openstack-tox-validate http://logs.openstack.org/22/658122/1/check/openstack-tox-validate/d2ab9a3/ : SUCCESS in 6m 34s\n- requirements-integration http://logs.openstack.org/22/658122/1/check/requirements-integration/784c3f4/ : FAILURE in 28m 18s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/22/658122/1/check/legacy-tempest-dsvm-neutron-full/76642e3/ : SUCCESS in 1h 28m 28s","accounts_in_message":[],"_revision_number":1},{"id":"dce466fcb54dcebcf3744ad6ba4550b44b5fb9e3","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-13 20:38:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2bb15a9efeb4acc4a3b070b0fa955d422926e695","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 00:18:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/22/658122/1/check/openstack-tox-pypy/c650d2c/ : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/22/658122/1/check/openstack-tox-pep8/f6debb0/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/22/658122/1/check/openstack-tox-py27/68b0827/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/22/658122/1/check/openstack-tox-py35/44e5bdf/ : SUCCESS in 3m 45s\n- openstack-tox-validate http://logs.openstack.org/22/658122/1/check/openstack-tox-validate/75c4fa1/ : SUCCESS in 4m 14s\n- requirements-integration http://logs.openstack.org/22/658122/1/check/requirements-integration/9a4ac02/ : FAILURE in 34m 57s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/22/658122/1/check/legacy-tempest-dsvm-neutron-full/e747538/ : SUCCESS in 1h 29m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e1e9864d2aa32ff3ed6f53724ff35fbedd0afe0d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-15 04:29:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc12a6c3065ee6c0e2c193d4565043c9ca47c32e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 06:07:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/22/658122/1/check/openstack-tox-pypy/248f3b9/ : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/22/658122/1/check/openstack-tox-pep8/f583277/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/22/658122/1/check/openstack-tox-py27/5def66b/ : SUCCESS in 3m 25s\n- openstack-tox-py35 http://logs.openstack.org/22/658122/1/check/openstack-tox-py35/c9f4f29/ : SUCCESS in 3m 17s\n- openstack-tox-validate http://logs.openstack.org/22/658122/1/check/openstack-tox-validate/a1ebdea/ : SUCCESS in 3m 10s\n- requirements-integration http://logs.openstack.org/22/658122/1/check/requirements-integration/44201ea/ : FAILURE in 21m 28s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/22/658122/1/check/legacy-tempest-dsvm-neutron-full/4abaa0c/ : SUCCESS in 1h 33m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4453dbdb04d5a342b1910e36ac0e908d8745cc1f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-23 06:18:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1496a37bd130f2bf555e78c2709ca76955ba7062","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 09:41:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/22/658122/2/check/openstack-tox-pypy/691c0db/ : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/22/658122/2/check/openstack-tox-pep8/ab7ae65/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/22/658122/2/check/openstack-tox-py27/f2e55fb/ : SUCCESS in 3m 47s\n- openstack-tox-py35 http://logs.openstack.org/22/658122/2/check/openstack-tox-py35/f1022c1/ : SUCCESS in 4m 44s\n- openstack-tox-validate http://logs.openstack.org/22/658122/2/check/openstack-tox-validate/122bee2/ : SUCCESS in 3m 38s\n- requirements-integration http://logs.openstack.org/22/658122/2/check/requirements-integration/efc4d6b/ : FAILURE in 28m 37s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/22/658122/2/check/legacy-tempest-dsvm-neutron-full/c2b0ab6/ : SUCCESS in 1h 23m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e6349891882412fb0cb3d8dc50d4e4c00c182cc5","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-05-24 00:10:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e26b880d8ff1c6b76fa3c7b579985fa3f15e8346","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 01:33:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/22/658122/3/check/openstack-tox-pypy/8510c1e/ : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/22/658122/3/check/openstack-tox-pep8/1c08177/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/22/658122/3/check/openstack-tox-py27/df56d46/ : SUCCESS in 3m 54s\n- openstack-tox-py35 http://logs.openstack.org/22/658122/3/check/openstack-tox-py35/72e7491/ : SUCCESS in 3m 26s\n- openstack-tox-validate http://logs.openstack.org/22/658122/3/check/openstack-tox-validate/8661f74/ : SUCCESS in 3m 22s\n- requirements-integration http://logs.openstack.org/22/658122/3/check/requirements-integration/020c849/ : SUCCESS in 46m 20s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/22/658122/3/check/legacy-tempest-dsvm-neutron-full/60d6e3f/ : SUCCESS in 1h 21m 36s","accounts_in_message":[],"_revision_number":3},{"id":"4b8c78b725afa85edcfd42317d067dcfb7416437","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-05-24 02:07:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"734da5c885025faa6e2d07e235ec739ed9f0b1c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 02:20:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"10cc522b1c99679060b6edfc5b39102f6b9a6fe6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 07:10:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"80ea5ceed19457be958d65feb82e7599d4580a74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 07:10:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/658122/3/gate/openstack-tox-pep8/0646aaa/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/22/658122/3/gate/openstack-tox-py27/89010eb/ : SUCCESS in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/22/658122/3/gate/openstack-tox-py35/655dc44/ : SUCCESS in 3m 59s\n- openstack-tox-validate http://logs.openstack.org/22/658122/3/gate/openstack-tox-validate/5b66e2b/ : SUCCESS in 3m 32s\n- requirements-integration http://logs.openstack.org/22/658122/3/gate/requirements-integration/68f6199/ : SUCCESS in 48m 28s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/22/658122/3/gate/legacy-tempest-dsvm-neutron-full/712bc28/ : SUCCESS in 1h 24m 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8ba5b2ef8f2aa2acdec017f05a24c22892c9a6d3","revisions":{"90445dc8d8f0e3f3292ef0181fce91db0210d4e1":{"kind":"REWORK","_number":1,"created":"2019-05-09 14:33:12.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/658122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/658122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/658122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/658122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/658122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/658122/1"}}},"commit":{"parents":[{"commit":"45b6c90d6a6cf66dbaa2b88437f323cd05ed68e1","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45b6c90d6a6cf66dbaa2b88437f323cd05ed68e1"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-09 14:31:54.000000000","tz":-240},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-09 14:33:12.000000000","tz":0},"subject":"Use the new opendev path for requirements-check","message":"Use the new opendev path for requirements-check\n\nChange-Id: Ib5865e1c36e90e3a6c4d7f4044c86ea878f7ca5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90445dc8d8f0e3f3292ef0181fce91db0210d4e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90445dc8d8f0e3f3292ef0181fce91db0210d4e1"}]},"branch":"refs/heads/stable/ocata"},"992e13c23613174c45fba9e9429381d4c72a1d6f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-23 06:18:29.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/658122/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/658122/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/658122/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/658122/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/658122/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/658122/2"}}},"commit":{"parents":[{"commit":"2484a9266c9f434af04654020fbab64781f5ed74","subject":"Cap keyring and scikit-learn on python2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2484a9266c9f434af04654020fbab64781f5ed74"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-09 14:31:54.000000000","tz":-240},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-23 06:18:29.000000000","tz":0},"subject":"Use the new opendev path for requirements-check","message":"Use the new opendev path for requirements-check\n\nChange-Id: Ib5865e1c36e90e3a6c4d7f4044c86ea878f7ca5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/992e13c23613174c45fba9e9429381d4c72a1d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/992e13c23613174c45fba9e9429381d4c72a1d6f"}]},"branch":"refs/heads/stable/ocata"},"8ba5b2ef8f2aa2acdec017f05a24c22892c9a6d3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-24 00:10:48.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/22/658122/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/658122/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/658122/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/658122/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/658122/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/658122/3"}}},"commit":{"parents":[{"commit":"d0e1c0a6ffacb50cb00d1bdb62fd66d1b4ee5fdf","subject":"Cap keyring and scikit-learn on python2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0e1c0a6ffacb50cb00d1bdb62fd66d1b4ee5fdf"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-09 14:31:54.000000000","tz":-240},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2019-05-24 00:10:48.000000000","tz":0},"subject":"Use the new opendev path for requirements-check","message":"Use the new opendev path for requirements-check\n\nChange-Id: Ib5865e1c36e90e3a6c4d7f4044c86ea878f7ca5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ba5b2ef8f2aa2acdec017f05a24c22892c9a6d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ba5b2ef8f2aa2acdec017f05a24c22892c9a6d3"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
