)]}'
{"id":"openstack%2Frequirements~658138","triplet_id":"openstack%2Frequirements~master~Ida13244fc71b6057fdd083e3640033dc74b2f4f5","project":"openstack/requirements","branch":"master","topic":"cap-grpcio","hashtags":[],"change_id":"Ida13244fc71b6057fdd083e3640033dc74b2f4f5","subject":"Bump grpcio upper-constraint","status":"ABANDONED","created":"2019-05-09 15:37:38.000000000","updated":"2019-06-16 20:49:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1e03419b7b466c095d4b2b871698705e25f3688b","_number":658138,"virtual_id_number":658138,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-06-03 21:08:15.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2019-05-29 08:13:55.000000000","_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":-1,"date":"2019-05-09 21:17:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 21:17:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-29 08:13:55.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-06-03 21:08:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"e2653aae12b3d9f79941a3171ef027c80f21fe15","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-09 15:37:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53f6c579434ec5e7b9439633d5b2678c3159a2df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 21:17:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/38/658138/1/check/openstack-tox-docs/2ddb629/html/ : SUCCESS in 3m 10s\n- openstack-tox-pypy http://logs.openstack.org/38/658138/1/check/openstack-tox-pypy/728aa64/ : FAILURE in 2m 40s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/38/658138/1/check/openstack-tox-pep8/df62f77/ : SUCCESS in 2m 50s\n- openstack-tox-py27 http://logs.openstack.org/38/658138/1/check/openstack-tox-py27/4becc60/ : FAILURE in 2m 33s\n- openstack-tox-py36 http://logs.openstack.org/38/658138/1/check/openstack-tox-py36/a465c78/ : SUCCESS in 3m 20s\n- openstack-tox-validate http://logs.openstack.org/38/658138/1/check/openstack-tox-validate/a0b4cdf/ : SUCCESS in 3m 04s\n- requirements-tox-babel http://logs.openstack.org/38/658138/1/check/requirements-tox-babel/28f623d/ : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/38/658138/1/check/requirements-tox-py27-check-uc/92ca81d/ : FAILURE in 6m 16s\n- requirements-tox-py36-check-uc http://logs.openstack.org/38/658138/1/check/requirements-tox-py36-check-uc/f7340e0/ : FAILURE in 7m 14s\n- cross-cinder-py27 http://logs.openstack.org/38/658138/1/check/cross-cinder-py27/652b1c6/ : SUCCESS in 9m 01s\n- cross-cinder-py36 http://logs.openstack.org/38/658138/1/check/cross-cinder-py36/227a37e/ : SUCCESS in 8m 33s\n- cross-glance-py27 http://logs.openstack.org/38/658138/1/check/cross-glance-py27/68d0bf7/ : SUCCESS in 4m 54s\n- cross-glance-py36 http://logs.openstack.org/38/658138/1/check/cross-glance-py36/541389b/ : SUCCESS in 4m 09s\n- cross-horizon-py27 http://logs.openstack.org/38/658138/1/check/cross-horizon-py27/970ad3e/ : SUCCESS in 8m 37s\n- cross-horizon-py36 http://logs.openstack.org/38/658138/1/check/cross-horizon-py36/f87ba78/ : SUCCESS in 8m 32s\n- cross-horizon-npm http://logs.openstack.org/38/658138/1/check/cross-horizon-npm/784edb4/ : SUCCESS in 5m 42s\n- cross-keystone-py27 http://logs.openstack.org/38/658138/1/check/cross-keystone-py27/01b144b/ : SUCCESS in 20m 57s\n- cross-keystone-py36 http://logs.openstack.org/38/658138/1/check/cross-keystone-py36/21c3a4c/ : SUCCESS in 22m 39s\n- cross-neutron-py27 http://logs.openstack.org/38/658138/1/check/cross-neutron-py27/39f8fee/ : SUCCESS in 25m 47s\n- cross-neutron-py36 http://logs.openstack.org/38/658138/1/check/cross-neutron-py36/39c32af/ : SUCCESS in 22m 51s\n- cross-nova-py27 http://logs.openstack.org/38/658138/1/check/cross-nova-py27/4fc31df/ : SUCCESS in 12m 12s\n- cross-nova-py36 http://logs.openstack.org/38/658138/1/check/cross-nova-py36/c3a3355/ : SUCCESS in 11m 55s\n- cross-nova-functional http://logs.openstack.org/38/658138/1/check/cross-nova-functional/14ccdaf/ : SUCCESS in 18m 08s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/38/658138/1/check/cross-osc-build-sphinx-docs/2f22fa2/html/ : SUCCESS in 6m 23s\n- cross-swift-py27 http://logs.openstack.org/38/658138/1/check/cross-swift-py27/eb0e5c8/ : SUCCESS in 10m 03s\n- cross-swift-py36 http://logs.openstack.org/38/658138/1/check/cross-swift-py36/1e77786/ : SUCCESS in 8m 05s\n- requirements-integration http://logs.openstack.org/38/658138/1/check/requirements-integration/07f35bb/ : FAILURE in 37m 54s\n- neutron-grenade-multinode http://logs.openstack.org/38/658138/1/check/neutron-grenade-multinode/c25e0bf/ : SUCCESS in 1h 14m 37s\n- tempest-full http://logs.openstack.org/38/658138/1/check/tempest-full/dc96e8d/ : SUCCESS in 2h 02m 47s\n- neutron-grenade http://logs.openstack.org/38/658138/1/check/neutron-grenade/6858a0b/ : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d408b31bbef26a4a29561ca70edeb11af1331147","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-05-29 08:13:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b63f9516f4b2a0d0ec9fede4ee05e9c151af468","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 10:33:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/38/658138/1/check/openstack-tox-docs/1ec420f/html/ : SUCCESS in 3m 06s\n- openstack-tox-pypy http://logs.openstack.org/38/658138/1/check/openstack-tox-pypy/45ee103/ : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/38/658138/1/check/openstack-tox-pep8/e7b24cc/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/38/658138/1/check/openstack-tox-py27/60e1d23/ : SUCCESS in 3m 40s\n- openstack-tox-py36 http://logs.openstack.org/38/658138/1/check/openstack-tox-py36/588ec86/ : SUCCESS in 3m 16s\n- openstack-tox-validate http://logs.openstack.org/38/658138/1/check/openstack-tox-validate/78b04c5/ : SUCCESS in 2m 57s\n- requirements-tox-babel http://logs.openstack.org/38/658138/1/check/requirements-tox-babel/17304cb/ : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc http://logs.openstack.org/38/658138/1/check/requirements-tox-py27-check-uc/5862d63/ : FAILURE in 6m 32s\n- requirements-tox-py36-check-uc http://logs.openstack.org/38/658138/1/check/requirements-tox-py36-check-uc/5b5dad8/ : FAILURE in 10m 02s\n- cross-cinder-py27 http://logs.openstack.org/38/658138/1/check/cross-cinder-py27/0b21199/ : FAILURE in 9m 49s\n- cross-cinder-py36 http://logs.openstack.org/38/658138/1/check/cross-cinder-py36/32116eb/ : SUCCESS in 9m 31s\n- cross-glance-py27 http://logs.openstack.org/38/658138/1/check/cross-glance-py27/f945aae/ : SUCCESS in 4m 30s\n- cross-glance-py36 http://logs.openstack.org/38/658138/1/check/cross-glance-py36/11f6334/ : SUCCESS in 4m 14s\n- cross-horizon-py27 http://logs.openstack.org/38/658138/1/check/cross-horizon-py27/265a031/ : SUCCESS in 10m 21s\n- cross-horizon-py36 http://logs.openstack.org/38/658138/1/check/cross-horizon-py36/1237921/ : SUCCESS in 7m 26s\n- cross-horizon-npm http://logs.openstack.org/38/658138/1/check/cross-horizon-npm/95e61be/ : SUCCESS in 6m 12s\n- cross-keystone-py27 http://logs.openstack.org/38/658138/1/check/cross-keystone-py27/95124ff/ : SUCCESS in 25m 23s\n- cross-keystone-py36 http://logs.openstack.org/38/658138/1/check/cross-keystone-py36/834465c/ : SUCCESS in 25m 28s\n- cross-neutron-py27 http://logs.openstack.org/38/658138/1/check/cross-neutron-py27/70798b5/ : SUCCESS in 23m 14s\n- cross-neutron-py36 http://logs.openstack.org/38/658138/1/check/cross-neutron-py36/fa8e5fb/ : SUCCESS in 19m 16s\n- cross-nova-py27 http://logs.openstack.org/38/658138/1/check/cross-nova-py27/2360710/ : SUCCESS in 19m 43s\n- cross-nova-py36 http://logs.openstack.org/38/658138/1/check/cross-nova-py36/7ba40ce/ : SUCCESS in 12m 18s\n- cross-nova-functional http://logs.openstack.org/38/658138/1/check/cross-nova-functional/94ebd99/ : SUCCESS in 19m 12s\n- cross-osc-tox-docs http://logs.openstack.org/38/658138/1/check/cross-osc-tox-docs/8f5f4c1/html/ : SUCCESS in 6m 21s\n- cross-swift-py27 http://logs.openstack.org/38/658138/1/check/cross-swift-py27/876934a/ : SUCCESS in 7m 49s\n- cross-swift-py36 http://logs.openstack.org/38/658138/1/check/cross-swift-py36/6a957b2/ : SUCCESS in 8m 05s\n- requirements-integration http://logs.openstack.org/38/658138/1/check/requirements-integration/a576b54/ : SUCCESS in 53m 30s\n- neutron-grenade-multinode http://logs.openstack.org/38/658138/1/check/neutron-grenade-multinode/775280b/ : SUCCESS in 1h 13m 18s\n- tempest-full http://logs.openstack.org/38/658138/1/check/tempest-full/7c303fb/ : SUCCESS in 1h 48m 11s\n- neutron-grenade http://logs.openstack.org/38/658138/1/check/neutron-grenade/1859a2b/ : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":1},{"id":"80f3f84d35a95027d11194b1d42b356a17df9fc3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-03 21:08:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ef92b4347eb433d41f6c90e1d0548289bdd89609","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 02:18:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/38/658138/1/check/openstack-tox-docs/19ccf3f/html/ : SUCCESS in 2m 35s\n- openstack-tox-pypy http://logs.openstack.org/38/658138/1/check/openstack-tox-pypy/89784eb/ : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/38/658138/1/check/openstack-tox-pep8/73233fa/ : SUCCESS in 2m 59s\n- openstack-tox-py27 http://logs.openstack.org/38/658138/1/check/openstack-tox-py27/520f54c/ : SUCCESS in 3m 04s\n- openstack-tox-py36 http://logs.openstack.org/38/658138/1/check/openstack-tox-py36/714e06e/ : SUCCESS in 3m 38s\n- openstack-tox-validate http://logs.openstack.org/38/658138/1/check/openstack-tox-validate/e42d4c7/ : SUCCESS in 2m 52s\n- requirements-tox-babel http://logs.openstack.org/38/658138/1/check/requirements-tox-babel/29eb4f4/ : SUCCESS in 2m 39s\n- requirements-tox-py27-check-uc http://logs.openstack.org/38/658138/1/check/requirements-tox-py27-check-uc/30d503d/ : FAILURE in 6m 54s\n- requirements-tox-py36-check-uc http://logs.openstack.org/38/658138/1/check/requirements-tox-py36-check-uc/5bab9a2/ : FAILURE in 6m 36s\n- cross-cinder-py27 http://logs.openstack.org/38/658138/1/check/cross-cinder-py27/082b389/ : SUCCESS in 12m 24s\n- cross-cinder-py36 http://logs.openstack.org/38/658138/1/check/cross-cinder-py36/e92054b/ : SUCCESS in 8m 37s\n- cross-glance-py27 http://logs.openstack.org/38/658138/1/check/cross-glance-py27/90bec2e/ : SUCCESS in 5m 10s\n- cross-glance-py36 http://logs.openstack.org/38/658138/1/check/cross-glance-py36/3b0ebea/ : SUCCESS in 5m 01s\n- cross-horizon-py27 http://logs.openstack.org/38/658138/1/check/cross-horizon-py27/875263b/ : SUCCESS in 10m 03s\n- cross-horizon-py36 http://logs.openstack.org/38/658138/1/check/cross-horizon-py36/c80b526/ : SUCCESS in 9m 10s\n- cross-horizon-npm http://logs.openstack.org/38/658138/1/check/cross-horizon-npm/1f02a8b/ : SUCCESS in 6m 02s\n- cross-keystone-py27 http://logs.openstack.org/38/658138/1/check/cross-keystone-py27/7d6f38f/ : SUCCESS in 23m 27s\n- cross-keystone-py36 http://logs.openstack.org/38/658138/1/check/cross-keystone-py36/c2fa5dc/ : SUCCESS in 25m 46s\n- cross-neutron-py27 http://logs.openstack.org/38/658138/1/check/cross-neutron-py27/04e2966/ : SUCCESS in 23m 08s\n- cross-neutron-py36 http://logs.openstack.org/38/658138/1/check/cross-neutron-py36/8f0de64/ : SUCCESS in 25m 34s\n- cross-nova-py27 http://logs.openstack.org/38/658138/1/check/cross-nova-py27/6e480ba/ : SUCCESS in 13m 02s\n- cross-nova-py36 http://logs.openstack.org/38/658138/1/check/cross-nova-py36/a3be1b5/ : SUCCESS in 11m 38s\n- cross-nova-functional http://logs.openstack.org/38/658138/1/check/cross-nova-functional/ed44126/ : SUCCESS in 18m 58s\n- cross-osc-tox-docs http://logs.openstack.org/38/658138/1/check/cross-osc-tox-docs/7838473/html/ : SUCCESS in 5m 08s\n- cross-swift-py27 http://logs.openstack.org/38/658138/1/check/cross-swift-py27/9fec1c2/ : SUCCESS in 7m 43s\n- cross-swift-py36 http://logs.openstack.org/38/658138/1/check/cross-swift-py36/9a2bdf8/ : SUCCESS in 8m 01s\n- requirements-integration http://logs.openstack.org/38/658138/1/check/requirements-integration/cfd6697/ : SUCCESS in 57m 00s\n- neutron-grenade-multinode http://logs.openstack.org/38/658138/1/check/neutron-grenade-multinode/329856e/ : SUCCESS in 1h 13m 08s\n- tempest-full http://logs.openstack.org/38/658138/1/check/tempest-full/1f76b8a/ : SUCCESS in 1h 35m 32s\n- neutron-grenade http://logs.openstack.org/38/658138/1/check/neutron-grenade/4a87c9b/ : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b433dfc4c83f0d8ff0c69535f33101be5d59e6b0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-16 20:49:19.000000000","message":"Abandoned\n\nbeen bumped","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f2b00fe4ee166d1b2e7ad9862c9033f946f8be3","revisions":{"9f2b00fe4ee166d1b2e7ad9862c9033f946f8be3":{"kind":"REWORK","_number":1,"created":"2019-05-09 15:37:38.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/38/658138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/658138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/658138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/658138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/658138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/658138/1"}}},"commit":{"parents":[{"commit":"75e7b4dc3710849125a3a1ef1cc43b7bbbaeee1e","subject":"Uncap grpcio","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75e7b4dc3710849125a3a1ef1cc43b7bbbaeee1e"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-09 15:37:08.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-09 15:37:08.000000000","tz":0},"subject":"Bump grpcio upper-constraint","message":"Bump grpcio upper-constraint\n\nChange-Id: Ida13244fc71b6057fdd083e3640033dc74b2f4f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f2b00fe4ee166d1b2e7ad9862c9033f946f8be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f2b00fe4ee166d1b2e7ad9862c9033f946f8be3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
